)]}'
{"id":"openstack%2Fkolla-ansible~844321","triplet_id":"openstack%2Fkolla-ansible~master~Ie34a883ac7ae989d7f74c19b9c9ee9f08a82a35f","project":"openstack/kolla-ansible","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-06-02 09:36:32.000000000","reason":"Change was submitted"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-06-02 09:36:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie34a883ac7ae989d7f74c19b9c9ee9f08a82a35f","subject":"Switch to Cortex-A72 cpu on AArch64","status":"MERGED","created":"2022-06-01 17:11:03.000000000","updated":"2022-06-02 09:37:44.000000000","submitted":"2022-06-02 09:36:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"844321","meta_rev_id":"7a7b4880bc2ba3a8b057db3469922dc61e5bb1e3","_number":844321,"virtual_id_number":844321,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-02 09:36:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-06-02 08:13:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2022-06-02 08:18:45.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2022-06-02 08:18:45.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-01 18:34:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-01 18:58:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-02 08:13:10.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-06-02 08:18:45.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"17289b07e3e52661562a25f59ee592e65c324fcd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-06-01 17:11:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b56c3c7fea4fdeca743a9f0893d2e46db1aa823f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-06-01 17:38:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"22f5a5d8cbdb57c3559ea5c249e5c55ce1f4132a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 18:34:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/30f0b770c0b94e2cbbf5879a34793e25 : SUCCESS in 1h 20m 04s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/bcbf3dc427b44a60a00d7d784c6e263d : SUCCESS in 1h 17m 38s","accounts_in_message":[],"_revision_number":1},{"id":"8b058da8f2c763ab7cae71013fdfa76e248351e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 18:58:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a1cc6b16b41644dbb8bc46578559ba87 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e1cc10505884a76b61cb7e46b08a01a : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0335f51c557a4b35a36742cac951a78c : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/204a15bb2fb048e188423f374629aec1 : SUCCESS in 6m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fc5bef5fa3ce4ef49fc8a95042c1174b : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec43a36717cb443cac4037fa7c5f5782 : SUCCESS in 6m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ffc1d079ea8646f4861d817589347a3d : SUCCESS in 8m 02s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/bd79d77261e941b28e634fcd91f80bbc : SUCCESS in 58m 35s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0f00bceaa30e4888ac6f7e751b061440 : SUCCESS in 48m 21s\n- kolla-ansible-openeuler-source https://zuul.opendev.org/t/openstack/build/9561a959bcd04e34b1b4371cd3a3c957 : SUCCESS in 1h 09m 28s (non-voting)\n- kolla-ansible-rocky8-source https://zuul.opendev.org/t/openstack/build/e194330989ac4376aeccb8d2701c4339 : SUCCESS in 54m 47s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/05eda23a0b574ea8a22c0a9c7eeface9 : SUCCESS in 43m 50s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/dd0caf8444704a9d933fe9f30818b450 : SUCCESS in 1h 03m 16s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/116b26fce40444f4a7c64fc5ff15a5db : SUCCESS in 49m 11s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d7b3128545ca4470b122867a19655709 : SUCCESS in 1h 09m 44s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/0b8fca7b2d0545ec97fc86634442754b : SUCCESS in 59m 02s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a45bd3839abc4c25be608cbf1ae8db58 : SUCCESS in 50m 59s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/dd83f2e1a2364879b6392743ebef68b3 : SUCCESS in 58m 47s\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/29ff77a38ee44cae8de0d44e4d3601d7 : SUCCESS in 1h 10m 54s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/f2021a784cfe4212a31aa0ee00bb7b62 : SUCCESS in 1h 08m 55s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/7c51af609992475bab153403337e9413 : SUCCESS in 1h 45m 44s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/43d8b9d8908d4a19b2c26b3cac7c477a : SUCCESS in 1h 28m 55s (non-voting)\n- kolla-ansible-centos8s-hashi-vault https://zuul.opendev.org/t/openstack/build/a713b50b4012465197ccee7c9498418f : SUCCESS in 3m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"43905b366c4ebcd0b23bb41b7e4cd0b271c80aba","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-02 08:13:10.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d8307494d56f1120e431af323531208d0f5306c8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-06-02 08:18:19.000000000","message":"Patch Set 1:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"98192e48a0787f750474c1f60296eeaa9339b6d0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-06-02 08:18:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"65fbeff6db40de3224b1b0191fb624cc5ee9e636","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 08:19:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4d6d6c9048dcfdefbe66736e362f89018811b5d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 09:36:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dc24811dcf346fd8634e298a1413d35 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/10c7ead185e1433aa7748354de2613e0 : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63a753f59058413eb35bd577405d9b7a : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/452b1c9cedd84f4c87f7bbc6030bcc99 : SUCCESS in 5m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/37d9e0edab7040ce9c67d11c0c4ebc63 : SUCCESS in 7m 10s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/38f3d001323d43b98028a4b336cfc796 : SUCCESS in 54m 15s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/b219520b6892480fbc4cdcfa93ceab52 : SUCCESS in 1h 15m 49s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/72a314dea6c44aa182348eddc47d5337 : SUCCESS in 44m 24s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/7e0808a0e2094540ba26e2abb83f6eed : SUCCESS in 1h 02m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/11c1c43b1e4c4b449edbd3c592eacaa9 : SUCCESS in 45m 03s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/431e69dc43b448e39664ef7903e4fb52 : SUCCESS in 58m 46s","accounts_in_message":[],"_revision_number":1},{"id":"891d97264288b809496160f639270f1f4fbca2f5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 09:36:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7a7b4880bc2ba3a8b057db3469922dc61e5bb1e3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 09:37:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/718747de3a5a4a75b97a05e15d57ebbc : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c9a3dd9cd3c004e61836a252aa34b65b89038bee","revisions":{"c9a3dd9cd3c004e61836a252aa34b65b89038bee":{"kind":"REWORK","_number":1,"created":"2022-06-01 17:11:03.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/21/844321/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/844321/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/844321/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/844321/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/844321/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/844321/1"}}},"commit":{"parents":[{"commit":"5d28a7c89bbac4a15f7d8f7a64c3767ea988efbd","subject":"masakari: support libvirt SASL in instance monitor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/5d28a7c89bbac4a15f7d8f7a64c3767ea988efbd"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-06-01 17:09:52.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-06-01 17:09:52.000000000","tz":120},"subject":"Switch to Cortex-A72 cpu on AArch64","message":"Switch to Cortex-A72 cpu on AArch64\n\nCPU model \"max\" fails to boot CirrOS and some other images when QEMU 7.0\nis used.\n\nChange-Id: Ie34a883ac7ae989d7f74c19b9c9ee9f08a82a35f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c9a3dd9cd3c004e61836a252aa34b65b89038bee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c9a3dd9cd3c004e61836a252aa34b65b89038bee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
