)]}'
{"id":"openstack%2Fkolla-ansible~863321","triplet_id":"openstack%2Fkolla-ansible~master~I6defbc0a914065d7f22d46d883cc4a71ead8f9a8","project":"openstack/kolla-ansible","branch":"master","topic":"bp/implement-automatic-deploy-of-trove","attention_set":{"28752":{"account":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"last_update":"2025-05-09 08:13:49.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-01-08 10:20:03.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-05-09 08:13:49.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"36201":{"account":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"last_update":"2025-05-09 08:13:49.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2025-05-09 10:48:05.000000000","reason":"A robot voted negatively on a label"},"32029":{"account":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"last_update":"2023-12-04 08:20:02.000000000","reason":"Vote got outdated and was removed: Code-Review+1"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I6defbc0a914065d7f22d46d883cc4a71ead8f9a8","subject":"Implement automatic deploy of trove","status":"NEW","created":"2022-11-02 15:22:26.000000000","updated":"2025-05-09 10:48:05.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":20,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"d3d268386cf4a3f062b83ee2f4c4d7423ccf8137","_number":863321,"virtual_id_number":863321,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-05-09 10:48:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"all":[{"value":1,"date":"2025-05-09 08:13:49.000000000","_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"value":-1,"date":"2025-05-09 08:13:49.000000000","_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-02 15:34:40.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2022-11-02 17:16:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-02 13:33:46.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-12-04 07:26:52.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2024-01-07 20:46:55.000000000","updated_by":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"reviewer":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"state":"REVIEWER"},{"updated":"2024-01-08 01:55:50.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"},{"updated":"2024-01-18 13:03:54.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"}],"messages":[{"id":"f83a186d41aa222feaa85d20a1d54a8fb61ab8c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-11-02 15:22:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"579806e2c2958896e486bb3254d930965f78733c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-11-02 15:34:40.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f9d11fcd9af0fdc744486e6d196e68362a818d4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 17:16:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/46cef53c029941c8a88be61d1a23f8cb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9bc233dbc8e4107ab5b5892c7afe3ae : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb7a632758ba4c34ae31bdc4d10e1516 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d466b8381854f8583f8b9c2fd126b02 : SUCCESS in 6m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac6787e8e9034b8f9620900975fc0702 : SUCCESS in 5m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3cb67b6669234491adb681966e5d2182 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/705d4fd9094a4f7db0faf836a2de1ef5 : SUCCESS in 4m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/15626f1b4666481183e251ddb1e609cb : FAILURE in 5m 27s\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/2df2f63474524dcaa0ca77b5a82a65e1 : SUCCESS in 1h 06m 44s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b0913feaf17f4a37b1ec548a3da58bd8 : SUCCESS in 49m 06s\n- kolla-ansible-openeuler-source https://zuul.opendev.org/t/openstack/build/0d0fbd8d5da74596a40b7e1a155dc0f7 : SUCCESS in 56m 27s (non-voting)\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/fdd53bf954454bb896e20c2e251406c5 : SUCCESS in 53m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f419b1344ddf44d9b3f87f358d0305c4 : SUCCESS in 45m 18s\n- kolla-ansible-rocky9-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/6e9077debe20430f9ab245c0a1354f73 : SUCCESS in 51m 32s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/dfe650deecb04b409587d28d3edc7946 : SUCCESS in 41m 05s (non-voting)\n- kolla-ansible-rocky9-source-magnum https://zuul.opendev.org/t/openstack/build/ad97d9ad2c5d464aa02d2d217a0690bb : SUCCESS in 53m 00s (non-voting)\n- kolla-ansible-ubuntu-source-magnum https://zuul.opendev.org/t/openstack/build/956119bffb144053bd4a18a4e8f13edc : SUCCESS in 41m 18s (non-voting)\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/199b6dbe65394bd386eebb5ea5a5958d : SUCCESS in 58m 28s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/21a3047b021a4fa68fdce86350e3bcbb : SUCCESS in 40m 34s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/ff62c60fac37417b99b279286b765373 : SUCCESS in 52m 56s\n- kolla-ansible-rocky9-source-cephadm https://zuul.opendev.org/t/openstack/build/7dfd30cb558046a6a2f8ace4dee42fe6 : SUCCESS in 1h 16m 34s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/10bc303435f8450f96375b3bea10b565 : SUCCESS in 50m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/b40298f49b1941afb70a13b2ada0aed3 : SUCCESS in 1h 14m 06s (non-voting)\n- kolla-ansible-rocky9-hashi-vault https://zuul.opendev.org/t/openstack/build/fa8d8b655f0742e68decb8b3645ec278 : SUCCESS in 3m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1fa772de2ee335b6ba6239b9e84d6dd80cbe5aa6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 17:21:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb61a2f69c5b46de88962fc93bd93907\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8de774072f9143dab722d290eef76f80 : SUCCESS in 1h 23m 54s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/1767df60ec95463483de94b719492b7f : SUCCESS in 1h 56m 49s","accounts_in_message":[],"_revision_number":1},{"id":"d562eda3502386e6c7ebd6b6b67d708b2cd510ac","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-11-03 11:28:34.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"1b3305e63f86612c64efd56bd4fdb70dedf50150","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-11-03 11:43:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"428de1345feffe647154663c215e0b716576f46e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 13:04:16.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52fd94b0bc2e4953b2acd513e3de9355\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0a7ed9d7263a4609bd45b398007c3216 : SUCCESS in 1h 01m 42s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/bf475c80f7434e27b9aaab9dd6504947 : FAILURE in 1h 15m 47s","accounts_in_message":[],"_revision_number":2},{"id":"14ee8dd3cb7c5a9b1db9200e604353c985bbf935","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 13:44:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd5a3e50655545aca697bc598de9384c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de4d3f17eb054c4fbbeff229c0d05d42 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e33537e93c4c42a9ad202d275ad8aaee : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d28da6f3638644d285194a8cc8e7bb5f : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3e57d0415304765bc55633fb9b7eb40 : SUCCESS in 5m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/409cb661b3cd4d2c846b566e72e6dc4d : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5424bee2e81b474bb186d80a8acbb82f : SUCCESS in 5m 26s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6a4cced5769143e2bd9378c0a1afa5fe : SUCCESS in 7m 13s\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/4f650f74df5a472da7c5ede0db6b4d38 : SUCCESS in 1h 07m 12s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/957bc88128a540e58709dca09bcaf46a : SUCCESS in 47m 14s\n- kolla-ansible-openeuler-source https://zuul.opendev.org/t/openstack/build/4f7d3eb0c3c44962a6982f46be6436cd : SUCCESS in 56m 26s (non-voting)\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/9746c064ec1f4dc489dcf0f04696d291 : SUCCESS in 1h 18m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/98a1417d8800490cadad0557e8cd4aba : SUCCESS in 39m 49s\n- kolla-ansible-rocky9-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/08a699a177324f6b85d0cc8f10a32c54 : SUCCESS in 55m 53s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/c4d012730eba4e92a2faed2775b69109 : SUCCESS in 43m 15s (non-voting)\n- kolla-ansible-rocky9-source-magnum https://zuul.opendev.org/t/openstack/build/900929e7485b45109105efadd0ea08a1 : SUCCESS in 53m 50s (non-voting)\n- kolla-ansible-ubuntu-source-magnum https://zuul.opendev.org/t/openstack/build/3a0d575fcbff4942982b96e304293a71 : SUCCESS in 39m 26s (non-voting)\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/57c2cb95a4c94d5eb084931167c6cbb9 : SUCCESS in 1h 01m 41s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c36b67f4941142dab283acd92d61de3d : SUCCESS in 54m 53s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/0ef567a6caa4431eb36d7fcdaa486fb8 : SUCCESS in 57m 55s\n- kolla-ansible-rocky9-source-cephadm https://zuul.opendev.org/t/openstack/build/44cee2744535491da4b6a13e667fb6f8 : SUCCESS in 1h 16m 49s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/6e8dad42ffc54089b6fa8ae4224bfe14 : SUCCESS in 1h 01m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/ca98b882db7e4e5fa54a3bbbe5e53638 : SUCCESS in 1h 55m 24s (non-voting)\n- kolla-ansible-rocky9-hashi-vault https://zuul.opendev.org/t/openstack/build/a41c3421530f46b9b899f9854ab159f7 : SUCCESS in 2m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"35edc0ca6a9c3c205eb72824465a5a6d36c7951c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-12-30 02:06:28.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"d258c98abac840e648199d428201fc948ae8a5d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-12-30 03:02:03.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"617666d30d602edacf21bf6bd8f867b20a2a4d09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-30 04:28:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7519a00b1614f7bb9c36b7ca04727eb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec30f36f13364fd195b018ea24283621 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/552c1d7f71d546ada95e013b7d289a85 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c3bdc6f17384bdea87f48c9b771b030 : SUCCESS in 5m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5574d872867748d5887cf17c805533dc : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f440ad4b539f4960b9933cb1c7246056 : SUCCESS in 4m 49s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5a0eca4a86084adc9a5e55e188339965 : SUCCESS in 8m 27s\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/57fb039c9f044815b14e5c5f0178c411 : SUCCESS in 1h 06m 09s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/90b7576f31d34444a8c53db29b0b7222 : SUCCESS in 57m 20s\n- kolla-ansible-openeuler-source https://zuul.opendev.org/t/openstack/build/ea500f6f69554e6e959b0ace5eb63b02 : FAILURE in 16m 45s (non-voting)\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/8157a44fb2ad4e8597cf59a9e542b606 : SUCCESS in 1h 02m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/078845e7136246e49d785678fd1b5b8a : SUCCESS in 44m 23s\n- kolla-ansible-rocky9-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/274225e9024e415eb704689c3310249b : SUCCESS in 1h 00m 55s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/367a0ea5956c40b3b7d4fa4807a1f09b : FAILURE in 9m 07s (non-voting)\n- kolla-ansible-rocky9-source-magnum https://zuul.opendev.org/t/openstack/build/9bd38c2fc43c4960b2d0d135551810b2 : FAILURE in 37m 22s (non-voting)\n- kolla-ansible-ubuntu-source-magnum https://zuul.opendev.org/t/openstack/build/f3d2248a1a0445d7a040fe2f9a3cc8fb : FAILURE in 29m 41s (non-voting)\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/0e42b9fdcbae4c9c82eee8be9860ac12 : SUCCESS in 1h 03m 48s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/61c00a8bf6504082915c415575bc569b : SUCCESS in 1h 00m 37s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/af4f7472785a46c397a78f1eb76ba0c7 : SUCCESS in 58m 27s\n- kolla-ansible-rocky9-source-cephadm https://zuul.opendev.org/t/openstack/build/63b26c559bc04be88176f4f264072120 : SUCCESS in 1h 17m 28s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/caa290161849475c9e64c69c029d9260 : SUCCESS in 1h 00m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/e009b8e9a6a1450a906ccb99655d6751 : SUCCESS in 1h 22m 32s (non-voting)\n- kolla-ansible-rocky9-hashi-vault https://zuul.opendev.org/t/openstack/build/e8de5564d3a945a8b85e582de4eddeb2 : SUCCESS in 3m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"719e8f3c660f07e41666116c371b968638389ed0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-30 04:57:15.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b69473529164747a39d635deba6921c\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5b1e4abfd7aa452ba867b622f3685d88 : SUCCESS in 1h 22m 31s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/9388012af0e8452d95107037814dac69 : SUCCESS in 1h 46m 11s","accounts_in_message":[],"_revision_number":4},{"id":"7fbfbc0649b5f04973cd3489081e810f2b7c7100","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-01-05 15:17:07.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7b47250eb3b8745ed8b298e194f8d666138e4cfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 17:00:05.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d15c9e5a86b4ae8a4ccd3df0c84db59\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a7dce00d28d43aea34155a9789bb881 : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d4a1b08fe304e50af4369d47681f9c7 : SUCCESS in 7m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df1e19de79bd41d38488ca3f207c8612 : SUCCESS in 5m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4277a47c59e24d8eb232761470c65a79 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5adc56a929e44023b7e3e5050c30f058 : SUCCESS in 4m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1deb36858f6f4d8d904e591ba0503106 : SUCCESS in 9m 23s\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/eb189f1cfae44dce8adb0574fbba7959 : SUCCESS in 1h 19m 27s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/49f249ad273e43ffb87c2c0dffb6ca38 : SUCCESS in 54m 21s\n- kolla-ansible-openeuler-source https://zuul.opendev.org/t/openstack/build/f03a3dc059bf4f84bbc4603f3ffb759e : SUCCESS in 58m 37s (non-voting)\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/a183da1f9b81448bb7bee1c4ba2508bf : SUCCESS in 57m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9c95829c5e9a424b9ea60ae744abc79b : SUCCESS in 44m 27s\n- kolla-ansible-rocky9-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/eb32792287b4444d8cde5a3962f300f1 : SUCCESS in 58m 11s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/8865577f9bf548e5963ca8ae08504cd2 : SUCCESS in 51m 04s (non-voting)\n- kolla-ansible-rocky9-source-magnum https://zuul.opendev.org/t/openstack/build/da9f425f51014ce29f2be56d405ed2ff : SUCCESS in 1h 04m 42s (non-voting)\n- kolla-ansible-ubuntu-source-magnum https://zuul.opendev.org/t/openstack/build/c947db13e7c64b68b756e5bf6147999f : SUCCESS in 38m 33s (non-voting)\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/1c0609f2c2414dd3b3fe21a9887dad19 : SUCCESS in 55m 25s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/143e1a5c15614e7192cf8880a79c60c1 : SUCCESS in 44m 45s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/746003768af14ca6aed31c31de407cec : SUCCESS in 1h 19m 50s\n- kolla-ansible-rocky9-source-cephadm https://zuul.opendev.org/t/openstack/build/110034f08a4f4c0aa8697bb909ac6a00 : FAILURE in 52m 50s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/7d180eaabfd948de82508b0931f3d043 : SUCCESS in 1h 05m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/96e3012aa8ff48e38f6771f1a854257f : SUCCESS in 1h 31m 09s (non-voting)\n- kolla-ansible-rocky9-hashi-vault https://zuul.opendev.org/t/openstack/build/39a4c50afc5d4e16bf4058268c65ae1e : SUCCESS in 2m 56s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c6f04cb5232cd777c97236ba0e277bea59e72e0e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 17:03:00.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9dac45056d44e1fb068f919214a42b5\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b0f7f8709f3c478ca26a2c8fe00c254a : SUCCESS in 1h 07m 42s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/6b4be4c1863c421ba83fb01b5ee25930 : SUCCESS in 1h 23m 32s","accounts_in_message":[],"_revision_number":5},{"id":"0ceee7af73cb21b6f800e2a9d6f93991526f6cc6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-17 07:24:52.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nThe following files contain Git conflicts:\n* etc/kolla/passwords.yml\n* kolla_ansible/cmd/genpwd.py\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"d3097587509e2e0cab67509b3bc7177583565c2b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-17 07:26:04.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"bd1ead14b10241178fb2d8cbd58bd583ede6e38e","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-17 07:26:26.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":7},{"id":"7ab2025eb4c7ee9765fde09c2ddf0f0049d91a0a","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-11-17 08:25:01.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"92bf16c053009a95760a3c9051748d6d2f482433","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 08:53:29.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42b8f761bc0d45f68ba2704774fdd432\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/44bd8f616ea0464fa34c61847f6c5b86 : SUCCESS in 1h 15m 16s\n- kolla-ansible-debian-aarch64-podman https://zuul.opendev.org/t/openstack/build/d3c18efe15604a51927bc24d58ad8921 : SUCCESS in 1h 09m 24s\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/f3ac524c032f44ce9d6ac9e21dbc88db : SUCCESS in 1h 26m 34s","accounts_in_message":[],"_revision_number":7},{"id":"e0cb7e66d7cb68bb39ae7c9e8b94e53408488c53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 09:54:17.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1924c54ce3d4d8b81fe9c1d2928c500\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4c7502ba68b4a62aa09d0a9af95ef5e : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2711697c79cc4b628e17be4304422918 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cda23da00c1a421eb8a03559303a8541 : SUCCESS in 3m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/681555538e4844be8d2ea91bbaca8114 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/664b4224b3db45fc97da2618ac1f99d3 : SUCCESS in 4m 57s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d8206667b9164108b651668d9bfd5d6d : SUCCESS in 6m 34s\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c3de15fac46f4d838684936cdc975496 : SUCCESS in 1h 02m 48s (non-voting)\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/7ef98ff77c02416fbfd4069a6dcdef79 : SUCCESS in 44m 42s\n- kolla-ansible-debian-podman https://zuul.opendev.org/t/openstack/build/cf4950e581e9487496803bfb6f64edab : SUCCESS in 1h 01m 35s\n- kolla-ansible-openeuler https://zuul.opendev.org/t/openstack/build/bdc4e75261cc4a4380043dcf84dbd04c : SUCCESS in 49m 38s (non-voting)\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ba45f54aa0a1442cab1ee49f38ff91c0 : SUCCESS in 50m 04s\n- kolla-ansible-rocky9-podman https://zuul.opendev.org/t/openstack/build/c39a86df0abb439aa05e2b9eee61206d : SUCCESS in 1h 02m 36s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/baac190050b74b40a77af984581d1ab9 : SUCCESS in 49m 08s\n- kolla-ansible-ubuntu-podman https://zuul.opendev.org/t/openstack/build/72c5af5efdb24b1c8d58b7e0ac3fec61 : SUCCESS in 1h 03m 43s\n- kolla-ansible-rocky9-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/13ac062ea5cc449ca4642a00917d489d : SUCCESS in 1h 13m 15s (non-voting)\n- kolla-ansible-ubuntu-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/db947f8757d049d28ffde39e4b99e1cf : SUCCESS in 56m 11s (non-voting)\n- kolla-ansible-rocky9-magnum https://zuul.opendev.org/t/openstack/build/f8daafe49a724eb193959a64aa22fcf4 : FAILURE in 33m 45s (non-voting)\n- kolla-ansible-ubuntu-magnum https://zuul.opendev.org/t/openstack/build/80da342e0faf4e9eaa488ae833f68b85 : FAILURE in 24m 52s (non-voting)\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b24e91d9a6294799aebefad720301536 : SUCCESS in 1h 25m 02s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8a6e4a9ec1e84d45bc993a7ed2390bfa : SUCCESS in 1h 05m 02s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/bc0e081d22fd4f43bda5b88b12b55feb : SUCCESS in 1h 19m 12s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/33908aa67ff44843b772db7435a5bd2e : SUCCESS in 1h 16m 22s (non-voting)\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/96783732000142099a8618affebe01ac : SUCCESS in 1h 23m 44s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/68caa9e24d18449ba4496b7d17085b8a : SUCCESS in 2h 21m 00s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/31d1c4be183447d3984bb6ebc3915cf3 : FAILURE in 40m 58s (non-voting)\n- kolla-ansible-rocky9-hashi-vault https://zuul.opendev.org/t/openstack/build/5c7b2679b13246388c5d0a1a4775220b : SUCCESS in 3m 46s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b11d73f16202b564223ad9a66a2f2567ad11e3d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-04 05:54:12.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"b95e8ae62f4451245e5a2aae7b01cbf8cfae6c6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-04 06:03:54.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"726f809dc42e16f140f2dc33aac8e0f90fa4aeab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 07:19:01.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2958e6529a24398b6d71cb78b4bd853\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/747139e383834b0f864f718cea4b94c3 : SUCCESS in 1h 00m 10s\n- kolla-ansible-debian-aarch64-podman https://zuul.opendev.org/t/openstack/build/57f62274690445a282d68059cded1e9f : SUCCESS in 1h 10m 24s\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/32b1ab8dff0c4be3813229881bfc06a8 : SUCCESS in 1h 14m 14s","accounts_in_message":[],"_revision_number":9},{"id":"9148fe335a4db3ba6db760c00131436847356354","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-12-04 07:26:52.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"1f7b930e63b12122e448e444bc140aa7918d8c95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-04 08:20:02.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":10},{"id":"f6e57cf7729d29333d4a2c060739b770119b1017","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 09:57:06.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f93bbd999e354743bf098f922cc86c18\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/915d891ac366416aa3b2c447282b9080 : SUCCESS in 54m 47s\n- kolla-ansible-debian-aarch64-podman https://zuul.opendev.org/t/openstack/build/cced4a4680064fd78eb65d6e07e405fc : SUCCESS in 1h 12m 39s\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/757aa74910c74d03846cca3fb78aed89 : SUCCESS in 1h 32m 33s","accounts_in_message":[],"_revision_number":10},{"id":"befc32a0e301e4be76b5aa92ff8d038b73742223","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 11:00:53.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53273b8afb0847a6b88e1d6ff951065a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c50018e1fa0c45628beb919d2398ccb9 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c7d1984889d442098720b4cae848672 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3990813b164142b68b0e4b4c130b64b6 : SUCCESS in 3m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1004681b71cb40239eb01f74bb2244e6 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65c4a8d217614f93b68a39831b71c804 : SUCCESS in 5m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7825af2eef524cdcab2bda3a3b6c9196 : SUCCESS in 6m 19s\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/1432c6b92708438da362c78da2151812 : SUCCESS in 55m 35s (non-voting)\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/15e14d6c58a74fb2bef71604a5d8b463 : SUCCESS in 45m 12s\n- kolla-ansible-debian-podman https://zuul.opendev.org/t/openstack/build/aa1d4fd9758e463597ce2921f0990200 : SUCCESS in 56m 40s\n- kolla-ansible-openeuler https://zuul.opendev.org/t/openstack/build/9aeefc78b12e4b8394ac4edaa8d6ea62 : SUCCESS in 49m 21s (non-voting)\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/172af91a31f44463881173e38ff6a92d : SUCCESS in 54m 50s\n- kolla-ansible-rocky9-podman https://zuul.opendev.org/t/openstack/build/2285c177011e4ac39c535c5c161335ae : SUCCESS in 59m 50s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/7e9a98f902314bea902a58930492644f : SUCCESS in 38m 53s\n- kolla-ansible-ubuntu-podman https://zuul.opendev.org/t/openstack/build/868c644de1ba4b12bdc7a1c06691ad90 : SUCCESS in 55m 12s\n- kolla-ansible-rocky9-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/537f02a66d304a6da7c06067e43d8ab7 : SUCCESS in 1h 00m 03s (non-voting)\n- kolla-ansible-ubuntu-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/072ac6042232479a870dffdd30b70de6 : SUCCESS in 42m 53s (non-voting)\n- kolla-ansible-rocky9-magnum https://zuul.opendev.org/t/openstack/build/0d3679c3047440448d1892ea25325caa : SUCCESS in 1h 13m 15s (non-voting)\n- kolla-ansible-ubuntu-magnum https://zuul.opendev.org/t/openstack/build/12fb446d76194dafb68fbc584b6866f4 : SUCCESS in 1h 01m 43s (non-voting)\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/684d23580ef04fc695cef4b5f873b0dd : SUCCESS in 1h 17m 13s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ca912d236f0e404382ce8802fa82ca8f : SUCCESS in 1h 07m 47s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ab38d1d62230481486b27257da77f896 : SUCCESS in 1h 21m 34s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/931186295f124e8286d88737741db845 : SUCCESS in 1h 39m 35s (non-voting)\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/fe32a28044da48ba91090a1facfebebb : SUCCESS in 1h 05m 29s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/14d788b58e6c417d80458c90d7b334e2 : FAILURE in 2h 19m 18s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/84f2eb7a13424a7a9f08c35af6e8d206 : FAILURE in 2h 29m 18s (non-voting)\n- kolla-ansible-rocky9-hashi-vault https://zuul.opendev.org/t/openstack/build/47e5aa7caba54b0ba7dfe8b70ea00ad0 : SUCCESS in 3m 14s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"3ac909e061beac98bc569707104339605e62cde7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-11 03:14:26.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"ca3f9c43f5d0af9de34a62231050dca5b6290072","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-11 05:21:25.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1828a9ef24ba466f8598357a88dd6d93\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/673af28238de495ba67a35695c1ea170 : SUCCESS in 1h 34m 17s\n- kolla-ansible-debian-aarch64-podman https://zuul.opendev.org/t/openstack/build/09639b2297414053a41dae1bb7cbdebc : SUCCESS in 1h 45m 52s\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/fddc2d8e6d7045079bb2fb02c665f344 : SUCCESS in 2h 04m 37s","accounts_in_message":[],"_revision_number":11},{"id":"c7306073a74b15433cd2e629e9cc58a3e918ea4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-11 05:33:45.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e19befbe0be4480cb9b86dc799508f5a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ebd04d6b37e4ef0aea1356ebef984c2 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06701e3003ed4092943673b5b299ad40 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c040d5ff3cf140caa0f8b1a8c1d7cfeb : SUCCESS in 3m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/859a0bbee9e1450caa26684e86891c38 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adb74000b00b4c49879e80241afcc945 : SUCCESS in 5m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7e3cc59ca29c4df3845571d8a24cc8b2 : FAILURE in 4m 42s\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/5a019f63535f4423945f0f01a17a935f : FAILURE in 9m 24s (non-voting)\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/7e1c0455de434d97839b14c25ed19dc4 : SUCCESS in 42m 53s\n- kolla-ansible-debian-podman https://zuul.opendev.org/t/openstack/build/6a4cc69a532144619afb19e23f37c3aa : SUCCESS in 1h 09m 37s\n- kolla-ansible-openeuler https://zuul.opendev.org/t/openstack/build/797d1bc12401485c872a6a6de5958f27 : SUCCESS in 47m 26s (non-voting)\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9caaff09ec344ac28e2fca91e626854a : SUCCESS in 58m 43s\n- kolla-ansible-rocky9-podman https://zuul.opendev.org/t/openstack/build/17f732dd35b34e41b4ef17cb9acd018f : SUCCESS in 1h 09m 34s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/27105707bea94295a15c779cd3d44262 : SUCCESS in 39m 50s\n- kolla-ansible-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b563765cedb94b9198ec266b6774ebe6 : SUCCESS in 55m 01s\n- kolla-ansible-rocky9-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/cec6baca6c144d46832beaf65edd817c : SUCCESS in 1h 10m 01s (non-voting)\n- kolla-ansible-ubuntu-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/86d5ffcf06764a3589977cd7843c8e17 : SUCCESS in 50m 22s (non-voting)\n- kolla-ansible-rocky9-magnum https://zuul.opendev.org/t/openstack/build/b3473ea4af44490b853af883aaabc6b1 : SUCCESS in 1h 03m 28s (non-voting)\n- kolla-ansible-ubuntu-magnum https://zuul.opendev.org/t/openstack/build/3ca4961a3f7141989672a557aa7bad4e : SUCCESS in 41m 27s (non-voting)\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/851111e01284440587ea4dae1a9b4ba9 : SUCCESS in 1h 25m 38s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/45523f4337b540879641b55151db48b8 : SUCCESS in 1h 19m 02s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/238e6ca005bf40d385a7164604d481cf : SUCCESS in 1h 14m 32s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/2d7edad57cfd4098837e2c9099291626 : SUCCESS in 1h 25m 47s (non-voting)\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/08e835d74f874c9d83dc1bb47de491de : SUCCESS in 1h 34m 26s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/a6b33b2b71214dd79b5fbd0f4c8dd49b : SUCCESS in 2h 09m 48s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/6172276431054729b5bf1459b6aa8235 : SUCCESS in 2h 08m 28s (non-voting)\n- kolla-ansible-rocky9-hashi-vault https://zuul.opendev.org/t/openstack/build/8fba36ba50b54689bba1bc8bbc60fff2 : SUCCESS in 4m 04s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"a2a17cb7076fb9b8e03219aa26230ac141112a88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-11 06:36:52.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"5fdb289359d0ddd525778683424f67b70ddbbba4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-11 08:08:40.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6082f769699d458487180509fd442494\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8399cf6855d948a68c17dd4b5a853d0b : SUCCESS in 50m 17s\n- kolla-ansible-debian-aarch64-podman https://zuul.opendev.org/t/openstack/build/1818cf8cba9a419e898896e1eb472603 : SUCCESS in 1h 10m 44s\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/b29dbb9b42b44c729cbfa09b60aaef57 : SUCCESS in 1h 30m 54s","accounts_in_message":[],"_revision_number":12},{"id":"7a892603e2871734ddd95cf772af2baba7b45888","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-11 09:16:30.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46ee6fd1ecb545ffaf1049ad9d0c9661\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ddb9199eec640d79ba7b60904c18150 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5784e599a8c45b2acd772e9b176164b : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0e355916796439a91453834baf8d64a : SUCCESS in 3m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/51210007c874469eb1c4e9e9c5bf07a8 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6303f39e691646c286f9eec31d356128 : SUCCESS in 4m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/da3068e9d8bb4b64bc945aba0bd71c5d : SUCCESS in 6m 19s\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/9939f7e98b814b3cbaaf31a72cba8648 : FAILURE in 9m 14s (non-voting)\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1a9887f8a45e43c29e47e77211c01863 : SUCCESS in 51m 56s\n- kolla-ansible-debian-podman https://zuul.opendev.org/t/openstack/build/f907440d13d148a680bee9d06a50116e : SUCCESS in 58m 09s\n- kolla-ansible-openeuler https://zuul.opendev.org/t/openstack/build/f6365a7e8ee748d5ba27d60d08dbc55d : SUCCESS in 1h 01m 28s (non-voting)\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/38c7b139552c44d4818abc38ee3ce5f4 : SUCCESS in 57m 23s\n- kolla-ansible-rocky9-podman https://zuul.opendev.org/t/openstack/build/2b4083c0be7b4ab99b2cfa2f62a5df0f : SUCCESS in 1h 03m 58s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/87ada255e5464d3f9e0704bb2a84c9b4 : SUCCESS in 51m 41s\n- kolla-ansible-ubuntu-podman https://zuul.opendev.org/t/openstack/build/672a30dc66fa479fb1dbe27dbc8b49d8 : SUCCESS in 1h 00m 13s\n- kolla-ansible-rocky9-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/4d4c85aac5644f04abcf2a7d0754e3a3 : FAILURE in 1h 04m 33s (non-voting)\n- kolla-ansible-ubuntu-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/94bf530aae294629b70bc2c994f80b2c : SUCCESS in 47m 11s (non-voting)\n- kolla-ansible-rocky9-magnum https://zuul.opendev.org/t/openstack/build/0fc49536ecac4ddfac8494bf298db0f7 : SUCCESS in 1h 06m 29s (non-voting)\n- kolla-ansible-ubuntu-magnum https://zuul.opendev.org/t/openstack/build/47eff301d83340199865c1e177e19bb7 : SUCCESS in 59m 45s (non-voting)\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f5a2d59c873045af8f5d22e8215a36b7 : SUCCESS in 1h 29m 27s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8444c2bb73dd4eb183aaaa664730b833 : SUCCESS in 1h 41m 09s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5bf5f3fb56ba4999b61fde10e09edadb : SUCCESS in 1h 26m 51s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/e0f9fee7df744812b646bb176d56e676 : SUCCESS in 1h 42m 29s (non-voting)\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/eb975b114b3c481d975135e4325a1423 : SUCCESS in 1h 13m 33s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/5dc83cebf23d429ea6787efabb6f114b : SUCCESS in 2h 33m 02s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/eca78fae962b46dd81a559ab8c4d6f3c : SUCCESS in 2h 08m 13s (non-voting)\n- kolla-ansible-rocky9-hashi-vault https://zuul.opendev.org/t/openstack/build/39c77f14986d4250840c5b40aac2d35a : SUCCESS in 3m 57s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"cce49bc555a6198d4da65a6ef3066b0362357869","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-26 03:35:19.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"a26c5d94957d7ec945aa05b4d2d2c55761350c33","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-01-07 00:36:54.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"d5c4f284d349c73c99f22620e89d5af12a7df465","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2024-01-07 20:46:55.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"0f1c9bd6d4faff48d1ad8ddba89b0789e71d2f4b","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2024-01-08 01:55:50.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"6a4391b5016a77f64d203ddd88b894ead2f555dd","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-01-08 08:56:22.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"3c8ed785f55f43a5854b5226f6dc427af697f184","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-01-08 10:20:03.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"d82c058c2be664078e47d0f6580c2858bd134589","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-01-08 16:07:48.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"66f0e1f550c5ad063d62e114e331348cb31a9f2c","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-01-09 03:25:54.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"ae4612a562e479023950f4ae51bb5a5b926f4374","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-01-09 03:26:03.000000000","message":"Patch Set 12: -Code-Review","accounts_in_message":[],"_revision_number":12},{"id":"22e92a5119621824a5bfb8739a8c9355c66ddbbe","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-01-09 07:59:41.000000000","message":"Patch Set 12: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"92ba71ecc9a2b274a391675c47309fe2be46f26e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-01-16 01:22:42.000000000","message":"Patch Set 13: Patch Set 12 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"5a569fc402c97d5d3c9a29af53c2de8a29f0fbc4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 02:50:05.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2be12bdc7cd0460695dce07b8b9d8872\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/35be1f78ad424fadbb393e4c93556299 : SUCCESS in 54m 44s\n- kolla-ansible-debian-aarch64-podman https://zuul.opendev.org/t/openstack/build/192066345b1a4b98824a2b27879057a9 : SUCCESS in 1h 26m 30s\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/1bdc297c98244032a476a9e22f907a4f : SUCCESS in 1h 10m 21s","accounts_in_message":[],"_revision_number":13},{"id":"5308b39e26d0d495c21d3894a1a569c7c2ff89dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 04:35:13.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfdf8b21b1a54b08965347db47797cfc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1eaee7d3819a4afbaccb38102c8130fa : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/963354e027c94b3f8d464fd8aac2f3d1 : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4cf0942aed834d8d9b549dd0d35718af : SUCCESS in 2m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ee1f31216c794f1ba2c65a74100f0fa9 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa8674f50bae4d9581149c1ddcbe38f6 : SUCCESS in 4m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/065dddfb7adc4848a410c9755b01ed5e : SUCCESS in 6m 13s\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ed42ce909b0347c29ceb88d2151ac322 : SUCCESS in 1h 14m 26s (non-voting)\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/dc2b6cfb0fa8466eb59873c3478a15fd : SUCCESS in 56m 33s\n- kolla-ansible-debian-podman https://zuul.opendev.org/t/openstack/build/a4bd401d7f444a53b562b7c9dbe27172 : SUCCESS in 1h 03m 01s\n- kolla-ansible-openeuler https://zuul.opendev.org/t/openstack/build/aa2164e2dc3f44a48d84a8693304c11b : SUCCESS in 1h 10m 54s (non-voting)\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f4407c28e4024b6ab905f5a04055c4dc : SUCCESS in 1h 00m 45s\n- kolla-ansible-rocky9-podman https://zuul.opendev.org/t/openstack/build/2fe6fcfe235c4e22bb59144ef4b0fa19 : SUCCESS in 1h 10m 40s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ac36ef7b816c4f36802794c71d84ecac : SUCCESS in 50m 23s\n- kolla-ansible-ubuntu-podman https://zuul.opendev.org/t/openstack/build/04aea177e3f24f4d97cf077e45b9ff8d : SUCCESS in 59m 34s\n- kolla-ansible-rocky9-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/fc20ff6a2c2f4616b486b4cb9f0e398d : SUCCESS in 1h 07m 23s (non-voting)\n- kolla-ansible-ubuntu-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/72a76b8155404a298c9cf2773e42009f : SUCCESS in 47m 08s (non-voting)\n- kolla-ansible-rocky9-magnum https://zuul.opendev.org/t/openstack/build/d4a85bf983374725926bba02822d5fa2 : SUCCESS in 1h 04m 47s (non-voting)\n- kolla-ansible-ubuntu-magnum https://zuul.opendev.org/t/openstack/build/d68701095e9c4615b79baf6f04aa7742 : SUCCESS in 39m 01s (non-voting)\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/38e8e0f05b234238963bcf5dc800e790 : SUCCESS in 1h 19m 49s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4c43f163553846a6acd349d446926e62 : SUCCESS in 1h 31m 27s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b485f444409b47bc81f9d0fc404fe864 : SUCCESS in 1h 17m 01s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/ddcc01c8f39b41de83cd77d7a9ae24ca : SUCCESS in 1h 54m 59s (non-voting)\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/54f0c16a2bc943e89d0c8a4c424a3840 : SUCCESS in 1h 34m 08s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/13d85b6c3b724067955ed49832d57509 : TIMED_OUT in 3h 04m 18s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/9cae11b2128049a7b776b976f71b47fc : SUCCESS in 2h 50m 56s (non-voting)\n- kolla-ansible-rocky9-hashi-vault https://zuul.opendev.org/t/openstack/build/b3e64d7b26ad45e4aed60924be71cf5a : SUCCESS in 3m 08s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"b704886e4d6867bfc148c0025eab4394abf64a03","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-01-16 05:26:40.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"065fbf58f29b33c589551809d56ad98f8d48c70a","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2024-01-25 06:13:45.000000000","message":"Patch Set 13: Backport-Candidate+1 Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"6f15296247ff806e3bdc0e373589609aeba2d272","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-01-25 08:36:07.000000000","message":"Patch Set 13: Backport-Candidate-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"011c4f51450729475b5d8c3dc7c1ae3b136d4e24","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-01-25 08:37:55.000000000","message":"Patch Set 13: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"b5fb0186eb0fee0ee60b4eba392b5e2a3a99aceb","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-02-20 03:21:35.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"297f40e12aa62dd4b17d8db4be67c513808e282f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-05-11 02:50:46.000000000","message":"Patch Set 14: Patch Set 13 was rebased\n\nCopied Votes:\n* Backport-Candidate+1, Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Code-Review+1, Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"7b12382084b999f4b4c80bcdb75f2a7e89979117","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-11 05:23:51.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f33a7e66bf543588dde46f09a2ce693\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d9c35901d924e6f920e1504421cbb83 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4d46551da3f4d4fb52117f8e4330a7d : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ef577a63c0945bc9cff237f31859809 : SUCCESS in 4m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ce64cd9f34de40029e171580d33ddf7a : SUCCESS in 4m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b52cfe6e1ef34126ae301ff38bf50fbf : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffb47e0d0b9a415d98e1004b2b3f08cb : SUCCESS in 5m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f261932b097e4c10a63010f118338fec : SUCCESS in 7m 27s\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/565c6739f1084256a192caf28e4419a1 : SUCCESS in 1h 07m 52s (non-voting)\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/946f2a8119af48a4b500a7cf37e97aa9 : SUCCESS in 1h 01m 52s\n- kolla-ansible-debian-podman https://zuul.opendev.org/t/openstack/build/78952740c72e4c9b8324c9f56374688c : SUCCESS in 1h 00m 10s\n- kolla-ansible-openeuler https://zuul.opendev.org/t/openstack/build/f767aa8e6ac14e01ac2d7ffdfeeca6f4 : SUCCESS in 46m 38s (non-voting)\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4e2467e4b861478b9d78a7e1acbc1c0e : SUCCESS in 53m 37s\n- kolla-ansible-rocky9-podman https://zuul.opendev.org/t/openstack/build/74e00c737eac4cf2afdbd0aaf5c68920 : SUCCESS in 1h 00m 44s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/34dd7e1379c94a41b40d379e2e2b0d5f : SUCCESS in 43m 57s\n- kolla-ansible-ubuntu-podman https://zuul.opendev.org/t/openstack/build/0443d3fa85ec4d2d96a3c439b552afc3 : SUCCESS in 51m 52s\n- kolla-ansible-rocky9-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/aa2f87c427c04dd9ac847d03f9d8e4b9 : SUCCESS in 1h 01m 51s (non-voting)\n- kolla-ansible-ubuntu-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/f9347a13a0a549a0a469cb5cc6d96374 : SUCCESS in 44m 31s (non-voting)\n- kolla-ansible-rocky9-magnum https://zuul.opendev.org/t/openstack/build/8b5d01170c9d4b81ada187a15154f0bc : SUCCESS in 1h 05m 15s (non-voting)\n- kolla-ansible-ubuntu-magnum https://zuul.opendev.org/t/openstack/build/3f3f5144125a42b8801bb61bc9cad616 : SUCCESS in 41m 31s (non-voting)\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/20487eefd3ab493e989d9f576e589b89 : SUCCESS in 1h 30m 18s\n- kolla-ansible-rocky9-slurp-upgrade https://zuul.opendev.org/t/openstack/build/eff5ecbcb8144c3a973a64f6b48d9c5e : SUCCESS in 1h 26m 31s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/02c001e627dd442d8463bf87f14719f5 : SUCCESS in 1h 20m 53s\n- kolla-ansible-debian-slurp-upgrade https://zuul.opendev.org/t/openstack/build/51c078ca4cfb4480a9c205669a7e9dcc : SUCCESS in 1h 21m 31s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0e5139c9a85b4a1eac5a415d138b46bc : SUCCESS in 1h 13m 38s\n- kolla-ansible-ubuntu-slurp-upgrade https://zuul.opendev.org/t/openstack/build/f3b0d778507a4655912fad1669de03b6 : SUCCESS in 1h 13m 11s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/4eba571f8743410684ebed590bc0ab16 : SUCCESS in 1h 34m 40s (non-voting)\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/ce9d20b9b1d345c38447ec29fe2ddc9d : SUCCESS in 1h 22m 52s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/3f95191d9ef64e9bb7fb22b679b5ad3e : SUCCESS in 2h 26m 08s (non-voting)\n- kolla-ansible-rocky9-slurp-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/664b9baa486a43feb4fccba885182a55 : SUCCESS in 1h 59m 50s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f6c7d97b9a894c08b8afc68a8c96f836 : SUCCESS in 2h 08m 24s (non-voting)\n- kolla-ansible-ubuntu-slurp-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/6f80f91f0eda44ed98e80e37fac281f3 : SUCCESS in 1h 50m 39s (non-voting)\n- kolla-ansible-rocky9-hashi-vault https://zuul.opendev.org/t/openstack/build/ccaf02d3fc324d3abcda3fad82024b63 : SUCCESS in 3m 59s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"8628a2a896b5364fb77f25dff395a43676e68933","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-11 05:55:29.000000000","message":"Patch Set 14:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d608b53e5bef44dd88e1f0fa772f8478\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6cd62dc8589542b3ac22cf3134cb6a62 : SUCCESS in 1h 04m 50s\n- kolla-ansible-debian-aarch64-podman https://zuul.opendev.org/t/openstack/build/6a9ea2035b18416fa3f0f8769ef1a3e2 : TIMED_OUT in 3h 02m 36s\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/fdad1fb14656493aada8b83454541aa1 : SUCCESS in 1h 18m 01s","accounts_in_message":[],"_revision_number":14},{"id":"1a8cdcc8777afb71e743693c3be6023779b378f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-05-09 08:13:49.000000000","message":"Uploaded patch set 15.\n\nCopied Votes:\n* Backport-Candidate+1, Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"f62461db8e840a0ac3be11ec134d10f6217a6f79","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-09 09:31:13.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6edb7521198145f78e4aa80de6b5d313\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/af526612eef7477b9a53426ac0f3b1b9 : SUCCESS in 1h 08m 43s (non-voting)\n- kolla-ansible-debian-aarch64-podman https://zuul.opendev.org/t/openstack/build/d148ecfd095c41e6a5a67f51067b13d1 : SUCCESS in 1h 09m 13s (non-voting)\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/db64947d503c4cf49d46a88bc95ba3ae : SUCCESS in 1h 15m 51s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"d3d268386cf4a3f062b83ee2f4c4d7423ccf8137","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-09 10:48:05.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d7a80940b8f4c2f994ef4566cd28687\n\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/ec0d24ef3bff4f938b548177b7572d34 : SUCCESS in 2m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a07a98bf282744d79cd961b30312d614 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49aa44fc67aa479b80877a56a54f4004 : SUCCESS in 5m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/85d64548c97145f88a6da52e595a391e : FAILURE in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c5d8528e7f94ad68e3a37f0c0744831 : FAILURE in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a23a68d27fb14958bc1b6e01a23b544b : SUCCESS in 4m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/563b5da56c414843826f25768186c4b8 : SUCCESS in 2m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9c7e9cf51e6a4836b1a65dd1b5f47acd : SUCCESS in 2m 58s\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/e9dbed524976497799fc4be3b59a32fd : SUCCESS in 1h 00m 42s (non-voting)\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/7bf96179d5004de798fb876c42020f3a : SUCCESS in 47m 54s\n- kolla-ansible-debian-podman https://zuul.opendev.org/t/openstack/build/b6edab3cdb164eb68f055540a89a6a5b : SUCCESS in 1h 00m 45s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c290b795d75544469e8d757dd179e9a4 : SUCCESS in 59m 25s\n- kolla-ansible-rocky9-podman https://zuul.opendev.org/t/openstack/build/b1dc4b4e9a0144f28c625fed062cb098 : SUCCESS in 40m 05s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5f92ab0c35ce43c684e68a0698c751a4 : SUCCESS in 49m 41s\n- kolla-ansible-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c054bf637dd04c819888ed82b5024a6b : SUCCESS in 43m 43s\n- kolla-ansible-rocky9-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/75fd198c4f894f53966ce3dbe28bfdac : SUCCESS in 1h 03m 41s (non-voting)\n- kolla-ansible-ubuntu-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/e8760c35857443b1b0d9d35e7734639b : DISK_FULL in 59m 37s (non-voting)\n- kolla-ansible-rocky9-magnum https://zuul.opendev.org/t/openstack/build/783175bf752848689b1959ac16082854 : SUCCESS in 55m 34s (non-voting)\n- kolla-ansible-ubuntu-magnum https://zuul.opendev.org/t/openstack/build/c8a8b4c3a94a4b4fb96a37bc1d2fd99c : SUCCESS in 34m 56s (non-voting)\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/172d9c4a06b24aaa97db3f194a4f1532 : SUCCESS in 42m 39s\n- kolla-ansible-rocky9-slurp-upgrade https://zuul.opendev.org/t/openstack/build/47135ca4e5164597999514c0aac3a93a : SUCCESS in 1h 16m 48s (non-voting)\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/91e164ed725a4b66b235caa59e6e98ef : SUCCESS in 1h 12m 15s\n- kolla-ansible-debian-slurp-upgrade https://zuul.opendev.org/t/openstack/build/b0af348f0cb44fa48bf29ac7ea7ae681 : SUCCESS in 1h 31m 46s (non-voting)\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/122d330d65a14e6eb3260f6e8b7de05d : SUCCESS in 1h 17m 36s\n- kolla-ansible-ubuntu-slurp-upgrade https://zuul.opendev.org/t/openstack/build/c17940e136f940e58ef9dcb7f903fcf3 : SUCCESS in 1h 29m 37s (non-voting)\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/b79c01a5d8bd49caa01449ae0acda952 : FAILURE in 1h 19m 22s (non-voting)\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/d1c48899851d44998b9d3ae74206875c : FAILURE in 1h 33m 08s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/2773aeccd0a0493898da03c68edb99c9 : FAILURE in 1h 03m 03s (non-voting)\n- kolla-ansible-rocky9-slurp-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/5a78b789d5e2484f8601b220a692195b : FAILURE in 58m 49s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/e5f43a45c6ce4ce097755fd2749eedd9 : FAILURE in 1h 54m 47s (non-voting)\n- kolla-ansible-ubuntu-slurp-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/2dbe8cedae4d451d8da59f0dbe51aced : FAILURE in 1h 17m 04s (non-voting)\n- kolla-ansible-rocky9-hashi-vault https://zuul.opendev.org/t/openstack/build/9eb57bff4c344389b5c870b7d31f9727 : SUCCESS in 4m 02s (non-voting)","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"3e64e274fc2b633e5205fa40f3f767de55a06303","revisions":{"253c1ad2aefee83b89b2a786a0974cc4f3df8314":{"kind":"REWORK","_number":1,"created":"2022-11-02 15:22:26.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863321/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863321/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/1"}}},"commit":{"parents":[{"commit":"4fa36cbe08d940305f2a6d4935291813f8e43d43","subject":"Merge \"Fix ironic interface option for ironic-inspector\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/4fa36cbe08d940305f2a6d4935291813f8e43d43"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-02 15:13:48.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-02 15:21:47.000000000","tz":480},"subject":"Implement automatic deploy of trove","message":"Implement automatic deploy of trove\n\nthis patchset has implemented:\n  - network (trove-mgmt-net)\n  - security groups and rules\n  - nova keypair\nAdd trove trove_docker_* variables which are\nused by trove guest-agent.\n\nImplements: blueprint implement-automatic-deploy-of-trove\nChange-Id: I6defbc0a914065d7f22d46d883cc4a71ead8f9a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/253c1ad2aefee83b89b2a786a0974cc4f3df8314"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/253c1ad2aefee83b89b2a786a0974cc4f3df8314"}]},"branch":"refs/heads/master"},"12cc96d4d4292ccb82c078c915c90ab740fdc6af":{"kind":"REWORK","_number":2,"created":"2022-11-03 11:43:36.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863321/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863321/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/2"}}},"commit":{"parents":[{"commit":"4fa36cbe08d940305f2a6d4935291813f8e43d43","subject":"Merge \"Fix ironic interface option for ironic-inspector\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/4fa36cbe08d940305f2a6d4935291813f8e43d43"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-02 15:13:48.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 11:43:09.000000000","tz":480},"subject":"Implement automatic deploy of trove","message":"Implement automatic deploy of trove\n\nthis patchset has implemented:\n  - network (trove-mgmt-net)\n  - security groups and rules\n  - nova keypair\nAdd trove trove_docker_* variables which are\nused by trove guest-agent.\n\nImplements: blueprint implement-automatic-deploy-of-trove\nChange-Id: I6defbc0a914065d7f22d46d883cc4a71ead8f9a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/12cc96d4d4292ccb82c078c915c90ab740fdc6af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/12cc96d4d4292ccb82c078c915c90ab740fdc6af"}]},"branch":"refs/heads/master"},"e722727c5d014038fc7ee2dfcbcd5b0b143ff618":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-12-30 02:06:28.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863321/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863321/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/3"}}},"commit":{"parents":[{"commit":"1f71aa563e626f70684fdc48966ddff3cfce518d","subject":"CI: Support tox4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1f71aa563e626f70684fdc48966ddff3cfce518d"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-02 15:13:48.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-12-30 02:04:31.000000000","tz":480},"subject":"Implement automatic deploy of trove","message":"Implement automatic deploy of trove\n\nthis patchset has implemented:\n  - network (trove-mgmt-net)\n  - security groups and rules\n  - nova keypair\nAdd trove trove_docker_* variables which are\nused by trove guest-agent.\n\nImplements: blueprint implement-automatic-deploy-of-trove\nChange-Id: I6defbc0a914065d7f22d46d883cc4a71ead8f9a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/e722727c5d014038fc7ee2dfcbcd5b0b143ff618"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/e722727c5d014038fc7ee2dfcbcd5b0b143ff618"}]},"branch":"refs/heads/master"},"21e59390e3bc5162c590cbd92f62c1dfb94ebe17":{"kind":"REWORK","_number":4,"created":"2022-12-30 03:02:03.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863321/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863321/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/4"}}},"commit":{"parents":[{"commit":"1f71aa563e626f70684fdc48966ddff3cfce518d","subject":"CI: Support tox4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1f71aa563e626f70684fdc48966ddff3cfce518d"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-02 15:13:48.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-12-30 03:01:49.000000000","tz":480},"subject":"Implement automatic deploy of trove","message":"Implement automatic deploy of trove\n\nthis patchset has implemented:\n  - network (trove-mgmt-net)\n  - security groups and rules\n  - nova keypair\nAdd trove trove_docker_* variables which are\nused by trove guest-agent.\n\nImplements: blueprint implement-automatic-deploy-of-trove\nChange-Id: I6defbc0a914065d7f22d46d883cc4a71ead8f9a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/21e59390e3bc5162c590cbd92f62c1dfb94ebe17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/21e59390e3bc5162c590cbd92f62c1dfb94ebe17"}]},"branch":"refs/heads/master"},"e1dc4b2faa2a7ddf30edca338e2d87dcb5d4ad35":{"kind":"REWORK","_number":5,"created":"2023-01-05 15:17:07.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863321/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863321/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/5"}}},"commit":{"parents":[{"commit":"1f71aa563e626f70684fdc48966ddff3cfce518d","subject":"CI: Support tox4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1f71aa563e626f70684fdc48966ddff3cfce518d"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-02 15:13:48.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-01-05 15:05:35.000000000","tz":480},"subject":"Implement automatic deploy of trove","message":"Implement automatic deploy of trove\n\nthis patchset has implemented:\n  - network (trove-mgmt-net)\n  - security groups and rules\n  - nova keypair\nAdd trove trove_docker_* variables which are\nused by trove guest-agent.\n\nImplements: blueprint implement-automatic-deploy-of-trove\nChange-Id: I6defbc0a914065d7f22d46d883cc4a71ead8f9a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/e1dc4b2faa2a7ddf30edca338e2d87dcb5d4ad35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/e1dc4b2faa2a7ddf30edca338e2d87dcb5d4ad35"}]},"branch":"refs/heads/master"},"8be6ed8eac716545d790b55e3d24c373c636fb65":{"kind":"REWORK","_number":6,"created":"2023-11-17 07:24:52.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/863321/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863321/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/6"}}},"commit":{"parents":[{"commit":"08e2b734ba4f437e2000120306ee8ad875fae79c","subject":"Merge \"[doc] Fix the incorrect URL of \u0027Quick Start for development\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/08e2b734ba4f437e2000120306ee8ad875fae79c"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-02 15:13:48.000000000","tz":480},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-11-17 07:24:52.000000000","tz":0},"subject":"Implement automatic deploy of trove","message":"Implement automatic deploy of trove\n\nthis patchset has implemented:\n  - network (trove-mgmt-net)\n  - security groups and rules\n  - nova keypair\nAdd trove trove_docker_* variables which are\nused by trove guest-agent.\n\nImplements: blueprint implement-automatic-deploy-of-trove\nChange-Id: I6defbc0a914065d7f22d46d883cc4a71ead8f9a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/8be6ed8eac716545d790b55e3d24c373c636fb65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/8be6ed8eac716545d790b55e3d24c373c636fb65"}]},"branch":"refs/heads/master","description":"Rebase"},"b5f8aa4ac6ee6012053f147348135405f23974d3":{"kind":"REWORK","_number":7,"created":"2023-11-17 07:26:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/863321/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863321/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/7"}}},"commit":{"parents":[{"commit":"08e2b734ba4f437e2000120306ee8ad875fae79c","subject":"Merge \"[doc] Fix the incorrect URL of \u0027Quick Start for development\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/08e2b734ba4f437e2000120306ee8ad875fae79c"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-02 15:13:48.000000000","tz":480},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-11-17 07:26:04.000000000","tz":0},"subject":"Implement automatic deploy of trove","message":"Implement automatic deploy of trove\n\nthis patchset has implemented:\n  - network (trove-mgmt-net)\n  - security groups and rules\n  - nova keypair\nAdd trove trove_docker_* variables which are\nused by trove guest-agent.\n\nImplements: blueprint implement-automatic-deploy-of-trove\nChange-Id: I6defbc0a914065d7f22d46d883cc4a71ead8f9a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/b5f8aa4ac6ee6012053f147348135405f23974d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/b5f8aa4ac6ee6012053f147348135405f23974d3"}]},"branch":"refs/heads/master"},"c5f9b7c7fa4328134f6ecd48fc634c6e206f3ab8":{"kind":"REWORK","_number":8,"created":"2023-12-04 05:54:12.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863321/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863321/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/8"}}},"commit":{"parents":[{"commit":"84f0a37e49dd3591a40d86349622ef61b2b4864b","subject":"Merge \"Fix grafana prometheus datasource\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/84f0a37e49dd3591a40d86349622ef61b2b4864b"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-02 15:13:48.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-04 05:42:31.000000000","tz":480},"subject":"Implement automatic deploy of trove","message":"Implement automatic deploy of trove\n\nthis patchset has implemented:\n  - network (trove-mgmt-net)\n  - security groups and rules\n  - nova keypair\n\nImplements: blueprint implement-automatic-deploy-of-trove\nChange-Id: I6defbc0a914065d7f22d46d883cc4a71ead8f9a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c5f9b7c7fa4328134f6ecd48fc634c6e206f3ab8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c5f9b7c7fa4328134f6ecd48fc634c6e206f3ab8"}]},"branch":"refs/heads/master"},"d5d4208b67260bf546513b622055ca2ff3f466e2":{"kind":"REWORK","_number":9,"created":"2023-12-04 06:03:54.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863321/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863321/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/9"}}},"commit":{"parents":[{"commit":"84f0a37e49dd3591a40d86349622ef61b2b4864b","subject":"Merge \"Fix grafana prometheus datasource\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/84f0a37e49dd3591a40d86349622ef61b2b4864b"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-02 15:13:48.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-04 06:03:09.000000000","tz":480},"subject":"Implement automatic deploy of trove","message":"Implement automatic deploy of trove\n\nthis patchset has implemented:\n  - network (trove-mgmt-net)\n  - security groups and rules\n  - nova keypair\n\nImplements: blueprint implement-automatic-deploy-of-trove\nChange-Id: I6defbc0a914065d7f22d46d883cc4a71ead8f9a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/d5d4208b67260bf546513b622055ca2ff3f466e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/d5d4208b67260bf546513b622055ca2ff3f466e2"}]},"branch":"refs/heads/master"},"9309a59fcd93b2f4f10d8381ff679f6b7d39a2bb":{"kind":"REWORK","_number":10,"created":"2023-12-04 08:20:02.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863321/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863321/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/10"}}},"commit":{"parents":[{"commit":"84f0a37e49dd3591a40d86349622ef61b2b4864b","subject":"Merge \"Fix grafana prometheus datasource\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/84f0a37e49dd3591a40d86349622ef61b2b4864b"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-02 15:13:48.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-04 08:19:40.000000000","tz":480},"subject":"Implement automatic deploy of trove","message":"Implement automatic deploy of trove\n\nthis patchset has implemented:\n  - network (trove-mgmt-net)\n  - security groups and rules\n  - nova keypair\n\nImplements: blueprint implement-automatic-deploy-of-trove\nChange-Id: I6defbc0a914065d7f22d46d883cc4a71ead8f9a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/9309a59fcd93b2f4f10d8381ff679f6b7d39a2bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/9309a59fcd93b2f4f10d8381ff679f6b7d39a2bb"}]},"branch":"refs/heads/master"},"5090c3b949d4115fb22fd5fa8cb064914b1e1142":{"kind":"REWORK","_number":11,"created":"2023-12-11 03:14:26.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863321/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863321/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/11"}}},"commit":{"parents":[{"commit":"84f0a37e49dd3591a40d86349622ef61b2b4864b","subject":"Merge \"Fix grafana prometheus datasource\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/84f0a37e49dd3591a40d86349622ef61b2b4864b"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-02 15:13:48.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-11 03:14:11.000000000","tz":480},"subject":"Implement automatic deploy of trove","message":"Implement automatic deploy of trove\n\nthis patchset has implemented:\n  - network (trove-mgmt-net)\n  - security groups and rules\n  - nova keypair\n\nImplements: blueprint implement-automatic-deploy-of-trove\nChange-Id: I6defbc0a914065d7f22d46d883cc4a71ead8f9a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/5090c3b949d4115fb22fd5fa8cb064914b1e1142"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/5090c3b949d4115fb22fd5fa8cb064914b1e1142"}]},"branch":"refs/heads/master"},"c331a12c48248275d4462a660633bb6a0ec06572":{"kind":"REWORK","_number":12,"created":"2023-12-11 06:36:52.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863321/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863321/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/12"}}},"commit":{"parents":[{"commit":"84f0a37e49dd3591a40d86349622ef61b2b4864b","subject":"Merge \"Fix grafana prometheus datasource\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/84f0a37e49dd3591a40d86349622ef61b2b4864b"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-02 15:13:48.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-11 06:36:42.000000000","tz":480},"subject":"Implement automatic deploy of trove","message":"Implement automatic deploy of trove\n\nthis patchset has implemented:\n  - network (trove-mgmt-net)\n  - security groups and rules\n  - nova keypair\n\nImplements: blueprint implement-automatic-deploy-of-trove\nChange-Id: I6defbc0a914065d7f22d46d883cc4a71ead8f9a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c331a12c48248275d4462a660633bb6a0ec06572"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c331a12c48248275d4462a660633bb6a0ec06572"}]},"branch":"refs/heads/master"},"c432aa68a1c7eec2805e754cdb3de1843445dc24":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2024-01-16 01:22:42.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863321/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863321/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/13"}}},"commit":{"parents":[{"commit":"77c18fa615cc592976ae65a52c0198a14e054876","subject":"Merge \"CI: Use 2023.2 as previous_release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/77c18fa615cc592976ae65a52c0198a14e054876"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-02 15:13:48.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2024-01-16 01:22:42.000000000","tz":0},"subject":"Implement automatic deploy of trove","message":"Implement automatic deploy of trove\n\nthis patchset has implemented:\n  - network (trove-mgmt-net)\n  - security groups and rules\n  - nova keypair\n\nImplements: blueprint implement-automatic-deploy-of-trove\nChange-Id: I6defbc0a914065d7f22d46d883cc4a71ead8f9a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c432aa68a1c7eec2805e754cdb3de1843445dc24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c432aa68a1c7eec2805e754cdb3de1843445dc24"}]},"branch":"refs/heads/master","description":"Rebase"},"734d789492dd726f514219d6e69f80f5ba67ed55":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2024-05-11 02:50:46.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863321/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863321/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/14"}}},"commit":{"parents":[{"commit":"b382471d02a9763aa135450e63a705cfad261d58","subject":"Merge \"loadbalancer: fix upgrade task when podman is used\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/b382471d02a9763aa135450e63a705cfad261d58"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-02 15:13:48.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2024-05-11 02:50:46.000000000","tz":0},"subject":"Implement automatic deploy of trove","message":"Implement automatic deploy of trove\n\nthis patchset has implemented:\n  - network (trove-mgmt-net)\n  - security groups and rules\n  - nova keypair\n\nImplements: blueprint implement-automatic-deploy-of-trove\nChange-Id: I6defbc0a914065d7f22d46d883cc4a71ead8f9a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/734d789492dd726f514219d6e69f80f5ba67ed55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/734d789492dd726f514219d6e69f80f5ba67ed55"}]},"branch":"refs/heads/master","description":"Rebase"},"3e64e274fc2b633e5205fa40f3f767de55a06303":{"kind":"REWORK","_number":15,"created":"2025-05-09 08:13:49.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863321/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863321/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863321/15"}}},"commit":{"parents":[{"commit":"f9d337a5819ee83019e59f4415786f26323e3333","subject":"Merge \"Add Pure Storage FlashBlade as Manila backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/f9d337a5819ee83019e59f4415786f26323e3333"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-02 15:13:48.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2025-05-09 08:13:47.000000000","tz":480},"subject":"Implement automatic deploy of trove","message":"Implement automatic deploy of trove\n\nthis patchset has implemented:\n  - network (trove-mgmt-net)\n  - security groups and rules\n  - nova keypair\n\nImplements: blueprint implement-automatic-deploy-of-trove\nChange-Id: I6defbc0a914065d7f22d46d883cc4a71ead8f9a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/3e64e274fc2b633e5205fa40f3f767de55a06303"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/3e64e274fc2b633e5205fa40f3f767de55a06303"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1"],"failing_atoms":[],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
