)]}'
{"id":"openstack%2Fkolla-ansible~863521","triplet_id":"openstack%2Fkolla-ansible~master~I37f225a2577c4a13bd87503b4a363ca1171f1513","project":"openstack/kolla-ansible","branch":"master","topic":"trove-deploy-ci","attention_set":{"28752":{"account":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"last_update":"2024-02-22 13:43:13.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"34579":{"account":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},"last_update":"2023-12-26 03:35:54.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-12-02 13:33:28.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"36201":{"account":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"last_update":"2024-02-22 13:43:13.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2025-05-09 10:43:18.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I37f225a2577c4a13bd87503b4a363ca1171f1513","subject":"CI: Trove: Create a database instance","status":"NEW","created":"2022-11-03 13:39:50.000000000","updated":"2025-05-09 10:43:18.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7aa6794513280bcf7aee237b0213106ebef89f25","_number":863521,"virtual_id_number":863521,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-05-09 10:43:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-03 13:53:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-02 13:33:28.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-12-06 15:26:23.000000000","updated_by":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},"reviewer":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},"state":"REVIEWER"},{"updated":"2024-01-07 17:33:28.000000000","updated_by":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"reviewer":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"state":"REVIEWER"},{"updated":"2024-01-08 02:38:04.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"}],"messages":[{"id":"32fb5fa00201657dd09aa72e49204c3c31cc0e04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-11-03 13:39:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"548068fdc0d400e53ce820f5ee9df7250df6dbe6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-11-03 13:42:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d1624a5842e6fee6bbbebb77da337c5ae65f7507","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 13:53:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3357ad7ad7e0482693e1849e3db1e528\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2feb5c56a7b451f935681bcfadfb6ab : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95572e7ae16c459bb0ae8ae7123256f7 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ea4c94ba6c24fb28b626cc9a0ee67a4 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25cd048dd0984e9fa7899bd08b3fdf8c : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e6e606347444f8d83d8f9b6638412e1 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f45e4ad0021140f38bad17392dcf8cfd : SUCCESS in 5m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d65daf0e1c8043f39688692815171739 : FAILURE in 4m 55s\n- kolla-ansible-rocky9-source-trove https://zuul.opendev.org/t/openstack/build/1bbf7c028286493582ef49be3824aec4 : FAILURE in 4m 30s (non-voting)\n- kolla-ansible-ubuntu-source-trove https://zuul.opendev.org/t/openstack/build/b9213a8583aa459e9bbb03406ca1a14c : FAILURE in 4m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d478184a73ed84f8d3c299593c7be81e20e8b0f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-11-03 15:05:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"56770fcdf2c3b432aae84c7114d491d79ebcbfd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 15:19:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f7ae6f22ac540c0ad158c73b2417b62\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59b320f4a2b946988cfa5c29a5b9475a : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/712c9814417f4d43b8059f36c4d5ba56 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a15e0e8b29849239be204d86feb9480 : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/131e1a8ed170499c9fbb832f5b1f2c85 : SUCCESS in 5m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e90543c2826048cda2a79af82cad0ed5 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b47beca5af2d4bc39da340c1e1c00f1b : SUCCESS in 5m 12s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8ca4d9b0d19741dba96a3f465a862838 : FAILURE in 6m 54s\n- kolla-ansible-rocky9-source-trove https://zuul.opendev.org/t/openstack/build/107d75c4a9d54d29baee92663cbc9646 : FAILURE in 4m 24s (non-voting)\n- kolla-ansible-ubuntu-source-trove https://zuul.opendev.org/t/openstack/build/db561203392843458373a061a895d7e5 : FAILURE in 3m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cbde880f13462999909aa1abde47c03d2e760410","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-11-03 15:40:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e428f621b0779712506b7531ab2768702d4fd19a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 16:17:39.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c828187ad4954c7e88de92ba2148ffcf\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6eaba841803c40e2b99d05a6a6280b65 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0da44da5cce24b9d9eee8184341bcaa3 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a99829291e4446caf134024d3e9fc18 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e71fbf82246344b897291c5c4551e6ce : SUCCESS in 5m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8f1df8f5fb7a4fbd98811d927cfb6228 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/350ec9039e4746dfae5afb3d1fadda8e : SUCCESS in 5m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a3be82f5b4db4ca28e50316342d1fdea : FAILURE in 5m 46s\n- kolla-ansible-rocky9-source-trove https://zuul.opendev.org/t/openstack/build/36dc7568f10342acabe351bf042e40c8 : FAILURE in 36m 29s (non-voting)\n- kolla-ansible-ubuntu-source-trove https://zuul.opendev.org/t/openstack/build/a7069ca11f7b43d499b5e77fc74c9088 : FAILURE in 29m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"24e4de652926c9286871389cb2d427c20eedfdf2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-11-04 01:07:19.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3d10146ead16cf7f64ab53b99124bfeb40a35c7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 01:44:57.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/82dac7f8b2a74023a97bdd9fad499408\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f6d4297dd8b4c0b8ca90a1610e1af59 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7916367fa58c4da999348db1c265e76c : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38809595e44f4f55bcc3398c7fd981f0 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c025777235f44adcad5bfec8560800b1 : SUCCESS in 5m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d3aec41938d84331b01554342b3215c2 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a56e1bffed5a450b9cdf9aa80123cb51 : SUCCESS in 4m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/76ebb6d41cf14c59adc3c3121903151d : FAILURE in 5m 26s\n- kolla-ansible-rocky9-source-trove https://zuul.opendev.org/t/openstack/build/d187afc052fe41f5a227e2bc7d5c1f3e : FAILURE in 31m 13s (non-voting)\n- kolla-ansible-ubuntu-source-trove https://zuul.opendev.org/t/openstack/build/21885b4788ee4dc3afdd495905e4e836 : FAILURE in 27m 41s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5dd45eb645ec846504ba9761d8809af48f3e68c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-11-04 02:19:08.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2f7df667c818322de8b6d09491044c568e857b26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 03:04:52.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1452620b6e5e497cbeb7258d3819ee79\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91d6b8af764b47b181f9e412538e2268 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/026903d25f474a33b3f33d0773a99dba : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38970bf7089147e3b696b4b0ace6f03b : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bfc962321f5344d496b2836a43636e57 : SUCCESS in 6m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/287bc82b52244d10935ad3ff18f1a4b9 : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0562ca9f89cc4281b67aa08abfffe2b3 : SUCCESS in 5m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a27360048c334f338d74df7ae11f48aa : FAILURE in 6m 13s\n- kolla-ansible-rocky9-source-trove https://zuul.opendev.org/t/openstack/build/dd36118f7c114465b0998b783622806c : FAILURE in 33m 44s (non-voting)\n- kolla-ansible-ubuntu-source-trove https://zuul.opendev.org/t/openstack/build/86ae5ffacf5b46daa42bd033cd74c079 : FAILURE in 28m 29s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"edbe59fa3c372eedea8cd86de6c7ceb91036cda5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-11-04 03:27:50.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"f483395772bc97b8fc8e76b57971ae4d6156c1b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 04:13:27.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/382debcd8d9d464f9e487f680c740391\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e78634c7e4640ee89f8195dc21269f1 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad1f8519739849e8b70da309b05bd0e2 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2bf1b7df48e04c3fa26ae14fed56f53e : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32c68ea0c9f24df88f548c2f0d607eb1 : SUCCESS in 5m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a0c0fa59fb84b52971b16c7ba0411d6 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f5593503a6746eab95fcf9b89693ed5 : SUCCESS in 5m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d683577a486244d98b8da8c30432cb35 : FAILURE in 5m 43s\n- kolla-ansible-rocky9-source-trove https://zuul.opendev.org/t/openstack/build/fb6d326f0d584331b53a36f96bcef29e : FAILURE in 39m 40s (non-voting)\n- kolla-ansible-ubuntu-source-trove https://zuul.opendev.org/t/openstack/build/178f1d5c29564740bf33340c31403aea : FAILURE in 30m 39s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d91cb5ea0c27c03cc8c0ada33de18f31a61d8a02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-11-04 05:49:09.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"e0410a8a9c649c5822844b29ebb79b33be163de2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 05:58:53.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/80ff11c993494283b303358a02339294\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e591cbc9dc744eac99aee3f5361484af : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1942f41260734819bbd628457fd46160 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e3ff16047074cc09c983bb39d73a5ca : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/361b74f3fc3d452ba758d80e6cb8d392 : SUCCESS in 4m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bfb06786fac0415ea57d377c96d766f6 : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b21d2c8d2d5f45718342b94a96714daf : SUCCESS in 4m 28s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e16b17a1050d4b6f9af4b8ad51bbc9c6 : FAILURE in 3m 56s\n- kolla-ansible-rocky9-source-trove https://zuul.opendev.org/t/openstack/build/daac18efe61640baa0d70d8fd8073677 : FAILURE in 3m 55s (non-voting)\n- kolla-ansible-ubuntu-source-trove https://zuul.opendev.org/t/openstack/build/ab4f1c4d97b042c8beca05985e31668e : FAILURE in 3m 37s (non-voting)\n\nWarning:\n  Job openstack-tox-linters: unable to map line for file comments: \n  stderr: \u0027fatal: file tests/setup_disks.sh has only 43 lines\u0027","accounts_in_message":[],"_revision_number":8},{"id":"d5c8ee88926f08a5eb18f049cf2bfc4a349fca9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-11-04 06:28:31.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"8dcc8305559d5f534fe6ff6ef4a7ace41d75fea8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 07:20:17.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/402813b7daa946cfb8c6e1a70634c3ee\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c8a95dbf2d749388672ae75d4b30046 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34c37ee752ec4696b9da170b05ab61a3 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c14a1917e8949e7b9ddfd84a6a77a4f : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b5e83f91ac844458d0b64cbaaa73d95 : SUCCESS in 5m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ee71091d98644c1e9089c1a3ff1bcd94 : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa80702bd2b64b1383d312df78580bd3 : SUCCESS in 5m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/338b25e5a4a24e4f9d670de2cd0822aa : FAILURE in 6m 15s\n- kolla-ansible-rocky9-source-trove https://zuul.opendev.org/t/openstack/build/21b9b54f9c6848e4b25931ca0a48c80e : FAILURE in 45m 45s (non-voting)\n- kolla-ansible-ubuntu-source-trove https://zuul.opendev.org/t/openstack/build/429a5961026641e4803014f80c5c9999 : FAILURE in 39m 04s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"8d291a3ece38783fdcd0e211b2549be6d1b76b6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-11-04 07:25:40.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"37be4880d4a19c56b504666642f924bf55e0696e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 08:21:55.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2132a3cdc4034829a262362515c6c848\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c21edfb081c4240b8c1e44abe7e1a98 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c09fb27ee18444c083b815883b1f17b9 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/acf169ab79dc46b4a7021a902550db5e : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/12061263578b4bc7b95b0b792315fab2 : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e67d111b8d334689a67aa0e6cdb256a0 : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fe99c40712c4095910dfaed33d52136 : SUCCESS in 4m 54s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f6c63e4973764eedad947a8b5e332d62 : FAILURE in 5m 27s\n- kolla-ansible-rocky9-source-trove https://zuul.opendev.org/t/openstack/build/95075fc5ff88409f9bbb9248cf055a58 : FAILURE in 48m 17s (non-voting)\n- kolla-ansible-ubuntu-source-trove https://zuul.opendev.org/t/openstack/build/a70d60e05a9e4c33bbecc82980ae7e75 : FAILURE in 42m 03s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"2328fe8dfdd2a5381c37a101944837adc8e76d39","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-11-04 08:46:39.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"32c78946b348a7a78dfb108d40d9cf4f0c0a514f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 09:53:33.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ff0de44130884ae3be6d08bad75a96a7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c25a9663ecdd459c97f5a59e47b09ad0 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7dba735dc07453e9bf8d4ada3dfbea9 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6be08cca65834f0db5bf530eca343936 : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2bf60c74963424aa228e84640e8d722 : SUCCESS in 5m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/64e2996fc2c849bd815302fcf57fd8c1 : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f8d5d2148474a98a71a7d1485962c82 : SUCCESS in 5m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/79fa78b7601448308f26fd3e2bccd3a7 : FAILURE in 5m 43s\n- kolla-ansible-rocky9-source-trove https://zuul.opendev.org/t/openstack/build/f29f3a4df8394124ab1042033647dbea : FAILURE in 1h 06m 02s (non-voting)\n- kolla-ansible-ubuntu-source-trove https://zuul.opendev.org/t/openstack/build/3021e2bf2f0c486883831af7742dd3ec : SUCCESS in 52m 56s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"ada217853f529f1682c9ab2965eba939f321e10b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-11-04 13:56:19.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"e9e3ad1865dc10d1e1f61665825acd4aa7e42401","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 15:20:12.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ccea9768669d417a9dff897a38b90701\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9fa33ff496be41c9a29b500249482996 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fab3135e4574250bd7b6bbec558c54e : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/006ece6c42e043a6ae0b4d7dab64716f : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c16a2ae61ee426d8491d60146154b4e : SUCCESS in 5m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/edc27a5b0788469194ac2caa233a05da : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c91b94b89ae74d7c92287c28392502fc : SUCCESS in 5m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bbf88fa58c484f4d9fb78dfc4071df15 : FAILURE in 5m 59s\n- kolla-ansible-rocky9-source-trove https://zuul.opendev.org/t/openstack/build/926e8222d68c4956bc575c81b4aabaa9 : SUCCESS in 1h 13m 39s (non-voting)\n- kolla-ansible-ubuntu-source-trove https://zuul.opendev.org/t/openstack/build/97ad93ffafa34309a6d0814374cec410 : SUCCESS in 57m 07s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"8019e18bb38b4a4de2c249c1540b49dc34bd6c19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-11-07 03:20:16.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"48001e42c11f8548a3a5f39c934dede0a83a10c1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 04:38:46.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21a4961416934942828483b53239ac7b\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/920f4ae9781841109e49311f416fa9a6 : SUCCESS in 1h 02m 24s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/74d9d44be1cb4b49bd07e7a2c201d579 : SUCCESS in 1h 12m 49s","accounts_in_message":[],"_revision_number":13},{"id":"4b056a8e82b01cfd9eda1fc66b1b67a5b4b17e70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 04:51:39.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36697892211441aba002d2acbd8748e6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c92615d5987a47e99746e8198acc7bbe : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaea1638c7b44f918a2cb929d2c27235 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad03cf639f3a4a97995add2d9070aab4 : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e20e58b4a9c14d37882d11da1110d660 : SUCCESS in 5m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7287ad054ba64ea7ab851ee7a175f524 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5a176053c7e4cf4a6c9d91268c74ef1 : SUCCESS in 4m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/112d731ec4144d13b35855445ecbc031 : SUCCESS in 7m 17s\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/570dbd0700bb4ba1a4f2efc0bd180b1d : SUCCESS in 52m 45s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5edbd05b351d427890b9ec03cb1afc47 : SUCCESS in 43m 43s\n- kolla-ansible-openeuler-source https://zuul.opendev.org/t/openstack/build/bafb0e3c5a1f4541ad6a05ee4e4e9c99 : SUCCESS in 34m 32s (non-voting)\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/d9b0638e37cc4eb18b6f1c9946dd9ff2 : SUCCESS in 50m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1da101b49cfe4b45abeb847e7b35bd52 : SUCCESS in 29m 11s\n- kolla-ansible-rocky9-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/adad0b2e061942258b76430492e444e6 : SUCCESS in 54m 53s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/408dcac160c941f5a2e864a3bbe63720 : SUCCESS in 38m 17s (non-voting)\n- kolla-ansible-rocky9-source-zun https://zuul.opendev.org/t/openstack/build/569cf40a6cfe4693a2c117bc33990f9d : SUCCESS in 1h 24m 08s (non-voting)\n- kolla-ansible-debian-source-zun https://zuul.opendev.org/t/openstack/build/8ec0ce79e2234559b98f91c63e9b0a0a : SUCCESS in 1h 17m 40s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/50136b8dcab64c48abfcf23e36259f01 : SUCCESS in 52m 50s (non-voting)\n- kolla-ansible-rocky9-source-swift https://zuul.opendev.org/t/openstack/build/73ffdf7027cd4ef4b7148feb8b8be629 : SUCCESS in 1h 17m 22s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/02f30968451d4b739b35e086f316f4e1 : SUCCESS in 57m 16s (non-voting)\n- kolla-ansible-rocky9-source-trove https://zuul.opendev.org/t/openstack/build/93848c5d10fe4474b01b25d48ef56c48 : FAILURE in 1h 03m 59s (non-voting)\n- kolla-ansible-ubuntu-source-trove https://zuul.opendev.org/t/openstack/build/f2f2b31a637346df9dfaca82c0253766 : SUCCESS in 1h 00m 19s (non-voting)\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/91fc6e0727104787a2cd0c6f8f48f6ed : SUCCESS in 54m 01s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bf3e58be9068407ba13f8db9a375a92a : SUCCESS in 1h 04m 30s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/ab422939b0ea49f28c0c4e3e064957cf : SUCCESS in 57m 21s\n- kolla-ansible-rocky9-source-cephadm https://zuul.opendev.org/t/openstack/build/f850e4f429c34fe69cbd0088f09ede95 : SUCCESS in 1h 19m 44s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/3a72fafc695e42c489da86ed53102842 : SUCCESS in 1h 04m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/09b0aa8e19524e558d26adb1de38dd5d : SUCCESS in 1h 29m 55s (non-voting)\n- kolla-ansible-rocky9-hashi-vault https://zuul.opendev.org/t/openstack/build/ae37150ed000496c8381e49008cbd4b4 : SUCCESS in 2m 53s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"ce63461c6d3fc8928d97ebf24b75ea42358252b3","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-11-08 02:26:47.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"aca1ed6a1104ee50fa20e6e95ecc85fe2e36b2bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-12-30 02:06:28.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"4590a625935042b059bcf02f4bbf3d2cf2d8a6af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-30 03:02:23.000000000","message":"Patch Set 14: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/863321 is needed.","accounts_in_message":[],"_revision_number":14},{"id":"d8f2576b60b21b00f8be1a492cb68baca74a42e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-30 03:02:27.000000000","message":"Patch Set 14:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/863321 is needed.","accounts_in_message":[],"_revision_number":14},{"id":"3a8ec9ab93e13851c6a1f9e87442c270ac84f7f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-12-30 03:08:21.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"2459f7aab2e5a641c35d3aab005ba4e3543e4971","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-30 04:55:59.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dad9b7d963674290abdde591231e1296\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5abbd789b4c149cdbbb8108512c06720 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a93ee66b1534f80a524f6217ba0a9c1 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42fe02307c284590880221bb55b3a9bf : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/395e8df222604819a90e65750acb32f0 : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2b4271ec0794f4692b1e1ce39f7bb7b : SUCCESS in 4m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fb1f0f848e70484abdc8f2676c8e44d1 : SUCCESS in 7m 31s\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/f2e250ce3c1e4c3c80a4ab7b060d1d01 : SUCCESS in 1h 11m 20s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/763f4d748ebb4e70b09a6de90956c406 : SUCCESS in 48m 48s\n- kolla-ansible-openeuler-source https://zuul.opendev.org/t/openstack/build/766f68127cbc4ca3bd712383e7544823 : SUCCESS in 1h 26m 41s (non-voting)\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/43eee0f8bcd24cc5aa55d1783d8eb88d : SUCCESS in 46m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2ec4bef66add4aa0bab3c9e8616b02e8 : SUCCESS in 38m 03s\n- kolla-ansible-rocky9-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/0023d6c9ee134ad89b912a42f9b38790 : SUCCESS in 49m 33s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/184b65b888624d0b9727630dd0d86730 : SUCCESS in 44m 20s (non-voting)\n- kolla-ansible-rocky9-source-zun https://zuul.opendev.org/t/openstack/build/51f8fc6448a24224b9da064ff2289bb7 : SUCCESS in 1h 01m 39s (non-voting)\n- kolla-ansible-debian-source-zun https://zuul.opendev.org/t/openstack/build/6dc9cf657a11446081d9f2a7cdbdbe5f : SUCCESS in 1h 17m 14s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/e741b54c8b414913a87350ed4876114c : SUCCESS in 54m 39s (non-voting)\n- kolla-ansible-rocky9-source-swift https://zuul.opendev.org/t/openstack/build/824334e36d344401b048ada85b5a6fd6 : SUCCESS in 1h 03m 28s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/bd5d93e9359f47548eab134a7aaced82 : FAILURE in 22m 07s (non-voting)\n- kolla-ansible-rocky9-source-trove https://zuul.opendev.org/t/openstack/build/d6e6dfce2155403091284ebaa1c689fb : FAILURE in 30m 28s (non-voting)\n- kolla-ansible-ubuntu-source-trove https://zuul.opendev.org/t/openstack/build/4b03a81dab0d4618ababcdb3629835af : FAILURE in 22m 41s (non-voting)\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/b7e20bdad2434fe59d5009aa47ff881d : SUCCESS in 53m 03s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/145600a41230458494bb7fc1aaaf06fd : SUCCESS in 53m 54s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/af24d3591fbc4ceaabe5200faf0d639c : SUCCESS in 1h 00m 33s\n- kolla-ansible-rocky9-source-cephadm https://zuul.opendev.org/t/openstack/build/6a129692142b447c947169fdb930a978 : SUCCESS in 1h 14m 09s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/e746cc64d40d4f7f861107f8927bc16b : SUCCESS in 1h 00m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/517a4cebdade4bad9bc94d0b5a664371 : SUCCESS in 1h 36m 38s (non-voting)\n- kolla-ansible-rocky9-hashi-vault https://zuul.opendev.org/t/openstack/build/e8ae7555a97c4ade8d9296b6d6c0ac49 : SUCCESS in 3m 07s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"8600982a4112507f9d5ecbf9a22934dfa871bd5b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-30 05:12:25.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7073ba49e404581885205e392356c6b\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c5ee16b54dae4ae78cccc239bb7cbe90 : SUCCESS in 1h 22m 56s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/7f435f125ee4461c8c208da832d796f5 : SUCCESS in 2h 00m 09s","accounts_in_message":[],"_revision_number":15},{"id":"b6402d540fd724fcb2014cb4ca184c6607fc218d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-01-05 15:17:07.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.","accounts_in_message":[],"_revision_number":16},{"id":"066f62f6a70795cc057291715bc38fd87a4f1b0b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 17:01:39.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/767979ec86af45a193c0fc42bd5eb8ab\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7eef26342af3475e82a62328a97f14da : SUCCESS in 1h 11m 14s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/5bd167857422475abc0f8cb90dee948a : SUCCESS in 1h 22m 29s","accounts_in_message":[],"_revision_number":16},{"id":"7673ae88d8a8f0cd860aa256ade3f208db22b48a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 17:08:25.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b852a1a7c2564aa293ad8acd91888d68\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d33e3269a9c840f694da80d11dd2f1cb : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9da887acb77b4dbe90a3d339c1442a94 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16b7a27134d849309ace6e657d86ac15 : SUCCESS in 6m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0d68939146b947cd8ac64f121d7e27f2 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe5264224a114ce897fc7d1be493c12a : SUCCESS in 7m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/aa1b5582282e4fc08e8372a6c5d7a8f6 : SUCCESS in 8m 52s\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/7435e252a0ef494bbd638bfbd56988dd : SUCCESS in 1h 03m 46s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/012057766116488bbe85c58edef9c969 : SUCCESS in 50m 08s\n- kolla-ansible-openeuler-source https://zuul.opendev.org/t/openstack/build/9f86e52bd63a4fb4879318de2917ba9c : SUCCESS in 51m 49s (non-voting)\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/d8c2bed03c0843c1b27ece51209cb82f : SUCCESS in 1h 02m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/162f705c9e124b2f8b84093770c8ff64 : SUCCESS in 42m 08s\n- kolla-ansible-rocky9-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/c05119b2d3e9493fb946ff6d4f18c474 : SUCCESS in 1h 01m 47s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/9d2b15d5caf44f94a8c2f047d626ca8d : SUCCESS in 52m 26s (non-voting)\n- kolla-ansible-rocky9-source-zun https://zuul.opendev.org/t/openstack/build/b7ee52e294e54abfba9ee5982814b028 : SUCCESS in 1h 26m 46s (non-voting)\n- kolla-ansible-debian-source-zun https://zuul.opendev.org/t/openstack/build/06225b45238b438491fb511466b254e4 : SUCCESS in 1h 10m 54s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/0554c4f797034cc88c9bc32793d6f1bc : SUCCESS in 1h 11m 29s (non-voting)\n- kolla-ansible-rocky9-source-swift https://zuul.opendev.org/t/openstack/build/04188462372740ea8b80094800bd7e23 : SUCCESS in 1h 04m 08s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/c6453e455ffb4800bc5356786153bbf2 : SUCCESS in 46m 09s (non-voting)\n- kolla-ansible-rocky9-source-trove https://zuul.opendev.org/t/openstack/build/a7f440354ecf420b8c466381efc98183 : FAILURE in 1h 12m 31s (non-voting)\n- kolla-ansible-ubuntu-source-trove https://zuul.opendev.org/t/openstack/build/c2fc1df5441843b49cd9ab111b894183 : SUCCESS in 54m 45s (non-voting)\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c0a5f7dbffc948a68d19e497e43646c0 : SUCCESS in 1h 04m 23s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c76dd02f45fb478c906d9fca6f121b2b : SUCCESS in 1h 00m 01s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/b6f4a1ab22a844638402c6416d55acb5 : SUCCESS in 58m 53s\n- kolla-ansible-rocky9-source-cephadm https://zuul.opendev.org/t/openstack/build/270099d83f764c61b1338388ea6407aa : FAILURE in 59m 40s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/fd7070631f9e4d5c81364dac76e6479e : SUCCESS in 1h 05m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/aec4507a674c4b83a680afa5a400e3b0 : SUCCESS in 1h 33m 24s (non-voting)\n- kolla-ansible-rocky9-hashi-vault https://zuul.opendev.org/t/openstack/build/c75d742fde0a485cbce16ad0185216b0 : SUCCESS in 3m 13s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"60b8612adb3994cd3821238556dac99e3f9a9e7b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-17 07:10:43.000000000","message":"Patch Set 17: Patch Set 16 was rebased\n\nThe following files contain Git conflicts:\n* tests/templates/globals-default.j2\n* zuul.d/jobs.yaml\n* zuul.d/project.yaml\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"74d0213c1c1d88a5c1c41ba5f0ddb12ddb32f889","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 07:11:01.000000000","message":"Patch Set 17: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla-ansible for 863321,5","accounts_in_message":[],"_revision_number":17},{"id":"d2e162e91aef27a23a600a47d76bc803ec909a05","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 07:11:01.000000000","message":"Patch Set 17:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla-ansible for 863321,5","accounts_in_message":[],"_revision_number":17},{"id":"b53f06d82b1ead11146fcf2ba7ca98ea364ad9b4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-17 07:12:23.000000000","message":"Patch Set 18: Published edit on patch set 17.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"02ab949005ddb5ef31152d51bb3815494b47dcff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 07:12:40.000000000","message":"Patch Set 18:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla-ansible for 863321,5","accounts_in_message":[],"_revision_number":18},{"id":"5116296342411448dd5803e12386bd22fc1096de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 07:12:41.000000000","message":"Patch Set 18: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla-ansible for 863321,5","accounts_in_message":[],"_revision_number":18},{"id":"e4007cb580d534cf5ed3fef20347b549b8ee0f0b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-17 07:12:56.000000000","message":"Patch Set 19: Published edit on patch set 18.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"aee9dec29b428a24e28c96244167cd322cd91fa6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 07:13:13.000000000","message":"Patch Set 19:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla-ansible for 863321,5","accounts_in_message":[],"_revision_number":19},{"id":"5b194f72e31dc5c684476580cee0ac03cff6c5c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 07:13:22.000000000","message":"Patch Set 19: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla-ansible for 863321,5","accounts_in_message":[],"_revision_number":19},{"id":"f823289ab3db181f1fda8e0b092d3817a219caf0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-17 07:13:34.000000000","message":"Patch Set 20: Published edit on patch set 19.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"dcc02e221faeec4d82b1034daf24d22428950c95","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 07:13:50.000000000","message":"Patch Set 20:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla-ansible for 863321,5","accounts_in_message":[],"_revision_number":20},{"id":"913293674b85124f1704399813b99edf4d6cc7cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 07:13:51.000000000","message":"Patch Set 20: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla-ansible for 863321,5","accounts_in_message":[],"_revision_number":20},{"id":"557c99a50ceac14f2ee6388675abb13d953f29af","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-17 07:26:32.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":20},{"id":"8ded86fa594486ff2fcccad3fe5e550234ce736d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-17 07:26:40.000000000","message":"Patch Set 21: Patch Set 20 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":21},{"id":"1a46f957f8510920c2e09c6990a3758693b57444","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 09:32:03.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04c8900956594fbf9291de2f9c8f88b3\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5eb08b6e37964b34b79862d00b39b855 : SUCCESS in 1h 15m 25s\n- kolla-ansible-debian-aarch64-podman https://zuul.opendev.org/t/openstack/build/0c31e666a00b4c8ca90a9b01f32ebb4b : SUCCESS in 1h 10m 37s\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/49ec2b2701314d01a62c6a96baec0205 : SUCCESS in 2h 02m 46s","accounts_in_message":[],"_revision_number":21},{"id":"8210350fb5e35045b122ed86f37d8f701987c1c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 10:06:01.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6da0e482d1304effa1332b1ea545f81b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89b2f435bf57493ba2ee418768f9065f : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d33bf50d9104cf49647212fe3fbf160 : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9fb27f6927d24fa38e257c3910173de8 : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/547f37fa75b941b9b8a69201c3c332ae : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b90e10c6876b4b4899c4f92b67e66916 : SUCCESS in 4m 59s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/689469f3288d46c889acbf73d6aaf22c : SUCCESS in 6m 22s\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/e78103e2a6524420b82b04e574dab1d1 : SUCCESS in 1h 01m 08s (non-voting)\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/56ed7892139747d6a1ff35f5945ba81f : SUCCESS in 44m 54s\n- kolla-ansible-debian-podman https://zuul.opendev.org/t/openstack/build/a602def4dd324e9da2914fbfe0d579e9 : SUCCESS in 1h 00m 52s\n- kolla-ansible-openeuler https://zuul.opendev.org/t/openstack/build/dfc2a8ca2e244f29ae9e8f5553884513 : SUCCESS in 1h 16m 19s (non-voting)\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d39a8269eccf438dbda4cc1744664768 : SUCCESS in 55m 17s\n- kolla-ansible-rocky9-podman https://zuul.opendev.org/t/openstack/build/cda76582de614aa8b569493b674a7434 : SUCCESS in 1h 01m 47s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/072646eec0a54d1ab27fe5b6e37ddffc : SUCCESS in 45m 06s\n- kolla-ansible-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c4d7412fa51944dbb1ebbc49bb271365 : SUCCESS in 1h 00m 34s\n- kolla-ansible-rocky9-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/d9a75fa382574c6dac4c5c09e59058f9 : SUCCESS in 57m 09s (non-voting)\n- kolla-ansible-ubuntu-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/9b9d23cc984447f1ae9d886063211ea9 : SUCCESS in 53m 12s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/e9ac20c37dbb4305afc35e4533129f9b : SUCCESS in 1h 13m 50s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/2b027902a33a4497a4588d3ab15d7c73 : SUCCESS in 1h 22m 36s (non-voting)\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/1f9990af36d94849a2ddf83d79b3bc97 : SUCCESS in 1h 15m 05s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ab0fdb37f6624f44b96021e881c90062 : SUCCESS in 1h 24m 03s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1bd059b17c824da7a2fbcfc089833fdd : SUCCESS in 1h 12m 55s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/dec58c09a31b47c590bf60c93c0ec82e : SUCCESS in 1h 26m 13s (non-voting)\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/87f2a98ce28f43ed871444d7096294cf : SUCCESS in 1h 09m 45s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/1dc369b6b8d14d18bd11776456cad1f6 : SUCCESS in 2h 18m 01s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/ed7525ce623e4112ad16b6e52e024907 : SUCCESS in 2h 33m 31s (non-voting)\n- kolla-ansible-rocky9-trove https://zuul.opendev.org/t/openstack/build/3109f23f7f7a47a6a24dc231b9665e10 : FAILURE in 4m 12s (non-voting)\n- kolla-ansible-ubuntu-trove https://zuul.opendev.org/t/openstack/build/bb534abb966f46d197a88f2352ab55d8 : FAILURE in 3m 50s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"f55c0aae9f5a0fe38a30d766a7ab348338ba22e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-04 06:56:15.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":22},{"id":"27ab6604c95e096b70126faccfd31705179759cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 07:01:20.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67abcbcba78643cbb74e8163341ac2b2\n\n- kolla-ansible-rocky9-trove https://zuul.opendev.org/t/openstack/build/46834e4d9656482f988addbf6cc764a2 : FAILURE in 4m 04s (non-voting)\n- kolla-ansible-ubuntu-trove https://zuul.opendev.org/t/openstack/build/f43d43c430614bd0ba52a6d2ddc8b292 : FAILURE in 4m 03s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"0625c546ac10215abf90556fe53b89ab61a66bdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-04 07:07:11.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":23},{"id":"24e829cd813b8a3e08f51e869f71badecfd6a9a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 07:46:27.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cda65e5491b243ea85473884d76a9909\n\n- kolla-ansible-rocky9-trove https://zuul.opendev.org/t/openstack/build/96bb5e3187144d5dab7aa769d4091154 : FAILURE in 34m 02s (non-voting)\n- kolla-ansible-ubuntu-trove https://zuul.opendev.org/t/openstack/build/07d9cf6d9b834945ac0b6ec8a0ab5266 : FAILURE in 25m 59s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"bc402b98482bd5e43d4b3f8e2a1bb6c62ef78111","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-04 08:23:23.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"9f969a34fd65e1540bfd70910a5b002d8654ff80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-04 08:31:18.000000000","message":"Uploaded patch set 24.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":24},{"id":"def97cb6f83316766582554414fa19a22dbb2e3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 09:22:06.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fea3931dab6d499dbbfebda2940097e2\n\n- kolla-ansible-rocky9-trove https://zuul.opendev.org/t/openstack/build/b55ce2d169d3432399308f0c9787687f : FAILURE in 43m 29s (non-voting)\n- kolla-ansible-ubuntu-trove https://zuul.opendev.org/t/openstack/build/09e35546eac146d294b26a6d656015c5 : FAILURE in 29m 22s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"2a008babbd26409dbef6562429d89538d1d64e1b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 10:36:59.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96fc4f97e02744a59b1327bae0a328b3\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1efadb45c7f04c37aeaea061cf6bee3a : SUCCESS in 55m 17s\n- kolla-ansible-debian-aarch64-podman https://zuul.opendev.org/t/openstack/build/31b1925e08214da8bc68ae36fd7a334b : SUCCESS in 1h 15m 23s\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/9afa682af27b464a8082e24bf1ee14fd : SUCCESS in 2h 03m 09s","accounts_in_message":[],"_revision_number":24},{"id":"5d784095978b0b32df4e0cc1fa1fe421e97fb09f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-04 13:37:05.000000000","message":"Uploaded patch set 25.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":25},{"id":"98180f297a435a55e7198e803b155bdad3a7c803","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 14:57:43.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fee03734131c4548893f21e008ce206b\n\n- kolla-ansible-rocky9-trove https://zuul.opendev.org/t/openstack/build/7f73f69ff3e549028e72730748e29712 : FAILURE in 1h 13m 19s (non-voting)\n- kolla-ansible-ubuntu-trove https://zuul.opendev.org/t/openstack/build/467c12c74748466c93e8372ab9ced722 : FAILURE in 1h 01m 30s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"08d2cce1d168c22143ad52882bbb15e89b24efe7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 15:54:18.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1a22833e58c4243836e7e17bc8fd6c4\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/66468aa6d4684d98824b60f4f9d7180b : SUCCESS in 48m 09s\n- kolla-ansible-debian-aarch64-podman https://zuul.opendev.org/t/openstack/build/5b09eda466874a098218883515196943 : SUCCESS in 1h 15m 37s\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/bad85c91641f43e081459d3d75521759 : SUCCESS in 2h 14m 59s","accounts_in_message":[],"_revision_number":25},{"id":"5490eb49c58104e30988b7496ebae6b6e472e65a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-05 03:25:33.000000000","message":"Uploaded patch set 26.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":26},{"id":"bc5c1a5331f7b7d6ff4f1e23a2641e74f81d931f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 04:55:44.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61911d9383b9476b96d19de023bbd65d\n\n- kolla-ansible-rocky9-trove https://zuul.opendev.org/t/openstack/build/392d92e6211d4230bff5b3c669994178 : FAILURE in 1h 11m 57s (non-voting)\n- kolla-ansible-ubuntu-trove https://zuul.opendev.org/t/openstack/build/613dca554d2d4395b5d2f4116201ceae : FAILURE in 1h 08m 43s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"6af2a913cc94ed2c896d381f75212b33c67d58f6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 04:57:28.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e672a782c3d5466681de27ae1e427a50\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/92bcc09a1d5b494c8dff84c86c60383c : SUCCESS in 46m 31s\n- kolla-ansible-debian-aarch64-podman https://zuul.opendev.org/t/openstack/build/12ece37de93741a086646e12e207fbad : SUCCESS in 1h 14m 03s\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/c65040d55d7a41cab0090d19724df04d : SUCCESS in 1h 29m 44s","accounts_in_message":[],"_revision_number":26},{"id":"74dfd7451893205f6942874e99020add99db534c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-05 06:21:37.000000000","message":"Uploaded patch set 27.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":27},{"id":"6c4064a3adfab5b0180c9c0e395d4436ddff6fd7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 07:41:06.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f5106fb1aa94fcea40b2ad08a0812b8\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/53a5dba766344301ba8e24bd26ea96eb : SUCCESS in 51m 38s\n- kolla-ansible-debian-aarch64-podman https://zuul.opendev.org/t/openstack/build/ea70631eb8d440848d5c983bbd6fb55f : SUCCESS in 1h 06m 21s\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/c30d4e0e75574ff9a1b49b2f493e6709 : SUCCESS in 1h 18m 17s","accounts_in_message":[],"_revision_number":27},{"id":"06ce1bff842f083f768518bffa79bb8a7bbe705b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 07:46:02.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fc40657b07e4381a3f5a780c4ecda74\n\n- kolla-ansible-rocky9-trove https://zuul.opendev.org/t/openstack/build/563068db88914df9bf5fc7acc792e903 : FAILURE in 1h 17m 59s (non-voting)\n- kolla-ansible-ubuntu-trove https://zuul.opendev.org/t/openstack/build/8cd4d5fb15fb4e30a2cd71b70ca6dd94 : FAILURE in 54m 58s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"04659157ebf79712bb9fcde6f7eba65423cba3d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-06 02:33:44.000000000","message":"Uploaded patch set 28.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":28},{"id":"905f28fa3436f0237754b0d9e1d9041243c51e87","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 03:55:03.000000000","message":"Patch Set 28:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d8c9b5f4d60495584b59b41cfe659b6\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5dcc31f21491429b9036dcce163fcef3 : SUCCESS in 50m 11s\n- kolla-ansible-debian-aarch64-podman https://zuul.opendev.org/t/openstack/build/e6664b83e8dd4fea994e753ba7dbba35 : SUCCESS in 1h 18m 46s\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/6bb4b78b6555425f80bc50e4a20ae176 : SUCCESS in 1h 18m 21s","accounts_in_message":[],"_revision_number":28},{"id":"19637d6ef4556e2627fdc308bbeb90029378fbcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 04:57:41.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c03bac578b64b6b9470de98d346b9ba\n\n- kolla-ansible-ubuntu-trove https://zuul.opendev.org/t/openstack/build/79898247c67a4c34a8796915c11bc447 : FAILURE in 1h 44m 43s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"1ec92b935ae2555c6fb8f3ec5addd37cb303b72b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-06 06:01:36.000000000","message":"Uploaded patch set 29.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":29},{"id":"d51a21cf39a38b31666cb1d5c1b4f8476cebcc5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 07:35:02.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1674ee7441ec4d29a8505d8d0f8b65f4\n\n- kolla-ansible-ubuntu-trove https://zuul.opendev.org/t/openstack/build/76a3ccca3e6547c18b084e24fffab92e : FAILURE in 1h 32m 16s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"3a4a3e8ba9995336dce24988d61dbd6d0e503e7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-06 07:44:07.000000000","message":"Uploaded patch set 30.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":30},{"id":"4a85ca1d4d92dcc3390c2460f25864f6bc318900","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 08:26:37.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c7d7d24c12e4c49980b603b5eaa0091\n\n- kolla-ansible-ubuntu-trove https://zuul.opendev.org/t/openstack/build/2feacb5ce6024fe583bf91c3696cb9f7 : FAILURE in 36m 29s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"2ee4a9cdef1bb8299327097870393ef886ecba26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-06 09:59:12.000000000","message":"Uploaded patch set 31.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":31},{"id":"c6ad60b222890955cf1012582a99dd71312af2e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 11:42:09.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c708ed5b7e81476089a55a4825d1fc64\n\n- kolla-ansible-ubuntu-trove https://zuul.opendev.org/t/openstack/build/4e05fac775fa43a0bed0e7667842fdd1 : FAILURE in 1h 41m 03s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"241b0aaffda00ffab83f1e0600c27ce355f97155","author":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},"date":"2023-12-06 15:26:23.000000000","message":"Patch Set 31: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"23d62681217734272282aa94c2d91a38769a2657","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-07 02:07:48.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"d9f53c6079823ee2ae68a2e65158ba6baef39ea8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-07 02:09:37.000000000","message":"Uploaded patch set 32.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":32},{"id":"7e4b7166ff0c6a031b7f5c4541358f07c50582d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 05:07:16.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25133248289047278532c51bc6c81588\n\n- kolla-ansible-ubuntu-trove https://zuul.opendev.org/t/openstack/build/f421bbb8499c4b338c8d57914319dcfb : FAILURE in 1h 45m 07s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"4dddbfa7ea0f3d516277c4b7ccaf627f3950e289","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-07 05:49:02.000000000","message":"Uploaded patch set 33.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":33},{"id":"4b8fb9f09e12758f423fda4cd7db8a751bf07ee9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 07:22:42.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5d9215975614555838f57a7fe728fb2\n\n- kolla-ansible-ubuntu-trove https://zuul.opendev.org/t/openstack/build/a5e6af2490584685a9a41aa4de056552 : FAILURE in 1h 33m 05s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"1d947d464c121e524e1402478130e41d4da511a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-07 10:05:48.000000000","message":"Uploaded patch set 34.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":34},{"id":"39de7ca8fa33e5dc5b976ea1927cd074a0921e79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 11:41:14.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3152541cbd845108068be4fda644905\n\n- kolla-ansible-ubuntu-trove https://zuul.opendev.org/t/openstack/build/7c58fbd4a0414ab08d8a5a96f20127dd : FAILURE in 1h 31m 52s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"756abd8012dfead55e7ddbdc5eebcdf42f530f9f","author":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},"date":"2023-12-07 14:13:35.000000000","message":"Patch Set 34: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":34},{"id":"bff5be736e89b1a453dfe219e4b7659d511480a0","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-08 00:59:22.000000000","message":"Patch Set 34:\n\n(1 comment)","accounts_in_message":[],"_revision_number":34},{"id":"542641921643a287a4c8ef6f0b56ad0b9031fc2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-08 01:10:47.000000000","message":"Uploaded patch set 35.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":35},{"id":"4056dd23ce632fa17aeda16537b5e8a0eab77411","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-08 01:15:16.000000000","message":"Patch Set 36: Patch Set 35 was rebased","accounts_in_message":[],"_revision_number":36},{"id":"0a9b4252a1daa7620c6bda61afb1d0d030341dcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 03:01:13.000000000","message":"Patch Set 36: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7ba40f90127473bae5cec228295bb15\n\n- kolla-ansible-ubuntu-trove https://zuul.opendev.org/t/openstack/build/2ea9322a684947b289991d472d58b7e6 : FAILURE in 1h 40m 23s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"033c3590652a322a74b7d4b84aa3ceecf2bee0ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-08 08:13:13.000000000","message":"Uploaded patch set 37.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":37},{"id":"7ba723823c053679857243b78b0bd072b51aa7fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-08 08:21:04.000000000","message":"Uploaded patch set 38.","accounts_in_message":[],"_revision_number":38},{"id":"90fb17d3620f0e7a1e2642570094b265f1d7d460","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 10:08:32.000000000","message":"Patch Set 38: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f6517070474428d8fedabc737db4a79\n\n- kolla-ansible-ubuntu-trove https://zuul.opendev.org/t/openstack/build/8916f15e6d43461ba02a6a3d71ccebfa : FAILURE in 1h 41m 23s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"7ed800ebf23569a306cad6f2f35dd986b39e221d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-11 01:44:56.000000000","message":"Uploaded patch set 39.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":39},{"id":"0de1bd05c89a79f33cd71bf27c98461c0131398c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-11 01:45:15.000000000","message":"Patch Set 39: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla-ansible for 863521,39","accounts_in_message":[],"_revision_number":39},{"id":"e4e3a490ae14d341c85d0a88087a17eefcfb4df5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-11 01:45:16.000000000","message":"Patch Set 39:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla-ansible for 863521,39","accounts_in_message":[],"_revision_number":39},{"id":"c11d1583f922435d3a21121e6758c2342f3d4aa9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-11 03:16:48.000000000","message":"Uploaded patch set 40.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":40},{"id":"42440edb805637a654c5dbe3a6305c19dcd6800c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-11 04:08:22.000000000","message":"Patch Set 40: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af6f5d1f8474404e99d326f20ce8829d\n\n- kolla-ansible-ubuntu-trove https://zuul.opendev.org/t/openstack/build/c64c75e8a42b4fbdbd98718a002107cd : SUCCESS in 50m 18s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"41d4c130dc2dce2677b1946c2f81a88601acbd18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-11 06:42:24.000000000","message":"Uploaded patch set 41.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":41},{"id":"c63bfedcf5e1f10915dbe2c181434c2195cd4625","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-11 06:49:34.000000000","message":"Patch Set 41:\n\n(1 comment)","accounts_in_message":[],"_revision_number":41},{"id":"340923b123781d76c281983920c889f9c2ecb3d5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-11 08:18:50.000000000","message":"Patch Set 41:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab5125c0f8bb481ca111b0c9545c7656\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9956c0f91a0e4bfb8c041343c19a9b83 : SUCCESS in 47m 25s\n- kolla-ansible-debian-aarch64-podman https://zuul.opendev.org/t/openstack/build/c920712ac3764995a4a1e32ac3c3a4be : SUCCESS in 1h 10m 07s\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/35fd9e680e2a41c9b6f49a48b44d720c : SUCCESS in 1h 35m 14s","accounts_in_message":[],"_revision_number":41},{"id":"67383e31fd0f15aea70958812f5212e31aafe884","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-11 09:31:27.000000000","message":"Patch Set 41: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bf94176fd7044358819d99c30ccc09b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/def487ffc43c45ebaae91b199ed5c47c : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d96464939ef48c68c5f946e39d13d9a : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0d35158778b4f59b4ab8c7bdd44a7fc : SUCCESS in 3m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1fe5e91ba05d4ba98050272ad7630751 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02c05ac890c74096870b178261926789 : SUCCESS in 4m 32s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b27e6852b0514e72870c4b4fdb8f9184 : SUCCESS in 6m 06s\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/592f2333c853463cb05486894d28afb3 : FAILURE in 9m 00s (non-voting)\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/511f7ea3d27c4714bb655509ef761a50 : SUCCESS in 50m 14s\n- kolla-ansible-debian-podman https://zuul.opendev.org/t/openstack/build/bad514a482e6464cbe3655c94db85e18 : SUCCESS in 1h 04m 54s\n- kolla-ansible-openeuler https://zuul.opendev.org/t/openstack/build/f3eb03e57d3942998f3ce92f2718fbd2 : SUCCESS in 1h 02m 19s (non-voting)\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/19f75f1a40e1402fa3d4dc83de0477d3 : SUCCESS in 1h 00m 15s\n- kolla-ansible-rocky9-podman https://zuul.opendev.org/t/openstack/build/cd94d52e9a1a4e6791207031c065cf7b : SUCCESS in 1h 03m 05s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/96274bdd601e4368a5012581726a500e : SUCCESS in 46m 57s\n- kolla-ansible-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e6eb0d215d334b4e9a6e803b6a3bcb1d : SUCCESS in 1h 03m 14s\n- kolla-ansible-rocky9-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/15537d40c26940499f9580d366cac5b4 : SUCCESS in 1h 03m 33s (non-voting)\n- kolla-ansible-ubuntu-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/e2233b1eacec42b3897ef5ab6cbd1a98 : SUCCESS in 52m 06s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/4c4164ae0fe3493494d5f072e01736a7 : SUCCESS in 1h 16m 09s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/9bf2a457210d49a6bd12e71f1def013e : SUCCESS in 53m 46s (non-voting)\n- kolla-ansible-rocky9-magnum https://zuul.opendev.org/t/openstack/build/8d1b8a34e79742d3b0b3e1c102e04e2d : SUCCESS in 1h 00m 29s (non-voting)\n- kolla-ansible-ubuntu-magnum https://zuul.opendev.org/t/openstack/build/f4fb3299573b465c94761682543987e5 : SUCCESS in 46m 16s (non-voting)\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/375422d83818409ab65a273f2f7ad404 : SUCCESS in 1h 33m 48s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0e86dab4e52a4dfbacef17e9aad4090d : SUCCESS in 1h 12m 44s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9a05215311934f06bede2377aff79c4e : SUCCESS in 1h 13m 40s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/334377d1bbbf489388ecec56e6bd5624 : SUCCESS in 1h 47m 41s (non-voting)\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/02fca82c529f4c4996e5897b96ee9843 : SUCCESS in 1h 15m 09s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/632fc806e89442b4819952b83cd9048a : SUCCESS in 2h 41m 20s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/1b113c10159f42f0a78c44ba6e4c0bba : SUCCESS in 2h 08m 17s (non-voting)\n- kolla-ansible-rocky9-trove https://zuul.opendev.org/t/openstack/build/5c0e2a11b4e0455992510e2caa637c78 : FAILURE in 1h 41m 53s (non-voting)\n- kolla-ansible-ubuntu-trove https://zuul.opendev.org/t/openstack/build/c9f3d79e86e24b7c9aebd1740adf0ed8 : SUCCESS in 55m 05s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"367b5ce8aed01866424ec1f9e5861bf75156aa82","author":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},"date":"2023-12-11 13:27:14.000000000","message":"Patch Set 41: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":41},{"id":"c04e1246e3404db2359c91a8958c408b9c07361e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-11 15:57:30.000000000","message":"Patch Set 41:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d661f1b7bf34419a8e7f94a93251a30\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e76db8457f0c4b0a9549c28d5e1e5654 : SUCCESS in 57m 59s\n- kolla-ansible-debian-aarch64-podman https://zuul.opendev.org/t/openstack/build/e9822d7b27784ab5b2920b8092e5cfe5 : SUCCESS in 1h 57m 20s\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/31103b7eed5e4b24941a1770e3345f52 : SUCCESS in 2h 07m 45s","accounts_in_message":[],"_revision_number":41},{"id":"b33ceca8afcefa0d4aeb1bc09335ba1f2d08cdab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-11 16:05:59.000000000","message":"Patch Set 41:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53232653106941d5b80392e2492b70c5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c54a90dbf1c84831a6126d39e84dbb8b : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce54fd220425439ab5eb0f29078d4932 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/163c40aeaaf34c7fbdeb00125d3732ea : SUCCESS in 3m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/08d4750eabf946c9957bd227eb8c49b1 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55b53648ebc54cdf8b07337b949126be : SUCCESS in 5m 54s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/15f65d434b074f7782a29d5b978e6c7d : SUCCESS in 7m 28s\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/4764c9e0070943c4a9345cc70388cc52 : FAILURE in 11m 32s (non-voting)\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4bee2e71fb91499facdd464b539e2edc : SUCCESS in 55m 36s\n- kolla-ansible-debian-podman https://zuul.opendev.org/t/openstack/build/122006017f274e648dfb3851da7814e6 : SUCCESS in 1h 03m 11s\n- kolla-ansible-openeuler https://zuul.opendev.org/t/openstack/build/b49b62e3db6f44c69c28af36efb7f534 : SUCCESS in 1h 03m 12s (non-voting)\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/682b0afb5d234c0e8d949787473f52f7 : SUCCESS in 58m 42s\n- kolla-ansible-rocky9-podman https://zuul.opendev.org/t/openstack/build/55fc77b7076343b8979ae57bb122f059 : SUCCESS in 1h 05m 32s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/981e292208c944f5a18cb459387e068a : SUCCESS in 44m 03s\n- kolla-ansible-ubuntu-podman https://zuul.opendev.org/t/openstack/build/58e2ad64d33a41a9994c7cd868835d0f : SUCCESS in 1h 07m 12s\n- kolla-ansible-rocky9-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/cf6fd044b8de4884afd7a13cd7803307 : SUCCESS in 1h 13m 04s (non-voting)\n- kolla-ansible-ubuntu-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/ea485ad3f84241ac9319b06ba090227b : SUCCESS in 48m 52s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/4a8fa5265bc44430b82acac8facdf3e4 : SUCCESS in 1h 14m 14s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/95ba254a988b4f96b44b870f78c9addd : SUCCESS in 1h 03m 53s (non-voting)\n- kolla-ansible-rocky9-magnum https://zuul.opendev.org/t/openstack/build/8fbb0c86d65f48ae87c2d4151a78d596 : SUCCESS in 56m 13s (non-voting)\n- kolla-ansible-ubuntu-magnum https://zuul.opendev.org/t/openstack/build/88bac07c57bb4e30a46997357e7343e9 : SUCCESS in 42m 56s (non-voting)\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/86b929c7d3504328a7e8a8d5845e63d4 : SUCCESS in 1h 33m 22s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/00a3f6394afe4223b2423c52d84b54a8 : SUCCESS in 1h 25m 04s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/cdbc5e77323d475e82c50125473e30fb : SUCCESS in 1h 21m 16s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/9557b631e152415ca17f1fa7b907d5e5 : SUCCESS in 1h 38m 39s (non-voting)\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/15b4e9f1030d4979acbdb85e38f8ecd7 : SUCCESS in 1h 24m 12s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/bb685fa0c4fb4353bf0c3fbaa5362a0a : SUCCESS in 2h 31m 27s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/60d4604875c644569235f7d8f8378e1a : SUCCESS in 2h 15m 54s (non-voting)\n- kolla-ansible-rocky9-trove https://zuul.opendev.org/t/openstack/build/222ed2e62e7a45c7916f0aa634c08ed1 : FAILURE in 1h 48m 15s (non-voting)\n- kolla-ansible-ubuntu-trove https://zuul.opendev.org/t/openstack/build/73e4fbe1604d434db59cd2d06f80a6f0 : SUCCESS in 1h 05m 03s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"e488287a024537aaafec2c270a21c2b96aa4114b","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-12 01:29:14.000000000","message":"Patch Set 41:\n\n(1 comment)","accounts_in_message":[],"_revision_number":41},{"id":"310dbbfe7cdcd857d377f623a84663143433489d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-14 09:01:42.000000000","message":"Uploaded patch set 42.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":42},{"id":"030f8f5b77170911db571e8f371cfedadf3b6ba8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 11:47:47.000000000","message":"Patch Set 42: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22cac5dc249248e79267f65c67270210\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe66527c53744fd7a9f376c86da62503 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee81c4a38ca44460bfece24cf606f52d : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5914124c35b4d989bc2443fcaed7385 : SUCCESS in 4m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9854acaac4114c7d92805e2fb88793f7 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ebbd57f7c444d6682d4efbbb0d5ce38 : SUCCESS in 5m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e18293d8131542d58b5df02cbd1b62ff : SUCCESS in 6m 18s\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/e2c84471bfb840d6b8b18de34de02298 : FAILURE in 9m 00s (non-voting)\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/aa539be2516c4944b3927db52d973240 : SUCCESS in 39m 41s\n- kolla-ansible-debian-podman https://zuul.opendev.org/t/openstack/build/079f87c862984c058101ba46f2ed0daf : SUCCESS in 57m 36s\n- kolla-ansible-openeuler https://zuul.opendev.org/t/openstack/build/5c20192d6e1a4512ad19265fd163dab1 : SUCCESS in 1h 03m 33s (non-voting)\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a3e8686f66234bc7b76d063091aa99a2 : SUCCESS in 57m 13s\n- kolla-ansible-rocky9-podman https://zuul.opendev.org/t/openstack/build/2c7d052548e340e2ba31cb7857c3d76d : SUCCESS in 56m 50s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e87b136d9dc74c37b1e7ab79dbf44b21 : SUCCESS in 43m 46s\n- kolla-ansible-ubuntu-podman https://zuul.opendev.org/t/openstack/build/d7c6080ce97e46bcb89d7db1876bd993 : SUCCESS in 1h 04m 17s\n- kolla-ansible-rocky9-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/a28f2e1b2e9044829c211066d122fac3 : SUCCESS in 1h 04m 00s (non-voting)\n- kolla-ansible-ubuntu-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/987c3367e6f949a1b4014300f17764e5 : SUCCESS in 53m 48s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/3cdb06cfce24451c913a195603067207 : SUCCESS in 1h 20m 22s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/c8d188fd83714e40a47353f6ef010b9c : SUCCESS in 1h 08m 46s (non-voting)\n- kolla-ansible-rocky9-magnum https://zuul.opendev.org/t/openstack/build/fdf9dd943fd54397b75b373601e8b06c : SUCCESS in 52m 58s (non-voting)\n- kolla-ansible-ubuntu-magnum https://zuul.opendev.org/t/openstack/build/dbb3e18803e449edbfde752f943557f3 : SUCCESS in 48m 01s (non-voting)\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/29172dd9f8474d5180e13b868814da0f : SUCCESS in 1h 26m 23s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5721695f948c410185b942420933b028 : SUCCESS in 1h 26m 45s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/71ca25e40c6a4be0a6d13cd30fd34e40 : SUCCESS in 1h 34m 13s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/5b5bca18fc41465b8c5a6ce041ab8af5 : SUCCESS in 1h 42m 33s (non-voting)\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/920ec86db2af4293a669efe810f47b19 : FAILURE in 28m 57s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/8a684c1a67f2423ea2aaa0ceecb93d9e : SUCCESS in 2h 38m 53s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/22f78b0ca2e84e538020128983509a0f : SUCCESS in 2h 23m 19s (non-voting)\n- kolla-ansible-rocky9-trove https://zuul.opendev.org/t/openstack/build/efa7bdf73425441f97f6c73ac3d1bf71 : FAILURE in 1h 42m 44s (non-voting)\n- kolla-ansible-ubuntu-trove https://zuul.opendev.org/t/openstack/build/7d8a200186bc4b0cb959ec4fe7b30c80 : SUCCESS in 1h 09m 30s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"2450c77a3242b783e6ab3922dcb5f9221ed19b9e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 12:05:06.000000000","message":"Patch Set 42:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec46f783880142cea1d619b3a25109c0\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5f1aea1c74384ec5b19ac96eb7489ac7 : SUCCESS in 58m 13s\n- kolla-ansible-debian-aarch64-podman https://zuul.opendev.org/t/openstack/build/1c333e1c378642faa0564d76f3fbd290 : SUCCESS in 1h 04m 45s\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/6e82216a313a47f7acf6dfccdb6456ee : TIMED_OUT in 3h 02m 07s","accounts_in_message":[],"_revision_number":42},{"id":"dedd335f62a47ebc577a543d8dcfb38c0da9122e","author":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},"date":"2023-12-15 13:32:52.000000000","message":"Patch Set 42: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":42},{"id":"84e4f0c47c78d70f83764b5531576211cc8132b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-18 02:14:35.000000000","message":"Uploaded patch set 43.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":43},{"id":"7daf2528697a95e5e7a252b34ff2831afff862bb","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-18 02:21:36.000000000","message":"Patch Set 43:\n\n(1 comment)","accounts_in_message":[],"_revision_number":43},{"id":"00624fb4ba7dbda94a7778eb7ad3eb3db03bd778","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-18 04:34:02.000000000","message":"Patch Set 43:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf7fb15656424362a512f5ab6e981abf\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6ebe8921d22543209009b4047c7896db : SUCCESS in 1h 30m 22s\n- kolla-ansible-debian-aarch64-podman https://zuul.opendev.org/t/openstack/build/8bf54837e12d4eee979671bd5be2500b : SUCCESS in 1h 55m 49s\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/66aef2517c444899aae2f2e24375b494 : SUCCESS in 2h 15m 34s","accounts_in_message":[],"_revision_number":43},{"id":"05aeaf53bca5b0b87be2ff485508da4ab81b9a5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-18 05:23:30.000000000","message":"Patch Set 43: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4f272b4b26d40059fca82872f118cdb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/334b6730585b4ee8bd6552445785faa6 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f27dabad91084efc8bd25c52c2b20b1f : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04ffa26cf9574a7cb182bc8e99f6b0e4 : SUCCESS in 4m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f141f68265934d29b47c9c37340f0bdd : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/079b636038234ad7bdb4ef657e3590ee : SUCCESS in 5m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/78b63ff9826243449942368fc8ef3880 : SUCCESS in 6m 33s\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ff09c4360a784088bfce40fa816f3516 : SUCCESS in 1h 01m 16s (non-voting)\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/dd63f85cffc4450ea1b6fd340a28d621 : SUCCESS in 47m 25s\n- kolla-ansible-debian-podman https://zuul.opendev.org/t/openstack/build/c049b3968e624f59adc40fd5ec89d6bc : SUCCESS in 1h 04m 48s\n- kolla-ansible-openeuler https://zuul.opendev.org/t/openstack/build/0ccfc0491b864f27b3e86b2bf300672d : SUCCESS in 1h 31m 30s (non-voting)\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/177af3dc5e074eeba44a391e40a58402 : SUCCESS in 59m 19s\n- kolla-ansible-rocky9-podman https://zuul.opendev.org/t/openstack/build/8f725db5e8d34d22be5dd05c58cbd363 : SUCCESS in 1h 08m 24s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4dfba9a05ed64c2bbc5a16a420ddf233 : SUCCESS in 52m 21s\n- kolla-ansible-ubuntu-podman https://zuul.opendev.org/t/openstack/build/7688e7d296cc4deca603d12e1da8cc56 : SUCCESS in 55m 29s\n- kolla-ansible-rocky9-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/1507c2cf53204afe9f7e1131f970074f : SUCCESS in 1h 01m 41s (non-voting)\n- kolla-ansible-ubuntu-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/a1b1da116a3d4f9da94d1ca4327d4604 : SUCCESS in 59m 52s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/d5e698a957ff418bbaa654504ba31909 : SUCCESS in 1h 10m 45s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/9006fec9ea1240b6bd25984902513c47 : SUCCESS in 1h 10m 04s (non-voting)\n- kolla-ansible-rocky9-magnum https://zuul.opendev.org/t/openstack/build/da5cc9b8de314e59a727cd344e307674 : SUCCESS in 1h 02m 51s (non-voting)\n- kolla-ansible-ubuntu-magnum https://zuul.opendev.org/t/openstack/build/7b7058cefcc940f38f31e51963c57887 : SUCCESS in 50m 58s (non-voting)\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/64f68baaa6d647c6b2b9ed256754863d : SUCCESS in 1h 31m 00s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0fe047a3eb2d4e8ba8736e6fc337ab3e : SUCCESS in 1h 21m 46s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/90069667a0a54fd2900db45ea09b69b3 : SUCCESS in 1h 29m 04s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/9d2efa278c03466e98b02007030c28ad : SUCCESS in 1h 35m 07s (non-voting)\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/a6cbcd99d6a84ba4bee80f8da833046f : SUCCESS in 1h 38m 59s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/b9c139e44bf64a8b8e8eec7117323914 : SUCCESS in 2h 09m 25s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/fc12f6a074594c5c81470d1577b6ccc9 : SUCCESS in 2h 23m 42s (non-voting)\n- kolla-ansible-ubuntu-trove https://zuul.opendev.org/t/openstack/build/89c028c29ae34b2d913b25d4cfabad5f : SUCCESS in 1h 05m 06s (non-voting)","accounts_in_message":[],"_revision_number":43},{"id":"4060d6afd06eab22fac229685576c1cefd726a74","author":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},"date":"2023-12-18 13:46:14.000000000","message":"Patch Set 43: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":43},{"id":"47d4d572e3744b8be7d29bce7b2a87b9220cce50","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-26 03:35:54.000000000","message":"Patch Set 43:\n\n(2 comments)","accounts_in_message":[],"_revision_number":43},{"id":"9df59191a77de6d35f3031d0acf8ae7e314870b0","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2024-01-07 17:33:28.000000000","message":"Patch Set 43: Code-Review+1","accounts_in_message":[],"_revision_number":43},{"id":"3b1917f03001b1a776ab957daec4a8c252abc861","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2024-01-08 02:38:04.000000000","message":"Patch Set 43: Code-Review+1","accounts_in_message":[],"_revision_number":43},{"id":"12c7de8f3d1073813d28152e8f1cdc4a286bd073","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-02-22 13:43:13.000000000","message":"Uploaded patch set 44.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":44},{"id":"e9227e808e093fc7667e7178b9e235b285bedd60","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 15:15:16.000000000","message":"Patch Set 44:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa681c6f205a405185f8ca1034c50aa4\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2868f539696a4a6db8ffb36dac930975 : SUCCESS in 58m 01s\n- kolla-ansible-debian-aarch64-podman https://zuul.opendev.org/t/openstack/build/076c1954ffd0406aa85e66a005682e16 : SUCCESS in 59m 34s\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/6d952515a4784562aceaa6b07b13a9e9 : SUCCESS in 1h 29m 46s","accounts_in_message":[],"_revision_number":44},{"id":"b7e5a19c95ca5a8ad04a713bd0ec776f033f52de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 15:38:47.000000000","message":"Patch Set 44: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1e3e63afa4c47cb8cf626697c7d6d0c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18e23d7178f14661b9ca58e52a262242 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2867932b20bc4c419e6d6faf6f764e89 : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd7d904bf78242c7919e4f19dfb61adc : SUCCESS in 3m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bc43e3920a57414581acd61792243b99 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33ab01d402754cbd9c4467a3c6a0a0f4 : SUCCESS in 5m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8560a62e40704cdc8bbe56d5bed932b5 : SUCCESS in 8m 00s\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/6a95e86a004b4d93ab7f93c8fe1e9acd : SUCCESS in 1h 01m 15s (non-voting)\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e06a95bbcd614b37aa3d61962c2cb294 : SUCCESS in 51m 12s\n- kolla-ansible-debian-podman https://zuul.opendev.org/t/openstack/build/d049d4fce6d4465c957f7425109d2d0d : SUCCESS in 58m 18s\n- kolla-ansible-openeuler https://zuul.opendev.org/t/openstack/build/f8cf0a5aafa24e07a24467a47092e9bc : SUCCESS in 51m 02s (non-voting)\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/85148e5998e94afcad87daa8c225c1ac : SUCCESS in 1h 03m 54s\n- kolla-ansible-rocky9-podman https://zuul.opendev.org/t/openstack/build/729b2e11efd841eeb4ffbb03de691e57 : SUCCESS in 1h 14m 31s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/85b84c9489f5494da3c9826b2e1b49be : SUCCESS in 45m 53s\n- kolla-ansible-ubuntu-podman https://zuul.opendev.org/t/openstack/build/3b0a982f8eeb4a4ebf6f32a585f33a21 : SUCCESS in 54m 01s\n- kolla-ansible-rocky9-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/ec38f2d1c4af46a6b691ec48973dadf3 : SUCCESS in 1h 08m 28s (non-voting)\n- kolla-ansible-ubuntu-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/eb0d9ec2b1c14222a0ca3e921577d44b : SUCCESS in 50m 35s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/b5832faf25b540f7bc776f9000b04cd0 : SUCCESS in 1h 20m 24s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/e72e6558052747b7b42237a9157ff0f1 : POST_FAILURE in 4m 10s (non-voting)\n- kolla-ansible-rocky9-magnum https://zuul.opendev.org/t/openstack/build/2e0e534a1828451782ac31c646301da8 : SUCCESS in 1h 13m 47s (non-voting)\n- kolla-ansible-ubuntu-magnum https://zuul.opendev.org/t/openstack/build/594bfc8ab37b4e8f81b7408172e7c10e : SUCCESS in 57m 41s (non-voting)\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/23d67ec912974845a0cbf4a25169dedc : SUCCESS in 1h 21m 01s\n- kolla-ansible-rocky9-slurp-upgrade https://zuul.opendev.org/t/openstack/build/b3d3cd69efe545c28f3a535b2efcdb1b : SUCCESS in 1h 34m 33s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ae84a1477e95455ca26275bbf2369d90 : SUCCESS in 1h 20m 39s\n- kolla-ansible-debian-slurp-upgrade https://zuul.opendev.org/t/openstack/build/b6b6756db0da4a639e74c79e0d633735 : SUCCESS in 1h 19m 06s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/dd4e658befde45689c350979f6225993 : SUCCESS in 1h 15m 02s\n- kolla-ansible-ubuntu-slurp-upgrade https://zuul.opendev.org/t/openstack/build/08fa7faf7f6e4254b811e357c047cfa6 : SUCCESS in 1h 14m 46s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/f5cd637bd6e448df85081b288794d893 : SUCCESS in 1h 37m 39s (non-voting)\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/875e40311013461fadff04f403752528 : SUCCESS in 1h 29m 47s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/8d5573fbb13d4ac294426c4a9680f61e : FAILURE in 48m 09s (non-voting)\n- kolla-ansible-rocky9-slurp-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d24fc598a2dd420c8721635852fc2603 : FAILURE in 1h 49m 42s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/15a29e846b4d487696fa46e0341d242b : FAILURE in 1h 18m 37s (non-voting)\n- kolla-ansible-ubuntu-slurp-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/4e1b912149c545aab40577446d6f43bc : FAILURE in 1h 41m 15s (non-voting)\n- kolla-ansible-ubuntu-trove https://zuul.opendev.org/t/openstack/build/23fc7aa92f094d159e2d8a80c1e275ff : SUCCESS in 57m 05s (non-voting)","accounts_in_message":[],"_revision_number":44},{"id":"f29a964845f4d79143ec6ccdd5a5910eccbb933c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-05-11 02:55:46.000000000","message":"Uploaded patch set 45.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":45},{"id":"ff571bf6a31d40697449b77ea033ccf49f2c55a2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-11 04:45:23.000000000","message":"Patch Set 45:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9729ffc7eaaf4d03ab8cb4a19f5737f7\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/42cdfc2212384d86b1b476bb7ea71d1b : SUCCESS in 1h 20m 45s\n- kolla-ansible-debian-aarch64-podman https://zuul.opendev.org/t/openstack/build/35ccb13a58c5409c99d5b300cf89f0cc : SUCCESS in 59m 12s\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/50c03e42ca1741bdbafe0a445d0d2881 : SUCCESS in 1h 47m 19s","accounts_in_message":[],"_revision_number":45},{"id":"4c7a6681c4a1e2ae76f7caab2ce7085d06d5ff59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-11 05:30:17.000000000","message":"Patch Set 45: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1fce0df2bef84ce594db4794710a05a5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b62ba96af3c045f5991205923627ae70 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fd7fd9568c5462e812d137864393bb9 : FAILURE in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/149757f4b3674da5b771f0b6fdcf0915 : SUCCESS in 3m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/45e0720a522e41459ae6b0bda283f548 : SUCCESS in 4m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6d68130922974fa489508aeb4c63b3d4 : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/905ea35901164194976a2c16de737e45 : SUCCESS in 4m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bbe5197ea6fe436d869d7652c29db4e5 : FAILURE in 4m 46s\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/5ecf06c2301c4185b85226582cbe259a : SUCCESS in 1h 03m 44s (non-voting)\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/843b42e01fb04a3ebdf45ef9ad39c060 : SUCCESS in 51m 12s\n- kolla-ansible-debian-podman https://zuul.opendev.org/t/openstack/build/d682b9125d81438eaf0b2671985bc6be : SUCCESS in 46m 12s\n- kolla-ansible-openeuler https://zuul.opendev.org/t/openstack/build/2e6e268c75fa4fd198c66e64e78b1bb9 : SUCCESS in 54m 37s (non-voting)\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d8be5f00326e44eea9ac49daaa1468cc : SUCCESS in 58m 57s\n- kolla-ansible-rocky9-podman https://zuul.opendev.org/t/openstack/build/7aa35c674ea245a59a0835bbe85da860 : SUCCESS in 1h 06m 50s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/578e949d92254d28a44f1882095f1128 : SUCCESS in 46m 55s\n- kolla-ansible-ubuntu-podman https://zuul.opendev.org/t/openstack/build/074b0cb0677e4b7a9859b1b74d40f584 : SUCCESS in 52m 30s\n- kolla-ansible-rocky9-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/bceb68ac6c494e3fa6c64eb9ed51e127 : SUCCESS in 1h 01m 35s (non-voting)\n- kolla-ansible-ubuntu-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/719ff0826cdf420f94d68552331fe0c1 : SUCCESS in 43m 34s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/5067a04a1b4a4882b2c7f6914d0a3758 : SUCCESS in 1h 25m 29s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/9fdab232cc524cf492292c6b26dfb110 : SUCCESS in 1h 15m 32s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/2edd95c12cf946cc99196ff99048a1c6 : SUCCESS in 1h 04m 43s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/f0c813076198438ab082074b215c4191 : SUCCESS in 53m 38s (non-voting)\n- kolla-ansible-rocky9-magnum https://zuul.opendev.org/t/openstack/build/4a49708652624359b9a674e821ffac36 : SUCCESS in 1h 05m 16s (non-voting)\n- kolla-ansible-ubuntu-magnum https://zuul.opendev.org/t/openstack/build/1558108af5834d71ab48230a82ed4d5d : SUCCESS in 49m 01s (non-voting)\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f143bfad9e604af6bb8d3fc2ec2061a9 : SUCCESS in 1h 21m 28s\n- kolla-ansible-rocky9-slurp-upgrade https://zuul.opendev.org/t/openstack/build/d9d7b28a2dce4bb980709c86a56443c8 : SUCCESS in 1h 26m 37s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6f35f54bdc0e47a4a7b0e23b093b86c8 : SUCCESS in 1h 19m 10s\n- kolla-ansible-debian-slurp-upgrade https://zuul.opendev.org/t/openstack/build/668f49e30875476cbea0942ba91a4438 : SUCCESS in 1h 24m 40s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f77bbe40123545bcbcb3cbc030d09a8e : SUCCESS in 1h 25m 22s\n- kolla-ansible-ubuntu-slurp-upgrade https://zuul.opendev.org/t/openstack/build/971380f2db144a8388b8f4568eb0bfd7 : SUCCESS in 1h 07m 14s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/3300fd84d8ee4afe86a0d8bd25460b1e : SUCCESS in 1h 25m 10s (non-voting)\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/e0494d41bc914785838df9f25b243097 : SUCCESS in 1h 34m 29s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/9ec01bdd5d654e54b7afa38e6204a08e : SUCCESS in 2h 14m 53s (non-voting)\n- kolla-ansible-rocky9-slurp-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/5d1ddc8305f04a13ac3ce99d36020c8f : SUCCESS in 2h 06m 37s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/8e4fe6da564a4dc1829ca818baefc270 : SUCCESS in 1h 59m 32s (non-voting)\n- kolla-ansible-ubuntu-slurp-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d79dea8fae8f45ed9293be48d2f14ca7 : SUCCESS in 2h 09m 22s (non-voting)\n- kolla-ansible-ubuntu-trove https://zuul.opendev.org/t/openstack/build/d468f255e2f54e1c936741fe3bef85cd : SUCCESS in 54m 04s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"33cd686421ebc3e8eec0d82b2dcc19dc282e5124","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-05-11 10:18:47.000000000","message":"Uploaded patch set 46.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":46},{"id":"422af78c47c0168a563925b97fba717fce5d7b6a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-11 12:07:27.000000000","message":"Patch Set 46:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/223a21bdadbc44dc9547cee5770425bd\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8dfec2d83443492fa8cf99cb76914575 : SUCCESS in 54m 30s\n- kolla-ansible-debian-aarch64-podman https://zuul.opendev.org/t/openstack/build/60f205b3ef3d479db429ab0052b55ef2 : SUCCESS in 52m 10s\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/c96678e06e384550b9ff30b03683cb98 : FAILURE in 1h 46m 27s","accounts_in_message":[],"_revision_number":46},{"id":"8e5a7e0d93bb12b7a7c912f0728c512e07895f4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-11 12:56:04.000000000","message":"Patch Set 46: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e8d4fcb11b94438ab730c1eb1cdbc089\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee54e47e1b8e4b8bbeb7eed1044bbece : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43606b35eb834b5abc273fb90fc716a9 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c37cfefe2f424adba97c62bf4de36926 : SUCCESS in 3m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ece4ae80143742ca988f1b6bca360d69 : SUCCESS in 4m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cec5063a7f20435392bb05ddbb901fd2 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0855fc48ab3946d88080c930acbc9305 : SUCCESS in 4m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fbe6785df24e4d91bda0377a3429abf5 : FAILURE in 3m 46s\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/3ba724bae57c4213a3baf382dddd981e : SUCCESS in 1h 06m 49s (non-voting)\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e1c11acd391b4f20b874c9eed944ef4b : SUCCESS in 52m 40s\n- kolla-ansible-debian-podman https://zuul.opendev.org/t/openstack/build/b15fb9721bdc4a8d9f83526379e2f9f4 : SUCCESS in 1h 00m 41s\n- kolla-ansible-openeuler https://zuul.opendev.org/t/openstack/build/8c8d2b2947194a8b914ae3d867c747a3 : SUCCESS in 44m 56s (non-voting)\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/35e24a407e6e4573aef14857a7aa4c9e : SUCCESS in 51m 57s\n- kolla-ansible-rocky9-podman https://zuul.opendev.org/t/openstack/build/5892fc8cba4b418e9e953359ecdc5729 : SUCCESS in 1h 00m 59s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6bd5cf55aac342d285a2afd043351bc0 : SUCCESS in 36m 57s\n- kolla-ansible-ubuntu-podman https://zuul.opendev.org/t/openstack/build/aa7913e2085d454381c86bdd39d04425 : SUCCESS in 44m 48s\n- kolla-ansible-rocky9-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/1947aef6da024c998bdd368762b47b12 : SUCCESS in 1h 02m 09s (non-voting)\n- kolla-ansible-ubuntu-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/b968d49fc24e493d90763dc617e87e34 : SUCCESS in 45m 33s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/c21713cd0cf04b0f986ac152847bd0bd : SUCCESS in 1h 34m 18s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/2012c6848bfa457db9f4838f7285e940 : SUCCESS in 1h 09m 36s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/2cb88f2014e447efb612b9d9ea3028a5 : SUCCESS in 1h 17m 58s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/7ea15f60f9d540a29384564d8245b93c : POST_FAILURE in 1h 08m 07s (non-voting)\n- kolla-ansible-rocky9-magnum https://zuul.opendev.org/t/openstack/build/52881c19443f445d9476a4fa4ee004e5 : SUCCESS in 55m 53s (non-voting)\n- kolla-ansible-ubuntu-magnum https://zuul.opendev.org/t/openstack/build/627584872e1a4c40abd561427b31ac71 : SUCCESS in 45m 57s (non-voting)\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/593d3181c63e4b9abc6cc781fd813988 : SUCCESS in 1h 07m 54s\n- kolla-ansible-rocky9-slurp-upgrade https://zuul.opendev.org/t/openstack/build/d9bf05ca93d44b4c997e3493e7ede873 : SUCCESS in 1h 11m 38s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d719a24c3807403f918d2e8a5d3fda4b : SUCCESS in 1h 25m 46s\n- kolla-ansible-debian-slurp-upgrade https://zuul.opendev.org/t/openstack/build/30f8573d5653477b9827b07e35321d95 : SUCCESS in 1h 02m 49s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f588a1aed17d4631bcfef2fb1e578bfe : SUCCESS in 1h 03m 13s\n- kolla-ansible-ubuntu-slurp-upgrade https://zuul.opendev.org/t/openstack/build/89e69c03e81d4e1794be30806eb8c510 : SUCCESS in 1h 09m 02s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/83ff7fa7ee18496a8472457f097ddfbb : SUCCESS in 1h 23m 10s (non-voting)\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/7f868870fe1b4a318a7e0b5a2c82fb40 : SUCCESS in 1h 18m 50s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d4bea8df05e14bf1b8ca3e0d554b8f33 : SUCCESS in 2h 15m 28s (non-voting)\n- kolla-ansible-rocky9-slurp-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/a68b6f526cf5433b981f549a18056833 : SUCCESS in 2h 24m 14s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/b1b95c15f8dc47f1a157b55b7cf02733 : SUCCESS in 2h 05m 20s (non-voting)\n- kolla-ansible-ubuntu-slurp-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/aad745c4ce2f4a2ba88ac94ff5b75240 : SUCCESS in 1h 41m 57s (non-voting)\n- kolla-ansible-ubuntu-trove https://zuul.opendev.org/t/openstack/build/18426c08e16749e28047dcb149489e87 : SUCCESS in 52m 24s (non-voting)","accounts_in_message":[],"_revision_number":46},{"id":"e6cd34e2f893232cb5d1900cad76c30319d2b12d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-05-09 08:25:13.000000000","message":"Uploaded patch set 47.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":47},{"id":"06ca0ae8f5eec2e06460ea92ff50a4efc4182ebe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-05-09 08:29:19.000000000","message":"Uploaded patch set 48.","accounts_in_message":[],"_revision_number":48},{"id":"0741a88e5604414ec97fe5ef9ab671b951b0dde4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-09 09:42:14.000000000","message":"Patch Set 48:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54a83cf0b6f345acb4df8a807b08e687\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/aaf8daeedbde4f669d97b727bba045b9 : SUCCESS in 43m 59s (non-voting)\n- kolla-ansible-debian-aarch64-podman https://zuul.opendev.org/t/openstack/build/9ac4149c05df48598c95b22e6bacf7d3 : SUCCESS in 1h 11m 12s (non-voting)\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/850255049b9d4f84bb19426256f46cb4 : SUCCESS in 1h 10m 54s (non-voting)","accounts_in_message":[],"_revision_number":48},{"id":"7aa6794513280bcf7aee237b0213106ebef89f25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-09 10:43:18.000000000","message":"Patch Set 48: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f6b2bf087bc44ec3ad65197662aae817\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3129226c3c3748ffbd0a8a46ff25ae2a : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddcd883d8da3472ca508e267279d39e6 : SUCCESS in 5m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/812fcceb8aa2415dbf2b69442b6875e9 : FAILURE in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f24a034b280245ce8744423185f04ef8 : FAILURE in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cdd78d0a303e4eec9a325292922d0148 : SUCCESS in 2m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e892fc341301412e86bb4212d7661f86 : SUCCESS in 2m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/30d4860cd19e4c3b9c26d9816a8f8fd3 : SUCCESS in 3m 04s\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/b2a7129ec6e84e14aa8250208d7ea472 : SUCCESS in 1h 04m 04s (non-voting)\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/255bbcf38b544a88873336fc1fb9dd15 : SUCCESS in 58m 16s\n- kolla-ansible-debian-podman https://zuul.opendev.org/t/openstack/build/6f915a33fed54bd39465f6c21b64adda : SUCCESS in 1h 00m 56s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/582fca6479f245eba4211764c4337cea : SUCCESS in 55m 17s\n- kolla-ansible-rocky9-podman https://zuul.opendev.org/t/openstack/build/03d09a0f88374882bc0c03d60b1bd480 : SUCCESS in 1h 04m 08s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1fa5e9c66cc6457b93d32d777b02e538 : SUCCESS in 57m 26s\n- kolla-ansible-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a739306f2ccc49bc858151b76ff2b78f : SUCCESS in 1h 11m 50s\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/65171416b7ee4590abfbb1d1cd36aaa1 : SUCCESS in 54m 46s (non-voting)\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/661c8773267c4eb5b34c7ce719672ec5 : SUCCESS in 41m 42s (non-voting)\n- kolla-ansible-rocky9-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/ec055a09bfe646b0988882b8b6098e18 : SUCCESS in 1h 12m 15s (non-voting)\n- kolla-ansible-ubuntu-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/a687731761d14b5a8aa90d7baf3a7c63 : DISK_FULL in 1h 03m 04s (non-voting)\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/2f4cc12f1cf246ca955522043a7fc6ef : SUCCESS in 17m 29s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/f4e593bf5ca24ed6b74526031b828b54 : SUCCESS in 15m 26s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/b4cf883977c64979b0c8004ed13915e0 : SUCCESS in 1h 42m 23s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/2648ea8df7254ed49925a14cdf4f0521 : SUCCESS in 1h 42m 08s (non-voting)\n- kolla-ansible-rocky9-scenario-nfv https://zuul.opendev.org/t/openstack/build/a7f505d79b4640149d191ce092a7a2e8 : FAILURE in 1h 18m 46s (non-voting)\n- kolla-ansible-rocky9-magnum https://zuul.opendev.org/t/openstack/build/41500186e35241e486f8304a3a93ac1a : SUCCESS in 55m 05s (non-voting)\n- kolla-ansible-ubuntu-magnum https://zuul.opendev.org/t/openstack/build/cde11d0cc1fe4041ba4b2e23ccf22d7c : SUCCESS in 57m 26s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/639ed95dc3ab48a88b0b7c500a2e80ed : FAILURE in 38m 45s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/cdb036f08981407783e806ee8a2e4401 : FAILURE in 38m 32s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/0d22f02a019844cbb460fa0f137595d7 : SUCCESS in 46m 23s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/ec46a1a981ed45189dde3012c68b1c1f : SUCCESS in 49m 38s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/1b563f20487c40cf95b3099b2f9046b1 : SUCCESS in 1h 00m 56s (non-voting)\n- kolla-ansible-debian-ironic https://zuul.opendev.org/t/openstack/build/e277607c9d994b4c926b446da4f5ae90 : SUCCESS in 52m 56s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/8f12e5bc51d84990a84b2e7376d4dd6a : SUCCESS in 1h 01m 32s (non-voting)\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/566f5694600942c9b2b57006dc4ba587 : SUCCESS in 1h 15m 23s\n- kolla-ansible-rocky9-slurp-upgrade https://zuul.opendev.org/t/openstack/build/ee6fe424a3b449d48d59e2de4e78868f : SUCCESS in 1h 16m 30s (non-voting)\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/aad38fb752ba4597940948b81588d7ea : SUCCESS in 1h 12m 54s\n- kolla-ansible-debian-slurp-upgrade https://zuul.opendev.org/t/openstack/build/a21d75c012ca41cdbbe90caa2470e1d3 : SUCCESS in 1h 17m 19s (non-voting)\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/228720c61b86428ebd3bb5cfa9340020 : SUCCESS in 1h 04m 26s\n- kolla-ansible-ubuntu-slurp-upgrade https://zuul.opendev.org/t/openstack/build/b9834c3d389c4dada01f217fa27a30e1 : SUCCESS in 1h 24m 05s (non-voting)\n- kolla-ansible-ubuntu-cells https://zuul.opendev.org/t/openstack/build/483b4eca2c9541bcb455a5881179c068 : SUCCESS in 1h 08m 37s (non-voting)\n- kolla-ansible-rocky9-cells https://zuul.opendev.org/t/openstack/build/dba254454788409c8654e4e20c466d61 : SUCCESS in 1h 24m 08s (non-voting)\n- kolla-ansible-debian-mariadb https://zuul.opendev.org/t/openstack/build/ac53f5f4a4054fe087f05f72fc8889de : SUCCESS in 16m 32s\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/0013ff9f6df646c2872cdfcf55ea94ee : SUCCESS in 20m 46s\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/7aef56713a094928ab626c0408184c33 : SUCCESS in 18m 00s\n- kolla-ansible-debian-ovn https://zuul.opendev.org/t/openstack/build/ba1a3f6cca034e2580b18173065b5efe : FAILURE in 52m 58s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/9004453847254b16a2d6236e124af3a0 : FAILURE in 58m 28s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/ed4ea18a572d4b9f8d12b1ef3c7332c2 : FAILURE in 55m 20s (non-voting)\n- kolla-ansible-debian-upgrade-ovn https://zuul.opendev.org/t/openstack/build/1c5133be2cc849498de123fbeef1d0e5 : SUCCESS in 1h 28m 40s (non-voting)\n- kolla-ansible-rocky9-upgrade-ovn https://zuul.opendev.org/t/openstack/build/dae10d2fdd654f3083404d27c96beb26 : SUCCESS in 1h 36m 50s (non-voting)\n- kolla-ansible-ubuntu-upgrade-ovn https://zuul.opendev.org/t/openstack/build/aec43892843f4e51b181aec022e03dad : SUCCESS in 1h 35m 25s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/23284e064660412a9375bc0ac4ef68b2 : SUCCESS in 22m 30s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/eeb5e3acf17e4d37b4cdddb114af7d0a : SUCCESS in 19m 31s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/9294554863f649dbb2338e934a9357c8 : SUCCESS in 30m 50s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/3e3980f8dca0428593ae98ca13025488 : SUCCESS in 34m 20s (non-voting)\n- kolla-ansible-rocky9-venus https://zuul.opendev.org/t/openstack/build/f98261a034ce4818860780b49665be1c : SUCCESS in 21m 00s (non-voting)\n- kolla-ansible-ubuntu-venus https://zuul.opendev.org/t/openstack/build/f90688dc820b495aba85e06255d11185 : SUCCESS in 21m 43s (non-voting)\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/d746215d78134d9ab694fc13973b208f : FAILURE in 1h 12m 32s (non-voting)\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/7cf4bafc7f8a4db384f7cf56ceff6af0 : SUCCESS in 1h 42m 36s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/36139daa6d8e40f8b60816021881ee25 : FAILURE in 1h 53m 51s (non-voting)\n- kolla-ansible-rocky9-slurp-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/93147889fe7c435ebfbc1e725d6fd621 : FAILURE in 54m 49s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/a95970e130f64b60aabefb09554cfa67 : FAILURE in 1h 22m 28s (non-voting)\n- kolla-ansible-ubuntu-slurp-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/4cea30898cb54d88ab5df5720e2bd016 : FAILURE in 6m 05s (non-voting)\n- kolla-ansible-rocky9-hashi-vault https://zuul.opendev.org/t/openstack/build/27d01a5c0a3041d99a4c334fc1d9f38e : SUCCESS in 3m 15s (non-voting)\n- kolla-ansible-rocky9-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/2b6145f691ed4f7898d7991d5b558485 : SUCCESS in 36m 58s (non-voting)\n- kolla-ansible-ubuntu-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/4735c7cfae2e410fbaa58b68a63d0939 : SUCCESS in 39m 17s (non-voting)\n- kolla-ansible-ubuntu-lets-encrypt https://zuul.opendev.org/t/openstack/build/7ae8eabf25e94505969f532360c62692 : SUCCESS in 1h 16m 51s (non-voting)\n- kolla-ansible-rocky9-lets-encrypt https://zuul.opendev.org/t/openstack/build/c9a76ec729b842e58745654dcc3777ee : SUCCESS in 1h 24m 21s (non-voting)\n- kolla-ansible-ubuntu-skyline https://zuul.opendev.org/t/openstack/build/c9e14bcb491644a39f729a930d9609cb : SUCCESS in 38m 25s (non-voting)\n- kolla-ansible-rocky9-skyline https://zuul.opendev.org/t/openstack/build/547f5420147749518c6e71cd95b569d3 : SUCCESS in 41m 56s (non-voting)\n- kolla-ansible-ubuntu-trove https://zuul.opendev.org/t/openstack/build/36be2a1d3f78491eb71a9531a52fe196 : FAILURE in 3m 44s (non-voting)","accounts_in_message":[],"_revision_number":48}],"current_revision_number":48,"current_revision":"718586ef92fe4cfca5714a01ca3dd145280e214f","revisions":{"a8d928809f007090639817c1d9d1ee1f72e79f1f":{"kind":"REWORK","_number":1,"created":"2022-11-03 13:39:50.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/1"}}},"commit":{"parents":[{"commit":"12cc96d4d4292ccb82c078c915c90ab740fdc6af","subject":"Implement automatic deploy of trove","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/12cc96d4d4292ccb82c078c915c90ab740fdc6af"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separated scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/a8d928809f007090639817c1d9d1ee1f72e79f1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/a8d928809f007090639817c1d9d1ee1f72e79f1f"}]},"branch":"refs/heads/master"},"16f600738ff06584e9ada13290e6e73c7bb29078":{"kind":"REWORK","_number":2,"created":"2022-11-03 13:42:43.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/2"}}},"commit":{"parents":[{"commit":"12cc96d4d4292ccb82c078c915c90ab740fdc6af","subject":"Implement automatic deploy of trove","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/12cc96d4d4292ccb82c078c915c90ab740fdc6af"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:42:06.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separated scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/16f600738ff06584e9ada13290e6e73c7bb29078"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/16f600738ff06584e9ada13290e6e73c7bb29078"}]},"branch":"refs/heads/master"},"e8e9f765e425f06637051022c09bc5b870dab112":{"kind":"REWORK","_number":3,"created":"2022-11-03 15:05:13.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/3"}}},"commit":{"parents":[{"commit":"12cc96d4d4292ccb82c078c915c90ab740fdc6af","subject":"Implement automatic deploy of trove","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/12cc96d4d4292ccb82c078c915c90ab740fdc6af"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 15:00:59.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separated scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/e8e9f765e425f06637051022c09bc5b870dab112"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/e8e9f765e425f06637051022c09bc5b870dab112"}]},"branch":"refs/heads/master"},"804d581dca0653bb4e4b4f01e8af16e389edfb6a":{"kind":"REWORK","_number":4,"created":"2022-11-03 15:40:30.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/4"}}},"commit":{"parents":[{"commit":"12cc96d4d4292ccb82c078c915c90ab740fdc6af","subject":"Implement automatic deploy of trove","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/12cc96d4d4292ccb82c078c915c90ab740fdc6af"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 15:40:01.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separated scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/804d581dca0653bb4e4b4f01e8af16e389edfb6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/804d581dca0653bb4e4b4f01e8af16e389edfb6a"}]},"branch":"refs/heads/master"},"1cc31e4e54833fa3eadc706c1b7493f3d4cb003c":{"kind":"REWORK","_number":5,"created":"2022-11-04 01:07:19.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/5"}}},"commit":{"parents":[{"commit":"12cc96d4d4292ccb82c078c915c90ab740fdc6af","subject":"Implement automatic deploy of trove","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/12cc96d4d4292ccb82c078c915c90ab740fdc6af"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-04 01:07:02.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separated scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1cc31e4e54833fa3eadc706c1b7493f3d4cb003c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1cc31e4e54833fa3eadc706c1b7493f3d4cb003c"}]},"branch":"refs/heads/master"},"dd0a73d250f1c0ad02be21f226d475d9758aaaab":{"kind":"REWORK","_number":6,"created":"2022-11-04 02:19:08.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/6"}}},"commit":{"parents":[{"commit":"12cc96d4d4292ccb82c078c915c90ab740fdc6af","subject":"Implement automatic deploy of trove","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/12cc96d4d4292ccb82c078c915c90ab740fdc6af"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-04 02:18:53.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separated scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/dd0a73d250f1c0ad02be21f226d475d9758aaaab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/dd0a73d250f1c0ad02be21f226d475d9758aaaab"}]},"branch":"refs/heads/master"},"76898e1065c95f9ba69f19f8c1d82ba748fbb50c":{"kind":"REWORK","_number":7,"created":"2022-11-04 03:27:50.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/7"}}},"commit":{"parents":[{"commit":"12cc96d4d4292ccb82c078c915c90ab740fdc6af","subject":"Implement automatic deploy of trove","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/12cc96d4d4292ccb82c078c915c90ab740fdc6af"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-04 03:26:29.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separated scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/76898e1065c95f9ba69f19f8c1d82ba748fbb50c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/76898e1065c95f9ba69f19f8c1d82ba748fbb50c"}]},"branch":"refs/heads/master"},"22d9821940e814e209a6fa92867d5735754d0500":{"kind":"REWORK","_number":8,"created":"2022-11-04 05:49:09.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/8"}}},"commit":{"parents":[{"commit":"12cc96d4d4292ccb82c078c915c90ab740fdc6af","subject":"Implement automatic deploy of trove","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/12cc96d4d4292ccb82c078c915c90ab740fdc6af"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-04 05:48:46.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separated scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/22d9821940e814e209a6fa92867d5735754d0500"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/22d9821940e814e209a6fa92867d5735754d0500"}]},"branch":"refs/heads/master"},"c904af53563abeb93775bccc5154b201c81ffca8":{"kind":"REWORK","_number":9,"created":"2022-11-04 06:28:31.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/9"}}},"commit":{"parents":[{"commit":"12cc96d4d4292ccb82c078c915c90ab740fdc6af","subject":"Implement automatic deploy of trove","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/12cc96d4d4292ccb82c078c915c90ab740fdc6af"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-04 06:28:11.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separated scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c904af53563abeb93775bccc5154b201c81ffca8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c904af53563abeb93775bccc5154b201c81ffca8"}]},"branch":"refs/heads/master"},"2b646db7f4f8401047546c8cb57c9594230cb06e":{"kind":"REWORK","_number":10,"created":"2022-11-04 07:25:40.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/10"}}},"commit":{"parents":[{"commit":"12cc96d4d4292ccb82c078c915c90ab740fdc6af","subject":"Implement automatic deploy of trove","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/12cc96d4d4292ccb82c078c915c90ab740fdc6af"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-04 07:25:24.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separated scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/2b646db7f4f8401047546c8cb57c9594230cb06e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/2b646db7f4f8401047546c8cb57c9594230cb06e"}]},"branch":"refs/heads/master"},"dcf5a24a2adee63f7f96b906e0eb38ebcb5e869f":{"kind":"REWORK","_number":11,"created":"2022-11-04 08:46:39.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/11"}}},"commit":{"parents":[{"commit":"12cc96d4d4292ccb82c078c915c90ab740fdc6af","subject":"Implement automatic deploy of trove","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/12cc96d4d4292ccb82c078c915c90ab740fdc6af"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-04 08:46:20.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separated scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/dcf5a24a2adee63f7f96b906e0eb38ebcb5e869f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/dcf5a24a2adee63f7f96b906e0eb38ebcb5e869f"}]},"branch":"refs/heads/master"},"d8af3d48c2d99997a1e69183e0e210f740932512":{"kind":"REWORK","_number":12,"created":"2022-11-04 13:56:19.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/12"}}},"commit":{"parents":[{"commit":"12cc96d4d4292ccb82c078c915c90ab740fdc6af","subject":"Implement automatic deploy of trove","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/12cc96d4d4292ccb82c078c915c90ab740fdc6af"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-04 13:27:51.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separated scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/d8af3d48c2d99997a1e69183e0e210f740932512"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/d8af3d48c2d99997a1e69183e0e210f740932512"}]},"branch":"refs/heads/master"},"66caef1218d8df200fb870a0afb7b229aada7f3e":{"kind":"REWORK","_number":13,"created":"2022-11-07 03:20:16.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/13"}}},"commit":{"parents":[{"commit":"12cc96d4d4292ccb82c078c915c90ab740fdc6af","subject":"Implement automatic deploy of trove","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/12cc96d4d4292ccb82c078c915c90ab740fdc6af"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-07 03:19:54.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separated scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/66caef1218d8df200fb870a0afb7b229aada7f3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/66caef1218d8df200fb870a0afb7b229aada7f3e"}]},"branch":"refs/heads/master"},"0bc51c0e71efd3a1f3ebb494fc87ea3b41621303":{"kind":"REWORK","_number":14,"created":"2022-12-30 02:06:28.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/14"}}},"commit":{"parents":[{"commit":"e722727c5d014038fc7ee2dfcbcd5b0b143ff618","subject":"Implement automatic deploy of trove","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/e722727c5d014038fc7ee2dfcbcd5b0b143ff618"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-12-30 02:06:04.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separated scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/0bc51c0e71efd3a1f3ebb494fc87ea3b41621303"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/0bc51c0e71efd3a1f3ebb494fc87ea3b41621303"}]},"branch":"refs/heads/master"},"9706da244233e523c9804eed7a1f1cc9cc8c362b":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2022-12-30 03:08:21.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/15"}}},"commit":{"parents":[{"commit":"21e59390e3bc5162c590cbd92f62c1dfb94ebe17","subject":"Implement automatic deploy of trove","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/21e59390e3bc5162c590cbd92f62c1dfb94ebe17"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-12-30 03:08:02.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separated scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/9706da244233e523c9804eed7a1f1cc9cc8c362b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/9706da244233e523c9804eed7a1f1cc9cc8c362b"}]},"branch":"refs/heads/master"},"c9504add55a7a50ebc8b06c9eda97c05ca596d74":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2023-01-05 15:17:07.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/16"}}},"commit":{"parents":[{"commit":"e1dc4b2faa2a7ddf30edca338e2d87dcb5d4ad35","subject":"Implement automatic deploy of trove","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/e1dc4b2faa2a7ddf30edca338e2d87dcb5d4ad35"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-01-05 15:05:54.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separated scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c9504add55a7a50ebc8b06c9eda97c05ca596d74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c9504add55a7a50ebc8b06c9eda97c05ca596d74"}]},"branch":"refs/heads/master"},"996c6950eba0349f9c366db9fcc606ee0bf8e760":{"kind":"REWORK","_number":17,"created":"2023-11-17 07:10:43.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/863521/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/17"}}},"commit":{"parents":[{"commit":"08e2b734ba4f437e2000120306ee8ad875fae79c","subject":"Merge \"[doc] Fix the incorrect URL of \u0027Quick Start for development\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/08e2b734ba4f437e2000120306ee8ad875fae79c"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-11-17 07:10:43.000000000","tz":0},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separated scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/996c6950eba0349f9c366db9fcc606ee0bf8e760"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/996c6950eba0349f9c366db9fcc606ee0bf8e760"}]},"branch":"refs/heads/master","description":"Rebase"},"2354e29566102c69142382f7c3b24bf232be70e7":{"kind":"REWORK","_number":18,"created":"2023-11-17 07:12:23.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/863521/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/18"}}},"commit":{"parents":[{"commit":"08e2b734ba4f437e2000120306ee8ad875fae79c","subject":"Merge \"[doc] Fix the incorrect URL of \u0027Quick Start for development\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/08e2b734ba4f437e2000120306ee8ad875fae79c"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-11-17 07:12:23.000000000","tz":0},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separated scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/2354e29566102c69142382f7c3b24bf232be70e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/2354e29566102c69142382f7c3b24bf232be70e7"}]},"branch":"refs/heads/master"},"ccb6b09a604b5c0e19f7bcdb9ed2993a3b7d3ae5":{"kind":"REWORK","_number":19,"created":"2023-11-17 07:12:56.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/863521/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/19"}}},"commit":{"parents":[{"commit":"08e2b734ba4f437e2000120306ee8ad875fae79c","subject":"Merge \"[doc] Fix the incorrect URL of \u0027Quick Start for development\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/08e2b734ba4f437e2000120306ee8ad875fae79c"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-11-17 07:12:56.000000000","tz":0},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separated scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ccb6b09a604b5c0e19f7bcdb9ed2993a3b7d3ae5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ccb6b09a604b5c0e19f7bcdb9ed2993a3b7d3ae5"}]},"branch":"refs/heads/master"},"c66d00b6946ff18e07b11214ddb54180184bd625":{"kind":"REWORK","_number":20,"created":"2023-11-17 07:13:34.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/863521/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/20"}}},"commit":{"parents":[{"commit":"08e2b734ba4f437e2000120306ee8ad875fae79c","subject":"Merge \"[doc] Fix the incorrect URL of \u0027Quick Start for development\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/08e2b734ba4f437e2000120306ee8ad875fae79c"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-11-17 07:13:34.000000000","tz":0},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separated scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c66d00b6946ff18e07b11214ddb54180184bd625"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c66d00b6946ff18e07b11214ddb54180184bd625"}]},"branch":"refs/heads/master"},"b8d19de850e212dad0502aef9886a33f43ad20c6":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2023-11-17 07:26:40.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/863521/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/21"}}},"commit":{"parents":[{"commit":"b5f8aa4ac6ee6012053f147348135405f23974d3","subject":"Implement automatic deploy of trove","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/b5f8aa4ac6ee6012053f147348135405f23974d3"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-11-17 07:26:40.000000000","tz":0},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separated scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/b8d19de850e212dad0502aef9886a33f43ad20c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/b8d19de850e212dad0502aef9886a33f43ad20c6"}]},"branch":"refs/heads/master","description":"Rebase"},"9957029986ea1102e73a885596c11596d9de3355":{"kind":"REWORK","_number":22,"created":"2023-12-04 06:56:15.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/22"}}},"commit":{"parents":[{"commit":"7112d5f117103845d55abd2a260a63312d43a6ec","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/7112d5f117103845d55abd2a260a63312d43a6ec"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-04 06:55:48.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separated scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/9957029986ea1102e73a885596c11596d9de3355"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/9957029986ea1102e73a885596c11596d9de3355"}]},"branch":"refs/heads/master"},"ce39dfa9548573cb93e6a06a3749f14a16c5b131":{"kind":"REWORK","_number":23,"created":"2023-12-04 07:07:11.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/23"}}},"commit":{"parents":[{"commit":"7112d5f117103845d55abd2a260a63312d43a6ec","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/7112d5f117103845d55abd2a260a63312d43a6ec"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-04 07:06:54.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separated scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ce39dfa9548573cb93e6a06a3749f14a16c5b131"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ce39dfa9548573cb93e6a06a3749f14a16c5b131"}]},"branch":"refs/heads/master"},"69322d7e508cc4da10fab369cf3cba5e4229bf9f":{"kind":"REWORK","_number":24,"created":"2023-12-04 08:31:18.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/24"}}},"commit":{"parents":[{"commit":"7112d5f117103845d55abd2a260a63312d43a6ec","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/7112d5f117103845d55abd2a260a63312d43a6ec"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-04 08:30:57.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separated scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/69322d7e508cc4da10fab369cf3cba5e4229bf9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/69322d7e508cc4da10fab369cf3cba5e4229bf9f"}]},"branch":"refs/heads/master"},"8593506f5956dcf918f0142f9869ebff9d9ed8f1":{"kind":"REWORK","_number":25,"created":"2023-12-04 13:37:05.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/25"}}},"commit":{"parents":[{"commit":"7112d5f117103845d55abd2a260a63312d43a6ec","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/7112d5f117103845d55abd2a260a63312d43a6ec"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-04 13:35:49.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separated scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/8593506f5956dcf918f0142f9869ebff9d9ed8f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/8593506f5956dcf918f0142f9869ebff9d9ed8f1"}]},"branch":"refs/heads/master"},"8d43ef90bbe7b5df430fa2376d762f7919ba3cb9":{"kind":"REWORK","_number":26,"created":"2023-12-05 03:25:33.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/26"}}},"commit":{"parents":[{"commit":"7112d5f117103845d55abd2a260a63312d43a6ec","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/7112d5f117103845d55abd2a260a63312d43a6ec"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-05 03:25:17.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separated scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/8d43ef90bbe7b5df430fa2376d762f7919ba3cb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/8d43ef90bbe7b5df430fa2376d762f7919ba3cb9"}]},"branch":"refs/heads/master"},"4f5f31c22e71fa81998185cb0f820a54a3dcba02":{"kind":"REWORK","_number":27,"created":"2023-12-05 06:21:37.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/27"}}},"commit":{"parents":[{"commit":"7112d5f117103845d55abd2a260a63312d43a6ec","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/7112d5f117103845d55abd2a260a63312d43a6ec"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-05 06:20:55.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separated scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/4f5f31c22e71fa81998185cb0f820a54a3dcba02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/4f5f31c22e71fa81998185cb0f820a54a3dcba02"}]},"branch":"refs/heads/master"},"85306e90d382ca86600fd7ffb54969ab1355a476":{"kind":"REWORK","_number":28,"created":"2023-12-06 02:33:44.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/28"}}},"commit":{"parents":[{"commit":"af6e1ca4fd3366c1ae936d225b2835e95a4191fc","subject":"Support Ansible max_fail_percentage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/af6e1ca4fd3366c1ae936d225b2835e95a4191fc"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-06 02:25:09.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separated scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/85306e90d382ca86600fd7ffb54969ab1355a476"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/85306e90d382ca86600fd7ffb54969ab1355a476"}]},"branch":"refs/heads/master"},"ccee2d624615a14091f92af818b9266bc601df69":{"kind":"REWORK","_number":29,"created":"2023-12-06 06:01:36.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/29"}}},"commit":{"parents":[{"commit":"af6e1ca4fd3366c1ae936d225b2835e95a4191fc","subject":"Support Ansible max_fail_percentage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/af6e1ca4fd3366c1ae936d225b2835e95a4191fc"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-06 06:00:17.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separated scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ccee2d624615a14091f92af818b9266bc601df69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ccee2d624615a14091f92af818b9266bc601df69"}]},"branch":"refs/heads/master"},"d5305c5b413d8ec66f3299d4e4e327ffcfde10cf":{"kind":"REWORK","_number":30,"created":"2023-12-06 07:44:07.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/30"}}},"commit":{"parents":[{"commit":"af6e1ca4fd3366c1ae936d225b2835e95a4191fc","subject":"Support Ansible max_fail_percentage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/af6e1ca4fd3366c1ae936d225b2835e95a4191fc"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-06 07:43:55.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separated scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/d5305c5b413d8ec66f3299d4e4e327ffcfde10cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/d5305c5b413d8ec66f3299d4e4e327ffcfde10cf"}]},"branch":"refs/heads/master"},"37211276bd8478cf3b7f710c0358f1e437596845":{"kind":"REWORK","_number":31,"created":"2023-12-06 09:59:12.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/31"}}},"commit":{"parents":[{"commit":"af6e1ca4fd3366c1ae936d225b2835e95a4191fc","subject":"Support Ansible max_fail_percentage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/af6e1ca4fd3366c1ae936d225b2835e95a4191fc"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-06 09:59:01.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separated scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/37211276bd8478cf3b7f710c0358f1e437596845"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/37211276bd8478cf3b7f710c0358f1e437596845"}]},"branch":"refs/heads/master"},"d6afd7a5fb78df65619a73c3470374486d76398f":{"kind":"REWORK","_number":32,"created":"2023-12-07 02:09:37.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/32"}}},"commit":{"parents":[{"commit":"af6e1ca4fd3366c1ae936d225b2835e95a4191fc","subject":"Support Ansible max_fail_percentage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/af6e1ca4fd3366c1ae936d225b2835e95a4191fc"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-07 02:09:28.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separated scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/d6afd7a5fb78df65619a73c3470374486d76398f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/d6afd7a5fb78df65619a73c3470374486d76398f"}]},"branch":"refs/heads/master"},"89434b960ac555ebfb6c5c9943e619569321b84e":{"kind":"REWORK","_number":33,"created":"2023-12-07 05:49:02.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/33"}}},"commit":{"parents":[{"commit":"af6e1ca4fd3366c1ae936d225b2835e95a4191fc","subject":"Support Ansible max_fail_percentage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/af6e1ca4fd3366c1ae936d225b2835e95a4191fc"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-07 05:48:47.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separated scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/89434b960ac555ebfb6c5c9943e619569321b84e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/89434b960ac555ebfb6c5c9943e619569321b84e"}]},"branch":"refs/heads/master"},"e3dc4ea34c028fcb8f96b7edfda3642aa6c3ce2e":{"kind":"REWORK","_number":34,"created":"2023-12-07 10:05:48.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/34"}}},"commit":{"parents":[{"commit":"af6e1ca4fd3366c1ae936d225b2835e95a4191fc","subject":"Support Ansible max_fail_percentage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/af6e1ca4fd3366c1ae936d225b2835e95a4191fc"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-07 10:05:39.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separated scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/e3dc4ea34c028fcb8f96b7edfda3642aa6c3ce2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/e3dc4ea34c028fcb8f96b7edfda3642aa6c3ce2e"}]},"branch":"refs/heads/master"},"9f5d233c6c01a17268321f7c3526d989ecbf4a03":{"kind":"REWORK","_number":35,"created":"2023-12-08 01:10:47.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/35"}}},"commit":{"parents":[{"commit":"af6e1ca4fd3366c1ae936d225b2835e95a4191fc","subject":"Support Ansible max_fail_percentage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/af6e1ca4fd3366c1ae936d225b2835e95a4191fc"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-08 01:10:33.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separated scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/9f5d233c6c01a17268321f7c3526d989ecbf4a03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/9f5d233c6c01a17268321f7c3526d989ecbf4a03"}]},"branch":"refs/heads/master"},"810b0ed87f847186187a3ef9ec49df46eebff906":{"kind":"TRIVIAL_REBASE","_number":36,"created":"2023-12-08 01:15:16.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/36"}}},"commit":{"parents":[{"commit":"7a29abb5907e2f6bbd10b1dd68ba6bdc1ff7f542","subject":"Merge \"CI: Move test-ovn before test-core-openstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/7a29abb5907e2f6bbd10b1dd68ba6bdc1ff7f542"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-12-08 01:15:16.000000000","tz":0},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separated scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/810b0ed87f847186187a3ef9ec49df46eebff906"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/810b0ed87f847186187a3ef9ec49df46eebff906"}]},"branch":"refs/heads/master","description":"Rebase"},"723d9531726eb2f3c6af6cc748a48bdeea61b327":{"kind":"REWORK","_number":37,"created":"2023-12-08 08:13:13.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/37"}}},"commit":{"parents":[{"commit":"af6e1ca4fd3366c1ae936d225b2835e95a4191fc","subject":"Support Ansible max_fail_percentage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/af6e1ca4fd3366c1ae936d225b2835e95a4191fc"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-08 08:13:00.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separated scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/723d9531726eb2f3c6af6cc748a48bdeea61b327"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/723d9531726eb2f3c6af6cc748a48bdeea61b327"}]},"branch":"refs/heads/master"},"7dccdc8f7bfd0dcaad80b13e7ce582331a01ad28":{"kind":"REWORK","_number":38,"created":"2023-12-08 08:21:04.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/38"}}},"commit":{"parents":[{"commit":"af6e1ca4fd3366c1ae936d225b2835e95a4191fc","subject":"Support Ansible max_fail_percentage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/af6e1ca4fd3366c1ae936d225b2835e95a4191fc"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-08 08:20:53.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separated scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/7dccdc8f7bfd0dcaad80b13e7ce582331a01ad28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/7dccdc8f7bfd0dcaad80b13e7ce582331a01ad28"}]},"branch":"refs/heads/master"},"5ddaad05cbf12003e3b1562637327119d0a1a63c":{"kind":"REWORK","_number":39,"created":"2023-12-11 01:44:56.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/39"}}},"commit":{"parents":[{"commit":"af6e1ca4fd3366c1ae936d225b2835e95a4191fc","subject":"Support Ansible max_fail_percentage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/af6e1ca4fd3366c1ae936d225b2835e95a4191fc"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-11 01:44:44.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separated scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/5ddaad05cbf12003e3b1562637327119d0a1a63c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/5ddaad05cbf12003e3b1562637327119d0a1a63c"}]},"branch":"refs/heads/master"},"37c24caec83d9416b5d38f083f602dba5ee386f3":{"kind":"REWORK","_number":40,"created":"2023-12-11 03:16:48.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/40"}}},"commit":{"parents":[{"commit":"af6e1ca4fd3366c1ae936d225b2835e95a4191fc","subject":"Support Ansible max_fail_percentage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/af6e1ca4fd3366c1ae936d225b2835e95a4191fc"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-11 03:16:36.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separated scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/37c24caec83d9416b5d38f083f602dba5ee386f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/37c24caec83d9416b5d38f083f602dba5ee386f3"}]},"branch":"refs/heads/master"},"823ab36f03dde8178e2a22ff8fb950b0d004dafc":{"kind":"REWORK","_number":41,"created":"2023-12-11 06:42:24.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/41"}}},"commit":{"parents":[{"commit":"af6e1ca4fd3366c1ae936d225b2835e95a4191fc","subject":"Support Ansible max_fail_percentage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/af6e1ca4fd3366c1ae936d225b2835e95a4191fc"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-11 06:42:02.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separate scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/823ab36f03dde8178e2a22ff8fb950b0d004dafc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/823ab36f03dde8178e2a22ff8fb950b0d004dafc"}]},"branch":"refs/heads/master"},"495f0c44dd1b04a5a9ff3579c3e0162fe4da8d90":{"kind":"REWORK","_number":42,"created":"2023-12-14 09:01:42.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/42"}}},"commit":{"parents":[{"commit":"af6e1ca4fd3366c1ae936d225b2835e95a4191fc","subject":"Support Ansible max_fail_percentage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/af6e1ca4fd3366c1ae936d225b2835e95a4191fc"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-14 09:01:27.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separate scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/495f0c44dd1b04a5a9ff3579c3e0162fe4da8d90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/495f0c44dd1b04a5a9ff3579c3e0162fe4da8d90"}]},"branch":"refs/heads/master"},"397f30748fe9d4f1d9a761ea790a354644c1ca43":{"kind":"REWORK","_number":43,"created":"2023-12-18 02:14:35.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/43"}}},"commit":{"parents":[{"commit":"d05a1560bc8b6d203ea083f8adeba89d9f200e50","subject":"CI: Drop duplicated variables in base.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/d05a1560bc8b6d203ea083f8adeba89d9f200e50"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-18 02:14:20.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separate scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/397f30748fe9d4f1d9a761ea790a354644c1ca43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/397f30748fe9d4f1d9a761ea790a354644c1ca43"}]},"branch":"refs/heads/master"},"c19317e60cc6ee1064c1e39486932b0a458679e2":{"kind":"REWORK","_number":44,"created":"2024-02-22 13:43:13.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/44"}}},"commit":{"parents":[{"commit":"ce3a6aff09f1880d77bdd70d9f029b0780d93dd9","subject":"Merge \"Fix gnocchi-metricd when TLS and Swift enabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ce3a6aff09f1880d77bdd70d9f029b0780d93dd9"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-02-22 13:43:01.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separate scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c19317e60cc6ee1064c1e39486932b0a458679e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c19317e60cc6ee1064c1e39486932b0a458679e2"}]},"branch":"refs/heads/master"},"4c187d8c273040e58efb82bbc07af0834487cd05":{"kind":"REWORK","_number":45,"created":"2024-05-11 02:55:46.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/45"}}},"commit":{"parents":[{"commit":"b382471d02a9763aa135450e63a705cfad261d58","subject":"Merge \"loadbalancer: fix upgrade task when podman is used\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/b382471d02a9763aa135450e63a705cfad261d58"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-05-11 02:55:17.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separate scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/4c187d8c273040e58efb82bbc07af0834487cd05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/4c187d8c273040e58efb82bbc07af0834487cd05"}]},"branch":"refs/heads/master"},"eee10631242fe78b54044ec4fc7e94e42dbde9f4":{"kind":"REWORK","_number":46,"created":"2024-05-11 10:18:47.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/46"}}},"commit":{"parents":[{"commit":"b382471d02a9763aa135450e63a705cfad261d58","subject":"Merge \"loadbalancer: fix upgrade task when podman is used\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/b382471d02a9763aa135450e63a705cfad261d58"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-05-11 10:18:36.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separate scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/eee10631242fe78b54044ec4fc7e94e42dbde9f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/eee10631242fe78b54044ec4fc7e94e42dbde9f4"}]},"branch":"refs/heads/master"},"20fd99d5a05c7986daccf1ee138ce1ffece32046":{"kind":"REWORK","_number":47,"created":"2025-05-09 08:25:13.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/47"}}},"commit":{"parents":[{"commit":"3e64e274fc2b633e5205fa40f3f767de55a06303","subject":"Implement automatic deploy of trove","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/3e64e274fc2b633e5205fa40f3f767de55a06303"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2025-05-09 08:24:57.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separate scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/20fd99d5a05c7986daccf1ee138ce1ffece32046"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/20fd99d5a05c7986daccf1ee138ce1ffece32046"}]},"branch":"refs/heads/master"},"718586ef92fe4cfca5714a01ca3dd145280e214f":{"kind":"REWORK","_number":48,"created":"2025-05-09 08:29:19.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/21/863521/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/863521/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/863521/48"}}},"commit":{"parents":[{"commit":"3e64e274fc2b633e5205fa40f3f767de55a06303","subject":"Implement automatic deploy of trove","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/3e64e274fc2b633e5205fa40f3f767de55a06303"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-11-03 13:32:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2025-05-09 08:29:03.000000000","tz":480},"subject":"CI: Trove: Create a database instance","message":"CI: Trove: Create a database instance\n\n1. Remove trove test from magnum to a separate scenario.\n2. Improve the trove tests\n\nImplements: blueprint implement-automatic-deploy-of-trove\n\nDepends-On: https://review.opendev.org/#/c/863321\nChange-Id: I37f225a2577c4a13bd87503b4a363ca1171f1513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/718586ef92fe4cfca5714a01ca3dd145280e214f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/718586ef92fe4cfca5714a01ca3dd145280e214f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
