)]}'
{"id":"openstack%2Fkolla-ansible~869821","triplet_id":"openstack%2Fkolla-ansible~master~Ie8a524cc944aa8cb9cf0999b1b8da79f30b40092","project":"openstack/kolla-ansible","branch":"master","topic":"bug/2002522","attention_set":{},"removed_from_attention_set":{"35010":{"account":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"last_update":"2023-01-18 16:26:19.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2023-01-18 15:00:09.000000000","reason":"\u003cGERRIT_ACCOUNT_27339\u003e replied on the change","reason_account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}}},"hashtags":[],"change_id":"Ie8a524cc944aa8cb9cf0999b1b8da79f30b40092","subject":"Stop firewalld config during kolla genconfig","status":"MERGED","created":"2023-01-11 13:02:05.000000000","updated":"2023-01-18 16:27:28.000000000","submitted":"2023-01-18 16:26:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869821-bug/2002522","meta_rev_id":"f913d3a8ff25f7f6cece6d7ffe750912d258e78e","_number":869821,"virtual_id_number":869821,"owner":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-18 16:26:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2023-01-18 14:59:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2023-01-11 13:28:14.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2023-01-18 15:00:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2023-01-18 15:00:09.000000000","_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-11 13:28:14.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-01-11 14:52:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-16 10:57:05.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2023-01-18 14:59:05.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"d436acc3c0f86d79308be368d55c7da24b93670e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2023-01-11 13:02:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"694ce70d71cbd9c7a956e13ed25b0d35220b5dbe","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-01-11 13:28:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3c5e411e2e956c93686c8bd7fbc1c1b2f0aeb75a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 14:52:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1845574a57934b59a33a909367905ac8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a7f1464acb343c7bf823412f781fdd0 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2c727b7cb614a75af7d77accef8b10f : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c84ccfcb1514b7685f4696602f3d400 : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3fd483a0cb484b2d9c62a68188cb6a35 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c582c21cabad491387711b1568d6a321 : SUCCESS in 6m 17s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/59fa9f2a61f44f6abca529a79f1016d9 : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e7c5c33bb5a4485932aa18249ea55ee : SUCCESS in 3m 57s\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/661398fe14a145dab4afcc1608c99e84 : SUCCESS in 1h 16m 16s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/67e652db18c04a609c159d68d93a5f76 : SUCCESS in 45m 36s\n- kolla-ansible-openeuler-source https://zuul.opendev.org/t/openstack/build/0bf58af0507f41afa7021c14187b547e : SUCCESS in 52m 37s (non-voting)\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/41dbe9098f4445cbbfb72711da259c83 : SUCCESS in 54m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d6a730507ff94bcda50914af8794c5dc : SUCCESS in 33m 08s\n- kolla-ansible-rocky9-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/5dfd4ec70fe74630831a2c228d7190bf : SUCCESS in 59m 07s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/fc2a84dfa6ff4dda90ecf432e2bc385f : SUCCESS in 35m 23s (non-voting)\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/ab13ffac1050481fad68100557488da4 : SUCCESS in 1h 05m 17s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a857933e3d044beda68bc656683ba05a : SUCCESS in 58m 01s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/ac1d33111b794906ace21c84f3cc9b91 : SUCCESS in 53m 38s\n- kolla-ansible-rocky9-source-cephadm https://zuul.opendev.org/t/openstack/build/96e501ba537d4f769204814e61b28490 : SUCCESS in 1h 44m 41s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/ac5be7af0b974bfab4de14a7feb72964 : SUCCESS in 1h 08m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/932c398aca0b41869224679a11aa257d : SUCCESS in 1h 31m 51s (non-voting)\n- kolla-ansible-rocky9-hashi-vault https://zuul.opendev.org/t/openstack/build/856b33aa92d1439581ed66aca42a21b1 : SUCCESS in 3m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aa9cb648f68dcfecdca74138f268cee2b0c4a9ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 15:14:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47c548b1564c4c7a9f8ce37f6523aee2\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/596764758c11474993b837c0a745e477 : SUCCESS in 1h 29m 22s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/dc99399bf8814adfb65eb15c941f62f2 : SUCCESS in 2h 09m 35s","accounts_in_message":[],"_revision_number":1},{"id":"1d318dd7c9cbb47651f6a54a18d0f3d8c94d93e7","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-01-18 14:59:05.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0ea901436fc8ea133195dc9aed27adb02fd0f0fb","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-01-18 15:00:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"691166ac3a7d067c9af45d16eef2e54f03f646f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 15:00:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"22396e6a0ed344ce5303396ad692b4e6e95cafe5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 16:26:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/848fe0a5a2f94c259f786920ced6da2b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1281873764bd427aae6424e86269c608 : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1c2b41a42504e078d5094e5be7fc20d : SUCCESS in 6m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/639be3e000a54e239dd99c08fa4dacc5 : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d20c9cb131ed4b7587268f8828fcfa98 : SUCCESS in 5m 54s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/13c8f6c9bed34e1c96dbe8d52a12ab44 : SUCCESS in 9m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/789d71f2e57a49da86faa2807d284b15 : SUCCESS in 3m 59s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e2b0bb3fb4a14d8da66fc79492ae414b : SUCCESS in 58m 00s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/45774943065440dc829b7e12ac8b9eb3 : SUCCESS in 1h 10m 00s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/d958f95ab0e0488f92c3b8eb25bb6bef : SUCCESS in 1h 09m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7d5ddefc3d7b43e0b3bfc63bde78370e : SUCCESS in 46m 24s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/445420b120574f06b4423e97979c2dfb : SUCCESS in 1h 05m 52s","accounts_in_message":[],"_revision_number":1},{"id":"f689d4c1af50d7d422a7d2da51966fb14fdcf1ba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 16:26:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f913d3a8ff25f7f6cece6d7ffe750912d258e78e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 16:27:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d312c444bdc4e02b98fdd3a16ca0133\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85129639ba7f4d31953ecfaa5a11cc0f : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8252ed7adb64682834dd63058d187da : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"86870bd7f84f2eeed14a7c0c24688c93b8d2ac1c","revisions":{"86870bd7f84f2eeed14a7c0c24688c93b8d2ac1c":{"kind":"REWORK","_number":1,"created":"2023-01-11 13:02:05.000000000","uploader":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"ref":"refs/changes/21/869821/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/869821/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/869821/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/869821/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/869821/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/869821/1"}}},"commit":{"parents":[{"commit":"2b88144c056bca649adc65b57def1cd77849037c","subject":"Merge \"Explicitly set the value of heartbeat_in_pthread\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/2b88144c056bca649adc65b57def1cd77849037c"}]}],"author":{"name":"Jack Hodgkiss","email":"identity@jackhodgkiss.uk","date":"2023-01-11 11:25:02.000000000","tz":0},"committer":{"name":"Jack Hodgkiss","email":"identity@jackhodgkiss.uk","date":"2023-01-11 12:03:49.000000000","tz":0},"subject":"Stop firewalld config during kolla genconfig","message":"Stop firewalld config during kolla genconfig\n\nPrevent the haproxy-config role from attempting to modify firewalld when\nrunning kolla-ansible genconfig.\n\nCloses-Bug: #2002522\nChange-Id: Ie8a524cc944aa8cb9cf0999b1b8da79f30b40092\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/86870bd7f84f2eeed14a7c0c24688c93b8d2ac1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/86870bd7f84f2eeed14a7c0c24688c93b8d2ac1c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
