)]}'
{"id":"openstack%2Fkolla-ansible~870521","triplet_id":"openstack%2Fkolla-ansible~master~I2546d714d1e6a1b648a4c6242ece18a2108b9183","project":"openstack/kolla-ansible","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-08-20 06:56:02.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-08-20 05:29:09.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2024-08-09 09:38:06.000000000","reason":"\u003cGERRIT_ACCOUNT_24072\u003e replied on the change","reason_account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-08-09 07:47:46.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2024-08-20 06:56:02.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2024-08-20 06:56:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2546d714d1e6a1b648a4c6242ece18a2108b9183","subject":"CI: improve get_logs.sh","status":"MERGED","created":"2023-01-16 10:19:24.000000000","updated":"2024-08-20 06:57:10.000000000","submitted":"2024-08-20 06:56:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"870521","meta_rev_id":"e110a3bfca7ade7f7b6045fb9d2bf78df94b3313","_number":870521,"virtual_id_number":870521,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-20 06:56:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-08-19 20:06:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2024-08-20 05:29:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-08-09 07:47:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-08-20 05:29:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-08-19 20:06:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-08-19 20:06:57.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-16 13:00:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-17 09:54:40.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-01-17 10:02:21.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2023-01-17 10:02:21.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2023-01-17 10:02:21.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-08-09 07:37:40.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-08-09 09:38:06.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REMOVED"}],"messages":[{"id":"41cebde0b6f1b4af918abf3044ef4d5d2a455781","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-16 10:19:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8470027854c5b7e477c3570f9c1b5dc99b3212f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 13:00:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b6c9ebd58c04074a06ec3a5db9802a0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/364c2fac4d214d628174af041fd41cad : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b114767e6cb1418788fd01beaa2c0b4a : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8308296e6c2541fe8ced979c2cd4b503 : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/031e61c226894e47a2b4377d978132dd : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f51630f2b57e4e35864dfb1d10df30c7 : SUCCESS in 5m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5d31825bbde3467bb848538de087f0e0 : SUCCESS in 7m 35s\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/88bc969042c4468fa0cfeae65ff3bfcb : SUCCESS in 1h 06m 51s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/689fe58f665749c7a9c62408b09a3738 : SUCCESS in 48m 51s\n- kolla-ansible-openeuler-source https://zuul.opendev.org/t/openstack/build/efb2b08a6a164e22818babf0339123b8 : SUCCESS in 46m 52s (non-voting)\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/4ee4091313024983b4d7c27834aaa40b : SUCCESS in 1h 00m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f0f65ead6c644a2db25730e5f60c60e3 : SUCCESS in 39m 48s\n- kolla-ansible-rocky9-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/e5432af24bec4bbe98ad8d9239dfb3e8 : SUCCESS in 1h 01m 39s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/e225343a5dca48cfac59e2d6c2a2c083 : SUCCESS in 50m 20s (non-voting)\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/f2bdc19164fa497eab9f96bd88f520a9 : SUCCESS in 57m 19s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/19ac0905fcf649368b58b97fef17be12 : SUCCESS in 58m 21s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/d8595b015db94d96bcb109e8b310e779 : SUCCESS in 1h 04m 31s\n- kolla-ansible-rocky9-source-cephadm https://zuul.opendev.org/t/openstack/build/63501e85692e4cf6a19a5f1f62d25f49 : SUCCESS in 1h 25m 35s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/f1c98663ea0148a58a595b539ea13689 : SUCCESS in 1h 07m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/ce10cc6204a24a1691beb6a330c97f4d : SUCCESS in 1h 37m 23s (non-voting)\n- kolla-ansible-rocky9-hashi-vault https://zuul.opendev.org/t/openstack/build/11dfe7959a37407da883c72752a8c038 : SUCCESS in 3m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"234acefbe42cd1593ef45adf92cf84f3409c0e04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-16 13:23:37.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"735914bc6df4ad798e23ec131e3a56a234d4f393","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 15:42:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfea12f45f634c04ab3688d41e010f87\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09b5e58841eb45e0bdf4b294abcb0f4d : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45e96e6d5a204b789996492ef3a54a55 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0688bb1944e54f0eaaf3502432962b4e : SUCCESS in 4m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/246fb1d49f644a808bdea39b7781530e : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eee31faf28ea4d7f9f418e7522455273 : SUCCESS in 4m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fd0e5bc481824c7fb5c8c432275f4934 : SUCCESS in 7m 01s\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/b6d01aa60e4249b0b2396b6ab127d38f : FAILURE in 48m 15s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f2b4e96927464d06ada9895eb614fb15 : SUCCESS in 51m 49s\n- kolla-ansible-openeuler-source https://zuul.opendev.org/t/openstack/build/ec63d92af05042b19c3898be0f86c1fb : SUCCESS in 52m 13s (non-voting)\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/d309188d734b4a8f9ad19fbb64ddad01 : SUCCESS in 48m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b2390132697b49ba8740d0a4f4efbed1 : SUCCESS in 41m 05s\n- kolla-ansible-rocky9-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/28e71200402043ebbfbb598e75548455 : SUCCESS in 59m 07s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/01b870b94051459b800145e921af26c6 : SUCCESS in 46m 23s (non-voting)\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/85e77ce6e7454c18858f8b369ba89f96 : SUCCESS in 59m 46s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9a078fa78d074a39945740535dee0a45 : SUCCESS in 56m 43s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/a3a00ffd992249fb805896b903b13ecf : SUCCESS in 1h 02m 04s\n- kolla-ansible-rocky9-source-cephadm https://zuul.opendev.org/t/openstack/build/067a838f23ff407bbd78c96c2ee64a3b : FAILURE in 1h 21m 51s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/28a01ed2657a453fb7bae13c5f0c75e5 : SUCCESS in 59m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/85689ba7bd9742f2a3dd3438190a9766 : SUCCESS in 1h 38m 00s (non-voting)\n- kolla-ansible-rocky9-hashi-vault https://zuul.opendev.org/t/openstack/build/2f35ac2057724692baee12fae498dbe9 : SUCCESS in 3m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9c6b3bcc9c5d3112464d7cba4880303f4c46d785","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 18:05:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1aabb0ea8e624a3c8d016f69fa43fe9d\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ca7bb8e8f5d74149aa721016fc482143 : SUCCESS in 1h 48m 26s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/758ef903230b4e1785e0b81165580ba0 : SUCCESS in 1h 50m 54s","accounts_in_message":[],"_revision_number":2},{"id":"f3319534721b75281db6c2c6606ec1e6909b56bd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-01-17 09:54:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e2df483a9636efb4d0d24a412bbd633e68ec392c","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-01-17 13:06:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"560295816a792fa549f520118819ebfc2d7beae4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 13:07:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d11d588499768e3a4f49358d61b9d66bde0fbb5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 14:17:58.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/68bb82d157dd408aaffbcae8c56be624\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31e2a20304d244fe8d2faaca0bae87ab : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3820ab2844db4d55bad75cf5afb2a7d3 : SUCCESS in 5m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d5adea1ff7b4d928ac97d7a2d309763 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/865038b467bf48178b32f18de8c36138 : SUCCESS in 5m 26s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c1aa0de9f55a46c88dd0a6a05969c485 : SUCCESS in 7m 33s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/36c237268ca14f27b037ad2b66215c4c : FAILURE in 30m 49s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/721858fea5384b4fba9b12b8ae787b30 : SUCCESS in 1h 02m 40s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/706f3c5fd7324922b442369142977bee : FAILURE in 29m 58s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c712b275c5944c86a3040d51feda159a : FAILURE in 20m 06s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e0763f204f7b4a44914ff41327173ef8 : SUCCESS in 1h 05m 07s","accounts_in_message":[],"_revision_number":2},{"id":"7f5f53566acaf7025b4e0680949592d8529085ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-10 06:23:31.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"9178a0641c6122b7550e12fc96c4976d46bf121c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 08:10:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d44891615d46481f8c30fc3f24bb5bad\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f1e2acf887964ae8a84d01895a464f82 : SUCCESS in 1h 15m 08s\n- kolla-ansible-debian-aarch64-podman https://zuul.opendev.org/t/openstack/build/5133000b13c2471db6b5d99f95e2c7ea : SUCCESS in 1h 07m 16s\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/949be846540142b2ab30cedc954b335d : SUCCESS in 1h 44m 40s","accounts_in_message":[],"_revision_number":3},{"id":"784c21fb87ae6135c0229e98852905de0ce058d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 08:38:10.000000000","message":"Patch Set 3: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e41ffb17614840d6bafd28d44f587fdc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/11542afb78ec42be81b37685583e4aa4 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14aa7a8e34834964bb63ba446487f8fd : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98dcf21d84b547578da5c8fa613b7438 : SUCCESS in 3m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/019b24de523f4a47925e28e01c79cac4 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7631cdb991664c5e94a92a3b2ee63de3 : SUCCESS in 5m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/51aa1a35733348ccbebfb4d6906c2b23 : FAILURE in 4m 19s\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8308169276674dab898306c27dd3b26a : SUCCESS in 1h 09m 42s (non-voting)\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6838240bec8140908e00f240bbfd1776 : SUCCESS in 55m 40s\n- kolla-ansible-debian-podman https://zuul.opendev.org/t/openstack/build/ca280bc352634f31b14dd1f17fa54125 : SUCCESS in 1h 02m 55s\n- kolla-ansible-openeuler https://zuul.opendev.org/t/openstack/build/750c86e854c0424283847fb37030777c : SUCCESS in 1h 03m 54s (non-voting)\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/be55b08c0da5421e9ff6b5100f5e9fab : SUCCESS in 57m 30s\n- kolla-ansible-rocky9-podman https://zuul.opendev.org/t/openstack/build/15f5e73f5dc548bf94d76b22c05735f1 : SUCCESS in 1h 07m 05s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1c797ef42611457faeaa0787123e351a : SUCCESS in 52m 27s\n- kolla-ansible-ubuntu-podman https://zuul.opendev.org/t/openstack/build/158f0d0d172f4529a3cc33b83e9c78a6 : SUCCESS in 59m 11s\n- kolla-ansible-rocky9-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/5d8d7446cca54af28b327d1b22a1e826 : SUCCESS in 1h 05m 36s (non-voting)\n- kolla-ansible-ubuntu-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/064faf4b12c14ac3b900600b087d3953 : SUCCESS in 1h 06m 59s (non-voting)\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/764e9278f9794f9f90b07b9ad75198b9 : SUCCESS in 1h 24m 38s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b714e956212f46558d616284da30aa38 : SUCCESS in 1h 20m 24s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/eebaa1bb66c3460a95e3ff5eeb5f8968 : SUCCESS in 1h 15m 51s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/119b3f25175746d881c4d206c9ddc4c5 : SUCCESS in 1h 35m 08s (non-voting)\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/aa3059b3b20b47f6af0867a3dd60ddb7 : FAILURE in 1h 10m 29s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/06ec7b2ad20c4ae08152d49e9e8ed91c : SUCCESS in 2h 05m 33s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/4438e05cbc0c47298cc0c7eb336b8884 : FAILURE in 50m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"dcda8ebd3984d0ec5e96ed30ed562cb25911c2ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-09 05:56:34.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"f067d58613e14bf22a6486e24da8a876f45d7c8f","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-08-09 07:47:46.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bf21f97302aed201651c30fc62270625c05ca0ba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 08:06:43.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d52a1f4473d44f5fad522ed381da30a1\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9af42f0e8de14d1f95c22ebc7063cccf : SUCCESS in 1h 32m 08s\n- kolla-ansible-debian-aarch64-podman https://zuul.opendev.org/t/openstack/build/e61e65ddcdb342b18bd2eed1928ca37e : SUCCESS in 2h 08m 15s\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/9ddc2b7ddd9d4853a3f119ad387a080f : FAILURE in 1h 20m 00s","accounts_in_message":[],"_revision_number":4},{"id":"f1e3f4d819814c167083c19932684f575323e37d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 08:36:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dff7a977e75a40bf9d4d33ac3f9eb8cd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ed821314f444a348e2f00df3fbd5853 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ce55436f01a46babac1b08c596afb63 : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1799578f69ff438d96e04dcb4c48a092 : SUCCESS in 3m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4d3c584e03f14347bbcd18140bedafbc : SUCCESS in 3m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f8d7acbdf210439ca523618685b8cb34 : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24aa99373d624167832638d06b285e43 : SUCCESS in 4m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0e930be400af48509dadb44af45582a0 : SUCCESS in 7m 09s\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8092be11a02d40658c6bbe891478ab7c : SUCCESS in 1h 00m 13s (non-voting)\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ba3e52412829434cb2303f18b01d3b4d : SUCCESS in 51m 26s\n- kolla-ansible-debian-podman https://zuul.opendev.org/t/openstack/build/acbac8ae19594e9485257d6dc4059c45 : SUCCESS in 59m 40s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c4ac2df00da843c2b230a625224f50d9 : SUCCESS in 1h 03m 38s\n- kolla-ansible-rocky9-podman https://zuul.opendev.org/t/openstack/build/2cfa811fc8784be98769dfc0810346bd : SUCCESS in 59m 44s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/81c876e3b12b40d7a54a7b38efc2c98a : SUCCESS in 42m 44s\n- kolla-ansible-ubuntu-podman https://zuul.opendev.org/t/openstack/build/36138da1ba9845c1bd66205b912e589f : SUCCESS in 52m 04s\n- kolla-ansible-rocky9-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/63e64aca348446bbbf954741a3e23164 : SUCCESS in 1h 05m 22s (non-voting)\n- kolla-ansible-ubuntu-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/0414d41f3c174ec7926b634eec9df3f7 : SUCCESS in 54m 18s (non-voting)\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c051edbcb46d49a0b76267af7716d5d5 : SUCCESS in 1h 20m 44s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7e7f2d9fee9445f6a60f793925b74587 : SUCCESS in 1h 14m 45s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/47dd019a15014391a018c3a45fd22dd3 : SUCCESS in 1h 19m 35s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/c915acc2e3a14fae8f2b70d436930f84 : SUCCESS in 1h 39m 18s (non-voting)\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/4dd02e32822d4029b79fe9e934fcd00a : SUCCESS in 1h 25m 02s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f3ad7c723e054839b4e34c6ffa93396a : SUCCESS in 2h 09m 30s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d40b6b73d45548cf9b8c14ebe1a7b8e3 : SUCCESS in 2h 18m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"01dada3d58e8eabb9e228f09dbdb2091bf9d09e4","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2024-08-09 09:38:06.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_24072\u003e.","accounts_in_message":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}],"_revision_number":4},{"id":"0d9a58fefa336f8993d0db376c4c76b8eb329512","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-19 20:06:57.000000000","message":"Patch Set 4: Backport-Candidate+1 Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"18bb92e1b5b4be9f342f1842a09ed9d51b9ac47e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-20 05:29:09.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3c3ebf25fdf228a53dd67966a42e63d86a35647e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 05:29:31.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d9a87fa7fae5d3ea7e4899c6d032ce9a65995494","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 06:56:02.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/078dbcae059041e3b90dfe066a547ed9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bb7f6b15e374fd9ae10e0112221c948 : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ffbd40c16c0848d88120cfc4a459a1f0 : SUCCESS in 3m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/59556f84a0304098a7bdc0b61d803f46 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7faf5ac1a9d443ab80041b766cfa4f44 : SUCCESS in 4m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d0370a37497b4ce88854e5b441d2eab8 : SUCCESS in 7m 03s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ba6a5a04a5cd4438b8deb62b7b7f4904 : SUCCESS in 41m 40s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b0e6c1e9d7e643028a32a47af43d3820 : SUCCESS in 1h 05m 23s\n- kolla-ansible-debian-podman https://zuul.opendev.org/t/openstack/build/0d9c8fa0fac647a48d919ab26b9dd74b : SUCCESS in 59m 27s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9164b71fba2645578759e83a01d9697b : SUCCESS in 53m 38s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/8fdd2270997a44ada160308cee956e4b : SUCCESS in 1h 17m 57s\n- kolla-ansible-rocky9-podman https://zuul.opendev.org/t/openstack/build/c303dd8c143445b7889363023eca0f2e : SUCCESS in 1h 05m 50s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/38c3aa6741834a089c8ed3c63100c60b : SUCCESS in 48m 49s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e93c820349b44257bc3dc797396bac8b : SUCCESS in 1h 07m 17s\n- kolla-ansible-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c9838e850c6d43d7ab5baaddcef55833 : SUCCESS in 57m 42s","accounts_in_message":[],"_revision_number":4},{"id":"316e98c9154e7a89b3e8c0c440c236776b2a3846","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 06:56:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"e110a3bfca7ade7f7b6045fb9d2bf78df94b3313","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 06:57:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8a517c0cdcf49d6bf10482c8d2797d5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19d8a766848d4ee7ade8fc8d2936be4f : SUCCESS in 45s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"68fd0cd2884e8c1f2c00037030f2b796fce018ad","revisions":{"973831a1d2a8e18faf196fcb4dc698accc2096b8":{"kind":"REWORK","_number":1,"created":"2023-01-16 10:19:24.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/870521/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/870521/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/870521/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/870521/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/870521/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/870521/1"}}},"commit":{"parents":[{"commit":"b2a7d2d7507ceeae5ef70a8a947e836803c9750f","subject":"Merge \"Drop remnants of install_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/b2a7d2d7507ceeae5ef70a8a947e836803c9750f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-16 10:19:10.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-16 10:19:41.000000000","tz":0},"subject":"CI: improve get_logs.sh","message":"CI: improve get_logs.sh\n\nChange-Id: I2546d714d1e6a1b648a4c6242ece18a2108b9183\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/973831a1d2a8e18faf196fcb4dc698accc2096b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/973831a1d2a8e18faf196fcb4dc698accc2096b8"}]},"branch":"refs/heads/master"},"34a8a85cd5bfc89eb25e40f2bab3258299d02c16":{"kind":"REWORK","_number":2,"created":"2023-01-16 13:23:37.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/870521/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/870521/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/870521/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/870521/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/870521/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/870521/2"}}},"commit":{"parents":[{"commit":"b2a7d2d7507ceeae5ef70a8a947e836803c9750f","subject":"Merge \"Drop remnants of install_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/b2a7d2d7507ceeae5ef70a8a947e836803c9750f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-16 10:19:10.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-16 13:23:37.000000000","tz":0},"subject":"CI: improve get_logs.sh","message":"CI: improve get_logs.sh\n\nChange-Id: I2546d714d1e6a1b648a4c6242ece18a2108b9183\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/34a8a85cd5bfc89eb25e40f2bab3258299d02c16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/34a8a85cd5bfc89eb25e40f2bab3258299d02c16"}]},"branch":"refs/heads/master"},"b26c33120d5cbc24c2dfa81bcbb9464b5df8dc6f":{"kind":"REWORK","_number":3,"created":"2024-01-10 06:23:31.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/21/870521/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/870521/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/870521/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/870521/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/870521/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/870521/3"}}},"commit":{"parents":[{"commit":"e30ef79d630a701a97157137190fa1190fa41d17","subject":"Merge \"CI: Test Nova server resize functionality\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/e30ef79d630a701a97157137190fa1190fa41d17"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-16 10:19:10.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-01-10 06:22:45.000000000","tz":60},"subject":"CI: improve get_logs.sh","message":"CI: improve get_logs.sh\n\nChange-Id: I2546d714d1e6a1b648a4c6242ece18a2108b9183\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/b26c33120d5cbc24c2dfa81bcbb9464b5df8dc6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/b26c33120d5cbc24c2dfa81bcbb9464b5df8dc6f"}]},"branch":"refs/heads/master"},"68fd0cd2884e8c1f2c00037030f2b796fce018ad":{"kind":"REWORK","_number":4,"created":"2024-08-09 05:56:34.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/21/870521/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/870521/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/870521/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/870521/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/870521/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/870521/4"}}},"commit":{"parents":[{"commit":"e30ef79d630a701a97157137190fa1190fa41d17","subject":"Merge \"CI: Test Nova server resize functionality\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/e30ef79d630a701a97157137190fa1190fa41d17"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-16 10:19:10.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-08-09 05:55:06.000000000","tz":120},"subject":"CI: improve get_logs.sh","message":"CI: improve get_logs.sh\n\nChange-Id: I2546d714d1e6a1b648a4c6242ece18a2108b9183\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/68fd0cd2884e8c1f2c00037030f2b796fce018ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/68fd0cd2884e8c1f2c00037030f2b796fce018ad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
