)]}'
{"id":"openstack%2Fkolla-ansible~872721","triplet_id":"openstack%2Fkolla-ansible~master~I39737e997fb308a781265c270668583222d42a50","project":"openstack/kolla-ansible","branch":"master","topic":"bug/1920095","attention_set":{},"removed_from_attention_set":{"34212":{"account":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"last_update":"2025-04-17 09:02:59.000000000","reason":"Change was abandoned"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-03-23 09:43:47.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2025-04-17 09:02:59.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I39737e997fb308a781265c270668583222d42a50","subject":"Don\u0027t template ceilometer event_definitions.yaml","status":"ABANDONED","created":"2023-02-05 14:38:02.000000000","updated":"2025-04-17 09:02:59.000000000","total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"eead9852bafc4697cf712cea79c0e0e8b27442d2","_number":872721,"virtual_id_number":872721,"owner":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-02-06 12:11:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-02-06 10:40:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":-1,"date":"2023-08-26 20:56:41.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-02-27 02:27:14.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2023-02-22 14:19:15.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-05 16:35:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-05 16:45:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-06 10:40:19.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-02-22 14:19:15.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-02-27 02:27:14.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-03-23 09:43:47.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2023-08-26 20:56:41.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"05d77cbe178ce02fde9dd53bb0bd2f8c6bcd519b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"date":"2023-02-05 14:38:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46a511f9c04d8cca0c80117b11cea1702fdf7b1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"date":"2023-02-05 14:42:33.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b0169fac8d582891e27b17faa12e8bfdc17d9a5e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-05 16:35:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/094131dc75534692982ae1237d1803c5\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7ab609d56aa241eca8bca4baa0d2bd7d : SUCCESS in 1h 25m 17s (non-voting)\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/60250dfa128c40a788f61f7a30a193fe : SUCCESS in 1h 51m 21s","accounts_in_message":[],"_revision_number":2},{"id":"5ce915bb90679893e42d9459b9ec1ad99200cf68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-05 16:45:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1e2d8d879b364c458a2d98bbf4e9f643\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/370e61041c664483b1c0ea4189852ca4 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f22d59711fbc4233be776d0d93a03c6a : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7bac57d44ce2418b906a8dab6c56fd1e : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/96edd6d8bdcf46bbba873193198dcca3 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/181760948cde450898b2bc28fbdb3370 : SUCCESS in 5m 04s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fd8b42e59e9e4456a3813fe69d6aaf26 : FAILURE in 6m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28143ec9b149494d827eb0d9a9695d89 : SUCCESS in 3m 30s\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/bcb2ce97e8824b0cb95350d79530936a : SUCCESS in 1h 02m 21s (non-voting)\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c49b3a3bc83840d0a9052a1a51d6ec0f : SUCCESS in 1h 23m 51s\n- kolla-ansible-openeuler https://zuul.opendev.org/t/openstack/build/eecfb7cf51f147e1b6db578a954aa7db : SUCCESS in 1h 14m 50s (non-voting)\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/7486cfca59cb418b9473f13f8e45fb3f : SUCCESS in 50m 52s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/eb938cd17a194fc4b12f6086ac8b8201 : SUCCESS in 34m 40s\n- kolla-ansible-rocky9-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/4df4750aae7943edab2d19c82293889c : SUCCESS in 54m 20s (non-voting)\n- kolla-ansible-ubuntu-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/d63d9217fd6b47899c283466fdc42def : SUCCESS in 41m 19s (non-voting)\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/343003c5a30943849ddf108cca77a925 : SUCCESS in 1h 55m 44s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7f7e61673ebf40bd8116980b4bacd327 : SUCCESS in 1h 01m 36s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/28c8d9c3927b47fabc21cfa137977fd8 : SUCCESS in 1h 17m 58s (non-voting)\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/2d183f6bc90644ff8cc7d5eac5f9870f : SUCCESS in 52m 49s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/c0c5ae8392ea4292bcf8cdc4285e34f6 : SUCCESS in 1h 36m 22s (non-voting)\n- kolla-ansible-rocky9-hashi-vault https://zuul.opendev.org/t/openstack/build/af77c9c92a774c1986b2738dbf8b2d96 : SUCCESS in 4m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3af88160a701e15893ca7a24d260ba039ef753c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"date":"2023-02-06 10:38:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d5fd72f1090ea35d73d61747e8c97ccab56eed67","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-06 10:40:19.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"08958e70e6e67b2dd20741217f3375f735c16239","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 12:11:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc9e7cda541848728e5eb0ca3894e0d6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2577ec5969d8489bb5caa1b3e3ad832b : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5db9369aa9ca430dbebc3f4e25852cab : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1570b0cafa240e789ff6f3e138ced1c : SUCCESS in 5m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/38838ad894924b06a97d3f4be09874ad : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64f9b21b88a94de096465c9f67823934 : SUCCESS in 4m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/533c6839ddff4fc19d4e1d9b30f45afb : SUCCESS in 7m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa926ed411764b42a4663bae1cef35fe : SUCCESS in 2m 56s\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/2d3258283b084b549841ed6b9fcbc68a : SUCCESS in 1h 20m 24s (non-voting)\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d9f2186fd0b748b48666b3c61dfc6600 : SUCCESS in 50m 27s\n- kolla-ansible-openeuler https://zuul.opendev.org/t/openstack/build/ca801c01a4fd4bfcbe4cf047b41e90e4 : SUCCESS in 50m 42s (non-voting)\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/701303ab4fef4b1cb773b4e006f204a3 : SUCCESS in 1h 05m 33s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/75b07337e2f84af189f58967331553ad : SUCCESS in 41m 54s\n- kolla-ansible-rocky9-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/a9c25daff6c842b091a0c6b477d1a622 : SUCCESS in 1h 00m 22s (non-voting)\n- kolla-ansible-ubuntu-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/c31d77c036c6477bb7f6270b62803cda : SUCCESS in 1h 08m 02s (non-voting)\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b1ad03058e294537bb5204bbd98ceaa0 : SUCCESS in 1h 00m 20s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b41951aab44c43b6964f422e30cff8f3 : SUCCESS in 54m 11s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/64307441aee840f596466995dd9bf6a9 : FAILURE in 4m 50s (non-voting)\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/2ce44b037b9347f18764b10aaa348059 : FAILURE in 4m 31s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/ed7cdb9b77e2479a8f8a245cc887b5c8 : FAILURE in 5m 50s (non-voting)\n- kolla-ansible-rocky9-hashi-vault https://zuul.opendev.org/t/openstack/build/90269eb5b8a443cdb4a76b54f2ffa013 : SUCCESS in 3m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"36d0f1a9c0dcb2af44463d03daabdadf120e38ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 12:42:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7daad76e7b204214b7d152ab09d2e202\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/de4b1cf065ac4fefb7f38559dbd0ef9b : SUCCESS in 55m 57s (non-voting)\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/8ea3f9c4c0d24688aa2a44f6c442a02c : SUCCESS in 2h 02m 06s","accounts_in_message":[],"_revision_number":3},{"id":"ef3efc1a8edc77f1cccb307d00345f4ac431400e","author":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"date":"2023-02-08 12:56:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"abaecf8efb87c3437ac6aaf9ddc63a101eea1517","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-02-22 14:19:15.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f5693b51d0552ec2d26da9a875d26528089b0b86","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-02-27 02:27:14.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"912297ec2e9545acd351ff864e2ccc3aec718e8c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-26 20:56:41.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"eead9852bafc4697cf712cea79c0e0e8b27442d2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"date":"2025-04-17 09:02:59.000000000","message":"Abandoned\n\nNot relevant","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"da7cd40c45bd4789aa3e441b6938656a7f5d25d9","revisions":{"138801981dd837b857dcae57213f841a1b9a0b79":{"kind":"REWORK","_number":1,"created":"2023-02-05 14:38:02.000000000","uploader":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"ref":"refs/changes/21/872721/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/872721/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/872721/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/872721/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/872721/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/872721/1"}}},"commit":{"parents":[{"commit":"f253f99c1216705cc52e65f6229f8ae42bfda178","subject":"Do not support dimensions:kernel_memory on Docker API 1.42","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/f253f99c1216705cc52e65f6229f8ae42bfda178"}]}],"author":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2023-02-05 14:08:07.000000000","tz":0},"committer":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2023-02-05 14:35:45.000000000","tz":0},"subject":"Don\u0027t template ceilometer event_definitions.yaml","message":"Don\u0027t template ceilometer event_definitions.yaml\n\nCeilometer provides its own event_definitions.yaml file which is periodically\nupdated by the project, and is used by Ceilometer services by default. Not\ntemplating this file ensures that kolla-ansible does not need to keep it\nin-sync, while retaining the ability for operators to override its content if\nrequired.\n\nCloses-Bug: #1920095\nChange-Id: I39737e997fb308a781265c270668583222d42a50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/138801981dd837b857dcae57213f841a1b9a0b79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/138801981dd837b857dcae57213f841a1b9a0b79"}]},"branch":"refs/heads/master"},"b1f835bae364331ac36cb8e4702d925f87ab8dc3":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-02-05 14:42:33.000000000","uploader":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"ref":"refs/changes/21/872721/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/872721/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/872721/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/872721/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/872721/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/872721/2"}}},"commit":{"parents":[{"commit":"f253f99c1216705cc52e65f6229f8ae42bfda178","subject":"Do not support dimensions:kernel_memory on Docker API 1.42","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/f253f99c1216705cc52e65f6229f8ae42bfda178"}]}],"author":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2023-02-05 14:08:07.000000000","tz":0},"committer":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2023-02-05 14:41:30.000000000","tz":0},"subject":"Don\u0027t template ceilometer event_definitions.yaml","message":"Don\u0027t template ceilometer event_definitions.yaml\n\nCeilometer provides its own event_definitions.yaml file which is\nperiodically updated by the project, and is used by Ceilometer services\nby default. Not templating this file ensures that kolla-ansible does\nnot need to keep it in-sync, while retaining the ability for operators\nto override its content if required.\n\nCloses-Bug: #1920095\nChange-Id: I39737e997fb308a781265c270668583222d42a50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/b1f835bae364331ac36cb8e4702d925f87ab8dc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/b1f835bae364331ac36cb8e4702d925f87ab8dc3"}]},"branch":"refs/heads/master"},"da7cd40c45bd4789aa3e441b6938656a7f5d25d9":{"kind":"REWORK","_number":3,"created":"2023-02-06 10:38:09.000000000","uploader":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"ref":"refs/changes/21/872721/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/872721/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/872721/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/872721/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/872721/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/872721/3"}}},"commit":{"parents":[{"commit":"f253f99c1216705cc52e65f6229f8ae42bfda178","subject":"Do not support dimensions:kernel_memory on Docker API 1.42","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/f253f99c1216705cc52e65f6229f8ae42bfda178"}]}],"author":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2023-02-05 14:08:07.000000000","tz":0},"committer":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2023-02-06 10:37:50.000000000","tz":0},"subject":"Don\u0027t template ceilometer event_definitions.yaml","message":"Don\u0027t template ceilometer event_definitions.yaml\n\nCeilometer provides its own event_definitions.yaml file which is\nperiodically updated by the project, and is used by Ceilometer services\nby default. Not templating this file ensures that kolla-ansible does\nnot need to keep it in-sync, while retaining the ability for operators\nto override its content if required.\n\nCloses-Bug: #1920095\nChange-Id: I39737e997fb308a781265c270668583222d42a50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/da7cd40c45bd4789aa3e441b6938656a7f5d25d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/da7cd40c45bd4789aa3e441b6938656a7f5d25d9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
