)]}'
{"id":"openstack%2Fkolla-ansible~889921","triplet_id":"openstack%2Fkolla-ansible~master~I68bf149c2df9cda078f5e3954773ba4e93178755","project":"openstack/kolla-ansible","branch":"master","hashtags":[],"change_id":"I68bf149c2df9cda078f5e3954773ba4e93178755","subject":"cephadm: Move osds to spec","status":"ABANDONED","created":"2023-07-28 07:42:04.000000000","updated":"2023-07-28 09:27:57.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"1e95f94e2a14c407b5ed2ce562a5f7d9542c0c21","_number":889921,"virtual_id_number":889921,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"CC":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2023-07-28 08:53:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"290907c509305e966d148573c57c331e2dbb62ad","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-07-28 07:42:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce5a6aee122dbe7f09c6f0505cc425d39c5829f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-28 08:53:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9797ccf87f1444a3a8c8730a7fc86c9b\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b3b4e315bb0143d2b2b2a8fd9dc76c06 : SUCCESS in 44m 52s\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/654d29bc04d647c9a1dee01624bb253a : SUCCESS in 1h 08m 57s","accounts_in_message":[],"_revision_number":1},{"id":"1e95f94e2a14c407b5ed2ce562a5f7d9542c0c21","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-07-28 09:27:57.000000000","message":"Abandoned\n\nnot going to work","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cdc7f3be85c9b67bdf6fd4192f68c92adaf901f1","revisions":{"cdc7f3be85c9b67bdf6fd4192f68c92adaf901f1":{"kind":"REWORK","_number":1,"created":"2023-07-28 07:42:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/889921/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/889921/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/889921/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/889921/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/889921/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/889921/1"}}},"commit":{"parents":[{"commit":"af2ccd15725885cbf03b812195b461903f1c7a52","subject":"Merge \"Trivial: Add deploy-containers for skyline\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/af2ccd15725885cbf03b812195b461903f1c7a52"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-07-28 07:41:00.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-07-28 07:42:01.000000000","tz":120},"subject":"cephadm: Move osds to spec","message":"cephadm: Move osds to spec\n\nChange-Id: I68bf149c2df9cda078f5e3954773ba4e93178755\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/cdc7f3be85c9b67bdf6fd4192f68c92adaf901f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/cdc7f3be85c9b67bdf6fd4192f68c92adaf901f1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
