)]}'
{"id":"openstack%2Fkolla-ansible~898721","triplet_id":"openstack%2Fkolla-ansible~master~Ieef0b3aa4e6b7eadd3ab835f2b1b20bafbee275c","project":"openstack/kolla-ansible","branch":"master","hashtags":[],"change_id":"Ieef0b3aa4e6b7eadd3ab835f2b1b20bafbee275c","subject":"WIP: Check OVN DB schema on container update","status":"NEW","created":"2023-10-18 08:38:35.000000000","updated":"2025-10-09 20:15:40.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"03be8d2a0081f8aef0e806b1b2c17ea6c3a81bc2","_number":898721,"virtual_id_number":898721,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-10-09 20:15:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"reviewer_updates":[{"updated":"2023-10-18 11:36:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-05 07:18:03.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"}],"messages":[{"id":"2c6aa70ff47b36dfd6fbbc1e6cc76c3be131328c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-18 08:38:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b4ba348c7fadb832010f1c15aa8d2bd2edfcf1e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-18 08:39:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"04d4b394a7465c45a063285cc0a932964a090055","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-18 10:06:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b21b41879d38f05d54eaea6438dac12488c7e237","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 11:36:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f65932f0cf3e4b438abaf5fed175556b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4ed25284d8f421fae44ff0306b91059 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e00b749ddf454f239ec062c6873b5d82 : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67e7941ee4c24fc18f7dd29f59ee6ff2 : SUCCESS in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ae6b8467fe240b7972f982c3ea2af4b : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/84c06709bc574c659d1f89d6cd3b44a6 : SUCCESS in 4m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c15e8566c9024205bcff04784f26a6bd : SUCCESS in 3m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10dc900aaadf42eb9eb7aceb0dee7cb6 : SUCCESS in 5m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4b370f3f702448d0a33f29e6cd3749c4 : SUCCESS in 5m 17s\n- kolla-ansible-debian-ovn https://zuul.opendev.org/t/openstack/build/0dc683941c2a46ef85c83c034b004cb7 : FAILURE in 34m 48s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/2cbd1ce4b92f4d6f8889a2e43c3af229 : FAILURE in 34m 19s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/2612cb39f48c4076adc6d3b1a8b791dd : FAILURE in 29m 23s (non-voting)\n- kolla-ansible-debian-upgrade-ovn https://zuul.opendev.org/t/openstack/build/75b9ee913ff247b78579dea85f647254 : FAILURE in 40m 19s (non-voting)\n- kolla-ansible-rocky9-upgrade-ovn https://zuul.opendev.org/t/openstack/build/caae9d61b54b4ff2a95ddd45c8b3e868 : FAILURE in 1h 29m 05s (non-voting)\n- kolla-ansible-ubuntu-upgrade-ovn https://zuul.opendev.org/t/openstack/build/37890ace49344f168871417d7e6eb54b : FAILURE in 1h 20m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"060988a41d6362447f96b9844067b975033d6ccd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-18 11:53:03.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"e13ac91934834a4df129faadb085cb6c418a437a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 13:24:09.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fdddb054b88f4873af9baf53e5be28fc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/11d1a7cd9c9a4acdb613f9ccea2f618f : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/814fb6555669454dade2181996658ebb : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4be36ca1d14949d5bffc202158c3e731 : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d08aefb9bb4f47aaa0772bdf8fe63d16 : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/69781cc0f6314e5bac5881ddd4a065d3 : SUCCESS in 3m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1e99659cacaf493cb02b63541ca4c8fe : SUCCESS in 3m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84b0887368164d4ca8a6c80577709bc9 : SUCCESS in 4m 49s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f240484d100e423587e5eea809472769 : FAILURE in 5m 53s\n- kolla-ansible-debian-ovn https://zuul.opendev.org/t/openstack/build/e1e4d6de885b47fd92fa6118f41b41bd : FAILURE in 29m 44s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/9dd6a5241a1e4c1c9e1b3d692b3102c2 : FAILURE in 35m 50s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/3b41cd5f76e545e3a28a99695831642b : FAILURE in 32m 25s (non-voting)\n- kolla-ansible-debian-upgrade-ovn https://zuul.opendev.org/t/openstack/build/3eff015fab2e406583249cd85c1d2ee2 : FAILURE in 43m 27s (non-voting)\n- kolla-ansible-rocky9-upgrade-ovn https://zuul.opendev.org/t/openstack/build/41dd09909ee3475b88bf4213a15a7d57 : FAILURE in 1h 24m 26s (non-voting)\n- kolla-ansible-ubuntu-upgrade-ovn https://zuul.opendev.org/t/openstack/build/970fe74133f044d8bedf770b713740fd : FAILURE in 1h 25m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3d8f8f0f51357ef9599b3bc214987da1b4439b41","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-10 09:39:34.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"bcb8c5342d92c0795e575e6ab90a761a90666cba","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-10 09:39:41.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nThe following files contain Git conflicts:\n* ansible/roles/ovn-db/handlers/main.yml\n* zuul.d/project.yaml\n","accounts_in_message":[],"_revision_number":6},{"id":"0bd57f4b6f753d620cd7763df2060e49b4758e9e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 09:39:55.000000000","message":"Patch Set 6:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla-ansible on branch master.  The\nproblem was:\n\n  while parsing a block collection\n    in \"openstack/kolla-ansible/zuul.d/project.yaml@master\", line 2, column 1\n  did not find expected \u0027-\u0027 indicator\n    in \"openstack/kolla-ansible/zuul.d/project.yaml@master\", line 14, column 1","accounts_in_message":[],"_revision_number":6},{"id":"4bcb51f6eaf20c7bab1cfb07dd7bb40441aecfd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 09:40:08.000000000","message":"Patch Set 6: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla-ansible on branch master.  The\nproblem was:\n\n  while parsing a block collection\n    in \"openstack/kolla-ansible/zuul.d/project.yaml@master\", line 2, column 1\n  did not find expected \u0027-\u0027 indicator\n    in \"openstack/kolla-ansible/zuul.d/project.yaml@master\", line 14, column 1","accounts_in_message":[],"_revision_number":6},{"id":"e343571c70f3e87f4972ae497eb401f3484ee020","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-10 09:40:21.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"947789f916ad79e2011db347abc904d04ed5360a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 11:04:38.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d75350c3b3c945c194555e5cab346acd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/806f93449fea4cfe951713532b210d5a : SUCCESS in 6m 22s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/83fb1eb8e81c428091755e6639947d3d : FAILURE in 7m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c96ab9fa2484627a890e7b545c12774 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c11b488412c44a93be40bf91894ba65f : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ec10dd2154644aa5acbe110f503a3bba : SUCCESS in 2m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b0c0aae138624727b3198eab2779c1ee : SUCCESS in 2m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9e45e6430ab24a8593be52a2dea15bdc : SUCCESS in 5m 09s (non-voting)\n- kolla-ansible-debian-ovn https://zuul.opendev.org/t/openstack/build/4cb4870cca45448c902cf4871a520e2f : FAILURE in 6m 17s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/f5c7b00ff6584b1baeddb06dbad1431c : FAILURE in 6m 39s (non-voting)\n- kolla-ansible-debian-upgrade-ovn https://zuul.opendev.org/t/openstack/build/83873dddaffe492a9c18b21dab65c560 : FAILURE in 58m 11s (non-voting)\n- kolla-ansible-ubuntu-upgrade-ovn https://zuul.opendev.org/t/openstack/build/04b708705873476a93297ca684665158 : FAILURE in 1h 03m 50s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"bba3e0c6fc416794d7efda9c925258c65992134f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-10 11:24:31.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"f034b8a9cb432b8447a626ec02f8f56ee181e246","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-10 12:31:41.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"5e815a6ac1d17c89ccf596f6b4f27e239d42c045","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 14:19:24.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/29106c7540124cedb8be3e7a04f30fc0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/100f09eecac54b8fb2b40f2e5e38347d : SUCCESS in 7m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e155e633ea3746588562420ea0b6f7b9 : FAILURE in 7m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18fe787ec3104d22995f3eba4eecfaee : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7d84910afd744978eb87e3b387f1f84 : SUCCESS in 5m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bb14a421b2bb43c7a11f9ecae92988c2 : SUCCESS in 4m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1d3a958c4c804cf3965f75923123f6c3 : SUCCESS in 4m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/594fa618cdbd4a7a9ee68af3dc44d0d8 : SUCCESS in 7m 44s (non-voting)\n- kolla-ansible-debian-ovn https://zuul.opendev.org/t/openstack/build/a3a1738ad3734051ade34bebd6dd1ecb : FAILURE in 41m 52s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/9795733d6c43493d9a49d529df690e62 : FAILURE in 42m 18s (non-voting)\n- kolla-ansible-debian-upgrade-ovn https://zuul.opendev.org/t/openstack/build/c8b2c7a7f7344944853a8e7eda4bda2f : FAILURE in 1h 39m 33s (non-voting)\n- kolla-ansible-ubuntu-upgrade-ovn https://zuul.opendev.org/t/openstack/build/0640d567a5a9488b9246b614b3202649 : FAILURE in 1h 06m 35s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"c98fadccb6c6a61365c8e7988394b8ea2fb59a28","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-10 14:24:35.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"76e9f5be95f7f34b2adb4907b344ea9209e05952","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-10 14:56:58.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"2bff488a5c15f32eab2692bf9a70532bf6998903","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 16:40:00.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e47c5a7015a4910a813fa9678ad7f92\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f2e118829574f298d106094abe73fc9 : SUCCESS in 7m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ee22d147b8c345caab7427cbfb9eca31 : SUCCESS in 7m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8c978065e6040d98f712803070b3173 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/594340a165f54fe18101ecaf69ad7a19 : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/49086584071a4330bc4774af48ab6496 : SUCCESS in 2m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/05cb857916f9450eb1dff792fb826773 : SUCCESS in 3m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/81778501a4d04c5bba38261ad2d37ad0 : SUCCESS in 7m 11s (non-voting)\n- kolla-ansible-debian-ovn https://zuul.opendev.org/t/openstack/build/6fd33af4f2f343bea99fb156e03d717f : FAILURE in 36m 55s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/594efea49bca414f84e687c84d865937 : FAILURE in 27m 33s (non-voting)\n- kolla-ansible-debian-upgrade-ovn https://zuul.opendev.org/t/openstack/build/bff8ac268395452cb4225562dced70c8 : FAILURE in 54m 34s (non-voting)\n- kolla-ansible-ubuntu-upgrade-ovn https://zuul.opendev.org/t/openstack/build/38e2413e88174cec8e4f157f82384df0 : FAILURE in 1h 37m 26s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"7cf6b563e68fedec99754530a7022b80625b29e3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-10 18:38:10.000000000","message":"Patch Set 12: Published edit on patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"9657ec6dd798b0569f3a33d6b632782ecb3debea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 18:38:16.000000000","message":"Patch Set 12: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla-ansible for 898721,12","accounts_in_message":[],"_revision_number":12},{"id":"81f7dcd66b8931a8a6dd02db9e1299de40c6f575","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 18:38:18.000000000","message":"Patch Set 12:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla-ansible for 898721,12","accounts_in_message":[],"_revision_number":12},{"id":"55b9a993b46069293b78b23aa30253509901859c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-10 18:38:27.000000000","message":"Patch Set 13: Patch Set 12 was rebased\n\nThe following files contain Git conflicts:\n* zuul.d/project.yaml\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"8ddb5cfe6a5cefd6aaad9b6a522f04cd3c6e0e48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 18:38:40.000000000","message":"Patch Set 13: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla-ansible on branch master.  The\nproblem was:\n\n  while parsing a block collection\n    in \"openstack/kolla-ansible/zuul.d/project.yaml@master\", line 2, column 1\n  did not find expected \u0027-\u0027 indicator\n    in \"openstack/kolla-ansible/zuul.d/project.yaml@master\", line 14, column 1","accounts_in_message":[],"_revision_number":13},{"id":"910e9ae2108ab7eb43766ee454d50efb78b544dc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 18:38:48.000000000","message":"Patch Set 13:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla-ansible on branch master.  The\nproblem was:\n\n  while parsing a block collection\n    in \"openstack/kolla-ansible/zuul.d/project.yaml@master\", line 2, column 1\n  did not find expected \u0027-\u0027 indicator\n    in \"openstack/kolla-ansible/zuul.d/project.yaml@master\", line 14, column 1","accounts_in_message":[],"_revision_number":13},{"id":"c4242cbada26c5070a5b868048074b72ad4e4e6e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-10 18:40:51.000000000","message":"Patch Set 14: Published edit on patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"abaf8e400642b935d360bb06c386fc770feeb6d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 20:21:49.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/753f9fef7d3c47b7a0330ed32358ac46\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/660d268b3bc044218330de83f6be3c5b : SUCCESS in 5m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/64ff319340374e1183b3b20dd413d8c4 : SUCCESS in 7m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43a131cc141245098b7bc7f5559cd7bb : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/649d75a753c34e31885c5d891355487c : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5b49755294004feab7b8eda0cfa694cd : SUCCESS in 3m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/31afa32ee76b44b4b207985f72322e4a : SUCCESS in 2m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b02b89f1aec447a7b0c0127ef3e2ad51 : SUCCESS in 5m 35s (non-voting)\n- kolla-ansible-debian-ovn https://zuul.opendev.org/t/openstack/build/5c9bd5433a314839b35a86e812b46c59 : FAILURE in 38m 34s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/81bfb9a949604c24b721cb22c58ecb5c : FAILURE in 36m 54s (non-voting)\n- kolla-ansible-debian-upgrade-ovn https://zuul.opendev.org/t/openstack/build/03c0c45f219d45a5aa3e43acd4395041 : FAILURE in 1h 28m 18s (non-voting)\n- kolla-ansible-ubuntu-upgrade-ovn https://zuul.opendev.org/t/openstack/build/7096f890ce794a99ba5c51c01a439c6f : FAILURE in 1h 27m 50s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"81117bb49b33a5f40accc3c572b7e8fc1b382aa4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 23:26:25.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fd5e12679204ab59a9689da909226ab\n\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/77e0049aeee943e7ac633a68df5ba6e2 : SUCCESS in 1h 54m 53s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/40c1225d19e244e2a40d3cf855b25633 : SUCCESS in 1h 04m 55s (non-voting)\n- kolla-ansible-debian-aarch64-podman https://zuul.opendev.org/t/openstack/build/34066b7e97284935b95bc284a449dc5f : SUCCESS in 1h 44m 04s (non-voting)\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/3a4b6319df7749bb904fefca903fabec : SUCCESS in 1h 20m 42s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"a45359b25d8f76f8a3368a65cb83c702cc28ef4e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-11 08:18:08.000000000","message":"Patch Set 15: Published edit on patch set 14.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"f4f3b0ff5897bfeef60ab725ef20016fee171fa4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 09:51:57.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a8cd9d3b2634866a65d3537c4a3904c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce2704e29eeb4a8496cb2eff70cf119b : SUCCESS in 6m 17s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/702f9687b33b4899b218b52b43995194 : SUCCESS in 3m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7728e76cb5642229b446beb86a66a8d : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf01112d345949aa9dd272e033755c76 : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7ec9d71bf2c9478f9609c7eb20f292ed : SUCCESS in 1m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b7832bcc19b14c82a8ad08ea63b1ee51 : SUCCESS in 2m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d65246a99fbb4a9b9b949b3f7a9228cd : SUCCESS in 5m 59s (non-voting)\n- kolla-ansible-debian-ovn https://zuul.opendev.org/t/openstack/build/6098fcc275cd4618a268a3ef035fc52a : FAILURE in 32m 23s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/362d1a5d7e45409a8b82c81132bfc00c : FAILURE in 38m 46s (non-voting)\n- kolla-ansible-debian-upgrade-ovn https://zuul.opendev.org/t/openstack/build/c90f35c1df9445eca502f469021b582b : FAILURE in 1h 28m 06s (non-voting)\n- kolla-ansible-ubuntu-upgrade-ovn https://zuul.opendev.org/t/openstack/build/1e06d6b8cb594c199c0df4130dcd8466 : FAILURE in 1h 29m 10s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"cbad017e5f9219bd7122eb2b9e0d3b5a32889a81","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 10:24:02.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77f9ee1f882a4c95807590c60cdaccdf\n\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/a89d57687c214c0f9c8ef3815bc02354 : SUCCESS in 1h 55m 19s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8b82fe86d3944fc3b5626aa037b88252 : SUCCESS in 1h 27m 21s (non-voting)\n- kolla-ansible-debian-aarch64-podman https://zuul.opendev.org/t/openstack/build/ef3807a51fb8451e988bbd4c89e53834 : SUCCESS in 1h 54m 12s (non-voting)\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/b3136616ca7f490e94bcc7313d847424 : SUCCESS in 2h 03m 25s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"7fade3c0bd2e7ad76c576d2a82eaf0ff5d620fa9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-14 11:09:08.000000000","message":"Patch Set 16: Published edit on patch set 15.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"b1c4cef9da6f31ae37bf9054824d03c2b9ec29b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 12:43:56.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6309a06d502d4133ac7afafbd7e0318f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23e6f33f343846cb8cc4e560de322f0e : SUCCESS in 5m 57s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a6305bcf15ad405e97f67d085150bc24 : SUCCESS in 7m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/894719c056404770a873347b74f72756 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94a3e665cf6e440b87b4a11d56ee0ec9 : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b80be3e1e6c7449bb683b011a2733591 : SUCCESS in 4m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/061bdf34ee3648998afb2f10e1516540 : SUCCESS in 1m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/00c8ab14144647a28e656033a730b9ef : SUCCESS in 6m 05s (non-voting)\n- kolla-ansible-debian-ovn https://zuul.opendev.org/t/openstack/build/22db9dc18858497ab73f5e88bfdb333a : FAILURE in 25m 04s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/8e26bb6373f1470b9994d54fa4174e1e : FAILURE in 41m 43s (non-voting)\n- kolla-ansible-debian-upgrade-ovn https://zuul.opendev.org/t/openstack/build/11ab2c7353474bf1a84835c63d9d19b0 : FAILURE in 1h 29m 46s (non-voting)\n- kolla-ansible-ubuntu-upgrade-ovn https://zuul.opendev.org/t/openstack/build/5bd7cffd65d44f0c9eeb630a84f5ad59 : FAILURE in 1h 29m 32s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"7cbf32f8ea5ef9a1ac6ac0fd87f289320c4e61d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 12:52:46.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/443366ee17d04851a079e8b5355fec81\n\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/88710293f01f45dd867118a582a3bbb2 : SUCCESS in 1h 27m 03s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fabcb1ebd23642238b8b0dede92d3d7d : SUCCESS in 1h 08m 13s (non-voting)\n- kolla-ansible-debian-aarch64-podman https://zuul.opendev.org/t/openstack/build/2f6a717701ea4dd1afc870e42544f771 : SUCCESS in 1h 19m 07s (non-voting)\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/887fe880f0e843f98d6b20a88cff1115 : SUCCESS in 1h 40m 46s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"89d4ddc505a308347f3e9fdcceb0118a83f6b68c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-09 08:51:48.000000000","message":"Patch Set 17: Patch Set 16 was rebased\n\nThe following files contain Git conflicts:\n* zuul.d/project.yaml\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"84b99c9902c2d89931fffba86b1efac00a2d2e84","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-09 08:51:57.000000000","message":"Patch Set 18: Published edit on patch set 17.","accounts_in_message":[],"_revision_number":18},{"id":"eb0bb127635096b83c9475e64e4f6411440efcde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 08:52:06.000000000","message":"Patch Set 17: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla-ansible on branch master.  The\nproblem was:\n\n  while parsing a block collection\n    in \"openstack/kolla-ansible/zuul.d/project.yaml@master\", line 2, column 1\n  did not find expected \u0027-\u0027 indicator\n    in \"openstack/kolla-ansible/zuul.d/project.yaml@master\", line 37, column 1","accounts_in_message":[],"_revision_number":17},{"id":"8e6b2f6ad9c6984a515caec6ae6ae401eed5795e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 08:52:19.000000000","message":"Patch Set 17:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla-ansible on branch master.  The\nproblem was:\n\n  while parsing a block collection\n    in \"openstack/kolla-ansible/zuul.d/project.yaml@master\", line 2, column 1\n  did not find expected \u0027-\u0027 indicator\n    in \"openstack/kolla-ansible/zuul.d/project.yaml@master\", line 37, column 1","accounts_in_message":[],"_revision_number":17},{"id":"a9ab9638b2cfb51a449a6cb8f39cdebe86f8ae98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 10:15:01.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d893a4356fff476683da97b6abbc0fcb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6205346bbfa9491d85c2cf262ac82b4a : SUCCESS in 7m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/622152f8196743aa83605253cfb88778 : SUCCESS in 8m 10s\n- kolla-ansible-debian-bookworm-ovn https://zuul.opendev.org/t/openstack/build/fb364a18e9f94223a995b36ded970ec2 : FAILURE in 37m 33s (non-voting)\n- kolla-ansible-debian-bookworm-ovn-upgrade https://zuul.opendev.org/t/openstack/build/bc4a364ad6b14a5793f4b9e9f6f3bcb1 : FAILURE in 1h 00m 59s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/8a9286942c164b2e8d1e47c72fcfc09e : FAILURE in 34m 08s\n- kolla-ansible-ubuntu-noble-ovn-upgrade https://zuul.opendev.org/t/openstack/build/99c48967133f416cb42583f574da2d3c : FAILURE in 1h 21m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bfd2549eea834e0485cb29a995c7871a : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c613c619bf14d55a7092698b579b228 : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dffde74135564755a917b3c69d9a6ac1 : SUCCESS in 2m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d148ee01fca645afa5e86ee47e7e960e : SUCCESS in 3m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/88661da7004543f5bf2527d94406cca5 : SUCCESS in 5m 58s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"69e4a60c013b1cafe9b401e2ee67ef9dba36fa04","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-09 19:01:45.000000000","message":"Patch Set 19: Published edit on patch set 18.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"03be8d2a0081f8aef0e806b1b2c17ea6c3a81bc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 20:15:40.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/416917a9b57e4f15a597aa030109f5e1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ed4425de7a3439f8a263519369a4698 : SUCCESS in 7m 16s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1222fc9dcaae47a08ba8b6ee6e0570a6 : SUCCESS in 8m 14s\n- kolla-ansible-debian-bookworm-ovn https://zuul.opendev.org/t/openstack/build/3451f05cb1d04f79a91d62408f410f27 : FAILURE in 36m 47s (non-voting)\n- kolla-ansible-debian-bookworm-ovn-upgrade https://zuul.opendev.org/t/openstack/build/29b267945d39451a9ec90c0854df44ba : FAILURE in 1h 12m 38s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/207407aa90324739978b4218c3624fef : FAILURE in 33m 44s\n- kolla-ansible-ubuntu-noble-ovn-upgrade https://zuul.opendev.org/t/openstack/build/661a5ab4f3f84012940333c501da3e41 : FAILURE in 1h 11m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d32a0cb6cde4435b8041172281215f62 : SUCCESS in 2m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab5b82c8457f45139b70330d6634deb5 : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b654dc2fd81941c3881ec59238bae079 : SUCCESS in 3m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0d2cd8704576431c95e4025255cdbaaa : SUCCESS in 3m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/96958362a12e4cd5a0c887a224e81c92 : SUCCESS in 5m 30s (non-voting)","accounts_in_message":[],"_revision_number":19}],"current_revision_number":19,"current_revision":"39117882173ae97f2d54dab9e74f859bc370dba6","revisions":{"5d4c98626fd143dcb5a1c28b501490f25619b31b":{"kind":"REWORK","_number":1,"created":"2023-10-18 08:38:35.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/898721/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/898721/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/1"}}},"commit":{"parents":[{"commit":"a6120f231a5985b95ac98c1abcc672622b3190f1","subject":"Merge \"Fix Cyborg endpoints again\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/a6120f231a5985b95ac98c1abcc672622b3190f1"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-10-18 08:38:14.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-10-18 08:38:32.000000000","tz":120},"subject":"WIP: Check OVN DB schema on container update","message":"WIP: Check OVN DB schema on container update\n\nChange-Id: Ieef0b3aa4e6b7eadd3ab835f2b1b20bafbee275c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/5d4c98626fd143dcb5a1c28b501490f25619b31b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/5d4c98626fd143dcb5a1c28b501490f25619b31b"}]},"branch":"refs/heads/master"},"f0c3a168979d7ea4c1f656b68b739c157a51bf90":{"kind":"REWORK","_number":2,"created":"2023-10-18 08:39:56.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/898721/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/898721/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/2"}}},"commit":{"parents":[{"commit":"a6120f231a5985b95ac98c1abcc672622b3190f1","subject":"Merge \"Fix Cyborg endpoints again\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/a6120f231a5985b95ac98c1abcc672622b3190f1"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-10-18 08:38:14.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-10-18 08:39:43.000000000","tz":120},"subject":"WIP: Check OVN DB schema on container update","message":"WIP: Check OVN DB schema on container update\n\nChange-Id: Ieef0b3aa4e6b7eadd3ab835f2b1b20bafbee275c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/f0c3a168979d7ea4c1f656b68b739c157a51bf90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/f0c3a168979d7ea4c1f656b68b739c157a51bf90"}]},"branch":"refs/heads/master"},"29abce3f018ad2df6e8180fccf795b3a31e6b877":{"kind":"REWORK","_number":3,"created":"2023-10-18 10:06:54.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/898721/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/898721/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/3"}}},"commit":{"parents":[{"commit":"a6120f231a5985b95ac98c1abcc672622b3190f1","subject":"Merge \"Fix Cyborg endpoints again\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/a6120f231a5985b95ac98c1abcc672622b3190f1"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-10-18 08:38:14.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-10-18 10:06:44.000000000","tz":120},"subject":"WIP: Check OVN DB schema on container update","message":"WIP: Check OVN DB schema on container update\n\nChange-Id: Ieef0b3aa4e6b7eadd3ab835f2b1b20bafbee275c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/29abce3f018ad2df6e8180fccf795b3a31e6b877"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/29abce3f018ad2df6e8180fccf795b3a31e6b877"}]},"branch":"refs/heads/master"},"0f1141b2ff0086b2845ae49c9212c9e3bb6bdd6b":{"kind":"REWORK","_number":4,"created":"2023-10-18 11:53:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/898721/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/898721/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/4"}}},"commit":{"parents":[{"commit":"a6120f231a5985b95ac98c1abcc672622b3190f1","subject":"Merge \"Fix Cyborg endpoints again\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/a6120f231a5985b95ac98c1abcc672622b3190f1"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-10-18 08:38:14.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-10-18 11:52:56.000000000","tz":120},"subject":"WIP: Check OVN DB schema on container update","message":"WIP: Check OVN DB schema on container update\n\nChange-Id: Ieef0b3aa4e6b7eadd3ab835f2b1b20bafbee275c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/0f1141b2ff0086b2845ae49c9212c9e3bb6bdd6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/0f1141b2ff0086b2845ae49c9212c9e3bb6bdd6b"}]},"branch":"refs/heads/master"},"28f0ee19bcb0cd5a675558cdad4841c57fb1e8f3":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2025-07-10 09:39:34.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/898721/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/898721/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/5"}}},"commit":{"parents":[{"commit":"a6120f231a5985b95ac98c1abcc672622b3190f1","subject":"Merge \"Fix Cyborg endpoints again\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/a6120f231a5985b95ac98c1abcc672622b3190f1"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-10-18 08:38:14.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-07-10 09:39:34.000000000","tz":0},"subject":"WIP: Check OVN DB schema on container update","message":"WIP: Check OVN DB schema on container update\n\nSigned-Off-By: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\nChange-Id: Ieef0b3aa4e6b7eadd3ab835f2b1b20bafbee275c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/28f0ee19bcb0cd5a675558cdad4841c57fb1e8f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/28f0ee19bcb0cd5a675558cdad4841c57fb1e8f3"}]},"branch":"refs/heads/master","description":"Edit commit message"},"5fa8416bcd4b75b802df43c310579f4dfac37494":{"kind":"REWORK","_number":6,"created":"2025-07-10 09:39:41.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/898721/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/898721/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/6"}}},"commit":{"parents":[{"commit":"07726215dc3734c488cfcc40138740cb0b684d96","subject":"CI: Bump instance creation timeout to 5 minutes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/07726215dc3734c488cfcc40138740cb0b684d96"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-10-18 08:38:14.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-07-10 09:39:41.000000000","tz":0},"subject":"WIP: Check OVN DB schema on container update","message":"WIP: Check OVN DB schema on container update\n\nSigned-Off-By: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\nChange-Id: Ieef0b3aa4e6b7eadd3ab835f2b1b20bafbee275c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/5fa8416bcd4b75b802df43c310579f4dfac37494"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/5fa8416bcd4b75b802df43c310579f4dfac37494"}]},"branch":"refs/heads/master","description":"Rebase"},"ebaecfbe7e322e3c084294495e1cbedccdcd6a44":{"kind":"REWORK","_number":7,"created":"2025-07-10 09:40:21.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/898721/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/898721/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/7"}}},"commit":{"parents":[{"commit":"07726215dc3734c488cfcc40138740cb0b684d96","subject":"CI: Bump instance creation timeout to 5 minutes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/07726215dc3734c488cfcc40138740cb0b684d96"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-10-18 08:38:14.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-07-10 09:40:21.000000000","tz":0},"subject":"WIP: Check OVN DB schema on container update","message":"WIP: Check OVN DB schema on container update\n\nSigned-Off-By: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\nChange-Id: Ieef0b3aa4e6b7eadd3ab835f2b1b20bafbee275c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ebaecfbe7e322e3c084294495e1cbedccdcd6a44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ebaecfbe7e322e3c084294495e1cbedccdcd6a44"}]},"branch":"refs/heads/master"},"0d1edae11791e9fdd3ef007df9cda50bbb4bf38b":{"kind":"REWORK","_number":8,"created":"2025-07-10 11:24:31.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/898721/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/898721/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/8"}}},"commit":{"parents":[{"commit":"07726215dc3734c488cfcc40138740cb0b684d96","subject":"CI: Bump instance creation timeout to 5 minutes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/07726215dc3734c488cfcc40138740cb0b684d96"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-10-18 08:38:14.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-07-10 11:24:31.000000000","tz":0},"subject":"WIP: Check OVN DB schema on container update","message":"WIP: Check OVN DB schema on container update\n\nSigned-Off-By: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\nChange-Id: Ieef0b3aa4e6b7eadd3ab835f2b1b20bafbee275c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/0d1edae11791e9fdd3ef007df9cda50bbb4bf38b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/0d1edae11791e9fdd3ef007df9cda50bbb4bf38b"}]},"branch":"refs/heads/master"},"c3508416d12e83f4eb3fb80526347465e0451d2e":{"kind":"REWORK","_number":9,"created":"2025-07-10 12:31:41.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/898721/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/898721/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/9"}}},"commit":{"parents":[{"commit":"07726215dc3734c488cfcc40138740cb0b684d96","subject":"CI: Bump instance creation timeout to 5 minutes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/07726215dc3734c488cfcc40138740cb0b684d96"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-10-18 08:38:14.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-07-10 12:31:41.000000000","tz":0},"subject":"WIP: Check OVN DB schema on container update","message":"WIP: Check OVN DB schema on container update\n\nSigned-Off-By: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\nChange-Id: Ieef0b3aa4e6b7eadd3ab835f2b1b20bafbee275c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c3508416d12e83f4eb3fb80526347465e0451d2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c3508416d12e83f4eb3fb80526347465e0451d2e"}]},"branch":"refs/heads/master"},"d273ac4337f2d7b71efe52ddf986808b63c78b6d":{"kind":"REWORK","_number":10,"created":"2025-07-10 14:24:35.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/898721/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/898721/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/10"}}},"commit":{"parents":[{"commit":"07726215dc3734c488cfcc40138740cb0b684d96","subject":"CI: Bump instance creation timeout to 5 minutes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/07726215dc3734c488cfcc40138740cb0b684d96"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-10-18 08:38:14.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-07-10 14:24:35.000000000","tz":0},"subject":"WIP: Check OVN DB schema on container update","message":"WIP: Check OVN DB schema on container update\n\nSigned-Off-By: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\nChange-Id: Ieef0b3aa4e6b7eadd3ab835f2b1b20bafbee275c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/d273ac4337f2d7b71efe52ddf986808b63c78b6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/d273ac4337f2d7b71efe52ddf986808b63c78b6d"}]},"branch":"refs/heads/master"},"7a1ea032d192c19c586f80dc46d8f74f0232ac44":{"kind":"REWORK","_number":11,"created":"2025-07-10 14:56:58.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/898721/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/898721/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/11"}}},"commit":{"parents":[{"commit":"07726215dc3734c488cfcc40138740cb0b684d96","subject":"CI: Bump instance creation timeout to 5 minutes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/07726215dc3734c488cfcc40138740cb0b684d96"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-10-18 08:38:14.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-07-10 14:56:58.000000000","tz":0},"subject":"WIP: Check OVN DB schema on container update","message":"WIP: Check OVN DB schema on container update\n\nSigned-Off-By: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\nChange-Id: Ieef0b3aa4e6b7eadd3ab835f2b1b20bafbee275c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/7a1ea032d192c19c586f80dc46d8f74f0232ac44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/7a1ea032d192c19c586f80dc46d8f74f0232ac44"}]},"branch":"refs/heads/master"},"726468679e8ecc741c7335d9c31e22a590abf2ad":{"kind":"REWORK","_number":12,"created":"2025-07-10 18:38:10.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/898721/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/898721/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/12"}}},"commit":{"parents":[{"commit":"07726215dc3734c488cfcc40138740cb0b684d96","subject":"CI: Bump instance creation timeout to 5 minutes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/07726215dc3734c488cfcc40138740cb0b684d96"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-10-18 08:38:14.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-07-10 18:38:10.000000000","tz":0},"subject":"WIP: Check OVN DB schema on container update","message":"WIP: Check OVN DB schema on container update\n\nSigned-Off-By: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\nChange-Id: Ieef0b3aa4e6b7eadd3ab835f2b1b20bafbee275c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/726468679e8ecc741c7335d9c31e22a590abf2ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/726468679e8ecc741c7335d9c31e22a590abf2ad"}]},"branch":"refs/heads/master"},"628c8410219f851c61515b63bacd7eeb0d25f245":{"kind":"REWORK","_number":13,"created":"2025-07-10 18:38:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/898721/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/898721/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/13"}}},"commit":{"parents":[{"commit":"9634064baa2bac2c87dd9eff6f63b5bf390b5217","subject":"Merge \"fluentd: Add ovn logs to input\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/9634064baa2bac2c87dd9eff6f63b5bf390b5217"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-10-18 08:38:14.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-07-10 18:38:27.000000000","tz":0},"subject":"WIP: Check OVN DB schema on container update","message":"WIP: Check OVN DB schema on container update\n\nSigned-Off-By: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\nChange-Id: Ieef0b3aa4e6b7eadd3ab835f2b1b20bafbee275c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/628c8410219f851c61515b63bacd7eeb0d25f245"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/628c8410219f851c61515b63bacd7eeb0d25f245"}]},"branch":"refs/heads/master","description":"Rebase"},"3908de971941080289f9bd8714f795e16925be35":{"kind":"REWORK","_number":14,"created":"2025-07-10 18:40:51.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/898721/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/898721/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/14"}}},"commit":{"parents":[{"commit":"9634064baa2bac2c87dd9eff6f63b5bf390b5217","subject":"Merge \"fluentd: Add ovn logs to input\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/9634064baa2bac2c87dd9eff6f63b5bf390b5217"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-10-18 08:38:14.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-07-10 18:40:51.000000000","tz":0},"subject":"WIP: Check OVN DB schema on container update","message":"WIP: Check OVN DB schema on container update\n\nSigned-Off-By: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\nChange-Id: Ieef0b3aa4e6b7eadd3ab835f2b1b20bafbee275c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/3908de971941080289f9bd8714f795e16925be35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/3908de971941080289f9bd8714f795e16925be35"}]},"branch":"refs/heads/master"},"abf5b185764569b1fe6b62a273f0343dc2ba5597":{"kind":"REWORK","_number":15,"created":"2025-07-11 08:18:08.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/898721/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/898721/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/15"}}},"commit":{"parents":[{"commit":"9634064baa2bac2c87dd9eff6f63b5bf390b5217","subject":"Merge \"fluentd: Add ovn logs to input\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/9634064baa2bac2c87dd9eff6f63b5bf390b5217"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-10-18 08:38:14.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-07-11 08:18:08.000000000","tz":0},"subject":"WIP: Check OVN DB schema on container update","message":"WIP: Check OVN DB schema on container update\n\nSigned-Off-By: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\nChange-Id: Ieef0b3aa4e6b7eadd3ab835f2b1b20bafbee275c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/abf5b185764569b1fe6b62a273f0343dc2ba5597"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/abf5b185764569b1fe6b62a273f0343dc2ba5597"}]},"branch":"refs/heads/master"},"e0ec81df12cd648525ba28962935f1cca263273a":{"kind":"REWORK","_number":16,"created":"2025-07-14 11:09:08.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/898721/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/898721/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/16"}}},"commit":{"parents":[{"commit":"9634064baa2bac2c87dd9eff6f63b5bf390b5217","subject":"Merge \"fluentd: Add ovn logs to input\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/9634064baa2bac2c87dd9eff6f63b5bf390b5217"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-10-18 08:38:14.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-07-14 11:09:08.000000000","tz":0},"subject":"WIP: Check OVN DB schema on container update","message":"WIP: Check OVN DB schema on container update\n\nSigned-Off-By: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\nChange-Id: Ieef0b3aa4e6b7eadd3ab835f2b1b20bafbee275c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/e0ec81df12cd648525ba28962935f1cca263273a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/e0ec81df12cd648525ba28962935f1cca263273a"}]},"branch":"refs/heads/master"},"5c266a9c284fb0bdfeccd5997bfef11525bef354":{"kind":"REWORK","_number":17,"created":"2025-10-09 08:51:48.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/898721/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/898721/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/17"}}},"commit":{"parents":[{"commit":"bdd98eca87890666f66b8b3d920f7ce0948ace44","subject":"ovn-db: Add OVN_NB_DB and OVN_SB_DB env variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/bdd98eca87890666f66b8b3d920f7ce0948ace44"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-10-18 08:38:14.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-09 08:51:48.000000000","tz":0},"subject":"WIP: Check OVN DB schema on container update","message":"WIP: Check OVN DB schema on container update\n\nSigned-Off-By: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\nChange-Id: Ieef0b3aa4e6b7eadd3ab835f2b1b20bafbee275c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/5c266a9c284fb0bdfeccd5997bfef11525bef354"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/5c266a9c284fb0bdfeccd5997bfef11525bef354"}]},"branch":"refs/heads/master","description":"Rebase"},"41c3a4dc415264e6f70598e077b1893bb9f7c8c6":{"kind":"REWORK","_number":18,"created":"2025-10-09 08:51:57.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/898721/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/898721/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/18"}}},"commit":{"parents":[{"commit":"bdd98eca87890666f66b8b3d920f7ce0948ace44","subject":"ovn-db: Add OVN_NB_DB and OVN_SB_DB env variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/bdd98eca87890666f66b8b3d920f7ce0948ace44"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-10-18 08:38:14.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-09 08:51:55.000000000","tz":0},"subject":"WIP: Check OVN DB schema on container update","message":"WIP: Check OVN DB schema on container update\n\nSigned-Off-By: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\nChange-Id: Ieef0b3aa4e6b7eadd3ab835f2b1b20bafbee275c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/41c3a4dc415264e6f70598e077b1893bb9f7c8c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/41c3a4dc415264e6f70598e077b1893bb9f7c8c6"}]},"branch":"refs/heads/master"},"39117882173ae97f2d54dab9e74f859bc370dba6":{"kind":"REWORK","_number":19,"created":"2025-10-09 19:01:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/898721/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/898721/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/898721/19"}}},"commit":{"parents":[{"commit":"bdd98eca87890666f66b8b3d920f7ce0948ace44","subject":"ovn-db: Add OVN_NB_DB and OVN_SB_DB env variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/bdd98eca87890666f66b8b3d920f7ce0948ace44"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-10-18 08:38:14.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-09 19:01:45.000000000","tz":0},"subject":"WIP: Check OVN DB schema on container update","message":"WIP: Check OVN DB schema on container update\n\nSigned-Off-By: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\nChange-Id: Ieef0b3aa4e6b7eadd3ab835f2b1b20bafbee275c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/39117882173ae97f2d54dab9e74f859bc370dba6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/39117882173ae97f2d54dab9e74f859bc370dba6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
