)]}'
{"id":"openstack%2Fkolla-ansible~904521","triplet_id":"openstack%2Fkolla-ansible~master~I638ba32e97234900745df62056710dcc37e7db77","project":"openstack/kolla-ansible","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-01-03 10:11:26.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2024-01-03 08:41:21.000000000","reason":"\u003cGERRIT_ACCOUNT_27339\u003e replied on the change","reason_account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}}},"hashtags":[],"change_id":"I638ba32e97234900745df62056710dcc37e7db77","subject":"haproxy: Fix single frontend after LE cert path change","status":"MERGED","created":"2024-01-02 17:30:45.000000000","updated":"2024-01-03 10:12:41.000000000","submitted":"2024-01-03 10:11:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"904521","meta_rev_id":"8191dee33f1ade6846dac23f21187a40031a32dd","_number":904521,"virtual_id_number":904521,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-03 10:11:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2024-01-03 08:41:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":2,"date":"2024-01-03 08:19:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-01-03 08:41:21.000000000","_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-01-03 07:47:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-01-03 07:47:08.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-02 19:08:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-02 20:20:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-03 08:19:41.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-01-03 08:28:51.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"}],"messages":[{"id":"8c416b41a35f268c7f639985f1a9f3141c62a498","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-01-02 17:30:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"05e52fc2966913b5c883c03575bdaf00a98505b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-02 19:08:59.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba19ec18178e4c0ca4258f4d70782c7b\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5012f49f079f4a2abd1c05cb905c9e6a : SUCCESS in 57m 20s\n- kolla-ansible-debian-aarch64-podman https://zuul.opendev.org/t/openstack/build/3b971dde4c8140dfb9994b931153af91 : FAILURE in 27m 04s\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/11af5bca9de847debcd33a5e69242990 : SUCCESS in 1h 32m 34s","accounts_in_message":[],"_revision_number":1},{"id":"0727334c4c642d6da332bf6499d8a233f5e1a02c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-02 20:20:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ae6ab89db8843ad8b4172deabb3e3f7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0d795ca55914aaa99556c60b21d37dd : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fd0484df4e844438dd1e2c9006aa7b3 : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f8c07529ae3413287de87ca6a8633bc : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e0c08730df4342ec81d2e4ac10e2a026 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afb3aaa643dd4622a7a0af77e8278d22 : SUCCESS in 6m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c0b30fbac5764cacbcb4ccf86f85b3f0 : SUCCESS in 6m 19s\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8d9a6e24bf4e4361a4c0068a165a9e56 : SUCCESS in 1h 10m 57s (non-voting)\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/bf0821c6ad5f4a3cab885e2bd1525293 : SUCCESS in 57m 44s\n- kolla-ansible-debian-podman https://zuul.opendev.org/t/openstack/build/52e706ef80e045aca649b6b989d83e9b : SUCCESS in 1h 01m 35s\n- kolla-ansible-openeuler https://zuul.opendev.org/t/openstack/build/bf5a1f6a54a94fc78b059c9c4e1d75ef : SUCCESS in 1h 00m 42s (non-voting)\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/aa267b8f7ba14a77a55559a7acfa2083 : SUCCESS in 58m 20s\n- kolla-ansible-rocky9-podman https://zuul.opendev.org/t/openstack/build/421866d68be6408996151bd2ddea93c3 : SUCCESS in 1h 10m 05s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6d4d92c9b11147c58c7818a1663e0042 : SUCCESS in 45m 10s\n- kolla-ansible-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b5b1dbd6076445a583f295a91ca28317 : SUCCESS in 1h 09m 33s\n- kolla-ansible-rocky9-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/add58264a47d4a4a860ff2a0589bed8b : SUCCESS in 1h 11m 25s (non-voting)\n- kolla-ansible-ubuntu-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/0d645918a70c4cc8bdb195e6e8541554 : SUCCESS in 48m 18s (non-voting)\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/861a580bf0aa4083b804349963b7a9d8 : SUCCESS in 1h 25m 13s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/57f2b943d25d4999bc15ab9b7a4a57c5 : SUCCESS in 1h 20m 57s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/53f16f19e2594b4eb9031f235b10b68f : SUCCESS in 1h 11m 09s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/0cbee42c2d0249a3af786a70898cd78d : SUCCESS in 1h 23m 02s (non-voting)\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/b5f19ef255cd455dae0c81cbfa074f08 : SUCCESS in 1h 25m 00s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/174685e666f24a2d84bd106ae480da28 : SUCCESS in 2h 40m 18s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/224f919cb6664d8c8e03d2c158d4ef23 : SUCCESS in 2h 25m 23s (non-voting)\n- kolla-ansible-ubuntu-lets-encrypt https://zuul.opendev.org/t/openstack/build/87663a32f2e84b99ab3a2becf874842a : SUCCESS in 1h 14m 05s (non-voting)\n- kolla-ansible-rocky9-lets-encrypt https://zuul.opendev.org/t/openstack/build/62835b6db1484ec79806ca8c7cb34725 : SUCCESS in 1h 18m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"01790959d89d921736730366b2be461b01cf848b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-01-03 07:47:08.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"e7bf0ba582079269c224c06d3f6c5ea2ed2ca07f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-01-03 07:47:19.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"e6013f69022929c536a512f49e75fa2affd425f1","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-01-03 08:19:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c8c0c353978113f2b69fee0e82da86e1a9f54994","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-01-03 08:41:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c62a3319ad6968b09a6049a232b2e0f261aa5de3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-03 08:41:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"84b34270f3de0e77693a21051da7dbe0173f6d09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-03 10:11:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92c875747d2a4938a39d8b0609ed949b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7429a1fe6e304600ac823f9ec5289e91 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40b7ca8841e7455fae185c576be0eef5 : SUCCESS in 4m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3c99de1d79e94213b864a5f908d74bab : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43dfab6ead254487a1506674a4ab43d6 : SUCCESS in 5m 22s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/000e80ad1d374e3797c6bdf4bc8f5ba1 : SUCCESS in 6m 06s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/60c7adda67384f34bb2641f10eee63d7 : SUCCESS in 55m 04s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ca2b8d3f508e4161843a800362d3b4ba : SUCCESS in 1h 25m 56s\n- kolla-ansible-debian-podman https://zuul.opendev.org/t/openstack/build/cb83b42f23af4d61a63c5aa242e41593 : SUCCESS in 1h 09m 32s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/09cd0f51db5142bca76c662c084a182d : SUCCESS in 59m 33s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d06158d8b5934ab89ad3b0e9948bafa0 : SUCCESS in 1h 24m 58s\n- kolla-ansible-rocky9-podman https://zuul.opendev.org/t/openstack/build/bd2dcd163fff46018f656c373d2ffd3e : SUCCESS in 1h 03m 41s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/050d2e5caea442668fc640e2d6855c05 : SUCCESS in 53m 18s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/bfb115b0f5eb44ff8787c9cd7dc60c16 : SUCCESS in 1h 21m 20s\n- kolla-ansible-ubuntu-podman https://zuul.opendev.org/t/openstack/build/67f09a490a314f4f87462f55c699b395 : SUCCESS in 1h 12m 44s","accounts_in_message":[],"_revision_number":1},{"id":"2686083fa3ab94435748a1a9b40f69570253ec55","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-03 10:11:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8191dee33f1ade6846dac23f21187a40031a32dd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-03 10:12:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a522f1a213a4a968cd51a915053ea75\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44e2f83323d64db097329cb5455f0a55 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"21e5b21f59a0824cf52b358a562ea962550e8d2d","revisions":{"21e5b21f59a0824cf52b358a562ea962550e8d2d":{"kind":"REWORK","_number":1,"created":"2024-01-02 17:30:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/904521/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/904521/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/904521/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/904521/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/904521/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/904521/1"}}},"commit":{"parents":[{"commit":"65886c1d4efc8f17fe870350b0a05d15c5eacac0","subject":"Merge \"Fix wsrep sync status task while switched to TCP/IP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/65886c1d4efc8f17fe870350b0a05d15c5eacac0"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-01-02 17:29:51.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-01-02 17:30:42.000000000","tz":60},"subject":"haproxy: Fix single frontend after LE cert path change","message":"haproxy: Fix single frontend after LE cert path change\n\nI35317ea0343f0db74ddc0e587862e95408e9e106 changed certificate path but omitted\nsingle frontend template.\n\nChange-Id: I638ba32e97234900745df62056710dcc37e7db77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/21e5b21f59a0824cf52b358a562ea962550e8d2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/21e5b21f59a0824cf52b358a562ea962550e8d2d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
