)]}'
{"id":"openstack%2Fkolla-ansible~911069","triplet_id":"openstack%2Fkolla-ansible~stable%2Fzed~I14ae2be2e19ad06e3190e2e948bac7ce77e80d4b","project":"openstack/kolla-ansible","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-03-06 18:34:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I14ae2be2e19ad06e3190e2e948bac7ce77e80d4b","subject":"Configure missing nova services to expose vendordata over configdrive","status":"MERGED","created":"2024-03-05 16:52:46.000000000","updated":"2024-03-06 18:35:33.000000000","submitted":"2024-03-06 18:34:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"911069","cherry_pick_of_change":905843,"cherry_pick_of_patch_set":9,"meta_rev_id":"c90adf6b74c312b283412718e5df9a6ab5a13bea","_number":911069,"virtual_id_number":911069,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-06 18:34:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-03-06 17:17:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-03-06 14:57:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-03-06 09:16:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-03-06 17:17:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"}],"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-05 16:52:46.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"state":"CC"},{"updated":"2024-03-05 18:02:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-05 18:33:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-06 09:16:32.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2024-03-06 14:57:20.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-03-06 17:17:19.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"501a1b2c81d254373a2f42c5985cb3e1d60384cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-03-05 16:52:46.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b6d6360213e1ec39aa96974f30e14f563fe093b4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 18:02:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a951b2c57de46de9b158c4b5f0edeb6\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/44532300f86245aabe925e67a32a9244 : SUCCESS in 1h 08m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8c4a3e94e95039658a0a255d8effaf61c47037a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 18:33:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fa0228b64d547009bb09fdd902d609b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0590c91794a41dca3f5a5637efc5efe : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9813f2560c93418caed8464950d3a411 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc635eaea5fa43af881764bbb5f2bc5a : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c37393067056483994b6090627ebc9d1 : SUCCESS in 5m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c9ef449eedb34106a458ea191598ca67 : SUCCESS in 9m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43622ce104df4e40a1ab0303218afd30 : SUCCESS in 4m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/55e63940a4564542ad651924c6061aad : SUCCESS in 10m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3d471cca85d4a03b365094fcce0417a : SUCCESS in 5m 14s\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/cb459cf64461476eb07db7d94a17b909 : FAILURE in 17m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e48760a3b9534079aa31cb735649e2b7 : SUCCESS in 51m 48s\n- kolla-ansible-openeuler-source https://zuul.opendev.org/t/openstack/build/bcbc196c93244759970419a538bc2aa9 : SUCCESS in 1h 02m 25s (non-voting)\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/b32db7e64309480796006f00e27f511d : SUCCESS in 56m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7d663db29e3e4582bf5b7e942fe5d2f2 : SUCCESS in 41m 03s\n- kolla-ansible-rocky9-source-kvm https://zuul.opendev.org/t/openstack/build/95f9bc9e734345ac9a409a74ed438e44 : SUCCESS in 47m 31s (non-voting)\n- kolla-ansible-ubuntu-source-kvm https://zuul.opendev.org/t/openstack/build/d9488ebc3d30403388cd2bd370a36c01 : SUCCESS in 27m 40s (non-voting)\n- kolla-ansible-rocky9-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/80d45089308c46c29872249c02e2bd77 : SUCCESS in 1h 13m 59s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/f61d7945f0d244dcb063bbbdba881208 : SUCCESS in 49m 29s (non-voting)\n- kolla-ansible-rocky9-source-ironic https://zuul.opendev.org/t/openstack/build/4f8466db079643e9a8ebd12adde4b40a : FAILURE in 40m 12s (non-voting)\n- kolla-ansible-debian-source-ironic https://zuul.opendev.org/t/openstack/build/9ab6e68c61a2495288b2c6fd984da6a1 : SUCCESS in 58m 16s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/a3c4ea5953a84130949290c7129e6972 : SUCCESS in 1h 24m 45s (non-voting)\n- kolla-ansible-rocky9-source-cephadm https://zuul.opendev.org/t/openstack/build/94c13debad0340c5bde9b5030406308d : SUCCESS in 1h 20m 30s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/deda1942e3d54ab7ace2b91bcd83986c : SUCCESS in 1h 34m 40s (non-voting)\n- kolla-ansible-rocky9-hashi-vault https://zuul.opendev.org/t/openstack/build/3f195ebcf37f43cf955eb5c7865c3ddc : SUCCESS in 4m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eef7c92ce4447dad8c6f4de3adbd1e4a473e55a5","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-03-06 09:16:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2383aaff551a166de727e5eaf5c63bac6a24de9f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-03-06 14:57:20.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e28b0aafb54cff1024860a7ce66ab8bd967c45b3","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-06 17:17:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"78b125bd315f54f5256c2514ca8873f33c4e915a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 17:17:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"09b423ee95a78390b20de89f51f48039bda2faba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 18:34:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8abb87b89c2a47dca9c098d2fd4acf21\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee07c4646e854251bcd1f7623e6ad798 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c24aa0ba78d40759bf5caac403d5633 : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f53e2de387d341fa8e8b2f0099dc64ac : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ae477eb73134778ae14905f27fecc32 : SUCCESS in 4m 08s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/256fbf74c3014586993db6e835894182 : SUCCESS in 6m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90ab72a58410498593a2c27a0d8f1e9b : SUCCESS in 3m 43s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/512be031263040159ea02a68e962e10c : SUCCESS in 53m 45s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/607dd3cafbb04934b46db5eff483a65d : SUCCESS in 1h 05m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/537d1ef2c54c414488c342af19a47eb1 : SUCCESS in 53m 39s","accounts_in_message":[],"_revision_number":1},{"id":"6d0fbc3a785a72a533617c4e6a74ee666b8e6c1c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 18:34:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c90adf6b74c312b283412718e5df9a6ab5a13bea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 18:35:33.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a7a2c00f398c4675b68cd68e5a6d2c20\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/393f448c853c468da02d7fdad12b7861 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6da8c2464d7b483289cfb55993103407 : FAILURE in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7abaa51c9cc3e12797b4f53750f64e53874be063","revisions":{"7abaa51c9cc3e12797b4f53750f64e53874be063":{"kind":"REWORK","_number":1,"created":"2024-03-05 16:52:46.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/69/911069/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/69/911069/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/69/911069/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/69/911069/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/69/911069/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/69/911069/1"}}},"commit":{"parents":[{"commit":"9d97f7f87c2902cd543af20aec51535693e9178d","subject":"Merge \"Fix gnocchi-metricd when TLS and Swift enabled\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/9d97f7f87c2902cd543af20aec51535693e9178d"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2024-01-17 09:56:13.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-03-05 16:52:46.000000000","tz":0},"subject":"Configure missing nova services to expose vendordata over configdrive","message":"Configure missing nova services to expose vendordata over configdrive\n\nCloses-Bug: #2049607\n\nChange-Id: I14ae2be2e19ad06e3190e2e948bac7ce77e80d4b\n(cherry picked from commit 0376f9dd8d03cf0fbae49156bb2d81ed32c03302)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/7abaa51c9cc3e12797b4f53750f64e53874be063"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/7abaa51c9cc3e12797b4f53750f64e53874be063"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
