)]}'
{"id":"openstack%2Fkolla-ansible~915533","triplet_id":"openstack%2Fkolla-ansible~stable%2F2023.1~I20dbaef29b2ef2d6ceffc21c156c6fa4b5e8d205","project":"openstack/kolla-ansible","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-04-15 14:46:54.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"I20dbaef29b2ef2d6ceffc21c156c6fa4b5e8d205","subject":"ironic: disable heartbeat_in_pthreads","status":"MERGED","created":"2024-04-12 18:51:58.000000000","updated":"2024-04-15 16:29:30.000000000","submitted":"2024-04-15 16:28:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"915533","cherry_pick_of_change":914107,"cherry_pick_of_patch_set":1,"meta_rev_id":"086b1916db6d80370fdfcb1988815e7bd66ed1c5","_number":915533,"virtual_id_number":915533,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-15 16:28:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-04-15 14:46:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-04-12 22:10:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-04-15 02:30:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":1,"date":"2024-04-15 08:42:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":2,"date":"2024-04-15 14:43:14.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-04-15 14:46:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-12 20:55:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-12 22:10:25.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-04-15 02:30:38.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"},{"updated":"2024-04-15 08:42:19.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-04-15 14:43:14.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"55dd204f44c80fc72b2ade661cbaaf49e2963027","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-12 18:51:58.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"3fb60580ec69315370ab88e455dc4d953b5114ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-12 20:55:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78d2b55a9d5c467c8bdbdde359a6059b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12a29e80d25c4316b6ce8a986c61b45b : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c0a6ca27d0c4b7a9228c1f512d8bd14 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d1f38d2cdcc44f1b1e6a143f976d6e8 : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c75a407173ce4e2a805d88cc99cac94a : SUCCESS in 7m 16s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/07b94f23afe4480d93d47f8cf734e298 : SUCCESS in 6m 05s\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/2aa4a0c1c7a545c4b29531fe07bd3601 : SUCCESS in 1h 01m 12s (non-voting)\n- kolla-ansible-debian-bullseye https://zuul.opendev.org/t/openstack/build/d4e3973ac52c4716bf76433d3bd8b588 : SUCCESS in 52m 17s\n- kolla-ansible-debian-bookworm https://zuul.opendev.org/t/openstack/build/571c01cda82d48f08fcbdfce637d8daf : SUCCESS in 42m 42s\n- kolla-ansible-openeuler https://zuul.opendev.org/t/openstack/build/dfa2d65b469b41ffac1699a7b2bf42a4 : SUCCESS in 1h 05m 27s (non-voting)\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/fc967a6466b8410b89211daead92edbd : SUCCESS in 52m 26s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e6ef244e3ac045df9d0e8b99d69505f9 : SUCCESS in 38m 01s\n- kolla-ansible-rocky9-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/a23d35b6cdfa413bb91ea9b3b46706e0 : SUCCESS in 1h 03m 28s (non-voting)\n- kolla-ansible-ubuntu-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/ad72297a324b48d8b626bd83fa7cb890 : SUCCESS in 40m 25s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/6005bacab71a4989ab8cecb3c11a905a : FAILURE in 57m 46s (non-voting)\n- kolla-ansible-debian-bullseye-ironic https://zuul.opendev.org/t/openstack/build/05e6c88515d84d8bb4be4ca168162ac7 : SUCCESS in 1h 03m 09s (non-voting)\n- kolla-ansible-debian-bookworm-ironic https://zuul.opendev.org/t/openstack/build/4352f5028d3f4d70b3739202da44a318 : SUCCESS in 58m 41s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/13e4bd222b4f461e8c918f693064f329 : SUCCESS in 56m 22s (non-voting)\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f037a6e0d59140e388e49adbd584356b : SUCCESS in 1h 17m 02s\n- kolla-ansible-debian-bullseye-upgrade https://zuul.opendev.org/t/openstack/build/c4e19b324e4b4574a763fea1b8af6ce1 : SUCCESS in 1h 04m 24s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/59ab492069434423912fe6df6a5b47ea : SUCCESS in 1h 13m 47s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/51ae025374194c0197dbf7907db3cc9d : SUCCESS in 1h 28m 27s (non-voting)\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/ac0f7376e5a74e7e8aa4e4d57d773136 : SUCCESS in 1h 22m 31s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/44c200f3c21d473b9482d23bf7c3366f : SUCCESS in 1h 58m 45s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/4ca4c60723164587a6e17669b5ab1d12 : SUCCESS in 1h 29m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"187194f17e55b8141d6c294e759dece0fc5ccaf7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-04-12 22:10:25.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"033e33652c53b4eefc69fc89df00d8207d9ccd84","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2024-04-15 02:30:38.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d1f0b278b46507e2e565253924cfc37b5d5458f6","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-04-15 08:42:19.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"db3341aea79d0697cc958752320a752da66b1255","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-04-15 14:43:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"93871d4ed8b4b20cc1d0c23bf53c05478625593c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-15 14:46:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5fcfdf1fa2534d91fc6cb8122df77952be5ef66d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 14:47:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9e9c2b0ad73f21dd2bf5dda42ba4ba3b0f2af75f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 16:28:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb0659ca48e742359cacad0cc7930b34\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cfea5bc2e2b4f63b4548d944c2c8ec4 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a95c73a8320a4068a311c5ff09dc9c92 : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f5e27dbc361429e9f48fda039f6140f : SUCCESS in 4m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2f53d8188374477ab969f17ec793e808 : SUCCESS in 6m 10s\n- kolla-ansible-debian-bullseye https://zuul.opendev.org/t/openstack/build/dae8d08c115e402a9604f116c2170954 : SUCCESS in 50m 48s\n- kolla-ansible-debian-bookworm https://zuul.opendev.org/t/openstack/build/35c5cd68297f49879e1a1a446a16aaf5 : SUCCESS in 53m 46s\n- kolla-ansible-debian-bullseye-upgrade https://zuul.opendev.org/t/openstack/build/aad1a5c000234407985636a0c0970112 : SUCCESS in 1h 03m 18s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/68b515a6c6234087ad2770586e7e2ae0 : SUCCESS in 52m 38s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e75977d1e4664693af859eec1e055e54 : SUCCESS in 1h 08m 19s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/341b45f963d84247bde1cb0c657b3b70 : SUCCESS in 44m 42s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3731808ad7ff4e78b8084ea476f2707f : SUCCESS in 1h 28m 42s","accounts_in_message":[],"_revision_number":1},{"id":"7d9cf447551a16ef6e41f2bc08eb77bec19246ae","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 16:28:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"086b1916db6d80370fdfcb1988815e7bd66ed1c5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 16:29:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63639d89353240bca301400a8f619ad9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74d7b758c9f9466890bd81c6b2d56e03 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f8d383128781235329b8a08ffd81beec7c596e0e","revisions":{"f8d383128781235329b8a08ffd81beec7c596e0e":{"kind":"REWORK","_number":1,"created":"2024-04-12 18:51:58.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/33/915533/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/33/915533/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/33/915533/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/33/915533/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/33/915533/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/33/915533/1"}}},"commit":{"parents":[{"commit":"10857e95014c29d26de158b586a4a7006638ef3f","subject":"Merge \"Add conditionals for IPv6 sysctl settings\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/10857e95014c29d26de158b586a4a7006638ef3f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-03-25 15:49:40.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-12 18:51:58.000000000","tz":0},"subject":"ironic: disable heartbeat_in_pthreads","message":"ironic: disable heartbeat_in_pthreads\n\ninspector is not running as a WSGI\n\nRelated-Bug: #2054705\nChange-Id: I20dbaef29b2ef2d6ceffc21c156c6fa4b5e8d205\n(cherry picked from commit 88aa51ac36923b3522cb314be73c9b16859f0446)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/f8d383128781235329b8a08ffd81beec7c596e0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/f8d383128781235329b8a08ffd81beec7c596e0e"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
