)]}'
{"id":"openstack%2Fkolla-ansible~924321","triplet_id":"openstack%2Fkolla-ansible~stable%2F2023.2~I50ce48c4e3c645e2f3aeee4913a9bc9ee506040a","project":"openstack/kolla-ansible","branch":"stable/2023.2","topic":"fix-opensearch-stable/2023.2","attention_set":{},"removed_from_attention_set":{"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-08-13 17:18:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I50ce48c4e3c645e2f3aeee4913a9bc9ee506040a","subject":"Add ca_path for module uri in opensearch role","status":"MERGED","created":"2024-07-17 10:48:07.000000000","updated":"2024-08-13 17:19:43.000000000","submitted":"2024-08-13 17:18:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924321-fix-opensearch-stable/2023.2","cherry_pick_of_change":924197,"cherry_pick_of_patch_set":5,"meta_rev_id":"35d4732af69c1d8b57b49dba510a9143f61eb7b4","_number":924321,"virtual_id_number":924321,"owner":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-13 17:18:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-08-13 15:45:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-07-22 05:14:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-08-05 14:38:57.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-08-13 15:45:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-17 13:16:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-17 13:22:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-22 05:14:47.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-08-05 14:38:57.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2024-08-13 15:45:07.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"a8255055d05007d015014c65e15fafb24a1b6079","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-07-17 10:48:07.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b0ea3300931c906ad83644027172e294cad06b95","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 13:16:51.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d17fde2447d466ab681a88bd8c35603\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fe62dcc95bc647cf955ccca288f65e3f : FAILURE in 54m 06s\n- kolla-ansible-debian-aarch64-podman https://zuul.opendev.org/t/openstack/build/77309f381612465aa9fd867d09941508 : FAILURE in 1h 19m 25s\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/c73b6f6f0b82450a9b353d2f2e7cee7a : SUCCESS in 2h 22m 41s","accounts_in_message":[],"_revision_number":1},{"id":"ebe0aa616bad7b187ba25eb7f6211108af4c43c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 13:22:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16db54fbf2084b3ebaa134a2ecfea176\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ed4b398c33644e5acd530cfb04714f4 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e2108d293c54958a25f40c419ed576d : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/391294c2ad694dfdbbb45aac4f7d951e : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/369f32e87c484ac78696d7130b54f7de : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/277ce63a7acc4bb8a6a15af89cb3141d : SUCCESS in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f7a361e9fb10453db58825940263f3a8 : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13ae393b42f047789f93e561fbf50145 : SUCCESS in 4m 49s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1a23579e0e154fb1b57de990f9a98e7f : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e0d54bba7684d4991100cd7265b3600 : SUCCESS in 5m 21s\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/11652ce992454a18b90f977676dd6867 : SUCCESS in 58m 23s (non-voting)\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e113d36ba784436c9baa453d36c18a8b : SUCCESS in 52m 23s\n- kolla-ansible-debian-podman https://zuul.opendev.org/t/openstack/build/c268f2cd391a4eedbe0450469bcca44e : SUCCESS in 56m 50s\n- kolla-ansible-openeuler https://zuul.opendev.org/t/openstack/build/960f80320aab4eea9837ed98431b4e33 : SUCCESS in 55m 08s (non-voting)\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e985e86221414442b9bd0cef4a7402d0 : SUCCESS in 56m 06s\n- kolla-ansible-rocky9-podman https://zuul.opendev.org/t/openstack/build/25b3d7ae52bd4396ba42d7eca2d9cebe : SUCCESS in 1h 12m 07s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f81c78aa54f947e08381e46bcc239cbc : SUCCESS in 52m 40s\n- kolla-ansible-ubuntu-podman https://zuul.opendev.org/t/openstack/build/d94b268892d240659e879763befbdd62 : SUCCESS in 58m 10s\n- kolla-ansible-rocky9-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/b052285eacc645c398b68c8993ceb133 : SUCCESS in 1h 04m 58s (non-voting)\n- kolla-ansible-ubuntu-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/711acdf52ca5470bbb88d5e582f4a624 : SUCCESS in 57m 29s (non-voting)\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/848b519c46874e848b4dc4a3717b718d : SUCCESS in 1h 27m 33s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/351971194c4e487cafb1b6f2286290c1 : SUCCESS in 1h 31m 08s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f59acf1d9b804a9c9b1babd8700fb4ca : SUCCESS in 1h 14m 42s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ef58f376f1a04bfbb4589bae6a1eb56c : SUCCESS in 28m 21s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/e337dda3fdca43a0ae744de6356b5b16 : SUCCESS in 20m 20s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/1e81a0bc77fe4b168066480c2eee2476 : SUCCESS in 42m 05s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/9ddac066c8814837b410a00b1828d7e5 : SUCCESS in 35m 14s (non-voting)\n- kolla-ansible-rocky9-venus https://zuul.opendev.org/t/openstack/build/77122568178a496ab174f9b2029e2586 : SUCCESS in 26m 49s (non-voting)\n- kolla-ansible-ubuntu-venus https://zuul.opendev.org/t/openstack/build/1e3071a14ad94f8f86d2bb4a31b8e3e8 : SUCCESS in 22m 40s (non-voting)\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/3825fce97e7847b3ace28e8bbe5eccf1 : SUCCESS in 1h 34m 49s (non-voting)\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/b1c77a717c074e129c82f2b9938adfe5 : SUCCESS in 1h 30m 35s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/0eec7aa38c5140fc99165e60de8fca0a : SUCCESS in 2h 22m 53s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/4399b4d34cb04110839b89cd437b5ae5 : SUCCESS in 2h 19m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fa42cab31960df5718f84d11f04d41c72e560df4","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-07-22 05:14:47.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9ca412c49600ef6d683da9d1890ed0e21714010d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-08-05 14:38:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"398a0bb161ed71eed7bf3c157b3d7f8df9616f1e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-13 15:45:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c5b342cdd31920b968d9ff305c64fa23ca5cfa05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 15:45:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"88cdd11ccd355a5ab3d62b40e7f20f7f84a9fd0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 17:18:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81f7ce5ba0324d98b258b1e2193a2286\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bb623e1c6164be7898d1b39406e679d : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6580ef46d7a849989a593972eec39975 : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c0cc552002594396905bf8b0b1e38c91 : SUCCESS in 5m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c8bce4aba04749c1865df85fcc0a3a16 : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54b9a7253b6a49a9a3c025dd5af3515a : SUCCESS in 4m 52s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/597b155414224a9bac84f53e20e15225 : SUCCESS in 4m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc568d5368784aeab793f812c01e06b9 : SUCCESS in 3m 57s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9a4347059388459f8b001f0236e0467b : SUCCESS in 54m 46s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/108f95c741b84a10834bc6f7fabc08fe : SUCCESS in 1h 05m 57s\n- kolla-ansible-debian-podman https://zuul.opendev.org/t/openstack/build/24d0654e7bed4714a298fe641051dfc1 : SUCCESS in 53m 23s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9a3b70fa061648d8b418f0bda772f66f : SUCCESS in 1h 00m 28s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b4b2f31ce90c47fca3524b3a7428a704 : SUCCESS in 1h 14m 28s\n- kolla-ansible-rocky9-podman https://zuul.opendev.org/t/openstack/build/59b7830a2b3a4de1be03e841ee429737 : SUCCESS in 1h 05m 39s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c3badb580eda44e2b3c55e0588478df8 : SUCCESS in 45m 14s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ecef2e580110486cad96af538e9050d3 : SUCCESS in 1h 21m 24s\n- kolla-ansible-ubuntu-podman https://zuul.opendev.org/t/openstack/build/dc01bc1a131e4e37a7d92ba8b203c4eb : SUCCESS in 51m 35s","accounts_in_message":[],"_revision_number":1},{"id":"d3ed91d28b88b93a625068c332312557878fa911","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 17:18:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"35d4732af69c1d8b57b49dba510a9143f61eb7b4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 17:19:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab4531e36e974a94a3aedc4ea087fd8b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd28ef2cfdab46b38551e428794c1a4d : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d222d5fa86e490fb2e382d3b836c7e1 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0594a284c64a397f176734ddc246b7a196911b36","revisions":{"0594a284c64a397f176734ddc246b7a196911b36":{"kind":"REWORK","_number":1,"created":"2024-07-17 10:48:07.000000000","uploader":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"ref":"refs/changes/21/924321/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/924321/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/924321/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/924321/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/924321/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/924321/1"}}},"commit":{"parents":[{"commit":"54019c56f12856ad1c87fac927a02b27a413ef70","subject":"Fix prechecks for interfaces with dashes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/54019c56f12856ad1c87fac927a02b27a413ef70"}]}],"author":{"name":"Petr Slavchenkov","email":"PSlavchenkov@itkey.com","date":"2024-07-16 04:58:13.000000000","tz":0},"committer":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-07-17 10:48:07.000000000","tz":0},"subject":"Add ca_path for module uri in opensearch role","message":"Add ca_path for module uri in opensearch role\n\nFixes an deploy opensearch whith enable TLS on the internal VIP\n\nCloses-Bug: #2073224\nChange-Id: I50ce48c4e3c645e2f3aeee4913a9bc9ee506040a\n(cherry picked from commit aed500c1405b17f1991333c5fe4d2faf11b213c4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/0594a284c64a397f176734ddc246b7a196911b36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/0594a284c64a397f176734ddc246b7a196911b36"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
