)]}'
{"id":"openstack%2Fkolla-ansible~924607","triplet_id":"openstack%2Fkolla-ansible~master~I00c38a6ff87f52b373e03363ab27c89d45a4a1b4","project":"openstack/kolla-ansible","branch":"master","topic":"heat-fix","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-08-08 09:09:33.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-09-09 15:34:38.000000000","reason":"Change was abandoned"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-09-09 15:34:38.000000000","reason":"Change was abandoned"},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-07-29 14:02:21.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2024-09-03 10:07:26.000000000","reason":"\u003cGERRIT_ACCOUNT_27339\u003e replied on the change","reason_account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-09-09 15:34:38.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I00c38a6ff87f52b373e03363ab27c89d45a4a1b4","subject":"Fix heat-engine start loop","status":"ABANDONED","created":"2024-07-22 12:15:47.000000000","updated":"2024-09-09 15:34:38.000000000","total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"4b5e5c9ea59e4298175db4f7fe66cec4c247b95b","_number":924607,"virtual_id_number":924607,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-07-23 18:11:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2024-07-24 08:09:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"date":"2024-09-03 10:07:26.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"date":"2024-09-03 10:07:26.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-22 12:17:07.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-07-22 12:17:07.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-07-22 12:17:07.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-07-22 12:17:07.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-07-22 13:50:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-22 15:26:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-29 09:08:41.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"6c91131f818ab421c9d8d376712310417b3f1c1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-07-22 12:15:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ecffd95f3600e420a05c39b4e5ffe993abf59eb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-07-22 12:16:42.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b72a3b4f5b786f740dd5596967a8de586e7845ba","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-07-22 12:20:18.000000000","message":"Patch Set 2: Backport-Candidate+1 Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"ee2b9e2f8757d47335ea30134a6e2c8a08ef00fc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 13:50:02.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7839adb8a23846d98e879e399eb79d2b\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8930d964a5e64942a7207c4a9d5b90a1 : FAILURE in 54m 13s\n- kolla-ansible-debian-aarch64-podman https://zuul.opendev.org/t/openstack/build/ece8ca9f39494e36925265d75f7d8ec9 : SUCCESS in 1h 32m 55s\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/b45e78b74f91474380ad5962dda50883 : FAILURE in 1h 01m 27s","accounts_in_message":[],"_revision_number":2},{"id":"a9852d60f4c7a747008d3e4568f4ed1306aaacd5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-22 15:23:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"80587dc16919f1c6ff18a03681febb3135aad389","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 15:26:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bb45a008fd24e249d84374a2e54d5a4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5ce8708e6f74ff2b39608856e6842f4 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c367a1a34b4642f39a48d9ef3a3cf33a : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c1e9ef860c4f41febdfdd52e3236f9a5 : SUCCESS in 3m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c44fe03561a44267b5cbed3b7519375f : SUCCESS in 3m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5e6ee9ba57154ae99d5b6b9f61a3e5e5 : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4306ce5aafe4fb5bd33552f04084a8c : SUCCESS in 5m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4ff37f5a59d4416fb3c1f08ae4efa5e2 : SUCCESS in 6m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/176025d2ea5948a8beac5deb0ad9817c : SUCCESS in 4m 58s\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/55309ff0247a4899a72f0d590ea7069a : SUCCESS in 1h 09m 27s (non-voting)\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/bc1c81a95d23427a8227249626b7ee46 : SUCCESS in 49m 39s\n- kolla-ansible-debian-podman https://zuul.opendev.org/t/openstack/build/c043da27954a417f94ce4963d0bd1964 : SUCCESS in 1h 00m 29s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d2c87ae6e8c142e69bbb8bce6c5387fb : SUCCESS in 1h 00m 18s\n- kolla-ansible-rocky9-podman https://zuul.opendev.org/t/openstack/build/33b841fc552f416183c436430340bb9f : SUCCESS in 1h 02m 31s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d0dffbf181464f869351e45e196dff64 : SUCCESS in 53m 14s\n- kolla-ansible-ubuntu-podman https://zuul.opendev.org/t/openstack/build/77cdd713d1624626b511861d4f7ff35c : SUCCESS in 52m 34s\n- kolla-ansible-rocky9-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/7993d1fe1e544e33a4c1b83ffb04f171 : SUCCESS in 1h 08m 27s (non-voting)\n- kolla-ansible-ubuntu-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/1c39a676abaf481ea3e1d8e4a90f46a8 : SUCCESS in 58m 55s (non-voting)\n- kolla-ansible-rocky9-scenario-nfv https://zuul.opendev.org/t/openstack/build/5df15c88a00447a6b63eb84b667d3a19 : FAILURE in 44m 46s (non-voting)\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/fff006c730bb425f8410252fa3d7a5e8 : SUCCESS in 1h 28m 08s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/45d1a3a32a704e558d2f03ffd417f943 : SUCCESS in 1h 11m 08s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9bd270f173a3456b869d68fe01c8a952 : SUCCESS in 1h 09m 31s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/4177afc4af3647999d9c2545bd7adbac : SUCCESS in 1h 29m 11s (non-voting)\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/a7fbc176c7b849a98c1a8d5e8f1a6563 : SUCCESS in 1h 19m 29s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/ce5b908e47dc4d00b97cadea5cb6fbcc : TIMED_OUT in 3h 03m 54s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d916c8c1842d4d6ead7006f67e158611 : SUCCESS in 2h 16m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2083007e2650f3431badfbb756a70325799b7a5f","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-07-22 15:34:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"914d15534b1c9465d34f17bb049f8e9bb17f4fa6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-07-23 14:56:38.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent tree, and commit message as Patch Set 2.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"35e6ff61046b27d33009d531285cc00f0e006bfa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 16:31:32.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6b75fd900a4471dbec180c3ef16d5d2\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/32f314066cd34fb98f2badfecda2737d : FAILURE in 48m 53s\n- kolla-ansible-debian-aarch64-podman https://zuul.opendev.org/t/openstack/build/51acb0fbac994c6b9076a7845162b765 : FAILURE in 50m 56s\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/cbcb45afc2b248698e27ec2227abfc7c : SUCCESS in 1h 33m 02s","accounts_in_message":[],"_revision_number":3},{"id":"8992cea4dfb14add6e6c1de4ce83a626cf1c2818","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 18:11:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ea2c505fde748aea9c233d97f8e0156\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04ff9ab10d72448780449eddfa7c57ca : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a74e09afa99543c98a5bcd7090036a60 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0502a71a1037443c9595781b0534eaab : SUCCESS in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5992bfe302144b06b205520a1d1fa792 : SUCCESS in 6m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/972fc0c951b94217991fd68484825fec : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2abc6d2d36f04ea2b79d17a966f1c5f0 : SUCCESS in 6m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4e70a7d7d7894c428365067bc950b730 : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1613ae657a6b4d78be4c6ba3c00a2179 : SUCCESS in 4m 43s\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/6a6124b4dc4d4a1296289e4e2a5eea00 : SUCCESS in 1h 08m 23s (non-voting)\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/7b4252d0494c4ab6810b12e11deb52fa : SUCCESS in 47m 51s\n- kolla-ansible-debian-podman https://zuul.opendev.org/t/openstack/build/642c6b0c7130401d813e7669043f6736 : SUCCESS in 1h 00m 25s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/1f46222b41d74f73b6fb037a81008081 : SUCCESS in 1h 00m 31s\n- kolla-ansible-rocky9-podman https://zuul.opendev.org/t/openstack/build/de75b1a06d39466ba8b9eae37653acd9 : SUCCESS in 1h 11m 18s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/3cea50f0b2e5454497c7387030558489 : SUCCESS in 47m 30s\n- kolla-ansible-ubuntu-podman https://zuul.opendev.org/t/openstack/build/3e2599edb3764d5ab3eb84592d7cab55 : SUCCESS in 55m 54s\n- kolla-ansible-rocky9-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/4d3ba5fcac894990a5ef2f58e77b6188 : SUCCESS in 1h 04m 32s (non-voting)\n- kolla-ansible-ubuntu-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/ddd53add599d47b592a5cff82e7b0480 : SUCCESS in 51m 18s (non-voting)\n- kolla-ansible-rocky9-scenario-nfv https://zuul.opendev.org/t/openstack/build/13bbab615e3b4dfcb35690e8b3f7b536 : FAILURE in 50m 04s (non-voting)\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/21a2ec1bd2cf479ca241f01785257fbe : SUCCESS in 1h 16m 39s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ce0d81d755124f9582d35e555165200c : SUCCESS in 1h 16m 02s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/78b9689d8f52475db02a62e215293939 : SUCCESS in 1h 10m 03s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/33616573f48949c8953030f49107a943 : SUCCESS in 1h 48m 10s (non-voting)\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/3e990d320d34443c8a76c2d0e3f232d0 : SUCCESS in 1h 21m 45s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f402d8f0c1b9490d881d2083298abef3 : TIMED_OUT in 3h 03m 40s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/9e9cf78c604f45aab68985f76dcec61c : FAILURE in 44m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0e3cb89077ffe713bb1ffffe7478262b64f35890","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-24 07:58:54.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a481e5dbc88a1769718e776cd605c637f2c981fa","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-24 08:09:40.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"adeb6618a4351c1517232fefbb6ce70389bd9b34","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-07-29 14:02:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"73020b65990e05931e436bdf973fc41451e55dd4","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-03 10:07:26.000000000","message":"Patch Set 3: -Backport-Candidate -Review-Priority","accounts_in_message":[],"_revision_number":3},{"id":"4b5e5c9ea59e4298175db4f7fe66cec4c247b95b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-09 15:34:38.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ec67e5a09b391246c469e83d31895f85ce2b210d","revisions":{"a9edc295613d268f684684b4f40810fab93c5653":{"kind":"REWORK","_number":1,"created":"2024-07-22 12:15:47.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/07/924607/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/07/924607/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/07/924607/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/07/924607/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/07/924607/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/07/924607/1"}}},"commit":{"parents":[{"commit":"fc44651d6297b760e5e39bf6635a0cc22e341ce9","subject":"Merge \"Fix mariadb recovery\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/fc44651d6297b760e5e39bf6635a0cc22e341ce9"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-07-22 12:10:48.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-07-22 12:15:45.000000000","tz":120},"subject":"Fix heat-engine start loop","message":"Fix heat-engine start loop\n\nThis patch fixes heat config, adds\noslo.concurrency\u0027s lock_path.\n\nCloses-Bug: #2073760\nChange-Id: I00c38a6ff87f52b373e03363ab27c89d45a4a1b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/a9edc295613d268f684684b4f40810fab93c5653"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/a9edc295613d268f684684b4f40810fab93c5653"}]},"branch":"refs/heads/master"},"35e0e7284c59d64a463179dc48690a732b515e69":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-07-22 12:16:42.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/07/924607/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/07/924607/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/07/924607/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/07/924607/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/07/924607/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/07/924607/2"}}},"commit":{"parents":[{"commit":"fc44651d6297b760e5e39bf6635a0cc22e341ce9","subject":"Merge \"Fix mariadb recovery\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/fc44651d6297b760e5e39bf6635a0cc22e341ce9"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-07-22 12:10:48.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-07-22 12:16:13.000000000","tz":120},"subject":"Fix heat-engine start loop","message":"Fix heat-engine start loop\n\nThis patch fixes heat\u0027s config by adding\noslo.concurrency\u0027s lock_path config option.\n\nCloses-Bug: #2073760\nChange-Id: I00c38a6ff87f52b373e03363ab27c89d45a4a1b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/35e0e7284c59d64a463179dc48690a732b515e69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/35e0e7284c59d64a463179dc48690a732b515e69"}]},"branch":"refs/heads/master"},"ec67e5a09b391246c469e83d31895f85ce2b210d":{"kind":"NO_CHANGE","_number":3,"created":"2024-07-23 14:56:38.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/07/924607/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/07/924607/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/07/924607/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/07/924607/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/07/924607/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/07/924607/3"}}},"commit":{"parents":[{"commit":"fc44651d6297b760e5e39bf6635a0cc22e341ce9","subject":"Merge \"Fix mariadb recovery\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/fc44651d6297b760e5e39bf6635a0cc22e341ce9"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-07-22 12:10:48.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-07-23 14:55:27.000000000","tz":120},"subject":"Fix heat-engine start loop","message":"Fix heat-engine start loop\n\nThis patch fixes heat\u0027s config by adding\noslo.concurrency\u0027s lock_path config option.\n\nCloses-Bug: #2073760\nChange-Id: I00c38a6ff87f52b373e03363ab27c89d45a4a1b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ec67e5a09b391246c469e83d31895f85ce2b210d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ec67e5a09b391246c469e83d31895f85ce2b210d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
