)]}'
{"id":"openstack%2Fkolla-ansible~926849","triplet_id":"openstack%2Fkolla-ansible~master~I25af4df0476b53b6bba6433c9abb88718eebc45a","project":"openstack/kolla-ansible","branch":"master","attention_set":{},"removed_from_attention_set":{"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2024-08-21 23:52:01.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I25af4df0476b53b6bba6433c9abb88718eebc45a","subject":"[DNM] Unify of cron and crond commands","status":"ABANDONED","created":"2024-08-21 20:23:23.000000000","updated":"2024-08-21 23:52:01.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7347a2fb8143cf5a1da00ad4317e478b76da88dc","_number":926849,"virtual_id_number":926849,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-08-21 22:15:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-21 22:15:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a8baa70dd0b16703646bffbe65cb69dbdd9e6270","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-08-21 20:23:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"041810c1fd810ea9faf933da6dc9361c362b7684","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-08-21 20:29:21.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"263d695273c36c5ec69c9102c6027df5a3191019","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-08-21 20:29:36.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"89bd6e2ccaa9f420d218b574409a22e1ca31af6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-08-21 20:29:55.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"6cbc419e330c3dee28ff6ea02ccabebb9701e393","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 22:15:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f9e0c76b891840f8b2e6747a9fbbdfa1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8774bce493b34f6b9d5ed3ab0f49fb0b : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a953985f2b04394bf3e362e2d1f5f53 : SUCCESS in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32a979d483654eb8b3fdaf3e43f05565 : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ef4d741591e740709ffd67f82dbd6899 : SUCCESS in 3m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dbf3942fb4a24191a8ff72719fa20bcc : SUCCESS in 3m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca07950b114c4f76a606b05c2e5bbe5b : SUCCESS in 6m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c2ad5295e3b742bab8db29c6cd7c5a23 : SUCCESS in 7m 44s\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8a22876d0db84e62b9d0aaf27edcbd8b : SUCCESS in 1h 11m 16s (non-voting)\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/173c3acfe82b4f6baa2d8bc294cc0cbc : FAILURE in 20m 31s\n- kolla-ansible-debian-podman https://zuul.opendev.org/t/openstack/build/6ead30abffd34f5287fdcdc8eb9348ec : FAILURE in 21m 19s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9a023cfe6df2470ba2556ee1922b98ce : SUCCESS in 1h 07m 10s\n- kolla-ansible-rocky9-podman https://zuul.opendev.org/t/openstack/build/d006c29c88be4884ac175b80eabcc372 : SUCCESS in 1h 20m 03s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/76eec03588e541ad9ac0306a17c68913 : FAILURE in 20m 32s\n- kolla-ansible-ubuntu-podman https://zuul.opendev.org/t/openstack/build/74fdc38c2b00474d96a131b5e38e0670 : FAILURE in 18m 06s\n- kolla-ansible-rocky9-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/f0249bb85bbe46dca877b58d12f04c52 : SUCCESS in 1h 09m 32s (non-voting)\n- kolla-ansible-ubuntu-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/5c43df6dea564e39b9cbcc9af0d2e521 : FAILURE in 27m 35s (non-voting)\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/43225f76e0d74c3299c12a05180c4c64 : SUCCESS in 1h 23m 16s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d1cfe26d56ad4d3f9027575e33021006 : FAILURE in 1h 18m 06s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/42fd98c686ac4145884a2cb3163b93c7 : FAILURE in 1h 14m 50s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/b64bbbaf1acc4837a5965b3c8b633c86 : SUCCESS in 29m 26s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/318f3ddf6d3348cc8c4463320a76892a : FAILURE in 24m 15s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/c9d1fc9f037245e6bfd7578eb1bd69dd : SUCCESS in 43m 09s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/65c56f63ebce4959831544ddba63d537 : FAILURE in 37m 49s (non-voting)\n- kolla-ansible-rocky9-venus https://zuul.opendev.org/t/openstack/build/d1410475a0ad4360aca1ab3bc9c7227a : SUCCESS in 30m 58s (non-voting)\n- kolla-ansible-ubuntu-venus https://zuul.opendev.org/t/openstack/build/4c4f7f26196a4d938cc11941280f68e1 : FAILURE in 19m 28s (non-voting)\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/403623d9a47548eb882d07a7ca25f72d : SUCCESS in 1h 35m 01s (non-voting)\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/3425b936f6f64a938036c8e7826e6003 : FAILURE in 32m 13s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/9ae6d901836745df9139fe8bf85d3177 : FAILURE in 1h 13m 04s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/a7c7dd6e66fb46cdbfddd79dce977512 : FAILURE in 1h 44m 38s (non-voting)\n- kolla-ansible-ubuntu-lets-encrypt https://zuul.opendev.org/t/openstack/build/43e123ea649343d5be7d4f0040c104db : FAILURE in 30m 38s (non-voting)\n- kolla-ansible-rocky9-lets-encrypt https://zuul.opendev.org/t/openstack/build/3d16bcdb87054f248facb6b863799d90 : SUCCESS in 1h 12m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7347a2fb8143cf5a1da00ad4317e478b76da88dc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-08-21 23:52:01.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3131368d4df377fa492a9b4a4138fffdd543ac72","revisions":{"d7b1e4409c088dbcbcc84378b3dab75edc65d00a":{"kind":"REWORK","_number":1,"created":"2024-08-21 20:23:23.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/49/926849/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/49/926849/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/49/926849/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/49/926849/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/49/926849/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/49/926849/1"}}},"commit":{"parents":[{"commit":"4685b972d84f277078985c5d72a057f13d2f6bb9","subject":"Remove distro_python_version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/4685b972d84f277078985c5d72a057f13d2f6bb9"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-08-21 15:48:50.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-08-21 20:23:02.000000000","tz":120},"subject":"Unify of cron and crond commands","message":"Unify of cron and crond commands\n\nChange-Id: I25af4df0476b53b6bba6433c9abb88718eebc45a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/d7b1e4409c088dbcbcc84378b3dab75edc65d00a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/d7b1e4409c088dbcbcc84378b3dab75edc65d00a"}]},"branch":"refs/heads/master"},"3131368d4df377fa492a9b4a4138fffdd543ac72":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-08-21 20:29:55.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/49/926849/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/49/926849/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/49/926849/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/49/926849/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/49/926849/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/49/926849/2"}}},"commit":{"parents":[{"commit":"4685b972d84f277078985c5d72a057f13d2f6bb9","subject":"Remove distro_python_version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/4685b972d84f277078985c5d72a057f13d2f6bb9"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-08-21 15:48:50.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-08-21 20:29:40.000000000","tz":120},"subject":"[DNM] Unify of cron and crond commands","message":"[DNM] Unify of cron and crond commands\n\nChange-Id: I25af4df0476b53b6bba6433c9abb88718eebc45a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/3131368d4df377fa492a9b4a4138fffdd543ac72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/3131368d4df377fa492a9b4a4138fffdd543ac72"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
