)]}'
{"id":"openstack%2Fkolla-ansible~934021","triplet_id":"openstack%2Fkolla-ansible~stable%2F2024.2~I548b22b27dff111d625361835029354557d8c9ca","project":"openstack/kolla-ansible","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-11-04 13:05:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I548b22b27dff111d625361835029354557d8c9ca","subject":"CI: Add pre/run.yml to files in all base jobs","status":"MERGED","created":"2024-11-04 09:19:18.000000000","updated":"2024-11-04 13:08:04.000000000","submitted":"2024-11-04 13:05:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"934021","cherry_pick_of_change":933846,"cherry_pick_of_patch_set":4,"meta_rev_id":"4c2e568653c477f6df8a95407f5977fe56a2e12d","_number":934021,"virtual_id_number":934021,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-04 13:05:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-11-04 10:04:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-11-04 11:06:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-11-04 09:55:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-11-04 10:04:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-04 09:55:14.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2024-11-04 11:06:53.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-11-04 11:35:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ac41340bd22502b0f810fe2dbf12c1079263f3f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-04 09:19:18.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"82440a733b997dfcf6850cec2658f28359f6027f","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-11-04 09:55:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4b2d57ca091cd9154dd6d3727597ad0ecd845c9d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-04 10:04:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3b79b93f09d603157a55184e01a19b13edfdd458","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-11-04 11:06:53.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f9d1f041d3e096f753630b3506e00bdf687a31b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 11:35:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eba0fc33587f47b5923ba44b2e73f53e\n\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/923329196dd34610aeb3dc5a065ee0c1 : SUCCESS in 2m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5dcdb29988d94d43bfcf3710a191a796 : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e18251c4edb4687adc5e5309aa0ef32 : SUCCESS in 3m 17s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6d8a97c69e804fefbf050a478f12e65a : SUCCESS in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7f50038ffaa4796827af187907f28e7 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f4929e815d5459eb1c03b84a14397ed : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fb604f18806b420db50c379191da5ce2 : SUCCESS in 2m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cabe2c59fa9c4ae4941b68bb548eff44 : SUCCESS in 2m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/87511d1d00744cb888d70a4a1c12e741 : SUCCESS in 3m 22s\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/5cb07251789343ecbdbcfcaceed4584e : SUCCESS in 1h 05m 58s (non-voting)\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ca6f8f11b94f404eaeae9c19c6920057 : SUCCESS in 49m 44s\n- kolla-ansible-debian-podman https://zuul.opendev.org/t/openstack/build/c55eca0b22d9436184552af095c64ba5 : SUCCESS in 54m 52s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9581e808a2f647c3905bc90b6d7c3382 : SUCCESS in 57m 43s\n- kolla-ansible-rocky9-podman https://zuul.opendev.org/t/openstack/build/41ae0076752e4349b09dc56c8e0181d3 : SUCCESS in 1h 03m 11s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/fdda1edf5baa4a75a794f3489e0053ed : SUCCESS in 54m 29s\n- kolla-ansible-ubuntu-podman https://zuul.opendev.org/t/openstack/build/dd0e9eca7fea463e8623691d81b98674 : SUCCESS in 1h 02m 17s\n- kolla-ansible-rocky9-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/f239858c11514fb2ae12989a1050a3f4 : SUCCESS in 55m 30s (non-voting)\n- kolla-ansible-ubuntu-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/e31911db8a904c56a2466a4825db3a25 : SUCCESS in 54m 47s (non-voting)\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/43ee745c80f9410eb0daa7485d20af1f : SUCCESS in 1h 10m 44s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a0bbffacc9a745ab9bc873fa2101b5f9 : SUCCESS in 1h 26m 20s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3b2b574f696f42a0be1c97404c08d0cf : SUCCESS in 1h 04m 04s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/c1b321df119747cf99fbfd56c8d54115 : SUCCESS in 1h 28m 08s (non-voting)\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/7df9c57f9f9148b3ae4934184b7f6a0c : SUCCESS in 1h 22m 32s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f510cfb275714b98b66cd1f0167a33af : SUCCESS in 2h 07m 44s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/97520a05ff4341538344fcd2c07c989f : SUCCESS in 2h 03m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"61eec172549e952226e70fd9499df141e5f9ac62","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 11:35:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e4f54b58f4c4726b16747ec69bb9776\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d799614546164050b54a5d3c1aaca611 : SUCCESS in 1h 54m 15s (non-voting)\n- kolla-ansible-debian-aarch64-podman https://zuul.opendev.org/t/openstack/build/fae7650a2ea14c82961be464848779a3 : SUCCESS in 2h 03m 44s (non-voting)\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/0c06ecf67cbd4b41a2dec56784059cf9 : SUCCESS in 1h 59m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a80c4a09584a821c036675cc9864a6aa8955f5e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 11:36:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"294c9569662c3d545a162e9139caf4e8d528018e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 13:05:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39c16fcb167544c894fa4178ffbe52f5\n\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/97c885405be64335ad09ae0ffe3597dd : SUCCESS in 2m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/200f04bd38b24b928c28dd268f3ded11 : SUCCESS in 4m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/54c6043163e8456b9b70170805c6c201 : SUCCESS in 4m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a75ea99c68e74a68b783a67743f411e6 : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7c8dbd2601984d9091b6a0553f04e53b : SUCCESS in 2m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/53b18e9197ff497d976d1519e5474e51 : SUCCESS in 3m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/717f0f83aa734151bba85ad259fc9af6 : SUCCESS in 3m 02s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9cfa2ee899fc4bcd8038dd6a0aa37dc0 : SUCCESS in 45m 13s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/683d27ec5a3a4ae1bb89358ab1dfd48d : SUCCESS in 1h 18m 36s\n- kolla-ansible-debian-podman https://zuul.opendev.org/t/openstack/build/172ad54ce8dc439d93dab8bcd5a0a177 : SUCCESS in 1h 03m 29s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9107003649c949c394cf0c41e54ba5b4 : SUCCESS in 55m 04s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/17b2754f719b4b22845d7cacd5b07638 : SUCCESS in 1h 06m 09s\n- kolla-ansible-rocky9-podman https://zuul.opendev.org/t/openstack/build/312cb67eb1ca4e77bbcfee4b4cbbf3a1 : SUCCESS in 1h 07m 25s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e2cb0a6ffd304b05b215af79238f476a : SUCCESS in 52m 07s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d54142ac266746438bb418bbc92ba80d : SUCCESS in 1h 09m 24s\n- kolla-ansible-ubuntu-podman https://zuul.opendev.org/t/openstack/build/13421ab198c7401aa4e0817d1855dcea : SUCCESS in 57m 24s","accounts_in_message":[],"_revision_number":1},{"id":"007cd559802971a66bb760de6fd0e2d812c3329e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 13:05:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4c2e568653c477f6df8a95407f5977fe56a2e12d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 13:08:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/537325481871469e8779e1463062ac7b\n\n- promote-deploy-guide https://zuul.opendev.org/t/openstack/build/23b9b10e4aa04f19893b61241b8a443f : SUCCESS in 43s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e3cb8af4d094a5880eee66cda33168e : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8808c6f8aec54fdba881b5c0b21f0781 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e76ae4199297a73109c111f67ca2b4d9c4996267","revisions":{"e76ae4199297a73109c111f67ca2b4d9c4996267":{"kind":"REWORK","_number":1,"created":"2024-11-04 09:19:18.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/934021/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/934021/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/934021/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/934021/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/934021/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/934021/1"}}},"commit":{"parents":[{"commit":"8e3995f65a30528a9adfa16bdd82447d51b26ffe","subject":"Fix destroy command with new python CLI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/8e3995f65a30528a9adfa16bdd82447d51b26ffe"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-31 10:38:35.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-11-04 09:19:18.000000000","tz":0},"subject":"CI: Add pre/run.yml to files in all base jobs","message":"CI: Add pre/run.yml to files in all base jobs\n\nRework the base jobs structure to include a mid-level\nkolla-ansible-scenario-base that includes common\nfiles: stanza.\n\nChange-Id: I548b22b27dff111d625361835029354557d8c9ca\n(cherry picked from commit 04873199efb940552da5507eaa75262114774857)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/e76ae4199297a73109c111f67ca2b4d9c4996267"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/e76ae4199297a73109c111f67ca2b4d9c4996267"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
