)]}'
{"id":"openstack%2Fkolla-ansible~937607","triplet_id":"openstack%2Fkolla-ansible~master~I316f6fc364bcff58c1f3ea1d884ec52de6e1675f","project":"openstack/kolla-ansible","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-12-12 15:01:33.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-12-12 17:53:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I316f6fc364bcff58c1f3ea1d884ec52de6e1675f","subject":"Remove tacker demos from contrib and their usage in nfv CI","status":"MERGED","created":"2024-12-12 10:32:50.000000000","updated":"2024-12-12 17:54:51.000000000","submitted":"2024-12-12 17:53:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937607","meta_rev_id":"3eccfe903e303a2da374b875071c3475ec395026","_number":937607,"virtual_id_number":937607,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-12 17:53:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-12-12 15:01:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-12-12 13:08:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-12-12 15:01:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2024-12-12 15:01:33.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-12 10:35:37.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-12-12 12:17:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-12-12 12:52:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-12 13:08:07.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"e7631fa7f01bdae8de7038ccf4c31cacc0aca031","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-12-12 10:32:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ddc479940112185c71fab79326222fc18bcd8779","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-12-12 10:35:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c2cd6110e4ad5d0914d07cd93033c1ed8b1a494e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 12:17:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0a8e34664b44199a93af95961bcfe4d\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5b086482a20b4c9c80b637a7e1342221 : SUCCESS in 1h 10m 50s (non-voting)\n- kolla-ansible-debian-aarch64-podman https://zuul.opendev.org/t/openstack/build/dfe9025078fa4d7980c693728fab5c0e : SUCCESS in 1h 42m 25s (non-voting)\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/b22bfae92d9b43c089e678e5181b5a67 : FAILURE in 1h 20m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c310bcabf988c4f74b0eec0d528d68779f8e87e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 12:52:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88e5350db33b43128ad3b0182193fe06\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d77f16673dcc4614bca3711136ca4839 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41a1888c0583403aa624a94fe6826c7f : SUCCESS in 7m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e1780c13f4bf45ef818a7124e9ef29fd : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1253f25f0d92478890e3c6307086944c : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a9f8901761384bcb81741894bf0f36fd : SUCCESS in 3m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/375e52578f4c4c99b1096ddd68e17f39 : SUCCESS in 4m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/832f9e2ad4da402291ceb574e5614efe : SUCCESS in 2m 23s\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/95e7dcb1bb294d0ca29d6d3bb858e812 : SUCCESS in 1h 07m 24s (non-voting)\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c048ec981afa42678c349d352e97b99c : SUCCESS in 47m 51s\n- kolla-ansible-debian-podman https://zuul.opendev.org/t/openstack/build/87494ce4945c40e0857028046e8bf756 : SUCCESS in 57m 08s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/011b8fbf4bb649d89c9e52864ec186a1 : SUCCESS in 52m 35s\n- kolla-ansible-rocky9-podman https://zuul.opendev.org/t/openstack/build/1426f1b943804b358d56868d27acfda5 : SUCCESS in 1h 05m 37s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1d96028d48db4bfe9c156b52a391a016 : SUCCESS in 53m 11s\n- kolla-ansible-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ab4f6dfd14314d76ac4aea714e4a4ea1 : SUCCESS in 1h 13m 22s\n- kolla-ansible-rocky9-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/f0a14ee130aa48459454011629763f28 : SUCCESS in 34m 08s (non-voting)\n- kolla-ansible-ubuntu-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/a7c7ae911cf045ccbeee719b2923196b : SUCCESS in 58m 14s (non-voting)\n- kolla-ansible-rocky9-scenario-nfv https://zuul.opendev.org/t/openstack/build/47cf876a1b7c4d498ddbfc75c4104c65 : FAILURE in 50m 10s (non-voting)\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c5f06d8555494f259e5347c17fdbe61c : SUCCESS in 1h 12m 15s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ed13f98de4c84ca7b59b1dc4667d62f1 : SUCCESS in 1h 24m 48s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8bab2a85c2af40379902a0572d1e8e62 : SUCCESS in 1h 19m 24s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/51e9c914b1df4e8c8087c449a5b60e4f : SUCCESS in 1h 24m 39s (non-voting)\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/08a5738b92fe4cbf905d189d255cc917 : SUCCESS in 1h 26m 06s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/268031df302b463faed45b8493f55159 : SUCCESS in 1h 17m 57s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/49adcb7158fe48109e1910ee8ca6def1 : SUCCESS in 2h 14m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0067e6ff5010426982946129ca9616588c7ce550","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-12-12 13:04:40.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"dda5c3feff91248a7f1be05b78506b6f9cdf0f45","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-12-12 13:08:07.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e84bf7dfe3aaa7b6b323755fafd2ad9cf96b2223","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-12-12 15:01:33.000000000","message":"Patch Set 2: Backport-Candidate-2 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f912223e78c773cd12b1d3cde8cf1eee4c29f828","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 15:15:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20d758840fb84d1ba4b80f09ca801e81\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d03af5c15db044f8a5add8cb1613ee51 : SUCCESS in 1h 06m 30s (non-voting)\n- kolla-ansible-debian-aarch64-podman https://zuul.opendev.org/t/openstack/build/feb91d3e455a43bd8566ecfabb5d923a : SUCCESS in 2h 09m 16s (non-voting)\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/6ad28c28ed5c431ea05d3f816dac5c79 : SUCCESS in 1h 52m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d225afa4b6b139bd29a49fadf0ebe3849a3a51fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 16:33:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da3bdbdddbb041e7a370bba283fc809d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/340a4cd9f0a74f61a99e4b6ae8f84623 : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d19637c386444b8b5daaf8dc52d765b : SUCCESS in 4m 40s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c27b118ab96d4cdc84946ee28390554b : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8e4b7ba08fe402d8d3e7eaf1ce6668a : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3cd36b7f63a7469e8998cd6e5eb46cf6 : SUCCESS in 2m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aa6ce74bec814c2dbc90df640e299b49 : SUCCESS in 2m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a1536dec5a254f2bbfa0c25761c5896b : SUCCESS in 3m 39s\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/bcffbf52fc6d428f9096fd48898b3d78 : SUCCESS in 1h 09m 46s (non-voting)\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f25d967a65f7427691a1a11491f67100 : SUCCESS in 27m 36s\n- kolla-ansible-debian-podman https://zuul.opendev.org/t/openstack/build/c24928e48ceb46cb8f92bc5c491c1264 : SUCCESS in 1h 00m 03s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/1e9698900ca34ed88176870ba1cf285a : SUCCESS in 55m 35s\n- kolla-ansible-rocky9-podman https://zuul.opendev.org/t/openstack/build/60f8d76291754a048d3b174015c6af65 : SUCCESS in 1h 04m 10s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b596cdca72324c618d4d9dcda3bcbbdb : SUCCESS in 58m 36s\n- kolla-ansible-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6b422aa6ceac4b6c9ea75a3a1d25c4f6 : SUCCESS in 1h 19m 01s\n- kolla-ansible-rocky9-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/f4dd7dd718f34c8e97a1445401766b56 : SUCCESS in 56m 20s (non-voting)\n- kolla-ansible-ubuntu-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/483a2717398d41b8b44b6a05e1376455 : FAILURE in 44m 22s (non-voting)\n- kolla-ansible-rocky9-scenario-nfv https://zuul.opendev.org/t/openstack/build/88faa8c55d314c1681a9ec2ac5d00599 : SUCCESS in 1h 21m 18s (non-voting)\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b5b9830abaf0409ea204f0641840ccb6 : SUCCESS in 1h 15m 34s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4462e84a175d49359e17f1e325cf344b : SUCCESS in 1h 12m 35s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7c225cf547af4811aebd952e99fb24d9 : SUCCESS in 1h 10m 35s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/ef19f69704994d8fad0f9113d74fc8c3 : SUCCESS in 1h 41m 29s (non-voting)\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/4f0056d0fd744ae7b5c161a194814aea : SUCCESS in 1h 29m 42s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/0b3b534c41a7436a920a27ece391877f : SUCCESS in 2h 21m 19s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/0897e9e7d94b4df2a395d1bada0d6fc0 : SUCCESS in 2h 03m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a94baedd1917cf3b7a8124b82d3adabaec586ebb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 16:34:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3df62b40763ec4993296cb97f7c2d41960b53255","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 17:53:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d2bfa0f5a304f3b8b79ee0d1cf60865\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0d109c154cd455a85458e68355bd380 : SUCCESS in 5m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d7459e3c41e74c1188ebaee5465a01ba : SUCCESS in 6m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/89a7784c50c74c1cb3465997bea7b83e : SUCCESS in 2m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e062d8f6594e43cda6d692300bc4db01 : SUCCESS in 3m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/82dc25a1f3694790b312274e459ad481 : SUCCESS in 3m 04s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/235566af9e4747cba4c58f3cfdb31698 : SUCCESS in 50m 26s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c0738f5c77b940c9a797ffb21adc54b2 : SUCCESS in 43m 27s\n- kolla-ansible-debian-podman https://zuul.opendev.org/t/openstack/build/7d337c21f4ca48088e1caf8187f88b5f : SUCCESS in 1h 02m 05s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/023e81aec70b4879bcaebedf6d308c11 : SUCCESS in 51m 25s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/cabf45c4e5574ec78193a839f4d37b50 : SUCCESS in 43m 21s\n- kolla-ansible-rocky9-podman https://zuul.opendev.org/t/openstack/build/9eff5879e7924cfd8c156402312e4791 : SUCCESS in 1h 11m 00s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e6956b927537402483a7e1b354f2b82e : SUCCESS in 54m 44s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/398993cea8af407dbe832adc5c235b46 : SUCCESS in 1h 07m 05s\n- kolla-ansible-ubuntu-podman https://zuul.opendev.org/t/openstack/build/1ddda287ab0c42ed84ab4a59ab797ed8 : SUCCESS in 1h 01m 31s","accounts_in_message":[],"_revision_number":2},{"id":"2769f64f5d54c8057977ae3fd31b4189ddc2cf19","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 17:53:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"3eccfe903e303a2da374b875071c3475ec395026","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 17:54:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8fa3994e9614448b5288227916e972a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d349c83779774fbcb639cf40caf786e8 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"81a5a7cf4d7a27a7c98cc968506b9f44cda257e7","revisions":{"1e21c520c5f00487ec379557fd9d980cefbf9a81":{"kind":"REWORK","_number":1,"created":"2024-12-12 10:32:50.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/937607/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/07/937607/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/07/937607/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/07/937607/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/07/937607/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/07/937607/1"}}},"commit":{"parents":[{"commit":"e628cbddb19fa26de7a5f1d1b0027c0bc70dab38","subject":"Merge \"magnum: Add k8s_cluster_api_rockylinux to CAPI drivers list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/e628cbddb19fa26de7a5f1d1b0027c0bc70dab38"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-12-12 10:30:37.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-12-12 10:32:48.000000000","tz":60},"subject":"Remove tacker demos from contrib and their usage in nfv CI","message":"Remove tacker demos from contrib and their usage in nfv CI\n\nThese are not usable for long time, based on legacy API [1],\nCI fails with non-existent openstack client commands.\n\n[1]: https://review.opendev.org/c/openstack/python-tackerclient/+/900977\n\nChange-Id: I316f6fc364bcff58c1f3ea1d884ec52de6e1675f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1e21c520c5f00487ec379557fd9d980cefbf9a81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1e21c520c5f00487ec379557fd9d980cefbf9a81"}]},"branch":"refs/heads/master"},"81a5a7cf4d7a27a7c98cc968506b9f44cda257e7":{"kind":"REWORK","_number":2,"created":"2024-12-12 13:04:40.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/937607/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/07/937607/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/07/937607/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/07/937607/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/07/937607/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/07/937607/2"}}},"commit":{"parents":[{"commit":"e628cbddb19fa26de7a5f1d1b0027c0bc70dab38","subject":"Merge \"magnum: Add k8s_cluster_api_rockylinux to CAPI drivers list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/e628cbddb19fa26de7a5f1d1b0027c0bc70dab38"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-12-12 10:30:37.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-12-12 13:04:40.000000000","tz":0},"subject":"Remove tacker demos from contrib and their usage in nfv CI","message":"Remove tacker demos from contrib and their usage in nfv CI\n\nThese are not usable for long time, based on legacy API [1],\nCI fails with non-existent openstack client commands.\n\n[1]: https://review.opendev.org/c/openstack/python-tackerclient/+/900977\n\nChange-Id: I316f6fc364bcff58c1f3ea1d884ec52de6e1675f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/81a5a7cf4d7a27a7c98cc968506b9f44cda257e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/81a5a7cf4d7a27a7c98cc968506b9f44cda257e7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
