)]}'
{"id":"openstack%2Fkolla-ansible~938021","triplet_id":"openstack%2Fkolla-ansible~stable%2F2024.1~Ice0eb62f2bb26ca6e3ac8d02c6ea787b60408a86","project":"openstack/kolla-ansible","branch":"stable/2024.1","topic":"public-openrc-system-stable/2024.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-12-19 10:04:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ice0eb62f2bb26ca6e3ac8d02c6ea787b60408a86","subject":"Generate system scoped public-openrc","status":"MERGED","created":"2024-12-18 22:42:32.000000000","updated":"2024-12-19 10:05:13.000000000","submitted":"2024-12-19 10:04:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938021-public-openrc-system-stable/2024.1","cherry_pick_of_change":937636,"cherry_pick_of_patch_set":2,"meta_rev_id":"b91f3c95f740c8054b891c03221f9943375b0473","_number":938021,"virtual_id_number":938021,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-19 10:04:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-12-19 08:07:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-12-19 08:07:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-19 00:50:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-19 08:07:37.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"9ab9b37303bdbe9ba6e7d42f660c8232ec64f8f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-12-18 22:42:32.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1d29b0abe4691c454399a1d05f4b85ba9a93d003","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 00:50:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56e5c55cd46544e98ace4e9180e89c43\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ebd03166a1ba441da0b8d60053bce531 : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55562597bf5b45b798c66c5dd1c18f99 : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0538347236eb4ba38134b5c50936b344 : SUCCESS in 3m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/484d5615c7764481bf99cffc572682c7 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/234d263478154a1eba413b18a0682dc9 : SUCCESS in 5m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d63c95571ada4b27aa9a81611caff216 : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2254be5a8ed64e808f35f18eb1c2cc76 : SUCCESS in 3m 48s\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/44c6d5f880c1461cbf3d82da3d844fdd : SUCCESS in 49m 53s (non-voting)\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/aa670f98296e4ac5828fc2e0b86932ae : SUCCESS in 43m 18s\n- kolla-ansible-debian-podman https://zuul.opendev.org/t/openstack/build/5d8e235cdfcb4bbea8d90117ea2a641a : SUCCESS in 1h 01m 26s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/3519161f8bbb43b1a5f19b9d97c0d5ed : SUCCESS in 47m 36s\n- kolla-ansible-rocky9-podman https://zuul.opendev.org/t/openstack/build/e83ffe5f37d249318677c01b59ac9e86 : SUCCESS in 1h 12m 39s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/7de1bc59a3564d858c9a4e945f7b7b46 : SUCCESS in 42m 20s\n- kolla-ansible-ubuntu-podman https://zuul.opendev.org/t/openstack/build/cb947f46eaa8421abb6c360e42b84c77 : SUCCESS in 1h 00m 55s\n- kolla-ansible-rocky9-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/37b2ae8ef5434125bd0bf2f3be1a62e5 : SUCCESS in 1h 12m 16s (non-voting)\n- kolla-ansible-ubuntu-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/b59407a56a3e490ca41b5396ae77171d : SUCCESS in 45m 56s (non-voting)\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7ef9f32762264447bad89b1771406b0f : SUCCESS in 1h 20m 23s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e910dd261e7644de893069b527761e93 : SUCCESS in 1h 21m 59s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/4aa93cc72ef6438a8beb731037e5c94a : SUCCESS in 1h 15m 19s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/a9d478ba1c184529a8c53020c26dff68 : SUCCESS in 26m 11s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/f6ed85de2a734b038bc82ee59863b0d2 : SUCCESS in 14m 09s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/9bdf5cfd007a40d1bbeaf70831a95c86 : SUCCESS in 38m 19s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/c190b14a832b4bd99f499b5e29ea345e : SUCCESS in 1h 21m 27s (non-voting)\n- kolla-ansible-rocky9-venus https://zuul.opendev.org/t/openstack/build/876562b2de5f4340b5874ba4c76a8c11 : SUCCESS in 26m 57s (non-voting)\n- kolla-ansible-ubuntu-venus https://zuul.opendev.org/t/openstack/build/b5fa4f1af9e94384aadabe40e2f0b523 : SUCCESS in 14m 09s (non-voting)\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/b8813a0f93bf4727b0282e6e20f5ca58 : SUCCESS in 1h 39m 25s (non-voting)\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/856915ab3f5b4b53970df198e19a38c0 : SUCCESS in 47m 18s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/c00cfb58ba4a4ec6adf6afe7d87e558c : FAILURE in 8m 04s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/c9c384c707ed46dfb4993c2d2260b16b : SUCCESS in 1h 46m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1c886d76337fec7c063b5364a65eb9aa9da488ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 00:59:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e27aa81df1974784b2a9990e1d42c2ab\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/006d12de3c5f46f3b6cef88c7076852b : SUCCESS in 1h 15m 55s (non-voting)\n- kolla-ansible-debian-aarch64-podman https://zuul.opendev.org/t/openstack/build/8f50b73e7b704f28b0da08ffeb5ba32f : SUCCESS in 2h 00m 16s (non-voting)\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/c62725755e754e45842f1ac1c1cf86ca : SUCCESS in 2h 09m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"16f6ee9bb08b1ae44165d37e6c18e6cbb8984316","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-12-19 08:07:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"da7aecb15771f5dc9ad8ef8bb2243b6c52be9aca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 08:08:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"80402917baba8858c535fda2b6d3ade2c8eb6063","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 10:04:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab6b609ace6e4708b347dd9db92eb1f9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50730ee1f03d425c84acd2d28015aef0 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25a8b8fdcfa748018e947448f59b2f33 : SUCCESS in 3m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8d11253d38894ae3b4ad980b29a86736 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ad8c4a44b6240de86d52817f7d81a34 : SUCCESS in 4m 31s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1bb5243700c04c639371667ba29daf38 : SUCCESS in 6m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f97fdd3e5844a9d88d2f0341938e66e : SUCCESS in 6m 46s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/aaf3d536e6db41b09066ae90041c9c78 : SUCCESS in 1h 04m 28s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5f20e612637e4c2eb7364197a657f4b9 : SUCCESS in 1h 27m 11s\n- kolla-ansible-debian-podman https://zuul.opendev.org/t/openstack/build/f33d7635c86a4d0c912922294fe59ad8 : SUCCESS in 1h 12m 01s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5ee8067cc66447d0af856126aa7d5f1b : SUCCESS in 1h 18m 09s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/4c0d70a989cb4a64acc8db3b7ae38fc4 : SUCCESS in 32m 42s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/69d5d31607654579a9480b7c698f79c3 : SUCCESS in 49m 04s (non-voting)\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/9150d3f9af274353bca94f86f88abd8a : SUCCESS in 1h 42m 02s\n- kolla-ansible-rocky9-podman https://zuul.opendev.org/t/openstack/build/7aa21fecd1f04b64986f99c872882a6b : SUCCESS in 1h 33m 08s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/fdcb0acbf0d2425a97623ffbb357ca75 : SUCCESS in 53m 26s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/3a9fdc3b5ac6453c9ffb2a6352037d7f : SUCCESS in 29m 56s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/03dac84d2cd34063b34ede03e5a036cb : SUCCESS in 41m 41s (non-voting)\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8ffb10670f354418817bb866f836b559 : SUCCESS in 1h 16m 08s\n- kolla-ansible-ubuntu-podman https://zuul.opendev.org/t/openstack/build/633887ff360443a09e5a6d5c219a5d87 : SUCCESS in 1h 01m 30s","accounts_in_message":[],"_revision_number":1},{"id":"43440f85c16e8199aadcbf034d1c7ae514d55252","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 10:04:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b91f3c95f740c8054b891c03221f9943375b0473","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 10:05:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5d65d6ced424a87b5e806a0c6704162\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98653d9041ee454bb8508d0414581ddf : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbe0331ac0724247b3cc3a6f79e3b865 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"edc101e7aaa9d0d577402399b9c6d1738bd5d79d","revisions":{"edc101e7aaa9d0d577402399b9c6d1738bd5d79d":{"kind":"REWORK","_number":1,"created":"2024-12-18 22:42:32.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/21/938021/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/938021/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/938021/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/938021/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/938021/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/938021/1"}}},"commit":{"parents":[{"commit":"1603a28a5cc0cbb237cd2d2bfbfebdb95756e074","subject":"Merge \"Fix nova-libvirt\u0027s secrets - device or resource is busy\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1603a28a5cc0cbb237cd2d2bfbfebdb95756e074"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-12 21:24:37.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-18 22:42:32.000000000","tz":0},"subject":"Generate system scoped public-openrc","message":"Generate system scoped public-openrc\n\nIt can be useful to run Ironic commands on the public API if access to\nthe internal API is not possible.\n\nRelated-Bug: #2051837\nChange-Id: Ice0eb62f2bb26ca6e3ac8d02c6ea787b60408a86\n(cherry picked from commit 1916b3c2ad1d261537ff05f608f805e5fefd4a54)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/edc101e7aaa9d0d577402399b9c6d1738bd5d79d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/edc101e7aaa9d0d577402399b9c6d1738bd5d79d"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
