)]}'
{"id":"openstack%2Fkolla-ansible~941021","triplet_id":"openstack%2Fkolla-ansible~stable%2F2024.1~I9fb694b7d70842d676b26704cb67e54415b29f74","project":"openstack/kolla-ansible","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"36698":{"account":{"_account_id":36698,"name":"hamid lotfi","display_name":"Hamid Lotfi","email":"hamid.lotfi@gmail.com","username":"hamidlotfi"},"last_update":"2025-03-08 07:27:12.000000000","reason":"\u003cGERRIT_ACCOUNT_36698\u003e replied on the change","reason_account":{"_account_id":36698,"name":"hamid lotfi","display_name":"Hamid Lotfi","email":"hamid.lotfi@gmail.com","username":"hamidlotfi"}},"36702":{"account":{"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"},"last_update":"2025-10-09 05:57:18.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I9fb694b7d70842d676b26704cb67e54415b29f74","subject":"Change the host from \"database_address\" to \"api_interface_address\"","status":"ABANDONED","created":"2025-02-08 10:51:07.000000000","updated":"2025-10-09 05:57:18.000000000","total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"28fe5fd87b4afc168dbe2fb2dd434f9597b079b1","_number":941021,"virtual_id_number":941021,"owner":{"_account_id":36698,"name":"hamid lotfi","display_name":"Hamid Lotfi","email":"hamid.lotfi@gmail.com","username":"hamidlotfi"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-02-19 14:31:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-08 12:27:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-08 12:32:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-19 10:21:32.000000000","updated_by":{"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"},"reviewer":{"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"},"state":"REVIEWER"}],"messages":[{"id":"992736d999d3bf84ad38128f35b5dc5cbfaeca75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36698,"name":"hamid lotfi","display_name":"Hamid Lotfi","email":"hamid.lotfi@gmail.com","username":"hamidlotfi"},"date":"2025-02-08 10:51:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e21524ebe489ceccae3d1a331cae122859ba6498","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":36698,"name":"hamid lotfi","display_name":"Hamid Lotfi","email":"hamid.lotfi@gmail.com","username":"hamidlotfi"},"date":"2025-02-08 12:03:51.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"59be0d0021ade049e68fcd9b938313816cd7317d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-08 12:27:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b56d6a396d3d4dce9df3050805a916c4\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f96f4b7423b24d48b06fa70d88056f47 : SUCCESS in 57m 44s (non-voting)\n- kolla-ansible-debian-aarch64-podman https://zuul.opendev.org/t/openstack/build/69f5ac89d87b455fa1c63212309bca82 : SUCCESS in 1h 34m 09s (non-voting)\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/d47f21f6e48c4610b26cc34140162cbb : SUCCESS in 1h 22m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a109e87a113ca8fa9ca6aa9770ad91271d94fdbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-08 12:32:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49841205465c44c9b9260020b02cc378\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef8e66bbe7ad4880ac3808ade000406f : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47e096867fd9421d886223ae1a7b9534 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e12fff30ba554c929dad37c73a40a66a : SUCCESS in 2m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f6237a02ebc04698811c00f8da34ee62 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d55a003fd904a5794f3bab216951e3a : SUCCESS in 3m 12s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9539f104162e47508c003a63c73ecd45 : SUCCESS in 3m 32s\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/7e2059e116fe45fc9d81870bf735ff4a : SUCCESS in 50m 26s (non-voting)\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/65033390438349c8a0fb242e43ed88ce : SUCCESS in 28m 59s\n- kolla-ansible-debian-podman https://zuul.opendev.org/t/openstack/build/4aa356d4555a497d8c160b8c2bec4c9c : SUCCESS in 49m 27s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0373357efa9a49c2b8c7ebdba8ef6768 : SUCCESS in 33m 15s\n- kolla-ansible-rocky9-podman https://zuul.opendev.org/t/openstack/build/36be80b033cb4b629e101c18c964f874 : SUCCESS in 56m 34s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/301cb9f8e6d14a889b7ef249b1d4076e : SUCCESS in 25m 18s\n- kolla-ansible-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c012d08ed35641c4b8f30dfeada2f186 : SUCCESS in 46m 13s\n- kolla-ansible-rocky9-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/27183959a0a34f83af40b1d4b52ae357 : SUCCESS in 50m 30s (non-voting)\n- kolla-ansible-ubuntu-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/47828821e849476fa4306954de65d271 : SUCCESS in 28m 30s (non-voting)\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/cd53ea17df664fa884baf50d8b31ee8a : SUCCESS in 1h 00m 30s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b0ce355453844cee91ef08dba9d39f6e : SUCCESS in 48m 09s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d6f81f1dee104951a0dba268a8118d25 : SUCCESS in 56m 46s\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/1e242617b33345a1aa83eb136aeb4bac : SUCCESS in 14m 34s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/c0a98d03aa5a4b3f89fc4314c01b0745 : SUCCESS in 14m 20s (non-voting)\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/142142b4cd4d4b0e9c017c9addb1017e : SUCCESS in 54m 48s (non-voting)\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/7a76ae6336af4315b5ff233619fa4713 : SUCCESS in 1h 06m 14s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/fe0cd7cfa70c4b4a9cb1875d989293e3 : FAILURE in 7m 27s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/b89cfdf95d564feb8cf02427a0edf081 : SUCCESS in 1h 39m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"df16ecaa9c70d1ab2a29a036374bb029fb7236f8","author":{"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"},"date":"2025-02-19 10:21:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"260ea74eccc4892da264c56ccf81e0e5b41c026c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36698,"name":"hamid lotfi","display_name":"Hamid Lotfi","email":"hamid.lotfi@gmail.com","username":"hamidlotfi"},"date":"2025-02-19 11:53:38.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"aa7941d8f9985fe8fb1b290f2b9672ebc729090d","tag":"autogenerated:gerrit:restore","author":{"_account_id":36698,"name":"hamid lotfi","display_name":"Hamid Lotfi","email":"hamid.lotfi@gmail.com","username":"hamidlotfi"},"date":"2025-02-19 11:55:14.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"b95ea3d66c6df194231844d1866afd2cbf21f3a8","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":36698,"name":"hamid lotfi","display_name":"Hamid Lotfi","email":"hamid.lotfi@gmail.com","username":"hamidlotfi"},"date":"2025-02-19 12:02:14.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"6465a4953829b5a85806805afb5b9b73f8b0b9a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36698,"name":"hamid lotfi","display_name":"Hamid Lotfi","email":"hamid.lotfi@gmail.com","username":"hamidlotfi"},"date":"2025-02-19 12:05:47.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ca4bbe93fcf9b36ecb56a4edae50fa7791d7bdd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36698,"name":"hamid lotfi","display_name":"Hamid Lotfi","email":"hamid.lotfi@gmail.com","username":"hamidlotfi"},"date":"2025-02-19 12:09:17.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"9820a37ba6b9280521aadc9cbd345516c1591111","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 14:31:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/237a7d473a7a423a9c3efe908c40b926\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d03038e26f24f389edaa406e500eb44 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b4d16953aca40089399aa4f01b3b2bc : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d937577aa9814986a0f5107a097475cb : SUCCESS in 3m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eede95bbf10a4742af2d2b22b35ac8f8 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87b04263cc3549e59eb0adbcaf0c6041 : SUCCESS in 3m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f87288be9ab2430296810fd1f16d28f1 : FAILURE in 4m 31s\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/3b3d4679a24c42e0afd37c79f2cd2c53 : SUCCESS in 54m 23s (non-voting)\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d248386363a547af896643d3ff878b31 : SUCCESS in 51m 47s\n- kolla-ansible-debian-podman https://zuul.opendev.org/t/openstack/build/05a0419af57b43139d5bce507bae0a41 : SUCCESS in 1h 00m 42s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/484326a1e7164220b0343dc19868de4c : SUCCESS in 1h 00m 00s\n- kolla-ansible-rocky9-podman https://zuul.opendev.org/t/openstack/build/379aec689a984aeb8889988ef5b6ef7d : FAILURE in 8m 43s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c48997c2cb2c4194bfe9970a7b51e4a0 : SUCCESS in 28m 08s\n- kolla-ansible-ubuntu-podman https://zuul.opendev.org/t/openstack/build/bf91a6d8def84c25b2c105d8ff0c5cc8 : FAILURE in 7m 22s\n- kolla-ansible-rocky9-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/a9e91fa335974e91ad7223d5c1db4b15 : SUCCESS in 1h 11m 57s (non-voting)\n- kolla-ansible-ubuntu-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/d4256506092a4027b0472047246fb694 : SUCCESS in 49m 18s (non-voting)\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/1e5cab197c9d49ebb635a72b92074e9d : SUCCESS in 1h 15m 10s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1e3dd8a16bc04fa285d2a44042234968 : SUCCESS in 1h 08m 25s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/cd41913342284834adf63e13e98a1869 : SUCCESS in 40m 41s\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/34b5acfa66f14e1c8760b97924a6079f : SUCCESS in 21m 28s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/c3e555380f60480e939e75ccc603e256 : SUCCESS in 18m 43s (non-voting)\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/1dc9a894103c4adeb62d2db59338ab09 : SUCCESS in 55m 47s (non-voting)\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/817fe13133f54246b882845858dd413f : SUCCESS in 1h 27m 39s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/913a97c5224f4fd9bfa76bdf5a5fd006 : FAILURE in 8m 03s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/fd2c210979f24204b999466c41b0fbed : SUCCESS in 2h 09m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"26e64f2d5e1767b125cada6c7bb270d035022b83","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 15:19:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08b57763e17043cb9d563a9c7d48e882\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/76791449b1af46e193520806788e19b9 : SUCCESS in 1h 06m 57s (non-voting)\n- kolla-ansible-debian-aarch64-podman https://zuul.opendev.org/t/openstack/build/d2f1f1bd2eda428598a9a903b0f04a94 : FAILURE in 3h 06m 43s (non-voting)\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/009ad86faf244d7da684f1b396de2850 : SUCCESS in 1h 52m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"907b409d5da0aa861f40176caedf0cd1e4758efd","author":{"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"},"date":"2025-03-05 07:44:26.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d0d4e69e7fbfb5c9c28d833a30d8e1dd136ca541","author":{"_account_id":36698,"name":"hamid lotfi","display_name":"Hamid Lotfi","email":"hamid.lotfi@gmail.com","username":"hamidlotfi"},"date":"2025-03-08 07:27:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"28fe5fd87b4afc168dbe2fb2dd434f9597b079b1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-09 05:57:18.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bcb7067a10c3187dca9f9d8a3f1331dc77b4c49a","revisions":{"ca672bdc96398cbbdd7bf9409aaa6984fab179fa":{"kind":"REWORK","_number":1,"created":"2025-02-08 10:51:07.000000000","uploader":{"_account_id":36698,"name":"hamid lotfi","display_name":"Hamid Lotfi","email":"hamid.lotfi@gmail.com","username":"hamidlotfi"},"ref":"refs/changes/21/941021/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/941021/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/941021/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/941021/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/941021/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/941021/1"}}},"commit":{"parents":[{"commit":"c39e0e199da99156b5100aa356cce2ab819caba8","subject":"Merge \"Fix cinder etcd3gw backend_url\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c39e0e199da99156b5100aa356cce2ab819caba8"}]}],"author":{"name":"hamidlotfi","email":"hamid.lotfi@gmail.com","date":"2025-02-08 10:48:31.000000000","tz":210},"committer":{"name":"hamidlotfi","email":"hamid.lotfi@gmail.com","date":"2025-02-08 10:51:03.000000000","tz":210},"subject":"Change restart_policy from oneshot to no","message":"Change restart_policy from oneshot to no\n\nIn release 2024.1, running maria_backup shows the following error:\n\n        \"Connecting to MariaDB server host: internal.host.com, user: backup_shard_0, password: set, port: 3306, socket: not set\",\n        \"Error: Unsupported server version: \u002710.11.10-MariaDB-ubu2204-log\u0027.\"\n\nTo resolve this issue:\nChange restart_policy from oneshot to no\n\nChange-Id: I9fb694b7d70842d676b26704cb67e54415b29f74\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ca672bdc96398cbbdd7bf9409aaa6984fab179fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ca672bdc96398cbbdd7bf9409aaa6984fab179fa"}]},"branch":"refs/heads/stable/2024.1"},"ca6a8e7ffc9f232c4a0ca7a4b0953e7738222516":{"kind":"REWORK","_number":2,"created":"2025-02-19 12:05:47.000000000","uploader":{"_account_id":36698,"name":"hamid lotfi","display_name":"Hamid Lotfi","email":"hamid.lotfi@gmail.com","username":"hamidlotfi"},"ref":"refs/changes/21/941021/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/941021/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/941021/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/941021/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/941021/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/941021/2"}}},"commit":{"parents":[{"commit":"c39e0e199da99156b5100aa356cce2ab819caba8","subject":"Merge \"Fix cinder etcd3gw backend_url\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c39e0e199da99156b5100aa356cce2ab819caba8"}]}],"author":{"name":"hamidlotfi","email":"hamid.lotfi@gmail.com","date":"2025-02-08 10:48:31.000000000","tz":210},"committer":{"name":"hamid lotfi","email":"hamid.lotfi@gmail.com","date":"2025-02-19 12:05:15.000000000","tz":0},"subject":"Change the host from \"database_address\" to \"api_interface_address\"","message":"Change the host from \"database_address\" to \"api_interface_address\"\n\nIn release 2024.1, running maria_backup shows the following error:\n\n        \"Connecting to MariaDB server host: internal.host.com, user: backup_shard_0, password: set, port: 3306, socket: not set\",\n        \"Error: Unsupported server version: \u002710.11.10-MariaDB-ubu2204-log\u0027.\"\n\nTo resolve this issue:\nChange the host from \"database_address\" to \"api_interface_address\"\n\nChange-Id: I9fb694b7d70842d676b26704cb67e54415b29f74\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ca6a8e7ffc9f232c4a0ca7a4b0953e7738222516"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ca6a8e7ffc9f232c4a0ca7a4b0953e7738222516"}]},"branch":"refs/heads/stable/2024.1"},"bcb7067a10c3187dca9f9d8a3f1331dc77b4c49a":{"kind":"REWORK","_number":3,"created":"2025-02-19 12:09:17.000000000","uploader":{"_account_id":36698,"name":"hamid lotfi","display_name":"Hamid Lotfi","email":"hamid.lotfi@gmail.com","username":"hamidlotfi"},"ref":"refs/changes/21/941021/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/941021/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/941021/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/941021/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/941021/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/941021/3"}}},"commit":{"parents":[{"commit":"c39e0e199da99156b5100aa356cce2ab819caba8","subject":"Merge \"Fix cinder etcd3gw backend_url\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c39e0e199da99156b5100aa356cce2ab819caba8"}]}],"author":{"name":"hamidlotfi","email":"hamid.lotfi@gmail.com","date":"2025-02-08 10:48:31.000000000","tz":210},"committer":{"name":"hamid lotfi","email":"hamid.lotfi@gmail.com","date":"2025-02-19 12:06:32.000000000","tz":0},"subject":"Change the host from \"database_address\" to \"api_interface_address\"","message":"Change the host from \"database_address\" to \"api_interface_address\"\n\nIn release 2024.1, running maria_backup shows the following error:\n\n        \"Connecting to MariaDB server host: internal.host.com, user: backup_shard_0, password: set, port: 3306, socket: not set\",\n        \"Error: Unsupported server version: \u002710.11.10-MariaDB-ubu2204-log\u0027.\"\n\nTo resolve this issue:\nChange the host from \"database_address\" to \"api_interface_address\"\n\nChange-Id: I9fb694b7d70842d676b26704cb67e54415b29f74\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/bcb7067a10c3187dca9f9d8a3f1331dc77b4c49a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/bcb7067a10c3187dca9f9d8a3f1331dc77b4c49a"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
