)]}'
{"id":"openstack%2Fkolla-ansible~945721","triplet_id":"openstack%2Fkolla-ansible~stable%2F2024.1~I2683156ea3f241381d58830153dd77b608b6fb6a","project":"openstack/kolla-ansible","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-03-28 06:22:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2683156ea3f241381d58830153dd77b608b6fb6a","subject":"Fix octavia.conf generation after jobboard fix","status":"MERGED","created":"2025-03-27 15:23:38.000000000","updated":"2025-03-28 06:23:22.000000000","submitted":"2025-03-28 06:22:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945721","cherry_pick_of_change":945632,"cherry_pick_of_patch_set":2,"meta_rev_id":"182be8ffcf41178c21846c1d04509afb08a8850f","_number":945721,"virtual_id_number":945721,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-28 06:22:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-03-28 04:44:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2025-03-27 15:24:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-03-28 04:44:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2025-03-27 15:24:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-27 17:29:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-27 17:43:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-28 04:44:40.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"3b55677601629599e54c661df2120973bde71207","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-27 15:23:38.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"98a80ca248510b3366b142b23a13b7c20cdc5d80","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-27 15:24:11.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"c46ffc6a236bbaa1316db377962849ca6bc61c50","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 17:29:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13a21774f2e246e49828c8a76306e973\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5b9352f09b2b40f1b03a3418fa17b6fc : SUCCESS in 1h 06m 39s (non-voting)\n- kolla-ansible-debian-aarch64-podman https://zuul.opendev.org/t/openstack/build/5aa8a66d0d2a4d3b99b065bc6cc896bd : SUCCESS in 1h 25m 07s (non-voting)\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/541d976ad66244d89848cda83e7f6700 : SUCCESS in 1h 36m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"173f3d7debc9f062e0b8debdee30d5e7f54e76ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 17:43:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd210ae16b3b4689935b25d53c83b7b7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d41179eca4c4937abc399d7dc120aab : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e2c13cc1e6644139223932be0ee2e3b : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7bf574a448b403689d080f4371db517 : SUCCESS in 3m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/22d2859428cf49af8da64f580fb0fb43 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd75af15e52949d2a919652714dcbe66 : SUCCESS in 4m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bc7da8a2971b421cae66f29093b68bd7 : SUCCESS in 5m 26s\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/2d37a254948444e2bd77edda9538a9a1 : SUCCESS in 28m 59s (non-voting)\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/38aaa32c607149adb0f98899c28ef1fc : SUCCESS in 57m 55s\n- kolla-ansible-debian-podman https://zuul.opendev.org/t/openstack/build/5aac743f9c7348c199613f65df3935d8 : SUCCESS in 53m 57s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9aabd3ef536e470bb00dc86187ea2b4f : SUCCESS in 26m 16s\n- kolla-ansible-rocky9-podman https://zuul.opendev.org/t/openstack/build/711cc6e1eca34ee6bf983e48cf98dab9 : SUCCESS in 1h 17m 17s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b68847bf654e4a5dab6e8ffff4f80f51 : SUCCESS in 39m 46s\n- kolla-ansible-ubuntu-podman https://zuul.opendev.org/t/openstack/build/3ceb149e98fd4d1789eb1da6fa39fb25 : SUCCESS in 30m 57s\n- kolla-ansible-rocky9-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/1f5243aac15d42e4881d2bb8fa1de73b : SUCCESS in 59m 38s (non-voting)\n- kolla-ansible-ubuntu-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/a7fa9671f49c4b2b95067900f7953c6b : SUCCESS in 43m 22s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/5213c95734a740c98cf1c564b5d6063b : SUCCESS in 35m 29s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/60f66133890743ab95249aa5d1c5cc1d : SUCCESS in 45m 03s (non-voting)\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/bede38ce40ec44b290cc4343a119dfd2 : SUCCESS in 1h 12m 27s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6be0904875a74368901fb12bb988ac43 : SUCCESS in 1h 15m 45s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/91cf553e4ca447af92f47e1a22a7c934 : SUCCESS in 1h 07m 51s\n- kolla-ansible-debian-ovn https://zuul.opendev.org/t/openstack/build/707c225901e7454aaea1f9c8829defda : SUCCESS in 1h 10m 24s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/5be752af83f948ecbf450827a5310491 : SUCCESS in 1h 18m 47s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/8d4d30da688349f9ba1856ecb7c404e5 : SUCCESS in 54m 20s (non-voting)\n- kolla-ansible-debian-upgrade-ovn https://zuul.opendev.org/t/openstack/build/fde47a04032a47dbbeb16692dc541c54 : SUCCESS in 1h 38m 33s (non-voting)\n- kolla-ansible-rocky9-upgrade-ovn https://zuul.opendev.org/t/openstack/build/b9e9fcd1abd04d04b7096ce13ae706bb : SUCCESS in 1h 43m 43s (non-voting)\n- kolla-ansible-ubuntu-upgrade-ovn https://zuul.opendev.org/t/openstack/build/f36241e3fa66410e90f9c3d350d24478 : SUCCESS in 1h 34m 31s (non-voting)\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/72e550002ffd445dbf961e12e997babd : SUCCESS in 1h 55m 10s (non-voting)\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/d6248ac856d84ea8acf016b01946835a : SUCCESS in 40m 21s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/92f7fdd51d454795a581c56d32f1ddcf : FAILURE in 8m 50s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/0b2f4d2829a849a0966aa4e5aa169b15 : FAILURE in 2h 13m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c18775649af1d93d9975dcd4d14e75262f045dae","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-03-28 04:44:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c7f4b9bc28bc27bef83411e5d3f4ad2f07a0e6c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 04:45:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a0bf1244501157219c42fbbccd3d7bf00da056c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 06:22:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/275f06b1e4f54e1db69a72bdff6bfdba\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b688af889bed45b78cf6b039228f3d7a : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/269018b7fa094f66bb57c5197d77f574 : SUCCESS in 3m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/039a498b0d1244db8a9b26046a320106 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dd076b9a025438c9c4d509769e0e9c1 : SUCCESS in 4m 11s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e746c72f3e434d8183f5817fdd61d6c0 : SUCCESS in 5m 57s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b63b60ac4d5f4ceca9363e3c57c58672 : SUCCESS in 31m 28s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8f434c0295474fbeb71ec50e3b43c6ad : SUCCESS in 1h 23m 26s\n- kolla-ansible-debian-podman https://zuul.opendev.org/t/openstack/build/12d33c6b7b314edfaeab52cba1f7e8b6 : SUCCESS in 1h 01m 33s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/bed4acd003bc4a4caeb2f740e96d91cb : SUCCESS in 49m 32s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e9c352c101d045358b0aa95a792766ff : SUCCESS in 1h 25m 51s\n- kolla-ansible-rocky9-podman https://zuul.opendev.org/t/openstack/build/3574bd328bb74ef9a0248d294f9c6c81 : SUCCESS in 1h 01m 30s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/41f00d73d0fb41cda905973249bb26f4 : SUCCESS in 50m 05s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7d9a5cd0182f4bf9a84290ca2e23d86f : SUCCESS in 1h 10m 38s\n- kolla-ansible-ubuntu-podman https://zuul.opendev.org/t/openstack/build/38a19144091547e597919dfd28713b2c : SUCCESS in 36m 18s","accounts_in_message":[],"_revision_number":1},{"id":"13cf733f21e9843bee50c18c93fe1a3fc26115f9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 06:22:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"182be8ffcf41178c21846c1d04509afb08a8850f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 06:23:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e753a54098a94321a2bf550c189c8720\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a286d5d300544e75ac895e7e3f4925bb : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6c1d12a5ad45a48b06d1abc6f3700e8a83453c6b","revisions":{"6c1d12a5ad45a48b06d1abc6f3700e8a83453c6b":{"kind":"REWORK","_number":1,"created":"2025-03-27 15:23:38.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/945721/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/945721/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/945721/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/945721/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/945721/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/945721/1"}}},"commit":{"parents":[{"commit":"edce8d9f4ad5ce691787f49fb9e50c4d49cb16a3","subject":"CI: Fix ironic scenario image builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/edce8d9f4ad5ce691787f49fb9e50c4d49cb16a3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-03-26 16:25:47.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-03-27 15:23:38.000000000","tz":0},"subject":"Fix octavia.conf generation after jobboard fix","message":"Fix octavia.conf generation after jobboard fix\n\nAdd Jinja2 option to not delete trailing newlines in for loops\nFollowup after [1]\n\n[1] https://review.opendev.org/c/openstack/kolla-ansible/+/942799\n\n\nChange-Id: I2683156ea3f241381d58830153dd77b608b6fb6a\n(cherry picked from commit 44c2fbf622fdcc7bc1b01f50fde0d2c81df67839)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/6c1d12a5ad45a48b06d1abc6f3700e8a83453c6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/6c1d12a5ad45a48b06d1abc6f3700e8a83453c6b"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
