)]}'
{"id":"openstack%2Fkolla-ansible~966021","triplet_id":"openstack%2Fkolla-ansible~master~I0e19cf109eb5665045806b7d925779e82164c2dc","project":"openstack/kolla-ansible","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-03-27 17:37:09.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-03-27 19:17:48.000000000","reason":"Change was submitted"},"37306":{"account":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"last_update":"2026-03-27 19:17:48.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-03-27 19:17:48.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2026-03-27 19:17:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0e19cf109eb5665045806b7d925779e82164c2dc","subject":"kolla-ansible: Add RabbitMQ 4.2 support and dynamic upgrade paths","status":"MERGED","created":"2025-11-04 10:24:27.000000000","updated":"2026-03-27 20:04:45.000000000","submitted":"2026-03-27 19:17:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":21,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"966021","meta_rev_id":"b199bedcb33bc46c6a44f34efeb8b932f7a7c7d0","_number":966021,"virtual_id_number":966021,"owner":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-27 19:17:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-03-27 17:37:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-03-27 17:37:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2026-03-27 17:37:09.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-03-27 17:37:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-04 10:32:31.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-11-04 12:26:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-04 13:28:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-25 10:51:29.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-11-25 15:42:46.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-11-26 16:11:01.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2026-02-10 11:58:32.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2026-02-11 09:38:45.000000000","updated_by":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"reviewer":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"state":"CC"}],"messages":[{"id":"a36d6ee8e2ed2be7929a893bd1cf81e1650b2ad9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-11-04 10:24:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"92a28e0a60190f57d3fcde7745b7b2a9003c7e15","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-04 10:32:31.000000000","message":"Patch Set 1: Review-Priority-1 Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"31c22574c8d03a6d900a3119a21f80134432ae05","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 12:26:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5f12da7d1984863862528e0c31df615\n\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/ef9975af0f7c43b4937ba8412c55a3a4 : FAILURE in 1h 41m 47s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64 https://zuul.opendev.org/t/openstack/build/434e9c5e8f73475ab84ffa05317d7dfd : FAILURE in 1h 02m 27s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64-podman https://zuul.opendev.org/t/openstack/build/38c7b1209a37425792cecc1121f571a6 : FAILURE in 1h 04m 08s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64-upgrade https://zuul.opendev.org/t/openstack/build/de8579fcfc5046868afd0b101aa3036d : FAILURE in 1h 27m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"505eddfe992ab330dc2f97e39f01fbc5d9de356f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 13:28:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/76bfe9989ab8416896b8d868c6bdabc8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00a82b5483ed44e5a7a21a9e4bcc2d9b : SUCCESS in 4m 57s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1c8890e29a7f4da0a11423e0c7e2928b : FAILURE in 3m 54s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/47066d9da1224c7688869e53e101760c : FAILURE in 35m 55s (non-voting)\n- kolla-ansible-debian-bookworm https://zuul.opendev.org/t/openstack/build/7c3694e7946e46658a17e00cf318c7f1 : FAILURE in 42m 47s\n- kolla-ansible-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/53b43d272e214423b7fbf70755259112 : FAILURE in 46m 07s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/d16bf713999e48b1b6cf0b0c750e7ac0 : FAILURE in 46m 12s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/6b28ef3917c54829824f189c14af8b9f : FAILURE in 48m 28s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/4254e7f58d4a4529b6e4ff055dd1c792 : FAILURE in 1h 02m 22s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e5e23063231a42688957ed3ee401f53b : FAILURE in 51m 10s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/410b10cf289e4c3391a456cc91805f16 : FAILURE in 52m 50s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/75be3e747ef345619afa9f99ecff76ea : FAILURE in 53m 47s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/14c478603ba1489ab80394cc7ac56d10 : FAILURE in 1h 20m 04s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/b81815d175164741b10e567b75da29b8 : TIMED_OUT in 3h 02m 45s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/4f64d6ed08574e338f57407d6cd8ae42 : FAILURE in 1h 31m 09s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/816ea4986b9e49dfbc8c00f19debd03e : FAILURE in 1h 31m 59s (non-voting)\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/d189736255f54ec88f934414e5c6483b : SUCCESS in 49m 39s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10fcc26daa1a432f9bc09ba974d60a38 : SUCCESS in 48m 10s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/53e69c12a99d4322a98fcec4bfcbafc6 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d4c8b9760474e78b8bac0c1a7dbe74a : SUCCESS in 4m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/78340943fe0b4697851ec67295d975e5 : SUCCESS in 3m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b73756db65be4d42821b59ad71085df6 : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":1},{"id":"650b9214974dc9c15d6259d2081a68f5ae7485d1","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-11-25 10:51:29.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d09721657f81b4c01015277e60d5003be35c97f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-11-25 14:51:40.000000000","message":"Uploaded patch set 2.\n\nCopied Votes:\n* Review-Priority-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"db4ac77e73541c8754fbd15309371d9db04d3fe7","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-11-25 14:55:36.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"62e1238b60fa5134141639ad37a215609f9e8fd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 16:48:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/352a7a4fad494170b3c213bd7f2c563a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acaf7afaa6994216a47c472efd5d238a : SUCCESS in 7m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/aa00d58ee4d145e4b82c52d770aede67 : FAILURE in 2m 50s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/2361f963c7144cd190912e7b4bea7713 : FAILURE in 34m 32s (non-voting)\n- kolla-ansible-debian-bookworm https://zuul.opendev.org/t/openstack/build/73163d037008459385d7ae145bb53057 : FAILURE in 59m 16s\n- kolla-ansible-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/3650ea5df47b46c681e19f95a2bd3631 : FAILURE in 44m 32s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/97a38088862f4da8b82b13dcad7f041f : FAILURE in 1h 03m 51s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/a7689e78ecf7423d9dc727796ff73dc4 : FAILURE in 51m 16s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/86426917ded74044951f94c4d8d164e2 : FAILURE in 1h 04m 19s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ebc3e2f9dbd1451b9ff7a5cbdd1db4d6 : FAILURE in 1h 08m 28s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/04a70e40f9154b949cc4cb5a11336a72 : FAILURE in 50m 04s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/f9eba1d9bea84fb59b212bc5014398d6 : FAILURE in 1h 09m 13s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/156f0b541a3d4b7a96338ca4b92d2f66 : FAILURE in 1h 23m 11s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/3401d385be014bed8722d649d43965ef : FAILURE in 1h 29m 48s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/fbbc77b2c78b4a7da841bf99da56a33d : FAILURE in 1h 29m 59s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/d7b9245652ec4a729e5571dbdb18d171 : FAILURE in 1h 52m 00s (non-voting)\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/885883837d0645789500d63b95171f36 : SUCCESS in 32m 57s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd2daba6d84c48c9b0a7bc5164e90e33 : SUCCESS in 59m 25s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0b6fc94b3c44b7fa8e1fda8b8442cd7 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ca302a0f5584fc6b87d307266d82469 : SUCCESS in 7m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/76991d5197ba42699d770b02f2e8cad0 : SUCCESS in 3m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38db726a566642c5aabb08f84003d59c : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":2},{"id":"12ef84451baf3588f8282b1892bc3f94acdd1d4c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 17:25:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9270fc82b19b4bfea5bcb6079eebba66\n\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/c1bfeebcd2294e608874c788b5a438cb : FAILURE in 1h 14m 54s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64 https://zuul.opendev.org/t/openstack/build/4e860c58a3064ec187c4fb4b0e9b07e7 : FAILURE in 2h 07m 25s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64-podman https://zuul.opendev.org/t/openstack/build/9dacfc07af6e4a5598c589d6c6b0287a : FAILURE in 1h 19m 34s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64-upgrade https://zuul.opendev.org/t/openstack/build/278cef9e1e7143c4aba13a2d28afa4b0 : FAILURE in 2h 18m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"28b6ef0d5809292d4c74e87fed03c071b5e80704","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-11-26 06:56:01.000000000","message":"Uploaded patch set 3.\n\nCopied Votes:\n* Review-Priority-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"ed27b60f1e26ebebf331ccbbf02637192f0e7dc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 08:32:38.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/358f3f550d094c729c7551bbf9dc648d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1198d3e6545d4a0a89f7287de4b04220 : SUCCESS in 6m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c5a4cd5811154aea9f0667f786c2cb1c : FAILURE in 3m 31s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/433f279124bc4a96b5e2d2ff982080ed : FAILURE in 31m 49s (non-voting)\n- kolla-ansible-debian-bookworm https://zuul.opendev.org/t/openstack/build/10c8fa738e874555846d7ebcf572f89c : FAILURE in 56m 42s\n- kolla-ansible-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/f2669b969fc9470e83520c9891ba8614 : FAILURE in 1h 05m 44s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/d3d258a276014792b55f3987492de451 : FAILURE in 1h 06m 36s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/984f466219f4413784fc552f6c57b30d : FAILURE in 49m 58s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/3fb02061bb094c16813434f891e7da11 : FAILURE in 1h 03m 36s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cbbac3537964484183903d8d6775fd84 : FAILURE in 48m 05s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a7b03821191040e7b37c41ebea026204 : FAILURE in 1h 05m 41s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/673f74073a1a4c25b1fa59df480c1f10 : FAILURE in 1h 07m 42s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/9c0748032cb2499285aa7ad69870a2d4 : FAILURE in 1h 31m 21s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/53c63225463b4739a98d1661621b010e : FAILURE in 1h 20m 26s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/138056bd4ec840ffa5aada4990a0e6c6 : FAILURE in 1h 13m 46s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/36023b24d7464e08a7be72e1865dc371 : FAILURE in 1h 29m 11s (non-voting)\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/954ae62e2cf046c3b689433f0f2108c1 : SUCCESS in 54m 45s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/741674c1c24d4b669186b85cf7fdb515 : SUCCESS in 30m 14s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c1acca8dc474e3eb8ded8dd1b392c6d : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9135cc6d6c984c09b7c394acb592b552 : SUCCESS in 4m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9fccf83848df4a9a847c49b3e2b88aee : SUCCESS in 2m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b9fea014c5a4641ba644cb2d1919b95 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":3},{"id":"93f8cc1dd6610cd3883c8c15d599f17cc34e72f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-11-26 09:44:16.000000000","message":"Uploaded patch set 4.\n\nCopied Votes:\n* Review-Priority-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"2f8f5064866f85598ef6f42b6cb728f6a32db1b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 11:16:28.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e240c0c9a3df4057bef872aa9eedee6d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b144193f407a47c99c02fcf10d64ba6a : SUCCESS in 4m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/699618c47d824e35ae5385108e1f3dfe : FAILURE in 3m 32s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/adf3aa578e2440e7a493122ffd6465b2 : FAILURE in 1h 00m 57s (non-voting)\n- kolla-ansible-debian-bookworm https://zuul.opendev.org/t/openstack/build/4f016e7cc2c94246ab3cb94684a65f6a : FAILURE in 42m 20s\n- kolla-ansible-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/00d73f4a1b0048069935cf0f8bc3c0d5 : FAILURE in 47m 36s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/7d9f24f02f244798ba11bc90b2a57cbe : FAILURE in 46m 03s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/6d6f2b501d974345b224837c2184a990 : FAILURE in 59m 11s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/60cea39ba35b45a6986bde4d4241327a : FAILURE in 1h 08m 55s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9180a0f9f20843118f4b4c30f373a247 : SUCCESS in 58m 38s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/53eee3f3675e47caaf9342d916f08327 : SUCCESS in 1h 14m 06s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/6f40aaddd7fb4ae3ac363ddb984036d8 : FAILURE in 49m 54s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/91fd729450444a10af841bd4cfbb560d : FAILURE in 1h 14m 38s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/d89469dd66d74146865e6d3797866e18 : FAILURE in 55m 09s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/2a9405290e2d43ffabb9a67e3629cd56 : SUCCESS in 1h 25m 54s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/c4d1154c20624b87928bcaea6ee6ebf6 : FAILURE in 1h 28m 43s (non-voting)\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/8adad8afe07d45d9a55054ba5f26e9c9 : SUCCESS in 1h 14m 58s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1bbf3d38dc7b4955a584199a8c18352e : SUCCESS in 1h 05m 10s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa07af8314834b819a95cea9bbf7fa44 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32f513a21f634d3ca513562b37ebdf1a : SUCCESS in 4m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e3a587b0d3404fbfa5bfa9760d1adeee : SUCCESS in 3m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbab99f7e31a4f05a0a0790a3d0f53bf : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":4},{"id":"0415fcf60e42e2d0c80130f16badc964912ea68c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-11-26 13:36:55.000000000","message":"Uploaded patch set 5.\n\nCopied Votes:\n* Review-Priority-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"a15e28861dfa6717cbeefad557456a0a8d3ad515","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 15:18:54.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c4f856024eda4253918c81ac62313e9e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da5b8bfea1c945f3b78d72c1a3358044 : SUCCESS in 4m 28s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ec86230674e84bf78c19eafd31b35daf : FAILURE in 4m 36s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/68355977e61d405795200b5f4e1a9b68 : FAILURE in 37m 56s (non-voting)\n- kolla-ansible-debian-bookworm https://zuul.opendev.org/t/openstack/build/aec2bc3f35b9499db01e96faed4b0757 : FAILURE in 1h 40m 53s\n- kolla-ansible-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/cccbe5cad1da494e965e13a56e91817d : SUCCESS in 1h 03m 36s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/d0ec08ee9f134598b56e3bbe5cac2a62 : FAILURE in 49m 57s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/1b91908789de4c93abd9436ae094274a : SUCCESS in 1h 04m 31s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/d06ff7a3179642fdb99e422c2fadc022 : SUCCESS in 1h 23m 07s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7faf2fe7147349eb92df8a4aac44281b : SUCCESS in 55m 11s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b70f99ce5f714d1d92f7b05f48b7aad9 : SUCCESS in 1h 08m 38s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/784d8880c2f0400a82caa6b9e66507d4 : FAILURE in 54m 27s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/7a221b12c6c64dd397ff4e260e0bb880 : SUCCESS in 1h 23m 23s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/084dd141fdee45ba94b4617618fea632 : FAILURE in 1h 25m 52s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/16870d9abb154262b49ccfbeed02c3ca : SUCCESS in 1h 26m 19s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/9efa0955262f4aadbcc0ee110aec21ea : FAILURE in 1h 32m 23s (non-voting)\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/8e0cfd1c977247b68d8486ad5df589c6 : SUCCESS in 54m 10s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3f04d85de934350aa4ca8eaf7339bd1 : SUCCESS in 28m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/980edf86255b46a5a8c1b334df384ac3 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3cdd3f7312446028dbfb77c1ada1def : SUCCESS in 5m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/96c471be5d1e4793b0ee59b966b22e78 : SUCCESS in 3m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3356c6c4ee345d19a9a36a5d03b7554 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":5},{"id":"140e7bd43c1564a3fa14bd984624c2c3cb37fa18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-11-26 16:02:39.000000000","message":"Uploaded patch set 6.\n\nCopied Votes:\n* Review-Priority-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"b745e9689eec41d5d63808181ea1b83ec1cd1b0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 17:37:53.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d7d7584a30e94dde91377bc2172a8e1a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef5a5bef063d41eca854ca59ca7a37b7 : SUCCESS in 6m 31s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a69f0a3fd8374c51bc2ad67a5b779b95 : FAILURE in 4m 43s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/4b73f6c8d60547a284216bee950eef84 : FAILURE in 1h 14m 59s (non-voting)\n- kolla-ansible-debian-bookworm https://zuul.opendev.org/t/openstack/build/2ed1f225e2b64bf7837f02ce2adb8f47 : SUCCESS in 48m 35s\n- kolla-ansible-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/54d89240a5ef4bcd89adc4a489b4397f : SUCCESS in 1h 06m 04s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/396acf8511fb428ba1b196a6fe4fb5b8 : FAILURE in 50m 01s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/90f3218f990645f7be3bffc20cb8c4d7 : SUCCESS in 1h 04m 41s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/e8e9c2b2f8434ca29e84759c9640d5c6 : SUCCESS in 1h 23m 35s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25a80471f2ec4eccbbc2654d75b926f8 : SUCCESS in 57m 45s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fb580ddabce64dcfbd655a78f709714d : SUCCESS in 1h 11m 10s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/206bd29c4c944695bb79ac985deeb2b6 : FAILURE in 48m 15s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/6183903e53d34fa0b483cdf519c4fea8 : SUCCESS in 1h 23m 00s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/02b255f3e5cd48e083fe2247576cc180 : FAILURE in 1h 21m 32s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/e7997e010d1c4e5b94cb3ea945535790 : SUCCESS in 1h 32m 00s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/cf0d0cee592a4fc890f9773e4be8c218 : FAILURE in 1h 29m 51s (non-voting)\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/9b399956331649d5b9d3555ab23859c4 : SUCCESS in 33m 41s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a8d4cae521d54798a7b29609aac17dd4 : SUCCESS in 57m 53s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a61dd4e7e2444ed8bfbb0e5c32b481c3 : SUCCESS in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc831f6f26744917abf3a1ff45903807 : SUCCESS in 3m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1f273b6e93db4f619c0288bb71fc2a65 : SUCCESS in 3m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a9108e1d7bd047808e1b0df529da6438 : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":6},{"id":"422a8022a4e48774c33c1ebd9e10f8ebb3465a8c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 23:28:39.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c9f0adc21e143bfaa86869cff4cc81a\n\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/5eed9b133384451d834027e4c960ca62 : FAILURE in 1h 43m 41s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64 https://zuul.opendev.org/t/openstack/build/5663090e6d874076ab894c4dd5e75f5a : FAILURE in 1h 00m 35s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64-podman https://zuul.opendev.org/t/openstack/build/90f38f687abf4c629c7cdc40568844b2 : FAILURE in 1h 23m 21s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64-upgrade https://zuul.opendev.org/t/openstack/build/66f445d8e8ca401b96efa37421e8553f : FAILURE in 2h 05m 52s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"95c7e0b9379d79eef881ef7e7863b76a5c2940dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-11-27 08:42:26.000000000","message":"Uploaded patch set 7.\n\nCopied Votes:\n* Review-Priority-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"d39ed15c993ae4bcd7bddd3744e2bcefc16ce0ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 10:22:33.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac84f29af177413c8b2a952254b34862\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a89513a7cad24127976925b8566c1374 : SUCCESS in 7m 17s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bd25f44e5a7f442cb50687b2c65606b2 : FAILURE in 8m 28s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/fc8c53aa28e64629a2bb2c0d353b8160 : FAILURE in 54m 36s (non-voting)\n- kolla-ansible-debian-bookworm https://zuul.opendev.org/t/openstack/build/443dbc4b481b4268a98ad266b82ae73e : FAILURE in 50m 28s\n- kolla-ansible-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/2eb997d174bb401e8903cbf2052370d1 : FAILURE in 47m 42s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/24202c6446824fe78faed9b9283f63af : FAILURE in 52m 12s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/b66c5cb898904514b296d0b1e147ef0f : FAILURE in 50m 55s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/a76ccc350e8b4f7da2aefee62e8999cb : FAILURE in 1h 03m 15s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0786117b4b7441fcb839b4f12e716506 : FAILURE in 50m 22s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3982f80627a344f38ebca9abe1b6b6ed : FAILURE in 50m 59s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/3252417a850a4c448066ee57e0d181a7 : FAILURE in 1h 00m 18s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/c764b8ae9dc34f83896d4f3a96d6916e : FAILURE in 1h 29m 14s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/f83c198a81bf4118aa8d6777c5ee206c : FAILURE in 1h 29m 37s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/dd01402ac273417e84c47a35cb8d14c7 : FAILURE in 1h 23m 24s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/7738141a45434d0381d387b84560d83c : FAILURE in 1h 16m 00s (non-voting)\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/bca55b8085c44f4ba3b1fd71109aa0de : SUCCESS in 1h 06m 23s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af52ed0cca1c40589fc4e36d658b61df : SUCCESS in 32m 18s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e267a6578d84bcfa6839fb73539b081 : SUCCESS in 2m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6aa4c489c0164df7bd76bb9fd8536952 : SUCCESS in 4m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aa832a094f37419db24d3faa0c355657 : SUCCESS in 3m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec8a83f9d76541a2a79a2f7cbc60aa7e : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":7},{"id":"94c1cc1027899eb559061d9ea1c14adc0d72bcd8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-11-27 10:34:46.000000000","message":"Uploaded patch set 8.\n\nCopied Votes:\n* Review-Priority-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"e7c1b7744413f27177af7573568507eb0ca80a14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 12:12:02.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a635a6aa0d5c41f49689dff032bd062a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bb56fd3d2e346669596e540aa219656 : SUCCESS in 4m 12s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4e1bbcac48cd47d085e3f9344de271a0 : FAILURE in 8m 00s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/a1d277e0829340cbb70d7b7945e26ea3 : FAILURE in 31m 52s (non-voting)\n- kolla-ansible-debian-bookworm https://zuul.opendev.org/t/openstack/build/be1d2cb5a31145a9942644ce0fcf69d9 : FAILURE in 48m 42s\n- kolla-ansible-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/dea802c297594c87962f3ca351cf9dda : FAILURE in 49m 49s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/3ac163fa779041959fe62aad63cf1db2 : FAILURE in 49m 08s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/f4d2c4c300af4f778f3c68d9854b0e3f : FAILURE in 54m 13s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/1b140acbca264f889f53737104cddbaf : FAILURE in 1h 10m 01s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cea518b53465474ea5f524d6f5644a78 : FAILURE in 55m 59s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9908e1ecd17c4f4c989b8dc396c5cca6 : FAILURE in 52m 26s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/21766139c0ef4131b84883a1fb9a8c92 : FAILURE in 55m 14s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/d803471fe1714c1db0e80aee051fc29f : FAILURE in 1h 17m 31s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/6710fb18d6b045529deca869d52aac74 : FAILURE in 1h 21m 42s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/03b99d57b17d4b4aa9d6a4a29d79f738 : FAILURE in 1h 20m 34s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/dc5352ec2ed14e67b564b58a25ef5e07 : FAILURE in 1h 35m 47s (non-voting)\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b0dc038ca7604020a35360064ac32f70 : SUCCESS in 38m 12s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4ef06b4863654748b8fdc670926db7d3 : SUCCESS in 1h 05m 03s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58a37b58f98d437aac408067243dced8 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fde292f3c2648eba4926d54d25f67a8 : SUCCESS in 4m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/838ffcf89aa84fe6965343e81cb63e8f : SUCCESS in 2m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/748518af18eb42b0bac990d8b42e4525 : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":8},{"id":"5cd28372cc9d37a4a84b6187ec506629d002645a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-11-27 13:29:23.000000000","message":"Uploaded patch set 9.\n\nCopied Votes:\n* Review-Priority-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"52f6bba45562eab1867d36e722907e7ab24d19eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 15:14:39.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e1e662b8e08c424b84a1c44f70ce358f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09878136c5b9475e8825f4bba9665bad : SUCCESS in 3m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6388e74e92ba4e60926628e9eb5649cb : FAILURE in 8m 10s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/3a48e097143f444798973ddca39dddde : FAILURE in 33m 41s (non-voting)\n- kolla-ansible-debian-bookworm https://zuul.opendev.org/t/openstack/build/a68022e16b224ab590c5a5aa516a27b3 : FAILURE in 41m 57s\n- kolla-ansible-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/41833106d5ad46f8b1efd53a4b5757a1 : FAILURE in 46m 18s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/9130abf51a0d4d04baaeeeecfef6e92f : FAILURE in 50m 59s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/b98ca0ea60384122947efb7f6102c994 : FAILURE in 58m 29s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/e3f1f0d6093247bfb11dcb5b041f4142 : FAILURE in 1h 01m 44s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03578a5b189b4d6884d4f1e81b1929cd : SUCCESS in 1h 04m 01s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/13dd927e61934b01adff4df75ded4fb9 : SUCCESS in 1h 15m 21s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/81730daa5b15459694bd4c1ad150ea38 : FAILURE in 53m 51s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/99d55c7a52874e5ca944c8b60a11409d : FAILURE in 59m 35s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/4f912e89526b469eb2764b24dd89c643 : FAILURE in 1h 34m 58s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/b2abd72fd4bb418281b3f82081b576ba : SUCCESS in 1h 43m 53s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/6353bceeff7a45b782cda447c48e7667 : FAILURE in 1h 30m 39s (non-voting)\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/630fbc1fb9db456eaa62efa875bee08c : SUCCESS in 1h 02m 06s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26782fd51e1540bc994f8b0e03300e8f : SUCCESS in 1h 09m 37s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4af84866bd3f4ff486c407da98defe71 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5481d57cbc354ec99c03141644cf3a1f : SUCCESS in 4m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8e60888ff1684f1cbd5f0b4341ce8194 : SUCCESS in 3m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e2066515edc48989da3a402309b96fb : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":9},{"id":"5039798d51fb3b849146473d264e7244d4983dd4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-11-27 15:33:51.000000000","message":"Uploaded patch set 10.\n\nCopied Votes:\n* Review-Priority-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"b895fafbfbb581c9f66ad4bca41dace9b22ad5a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 17:06:44.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9e19e5a93eb14a70ae40a4dc12c2b9dd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7793dcf9b3bd4f0ca52c6e879c9bab32 : SUCCESS in 4m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0bc9da2a7f2d4e79b67c9c6223a30103 : FAILURE in 8m 17s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/0987fcef5ab3448db5eba0d05dd07b8c : FAILURE in 34m 01s (non-voting)\n- kolla-ansible-debian-bookworm https://zuul.opendev.org/t/openstack/build/007af298410645b4bb525362d2f66f4c : FAILURE in 43m 19s\n- kolla-ansible-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/9a3cf14e13174baf9c912799d606ab12 : FAILURE in 49m 00s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/dfed0baceec84297bc5bda784f8b5203 : FAILURE in 45m 58s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/ac1bac6249a74cc3945324315919dbe5 : FAILURE in 53m 02s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/261d748b8f62404490213e88a8b4dd39 : FAILURE in 1h 06m 40s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c524702f32546dd9bf7c9be0c14dd1f : SUCCESS in 59m 38s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4a3cf993a0104720af2a86f77e0d94c8 : SUCCESS in 1h 16m 08s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/110d9bd978bd47c2be636867b1212a6f : FAILURE in 55m 34s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/e2141cc0952848ca88409ea1ac302f82 : FAILURE in 1h 12m 48s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/254343aea00e4515bb35a777196e3837 : FAILURE in 1h 29m 41s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/bb0d65d3cf83411b996b96f8560751ff : SUCCESS in 1h 29m 06s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/b3f24d60e76642788fe88dcb64589e8d : FAILURE in 1h 29m 37s (non-voting)\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/35e2c72310e94a6d8085896fbe544e09 : SUCCESS in 54m 22s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1bee363bf5594ca9ad5987b56983c9a8 : SUCCESS in 1h 00m 24s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f63c281cdc854b2f9cd913737b0dc12b : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e92f7aafbe574715a99d0dcb64086059 : SUCCESS in 4m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a68d4488ded34490928d0348e5046a1a : SUCCESS in 3m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b1ba448c24f4240ab70a5e070927b20 : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":10},{"id":"38379925792ed99e427db6ed38c16fb6292725ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-11-27 17:11:24.000000000","message":"Uploaded patch set 11.\n\nCopied Votes:\n* Review-Priority-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"232b2c8328cdd38368c7cc7f6063e15646efeb7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 18:58:21.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cc1321211ed24ef4ac24a00d520214a9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95c17559084a4673ad0219736c9d80ab : SUCCESS in 6m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3538c2674c7e4f23a1f48925ce090712 : FAILURE in 3m 55s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/d0053c721a714abeba1e53affbbd9826 : FAILURE in 32m 39s (non-voting)\n- kolla-ansible-debian-bookworm https://zuul.opendev.org/t/openstack/build/d2a675ecadf342e892dc3debb2f9e985 : FAILURE in 57m 32s\n- kolla-ansible-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/9cf855d43f2b4a77b9158166f34dac7a : FAILURE in 46m 32s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/898c88e0667c466b97e9ef08ba5bc9b2 : FAILURE in 37m 44s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/ba88251bc93141e5bc16f4e729a0033e : FAILURE in 48m 38s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/0a40f687e9354977805f068ee16a4ac7 : FAILURE in 1h 15m 48s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5772210804164d49b64a0ccbbebf5bee : SUCCESS in 1h 14m 28s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8f3220126aa249c4baa0d24d3e6902da : SUCCESS in 1h 11m 10s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/7a32310b21af4cffb5f69d42432e3f43 : FAILURE in 1h 03m 58s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/d5acd27bf3c04057b8dcc918302163a5 : FAILURE in 1h 26m 07s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/abf2b630c1ea45229baa5889c524a3fd : FAILURE in 1h 20m 39s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/ca4870a3cc144eed960abbc59813e8aa : SUCCESS in 1h 45m 27s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/0968d7583c604488bb94f7942b0be7b5 : FAILURE in 1h 31m 26s (non-voting)\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/a1e5656677e844d19fcef44d2b65e36d : SUCCESS in 37m 14s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1dd2e494c1fd4d71a5161111de1a4311 : SUCCESS in 1h 08m 20s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9474a487038c474eab30ee80254dcf4f : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24476ff7ace942d28b852ac3ece793cd : SUCCESS in 2m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/798dd678fdfe46aaab9c69dae5eea206 : SUCCESS in 2m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a7bf5c7690b4279b1bf7e7bbd667c07 : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":11},{"id":"1d90a138d56280059bc538492c1dafce5d662d55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-11-27 19:10:00.000000000","message":"Uploaded patch set 12.\n\nCopied Votes:\n* Review-Priority-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"e815eec8c92fc44eeb7b66fe9d3f14aabd66a308","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 20:36:24.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/415f796fb6b34e0482293e6d4df5209d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0c9a4556cab47a782babe9574b31fd1 : SUCCESS in 5m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/13c436f157494b07a504f5aa6253d511 : FAILURE in 7m 33s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/32b4368ce87d4e029aa8e440c5517949 : FAILURE in 49m 35s (non-voting)\n- kolla-ansible-debian-bookworm https://zuul.opendev.org/t/openstack/build/59424d6bffa84ecc8975bce2c0f53888 : FAILURE in 41m 42s\n- kolla-ansible-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/32d15f88eca44f1f8b5b5926b19394ff : FAILURE in 46m 37s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/9250b9be1d3f45c2aa118564a7a61c9a : FAILURE in 44m 53s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/880677a54eb746f7909594651a744da0 : FAILURE in 52m 48s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/984a8fd6bdcf49048428d4949ffe92a6 : FAILURE in 1h 05m 19s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e70e6eff73894ca69a1edb638f33c19e : FAILURE in 50m 19s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8ba3aaf8ee73423083dbc9c78dc37e7d : FAILURE in 53m 22s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/cd8c8aac463340e78bc2580a46fdc218 : FAILURE in 54m 55s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/a25731dcc3a24d4c8e7c2c6c06b6e350 : FAILURE in 1h 06m 33s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/e09c197903284b54bcc6565a1105c420 : FAILURE in 1h 23m 05s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/8e70f435c5654a6dbc7ba005f8944942 : FAILURE in 1h 14m 18s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/dcdd728a2b454869b290ec8b5bd16bf0 : FAILURE in 1h 22m 51s (non-voting)\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/4cca9337082d47a6be3701ae070b258e : SUCCESS in 52m 35s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ef44061a1424b72ada22d05d40c8c70 : SUCCESS in 58m 57s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87d3302a966a4285b9eaed04d3178e1a : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce746679ee7b472691eea0253bc413b8 : SUCCESS in 3m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/76bc16e27b7e4be9b7f9a61156246ac7 : SUCCESS in 3m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2f97298610c4fa2a9cf8bf38db24b0a : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":12},{"id":"759e7e5172cdb90e1605eb79ccdb2f75b242f22d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 23:33:05.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38e4eaa469bd4b978e9c2387994663f0\n\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/547edda44c4a49018b02db4385c5135e : FAILURE in 20m 59s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64 https://zuul.opendev.org/t/openstack/build/ca196240ad354b8b8606a2ac872bd7ba : FAILURE in 1h 48m 45s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64-podman https://zuul.opendev.org/t/openstack/build/5af949444b7e42ad966236d7fea715cd : FAILURE in 1h 34m 43s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64-upgrade https://zuul.opendev.org/t/openstack/build/97f9257a27e24c9f8ec2eb60ee9a130e : FAILURE in 1h 19m 49s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"a0425348cff258fc8c2e42e74ae556b93a1feb0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-11-28 09:29:24.000000000","message":"Uploaded patch set 13.\n\nCopied Votes:\n* Review-Priority-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"523a73dfc9fc41d32b241a68ef89e7b0ce82b702","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 11:09:10.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/27684bc850e74db4aed6064e8e0d15cf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/373e7abc5b364bb185f163a2e70c5adb : SUCCESS in 5m 28s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7859c036d9714e5c9348560d429ea1f3 : FAILURE in 4m 20s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/342adbab9f9b40958dfdbf1ff674f0ba : FAILURE in 33m 12s (non-voting)\n- kolla-ansible-debian-bookworm https://zuul.opendev.org/t/openstack/build/71717b98ad7e49bfbb604cbcff91f70f : FAILURE in 46m 57s\n- kolla-ansible-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/f0ba88f11e3846fcb97043e1f0decb00 : FAILURE in 54m 45s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/61b07754fae440798c713ca63651cede : FAILURE in 49m 53s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/28354e3eeaaa416cad16ee62af8a328e : FAILURE in 59m 16s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/f4276c9a9ee6417eb98a0b7121fc99d1 : FAILURE in 1h 07m 01s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94f36310efe44eb2965825eefc69d99a : SUCCESS in 1h 06m 50s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4a66543f12ed415cba962b4522d6e6b8 : SUCCESS in 1h 18m 59s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/bc18e05155064cf583bc78e22d9c4018 : FAILURE in 59m 00s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/2e63fb75d7e94f158f1424e6b1b99d02 : FAILURE in 1h 11m 53s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/420422bca336440ea7fbbf745c36e5eb : FAILURE in 1h 34m 36s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/c0ee1d3324ba482d821461f96d9db763 : SUCCESS in 1h 28m 52s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/5460d8b3e6de400f8147e5d98b684580 : FAILURE in 1h 35m 52s (non-voting)\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/3df64b5aa33c4c4699a5e7af6961dac1 : SUCCESS in 39m 30s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67ed3a02defa4f36a93226b55a7fd258 : SUCCESS in 59m 46s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d59ead0b0d14520ab225ce188b19c87 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb67e971a5e64d1cb5f986e060f64e4f : SUCCESS in 4m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/52550e18543a46748933a4a0328dba89 : SUCCESS in 3m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95f52ce10e554f8e957a60ec642cda34 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":13},{"id":"76253de3f63d527c2dd32abd34e5ba3110fb61fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-11-28 11:42:25.000000000","message":"Uploaded patch set 14.\n\nCopied Votes:\n* Review-Priority-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"3fd0076f61d472c346c8415551c8a67252de8bce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 13:22:26.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/484901ec1b0d4f378d15e7624bd39e39\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55782f5f5d774ee8b5f20c4afc870fb1 : SUCCESS in 6m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/dd98c4f666fb4e43b90a0ab4e0317d7e : FAILURE in 8m 04s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/2399600e58b1433892b2f6bc59a56147 : FAILURE in 30m 36s (non-voting)\n- kolla-ansible-debian-bookworm https://zuul.opendev.org/t/openstack/build/39cee4224450472fa4ddd71b5f3ed1e4 : FAILURE in 47m 39s\n- kolla-ansible-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/1f4b7fac78574e19ade305452aa4cd2f : FAILURE in 51m 42s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/a16e8de8a85849a8a64e37e8932318ae : FAILURE in 52m 02s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/cf01c855705243e7b033d0a5217d12f2 : FAILURE in 53m 36s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/e8972abdc4c343cc840fd9f13eef6cd8 : FAILURE in 1h 05m 32s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/38cd1de6abdc47899c238f77e7bca4e5 : SUCCESS in 1h 06m 55s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/242dedf723d8449ead7196834485326e : SUCCESS in 1h 19m 48s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/69e321b595064325b577f77aa42e61a0 : FAILURE in 54m 14s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/d1e98228eda743f8b083e64004895ed8 : FAILURE in 1h 21m 19s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/c2c9c63a0fda4ed3bfe08233324da2c7 : FAILURE in 1h 04m 44s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/f0acb1b42b1e4e1e83b7ab904ff08862 : SUCCESS in 1h 38m 48s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/f3d27a11563946a0a8ea9f3ffedd5dc6 : FAILURE in 1h 25m 08s (non-voting)\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/eeab6af3a4be4295a91cf1b3d399e332 : SUCCESS in 1h 18m 24s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6558b6911c04ad49f51787ebb73000e : SUCCESS in 1h 15m 55s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6239ada499674b9096be6c5658c2a150 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ee2fb91d8b5470e8e0a64e63140aa3f : SUCCESS in 4m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/58bc5a9d1168486a979931e2535646e7 : SUCCESS in 3m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c12b9bdea9454d9e9ef26581f95bbdff : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":14},{"id":"10b29de86e9e97d0dd6fcab38d1b881e4548f5ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-11-28 14:02:36.000000000","message":"Uploaded patch set 15.\n\nCopied Votes:\n* Review-Priority-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"f3b4f5700409fc63493b0582a14413c0b6aa044f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 16:13:26.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f0559d77b5e44220898182a3913e4147\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73f25ed781b0488981a1b07970265aa6 : SUCCESS in 4m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3b2024dd35b74312877f68f7c7c15ecd : FAILURE in 8m 50s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/f313201236be472b85a7bd802623116e : FAILURE in 1h 06m 40s (non-voting)\n- kolla-ansible-debian-bookworm https://zuul.opendev.org/t/openstack/build/b3b35740dff445949933edc290d41fe5 : FAILURE in 1h 18m 01s\n- kolla-ansible-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/08833619c1da4f1f964e2ba84ea81105 : FAILURE in 1h 17m 46s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/d12afc322c6a4d07baaee1eeb210f124 : FAILURE in 1h 22m 32s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/e0b005d132f34382a71c0e7124bcb684 : FAILURE in 1h 23m 08s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/c8a4d0be082a4c6ba9cc56ac5ffbe3d5 : FAILURE in 1h 25m 54s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a3c8f075a1c2477c883be1b3225fed48 : SUCCESS in 1h 33m 52s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2f71c13b141e456cbb37238894ec94d4 : SUCCESS in 1h 43m 08s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/7a6a0bf7627b43ff9799e36f222cf148 : FAILURE in 1h 26m 41s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/e7ce14340d2b4752b7e1bdc7c51e590f : FAILURE in 1h 41m 55s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/8c933dcfac554bd3869782e5cc53282b : FAILURE in 1h 33m 38s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/a672bcbfbd4542c1826a16637bd75df4 : SUCCESS in 1h 25m 06s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/dbfdf7ec20824d5fbb102c5dd55659c3 : FAILURE in 1h 58m 35s (non-voting)\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/67eb770ed78f48d29e56fa345fced9b9 : SUCCESS in 1h 16m 09s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b310e237abda449a83a3770d89cc278c : SUCCESS in 32m 22s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d09400706ab34866a27810ef4e95fb20 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef7c1825c61b48fdb6e64ac58315f2f1 : SUCCESS in 4m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/73ec7498b7194a3c8c33c12aae0fb8e3 : SUCCESS in 3m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0fdec586dd94ceea70c5ec4f1687895 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":15},{"id":"5c7522f6c260a25ce94e8ed7df8514624dd16118","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 16:37:17.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba8f929231154fab83cad0a6c7763922\n\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/9fc8d8f304044a259bbd3a2a5e199f16 : FAILURE in 2h 29m 43s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64 https://zuul.opendev.org/t/openstack/build/6255009c0b524c339167f84596a85a58 : FAILURE in 2h 03m 55s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64-podman https://zuul.opendev.org/t/openstack/build/4bdd2a0c99824e0c9d6e31d9652372fe : FAILURE in 2h 23m 14s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64-upgrade https://zuul.opendev.org/t/openstack/build/98b5f1b91e0540619e91b9606e81a934 : FAILURE in 1h 03m 19s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"f32b9debaa36ee9896e6f7e525471e78b78d32fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-11-28 16:48:16.000000000","message":"Uploaded patch set 16.\n\nCopied Votes:\n* Review-Priority-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"50ff287d0645110edf8d1384bbc15517d56c8315","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 18:57:55.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e83e857330a443c8355b63e5e9c1c62\n\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/090577a8373d46669d1004e209fd52d3 : FAILURE in 2h 07m 10s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64 https://zuul.opendev.org/t/openstack/build/7683adf2d32a4994bc0e2e8f24bcdfba : FAILURE in 55m 21s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64-podman https://zuul.opendev.org/t/openstack/build/931879c41e99492a81fde5a7ade98176 : FAILURE in 1h 59m 56s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64-upgrade https://zuul.opendev.org/t/openstack/build/e2fdd44e53f24d16971a2e3bfdd16e1f : FAILURE in 1h 08m 56s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"ec1a14d511b94f8326339fa4c8d3133edef4fb9b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-28 19:00:10.000000000","message":"Patch Set 16: Code-Review-1 Review-Priority-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"564f73f93a1df0e29a0a08598845e9fbb3cba0ec","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-28 19:01:26.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"7de8d37b8105ba2d4cf83b3846816485450bb29d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 19:06:30.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db1454d783da4179bd82cb7e8eb9c34a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87cd27a49b594de0b55a57f80a9197e1 : SUCCESS in 7m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cb859ef00dc44d78bb35dadf4b5eb223 : SUCCESS in 3m 52s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/01bf90bc876749b8a9957893e3ca1ea2 : SUCCESS in 36m 53s (non-voting)\n- kolla-ansible-debian-bookworm https://zuul.opendev.org/t/openstack/build/2d4ce54f06ae4527b8cdec4c4bee0c03 : SUCCESS in 53m 38s\n- kolla-ansible-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/61f140fd89c94f4997edde3e97066551 : SUCCESS in 1h 08m 45s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/b8ad5d3ae043421a8972680506ee18f3 : SUCCESS in 1h 13m 37s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/4144946a2f9549338ff680531064b067 : SUCCESS in 1h 06m 08s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/76940a156f694665a49da3cbc3929679 : SUCCESS in 1h 32m 39s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7697634be73949fd86babc74ae325d89 : SUCCESS in 1h 04m 16s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/94deac037adb42a896c8b9fe1e3cda34 : SUCCESS in 1h 17m 10s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/178e5997e93145b39464359b86f8d77a : SUCCESS in 1h 21m 19s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/b84275f139014159897afb060801d624 : SUCCESS in 1h 26m 14s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/12fe9b856e18427e9765ff20283da6d3 : FAILURE in 1h 30m 43s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/fbabce5a21384686800b21725c6b5246 : SUCCESS in 1h 40m 52s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/2fb240ae58f143a5a03fc55645f778d0 : SUCCESS in 2h 16m 37s (non-voting)\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/054fc2c4136549a1b4a0f0fb3fd79537 : SUCCESS in 1h 16m 53s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/989e06bb45e24e37adddc15f22f7810c : SUCCESS in 1h 07m 05s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b66a23125198490b8bf8006c1080651e : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a56381f018bb467dba29f780aae373fa : SUCCESS in 4m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0ab0cceacf8a4f7ea3553585a2b1c7a8 : SUCCESS in 3m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/acac25a79c754f3a99f41f6e0ae0a78b : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":16},{"id":"3252bfdd57cbc8e958f5961a9770dc467d57d876","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-11-28 19:19:43.000000000","message":"Uploaded patch set 17.\n\nCopied Votes:\n* Review-Priority-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"49c5115ee75e7cfc69471f144d4675e2501fc67c","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-11-28 19:20:03.000000000","message":"Patch Set 17:\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"9b365a59425e6da7a4edc1d8ade5b262d14c38c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-11-28 19:33:54.000000000","message":"Uploaded patch set 18.\n\nCopied Votes:\n* Review-Priority-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":18},{"id":"8dca5ca6fd722a6931de17372a6a8c800f83c230","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 21:01:41.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c24b6ee3ed0c447892ae7840c4d8fc07\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d39952b861de43d3bac059885f159529 : SUCCESS in 4m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/eaaf6fe240024a0781f8c75daf108090 : SUCCESS in 7m 57s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/702c90e3110c4895a446620ec76b531d : FAILURE in 32m 32s (non-voting)\n- kolla-ansible-debian-bookworm https://zuul.opendev.org/t/openstack/build/1d6a5a04f8534aeab91b47ebaea3ee0b : FAILURE in 42m 09s\n- kolla-ansible-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/c26bda3e4bf64ef8b8bfe5a832367471 : FAILURE in 49m 08s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/66122b55fe80446e9695ed7e1ca674fc : FAILURE in 49m 15s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/033ebfe210944dc39fa063b4d93616ee : FAILURE in 48m 30s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/02cda137a68a439bb8c083853c4e79f3 : FAILURE in 1h 05m 16s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/adfc5a861ba54f63935ec425e6bf45ee : FAILURE in 47m 24s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1fe3c04317514d46981154215374bddf : FAILURE in 51m 02s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/91beebc499014ac5bb529eece7f3a1b0 : FAILURE in 49m 39s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/8e90def71cf7468ba8402f3f908d9ce2 : FAILURE in 1h 12m 59s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/09752f1b60394f7cbad0ceb628fbb6b6 : FAILURE in 1h 21m 08s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/d23218a098634906ac67d2e1166e9914 : FAILURE in 55m 20s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/e12e0a2859e54a68a8a1dc088b915138 : FAILURE in 1h 26m 10s (non-voting)\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/313321cba9e9429f9ad0ff5eb3837c2a : SUCCESS in 39m 21s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/06a22dee264149a085648c3d1b033ef7 : SUCCESS in 1h 21m 00s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c558897f7c7645c784d372df8dbe4abd : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84459314c223435390aaf1608e379668 : SUCCESS in 2m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3687bbb0608347fdaa3f88287ab588f3 : SUCCESS in 3m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa8f143cc1a744248485507671870fc1 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":18},{"id":"42638c7899b6b42782be2d18a761ba2966e4fef7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 22:23:33.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/583942e89d8a439fb3eb29a0a57f13ca\n\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/33f4673e8e8a44d6a51771d62eba038f : FAILURE in 1h 20m 33s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64 https://zuul.opendev.org/t/openstack/build/43925ef2659a4ad0992c0d52f120b8ac : FAILURE in 1h 06m 14s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64-podman https://zuul.opendev.org/t/openstack/build/d050fc91b12d472ea065a72bb0e335a8 : FAILURE in 1h 08m 02s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64-upgrade https://zuul.opendev.org/t/openstack/build/40cd23f15b874c68835062beccbde394 : FAILURE in 1h 38m 09s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"9c4e0ca74d02ed4b5a886587e4fe218131e7f676","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-12-01 14:06:19.000000000","message":"Uploaded patch set 19.\n\nCopied Votes:\n* Review-Priority-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"e125012da0a5bb9bd838d92dc2b1445126168ce4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 16:16:58.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b5d70db6d3346cd8b8e619d03da4115\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b160fe54724a4003a578edf622c1ef11 : SUCCESS in 3m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7c252f090d044cb08c74b82bd70682ee : SUCCESS in 7m 55s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/0cc35284103944cb9cb4b20ce9908824 : SUCCESS in 43m 58s (non-voting)\n- kolla-ansible-debian-bookworm https://zuul.opendev.org/t/openstack/build/83e094e5d2fa46038b7664a180bf222a : SUCCESS in 56m 37s\n- kolla-ansible-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/3c01ebc9f6c640dabc4d014786604fc1 : SUCCESS in 1h 08m 39s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/805ac004c0d549e88729fba5f9d8d945 : SUCCESS in 1h 10m 15s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/a3b5620cdfe34163bc23e4bd138cd172 : SUCCESS in 1h 04m 18s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/0eeffc89380c4baeaf04d6f525a2c71b : SUCCESS in 1h 28m 28s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/125dabce82e44aa6b0f35d9615d4fad2 : SUCCESS in 1h 04m 09s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b0cf28e0f82d4474820dc058c3051839 : SUCCESS in 1h 14m 36s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/b9eeb7cd1b4d4f668feb6eab21aa1cce : SUCCESS in 1h 18m 02s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/9d0f0a0340c6439ea951608fa57f7c62 : FAILURE in 58m 52s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/d6772175d16240608f9917feef2b500d : FAILURE in 1h 13m 01s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/dc66936f10b143e797de2311d0ff8b3a : SUCCESS in 1h 31m 47s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/cdbc322f59634296b87532382ae11c4e : SUCCESS in 2h 09m 05s (non-voting)\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/e52482ecd11e4e6581e145bfdd15e1d6 : SUCCESS in 38m 48s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f742b95c39247bf88aedb6009b61fc1 : SUCCESS in 1h 20m 22s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ec8933fb24f47ad94836db52f8b7daf : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9558828585eb47e5b29a1d2cd6471f89 : SUCCESS in 3m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c352e935595a4782ae863ba34f482b81 : SUCCESS in 3m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a10b089bd045424ca199f2a4067314d5 : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":19},{"id":"01f6b973a92d1d9be143b7672377ed5f0e84d521","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 21:37:39.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f94b087fbed94290994ee63ee9030548\n\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/628cb4cddb0b43f2b1d69e4716cf30a7 : FAILURE in 1h 17m 57s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64 https://zuul.opendev.org/t/openstack/build/de1b4e2c640a492c865b74ce03b336ee : FAILURE in 1h 23m 06s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64-podman https://zuul.opendev.org/t/openstack/build/e40a9eb36b164ffea1ee0df6b58dd019 : FAILURE in 1h 59m 09s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64-upgrade https://zuul.opendev.org/t/openstack/build/5cdfb332d69c43afbc3f785f278d7b85 : FAILURE in 1h 05m 07s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"1e64e4c324df0140f1f8ac1648c4881007a4692d","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-01-14 09:47:07.000000000","message":"Removed Review-Priority-1 by \u003cGERRIT_ACCOUNT_22629\u003e\n","accounts_in_message":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"_revision_number":19},{"id":"321262032f203a63f84066592cb19a79dea71c5b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-10 11:58:32.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"e7cb55971ef16c6659affc5961139213424f7ef7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-02-10 13:28:56.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":20},{"id":"88c980e0a33114b6ff74696d43460ba4fa4ac3b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 15:05:22.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e725c8ad2c704ab7a4253702c7b10e12\n\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/d717063d02d34133a7a0687145ddd9e1 : FAILURE in 1h 00m 38s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/cd99f14cfe3c4bcf85aca88016a46379 : FAILURE in 1h 15m 27s (non-voting)\n- kolla-ansible-debian-trixie-aarch64-podman https://zuul.opendev.org/t/openstack/build/33ee65af53894329abe7dfd17142ac32 : FAILURE in 1h 25m 30s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64-upgrade https://zuul.opendev.org/t/openstack/build/d1a661c4d3f14e58a39ae66006f3c43d : FAILURE in 1h 12m 12s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"f60a920fbb9c826c49c13ae4b6f76982628fa17e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 15:30:08.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/144ee8bd39044013915635242f8b6342\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06e0092c65094faf948f171350a42858 : SUCCESS in 6m 56s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b00ad9174f264f658abc4ad64eda9c09 : SUCCESS in 10m 02s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/564d4f28a0254077963431ac842ad859 : FAILURE in 35m 18s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/b8a6e017fa6b40acb00073ea763de2bf : FAILURE in 1h 20m 17s\n- kolla-ansible-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/e7c9e8899dc74ee0acaca23c199e2cb8 : SUCCESS in 1h 20m 00s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/6f676916c8a94676b7a51ea766881c14 : SUCCESS in 1h 15m 38s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/52da8a5c86044f489166cd1a8716cd18 : SUCCESS in 1h 28m 01s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/7d9d3e13d56b43f39f4f4ce615577014 : SUCCESS in 1h 23m 38s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/5b4d0781298f407c90ca1c39ca341997 : SUCCESS in 1h 09m 08s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e2ab3caa4ae849e88d8152fe354157b6 : SUCCESS in 1h 31m 26s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/64f5ec5613fa47858117cdbcfda8d0c6 : SUCCESS in 1h 05m 34s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/7daae1e82f0d42aa9e944cf4c576a968 : SUCCESS in 1h 18m 07s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/09d451d9bbf547e39503d5826f56c820 : FAILURE in 1h 22m 57s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/40fdd968c07f470184c93c3533bc7b43 : FAILURE in 1h 47m 24s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/2b6c828203c84512a44fa202bfad6653 : FAILURE in 1h 55m 40s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/5b25242bd6454ceba2040daa62e45c25 : SUCCESS in 1h 59m 09s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/efc1cd0490434cf8a2d7174ae715b7d3 : FAILURE in 1h 51m 33s (non-voting)\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/9a04727fc8844e7c86a5de8ca2aa6d49 : SUCCESS in 1h 15m 34s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7499c611e0c948ef83d726cd0d0332d0 : SUCCESS in 1h 01m 34s (non-voting)\n- kolla-ansible-rocky-10-upgrade-slurp https://zuul.opendev.org/t/openstack/build/c84a1c5455904beeb2ddce99c5dc6ab1 : FAILURE in 51m 32s\n- kolla-ansible-ubuntu-noble-upgrade-slurp https://zuul.opendev.org/t/openstack/build/516e6c3551804e0493c23ddd91e2337e : FAILURE in 44m 43s\n- kolla-ansible-ubuntu-noble-cephadm-upgrade-slurp https://zuul.opendev.org/t/openstack/build/b88884063cce4b27b13bef386ae22268 : FAILURE in 1h 22m 19s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64b70ddbb66345d9929c263b2300c488 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fb68b39b2314bc4b489da18521c83c5 : SUCCESS in 4m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/de707bc4bfdb4cfba58ad97ce2c54605 : SUCCESS in 3m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e9664e781ff487aa7ef3b10cb0c7564 : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":20},{"id":"49533705c94a743d51f92b95fedaffe6dd24a7ca","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-02-10 15:33:36.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"b843a9a4b0188e686d6d05972187f4053d0123be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 17:15:51.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a3a778e19d145799abfb7eca5b69fb7\n\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/c761421fec1f44b2a48a14f37c92a41a : FAILURE in 1h 16m 40s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/8436144d096a440e9c96185453408f48 : FAILURE in 1h 40m 10s (non-voting)\n- kolla-ansible-debian-trixie-aarch64-podman https://zuul.opendev.org/t/openstack/build/42a6aba0d35f4ea482ef651dadc55332 : FAILURE in 1h 39m 48s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64-upgrade https://zuul.opendev.org/t/openstack/build/62ec3914d25e484b8e2d49c2d9bc5dd8 : FAILURE in 1h 13m 45s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"481b8cdfe3326c63469d4401d9dbe1affb57cf08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 17:51:03.000000000","message":"Patch Set 20:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b634ff1459448dc97d0615cb9528693\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b08a106d4df47d5a1213d6c9233cd5a : SUCCESS in 5m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/64abbaed5d594295bc6087cf6d4edffb : SUCCESS in 5m 18s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/d9644f1a73294faf963bee8ba6f0b73b : FAILURE in 46m 33s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/664f7bfaf8a24622b5c0817a3b09b396 : SUCCESS in 58m 05s\n- kolla-ansible-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/807608bde7af496ba3cc3d1036d599a0 : SUCCESS in 54m 26s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/a0b3b6d3f2764f8ea31e978fd499db6f : SUCCESS in 1h 14m 42s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/8d1288a219714dd4be066cf3f8b57ae0 : SUCCESS in 59m 11s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/78d35c65b89e419aaf2aef56a270fc35 : SUCCESS in 1h 22m 32s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/f0431c2af6ed4408b6b2fab3f31857b6 : SUCCESS in 1h 22m 10s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d565fec3b434c7185bcb73d3fe35729 : SUCCESS in 1h 12m 54s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/16cb30788b7f4325aaea218ea1aad7d5 : SUCCESS in 1h 11m 47s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/b7570922a9d64da094fb505fbc0daea1 : SUCCESS in 1h 26m 15s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/4810580d7ef94d6fb1776c0d4eaf8338 : SUCCESS in 1h 32m 27s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/c5c582617eca4d3fbe15d8bc124dba1e : SUCCESS in 1h 49m 45s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/cd8432bc079347888098ea89a7e80009 : SUCCESS in 2h 15m 50s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/322e9686d38b4fef8f1486510537788a : SUCCESS in 1h 54m 32s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/defd2d1470d84c8fb11fc601ad2fbe56 : FAILURE in 1h 35m 28s (non-voting)\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/6d6350ae3ea84f559395cf7294323dcb : SUCCESS in 1h 07m 01s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db3c8ee2a2594fd4be4b9df275c3d570 : SUCCESS in 58m 33s (non-voting)\n- kolla-ansible-rocky-10-upgrade-slurp https://zuul.opendev.org/t/openstack/build/ab87eba629e84884bfddf41b7605a92e : FAILURE in 1h 01m 10s\n- kolla-ansible-ubuntu-noble-upgrade-slurp https://zuul.opendev.org/t/openstack/build/b79782250d754bef93651e01799a2d3a : FAILURE in 43m 25s\n- kolla-ansible-ubuntu-noble-cephadm-upgrade-slurp https://zuul.opendev.org/t/openstack/build/bf52f471926742d3882f1fdf4816f989 : FAILURE in 1h 28m 15s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bef2f05fda0040fb982725646025e23f : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46fc61f449b84dd2afc7341bba74d4f9 : SUCCESS in 4m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/05ba8b45d85a416b9be98257095d5bc6 : SUCCESS in 3m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e04bb9b7869c4818819daa0592e5ac4d : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":20},{"id":"6d3bb12cf606b911fc11a68353fa05d9203f8926","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-02-12 07:30:13.000000000","message":"Uploaded patch set 21: Patch Set 20 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":21},{"id":"7560afcbe7c68a29572934764b10c92f24ab765d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 10:29:51.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9e393335e66b4710ab3c15ef43917d42\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04d544e7df0e47589bf65fc65aac005e : SUCCESS in 7m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8654704e4a5b4657a6961905796af0d0 : FAILURE in 4m 25s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/6b9c9a57419b48038c0531d3b85e7fec : FAILURE in 34m 33s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/ccbfac2f33564a948a319fde7dcd4997 : SUCCESS in 56m 24s\n- kolla-ansible-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/3afb47ab20634a5a80be57550face4e0 : SUCCESS in 1h 04m 01s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/a0732c610e8f4740b07f786dbac43b2a : SUCCESS in 1h 17m 13s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/e7234cf404b74b5f927dc2fc10460ba1 : SUCCESS in 1h 02m 55s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/64ea4d8c7b5746c5a5e448f1e1b0314b : SUCCESS in 1h 39m 47s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/5276750177564d509e28b66e1b95eb2e : SUCCESS in 1h 26m 37s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2c9574217294dc0bf62a5da5afa6d84 : SUCCESS in 1h 02m 54s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/31ce175ede3049c5a757d54beb66b3ff : SUCCESS in 1h 10m 58s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/e1f243b144a74a879496b8800bc68ae4 : SUCCESS in 1h 25m 13s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/7e6a167d7e674ad5b21983dffb6314c2 : SUCCESS in 1h 50m 20s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/e724f6b5141c48a0ba90cf0532b8b340 : FAILURE in 2h 58m 08s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/369207bfb74547b2a2f03fdb9573ab60 : SUCCESS in 2h 31m 10s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/386bb427da67460c89d7682f0e537d47 : SUCCESS in 1h 45m 02s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/75c16dd7fc9f42c9a7a698b89dd1ce78 : SUCCESS in 2h 31m 09s (non-voting)\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/7d6ac9d51cf4423bbf506ddb6319c8fc : SUCCESS in 1h 07m 54s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/efb83d8b16444369900e52effb34150e : SUCCESS in 1h 06m 41s (non-voting)\n- kolla-ansible-rocky-10-upgrade-slurp https://zuul.opendev.org/t/openstack/build/3a48a999865249249843ce8a7886fe81 : SUCCESS in 1h 19m 23s\n- kolla-ansible-ubuntu-noble-upgrade-slurp https://zuul.opendev.org/t/openstack/build/c8fb46e46dad4bfa8292e5b624e86e5e : SUCCESS in 1h 32m 41s\n- kolla-ansible-ubuntu-noble-cephadm-upgrade-slurp https://zuul.opendev.org/t/openstack/build/f3fd63d0c920403ebefa417daf847979 : SUCCESS in 2h 33m 07s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1fd85ac880bc49528d6d24bb788d7d3c : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b41263bc31ae418887fb48d1f54faf0e : SUCCESS in 3m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1c657a7565ae4604bee77890de7b4e44 : SUCCESS in 3m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ea07e6f7e3545c39de7a8c336ae5abd : SUCCESS in 7m 45s","accounts_in_message":[],"_revision_number":21},{"id":"811da8545d65a03bc994a99d08644b196ce360bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 11:02:14.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98bd1213690c42feaab5fd2dcb7eab06\n\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/2519d60d0c4b4c54b7633c74f5125a5a : FAILURE in 1h 56m 19s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/b67cc67fa997472d9a62336f9503e00b : FAILURE in 1h 33m 29s (non-voting)\n- kolla-ansible-debian-trixie-aarch64-podman https://zuul.opendev.org/t/openstack/build/610ebf49fca9469dbd0b7ca417002c7d : FAILURE in 1h 40m 21s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64-upgrade https://zuul.opendev.org/t/openstack/build/7b4e885c1160488faafe82765eb1644a : FAILURE in 1h 49m 02s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"ebb45093b7848753039c7f4c30435d0f212e75ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-03-17 09:46:14.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":22},{"id":"d5fd4e44a70d8f8a91d8c9676f0d5f28660e993a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-17 09:49:18.000000000","message":"Patch Set 22: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"9f12591aa47010c56b61fadfc77421ab9dd1169b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-03-17 10:20:01.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":23},{"id":"3f03e9a773836a7905894f2f4003ae5ee254bc99","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-03-17 10:20:06.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"13ec21f7285d5d7db6fcfa6320a2439eba7ddd95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-03-17 10:33:09.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"0462f127b7de22d2840729b374d0782f00ec01dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 12:49:32.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4ecf9ed8a5c465099dca33ed7f2e8a3\n\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/8e35a520296b4467bb25ddc9e7137d07 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fad4c13aea7407b95259413381f65cf : SUCCESS in 6m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3a0da76c79c9485aade333a4aa3c2945 : SUCCESS in 10m 12s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/ba5a550db48b44bcb2387181d92b0cd2 : FAILURE in 38m 33s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/c0b449bb4cd645c3a25d3609b2a75dc4 : SUCCESS in 53m 31s\n- kolla-ansible-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/2020afdd5cf34605a915a922d8f75bf8 : SUCCESS in 53m 41s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/39de8657776144e3bd6b7513a104ee0d : SUCCESS in 1h 23m 48s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/aa469d4e8538478183ba1e51197fe662 : SUCCESS in 1h 04m 00s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/8d166a76ddaf4fdf825b1c35608a6233 : SUCCESS in 1h 27m 29s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/49ee1bb12f7049b19829310abdc24f18 : SUCCESS in 1h 18m 42s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3147a63723794ea088f0ff2def7d9f60 : SUCCESS in 1h 01m 09s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cc5ac50813994489b35ba9aa0cadee34 : SUCCESS in 57m 39s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/6eb351e01501438ab39fade636e99543 : SUCCESS in 1h 28m 03s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/2f5d041a954d40eca8c409da34208ddf : SUCCESS in 1h 24m 37s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/46553c834199463f8e637ea8c3ec5d94 : SUCCESS in 1h 44m 56s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/9b6f62a96dec4613a127da8701f460a7 : SUCCESS in 2h 02m 36s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/e2abee193df243b78a9104fcb17e5a20 : SUCCESS in 1h 42m 23s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/cb750dfe7af54ef886038ccc81b536c7 : SUCCESS in 2h 14m 45s (non-voting)\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/ed88ed9cab564c51ad397d332c15383e : SUCCESS in 57m 31s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de91860a17334551aa8729398da40865 : SUCCESS in 1h 02m 55s (non-voting)\n- kolla-ansible-rocky-10-upgrade-slurp https://zuul.opendev.org/t/openstack/build/21cb9f9303344925a0e5124792d31b6d : SUCCESS in 1h 27m 52s\n- kolla-ansible-ubuntu-noble-upgrade-slurp https://zuul.opendev.org/t/openstack/build/59c7f115cbf743098485eb227196f5cc : SUCCESS in 1h 19m 12s\n- kolla-ansible-ubuntu-noble-cephadm-upgrade-slurp https://zuul.opendev.org/t/openstack/build/2813906165654987af71861a9c772826 : SUCCESS in 2h 08m 46s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/53147da6c6734ef58421cd123f3ad32b : SUCCESS in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c9fcddf8cba40deada1a898c7201ed7 : SUCCESS in 4m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8da720fadc9b4a168df0d5037b61173d : SUCCESS in 2m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c1cc37633444a8e8a7df1bbc7551571 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":24},{"id":"21bfcad08b9dd580e1d7bc42ce7217af3be3e2dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 14:49:59.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eebcf0f1ea9e4d4e99c01668d3024173\n\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/7be0cbd53f2e421ebcd002649c00be8e : FAILURE in 1h 59m 43s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/d67e96e2cd3847b389fe2944b16c2b69 : POST_FAILURE in 1h 03m 39s (non-voting)\n- kolla-ansible-debian-trixie-aarch64-podman https://zuul.opendev.org/t/openstack/build/6b0914f6aef649fda439d1fd5aa4409a : FAILURE in 1h 26m 09s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64-upgrade https://zuul.opendev.org/t/openstack/build/611cdcd5e3b54ed99cd1a35a30fc2398 : TIMED_OUT in 3h 07m 25s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"d23eff045657a8b794451ba9320b7bc4930812d9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-27 10:57:27.000000000","message":"Patch Set 25: Patch Set 24 was rebased\n\nThe following files contain Git conflicts:\n* ansible/roles/rabbitmq/tasks/version-check.yml\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":25},{"id":"8731933cc72380c26d231bd37c5c910b008ec9bb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-27 10:58:32.000000000","message":"Patch Set 26: Published edit on patch set 25.","accounts_in_message":[],"_revision_number":26},{"id":"d8fe06bf2a3f5e14a8dc7f197e9c75dbd6908b34","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-27 10:59:46.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":26},{"id":"7f18d24a67da0b8de564af42442eaa79caf3c5e1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-27 10:59:49.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"b1d13659234e7f2bca04be85416b9821c6b324eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-27 11:00:30.000000000","message":"Patch Set 27: Published edit on patch set 26.","accounts_in_message":[],"_revision_number":27},{"id":"e3a0db8bf89cbf0e1f0d106f01b0ec422b0baf30","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-27 11:57:30.000000000","message":"Patch Set 28: Patch Set 27 was rebased\n\nThe following files contain Git conflicts:\n* doc/source/reference/message-queues/rabbitmq.rst\n","accounts_in_message":[],"_revision_number":28},{"id":"ed436868daad29465f749d31ed4c0aea9a294808","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-27 11:58:03.000000000","message":"Patch Set 29: Published edit on patch set 28.","accounts_in_message":[],"_revision_number":29},{"id":"638a558814741f726b662888ab0b6e8d3adb4829","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-27 11:58:13.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":29},{"id":"28079cf75a95fe010deafce5faf9e4abf63a8fab","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-03-27 13:00:18.000000000","message":"Patch Set 29:\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"2e8126ebb199f6216e96ccac0167eea2e9478a9d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-27 13:50:51.000000000","message":"Patch Set 29:\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"177ef6231c85e62b1710207e47faac39ea0ad431","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-03-27 14:00:49.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"8dd90e4b1ef51353312b575a0ed20853ff014438","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 16:19:15.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8742cf5292d14b0c9ddeb9b7d3f018bb\n\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/a63284c91bf149cb9a3b16881a56a9c7 : SUCCESS in 2m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19d34f40fd204c1e9db229a33e2f8dcb : SUCCESS in 3m 33s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b245c0b6eae84457a585869e30287ab5 : SUCCESS in 11m 05s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/5ea4a6fe5bfe43239826ecd5adbb141c : SUCCESS in 1h 30m 40s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/c88ea971935242a294339cadae2689c2 : SUCCESS in 1h 05m 16s\n- kolla-ansible-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/87e2215ade9245c4ab61df49fd51fc2e : SUCCESS in 49m 36s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/b02b673f86b74175bbf82ea4b00a8c6e : SUCCESS in 1h 16m 59s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/aeccc68feab1413aa09a024fc8f4523a : SUCCESS in 58m 40s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/b3f95f4bd7494cbba13191da24ad4f83 : SUCCESS in 1h 27m 10s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/794ca447025e4d78a20d6ba526cec325 : SUCCESS in 55m 35s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/499d25562989466991f4ecf9584e3d30 : SUCCESS in 56m 31s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/82ef4fed6e044213b5b16bb32fd58a86 : SUCCESS in 57m 26s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/17966b08fe0f4d5697a6c49b542af41d : SUCCESS in 1h 25m 15s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/ae1ce3adea614fce81b7b914e2265cac : SUCCESS in 1h 24m 07s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/35ad58837dbe4196be1619f2d4341540 : SUCCESS in 1h 29m 48s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/bed10a5359f74787b43b194cd550fe25 : SUCCESS in 2h 07m 05s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/1fc0280ff5ef49c0a53814934ada4812 : SUCCESS in 1h 25m 09s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/7b56393e7a90409781efa93fdb569064 : SUCCESS in 2h 13m 03s (non-voting)\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/6e31994394964de497ba450a4c0e8572 : SUCCESS in 1h 06m 30s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a10a3a7cb44f49ddb1c1db4869cf59e2 : SUCCESS in 1h 07m 15s (non-voting)\n- kolla-ansible-rocky-10-upgrade-slurp https://zuul.opendev.org/t/openstack/build/ad6aecf49c614623bbcd1e1e798473ee : SUCCESS in 1h 20m 21s\n- kolla-ansible-ubuntu-noble-upgrade-slurp https://zuul.opendev.org/t/openstack/build/42bcb774a78349e8863649e0c2323e50 : SUCCESS in 1h 19m 42s\n- kolla-ansible-ubuntu-noble-cephadm-upgrade-slurp https://zuul.opendev.org/t/openstack/build/822e8ce7e3df48f0b2d277418d56f146 : SUCCESS in 2h 16m 43s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3186a25d64024586acb39c8994d33417 : SUCCESS in 2m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af2f884a50c347a29c44972c8933fa6b : SUCCESS in 3m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0eb7ab9bf33245149bdd4a14cdcb8325 : SUCCESS in 2m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b0931ffd1384fe596f2bfe245ae32b5 : SUCCESS in 8m 40s","accounts_in_message":[],"_revision_number":30},{"id":"a392a57884d8bbd7eacc65b1dec90cf4fa0d4117","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-27 17:37:09.000000000","message":"Patch Set 30: Backport-Candidate-1 Code-Review+2 Review-Priority+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":30},{"id":"8a7dedcbc6d8a18745da72a7c0e24b827deb14fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 17:37:23.000000000","message":"Patch Set 30: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":30},{"id":"58ea4af95dfc2e853fb5f7a50c4178ab9a0056c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 19:17:47.000000000","message":"Patch Set 30: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9795eee07b664c7babbdb39a5795ec6a\n\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/813ecabfe4a54246a09ebb3e9e628653 : SUCCESS in 2m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3717dd6e71c475ba946610d1d20c5d8 : SUCCESS in 5m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e18c908c1cdb4e7b905a8dde4adc0121 : SUCCESS in 5m 28s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/426984c5fd844c23ac5af200cbf36205 : SUCCESS in 1h 14m 09s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/0839818324fe405c8316505b9d334410 : SUCCESS in 1h 04m 59s\n- kolla-ansible-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/4f5a425690484136b40cb1a788a2f123 : SUCCESS in 47m 36s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/32c3376a7b3d489cb400b1967db7a6c1 : SUCCESS in 55m 05s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/c00505b55f9f4496b77b51fd936d7ce7 : SUCCESS in 1h 38m 58s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/db942db1b2694de8a5ec5d22dd167d1f : SUCCESS in 1h 06m 59s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/30db973defa54790bad30b13516bc555 : SUCCESS in 1h 08m 02s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/5df67e48fec845eabfb6b053ab6e3c96 : SUCCESS in 1h 38m 11s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f34ffd7d96e44e519124b575b5dfc743 : SUCCESS in 1h 08m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c5a047492ea4123980d16ccd4907ff0 : SUCCESS in 2m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0cc1203002474406b5764e94cc6ad624 : SUCCESS in 6m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f804710c84cd410b95fe4bc502a00038 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":30},{"id":"b8c1f8e5d4f68116216455fe43257a6063d48f23","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 19:17:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":30},{"id":"6b58c5db6f39a16c8de206a11c9a2f0aa4403988","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 19:18:45.000000000","message":"Patch Set 30:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b291603ab36543819ee474355860dafd\n\n- promote-deploy-guide https://zuul.opendev.org/t/openstack/build/32df40c0ecab415f85cf44beb4c7bca3 : SUCCESS in 48s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15190ef58e2d42deb1a2bf50af534cd8 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2944844cbf3842ffa87b5b8a55bb0fe9 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":30},{"id":"b199bedcb33bc46c6a44f34efeb8b932f7a7c7d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 20:04:45.000000000","message":"Patch Set 30:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d420aecbb565444a989df62168c58f5d\n\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/fd4c0fe70ab44737bb947daed910736f : FAILURE in 1h 29m 29s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/fa5af88e78f6443a96a16796562cbe2c : TIMED_OUT in 4h 03m 22s (non-voting)\n- kolla-ansible-debian-trixie-aarch64-podman https://zuul.opendev.org/t/openstack/build/f848eafb96f74ad4b172dbba032ad753 : FAILURE in 1h 12m 45s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64-upgrade https://zuul.opendev.org/t/openstack/build/61a324df2c9445efab647c0554e00293 : FAILURE in 58m 05s (non-voting)","accounts_in_message":[],"_revision_number":30}],"current_revision_number":30,"current_revision":"04f1276d652afbe6eb03b184902594666117d631","revisions":{"40af895c51e19d4ac5518ccc1f93c76898048c98":{"kind":"REWORK","_number":1,"created":"2025-11-04 10:24:27.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/21/966021/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/966021/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/1"}}},"commit":{"parents":[{"commit":"cf02acb1a5b370fd4255073c26140646de324eaf","subject":"CI: ironic - stop using tinyipa","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/cf02acb1a5b370fd4255073c26140646de324eaf"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-04 10:22:45.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-04 10:22:56.000000000","tz":60},"subject":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition","message":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition\n\nThe RabbitMQ version compatibility check in the `rabbitmq-version-check`\ntask currently hardcodes the list of explicitly allowed version jumps\n(e.g., 3.13 to 4.0, 4.0 to 4.1).\n\nThis commit introduces a dynamic method to define these allowed upgrade\npaths by loading them from an external file: `vars/rabbitmq_upgrade_paths.yml`.\nThis file is initialized to include the necessary path to support **RabbitMQ 4.2**\nupgrades (e.g., 4.1.x -\u003e 4.2.x).\n\nThis enhancement improves the maintainability of Kolla Ansible by allowing\noperators to update the list of supported version jumps for future RabbitMQ\nreleases without the need to modify the core logic in the Ansible task file.\n\nThe version validation logic now works as follows:\n1. Load explicit version jumps from `rabbitmq_upgrade_paths.yml`.\n2. Check for one minor version increment (e.g., 4.1 -\u003e 4.2).\n3. Check for explicitly defined jumps loaded from the YAML file.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/966017\n\nChange-Id: I0e19cf109eb5665045806b7d925779e82164c2dc\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/40af895c51e19d4ac5518ccc1f93c76898048c98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/40af895c51e19d4ac5518ccc1f93c76898048c98"}]},"branch":"refs/heads/master"},"02f132e8bca35164e3aacd405bb812f216aff5c1":{"kind":"REWORK","_number":2,"created":"2025-11-25 14:51:40.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/21/966021/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/966021/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/2"}}},"commit":{"parents":[{"commit":"1038c3b2f311b969a481af9dd05662e6b3d138ec","subject":"Merge \"ansible-lint: Fix yaml[truthy] in zun role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1038c3b2f311b969a481af9dd05662e6b3d138ec"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-04 10:22:45.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-25 14:51:29.000000000","tz":60},"subject":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition","message":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition\n\nThe RabbitMQ version compatibility check in the `rabbitmq-version-check`\ntask currently hardcodes the list of explicitly allowed version jumps\n(e.g., 3.13 to 4.0, 4.0 to 4.1).\n\nThis commit introduces a dynamic method to define these allowed upgrade\npaths by loading them from an external file: `vars/rabbitmq_upgrade_paths.yml`.\nThis file is initialized to include the necessary path to support **RabbitMQ 4.2**\nupgrades (e.g., 4.1.x -\u003e 4.2.x).\n\nThis enhancement improves the maintainability of Kolla Ansible by allowing\noperators to update the list of supported version jumps for future RabbitMQ\nreleases without the need to modify the core logic in the Ansible task file.\n\nThe version validation logic now works as follows:\n1. Load explicit version jumps from `rabbitmq_upgrade_paths.yml`.\n2. Check for one minor version increment (e.g., 4.1 -\u003e 4.2).\n3. Check for explicitly defined jumps loaded from the YAML file.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/966017\n\nChange-Id: I0e19cf109eb5665045806b7d925779e82164c2dc\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/02f132e8bca35164e3aacd405bb812f216aff5c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/02f132e8bca35164e3aacd405bb812f216aff5c1"}]},"branch":"refs/heads/master"},"e509ba4a00bdaa3f1383765660abfa238ad4c675":{"kind":"REWORK","_number":3,"created":"2025-11-26 06:56:01.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/21/966021/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/966021/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/3"}}},"commit":{"parents":[{"commit":"1038c3b2f311b969a481af9dd05662e6b3d138ec","subject":"Merge \"ansible-lint: Fix yaml[truthy] in zun role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1038c3b2f311b969a481af9dd05662e6b3d138ec"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-04 10:22:45.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-26 06:55:42.000000000","tz":60},"subject":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition","message":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition\n\nThe RabbitMQ version compatibility check in the `rabbitmq-version-check`\ntask currently hardcodes the list of explicitly allowed version jumps\n(e.g., 3.13 to 4.0, 4.0 to 4.1).\n\nThis commit introduces a dynamic method to define these allowed upgrade\npaths by loading them from an external file: `vars/rabbitmq_upgrade_paths.yml`.\nThis file is initialized to include the necessary path to support **RabbitMQ 4.2**\nupgrades (e.g., 4.1.x -\u003e 4.2.x).\n\nThis enhancement improves the maintainability of Kolla Ansible by allowing\noperators to update the list of supported version jumps for future RabbitMQ\nreleases without the need to modify the core logic in the Ansible task file.\n\nThe version validation logic now works as follows:\n1. Load explicit version jumps from `rabbitmq_upgrade_paths.yml`.\n2. Check for one minor version increment (e.g., 4.1 -\u003e 4.2).\n3. Check for explicitly defined jumps loaded from the YAML file.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/966017\n\nChange-Id: I0e19cf109eb5665045806b7d925779e82164c2dc\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/e509ba4a00bdaa3f1383765660abfa238ad4c675"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/e509ba4a00bdaa3f1383765660abfa238ad4c675"}]},"branch":"refs/heads/master"},"9cd8e62442ecdeea255c5c3a9617ee3ad95041bc":{"kind":"REWORK","_number":4,"created":"2025-11-26 09:44:16.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/21/966021/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/966021/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/4"}}},"commit":{"parents":[{"commit":"1038c3b2f311b969a481af9dd05662e6b3d138ec","subject":"Merge \"ansible-lint: Fix yaml[truthy] in zun role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1038c3b2f311b969a481af9dd05662e6b3d138ec"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-04 10:22:45.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-26 09:44:03.000000000","tz":60},"subject":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition","message":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition\n\nThe RabbitMQ version compatibility check in the `rabbitmq-version-check`\ntask currently hardcodes the list of explicitly allowed version jumps\n(e.g., 3.13 to 4.0, 4.0 to 4.1).\n\nThis commit introduces a dynamic method to define these allowed upgrade\npaths by loading them from an external file: `vars/rabbitmq_upgrade_paths.yml`.\nThis file is initialized to include the necessary path to support **RabbitMQ 4.2**\nupgrades (e.g., 4.1.x -\u003e 4.2.x).\n\nThis enhancement improves the maintainability of Kolla Ansible by allowing\noperators to update the list of supported version jumps for future RabbitMQ\nreleases without the need to modify the core logic in the Ansible task file.\n\nThe version validation logic now works as follows:\n1. Load explicit version jumps from `rabbitmq_upgrade_paths.yml`.\n2. Check for one minor version increment (e.g., 4.1 -\u003e 4.2).\n3. Check for explicitly defined jumps loaded from the YAML file.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/966017\n\nChange-Id: I0e19cf109eb5665045806b7d925779e82164c2dc\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/9cd8e62442ecdeea255c5c3a9617ee3ad95041bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/9cd8e62442ecdeea255c5c3a9617ee3ad95041bc"}]},"branch":"refs/heads/master"},"8dbefc6b0e5ae64399c8e5e104d70fd1f18b07eb":{"kind":"REWORK","_number":5,"created":"2025-11-26 13:36:55.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/21/966021/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/966021/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/5"}}},"commit":{"parents":[{"commit":"1038c3b2f311b969a481af9dd05662e6b3d138ec","subject":"Merge \"ansible-lint: Fix yaml[truthy] in zun role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1038c3b2f311b969a481af9dd05662e6b3d138ec"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-04 10:22:45.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-26 13:36:43.000000000","tz":60},"subject":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition","message":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition\n\nThe RabbitMQ version compatibility check in the `rabbitmq-version-check`\ntask currently hardcodes the list of explicitly allowed version jumps\n(e.g., 3.13 to 4.0, 4.0 to 4.1).\n\nThis commit introduces a dynamic method to define these allowed upgrade\npaths by loading them from an external file: `vars/rabbitmq_upgrade_paths.yml`.\nThis file is initialized to include the necessary path to support **RabbitMQ 4.2**\nupgrades (e.g., 4.1.x -\u003e 4.2.x).\n\nThis enhancement improves the maintainability of Kolla Ansible by allowing\noperators to update the list of supported version jumps for future RabbitMQ\nreleases without the need to modify the core logic in the Ansible task file.\n\nThe version validation logic now works as follows:\n1. Load explicit version jumps from `rabbitmq_upgrade_paths.yml`.\n2. Check for one minor version increment (e.g., 4.1 -\u003e 4.2).\n3. Check for explicitly defined jumps loaded from the YAML file.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/966017\n\nChange-Id: I0e19cf109eb5665045806b7d925779e82164c2dc\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/8dbefc6b0e5ae64399c8e5e104d70fd1f18b07eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/8dbefc6b0e5ae64399c8e5e104d70fd1f18b07eb"}]},"branch":"refs/heads/master"},"19e90608a8045fecaf1b8e30940e5ff7a8e3a94b":{"kind":"REWORK","_number":6,"created":"2025-11-26 16:02:39.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/21/966021/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/966021/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/6"}}},"commit":{"parents":[{"commit":"1038c3b2f311b969a481af9dd05662e6b3d138ec","subject":"Merge \"ansible-lint: Fix yaml[truthy] in zun role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1038c3b2f311b969a481af9dd05662e6b3d138ec"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-04 10:22:45.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-26 16:02:23.000000000","tz":60},"subject":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition","message":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition\n\nThe RabbitMQ version compatibility check in the `rabbitmq-version-check`\ntask currently hardcodes the list of explicitly allowed version jumps\n(e.g., 3.13 to 4.0, 4.0 to 4.1).\n\nThis commit introduces a dynamic method to define these allowed upgrade\npaths by loading them from an external file: `vars/rabbitmq_upgrade_paths.yml`.\nThis file is initialized to include the necessary path to support **RabbitMQ 4.2**\nupgrades (e.g., 4.1.x -\u003e 4.2.x).\n\nThis enhancement improves the maintainability of Kolla Ansible by allowing\noperators to update the list of supported version jumps for future RabbitMQ\nreleases without the need to modify the core logic in the Ansible task file.\n\nThe version validation logic now works as follows:\n1. Load explicit version jumps from `rabbitmq_upgrade_paths.yml`.\n2. Check for one minor version increment (e.g., 4.1 -\u003e 4.2).\n3. Check for explicitly defined jumps loaded from the YAML file.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/966017\n\nChange-Id: I0e19cf109eb5665045806b7d925779e82164c2dc\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/19e90608a8045fecaf1b8e30940e5ff7a8e3a94b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/19e90608a8045fecaf1b8e30940e5ff7a8e3a94b"}]},"branch":"refs/heads/master"},"bc0870c431cd5b884a8ed5542fbe6f542d96861c":{"kind":"REWORK","_number":7,"created":"2025-11-27 08:42:26.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/21/966021/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/966021/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/7"}}},"commit":{"parents":[{"commit":"1038c3b2f311b969a481af9dd05662e6b3d138ec","subject":"Merge \"ansible-lint: Fix yaml[truthy] in zun role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1038c3b2f311b969a481af9dd05662e6b3d138ec"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-04 10:22:45.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-27 08:42:14.000000000","tz":60},"subject":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition","message":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition\n\nThe RabbitMQ version compatibility check in the `rabbitmq-version-check`\ntask currently hardcodes the list of explicitly allowed version jumps\n(e.g., 3.13 to 4.0, 4.0 to 4.1).\n\nThis commit introduces a dynamic method to define these allowed upgrade\npaths by loading them from an external file: `vars/rabbitmq_upgrade_paths.yml`.\nThis file is initialized to include the necessary path to support **RabbitMQ 4.2**\nupgrades (e.g., 4.1.x -\u003e 4.2.x).\n\nThis enhancement improves the maintainability of Kolla Ansible by allowing\noperators to update the list of supported version jumps for future RabbitMQ\nreleases without the need to modify the core logic in the Ansible task file.\n\nThe version validation logic now works as follows:\n1. Load explicit version jumps from `rabbitmq_upgrade_paths.yml`.\n2. Check for one minor version increment (e.g., 4.1 -\u003e 4.2).\n3. Check for explicitly defined jumps loaded from the YAML file.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/966017\n\nChange-Id: I0e19cf109eb5665045806b7d925779e82164c2dc\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/bc0870c431cd5b884a8ed5542fbe6f542d96861c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/bc0870c431cd5b884a8ed5542fbe6f542d96861c"}]},"branch":"refs/heads/master"},"735de116f686d8a47b0f97e0f58883ce6e90bf6d":{"kind":"REWORK","_number":8,"created":"2025-11-27 10:34:46.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/21/966021/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/966021/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/8"}}},"commit":{"parents":[{"commit":"1038c3b2f311b969a481af9dd05662e6b3d138ec","subject":"Merge \"ansible-lint: Fix yaml[truthy] in zun role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1038c3b2f311b969a481af9dd05662e6b3d138ec"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-04 10:22:45.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-27 10:34:35.000000000","tz":60},"subject":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition","message":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition\n\nThe RabbitMQ version compatibility check in the `rabbitmq-version-check`\ntask currently hardcodes the list of explicitly allowed version jumps\n(e.g., 3.13 to 4.0, 4.0 to 4.1).\n\nThis commit introduces a dynamic method to define these allowed upgrade\npaths by loading them from an external file: `vars/rabbitmq_upgrade_paths.yml`.\nThis file is initialized to include the necessary path to support **RabbitMQ 4.2**\nupgrades (e.g., 4.1.x -\u003e 4.2.x).\n\nThis enhancement improves the maintainability of Kolla Ansible by allowing\noperators to update the list of supported version jumps for future RabbitMQ\nreleases without the need to modify the core logic in the Ansible task file.\n\nThe version validation logic now works as follows:\n1. Load explicit version jumps from `rabbitmq_upgrade_paths.yml`.\n2. Check for one minor version increment (e.g., 4.1 -\u003e 4.2).\n3. Check for explicitly defined jumps loaded from the YAML file.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/966017\n\nChange-Id: I0e19cf109eb5665045806b7d925779e82164c2dc\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/735de116f686d8a47b0f97e0f58883ce6e90bf6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/735de116f686d8a47b0f97e0f58883ce6e90bf6d"}]},"branch":"refs/heads/master"},"7ec03e3d8a82c9a38c354c6e46336fcc8501c1c6":{"kind":"REWORK","_number":9,"created":"2025-11-27 13:29:23.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/21/966021/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/966021/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/9"}}},"commit":{"parents":[{"commit":"1038c3b2f311b969a481af9dd05662e6b3d138ec","subject":"Merge \"ansible-lint: Fix yaml[truthy] in zun role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1038c3b2f311b969a481af9dd05662e6b3d138ec"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-04 10:22:45.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-27 13:29:11.000000000","tz":60},"subject":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition","message":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition\n\nThe RabbitMQ version compatibility check in the `rabbitmq-version-check`\ntask currently hardcodes the list of explicitly allowed version jumps\n(e.g., 3.13 to 4.0, 4.0 to 4.1).\n\nThis commit introduces a dynamic method to define these allowed upgrade\npaths by loading them from an external file: `vars/rabbitmq_upgrade_paths.yml`.\nThis file is initialized to include the necessary path to support **RabbitMQ 4.2**\nupgrades (e.g., 4.1.x -\u003e 4.2.x).\n\nThis enhancement improves the maintainability of Kolla Ansible by allowing\noperators to update the list of supported version jumps for future RabbitMQ\nreleases without the need to modify the core logic in the Ansible task file.\n\nThe version validation logic now works as follows:\n1. Load explicit version jumps from `rabbitmq_upgrade_paths.yml`.\n2. Check for one minor version increment (e.g., 4.1 -\u003e 4.2).\n3. Check for explicitly defined jumps loaded from the YAML file.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/966017\n\nChange-Id: I0e19cf109eb5665045806b7d925779e82164c2dc\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/7ec03e3d8a82c9a38c354c6e46336fcc8501c1c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/7ec03e3d8a82c9a38c354c6e46336fcc8501c1c6"}]},"branch":"refs/heads/master"},"5489e3078e2cba1d5f1f66c23eac54093bc33b83":{"kind":"REWORK","_number":10,"created":"2025-11-27 15:33:51.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/21/966021/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/966021/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/10"}}},"commit":{"parents":[{"commit":"1038c3b2f311b969a481af9dd05662e6b3d138ec","subject":"Merge \"ansible-lint: Fix yaml[truthy] in zun role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1038c3b2f311b969a481af9dd05662e6b3d138ec"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-04 10:22:45.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-27 15:33:41.000000000","tz":60},"subject":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition","message":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition\n\nThe RabbitMQ version compatibility check in the `rabbitmq-version-check`\ntask currently hardcodes the list of explicitly allowed version jumps\n(e.g., 3.13 to 4.0, 4.0 to 4.1).\n\nThis commit introduces a dynamic method to define these allowed upgrade\npaths by loading them from an external file: `vars/rabbitmq_upgrade_paths.yml`.\nThis file is initialized to include the necessary path to support **RabbitMQ 4.2**\nupgrades (e.g., 4.1.x -\u003e 4.2.x).\n\nThis enhancement improves the maintainability of Kolla Ansible by allowing\noperators to update the list of supported version jumps for future RabbitMQ\nreleases without the need to modify the core logic in the Ansible task file.\n\nThe version validation logic now works as follows:\n1. Load explicit version jumps from `rabbitmq_upgrade_paths.yml`.\n2. Check for one minor version increment (e.g., 4.1 -\u003e 4.2).\n3. Check for explicitly defined jumps loaded from the YAML file.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/966017\n\nChange-Id: I0e19cf109eb5665045806b7d925779e82164c2dc\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/5489e3078e2cba1d5f1f66c23eac54093bc33b83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/5489e3078e2cba1d5f1f66c23eac54093bc33b83"}]},"branch":"refs/heads/master"},"041aa87ffbc216522ebe0a7103ae7b8912ac0d22":{"kind":"REWORK","_number":11,"created":"2025-11-27 17:11:24.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/21/966021/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/966021/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/11"}}},"commit":{"parents":[{"commit":"1038c3b2f311b969a481af9dd05662e6b3d138ec","subject":"Merge \"ansible-lint: Fix yaml[truthy] in zun role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1038c3b2f311b969a481af9dd05662e6b3d138ec"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-04 10:22:45.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-27 17:11:12.000000000","tz":60},"subject":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition","message":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition\n\nThe RabbitMQ version compatibility check in the `rabbitmq-version-check`\ntask currently hardcodes the list of explicitly allowed version jumps\n(e.g., 3.13 to 4.0, 4.0 to 4.1).\n\nThis commit introduces a dynamic method to define these allowed upgrade\npaths by loading them from an external file: `vars/rabbitmq_upgrade_paths.yml`.\nThis file is initialized to include the necessary path to support **RabbitMQ 4.2**\nupgrades (e.g., 4.1.x -\u003e 4.2.x).\n\nThis enhancement improves the maintainability of Kolla Ansible by allowing\noperators to update the list of supported version jumps for future RabbitMQ\nreleases without the need to modify the core logic in the Ansible task file.\n\nThe version validation logic now works as follows:\n1. Load explicit version jumps from `rabbitmq_upgrade_paths.yml`.\n2. Check for one minor version increment (e.g., 4.1 -\u003e 4.2).\n3. Check for explicitly defined jumps loaded from the YAML file.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/966017\n\nChange-Id: I0e19cf109eb5665045806b7d925779e82164c2dc\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/041aa87ffbc216522ebe0a7103ae7b8912ac0d22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/041aa87ffbc216522ebe0a7103ae7b8912ac0d22"}]},"branch":"refs/heads/master"},"45bbc669eccd47d89db6e0ad85c2477d2e01bb6e":{"kind":"REWORK","_number":12,"created":"2025-11-27 19:10:00.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/21/966021/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/966021/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/12"}}},"commit":{"parents":[{"commit":"1038c3b2f311b969a481af9dd05662e6b3d138ec","subject":"Merge \"ansible-lint: Fix yaml[truthy] in zun role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1038c3b2f311b969a481af9dd05662e6b3d138ec"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-04 10:22:45.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-27 19:09:48.000000000","tz":60},"subject":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition","message":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition\n\nThe RabbitMQ version compatibility check in the `rabbitmq-version-check`\ntask currently hardcodes the list of explicitly allowed version jumps\n(e.g., 3.13 to 4.0, 4.0 to 4.1).\n\nThis commit introduces a dynamic method to define these allowed upgrade\npaths by loading them from an external file: `vars/rabbitmq_upgrade_paths.yml`.\nThis file is initialized to include the necessary path to support **RabbitMQ 4.2**\nupgrades (e.g., 4.1.x -\u003e 4.2.x).\n\nThis enhancement improves the maintainability of Kolla Ansible by allowing\noperators to update the list of supported version jumps for future RabbitMQ\nreleases without the need to modify the core logic in the Ansible task file.\n\nThe version validation logic now works as follows:\n1. Load explicit version jumps from `rabbitmq_upgrade_paths.yml`.\n2. Check for one minor version increment (e.g., 4.1 -\u003e 4.2).\n3. Check for explicitly defined jumps loaded from the YAML file.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/966017\n\nChange-Id: I0e19cf109eb5665045806b7d925779e82164c2dc\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/45bbc669eccd47d89db6e0ad85c2477d2e01bb6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/45bbc669eccd47d89db6e0ad85c2477d2e01bb6e"}]},"branch":"refs/heads/master"},"6401db9b39f43ea72645295ed0e8cce779f856f5":{"kind":"REWORK","_number":13,"created":"2025-11-28 09:29:24.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/21/966021/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/966021/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/13"}}},"commit":{"parents":[{"commit":"1038c3b2f311b969a481af9dd05662e6b3d138ec","subject":"Merge \"ansible-lint: Fix yaml[truthy] in zun role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1038c3b2f311b969a481af9dd05662e6b3d138ec"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-04 10:22:45.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-28 09:29:12.000000000","tz":60},"subject":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition","message":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition\n\nThe RabbitMQ version compatibility check in the `rabbitmq-version-check`\ntask currently hardcodes the list of explicitly allowed version jumps\n(e.g., 3.13 to 4.0, 4.0 to 4.1).\n\nThis commit introduces a dynamic method to define these allowed upgrade\npaths by loading them from an external file: `vars/rabbitmq_upgrade_paths.yml`.\nThis file is initialized to include the necessary path to support **RabbitMQ 4.2**\nupgrades (e.g., 4.1.x -\u003e 4.2.x).\n\nThis enhancement improves the maintainability of Kolla Ansible by allowing\noperators to update the list of supported version jumps for future RabbitMQ\nreleases without the need to modify the core logic in the Ansible task file.\n\nThe version validation logic now works as follows:\n1. Load explicit version jumps from `rabbitmq_upgrade_paths.yml`.\n2. Check for one minor version increment (e.g., 4.1 -\u003e 4.2).\n3. Check for explicitly defined jumps loaded from the YAML file.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/966017\n\nChange-Id: I0e19cf109eb5665045806b7d925779e82164c2dc\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/6401db9b39f43ea72645295ed0e8cce779f856f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/6401db9b39f43ea72645295ed0e8cce779f856f5"}]},"branch":"refs/heads/master"},"d9af76c6492e0775bf02cf3626d2a2459604086b":{"kind":"REWORK","_number":14,"created":"2025-11-28 11:42:25.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/21/966021/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/966021/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/14"}}},"commit":{"parents":[{"commit":"1038c3b2f311b969a481af9dd05662e6b3d138ec","subject":"Merge \"ansible-lint: Fix yaml[truthy] in zun role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1038c3b2f311b969a481af9dd05662e6b3d138ec"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-04 10:22:45.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-28 11:42:14.000000000","tz":60},"subject":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition","message":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition\n\nThe RabbitMQ version compatibility check in the `rabbitmq-version-check`\ntask currently hardcodes the list of explicitly allowed version jumps\n(e.g., 3.13 to 4.0, 4.0 to 4.1).\n\nThis commit introduces a dynamic method to define these allowed upgrade\npaths by loading them from an external file: `vars/rabbitmq_upgrade_paths.yml`.\nThis file is initialized to include the necessary path to support **RabbitMQ 4.2**\nupgrades (e.g., 4.1.x -\u003e 4.2.x).\n\nThis enhancement improves the maintainability of Kolla Ansible by allowing\noperators to update the list of supported version jumps for future RabbitMQ\nreleases without the need to modify the core logic in the Ansible task file.\n\nThe version validation logic now works as follows:\n1. Load explicit version jumps from `rabbitmq_upgrade_paths.yml`.\n2. Check for one minor version increment (e.g., 4.1 -\u003e 4.2).\n3. Check for explicitly defined jumps loaded from the YAML file.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/966017\n\nChange-Id: I0e19cf109eb5665045806b7d925779e82164c2dc\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/d9af76c6492e0775bf02cf3626d2a2459604086b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/d9af76c6492e0775bf02cf3626d2a2459604086b"}]},"branch":"refs/heads/master"},"68a49e607f19e52c8b32e7ff9e1cdf82fd5529c6":{"kind":"REWORK","_number":15,"created":"2025-11-28 14:02:36.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/21/966021/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/966021/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/15"}}},"commit":{"parents":[{"commit":"1038c3b2f311b969a481af9dd05662e6b3d138ec","subject":"Merge \"ansible-lint: Fix yaml[truthy] in zun role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1038c3b2f311b969a481af9dd05662e6b3d138ec"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-04 10:22:45.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-28 14:02:26.000000000","tz":60},"subject":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition","message":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition\n\nThe RabbitMQ version compatibility check in the `rabbitmq-version-check`\ntask currently hardcodes the list of explicitly allowed version jumps\n(e.g., 3.13 to 4.0, 4.0 to 4.1).\n\nThis commit introduces a dynamic method to define these allowed upgrade\npaths by loading them from an external file: `vars/rabbitmq_upgrade_paths.yml`.\nThis file is initialized to include the necessary path to support **RabbitMQ 4.2**\nupgrades (e.g., 4.1.x -\u003e 4.2.x).\n\nThis enhancement improves the maintainability of Kolla Ansible by allowing\noperators to update the list of supported version jumps for future RabbitMQ\nreleases without the need to modify the core logic in the Ansible task file.\n\nThe version validation logic now works as follows:\n1. Load explicit version jumps from `rabbitmq_upgrade_paths.yml`.\n2. Check for one minor version increment (e.g., 4.1 -\u003e 4.2).\n3. Check for explicitly defined jumps loaded from the YAML file.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/966017\n\nChange-Id: I0e19cf109eb5665045806b7d925779e82164c2dc\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/68a49e607f19e52c8b32e7ff9e1cdf82fd5529c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/68a49e607f19e52c8b32e7ff9e1cdf82fd5529c6"}]},"branch":"refs/heads/master"},"27280ea0467b7f30d64971d09cdb3137dfeaa710":{"kind":"REWORK","_number":16,"created":"2025-11-28 16:48:16.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/21/966021/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/966021/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/16"}}},"commit":{"parents":[{"commit":"1038c3b2f311b969a481af9dd05662e6b3d138ec","subject":"Merge \"ansible-lint: Fix yaml[truthy] in zun role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1038c3b2f311b969a481af9dd05662e6b3d138ec"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-04 10:22:45.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-28 16:48:05.000000000","tz":60},"subject":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition","message":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition\n\nThe RabbitMQ version compatibility check in the `rabbitmq-version-check`\ntask currently hardcodes the list of explicitly allowed version jumps\n(e.g., 3.13 to 4.0, 4.0 to 4.1).\n\nThis commit introduces a dynamic method to define these allowed upgrade\npaths by loading them from an external file: `vars/rabbitmq_upgrade_paths.yml`.\nThis file is initialized to include the necessary path to support **RabbitMQ 4.2**\nupgrades (e.g., 4.1.x -\u003e 4.2.x).\n\nThis enhancement improves the maintainability of Kolla Ansible by allowing\noperators to update the list of supported version jumps for future RabbitMQ\nreleases without the need to modify the core logic in the Ansible task file.\n\nThe version validation logic now works as follows:\n1. Load explicit version jumps from `rabbitmq_upgrade_paths.yml`.\n2. Check for one minor version increment (e.g., 4.1 -\u003e 4.2).\n3. Check for explicitly defined jumps loaded from the YAML file.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/966017\n\nChange-Id: I0e19cf109eb5665045806b7d925779e82164c2dc\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/27280ea0467b7f30d64971d09cdb3137dfeaa710"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/27280ea0467b7f30d64971d09cdb3137dfeaa710"}]},"branch":"refs/heads/master"},"5a6c33cb9833c8853f136468ccf151d078b4a35c":{"kind":"REWORK","_number":17,"created":"2025-11-28 19:19:43.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/21/966021/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/966021/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/17"}}},"commit":{"parents":[{"commit":"1038c3b2f311b969a481af9dd05662e6b3d138ec","subject":"Merge \"ansible-lint: Fix yaml[truthy] in zun role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1038c3b2f311b969a481af9dd05662e6b3d138ec"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-04 10:22:45.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-28 19:19:23.000000000","tz":60},"subject":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition","message":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition\n\nThe RabbitMQ version compatibility check in the `rabbitmq-version-check`\ntask currently hardcodes the list of explicitly allowed version jumps\n(e.g., 3.13 to 4.0, 4.0 to 4.1).\n\nThis commit introduces a dynamic method to define these allowed upgrade\npaths by loading them from an external file: `vars/rabbitmq_upgrade_paths.yml`.\nThis file is initialized to include the necessary path to support **RabbitMQ 4.2**\nupgrades (e.g., 4.1.x -\u003e 4.2.x).\n\nThis enhancement improves the maintainability of Kolla Ansible by allowing\noperators to update the list of supported version jumps for future RabbitMQ\nreleases without the need to modify the core logic in the Ansible task file.\n\nThe version validation logic now works as follows:\n1. Load explicit version jumps from `rabbitmq_upgrade_paths.yml`.\n2. Check for one minor version increment (e.g., 4.1 -\u003e 4.2).\n3. Check for explicitly defined jumps loaded from the YAML file.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/966017\n\nChange-Id: I0e19cf109eb5665045806b7d925779e82164c2dc\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/5a6c33cb9833c8853f136468ccf151d078b4a35c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/5a6c33cb9833c8853f136468ccf151d078b4a35c"}]},"branch":"refs/heads/master"},"107ad4ec277d23755bfc19cd3ce77b4b450a22cf":{"kind":"REWORK","_number":18,"created":"2025-11-28 19:33:54.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/21/966021/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/966021/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/18"}}},"commit":{"parents":[{"commit":"1038c3b2f311b969a481af9dd05662e6b3d138ec","subject":"Merge \"ansible-lint: Fix yaml[truthy] in zun role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1038c3b2f311b969a481af9dd05662e6b3d138ec"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-04 10:22:45.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-28 19:33:41.000000000","tz":60},"subject":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition","message":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition\n\nThe RabbitMQ version compatibility check in the `rabbitmq-version-check`\ntask currently hardcodes the list of explicitly allowed version jumps\n(e.g., 3.13 to 4.0, 4.0 to 4.1).\n\nThis commit introduces a dynamic method to define these allowed upgrade\npaths by loading them from an external file: `vars/rabbitmq_upgrade_paths.yml`.\nThis file is initialized to include the necessary path to support **RabbitMQ 4.2**\nupgrades (e.g., 4.1.x -\u003e 4.2.x).\n\nThis enhancement improves the maintainability of Kolla Ansible by allowing\noperators to update the list of supported version jumps for future RabbitMQ\nreleases without the need to modify the core logic in the Ansible task file.\n\nThe version validation logic now works as follows:\n1. Load explicit version jumps from `rabbitmq_upgrade_paths.yml`.\n2. Check for one minor version increment (e.g., 4.1 -\u003e 4.2).\n3. Check for explicitly defined jumps loaded from the YAML file.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/966017\n\nChange-Id: I0e19cf109eb5665045806b7d925779e82164c2dc\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/107ad4ec277d23755bfc19cd3ce77b4b450a22cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/107ad4ec277d23755bfc19cd3ce77b4b450a22cf"}]},"branch":"refs/heads/master"},"30b82da4fb46edd3557467b06faa37d088bdb79b":{"kind":"REWORK","_number":19,"created":"2025-12-01 14:06:19.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/21/966021/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/966021/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/19"}}},"commit":{"parents":[{"commit":"1038c3b2f311b969a481af9dd05662e6b3d138ec","subject":"Merge \"ansible-lint: Fix yaml[truthy] in zun role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1038c3b2f311b969a481af9dd05662e6b3d138ec"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-04 10:22:45.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-12-01 14:05:49.000000000","tz":60},"subject":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition","message":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition\n\nThe RabbitMQ version compatibility check in the `rabbitmq-version-check`\ntask currently hardcodes the list of explicitly allowed version jumps\n(e.g., 3.13 to 4.0, 4.0 to 4.1).\n\nThis commit introduces a dynamic method to define these allowed upgrade\npaths by loading them from an external file: `vars/rabbitmq_upgrade_paths.yml`.\nThis file is initialized to include the necessary path to support **RabbitMQ 4.2**\nupgrades (e.g., 4.1.x -\u003e 4.2.x).\n\nThis enhancement improves the maintainability of Kolla Ansible by allowing\noperators to update the list of supported version jumps for future RabbitMQ\nreleases without the need to modify the core logic in the Ansible task file.\n\nThe version validation logic now works as follows:\n1. Load explicit version jumps from `rabbitmq_upgrade_paths.yml`.\n2. Check for one minor version increment (e.g., 4.1 -\u003e 4.2).\n3. Check for explicitly defined jumps loaded from the YAML file.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/966017\n\nChange-Id: I0e19cf109eb5665045806b7d925779e82164c2dc\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/30b82da4fb46edd3557467b06faa37d088bdb79b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/30b82da4fb46edd3557467b06faa37d088bdb79b"}]},"branch":"refs/heads/master"},"7eff3aa7fc7dc4d9056c392bad063211cf954834":{"kind":"REWORK","_number":20,"created":"2026-02-10 13:28:56.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/21/966021/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/966021/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/20"}}},"commit":{"parents":[{"commit":"120aa1eef97cdfe8da6a769fcbcd0c37414d66c9","subject":"Merge \"ironic: Drop legacy upgrade option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/120aa1eef97cdfe8da6a769fcbcd0c37414d66c9"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-04 10:22:45.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-02-10 13:28:53.000000000","tz":60},"subject":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition","message":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition\n\nThe RabbitMQ version compatibility check in the `rabbitmq-version-check`\ntask currently hardcodes the list of explicitly allowed version jumps\n(e.g., 3.13 to 4.0, 4.0 to 4.1).\n\nThis commit introduces a dynamic method to define these allowed upgrade\npaths by loading them from an external file: `vars/rabbitmq_upgrade_paths.yml`.\nThis file is initialized to include the necessary path to support **RabbitMQ 4.2**\nupgrades (e.g., 4.1.x -\u003e 4.2.x).\n\nThis enhancement improves the maintainability of Kolla Ansible by allowing\noperators to update the list of supported version jumps for future RabbitMQ\nreleases without the need to modify the core logic in the Ansible task file.\n\nThe version validation logic now works as follows:\n1. Load explicit version jumps from `rabbitmq_upgrade_paths.yml`.\n2. Check for one minor version increment (e.g., 4.1 -\u003e 4.2).\n3. Check for explicitly defined jumps loaded from the YAML file.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/966017\n\nChange-Id: I0e19cf109eb5665045806b7d925779e82164c2dc\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/7eff3aa7fc7dc4d9056c392bad063211cf954834"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/7eff3aa7fc7dc4d9056c392bad063211cf954834"}]},"branch":"refs/heads/master"},"7737269c7d53a16eb595c757847a8e52403b533f":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2026-02-12 07:30:13.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/21/966021/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/966021/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/21"}}},"commit":{"parents":[{"commit":"179c4ba45f0f912e319ce114bbf0a0835506c697","subject":"Merge \"CI: use 3 ovn sb db relays for better coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/179c4ba45f0f912e319ce114bbf0a0835506c697"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-04 10:22:45.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-02-12 07:30:09.000000000","tz":60},"subject":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition","message":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition\n\nThe RabbitMQ version compatibility check in the `rabbitmq-version-check`\ntask currently hardcodes the list of explicitly allowed version jumps\n(e.g., 3.13 to 4.0, 4.0 to 4.1).\n\nThis commit introduces a dynamic method to define these allowed upgrade\npaths by loading them from an external file: `vars/rabbitmq_upgrade_paths.yml`.\nThis file is initialized to include the necessary path to support **RabbitMQ 4.2**\nupgrades (e.g., 4.1.x -\u003e 4.2.x).\n\nThis enhancement improves the maintainability of Kolla Ansible by allowing\noperators to update the list of supported version jumps for future RabbitMQ\nreleases without the need to modify the core logic in the Ansible task file.\n\nThe version validation logic now works as follows:\n1. Load explicit version jumps from `rabbitmq_upgrade_paths.yml`.\n2. Check for one minor version increment (e.g., 4.1 -\u003e 4.2).\n3. Check for explicitly defined jumps loaded from the YAML file.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/966017\n\nChange-Id: I0e19cf109eb5665045806b7d925779e82164c2dc\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/7737269c7d53a16eb595c757847a8e52403b533f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/7737269c7d53a16eb595c757847a8e52403b533f"}]},"branch":"refs/heads/master"},"79b7a06ef46a093a4a9d5945cb22e6cba616c80b":{"kind":"REWORK","_number":22,"created":"2026-03-17 09:46:14.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/21/966021/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/966021/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/22"}}},"commit":{"parents":[{"commit":"4d0c1d0f29aef63c860b72dffa0eabb1011108ea","subject":"Merge \"Ironic parameter rework and default NTP server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/4d0c1d0f29aef63c860b72dffa0eabb1011108ea"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-04 10:22:45.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-03-17 09:46:10.000000000","tz":60},"subject":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition","message":"kolla-ansible: Add RabbitMQ 4.2 support by allowing external upgrade path definition\n\nThe RabbitMQ version compatibility check in the `rabbitmq-version-check`\ntask currently hardcodes the list of explicitly allowed version jumps\n(e.g., 3.13 to 4.0, 4.0 to 4.1).\n\nThis commit introduces a dynamic method to define these allowed upgrade\npaths by loading them from an external file: `vars/rabbitmq_upgrade_paths.yml`.\nThis file is initialized to include the necessary path to support **RabbitMQ 4.2**\nupgrades (e.g., 4.1.x -\u003e 4.2.x).\n\nThis enhancement improves the maintainability of Kolla Ansible by allowing\noperators to update the list of supported version jumps for future RabbitMQ\nreleases without the need to modify the core logic in the Ansible task file.\n\nThe version validation logic now works as follows:\n1. Load explicit version jumps from `rabbitmq_upgrade_paths.yml`.\n2. Check for one minor version increment (e.g., 4.1 -\u003e 4.2).\n3. Check for explicitly defined jumps loaded from the YAML file.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/966017\n\nChange-Id: I0e19cf109eb5665045806b7d925779e82164c2dc\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/79b7a06ef46a093a4a9d5945cb22e6cba616c80b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/79b7a06ef46a093a4a9d5945cb22e6cba616c80b"}]},"branch":"refs/heads/master"},"bcac93fbac71de0ea0b77235aa076f0df3efab4a":{"kind":"REWORK","_number":23,"created":"2026-03-17 10:20:01.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/21/966021/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/966021/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/23"}}},"commit":{"parents":[{"commit":"4d0c1d0f29aef63c860b72dffa0eabb1011108ea","subject":"Merge \"Ironic parameter rework and default NTP server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/4d0c1d0f29aef63c860b72dffa0eabb1011108ea"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-04 10:22:45.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-03-17 10:19:46.000000000","tz":60},"subject":"kolla-ansible: Add RabbitMQ 4.2 support and dynamic upgrade paths","message":"kolla-ansible: Add RabbitMQ 4.2 support and dynamic upgrade paths\n\nThe RabbitMQ version check currently hardcodes allowed version jumps.\nThis makes supporting new major/minor releases (like 4.2) rigid and\nrequires modifying the core task file.\n\nThis commit introduces the `rabbitmq_allowed_upgrades` variable in\nthe role\u0027s defaults. This allows operators to define explicit version\njumps (e.g., 3.13 to 4.2) and makes the upgrade logic more flexible.\n\nThe version validation logic now:\n1. Checks for standard minor increments (e.g., 4.1 -\u003e 4.2).\n2. Checks for explicitly defined jumps in `rabbitmq_allowed_upgrades`.\n\nBy moving these definitions to defaults, operators can easily override\nthem in globals.yml if necessary.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/966017\n\nChange-Id: I0e19cf109eb5665045806b7d925779e82164c2dc\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/bcac93fbac71de0ea0b77235aa076f0df3efab4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/bcac93fbac71de0ea0b77235aa076f0df3efab4a"}]},"branch":"refs/heads/master"},"84e73a3f6ce1b1f548ad4a3e4dcc294ce03db92d":{"kind":"REWORK","_number":24,"created":"2026-03-17 10:33:09.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/21/966021/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/966021/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/24"}}},"commit":{"parents":[{"commit":"4d0c1d0f29aef63c860b72dffa0eabb1011108ea","subject":"Merge \"Ironic parameter rework and default NTP server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/4d0c1d0f29aef63c860b72dffa0eabb1011108ea"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-04 10:22:45.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-03-17 10:32:59.000000000","tz":60},"subject":"kolla-ansible: Add RabbitMQ 4.2 support and dynamic upgrade paths","message":"kolla-ansible: Add RabbitMQ 4.2 support and dynamic upgrade paths\n\nThe RabbitMQ version check currently hardcodes allowed version jumps.\nThis makes supporting new major/minor releases (like 4.2) rigid and\nrequires modifying the core task file.\n\nThis commit introduces the `rabbitmq_allowed_upgrades` variable in\nthe role\u0027s defaults. This allows operators to define explicit version\njumps (e.g., 3.13 to 4.2) and makes the upgrade logic more flexible.\n\nThe version validation logic now:\n1. Checks for standard minor increments (e.g., 4.1 -\u003e 4.2).\n2. Checks for explicitly defined jumps in `rabbitmq_allowed_upgrades`.\n\nBy moving these definitions to defaults, operators can easily override\nthem in globals.yml if necessary.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/966017\n\nChange-Id: I0e19cf109eb5665045806b7d925779e82164c2dc\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/84e73a3f6ce1b1f548ad4a3e4dcc294ce03db92d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/84e73a3f6ce1b1f548ad4a3e4dcc294ce03db92d"}]},"branch":"refs/heads/master"},"1e2a0b62066b084d3767756e26cce384d11b5224":{"kind":"REWORK","_number":25,"created":"2026-03-27 10:57:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/966021/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/966021/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/25"}}},"commit":{"parents":[{"commit":"0812604cebab71eb804fbf50d5d6e1927b2629df","subject":"Merge \"cloudkitty: Switch to uWSGI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/0812604cebab71eb804fbf50d5d6e1927b2629df"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-04 10:22:45.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-03-27 10:57:27.000000000","tz":0},"subject":"kolla-ansible: Add RabbitMQ 4.2 support and dynamic upgrade paths","message":"kolla-ansible: Add RabbitMQ 4.2 support and dynamic upgrade paths\n\nThe RabbitMQ version check currently hardcodes allowed version jumps.\nThis makes supporting new major/minor releases (like 4.2) rigid and\nrequires modifying the core task file.\n\nThis commit introduces the `rabbitmq_allowed_upgrades` variable in\nthe role\u0027s defaults. This allows operators to define explicit version\njumps (e.g., 3.13 to 4.2) and makes the upgrade logic more flexible.\n\nThe version validation logic now:\n1. Checks for standard minor increments (e.g., 4.1 -\u003e 4.2).\n2. Checks for explicitly defined jumps in `rabbitmq_allowed_upgrades`.\n\nBy moving these definitions to defaults, operators can easily override\nthem in globals.yml if necessary.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/966017\n\nChange-Id: I0e19cf109eb5665045806b7d925779e82164c2dc\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1e2a0b62066b084d3767756e26cce384d11b5224"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1e2a0b62066b084d3767756e26cce384d11b5224"}]},"branch":"refs/heads/master","description":"Rebase"},"423ce9388becee87f96aa3c16bb081a4f2c18335":{"kind":"REWORK","_number":26,"created":"2026-03-27 10:58:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/966021/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/966021/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/26"}}},"commit":{"parents":[{"commit":"0812604cebab71eb804fbf50d5d6e1927b2629df","subject":"Merge \"cloudkitty: Switch to uWSGI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/0812604cebab71eb804fbf50d5d6e1927b2629df"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-04 10:22:45.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-03-27 10:58:32.000000000","tz":0},"subject":"kolla-ansible: Add RabbitMQ 4.2 support and dynamic upgrade paths","message":"kolla-ansible: Add RabbitMQ 4.2 support and dynamic upgrade paths\n\nThe RabbitMQ version check currently hardcodes allowed version jumps.\nThis makes supporting new major/minor releases (like 4.2) rigid and\nrequires modifying the core task file.\n\nThis commit introduces the `rabbitmq_allowed_upgrades` variable in\nthe role\u0027s defaults. This allows operators to define explicit version\njumps (e.g., 3.13 to 4.2) and makes the upgrade logic more flexible.\n\nThe version validation logic now:\n1. Checks for standard minor increments (e.g., 4.1 -\u003e 4.2).\n2. Checks for explicitly defined jumps in `rabbitmq_allowed_upgrades`.\n\nBy moving these definitions to defaults, operators can easily override\nthem in globals.yml if necessary.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/966017\n\nChange-Id: I0e19cf109eb5665045806b7d925779e82164c2dc\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/423ce9388becee87f96aa3c16bb081a4f2c18335"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/423ce9388becee87f96aa3c16bb081a4f2c18335"}]},"branch":"refs/heads/master"},"3c88d77fad32e2ebca657843178b36340c411e8e":{"kind":"REWORK","_number":27,"created":"2026-03-27 11:00:30.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/966021/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/966021/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/27"}}},"commit":{"parents":[{"commit":"0812604cebab71eb804fbf50d5d6e1927b2629df","subject":"Merge \"cloudkitty: Switch to uWSGI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/0812604cebab71eb804fbf50d5d6e1927b2629df"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-04 10:22:45.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-03-27 11:00:29.000000000","tz":0},"subject":"kolla-ansible: Add RabbitMQ 4.2 support and dynamic upgrade paths","message":"kolla-ansible: Add RabbitMQ 4.2 support and dynamic upgrade paths\n\nThe RabbitMQ version check currently hardcodes allowed version jumps.\nThis makes supporting new major/minor releases (like 4.2) rigid and\nrequires modifying the core task file.\n\nThis commit introduces the `rabbitmq_allowed_upgrades` variable in\nthe role\u0027s defaults. This allows operators to define explicit version\njumps (e.g., 3.13 to 4.2) and makes the upgrade logic more flexible.\n\nThe version validation logic now:\n1. Checks for standard minor increments (e.g., 4.1 -\u003e 4.2).\n2. Checks for explicitly defined jumps in `rabbitmq_allowed_upgrades`.\n\nBy moving these definitions to defaults, operators can easily override\nthem in globals.yml if necessary.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/966017\n\nChange-Id: I0e19cf109eb5665045806b7d925779e82164c2dc\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/3c88d77fad32e2ebca657843178b36340c411e8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/3c88d77fad32e2ebca657843178b36340c411e8e"}]},"branch":"refs/heads/master"},"681f90526695366bf115bb70f94a0ba5924103bb":{"kind":"REWORK","_number":28,"created":"2026-03-27 11:57:30.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/966021/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/966021/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/28"}}},"commit":{"parents":[{"commit":"eeca18e402a245fdb90914f17f49e42f52ede286","subject":"Merge \"RabbitMQ: document handling stream replicas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/eeca18e402a245fdb90914f17f49e42f52ede286"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-04 10:22:45.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-03-27 11:57:30.000000000","tz":0},"subject":"kolla-ansible: Add RabbitMQ 4.2 support and dynamic upgrade paths","message":"kolla-ansible: Add RabbitMQ 4.2 support and dynamic upgrade paths\n\nThe RabbitMQ version check currently hardcodes allowed version jumps.\nThis makes supporting new major/minor releases (like 4.2) rigid and\nrequires modifying the core task file.\n\nThis commit introduces the `rabbitmq_allowed_upgrades` variable in\nthe role\u0027s defaults. This allows operators to define explicit version\njumps (e.g., 3.13 to 4.2) and makes the upgrade logic more flexible.\n\nThe version validation logic now:\n1. Checks for standard minor increments (e.g., 4.1 -\u003e 4.2).\n2. Checks for explicitly defined jumps in `rabbitmq_allowed_upgrades`.\n\nBy moving these definitions to defaults, operators can easily override\nthem in globals.yml if necessary.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/966017\n\nChange-Id: I0e19cf109eb5665045806b7d925779e82164c2dc\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/681f90526695366bf115bb70f94a0ba5924103bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/681f90526695366bf115bb70f94a0ba5924103bb"}]},"branch":"refs/heads/master","description":"Rebase"},"5e18d6803fd8a725dc7411017bccea63826b32fe":{"kind":"REWORK","_number":29,"created":"2026-03-27 11:58:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/966021/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/966021/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/29"}}},"commit":{"parents":[{"commit":"eeca18e402a245fdb90914f17f49e42f52ede286","subject":"Merge \"RabbitMQ: document handling stream replicas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/eeca18e402a245fdb90914f17f49e42f52ede286"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-04 10:22:45.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-03-27 11:58:03.000000000","tz":0},"subject":"kolla-ansible: Add RabbitMQ 4.2 support and dynamic upgrade paths","message":"kolla-ansible: Add RabbitMQ 4.2 support and dynamic upgrade paths\n\nThe RabbitMQ version check currently hardcodes allowed version jumps.\nThis makes supporting new major/minor releases (like 4.2) rigid and\nrequires modifying the core task file.\n\nThis commit introduces the `rabbitmq_allowed_upgrades` variable in\nthe role\u0027s defaults. This allows operators to define explicit version\njumps (e.g., 3.13 to 4.2) and makes the upgrade logic more flexible.\n\nThe version validation logic now:\n1. Checks for standard minor increments (e.g., 4.1 -\u003e 4.2).\n2. Checks for explicitly defined jumps in `rabbitmq_allowed_upgrades`.\n\nBy moving these definitions to defaults, operators can easily override\nthem in globals.yml if necessary.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/966017\n\nChange-Id: I0e19cf109eb5665045806b7d925779e82164c2dc\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/5e18d6803fd8a725dc7411017bccea63826b32fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/5e18d6803fd8a725dc7411017bccea63826b32fe"}]},"branch":"refs/heads/master"},"04f1276d652afbe6eb03b184902594666117d631":{"kind":"REWORK","_number":30,"created":"2026-03-27 14:00:49.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/21/966021/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/966021/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966021/30"}}},"commit":{"parents":[{"commit":"eeca18e402a245fdb90914f17f49e42f52ede286","subject":"Merge \"RabbitMQ: document handling stream replicas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/eeca18e402a245fdb90914f17f49e42f52ede286"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-04 10:22:45.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-03-27 14:00:46.000000000","tz":60},"subject":"kolla-ansible: Add RabbitMQ 4.2 support and dynamic upgrade paths","message":"kolla-ansible: Add RabbitMQ 4.2 support and dynamic upgrade paths\n\nThe RabbitMQ version check currently hardcodes allowed version jumps.\nThis makes supporting new major/minor releases (like 4.2) rigid and\nrequires modifying the core task file.\n\nThis commit introduces the `rabbitmq_allowed_upgrades` variable in\nthe role\u0027s defaults. This allows operators to define explicit version\njumps (e.g., 3.13 to 4.2) and makes the upgrade logic more flexible.\n\nThe version validation logic now:\n1. Checks for standard minor increments (e.g., 4.1 -\u003e 4.2).\n2. Checks for explicitly defined jumps in `rabbitmq_allowed_upgrades`.\n\nBy moving these definitions to defaults, operators can easily override\nthem in globals.yml if necessary.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/966017\n\nChange-Id: I0e19cf109eb5665045806b7d925779e82164c2dc\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/04f1276d652afbe6eb03b184902594666117d631"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/04f1276d652afbe6eb03b184902594666117d631"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
