)]}'
{"id":"openstack%2Fkolla-ansible~966821","triplet_id":"openstack%2Fkolla-ansible~master~Ia3d5ddd102dfb4e34237bda0016b93e329fd1a86","project":"openstack/kolla-ansible","branch":"master","topic":"kolla-ansible-lint","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-11-24 11:01:02.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"Ia3d5ddd102dfb4e34237bda0016b93e329fd1a86","subject":"ansible-lint: Fix yaml[truthy] in loadbalancer role","status":"MERGED","created":"2025-11-12 13:29:18.000000000","updated":"2025-11-24 14:13:38.000000000","submitted":"2025-11-24 14:12:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"966821-kolla-ansible-lint","meta_rev_id":"bc95f4dc42835193298f2831affa2187e3d9782d","_number":966821,"virtual_id_number":966821,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-24 14:12:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-11-24 09:03:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-11-24 09:03:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2025-11-24 09:03:45.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-12 15:50:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f9e0572592d7caf7b36142311d7c1a6e48f4afe0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-12 13:29:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e9b9c6f7668b0b1dc543a456a7ed621c639ab9be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 15:50:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1b0570a329764e629d94ab325976abcd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61dafc8145be4972895015eb4e807950 : POST_FAILURE in 7m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9564a40f233d4d93a68a252373929c50 : SUCCESS in 9m 35s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/51fef8e444884ea499789e875b231620 : POST_FAILURE in 1h 03m 46s (non-voting)\n- kolla-ansible-debian-bookworm https://zuul.opendev.org/t/openstack/build/c889c83609ce4676b8fa7c0d362dd7ab : FAILURE in 17m 08s\n- kolla-ansible-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/31c336ba2b0b4a0884e3b8c41ec9ecba : SUCCESS in 56m 57s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/7f2dc9bf659e426696d6e1ab92db7ec2 : POST_FAILURE in 8m 40s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/12c30f5ef53242afab04d7a3561c8769 : FAILURE in 11m 20s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/96ad0a37dcb34ef7ae4e023fa15d260a : SUCCESS in 1h 06m 03s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33d50efdfbd145d7a74dd82d8c897ac0 : POST_FAILURE in 18m 17s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fd5c3d8015324218a8a498ba81415aa0 : SUCCESS in 1h 07m 36s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/0d4949211c6046d5bb9a0f3073ae839a : FAILURE in 10m 45s\n- kolla-ansible-debian-bookworm-cells https://zuul.opendev.org/t/openstack/build/b2d5f8202d7e456f9a650ac1d8c97539 : FAILURE in 12m 46s (non-voting)\n- kolla-ansible-rocky-10-cells https://zuul.opendev.org/t/openstack/build/123fe6a11fb54f4588a254ef2fd804cc : SUCCESS in 54m 06s (non-voting)\n- kolla-ansible-ubuntu-noble-cells https://zuul.opendev.org/t/openstack/build/e9d838d85f714a49981b704555d2daf6 : POST_FAILURE in 14m 49s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/d885e10b5b404e6b810d43342417938c : FAILURE in 17m 14s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/7c2f07a4a44d4e41af72e484bc71cb3d : FAILURE in 18m 09s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/683699785de547c1899a7826106e9468 : SUCCESS in 1h 53m 54s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/48800869ccb04a85a0defab6def636e9 : FAILURE in 17m 47s (non-voting)\n- kolla-ansible-debian-bookworm-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/ea2316cc22bd41a38b2630c6732d162e : POST_FAILURE in 8m 25s (non-voting)\n- kolla-ansible-rocky-10-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/9e287b23b3c946d597f3c70936def016 : POST_FAILURE in 13m 48s (non-voting)\n- kolla-ansible-ubuntu-noble-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/37ef33f0dcc344519f124eea5cab8d4f : POST_FAILURE in 55m 31s (non-voting)\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/dbd52735d53f44dba7fcaee3dc2e061c : POST_FAILURE in 53m 35s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1e3934b006de4d1186cd19fbeb9e06d5 : SUCCESS in 53m 28s (non-voting)\n- kolla-ansible-debian-bookworm-lets-encrypt https://zuul.opendev.org/t/openstack/build/e6ccd2b3b86c4946a5314ab38f6dd618 : POST_FAILURE in 10m 03s (non-voting)\n- kolla-ansible-ubuntu-noble-lets-encrypt https://zuul.opendev.org/t/openstack/build/a6d1f647738a43628c88dba4decbdbbe : SUCCESS in 1h 36m 32s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/e25025f969bd43f4944ac7d32e0ea742 : POST_FAILURE in 8m 54s\n- kolla-ansible-debian-bookworm-mariadb-upgrade https://zuul.opendev.org/t/openstack/build/f838599b6dd44901bbe2486145f8483c : FAILURE in 10m 27s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/f5cf4cdce8df45e0bbf6040bd2d3cac8 : SUCCESS in 18m 14s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/dff5e780a694469080fd05c424679ab0 : SUCCESS in 23m 29s\n- kolla-ansible-ubuntu-noble-mariadb-upgrade https://zuul.opendev.org/t/openstack/build/c441ada2fe194824904f0dd1220b2d60 : FAILURE in 10m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4761bf4ec4d4b32b36a011bb743b54b : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/caed27c0a5334bacaf82c1bdbbde1707 : POST_FAILURE in 5m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a2300a4b41394d9f930d621b4631cc03 : POST_FAILURE in 4m 23s","accounts_in_message":[],"_revision_number":1},{"id":"b5a18aa0e566dc2d15d2752655681de90b53210d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 21:06:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d0090112e914909b2ac122744b3332b\n\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/b30fdddcf349449fa634a07aae477aca : FAILURE in 1h 20m 42s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64 https://zuul.opendev.org/t/openstack/build/263db2970f5847118b8926f7650d6e1c : FAILURE in 1h 47m 39s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64-podman https://zuul.opendev.org/t/openstack/build/52a9d370ca29479ea827193ead5bb4c4 : FAILURE in 1h 49m 58s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64-upgrade https://zuul.opendev.org/t/openstack/build/d0aee0e942154f939dc38fc92966d2cc : FAILURE in 16m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cdaf4adb1c4851413b1a572480682a742b620280","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-24 07:33:42.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"78c7656bc92f66c0ba7fffced1f76d13628aecc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-24 07:33:48.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"a9fe0d6bfd7d47327a150b137b1c7a41a8047fdd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-24 07:34:24.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"3ecc10f32bcd19600149a6050379438a82ee8309","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-24 07:34:54.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"53c49e40b56bd9de9a0fd1e26638ede22f3859de","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-24 09:03:45.000000000","message":"Patch Set 5: Backport-Candidate-2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"be7f9b210fadfb24cd082ed9968df86c8ebf7d93","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 09:44:17.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2711183ba310469f8b99306d83c57f6a\n\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/b4d2656be9564728a2ec1fbe68a282e6 : FAILURE in 2h 07m 35s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64 https://zuul.opendev.org/t/openstack/build/51e4a8de26ae45d89a0aed3feafdf5ae : FAILURE in 1h 29m 11s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64-podman https://zuul.opendev.org/t/openstack/build/d6f0a74a9c36461798ec54e5710e2e92 : FAILURE in 1h 19m 13s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64-upgrade https://zuul.opendev.org/t/openstack/build/12d67a7552874212bc8491d841241160 : FAILURE in 45m 53s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8358424fb2e47592a3f6ca0113359c5f39ec1f6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 10:00:56.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4631de2d63c54c0d957ae5246a38fa08\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d353e3c1db72454b8b24c59915fe65da : SUCCESS in 5m 40s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c8d0ea9379794da1bd6a121baa330cb2 : SUCCESS in 6m 40s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/698cfb5bc1d34230b6ed09eeb70f551e : FAILURE in 48m 12s (non-voting)\n- kolla-ansible-debian-bookworm https://zuul.opendev.org/t/openstack/build/acfa6120f0c045438633975b79fc4017 : SUCCESS in 41m 48s\n- kolla-ansible-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/06dba3312b034f10840085edf6af202c : SUCCESS in 1h 14m 53s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/1ad5af5498944aec9d886394508b2da1 : SUCCESS in 1h 16m 24s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/7acf565a0f154cfb8c1b8b515c3426f3 : SUCCESS in 1h 07m 21s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/50492071e9ec401087e505e17d0559d1 : SUCCESS in 1h 06m 21s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cdf329088b7f445ebc1689140ef33a63 : SUCCESS in 1h 07m 49s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c617910b79584f08acb09db20e0a1c9d : SUCCESS in 1h 27m 57s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/99a66ae1f4c348daa93fa7b9e9a0ad39 : SUCCESS in 1h 24m 32s\n- kolla-ansible-debian-bookworm-cells https://zuul.opendev.org/t/openstack/build/f79affbe7f684bbb86d75739b4462c7f : SUCCESS in 1h 05m 28s (non-voting)\n- kolla-ansible-rocky-10-cells https://zuul.opendev.org/t/openstack/build/0a881f2e3f304beda814a306db11cd7f : SUCCESS in 1h 19m 20s (non-voting)\n- kolla-ansible-ubuntu-noble-cells https://zuul.opendev.org/t/openstack/build/d68549fbaeee48e8a1dee8c00abb6a3b : SUCCESS in 1h 16m 06s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/936cb51d95f94d0f94f0a4bf0c2624f7 : SUCCESS in 1h 11m 29s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/55f3203ae29d40f18eaca8a7eba628d6 : SUCCESS in 2h 25m 20s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/a519ad48938944cdaba380c585b2c5ff : SUCCESS in 1h 18m 57s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/ea67c0b0e51c4ae9bb35986a5b735643 : SUCCESS in 2h 22m 43s (non-voting)\n- kolla-ansible-debian-bookworm-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/b92f9478c32e499ca072abbf7f85dbb6 : SUCCESS in 47m 51s (non-voting)\n- kolla-ansible-rocky-10-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/3c2e5e5a738346b0b497d28056ad042d : SUCCESS in 59m 55s (non-voting)\n- kolla-ansible-ubuntu-noble-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/250b35d2540140e68d3df781679e3021 : SUCCESS in 25m 29s (non-voting)\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/35a07efc3c7d49f382487140d2b85748 : SUCCESS in 32m 52s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a855189777d64f318fc2494c01b8971a : SUCCESS in 29m 58s (non-voting)\n- kolla-ansible-debian-bookworm-lets-encrypt https://zuul.opendev.org/t/openstack/build/9221b96f5f874d19a59e4d3e3bdfd72c : SUCCESS in 55m 52s (non-voting)\n- kolla-ansible-ubuntu-noble-lets-encrypt https://zuul.opendev.org/t/openstack/build/3508c03c777c4ec7b378e3a7fb50b572 : SUCCESS in 1h 01m 26s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/2dc886be5cc7456fa3c2992cfa84f5d8 : SUCCESS in 17m 55s\n- kolla-ansible-debian-bookworm-mariadb-upgrade https://zuul.opendev.org/t/openstack/build/7f5fc50980274b99ac882a61e2076a9d : SUCCESS in 34m 09s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/510617dd5c3c46bca7841da6da496170 : SUCCESS in 20m 52s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/b6928dd15006468499c2c71c79a7de8c : SUCCESS in 23m 55s\n- kolla-ansible-ubuntu-noble-mariadb-upgrade https://zuul.opendev.org/t/openstack/build/261b0dd4afb248daab4d9b37dc4e10c9 : SUCCESS in 23m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc62160eb3fc458ba61a010abc78467a : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7563df0c4d4041dd95adafb517059c83 : SUCCESS in 4m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4105b2e581e24c8da2d8c71bd64eca9b : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":5},{"id":"f2c00f13512809f1e331bd8c7e359724bf880761","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 10:05:08.000000000","message":"Patch Set 5: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/kolla-ansible/+/966820 is needed.","accounts_in_message":[],"_revision_number":5},{"id":"b218f54881e62cc742a6d8ff26c3ecef30f2d4a2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-24 11:01:02.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ab10978fec4c99c6b5374817516276686eb5402f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 12:11:44.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ac0112814a7486690590afa048fdfff\n\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/89ce9e7d1593458d88d222a31e0a3d25 : NODE_FAILURE Node(set) request e387b0d6b4154405acc6e951829bf722 failed in 0s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64 https://zuul.opendev.org/t/openstack/build/003a1a3e20224bd4a73b4fcaacd184c1 : FAILURE in 1h 06m 53s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64-podman https://zuul.opendev.org/t/openstack/build/cfc9dcd8e7a54e61b1a3117dac6bfdb2 : NODE_FAILURE Node(set) request 60699f524e8645b1a40fb72390cde406 failed in 0s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64-upgrade https://zuul.opendev.org/t/openstack/build/8f023499625649dea8588c4a7afc15a2 : FAILURE in 55m 56s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"777f298aa0411dcf644343017ee39932c6a2382a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 13:03:46.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3650b47095464bd4a438c50978fe9b51\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48f94ae818224adfbe528b959dbf164e : SUCCESS in 6m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/584ec7e9812f472f981352444e7bafcb : SUCCESS in 4m 01s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/63ecd2ac8bd1431cb8fbcd92f11cca23 : SUCCESS in 38m 33s (non-voting)\n- kolla-ansible-debian-bookworm https://zuul.opendev.org/t/openstack/build/0227df5686ca450194e30828176af59c : SUCCESS in 40m 41s\n- kolla-ansible-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/e4f935c41386437a80b83abb1f6fbdbc : SUCCESS in 55m 46s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/fcd32293864d445fb33b75e7061d9467 : SUCCESS in 58m 44s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/a60be9864a62490482e1e3f68091a7fe : SUCCESS in 53m 27s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/c8a9c820a0514310b7a63eba322bd7a6 : SUCCESS in 1h 05m 14s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94470b6d4963470492960dc02fc98643 : SUCCESS in 53m 26s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9ceeed7d7b234025b72af73141939e2f : SUCCESS in 1h 07m 37s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/c79c47a0ab004f9d8122e653933ec9ce : SUCCESS in 1h 10m 36s\n- kolla-ansible-debian-bookworm-cells https://zuul.opendev.org/t/openstack/build/acd8ea9bed4e454682cf402db73d23e3 : FAILURE in 42m 01s (non-voting)\n- kolla-ansible-rocky-10-cells https://zuul.opendev.org/t/openstack/build/95e3d8c1b5314933aa72aea859c0464d : SUCCESS in 59m 45s (non-voting)\n- kolla-ansible-ubuntu-noble-cells https://zuul.opendev.org/t/openstack/build/da67274418004175972dae4bd528ed33 : SUCCESS in 59m 10s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/0d607e2c05c044ee98370482ed4d00c2 : FAILURE in 56m 07s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/d336ee74272d4c10ba24b1badd71b49c : SUCCESS in 1h 47m 36s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/bc71b41c5ce24e158d36de036c91edc3 : SUCCESS in 1h 25m 51s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/07cb8f61a2ac458db78c42993fb28341 : SUCCESS in 1h 58m 47s (non-voting)\n- kolla-ansible-debian-bookworm-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/57e52f9609db456ca800e80a61298bd7 : SUCCESS in 21m 31s (non-voting)\n- kolla-ansible-rocky-10-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/caa79d0830994fcd8036b418d75edaa8 : SUCCESS in 25m 50s (non-voting)\n- kolla-ansible-ubuntu-noble-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/d11ccc17eda5445691062fd6e86d7bab : SUCCESS in 53m 02s (non-voting)\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/02a0ca0801094330a77e396fbd5cd362 : SUCCESS in 47m 54s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/336a410ec86b42d9ba0f4887fdf57b09 : SUCCESS in 30m 08s (non-voting)\n- kolla-ansible-debian-bookworm-lets-encrypt https://zuul.opendev.org/t/openstack/build/d52ad10d17884026a50a25f7da31cd42 : SUCCESS in 1h 01m 41s (non-voting)\n- kolla-ansible-ubuntu-noble-lets-encrypt https://zuul.opendev.org/t/openstack/build/bcfdaa01a10c4da3be05d6b6c2a1a504 : SUCCESS in 1h 02m 48s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/ba477289df8243818524f5ca1fbcfe69 : SUCCESS in 16m 54s\n- kolla-ansible-debian-bookworm-mariadb-upgrade https://zuul.opendev.org/t/openstack/build/499067807fef4c93ab7e96fa37fadc34 : SUCCESS in 21m 16s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/faa5512d47844615a330e0f5f18a9b16 : SUCCESS in 21m 12s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/4659818698ff46afac9e4f1c6402dbfe : SUCCESS in 24m 34s\n- kolla-ansible-ubuntu-noble-mariadb-upgrade https://zuul.opendev.org/t/openstack/build/35d2bc0381ab4cc38aba5a74c04d8459 : SUCCESS in 17m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6f7857e66a14b27a15798e9979167bc : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be964b4d3f0f422598409330942eac40 : SUCCESS in 7m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/460c323b80724810b3942bd14f109bf2 : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":5},{"id":"6dc25ff1e34895b7c9bf33f847cb823e7e5be55f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 13:04:04.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"761b860a6ec8f8962c4a74521cd5116533f9ded4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 14:12:44.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6adf59cc093f42e2b30909d885398b33\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b66593a31814f0d9c111133273a03af : SUCCESS in 6m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f89f489ce56e404b8f4ab83a0c016ce4 : SUCCESS in 6m 32s\n- kolla-ansible-debian-bookworm https://zuul.opendev.org/t/openstack/build/e3572aa2c00f415ca46f6ba40e363446 : SUCCESS in 42m 14s\n- kolla-ansible-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/37013944963b459abf19774d620538ce : SUCCESS in 56m 50s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/431e7eb2024e439a9965a9073a81fdea : SUCCESS in 56m 20s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/1a1e089294cd488fbf45c0c5fb4c1f1b : SUCCESS in 49m 41s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/be910592681549339b6338fd93a9c0d7 : SUCCESS in 1h 05m 49s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/90dcb29c36aa4b2197ce46e79c1c94d0 : SUCCESS in 46m 18s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/f0da3e451ccb4c9688aa8a6c0ec366fa : SUCCESS in 1h 03m 00s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ef2e219267354e43aa45c433877e2504 : SUCCESS in 1h 03m 54s\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/d086a7882f744c698a53b31723571bc9 : SUCCESS in 18m 23s\n- kolla-ansible-debian-bookworm-mariadb-upgrade https://zuul.opendev.org/t/openstack/build/117921b750064bb9b3a952985206d392 : SUCCESS in 22m 57s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/0bacc9f325c64acd91930e217dfe2107 : SUCCESS in 21m 16s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/29527e658df24b6f9fb46a13ca6b6448 : SUCCESS in 18m 37s\n- kolla-ansible-ubuntu-noble-mariadb-upgrade https://zuul.opendev.org/t/openstack/build/5c7c5d0bd9b045e3b6abf2be2037b612 : SUCCESS in 32m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa93934a87bb4b0d8d55ab8dfc292d54 : SUCCESS in 2m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1218b4546cc148bfb67a70d38cc3b188 : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":5},{"id":"111c13265f22fa41526286a39ee679ab4e635388","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 14:12:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"bc95f4dc42835193298f2831affa2187e3d9782d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 14:13:38.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91dfcee7e54a4c5fbe0af9e05cfb168c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/380f2f2ad60e40698339694eb1aee08e : SUCCESS in 41s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"9ae77af3be44d74c7446eec6db318db5d7ec219f","revisions":{"a6dc906f6267605680a70ef388eafa2ea7bddd69":{"kind":"REWORK","_number":1,"created":"2025-11-12 13:29:18.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/966821/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/966821/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966821/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966821/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966821/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966821/1"}}},"commit":{"parents":[{"commit":"95b2163f5aa318fdcc533a2f96ef0c7b2bb8aa1e","subject":"ansible-lint: Fix yaml[truthy] in kuryr role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/95b2163f5aa318fdcc533a2f96ef0c7b2bb8aa1e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-12 13:29:08.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-12 13:29:08.000000000","tz":60},"subject":"ansible-lint: Fix yaml[truthy] in loadbalancer role","message":"ansible-lint: Fix yaml[truthy] in loadbalancer role\n\nChange-Id: Ia3d5ddd102dfb4e34237bda0016b93e329fd1a86\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/a6dc906f6267605680a70ef388eafa2ea7bddd69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/a6dc906f6267605680a70ef388eafa2ea7bddd69"}]},"branch":"refs/heads/master"},"616ecc80aeff3bf7fa7f96f5ad67ee7007384820":{"kind":"REWORK","_number":2,"created":"2025-11-24 07:33:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/966821/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/966821/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966821/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966821/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966821/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966821/2"}}},"commit":{"parents":[{"commit":"95b2163f5aa318fdcc533a2f96ef0c7b2bb8aa1e","subject":"ansible-lint: Fix yaml[truthy] in kuryr role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/95b2163f5aa318fdcc533a2f96ef0c7b2bb8aa1e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-12 13:29:08.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-24 07:33:41.000000000","tz":0},"subject":"ansible-lint: Fix yaml[truthy] in loadbalancer role","message":"ansible-lint: Fix yaml[truthy] in loadbalancer role\n\nChange-Id: Ia3d5ddd102dfb4e34237bda0016b93e329fd1a86\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/616ecc80aeff3bf7fa7f96f5ad67ee7007384820"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/616ecc80aeff3bf7fa7f96f5ad67ee7007384820"}]},"branch":"refs/heads/master"},"b10b34a58e25c280cb0dc981656cd9e5afa5a98f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-11-24 07:33:48.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/966821/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/966821/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966821/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966821/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966821/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966821/3"}}},"commit":{"parents":[{"commit":"43f255d9dfd8d2f2cf0b90281cd12c577136b5bb","subject":"ansible-lint: Fix yaml[truthy] in kuryr role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/43f255d9dfd8d2f2cf0b90281cd12c577136b5bb"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-12 13:29:08.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-24 07:33:48.000000000","tz":0},"subject":"ansible-lint: Fix yaml[truthy] in loadbalancer role","message":"ansible-lint: Fix yaml[truthy] in loadbalancer role\n\nChange-Id: Ia3d5ddd102dfb4e34237bda0016b93e329fd1a86\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/b10b34a58e25c280cb0dc981656cd9e5afa5a98f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/b10b34a58e25c280cb0dc981656cd9e5afa5a98f"}]},"branch":"refs/heads/master","description":"Rebase"},"dc8074b5c9e70d38fabb115750013ac04ee4e553":{"kind":"REWORK","_number":4,"created":"2025-11-24 07:34:24.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/966821/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/966821/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966821/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966821/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966821/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966821/4"}}},"commit":{"parents":[{"commit":"43f255d9dfd8d2f2cf0b90281cd12c577136b5bb","subject":"ansible-lint: Fix yaml[truthy] in kuryr role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/43f255d9dfd8d2f2cf0b90281cd12c577136b5bb"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-12 13:29:08.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-24 07:34:23.000000000","tz":0},"subject":"ansible-lint: Fix yaml[truthy] in loadbalancer role","message":"ansible-lint: Fix yaml[truthy] in loadbalancer role\n\nChange-Id: Ia3d5ddd102dfb4e34237bda0016b93e329fd1a86\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/dc8074b5c9e70d38fabb115750013ac04ee4e553"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/dc8074b5c9e70d38fabb115750013ac04ee4e553"}]},"branch":"refs/heads/master"},"9ae77af3be44d74c7446eec6db318db5d7ec219f":{"kind":"REWORK","_number":5,"created":"2025-11-24 07:34:54.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/966821/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/966821/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966821/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966821/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966821/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/966821/5"}}},"commit":{"parents":[{"commit":"43f255d9dfd8d2f2cf0b90281cd12c577136b5bb","subject":"ansible-lint: Fix yaml[truthy] in kuryr role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/43f255d9dfd8d2f2cf0b90281cd12c577136b5bb"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-12 13:29:08.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-24 07:34:52.000000000","tz":0},"subject":"ansible-lint: Fix yaml[truthy] in loadbalancer role","message":"ansible-lint: Fix yaml[truthy] in loadbalancer role\n\nChange-Id: Ia3d5ddd102dfb4e34237bda0016b93e329fd1a86\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/9ae77af3be44d74c7446eec6db318db5d7ec219f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/9ae77af3be44d74c7446eec6db318db5d7ec219f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
