)]}'
{"id":"openstack%2Fkolla-ansible~972521","triplet_id":"openstack%2Fkolla-ansible~stable%2F2025.1~I5782a5da30434cf5c5ae0fcb60b3c6f178351fb7","project":"openstack/kolla-ansible","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"37203":{"account":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"last_update":"2026-03-17 21:42:58.000000000","reason":"\u003cGERRIT_ACCOUNT_37203\u003e replied on the change","reason_account":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-01-23 08:22:03.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"36698":{"account":{"_account_id":36698,"name":"hamid lotfi","display_name":"Hamid Lotfi","email":"hamid.lotfi@gmail.com","username":"hamidlotfi"},"last_update":"2026-01-23 08:03:52.000000000","reason":"Reviewer/Cc was removed"}},"hashtags":[],"change_id":"I5782a5da30434cf5c5ae0fcb60b3c6f178351fb7","subject":"Fix trove_api healthcheck with TLS backend","status":"NEW","created":"2026-01-07 17:18:25.000000000","updated":"2026-04-24 09:00:48.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d6dbb8bae26c688994de7f4276ec3e822df4e0a8","_number":972521,"virtual_id_number":972521,"owner":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-01-23 10:08:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2026-04-24 09:00:48.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},{"value":0,"date":"2026-03-17 21:43:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":1,"date":"2026-01-23 08:22:03.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"}],"CC":[{"_account_id":36698,"name":"hamid lotfi","display_name":"Hamid Lotfi","email":"hamid.lotfi@gmail.com","username":"hamidlotfi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-07 17:18:25.000000000","updated_by":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"reviewer":{"_account_id":36698,"name":"hamid lotfi","display_name":"Hamid Lotfi","email":"hamid.lotfi@gmail.com","username":"hamidlotfi"},"state":"REVIEWER"},{"updated":"2026-01-07 19:19:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-01-07 19:29:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-07 22:31:16.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-01-23 08:03:52.000000000","updated_by":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"reviewer":{"_account_id":36698,"name":"hamid lotfi","display_name":"Hamid Lotfi","email":"hamid.lotfi@gmail.com","username":"hamidlotfi"},"state":"CC"},{"updated":"2026-04-24 09:00:48.000000000","updated_by":{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},"reviewer":{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},"state":"REVIEWER"}],"messages":[{"id":"865049c6ab7f18b96c07fdfda1c219ab0737b140","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"date":"2026-01-07 17:18:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"40137a515977a2c1b9810afccfe1ccb70d34b2cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 19:19:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad55fc30bf6b4c05a0d63da8929044c5\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9ff7c2ed83384c80acd31bbd486a56b2 : SUCCESS in 1h 15m 46s (non-voting)\n- kolla-ansible-debian-aarch64-podman https://zuul.opendev.org/t/openstack/build/5902542aea234f63b2115d1535eca916 : SUCCESS in 57m 43s (non-voting)\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/70aaa58606524c1388e235e7175b68ee : SUCCESS in 1h 50m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"59f94ffd4d2d7662efa53ca4807bf2261eac298e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 19:29:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63b85a461d4c4a5083a3c9519babc459\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2e3a7e303374277b9fdfd7b4c1c2c08 : SUCCESS in 2m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a138118fb354434b5ab431a77134fb2 : SUCCESS in 3m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/00ae1d2990e64a1ca39e6f4d2e346f23 : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/320ef4e0ee62418eaa0cd5237c30fad3 : SUCCESS in 2m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aff9ac63d693412bb977d9e2ad64107d : SUCCESS in 3m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/593c55744222473f9d0dfa4a11c996e3 : SUCCESS in 3m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/690d86ab30c14414a88c0ecb45e85f8a : SUCCESS in 2m 42s\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/bbfc80070ab048dea429b4c7724bde5e : SUCCESS in 1h 27m 33s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/27471c1ee1bf4db6b6db5e1ed11bc988 : SUCCESS in 1h 29m 36s (non-voting)\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f5e3928747dd4ceea6cf6571912e22a1 : SUCCESS in 37m 39s\n- kolla-ansible-debian-podman https://zuul.opendev.org/t/openstack/build/a972523c8f0f4873bc91230860ebfd79 : SUCCESS in 52m 10s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5124bb5ab0774f959499a0f69174a621 : SUCCESS in 33m 56s\n- kolla-ansible-rocky9-podman https://zuul.opendev.org/t/openstack/build/c2da19e7edad4bb683e692e3d9f6ec12 : SUCCESS in 1h 01m 08s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/44b9aac03eb7411db15ca21add127401 : SUCCESS in 32m 33s\n- kolla-ansible-rocky10-podman https://zuul.opendev.org/t/openstack/build/2b76b3d4d9f74c5d8674cb459eda3459 : SUCCESS in 1h 08m 12s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b5baebe5fff046929f39e269721925ea : SUCCESS in 43m 46s\n- kolla-ansible-ubuntu-podman https://zuul.opendev.org/t/openstack/build/1275f46acfe6479796d5166d7e2f3f7f : SUCCESS in 59m 05s\n- kolla-ansible-rocky9-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/21606bba376b4628b0659be6597b94fb : SUCCESS in 1h 08m 10s (non-voting)\n- kolla-ansible-ubuntu-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/69325d3b2c8a4222bbe43afca7e46e4d : DISK_FULL in 1h 24m 11s (non-voting)\n- kolla-ansible-rocky9-magnum https://zuul.opendev.org/t/openstack/build/5a73c0bec1c84b8ea39d3d9a0616b7d8 : SUCCESS in 1h 08m 45s (non-voting)\n- kolla-ansible-ubuntu-magnum https://zuul.opendev.org/t/openstack/build/1671a62d3e8745ef99f91d33c553f1a8 : SUCCESS in 34m 21s (non-voting)\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/98477e88cd114ed49561f9bc7e6a4bf0 : SUCCESS in 49m 44s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3f14723a72a049f3a3939c47df442e9d : SUCCESS in 1h 09m 23s (non-voting)\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f09a495648664fd69830168ec43a63ba : SUCCESS in 1h 15m 09s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/718df191d5dd450c89808408456df1a8 : SUCCESS in 1h 04m 55s (non-voting)\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/22ba0a7f575a4950acb8dac5d4c0eb3c : SUCCESS in 1h 34m 07s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f5d34d1cf72f4d4dae3ccfaaa545c3f7 : FAILURE in 51m 59s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/5a74aa6f3712416bab782bad6cd0a874 : SUCCESS in 2h 09m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"67c11a8ab009677493d480d989e6439d6d7455dd","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-01-07 22:31:16.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0eb6cef466d04261eab53e26e6ff7d439d03aeab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"date":"2026-01-23 08:03:52.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"21bd89595babfe0387d5de01f2d34b91801f6422","author":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"date":"2026-01-23 08:05:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"70e214797202f9be67ebfe5c415cdcc3d37e9fce","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-01-23 08:22:03.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b2f13b4889b625b23aa09f56f69a6f71fd0b7e6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 10:08:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/201ff35d54a24b20b3c9b2398cc34d5d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04502399323843eca02103ffd55d5102 : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71684d3f160f41079643ddc550aa1dd9 : SUCCESS in 6m 37s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ae06dd364e5d45a2b6983f735494f0ed : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a357098bd184b0d9ea245eb7882bdd2 : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9126f014a8e84a6f80105067b6b42501 : SUCCESS in 3m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/40e38d2d3f734a8bb8a86d037f2aa089 : SUCCESS in 3m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d50f425dae814640a1f0aa6af1ab3f67 : SUCCESS in 3m 16s\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8b3dc18dd43a46c3a4edc30e44bc71cf : SUCCESS in 46m 20s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/9f9c5acfda6c454db36b3088253ac648 : SUCCESS in 54m 11s (non-voting)\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/99c36be4a56946b780ef34dfe1602ee0 : SUCCESS in 37m 12s\n- kolla-ansible-debian-podman https://zuul.opendev.org/t/openstack/build/3b6e41d41a6e4976a45658370bbb4caf : SUCCESS in 49m 58s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6b8de84531554ce5823933a82398e420 : SUCCESS in 1h 03m 03s\n- kolla-ansible-rocky9-podman https://zuul.opendev.org/t/openstack/build/d14c086b05274e85a49e3fc3def4e6b2 : SUCCESS in 1h 12m 07s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/b6cd47e0a3254039b42bbb2167ee3103 : SUCCESS in 32m 09s\n- kolla-ansible-rocky10-podman https://zuul.opendev.org/t/openstack/build/4feaf25b0d5547ab8ccd860b0ac9ca59 : SUCCESS in 47m 29s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/dfacce56b5894d23b18c4bd408e7de91 : SUCCESS in 44m 05s\n- kolla-ansible-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c1f417fe8cb44ca987fe73153d3e4687 : SUCCESS in 54m 19s\n- kolla-ansible-rocky9-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/55bd269699884678968c431a96feae62 : SUCCESS in 1h 07m 27s (non-voting)\n- kolla-ansible-ubuntu-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/9ad87f4fd3d34a84b1f85c93502cb871 : DISK_FULL in 1h 10m 55s (non-voting)\n- kolla-ansible-rocky9-magnum https://zuul.opendev.org/t/openstack/build/8951c5da68d0482794646a0f5a1e4961 : SUCCESS in 1h 16m 21s (non-voting)\n- kolla-ansible-ubuntu-magnum https://zuul.opendev.org/t/openstack/build/15a2b071faf34096a6a2d911f00cd2bd : SUCCESS in 1h 12m 23s (non-voting)\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/4ce0feafbfa446fe91d405203672e3ec : SUCCESS in 1h 42m 29s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6fdde5a1944c4c12a2d1d8870d452838 : SUCCESS in 1h 08m 42s (non-voting)\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7ff5c7723cdb4622bade582460e72815 : SUCCESS in 1h 12m 20s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/64e1248a722e42ba98c9f03b69054c7b : SUCCESS in 1h 52m 22s (non-voting)\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/9db000d919f843fda8fe7e28400ae6f3 : SUCCESS in 1h 25m 54s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/7997600947a646ce961433133cd08767 : FAILURE in 1h 20m 30s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f7e2a8d14ae149ddbb39890d1d91a23d : SUCCESS in 2h 03m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6051211c3ca6c4f4755569f301476e5b211b0ece","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 12:25:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04f69b19311a48468a718baa5071b1ed\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e9437614ed434283bebdde7b0be20deb : SUCCESS in 1h 21m 14s (non-voting)\n- kolla-ansible-debian-aarch64-podman https://zuul.opendev.org/t/openstack/build/0c4bf74d3bda454799429fa343bdc729 : SUCCESS in 1h 50m 31s (non-voting)\n- kolla-ansible-debian-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/be1ff200faa649b082e286c225a0b551 : SUCCESS in 2h 05m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cd6fe2dc60ef1eca75ce362a05ffd19d7c1c730f","author":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"date":"2026-03-17 21:42:58.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"917671a97e36004c745e8cafcd414791d2b39d3f","author":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"date":"2026-03-17 21:43:06.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"d6dbb8bae26c688994de7f4276ec3e822df4e0a8","author":{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},"date":"2026-04-24 09:00:48.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"dfa5c42c596a859b1fa37c1249aa8eef0335be5b","revisions":{"39411df963908deb5cc26db4e68b4957ff95ddd0":{"kind":"REWORK","_number":1,"created":"2026-01-07 17:18:25.000000000","uploader":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"ref":"refs/changes/21/972521/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/972521/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/972521/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/972521/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/972521/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/972521/1"}}},"commit":{"parents":[{"commit":"f74e0444a0bf341d76caeb798f2f463493e8cc16","subject":"Merge \"Fix incorrect mariadb-recovery task name in error message\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/f74e0444a0bf341d76caeb798f2f463493e8cc16"}]}],"author":{"name":"hamidlotfi","email":"hamid.lotfi@gmail.com","date":"2025-12-21 12:19:26.000000000","tz":210},"committer":{"name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","date":"2026-01-07 17:18:16.000000000","tz":60},"subject":"Fix trove_api healthcheck with TLS backend","message":"Fix trove_api healthcheck with TLS backend\n\nWhen trove_enable_tls_backend is enabled, the API healthcheck\nmust use https scheme instead of plain http.\nThis change updates the healthcheck command accordingly.\n\nChange-Id: I5782a5da30434cf5c5ae0fcb60b3c6f178351fb7\nCloses-Bug: #2136977\nSigned-off-by: hamidlotfi \u003chamid.lotfi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/39411df963908deb5cc26db4e68b4957ff95ddd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/39411df963908deb5cc26db4e68b4957ff95ddd0"}]},"branch":"refs/heads/stable/2025.1"},"dfa5c42c596a859b1fa37c1249aa8eef0335be5b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-01-23 08:03:52.000000000","uploader":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"ref":"refs/changes/21/972521/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/972521/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/972521/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/972521/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/972521/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/972521/2"}}},"commit":{"parents":[{"commit":"f74e0444a0bf341d76caeb798f2f463493e8cc16","subject":"Merge \"Fix incorrect mariadb-recovery task name in error message\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/f74e0444a0bf341d76caeb798f2f463493e8cc16"}]}],"author":{"name":"hamidlotfi","email":"hamid.lotfi@gmail.com","date":"2025-12-21 12:19:26.000000000","tz":210},"committer":{"name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","date":"2026-01-23 08:03:30.000000000","tz":60},"subject":"Fix trove_api healthcheck with TLS backend","message":"Fix trove_api healthcheck with TLS backend\n\nWhen trove_enable_tls_backend is enabled, the API healthcheck\nmust use https scheme instead of plain http.\nThis change updates the healthcheck command accordingly.\n\nChange-Id: I5782a5da30434cf5c5ae0fcb60b3c6f178351fb7\nCloses-Bug: #2136977\nSigned-off-by: hamidlotfi \u003chamid.lotfi@gmail.com\u003e\n(cherry picked from commit 0079f20030ec52d43a8ea4503cd4fad60dcb35fb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/dfa5c42c596a859b1fa37c1249aa8eef0335be5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/dfa5c42c596a859b1fa37c1249aa8eef0335be5b"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
