)]}'
{"id":"openstack%2Fkolla-ansible~972725","triplet_id":"openstack%2Fkolla-ansible~master~I6f51dedc6663a792b402422a0ff737d58506601e","project":"openstack/kolla-ansible","branch":"master","topic":"local-registry-explanation","attention_set":{"26673":{"account":{"_account_id":26673,"name":"VinceLe","email":"vincent.legoll@gmail.com","username":"vincentlegoll"},"last_update":"2026-01-28 07:14:25.000000000","reason":"\u003cGERRIT_ACCOUNT_37855\u003e replied on the change","reason_account":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}},"37855":{"account":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"last_update":"2026-01-28 09:26:39.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I6f51dedc6663a792b402422a0ff737d58506601e","subject":"Add explanation for using a local Docker registry in multinode deployment","status":"NEW","created":"2026-01-08 15:13:51.000000000","updated":"2026-01-28 09:26:39.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4fbe9d8d3a4d80e1dbe2a37b74479a1f21a21df6","_number":972725,"virtual_id_number":972725,"owner":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":26673,"name":"VinceLe","email":"vincent.legoll@gmail.com","username":"vincentlegoll"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-01-08 15:30:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":26673,"name":"VinceLe","email":"vincent.legoll@gmail.com","username":"vincentlegoll"},"all":[{"value":1,"date":"2026-01-08 17:23:53.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26673,"name":"VinceLe","email":"vincent.legoll@gmail.com","username":"vincentlegoll"},{"value":1,"date":"2026-01-28 09:26:39.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":26673,"name":"VinceLe","email":"vincent.legoll@gmail.com","username":"vincentlegoll"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":26673,"name":"VinceLe","email":"vincent.legoll@gmail.com","username":"vincentlegoll"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":26673,"name":"VinceLe","email":"vincent.legoll@gmail.com","username":"vincentlegoll"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26673,"name":"VinceLe","email":"vincent.legoll@gmail.com","username":"vincentlegoll"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-08 15:30:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-08 17:23:53.000000000","updated_by":{"_account_id":26673,"name":"VinceLe","email":"vincent.legoll@gmail.com","username":"vincentlegoll"},"reviewer":{"_account_id":26673,"name":"VinceLe","email":"vincent.legoll@gmail.com","username":"vincentlegoll"},"state":"REVIEWER"},{"updated":"2026-01-28 09:26:39.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"66231e6f98d4a8c033d83d3155f76514a2f0118e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"date":"2026-01-08 15:13:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1204e0b70b8593631259be589829daad587316c3","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"date":"2026-01-08 15:14:54.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"f7299b4031751bb55b6220c54dcb146bd38a0cd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 15:30:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b11993a04fad4bd596651ca6a6660e08\n\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/fb53b9841ff447f38087339c19223020 : SUCCESS in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9fd72274a35467cb39c339485a780c2 : SUCCESS in 6m 33s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b095739278994fcca5755c399ab551cb : SUCCESS in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f44573f3e2a84222b15c5b6f1062f4c9 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":1},{"id":"bba2de28db4cff4dd0f2e310ea36e8fa503a5e38","author":{"_account_id":26673,"name":"VinceLe","email":"vincent.legoll@gmail.com","username":"vincentlegoll"},"date":"2026-01-08 17:23:53.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"29c3048f3a6d437fb718f8c84dcb0bd235ea94d8","author":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"date":"2026-01-28 07:14:25.000000000","message":"Patch Set 1:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"4fbe9d8d3a4d80e1dbe2a37b74479a1f21a21df6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-01-28 09:26:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ebd8df3d9ce31ec0b53dc66e4c62f7537e70cd0c","revisions":{"ebd8df3d9ce31ec0b53dc66e4c62f7537e70cd0c":{"kind":"REWORK","_number":1,"created":"2026-01-08 15:13:51.000000000","uploader":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"ref":"refs/changes/25/972725/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/25/972725/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/25/972725/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/25/972725/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/25/972725/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/25/972725/1"}}},"commit":{"parents":[{"commit":"f699a56a8ac9706c6c33edf00e5056f8ce1dc896","subject":"Merge \"Bump ansible-core to 2.19/2.20\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/f699a56a8ac9706c6c33edf00e5056f8ce1dc896"}]}],"author":{"name":"amir58118","email":"sahahmadi96@gmail.com","date":"2026-01-08 15:12:29.000000000","tz":0},"committer":{"name":"amir58118","email":"sahahmadi96@gmail.com","date":"2026-01-08 15:13:48.000000000","tz":0},"subject":"Add explanation for using a local Docker registry in multinode deployment","message":"Add explanation for using a local Docker registry in multinode deployment\n\nThis adds a short explanation in the multinode deployment docs about why a\nlocal registry is recommended: to ensure all nodes get the same Docker image\nversions and to reduce network-related issues.\n\nCloses-Bug: #2129795\n\nChange-Id: I6f51dedc6663a792b402422a0ff737d58506601e\nSigned-off-by: amir58118 \u003csahahmadi96@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ebd8df3d9ce31ec0b53dc66e4c62f7537e70cd0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ebd8df3d9ce31ec0b53dc66e4c62f7537e70cd0c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26673,"name":"VinceLe","email":"vincent.legoll@gmail.com","username":"vincentlegoll"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
