)]}'
{"id":"openstack%2Fkolla-ansible~974721","triplet_id":"openstack%2Fkolla-ansible~master~I7a996387ea31702e1a02567c6ac808cdb820a323","project":"openstack/kolla-ansible","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-01-26 16:44:57.000000000","reason":"Change was abandoned"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-01-26 16:44:57.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I7a996387ea31702e1a02567c6ac808cdb820a323","subject":"TEMP: Switch requirements-core.yml to git","status":"ABANDONED","created":"2026-01-26 12:46:35.000000000","updated":"2026-01-26 16:44:57.000000000","total_comment_count":4,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"4122725e4a32a5867d8f34dec49bfe8633bf9e6d","_number":974721,"virtual_id_number":974721,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-01-26 15:21:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-01-26 13:40:41.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-26 13:59:34.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2026-01-26 14:23:43.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-01-26 15:16:25.000000000","updated_by":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"reviewer":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"state":"CC"},{"updated":"2026-01-26 15:21:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cb0f37ed0d8322b9641ac5d65e101f0db21b89bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-01-26 12:46:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1cc9e29ca6b5a4f1b54be1b05d7cba579ec84cc","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-01-26 13:00:37.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"c22a44631827abe70556060a180ab3d81f33ce47","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-01-26 13:01:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ed0edfa801e487d8ed9ea5d34784d374b763cc8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-01-26 13:40:41.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":2},{"id":"85a7a0514d40789cf749c7ffe70ecec66f64f7a3","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-01-26 13:59:34.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"1552ce374dd04ef943fec99873b24eb9abe1b0d3","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2026-01-26 15:16:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d753c96a046398499aebc77f69d5104a95f3f8dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 15:21:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5466d12748b14b0eaad29e0ec49b6a6f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7844dc2d3c0453a8a391f5778879cb5 : SUCCESS in 3m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7610748fec404d13824e37cda3a63d5b : SUCCESS in 10m 53s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/2a5ee4476255463f92d3c97db31b87b9 : FAILURE in 1h 09m 50s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/e90b51a64712479289e467d4d8e3e87f : SUCCESS in 52m 14s\n- kolla-ansible-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/0ae2b438af3e4e8999d1de10f57fb3ed : SUCCESS in 49m 48s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/d8df5c04673545088c93818cebc4f918 : SUCCESS in 1h 07m 26s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/90d22eab54504d42a596a00f8e86f954 : SUCCESS in 56m 49s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/fa62262e7b7348609a45f906b2132bd2 : SUCCESS in 1h 16m 12s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/ba99993928f0481e9bbc3b400554660c : SUCCESS in 57m 16s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/35756b67fa124681b98ee64abf22e582 : SUCCESS in 52m 18s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3b639a5e13c5452baa8b9f2ecb63468c : SUCCESS in 54m 11s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/2439ff465f4a4cf5a22ce9e016e4d2a0 : SUCCESS in 1h 11m 46s\n- kolla-ansible-debian-trixie-bifrost https://zuul.opendev.org/t/openstack/build/e88a22787a0647949cfa5d670d8465b0 : SUCCESS in 14m 05s (non-voting)\n- kolla-ansible-rocky-10-bifrost https://zuul.opendev.org/t/openstack/build/dacce007044a4bc286017cb03537ca4e : SUCCESS in 15m 59s (non-voting)\n- kolla-ansible-ubuntu-noble-bifrost https://zuul.opendev.org/t/openstack/build/3f65c951903d4d618ae49abf742e3343 : SUCCESS in 14m 33s (non-voting)\n- kolla-ansible-debian-trixie-cells https://zuul.opendev.org/t/openstack/build/f3674eb15f7d49078d4e96a4eda1b3a0 : SUCCESS in 1h 02m 56s (non-voting)\n- kolla-ansible-rocky-10-cells https://zuul.opendev.org/t/openstack/build/f2bf1d79558d4cbdada3536eb1ac5c7e : SUCCESS in 1h 08m 22s (non-voting)\n- kolla-ansible-ubuntu-noble-cells https://zuul.opendev.org/t/openstack/build/3473531eb50b4f0685129aa878845b67 : SUCCESS in 1h 03m 59s (non-voting)\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/7bfeed235dfd4e3abed87287f341987e : FAILURE in 1h 05m 59s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/b9281ceaccc64ec68aefdc90fe6749b9 : FAILURE in 1h 02m 47s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/9d23f4891b15441f82749b4007231f55 : FAILURE in 1h 22m 04s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/140185c67e7d43098367d47451adc09a : FAILURE in 1h 07m 13s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/77e7599750a646069e4db8359a18229b : FAILURE in 1h 37m 52s (non-voting)\n- kolla-ansible-debian-trixie-federation https://zuul.opendev.org/t/openstack/build/1fcc4308b12e4ec8b5f0be11fda70cc2 : SUCCESS in 1h 07m 38s (non-voting)\n- kolla-ansible-rocky-10-federation https://zuul.opendev.org/t/openstack/build/24c10605989241b89ef6104f370068bc : SUCCESS in 59m 04s (non-voting)\n- kolla-ansible-ubuntu-noble-federation https://zuul.opendev.org/t/openstack/build/84d75a1b8d094fecadcda55bb1d475cc : SUCCESS in 1h 01m 52s (non-voting)\n- kolla-ansible-debian-trixie-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/a500b6678fbd43ab9ea578b79c6908bb : SUCCESS in 57m 13s (non-voting)\n- kolla-ansible-rocky-10-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/63145407022c4d008f246fcf6fd5e5ad : SUCCESS in 43m 58s (non-voting)\n- kolla-ansible-ubuntu-noble-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/933d37c926d3441cbec7e842c27b15c7 : SUCCESS in 1h 00m 08s (non-voting)\n- kolla-ansible-debian-trixie-openbao https://zuul.opendev.org/t/openstack/build/7a6b5a3ed0544910ad77497754f805fe : SUCCESS in 4m 49s (non-voting)\n- kolla-ansible-rocky-10-openbao https://zuul.opendev.org/t/openstack/build/567b87d341a544529f2fe8957b8e6921 : SUCCESS in 3m 33s (non-voting)\n- kolla-ansible-ubuntu-noble-openbao https://zuul.opendev.org/t/openstack/build/020bb0c0eac04283b74f24bdceea1834 : SUCCESS in 4m 25s (non-voting)\n- kolla-ansible-debian-trixie-kvm https://zuul.opendev.org/t/openstack/build/09f499c9f4814fec99bac1beebc539fc : SUCCESS in 50m 05s (non-voting)\n- kolla-ansible-rocky-10-kvm https://zuul.opendev.org/t/openstack/build/683e734b02d247acad7c087f1a1737f2 : FAILURE in 26m 51s (non-voting)\n- kolla-ansible-ubuntu-noble-kvm https://zuul.opendev.org/t/openstack/build/de1e47f43d4b4f898d663322d72b6bab : SUCCESS in 54m 06s (non-voting)\n- kolla-ansible-debian-trixie-lets-encrypt https://zuul.opendev.org/t/openstack/build/628a72406bec44aa93f613cabb99a4dd : FAILURE in 33m 59s (non-voting)\n- kolla-ansible-ubuntu-noble-lets-encrypt https://zuul.opendev.org/t/openstack/build/99925ca85b5a44a69cbbe04edf3deb08 : SUCCESS in 1h 10m 54s (non-voting)\n- kolla-ansible-debian-trixie-magnum https://zuul.opendev.org/t/openstack/build/f60a4dc8c6b341218af0e16fc9ec3c73 : FAILURE in 30m 40s (non-voting)\n- kolla-ansible-rocky-10-magnum https://zuul.opendev.org/t/openstack/build/c8e0ed23a397408280065c84c396e8b1 : SUCCESS in 1h 14m 22s (non-voting)\n- kolla-ansible-ubuntu-noble-magnum https://zuul.opendev.org/t/openstack/build/c792a7a1f2154f58858790f1e45a9560 : SUCCESS in 41m 13s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/7d081ca50bb646a288b6f6594ee53419 : SUCCESS in 22m 59s\n- kolla-ansible-debian-bookworm-mariadb-upgrade https://zuul.opendev.org/t/openstack/build/a71e4bbf78c747aba95ecff8c205ec06 : SUCCESS in 25m 02s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/4053fd6582d64fb5af47dc207785a930 : SUCCESS in 21m 49s\n- kolla-ansible-rocky-10-mariadb-upgrade https://zuul.opendev.org/t/openstack/build/17dd348564ed4d0480fdb8fd067e04f7 : SUCCESS in 29m 16s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/6fca453427994276b10c6c68993e53c1 : SUCCESS in 25m 11s\n- kolla-ansible-ubuntu-noble-mariadb-upgrade https://zuul.opendev.org/t/openstack/build/33bca3946f934fe7b6e5cff42495098b : SUCCESS in 30m 40s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/8b0cd1518d7c4611a0ca5b07fe122bb1 : FAILURE in 19m 13s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/9514d2b263ea47568c03bb9316f6150e : SUCCESS in 35m 42s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/56c75390655745dba223fdf7f5e020bc : SUCCESS in 1h 04m 50s (non-voting)\n- kolla-ansible-debian-trixie-nfv https://zuul.opendev.org/t/openstack/build/84ba92a99c5147638a42cee78f21fb75 : FAILURE in 7m 12s (non-voting)\n- kolla-ansible-ubuntu-noble-nfv https://zuul.opendev.org/t/openstack/build/06175e6f7d8145539df1922d3f58dd35 : SUCCESS in 1h 28m 04s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/60260d91bcd84205bb1b4a40c4ad140f : FAILURE in 59m 08s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/53277024201040afbf7fc1c111732943 : FAILURE in 1h 03m 08s (non-voting)\n- kolla-ansible-rocky-10-ironic-upgrade https://zuul.opendev.org/t/openstack/build/e6772678c4a34bc790f03bb51ea63deb : SUCCESS in 1h 19m 00s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/1f3cddf0c78644e0b44488fe1114ce74 : FAILURE in 1h 03m 27s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic-upgrade https://zuul.opendev.org/t/openstack/build/3dcc9e40ecf24a91a81eea49482b7f44 : SUCCESS in 1h 26m 20s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/8ed4ea5401d44255b71c67f27029badd : SUCCESS in 47m 52s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/04ec9c626ba849c0bbcbbd096f3374f9 : SUCCESS in 50m 35s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/74122fa4885246579477c9ec9f3c7a1c : FAILURE in 29m 05s (non-voting)\n- kolla-ansible-debian-bookworm-ovn-upgrade https://zuul.opendev.org/t/openstack/build/1f490254349348feb1d49be3f2adc1cb : FAILURE in 1h 15m 15s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/2c2f90a98aa14dfabfe5d07dae646871 : FAILURE in 30m 31s (non-voting)\n- kolla-ansible-rocky-10-ovn-upgrade https://zuul.opendev.org/t/openstack/build/47b3ddd144c043769f9f24b22fe6fda4 : FAILURE in 1h 19m 37s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/75b5d3e4ddd34a5798f234582ebc2db0 : FAILURE in 29m 54s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn-upgrade https://zuul.opendev.org/t/openstack/build/98cf6744c9444a37bea19ea0da1c8e47 : FAILURE in 1h 12m 11s (non-voting)\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ceb8601d5a9045cabed547e96132c8e9 : FAILURE in 16m 55s (non-voting)\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/93ebfdee6116431a9e358345116ab94f : FAILURE in 21m 18s (non-voting)\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/3a22372aa1bf461ba576b964d9cacce6 : SUCCESS in 52m 36s (non-voting)\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/096cd2f2c8f641c2a10f9830934e55ae : FAILURE in 35m 08s (non-voting)\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/af56da7b90de463caa0f1b6fa099eb7a : SUCCESS in 42m 15s (non-voting)\n- kolla-ansible-rocky-10-upgrade-slurp https://zuul.opendev.org/t/openstack/build/117598728ccf43ea992d8586d15aa9aa : FAILURE in 33m 13s\n- kolla-ansible-ubuntu-noble-upgrade-slurp https://zuul.opendev.org/t/openstack/build/0c296f0da1244a7fb3f1c4385cfbf232 : FAILURE in 35m 18s\n- kolla-ansible-ubuntu-noble-cephadm-upgrade-slurp https://zuul.opendev.org/t/openstack/build/e91191254a184e579cecf217824afdd5 : FAILURE in 1h 04m 11s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/534d942cd97041cc8a97c71b756a5e5c : SUCCESS in 44m 11s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/23cc3e9aa5db4161baa4ceb802475ed9 : SUCCESS in 1h 16m 19s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/8adef277d18d41dba9625288872a14f3 : SUCCESS in 48m 25s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2156833424de4812aa61631fa0028b1d : SUCCESS in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/632df8786b044e168f62f09fc6c28ffa : SUCCESS in 5m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8b35fe3a5e694c8fb382443fe369d90d : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":2},{"id":"4122725e4a32a5867d8f34dec49bfe8633bf9e6d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-01-26 16:44:57.000000000","message":"Abandoned\n\nnot needed anymore","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"892acb1c5d012e4d816bd3b66801c5ef41d7c3f6","revisions":{"d1ef1d68e9db97d20e30517ea3775e12b83102a1":{"kind":"REWORK","_number":1,"created":"2026-01-26 12:46:35.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/974721/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/974721/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/974721/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/974721/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/974721/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/974721/1"}}},"commit":{"parents":[{"commit":"77e9d23a20fa39fe49f03210b6a20bc463c31078","subject":"CI: Fix parent of bookworm ovn upgrade job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/77e9d23a20fa39fe49f03210b6a20bc463c31078"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-01-26 12:46:04.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-01-26 12:46:33.000000000","tz":60},"subject":"TEMP: Switch requirements-core.yml to git","message":"TEMP: Switch requirements-core.yml to git\n\nCurrently Ansible Galaxy is down, let\u0027s unblock some work.\n\nChange-Id: I7a996387ea31702e1a02567c6ac808cdb820a323\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/d1ef1d68e9db97d20e30517ea3775e12b83102a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/d1ef1d68e9db97d20e30517ea3775e12b83102a1"}]},"branch":"refs/heads/master"},"892acb1c5d012e4d816bd3b66801c5ef41d7c3f6":{"kind":"REWORK","_number":2,"created":"2026-01-26 13:40:41.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/974721/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/974721/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/974721/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/974721/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/974721/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/974721/2"}}},"commit":{"parents":[{"commit":"77e9d23a20fa39fe49f03210b6a20bc463c31078","subject":"CI: Fix parent of bookworm ovn upgrade job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/77e9d23a20fa39fe49f03210b6a20bc463c31078"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-01-26 12:46:04.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-01-26 13:40:40.000000000","tz":0},"subject":"TEMP: Switch requirements-core.yml to git","message":"TEMP: Switch requirements-core.yml to git\n\nCurrently Ansible Galaxy is down, let\u0027s unblock some work.\n\nChange-Id: I7a996387ea31702e1a02567c6ac808cdb820a323\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/892acb1c5d012e4d816bd3b66801c5ef41d7c3f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/892acb1c5d012e4d816bd3b66801c5ef41d7c3f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
