)]}'
{"id":"openstack%2Fkolla-ansible~975421","triplet_id":"openstack%2Fkolla-ansible~master~Ie41d001671cde41e6574ff18d1d06c9d8966a4a3","project":"openstack/kolla-ansible","branch":"master","attention_set":{},"removed_from_attention_set":{"37203":{"account":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"last_update":"2026-03-02 18:37:51.000000000","reason":"\u003cGERRIT_ACCOUNT_37203\u003e replied on the change","reason_account":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-03-02 21:57:21.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-02-23 19:42:32.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2026-02-23 09:38:42.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},"37855":{"account":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"last_update":"2026-03-02 21:57:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie41d001671cde41e6574ff18d1d06c9d8966a4a3","subject":"Allow SNI frontend when using a single haproxy VIP","status":"MERGED","created":"2026-02-02 16:38:07.000000000","updated":"2026-03-02 21:58:20.000000000","submitted":"2026-03-02 21:57:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975421","meta_rev_id":"19438c67c1223d160f09cf248124b3f0a0232ed1","_number":975421,"virtual_id_number":975421,"owner":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-02 21:57:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-02-23 19:42:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2026-02-23 09:38:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2026-02-23 09:38:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":1,"date":"2026-02-20 16:54:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":1,"date":"2026-02-20 16:54:40.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2026-02-23 09:38:42.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-02 16:54:02.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2026-02-02 18:21:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-02 18:38:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-04 11:18:51.000000000","updated_by":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"reviewer":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"state":"REVIEWER"},{"updated":"2026-02-04 12:05:14.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-02-04 14:26:55.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-02-16 14:00:09.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2026-02-16 14:15:34.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"472f184867678167e3fcb153e8fc10afd23e2791","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"date":"2026-02-02 16:38:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"28ceeb9c34b10d70e0bfdef251a750d87dee116d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 18:21:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43e2c0291b324ade961516c690401d61\n\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/980a10f81a1e40ea8a3a58669535267b : FAILURE in 1h 40m 21s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/7b8020eaedb24239b379d2eef82ee47f : FAILURE in 1h 32m 55s (non-voting)\n- kolla-ansible-debian-trixie-aarch64-podman https://zuul.opendev.org/t/openstack/build/8ea5ad6f78f44e7393464872b29630d7 : FAILURE in 1h 28m 58s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64-upgrade https://zuul.opendev.org/t/openstack/build/7e1ea7158b60430298397a212ce015b6 : FAILURE in 1h 15m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3c0124f937f83e768bf53578982933c40d8e0885","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 18:38:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6465f95987444058279875d05d6dc8b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6184fb32752544f093da7f144a9040d9 : SUCCESS in 3m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fde88ee153a94017bf6759e1bada2393 : SUCCESS in 10m 16s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/e6f9dfae337f42bf9301716d559c6d70 : FAILURE in 40m 56s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/f95a9b00f8024ea19e7568cbe608d642 : SUCCESS in 53m 22s\n- kolla-ansible-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/ef4d2c7ed3dc4cd2bd13a9fc26067039 : SUCCESS in 52m 38s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/0ed75fdd47ab43ea98db66e121f556e1 : SUCCESS in 1h 12m 26s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/826961b3dde744b6be26ece15db140f9 : SUCCESS in 54m 37s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/c27a5f3009b141118db7a33e3eaff39f : SUCCESS in 1h 12m 04s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/79eef5f41c6e4be6b98de8cbb871e6b1 : SUCCESS in 59m 52s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db4a8b2824e74ea6a79793fdc9b6d32a : SUCCESS in 56m 24s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3c3ba8d5cc8c46b796a829bea2dfd258 : SUCCESS in 55m 47s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/8d80d6b74d0042d6a21b50d76c24d58b : SUCCESS in 1h 16m 49s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/dd6e5147e5cd40c59f1b8d330c00c756 : SUCCESS in 1h 26m 11s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/dd60c02b23244b8da9c13a7441e1df0a : SUCCESS in 1h 25m 46s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/e410c32528d545f4804ee5daf9be2e32 : SUCCESS in 1h 53m 34s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/46b177e9fd0e4bd5ba56ca9713183a50 : SUCCESS in 1h 26m 54s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/d8be3431ed0b4c018559e27449fdd4b6 : SUCCESS in 1h 48m 20s (non-voting)\n- kolla-ansible-debian-trixie-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/2ab44e259d094c5085c173256ace9d55 : SUCCESS in 51m 35s (non-voting)\n- kolla-ansible-rocky-10-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/4c8b98c11caa484784112ef1ee878d57 : SUCCESS in 31m 54s (non-voting)\n- kolla-ansible-ubuntu-noble-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/b4141c2d23574075bc91bff6bc03f177 : SUCCESS in 52m 17s (non-voting)\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/dac56b7e578b43b6ad5dbf7f1cbdaab2 : SUCCESS in 57m 05s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7121de73d8ca4d6eab8d0e282ca43739 : SUCCESS in 55m 24s (non-voting)\n- kolla-ansible-debian-trixie-lets-encrypt https://zuul.opendev.org/t/openstack/build/5b70d3ddf8ad4371aebaa82455099857 : FAILURE in 32m 45s (non-voting)\n- kolla-ansible-ubuntu-noble-lets-encrypt https://zuul.opendev.org/t/openstack/build/8e53b523da3347268779386f1cc0bc11 : SUCCESS in 1h 08m 50s (non-voting)\n- kolla-ansible-rocky-10-upgrade-slurp https://zuul.opendev.org/t/openstack/build/ff99e96b742d43b7a8468fac19fa730f : SUCCESS in 1h 04m 27s\n- kolla-ansible-ubuntu-noble-upgrade-slurp https://zuul.opendev.org/t/openstack/build/b71832aa9fed47c0abfe8716c81683b5 : SUCCESS in 1h 07m 46s\n- kolla-ansible-ubuntu-noble-cephadm-upgrade-slurp https://zuul.opendev.org/t/openstack/build/160fd075a0df49979eb301d733a17055 : SUCCESS in 1h 54m 52s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3421d6e278047eb95157e6ce13f54fd : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8dac53e1ca1c46ea936510b9e88d182c : SUCCESS in 2m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/450f82e6bdcf4f9588da2bd469bca5ff : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":1},{"id":"61cf395e147dbf2f2c0920476a08468e04d32ab4","author":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"date":"2026-02-04 11:18:51.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7b8ce2d5d94eb9d50f2efab28e73144ebd101bbf","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-02-04 12:05:14.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8451bcce600e91e07742e784f70aa4a5472affce","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-02-04 14:26:55.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"ce874003798e4177d23bea8057fc455d32b685c5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-02-16 14:00:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"be5cab9c8d9249bcdf4347aba5712db0fef80799","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"date":"2026-02-16 14:13:15.000000000","message":"Uploaded patch set 2.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"7bebbfec72830efbf478922022b94e4b0d456df2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-02-16 14:15:34.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fa5354acff1e3dad0adf7b759f75b3492be91b1d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-02-16 14:44:13.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"18604177f53e60e0c5c923569e2ff101dfa9edd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 16:24:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28ae376c266f4e7788df9590b63480d8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd6372c4a1b94a0ea20543d102d5e138 : SUCCESS in 6m 40s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d6abfb870c984deda5060fb40a0c4a0f : SUCCESS in 11m 16s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/75f212b12dd049d29b057bcafd4f0cb5 : FAILURE in 53m 27s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/a5a275763e844494b163504d13a020c1 : SUCCESS in 49m 44s\n- kolla-ansible-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/326144d9c69543bfa228af1a2ef88b3f : SUCCESS in 50m 47s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/c523d61579384455a91888f53c6e9a4f : SUCCESS in 1h 03m 51s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/fc14425208724fd29ab7346355e39741 : SUCCESS in 57m 53s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/56ff71f7d5f9418f8b7c31dd69e31900 : SUCCESS in 1h 12m 19s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/44025a0e74e6480b9df3dfd963bb13d4 : SUCCESS in 56m 07s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de79d705a8764edf88baffbc5950009f : SUCCESS in 51m 44s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f3fea53e1fba477fa946ab63ae866af1 : SUCCESS in 53m 16s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/88e14424cac7461cb7b58517b86e277a : SUCCESS in 1h 17m 08s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/cf082e7cac924aa7845dcf3b53b4224a : FAILURE in 1h 57m 32s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/d886bca0adb44002ac02866fbc88fa17 : SUCCESS in 1h 19m 09s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/3e14c6747d0746f596a92a6262eff334 : SUCCESS in 1h 54m 11s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/0e79567344bf43c6a07027ca5594b2c5 : SUCCESS in 1h 24m 10s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/1c50a14fa5454b02a9188c6a6ff81385 : FAILURE in 44m 05s (non-voting)\n- kolla-ansible-debian-trixie-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/25bf9ff046974eb9a243eb0ee79728fc : SUCCESS in 56m 16s (non-voting)\n- kolla-ansible-rocky-10-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/9f0c2d18f7e04afaacf09593b91d255a : SUCCESS in 49m 20s (non-voting)\n- kolla-ansible-ubuntu-noble-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/7c6133e947374aa9b58554bba8080855 : SUCCESS in 42m 24s (non-voting)\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/abdfa5d3e03549eabaeccbb6def400fd : SUCCESS in 55m 21s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9662148f18484a0c85a04b98aa0d2a6d : SUCCESS in 56m 21s (non-voting)\n- kolla-ansible-debian-trixie-lets-encrypt https://zuul.opendev.org/t/openstack/build/0d8e2f4697414f47b4d1dff3c0e6fdc3 : FAILURE in 31m 37s (non-voting)\n- kolla-ansible-ubuntu-noble-lets-encrypt https://zuul.opendev.org/t/openstack/build/922c2bc305f2484dbadae1a5beddc21f : SUCCESS in 1h 12m 53s (non-voting)\n- kolla-ansible-rocky-10-upgrade-slurp https://zuul.opendev.org/t/openstack/build/f7e613ff6cef423187af6f7c09ddf327 : SUCCESS in 1h 08m 57s\n- kolla-ansible-ubuntu-noble-upgrade-slurp https://zuul.opendev.org/t/openstack/build/e82812c8edf241a89d0ccdb7f4a2f804 : SUCCESS in 1h 03m 46s\n- kolla-ansible-ubuntu-noble-cephadm-upgrade-slurp https://zuul.opendev.org/t/openstack/build/370e59d38a9e410fa21a108509c249b7 : SUCCESS in 2h 01m 19s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ae90321b04f4767be9436feff6b5614 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92796e96d0fc43d6a7e253c3cf3552f8 : SUCCESS in 4m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d903c2d7fb744f319252c02435ce327f : SUCCESS in 2m 03s","accounts_in_message":[],"_revision_number":2},{"id":"453f4cab7c179748344b7ab6c60087a09bd08319","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 19:59:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eada97bdd3a147f38b1575462d10ff52\n\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/5eddd73156664b4cafa1b2933ce47a47 : FAILURE in 1h 20m 09s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/a3b7a01df0df41b8bd4c8d3643463a7d : FAILURE in 1h 09m 25s (non-voting)\n- kolla-ansible-debian-trixie-aarch64-podman https://zuul.opendev.org/t/openstack/build/2b5471fd39054b36afef50238b32aa82 : FAILURE in 1h 09m 00s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64-upgrade https://zuul.opendev.org/t/openstack/build/b23aa511f6084bd7bc22f836d8a74588 : FAILURE in 37m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2653e534418658617baac028e27502620d5380c9","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-02-17 12:31:59.000000000","message":"Patch Set 2: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6f4e030fadf6c83146380e64fb3ea13ac60cd1ed","author":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"date":"2026-02-17 14:46:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9155ed92770895be9829f7b221efd5183dc91135","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-02-17 15:53:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d61ef796b6b94f382c2443cf490425bf026d9330","author":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"date":"2026-02-18 07:47:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aa93d37bd6dc13e00ecb7af043e0264d0db391f5","author":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"date":"2026-02-19 10:29:27.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"969b6b29785f0fef65d7b57b3aeebd03f67efdd1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"date":"2026-02-20 11:58:12.000000000","message":"Uploaded patch set 3.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"ecade0bc3035c733559bd6d3233df70922ff9f38","author":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"date":"2026-02-20 11:58:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f7b831464c2445f84170f427feb2c1c2fdab64d9","author":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"date":"2026-02-20 13:01:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3ad7439b211a5b7c9a2e534ec1be891c2e82ccfd","author":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"date":"2026-02-20 13:12:50.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f89b0e6e1d5eebe0bc3dc8a2b05e30ce60563efa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 14:04:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5351a15b2b0d4f08b23b8c7029d7911d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31fd8d58a7fe4027a88a77bb65d31132 : SUCCESS in 4m 03s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4f6fdeb57dff47d3b4a7ee73e290cf10 : SUCCESS in 4m 23s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/9319b86fe67a43f0b6e59e23d3a33d63 : FAILURE in 40m 59s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/429ddba442cb483d86bfd2b093e2e825 : SUCCESS in 53m 41s\n- kolla-ansible-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/1a7e6252bc994f93bdafeba88ce221fd : SUCCESS in 52m 42s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/2d454295f2b542eab2ba373bd6d23a1c : SUCCESS in 1h 19m 39s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/9ee1db1c6b884243a94390bb8ca33dbc : SUCCESS in 55m 03s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/9a2899229adc4774bbfb6a5a9d025daa : SUCCESS in 1h 14m 09s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/872e970c0bd242358ea15c2726eee621 : SUCCESS in 1h 06m 42s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9da64200a0a740de86690483a4581fba : SUCCESS in 55m 43s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1d50e429a6234fbbbd3447772b25c738 : SUCCESS in 53m 08s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/68be7885eb2e40fc82387a0b0b2c5861 : SUCCESS in 1h 14m 52s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/8c6db5457021411ab18d70bc431bc9e8 : SUCCESS in 1h 23m 57s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/aed8276b1bb34a60a1ee7a4e7e6856a6 : SUCCESS in 1h 23m 24s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/589ebfd32ee94c6da57795717868e00f : SUCCESS in 2h 04m 34s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/0883aca995d44f21b11c98a2909a58bb : SUCCESS in 1h 31m 09s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/ba09140539d84d3d98b1e3317843c524 : SUCCESS in 2h 02m 14s (non-voting)\n- kolla-ansible-debian-trixie-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/45bd05ea30e54d2a96d1d5121e49edff : SUCCESS in 40m 02s (non-voting)\n- kolla-ansible-rocky-10-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/11df2cbd49f346ff8dc0e4b1d08ab4ec : SUCCESS in 28m 09s (non-voting)\n- kolla-ansible-ubuntu-noble-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/e30c286d607a405ebfccaf420b1d0423 : SUCCESS in 39m 09s (non-voting)\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/e3769ebde616470687cb11cba2abae3f : SUCCESS in 1h 07m 32s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c7ce1fee773472c99d701be695c112e : SUCCESS in 1h 08m 43s (non-voting)\n- kolla-ansible-debian-trixie-lets-encrypt https://zuul.opendev.org/t/openstack/build/c82d53f14b324bbaa9d06007cd13a8cc : FAILURE in 42m 05s (non-voting)\n- kolla-ansible-ubuntu-noble-lets-encrypt https://zuul.opendev.org/t/openstack/build/a6f56a0662d34f059e82845d0b955afc : SUCCESS in 1h 07m 05s (non-voting)\n- kolla-ansible-rocky-10-upgrade-slurp https://zuul.opendev.org/t/openstack/build/1b06f7dd24104046b801bcce49612a8c : SUCCESS in 1h 20m 10s\n- kolla-ansible-ubuntu-noble-upgrade-slurp https://zuul.opendev.org/t/openstack/build/d25750286aa746b0995d17017e240f79 : SUCCESS in 1h 23m 08s\n- kolla-ansible-ubuntu-noble-cephadm-upgrade-slurp https://zuul.opendev.org/t/openstack/build/c4bedcae919e411ab715022119c63e84 : SUCCESS in 1h 46m 55s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae1775bc9903491ca550d9a3bb054185 : SUCCESS in 2m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/236ea85428b34d4d8be24c1a7c0c22ff : SUCCESS in 4m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/61a20290bb374572adf4c3164b41d3db : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":3},{"id":"ce538ec3dd5c1082d5c586aaef729e1914bdeec4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 14:07:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15783f28dba94016a5292b09bce85363\n\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/948a956e99f04c78b87a2a0a507ae5e1 : FAILURE in 1h 47m 11s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/03b6603c3f2b429ea0bbb84d4032848a : FAILURE in 1h 27m 02s (non-voting)\n- kolla-ansible-debian-trixie-aarch64-podman https://zuul.opendev.org/t/openstack/build/14cb6c1589af4ee1a4982bc075e33a17 : FAILURE in 2h 07m 59s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64-upgrade https://zuul.opendev.org/t/openstack/build/f87516e7bb1c4d47ad603c898bf63566 : FAILURE in 44m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7a95a3b45781dd6f939b323ecd561670ba258930","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"date":"2026-02-20 16:54:40.000000000","message":"Uploaded patch set 4.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"f7911ff5033abd968adaffd74b85e4586750fb66","author":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"date":"2026-02-20 16:56:24.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a9425b45c43ae7d8c25f05532a889c4bbfab329a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 19:09:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db24d7b2ac4d44b88a95670670b19be7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb867fc447fb458bac7b942703d71e75 : SUCCESS in 6m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/aa10c5c566ad4e8f9b59fb2eeda50374 : SUCCESS in 4m 19s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/2a7df7a152254c0c8dff40fa5326171e : FAILURE in 1h 11m 54s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/5988e4729b944240bb35eb6a51734943 : SUCCESS in 49m 09s\n- kolla-ansible-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/efbe51d949b24767960cd577a8873c8d : SUCCESS in 48m 44s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/34070442dd7d4ea089f96267ac39567b : SUCCESS in 1h 05m 59s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/a87b64ee150a4082b262ebe582eee205 : SUCCESS in 50m 43s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/61265b4808484624a7bd385121d231bf : SUCCESS in 1h 09m 40s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/0f2ecb25a9b24cf294af43b34d9ec17b : SUCCESS in 56m 58s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/59892b1bb5f1475195579fa97e20a7f0 : SUCCESS in 53m 38s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f4fed45611804d6e82a6de5f882349dd : SUCCESS in 48m 39s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/02036e73a2ab4c2ea5c7304d80f01906 : SUCCESS in 1h 23m 57s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/bff56b9b26e349bb8591aef0252a7ac0 : SUCCESS in 1h 17m 45s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/e8847830472749a0bb43ac59f7e502f1 : SUCCESS in 1h 30m 30s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/c86bc5d1a0b24450bc054305c1374554 : SUCCESS in 2h 12m 39s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/556b9de928804c149dba04c1fbd7fc91 : FAILURE in 53m 56s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/aafc0f9cbe1d49c1ad0cd81fbd5c3b91 : SUCCESS in 2h 09m 38s (non-voting)\n- kolla-ansible-debian-trixie-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/b73ba2e9eb44454e8fc0cb55c07312e1 : SUCCESS in 50m 14s (non-voting)\n- kolla-ansible-rocky-10-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/b0b6c66d904b4ffebb5692d96b3cc011 : SUCCESS in 46m 09s (non-voting)\n- kolla-ansible-ubuntu-noble-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/8a49d44b780a42a2a5869c90b47fdf1d : SUCCESS in 25m 23s (non-voting)\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/90490ab9e3274daa89b73b9660da3648 : SUCCESS in 1h 02m 52s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/06ff63269330413cbf5d0ac78f494ec2 : SUCCESS in 1h 08m 02s (non-voting)\n- kolla-ansible-debian-trixie-lets-encrypt https://zuul.opendev.org/t/openstack/build/33845b42f0db48a587cd1e49fcd337d0 : FAILURE in 34m 52s (non-voting)\n- kolla-ansible-ubuntu-noble-lets-encrypt https://zuul.opendev.org/t/openstack/build/890799764e8747b5a87cabc22fe9f964 : SUCCESS in 1h 19m 08s (non-voting)\n- kolla-ansible-rocky-10-upgrade-slurp https://zuul.opendev.org/t/openstack/build/36deed941c3244c68f2b6f6558ae2c66 : SUCCESS in 1h 07m 22s\n- kolla-ansible-ubuntu-noble-upgrade-slurp https://zuul.opendev.org/t/openstack/build/355f81a7fee6481593eca81babb93cf8 : SUCCESS in 1h 10m 11s\n- kolla-ansible-ubuntu-noble-cephadm-upgrade-slurp https://zuul.opendev.org/t/openstack/build/c6a2f3062007475f924645e0c8e1e849 : SUCCESS in 2h 04m 58s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc71c7aaec024979888777394960357a : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d651c806a70241c296dae4673b91a878 : SUCCESS in 4m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/614cc0cc05b84ebc9c6a16066b8b3057 : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":4},{"id":"313f5be02aa33402f5fc2ac307a01382f4ac3bea","author":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"date":"2026-02-20 21:21:44.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0e3ec337f0f2180bc979a2078cf9aa74f932a9a7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 23:51:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a47430755894750917de9c9a61b4d62\n\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/0e13e53a7cb14563aa0ed8f8068a989c : TIMED_OUT in 3h 39m 00s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/6f9afb7b680943bf9d59dd7f8a9e62d7 : FAILURE in 54m 16s (non-voting)\n- kolla-ansible-debian-trixie-aarch64-podman https://zuul.opendev.org/t/openstack/build/4163f8692a514ebb95ebfa42e3a9550e : FAILURE in 1h 56m 33s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64-upgrade https://zuul.opendev.org/t/openstack/build/97800ef32ec4497eb438886443155676 : FAILURE in 51m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bb31058a7f0043d0b02415ce48d15113079a36c2","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-02-23 09:38:42.000000000","message":"Patch Set 4: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"632c4b11fe299e3101e3a4336759b6883531b434","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-02-23 09:38:55.000000000","message":"Patch Set 4: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6370f4bd7fdb9b615bd5f3c62342cfb79ad64021","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 09:39:11.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b8c6c598d8e4647d55b14d8efe38a6e4cd0ad99d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 11:15:22.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2512b28a3c624e44bd6e6520846e23c7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a52a1011e6346888c71de7fac9bac14 : SUCCESS in 7m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a4a351b768ad446182f507b0a8296848 : SUCCESS in 4m 22s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/534d55d9974544a9ae6b70ae8fe24352 : FAILURE in 1h 19m 29s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/ee36f48c87b740ca92e12158abf4bdf4 : FAILURE in 48m 12s\n- kolla-ansible-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/ba5d245ea97f4a2bb7758e8c00985bec : FAILURE in 48m 25s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/d698b8afb8f3408795c79aea582daa42 : FAILURE in 52m 04s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/1922ad73550b46a4b55d648e4f01a8f2 : FAILURE in 1h 28m 31s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/7e8b7a40cf794cf6ae3225e54a3b5827 : FAILURE in 1h 02m 03s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3aed83498d2a4555b59d6b9fb2bee0ee : FAILURE in 50m 25s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/9d206bdb5dd34fe7b7a4ecc2372372cb : FAILURE in 1h 26m 31s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a00794c08c26407fa48d66493a863b48 : FAILURE in 47m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea3d5e01afb54310b750310f675bf6fd : SUCCESS in 2m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dcbd8760ffc749c5bdfb007741989a47 : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":4},{"id":"7eee0f09b9929a3c2f364ba5aaa42ff284759298","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-02-23 19:42:32.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"444012cec4f48c7f924fdf2420e6ac245fd4b2c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 20:25:06.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a02badcb65064eeabcd0dd8f0ea6f908\n\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/b0c5c1a4236b47f3a0fffe3b70399bac : FAILURE in 6m 18s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/b34d02dea93a49f0894674db0a85e3ee : FAILURE in 6m 08s (non-voting)\n- kolla-ansible-debian-trixie-aarch64-podman https://zuul.opendev.org/t/openstack/build/5093c7188e1a41eaa33965af2975805d : FAILURE in 6m 16s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64-upgrade https://zuul.opendev.org/t/openstack/build/a6435e658ee54aae877d0b1e134c585b : FAILURE in 40m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5c650b0e1bfc10c6d1ec0ad0aa0fcfcfa5292ec4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 20:52:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1bd98d7b0a4f464f9dc20e676e4c0381\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb7dd02f1a524bcf966343667d12fa9d : SUCCESS in 4m 14s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/eeadf1b3cb3a4e59a73ce17161e421e8 : FAILURE in 3m 33s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/f8210041160344c8b3909a837c24e1f3 : FAILURE in 4m 27s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/b852e885d8bf4dd5af9062eb193fa30a : FAILURE in 4m 11s\n- kolla-ansible-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/fb237810ae904c39947c21bae4fa8856 : FAILURE in 4m 21s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/46129f1aa8a340e899092413f8580eff : FAILURE in 34m 43s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/fee36581ef7e4819806bed8217c8ef36 : FAILURE in 4m 49s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/124079d464b946c2b511ff990c97ed4f : FAILURE in 37m 34s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/5f3e3f82654c429c976316dc489aac08 : FAILURE in 3m 56s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eddd086eb60d43c6bdc6ad1f9c20b4ce : FAILURE in 3m 46s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6e536925d6ec43e3a9c9ffadc22ecc22 : FAILURE in 3m 55s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/0f488cb4916a45e38e21d55a99637b1c : FAILURE in 39m 33s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/c442618ed6fd4c3aba90c52f1a957d59 : FAILURE in 5m 35s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/a780bcd9e73e4225bc88885ee4dbcd31 : FAILURE in 5m 42s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/0720b2c16cf64bf5ba3d06da94ff13ad : FAILURE in 57m 08s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/8936f9a92760443182ba35085cb41c40 : FAILURE in 7m 23s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/e5a190a8d12643ee9b8f0152efc51a68 : FAILURE in 58m 19s (non-voting)\n- kolla-ansible-debian-trixie-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/b73a27ac13014174bcfaff67e9f0e401 : FAILURE in 6m 44s (non-voting)\n- kolla-ansible-rocky-10-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/3dbbd08e5b6c42769b325c76a974817b : FAILURE in 3m 09s (non-voting)\n- kolla-ansible-ubuntu-noble-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/cd071ae1c0084741ad4139a153762bb6 : FAILURE in 6m 12s (non-voting)\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c6e6df9748e240a6855a741e9e64159d : FAILURE in 10m 46s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b60853673ae4806824bd4d4402d6c75 : FAILURE in 9m 30s (non-voting)\n- kolla-ansible-debian-trixie-lets-encrypt https://zuul.opendev.org/t/openstack/build/7f1ca3e3e63a405884e3b753ca72d976 : FAILURE in 6m 44s (non-voting)\n- kolla-ansible-ubuntu-noble-lets-encrypt https://zuul.opendev.org/t/openstack/build/70b79ed146fb48deaa77a5b4bf16da14 : FAILURE in 5m 13s (non-voting)\n- kolla-ansible-rocky-10-upgrade-slurp https://zuul.opendev.org/t/openstack/build/84ce0911ddfe45dfb37bab494a475286 : FAILURE in 32m 17s\n- kolla-ansible-ubuntu-noble-upgrade-slurp https://zuul.opendev.org/t/openstack/build/f2b71f0291cf4ffa8263ab5b5c5ddda8 : FAILURE in 32m 02s\n- kolla-ansible-ubuntu-noble-cephadm-upgrade-slurp https://zuul.opendev.org/t/openstack/build/28b81f12a01c48b0a417d012bf84ea20 : FAILURE in 1h 05m 50s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db2648b74adc403682a9ceb4f3bcefeb : FAILURE in 2m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b97a96155c214b52ad0387d6709e6499 : FAILURE in 1m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/79a5a34eb1664283b642fbd9265332c9 : FAILURE in 3m 46s","accounts_in_message":[],"_revision_number":4},{"id":"a4b9457bdffd4178dde5a6ca7f46788eb463700d","author":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"date":"2026-02-27 09:06:48.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8c6e641b929c625bed262689bd672375d3d702b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 11:14:12.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3604f0b41bcc4d889508c4d0a34da3b3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5796208d07c47a288c77c5d6df44fb8 : SUCCESS in 7m 14s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5dc2fce134fc4ac4b66d7052e03c2fef : SUCCESS in 10m 48s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/318531bc49f446eeb218fed7172bb703 : FAILURE in 26m 12s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/e9e4672602b54d96a25d43dceb99ba41 : SUCCESS in 53m 26s\n- kolla-ansible-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/4c4ab3c478f14228bd642e4507a2c559 : SUCCESS in 46m 00s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/588e5691154e4ef18390bb4459138d94 : SUCCESS in 1h 20m 05s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/40598ba5718c4f13aa6bb840edba98c4 : SUCCESS in 1h 00m 42s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/59f45e6676ff4709884f58949fb8417e : SUCCESS in 1h 37m 53s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/774979f1a4ad46e89abf6f3910837f07 : SUCCESS in 55m 34s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e3273116c4246bd961b30a53f2e217a : SUCCESS in 1h 02m 07s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/19e9867eb6e1417181b7377392792778 : SUCCESS in 1h 06m 14s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/252b01d20dd64753baf2a05f58683257 : SUCCESS in 1h 26m 23s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/7a6accff20f640eaadc5c7877a09911e : SUCCESS in 1h 30m 30s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/690ab84c34554ed6bec126612b667346 : SUCCESS in 1h 34m 51s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/44db526cd5894f19a9e1537d816778ef : FAILURE in 2h 02m 07s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/37885a7d94914a8c891473fb78287236 : SUCCESS in 1h 33m 05s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/4595df3a93e64ecfbae4f80bc24555e5 : FAILURE in 58m 39s (non-voting)\n- kolla-ansible-debian-trixie-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/de43e379a5804f6883ac63dd7febe223 : SUCCESS in 53m 33s (non-voting)\n- kolla-ansible-rocky-10-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/9d28b7747ce64e65bd608c07440b6739 : SUCCESS in 51m 49s (non-voting)\n- kolla-ansible-ubuntu-noble-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/74edff91e57f4d47bf91424876be9c29 : SUCCESS in 59m 48s (non-voting)\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f8b560b2f24f4408b650909eba2dd2f1 : SUCCESS in 56m 41s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/63c8357663044f918d82941476393d0f : SUCCESS in 1h 03m 54s (non-voting)\n- kolla-ansible-debian-trixie-lets-encrypt https://zuul.opendev.org/t/openstack/build/8b5e3ce132704778bb7d03cee3d2fdb7 : FAILURE in 23m 57s (non-voting)\n- kolla-ansible-ubuntu-noble-lets-encrypt https://zuul.opendev.org/t/openstack/build/21791394169949b88510761fcbc3c9e4 : FAILURE in 1h 07m 08s (non-voting)\n- kolla-ansible-rocky-10-upgrade-slurp https://zuul.opendev.org/t/openstack/build/b88f8fe4bd3942079961c6a582512e81 : FAILURE in 30m 33s\n- kolla-ansible-ubuntu-noble-upgrade-slurp https://zuul.opendev.org/t/openstack/build/21b6d1d4bdba4efc96be2f98d0a5367d : SUCCESS in 1h 13m 54s\n- kolla-ansible-ubuntu-noble-cephadm-upgrade-slurp https://zuul.opendev.org/t/openstack/build/d9784ba799e94550869aeadce86b9efa : SUCCESS in 2h 05m 47s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fbd93548135e460d8d8e1060f6079409 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5058d6ebd22a41fb95fe853952724074 : SUCCESS in 3m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9cb0def7d5d445e1bc4881d625f0773a : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":4},{"id":"4e535a622dcebf871d855aa8d308b8a6b315f234","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 13:15:05.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c58c6bb22d2468a83035f5344fd17ee\n\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/cf3b2ef22f954cdb891336de74e72bf5 : TIMED_OUT in 4h 02m 19s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/a42325c1503c4e8d9b243d919d0db865 : TIMED_OUT in 3h 29m 51s (non-voting)\n- kolla-ansible-debian-trixie-aarch64-podman https://zuul.opendev.org/t/openstack/build/1f15a9c0830f41a49c33a671e0d20a60 : FAILURE in 1h 39m 28s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64-upgrade https://zuul.opendev.org/t/openstack/build/8c2525abc17b4df7944d09330a5d1ba8 : FAILURE in 47m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a840cfb08d3555ceea8581dce2c8283d8fdc8570","author":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"date":"2026-02-27 23:07:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f275678d36c2c12c26ed3dfe8a8822699507ea3a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 00:30:52.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d801d35b70184390ba4b876d8f1aac0c\n\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/23204f4cb75f46169217d821bc40d399 : FAILURE in 1h 21m 10s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/90ca0100e84748cebc5d63d8f578dd3b : FAILURE in 1h 20m 35s (non-voting)\n- kolla-ansible-debian-trixie-aarch64-podman https://zuul.opendev.org/t/openstack/build/7a472cba583e41079216b5582b4b5b09 : FAILURE in 1h 15m 17s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64-upgrade https://zuul.opendev.org/t/openstack/build/19715b4ace5b4a1d940f68664ebd282f : FAILURE in 1h 11m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9c94263a67f98c2d054ccdd5e454f1f09e132147","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 01:14:10.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/269e6a1cc8004bf1be41ae73afa9030a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cabc3813ac024887b340126cdf5ab413 : SUCCESS in 3m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5302d399898b483d823f460b329dfbb3 : SUCCESS in 4m 16s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/e0c44151723a4960803b36053b775637 : FAILURE in 38m 58s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/66f0a2e808e54c7ba21d82b3c4e63b4d : SUCCESS in 53m 14s\n- kolla-ansible-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/f52b55d6749b4a619588460d5e77c688 : SUCCESS in 53m 30s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/da3f430ce995480ca70a94228943771a : SUCCESS in 1h 08m 30s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/7957f7dd46b440909b168a5dc5a352c7 : SUCCESS in 59m 41s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/834311bddb134d59b9e4743588fef2b0 : SUCCESS in 1h 21m 58s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/475ec551e4374f50b2670348558ae6cd : SUCCESS in 59m 09s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a71fb702afaa4774b542b8ba9e9e9284 : SUCCESS in 57m 15s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/44e5e0b676fa477fb5706d8b66c0e919 : SUCCESS in 54m 49s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/365a16e5329144c5a2e93548ad0e02fc : SUCCESS in 1h 12m 39s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/ef0ad0f7fc894fd99683724ca47bbc7c : FAILURE in 1h 04m 24s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/45fafa08ea384cbeac4d8601f2e993d5 : FAILURE in 1h 29m 04s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/2626b1073fc540c88d40e3704b9061d0 : FAILURE in 1h 07m 35s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/cea31bfdb9ea47fb9e430554ce7989d3 : SUCCESS in 1h 21m 30s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/3cef746e74d04161addf9283609f7650 : SUCCESS in 2h 05m 00s (non-voting)\n- kolla-ansible-debian-trixie-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/3cfcc854be434397ba915800f3d982bb : SUCCESS in 42m 03s (non-voting)\n- kolla-ansible-rocky-10-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/fa8d88dbf63e482290c750c677f4efdd : SUCCESS in 48m 15s (non-voting)\n- kolla-ansible-ubuntu-noble-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/b6ad47cb217040a1ba5df00bef0a55e6 : SUCCESS in 27m 35s (non-voting)\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/a50306b7c25b4aebaadd07ef3cf10a92 : SUCCESS in 1h 05m 09s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/81e5a8b776bb4ec1af7be4e06ca7ae22 : SUCCESS in 1h 04m 12s (non-voting)\n- kolla-ansible-debian-trixie-lets-encrypt https://zuul.opendev.org/t/openstack/build/c302e035a4d445529f6bde5eb99678a6 : FAILURE in 33m 39s (non-voting)\n- kolla-ansible-ubuntu-noble-lets-encrypt https://zuul.opendev.org/t/openstack/build/e9f2ae9e8d7449ca9c99629ef0f34b9f : SUCCESS in 1h 17m 21s (non-voting)\n- kolla-ansible-rocky-10-upgrade-slurp https://zuul.opendev.org/t/openstack/build/551b0b8400d24c9199160011ebb6b5e2 : SUCCESS in 1h 17m 19s\n- kolla-ansible-ubuntu-noble-upgrade-slurp https://zuul.opendev.org/t/openstack/build/1b1cb8d7d5d74f808d1af5e5b2c81cd3 : SUCCESS in 1h 17m 37s\n- kolla-ansible-ubuntu-noble-cephadm-upgrade-slurp https://zuul.opendev.org/t/openstack/build/6fad9ebac4204679aa53bf798f751bd3 : FAILURE in 1h 24m 27s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3379d54754db4a85babf17afba60f6ad : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/856ccbc92d68494fa9de89fb19085a97 : SUCCESS in 2m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/40fdfaab62794a079dafbaf01da49be8 : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":4},{"id":"719de20d84a27d046091d1f2bc1e66dfe4f26358","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 01:14:22.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"220d89244909840a30aa075be288450ac00a7491","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 02:29:37.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b6dc8da994114bd7b71180378e1b60ea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47b5891c2596448ebc91fc1e85adfdcc : SUCCESS in 8m 26s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/30d3b0d230984fdc9f135b9aed9ab2aa : SUCCESS in 11m 15s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/b467e06bb7994eab9b8e12e65ae6c1b2 : SUCCESS in 1h 10m 46s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/157174a2436c42f5b8f5d9af1b923171 : SUCCESS in 47m 27s\n- kolla-ansible-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/c54f5e2c2f7f44c7bfdb7887b131406b : SUCCESS in 49m 42s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/7e755670450843df943a3c69d1cdc191 : SUCCESS in 51m 35s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/e3a94d0bc355480eaacb7ea9cd23a056 : SUCCESS in 1h 13m 35s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/dabe2b9155094796a20614017f903843 : SUCCESS in 53m 34s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/89666bf9a982417789c501faf52494ce : SUCCESS in 52m 13s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/5a42537f7da34470ab271f301f6e026c : FAILURE in 1h 14m 21s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/935937d2b8a24d65a93a86210ad5437b : SUCCESS in 52m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/297bda06729b45c4a9ff28ff79386752 : SUCCESS in 3m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1f3321bd6f4b41caafd5315bb5a94c2d : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":4},{"id":"fcc3bd8bd94e3bb18ccec17de13215949f2ad3f3","author":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"date":"2026-03-02 18:37:51.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"112b77a7f4aef9b7cd3ca9d78e89bc921835b4c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 20:38:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c42044338374264978299770ae35c84\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/511c5cc5db214ab2a008f70ad37eee1f : SUCCESS in 5m 31s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/12c5d21fc19e4baabc39b073f53415c8 : SUCCESS in 9m 10s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/94aa8256f009464d964397fd5dd99dd2 : FAILURE in 55m 38s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/de7aedb993cb43be832742ff71b998f3 : SUCCESS in 52m 27s\n- kolla-ansible-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/f0c1462a66164820a4852defd5ca63ef : SUCCESS in 49m 37s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/dd1befe6f9d541549be7d114e1a368c1 : SUCCESS in 1h 09m 32s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/dd69773ea7b24404a1640db0fc8ada91 : SUCCESS in 54m 28s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/c526fb3815814ceaa2451e63388a59cc : SUCCESS in 1h 20m 06s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/aaa1615a98f34884964eab30849084fe : SUCCESS in 53m 53s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dea88e1b55e9484c8ee1358240e2236b : SUCCESS in 52m 39s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/77a08d1a495146a9858f59a350fb5eec : SUCCESS in 51m 59s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/3548192541f045dbabdb0ac5b7838d55 : SUCCESS in 1h 14m 50s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/14568fd5c4ec4a279c91c82dfb371866 : SUCCESS in 1h 16m 37s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/b87400a8be344a6c87a854c835417b45 : SUCCESS in 1h 22m 05s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/7f704fcbf4c2447f9e7cce1743053cb2 : FAILURE in 1h 53m 26s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/499e2298a9c041bc95b5b7742251f720 : FAILURE in 1h 06m 40s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/b1c34e14bfee4076ac06d3988d285987 : SUCCESS in 1h 59m 17s (non-voting)\n- kolla-ansible-debian-trixie-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/e536c0a8eded4aeea455a0516eab62ee : SUCCESS in 39m 35s (non-voting)\n- kolla-ansible-rocky-10-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/a213453c81e941ac99a095dc76e2546f : SUCCESS in 46m 31s (non-voting)\n- kolla-ansible-ubuntu-noble-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/ac33e36f580141f184c862c2c18749a8 : SUCCESS in 53m 52s (non-voting)\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/48b254f7e0c04ab59de862895c87b480 : SUCCESS in 1h 00m 06s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f08999f2da94c099e319ce292eeedce : SUCCESS in 1h 00m 56s (non-voting)\n- kolla-ansible-debian-trixie-lets-encrypt https://zuul.opendev.org/t/openstack/build/883ce1e62ad2487bab55692cd5067c4c : FAILURE in 31m 16s (non-voting)\n- kolla-ansible-ubuntu-noble-lets-encrypt https://zuul.opendev.org/t/openstack/build/d421bff644d74a3da7d241f2cd81dbd6 : SUCCESS in 1h 08m 50s (non-voting)\n- kolla-ansible-rocky-10-upgrade-slurp https://zuul.opendev.org/t/openstack/build/ba04a19a8a7d440c95bcfd90d446d554 : SUCCESS in 1h 06m 26s\n- kolla-ansible-ubuntu-noble-upgrade-slurp https://zuul.opendev.org/t/openstack/build/7ef5649db4804dcdbba6f64db2ca3d0f : SUCCESS in 1h 10m 21s\n- kolla-ansible-ubuntu-noble-cephadm-upgrade-slurp https://zuul.opendev.org/t/openstack/build/4919d59f6afa4d2ea990a811f7d6b3f3 : FAILURE in 1h 05m 57s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31742ff5ca09479db5a980297507d401 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c38b6b5f1784d4ca9e96ac2d302f1b9 : SUCCESS in 3m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/734d6c9f83a34014983267fb096f629b : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":4},{"id":"7776ef28878351db178943c9fc5fb31baf719d30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 20:38:58.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"cc76d93dcfd36f96abb752e27c5d4bb98e072e7c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 21:48:08.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54354f5340ff4d35b747ccfc7c26c213\n\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/56a79e2b2ba84fedb568808f17156c3f : FAILURE in 1h 58m 13s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/79388c4b5fbc4f229546ec2dc85d1fac : FAILURE in 1h 06m 56s (non-voting)\n- kolla-ansible-debian-trixie-aarch64-podman https://zuul.opendev.org/t/openstack/build/aa8af0111d8c48069e6ea7b262115289 : FAILURE in 1h 38m 04s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64-upgrade https://zuul.opendev.org/t/openstack/build/6f9e9f0fa8d94ca3a637bf0870c9c98b : FAILURE in 49m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1660d0e58613fec8ae3b52f65ef12f047b5a120e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 21:57:20.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8753b4671ae44abb8515978d00cca26e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2e15700983e43dcb523300e65a1fecc : SUCCESS in 7m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d1284b5420804b889cc22712cefb2dee : SUCCESS in 10m 20s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/b9ec99ce0dc843619ced7a833cf5e9f9 : SUCCESS in 1h 14m 47s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/4473031833ff4f4eade4abcfd0e2f835 : SUCCESS in 50m 30s\n- kolla-ansible-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/58bac74878334b81b477ffddc25890dd : SUCCESS in 51m 04s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/86d15942e01749249a6dc64118df006b : SUCCESS in 52m 15s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/1f917c86168f4e03b421fce83e7255a9 : SUCCESS in 1h 17m 45s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/e2eef6bfafe945159d51187242fe211a : SUCCESS in 55m 30s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/842567c600ee469ca0215cb8fa254aa3 : SUCCESS in 52m 10s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/aa783f0832894454b56eec44b34632d5 : SUCCESS in 1h 15m 05s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/48fb2c55b2954cdda02a62978514c207 : SUCCESS in 51m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72495bd33bce4edf9aa0b2e526a753ac : SUCCESS in 4m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e2106ddf55554505bd927f060176fa5d : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":4},{"id":"a5b68d7b43c64cee3133b5e2e7870d4413fb5749","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 21:57:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"19438c67c1223d160f09cf248124b3f0a0232ed1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 21:58:20.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c82507d6613e4e0eb4c0a8f430b8526e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cc3ccccae6145d5acaa30d2ef08610e : SUCCESS in 46s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4939a11e6f59de9c5e8512800b8c0ac659d109ff","revisions":{"1ef7d4c83e726295de7a5db2beccec40f5120f04":{"kind":"REWORK","_number":1,"created":"2026-02-02 16:38:07.000000000","uploader":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"ref":"refs/changes/21/975421/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/975421/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/975421/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/975421/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/975421/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/975421/1"}}},"commit":{"parents":[{"commit":"2157483456a1c0b2a9d67f600c08e47b807b4edd","subject":"Merge \"Try not to return loopback addresses with kolla_address in ipv4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/2157483456a1c0b2a9d67f600c08e47b807b4edd"}]}],"author":{"name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","date":"2026-02-02 16:29:45.000000000","tz":60},"committer":{"name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","date":"2026-02-02 16:37:45.000000000","tz":60},"subject":"Allow SNI frontend when using a single haproxy VIP","message":"Allow SNI frontend when using a single haproxy VIP\n\nCurrently, if there is only one VIP defined,\nthe external backends for haproxy are not defined\nin the configuration.\n\nIf haproxy_single_external_frontend is enabled,\nthis causes issues because the map for external\nfrontends points to those external backends.\n\nThis fixes restore the rendering of external\nbackends if haproxy_single_external_frontend is\nenabled to make public interfaces work in this\nscenario.\n\nChange-Id: Ie41d001671cde41e6574ff18d1d06c9d8966a4a3\nSigned-off-by: Bertrand Lanson \u003cbertrand.lanson@infomaniak.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1ef7d4c83e726295de7a5db2beccec40f5120f04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1ef7d4c83e726295de7a5db2beccec40f5120f04"}]},"branch":"refs/heads/master"},"d2e9ee75b818ab00a757485f3ef6d505290074a7":{"kind":"REWORK","_number":2,"created":"2026-02-16 14:13:15.000000000","uploader":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"ref":"refs/changes/21/975421/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/975421/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/975421/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/975421/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/975421/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/975421/2"}}},"commit":{"parents":[{"commit":"2157483456a1c0b2a9d67f600c08e47b807b4edd","subject":"Merge \"Try not to return loopback addresses with kolla_address in ipv4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/2157483456a1c0b2a9d67f600c08e47b807b4edd"}]}],"author":{"name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","date":"2026-02-02 16:29:45.000000000","tz":60},"committer":{"name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","date":"2026-02-16 14:12:39.000000000","tz":60},"subject":"Allow SNI frontend when using a single haproxy VIP","message":"Allow SNI frontend when using a single haproxy VIP\n\nCurrently, if there is only one VIP defined,\nthe external backends for haproxy are not defined\nin the configuration.\n\nIf haproxy_single_external_frontend is enabled,\nthis causes issues because the map for external\nfrontends points to those external backends.\n\nThis fixes restore the rendering of external\nbackends if haproxy_single_external_frontend is\nenabled to make public interfaces work in this\nscenario.\n\nChange-Id: Ie41d001671cde41e6574ff18d1d06c9d8966a4a3\nSigned-off-by: Bertrand Lanson \u003cbertrand.lanson@infomaniak.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/d2e9ee75b818ab00a757485f3ef6d505290074a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/d2e9ee75b818ab00a757485f3ef6d505290074a7"}]},"branch":"refs/heads/master"},"f4640e4b4bee816831a036e12c383e234538239a":{"kind":"REWORK","_number":3,"created":"2026-02-20 11:58:12.000000000","uploader":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"ref":"refs/changes/21/975421/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/975421/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/975421/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/975421/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/975421/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/975421/3"}}},"commit":{"parents":[{"commit":"2157483456a1c0b2a9d67f600c08e47b807b4edd","subject":"Merge \"Try not to return loopback addresses with kolla_address in ipv4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/2157483456a1c0b2a9d67f600c08e47b807b4edd"}]}],"author":{"name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","date":"2026-02-02 16:29:45.000000000","tz":60},"committer":{"name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","date":"2026-02-20 11:58:06.000000000","tz":60},"subject":"Allow SNI frontend when using a single haproxy VIP","message":"Allow SNI frontend when using a single haproxy VIP\n\nCurrently, if there is only one VIP defined,\nthe external backends for haproxy are not defined\nin the configuration.\n\nIf haproxy_single_external_frontend is enabled,\nthis causes issues because the map for external\nfrontends points to those external backends.\n\nThis fixes restore the rendering of external\nbackends if haproxy_single_external_frontend is\nenabled to make public interfaces work in this\nscenario.\n\nChange-Id: Ie41d001671cde41e6574ff18d1d06c9d8966a4a3\nSigned-off-by: Bertrand Lanson \u003cbertrand.lanson@infomaniak.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/f4640e4b4bee816831a036e12c383e234538239a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/f4640e4b4bee816831a036e12c383e234538239a"}]},"branch":"refs/heads/master"},"4939a11e6f59de9c5e8512800b8c0ac659d109ff":{"kind":"REWORK","_number":4,"created":"2026-02-20 16:54:40.000000000","uploader":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"ref":"refs/changes/21/975421/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/975421/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/975421/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/975421/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/975421/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/975421/4"}}},"commit":{"parents":[{"commit":"2157483456a1c0b2a9d67f600c08e47b807b4edd","subject":"Merge \"Try not to return loopback addresses with kolla_address in ipv4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/2157483456a1c0b2a9d67f600c08e47b807b4edd"}]}],"author":{"name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","date":"2026-02-02 16:29:45.000000000","tz":60},"committer":{"name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","date":"2026-02-20 16:54:33.000000000","tz":60},"subject":"Allow SNI frontend when using a single haproxy VIP","message":"Allow SNI frontend when using a single haproxy VIP\n\nCurrently, if there is only one VIP defined,\nthe external backends for haproxy are not defined\nin the configuration.\n\nIf haproxy_single_external_frontend is enabled,\nthis causes issues because the map for external\nfrontends points to those external backends.\n\nThis fixes restore the rendering of external\nbackends if haproxy_single_external_frontend is\nenabled to make public interfaces work in this\nscenario.\n\nChange-Id: Ie41d001671cde41e6574ff18d1d06c9d8966a4a3\nSigned-off-by: Bertrand Lanson \u003cbertrand.lanson@infomaniak.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/4939a11e6f59de9c5e8512800b8c0ac659d109ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/4939a11e6f59de9c5e8512800b8c0ac659d109ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
