)]}'
{"id":"openstack%2Fkolla-ansible~984021","triplet_id":"openstack%2Fkolla-ansible~master~I61220404ab926cbb13b1daaf5d9395caa0392483","project":"openstack/kolla-ansible","branch":"master","topic":"kolla-yaml-config","hashtags":[],"change_id":"I61220404ab926cbb13b1daaf5d9395caa0392483","subject":"fluentd: migrate container configuration to YAML format","status":"NEW","created":"2026-04-10 13:11:56.000000000","updated":"2026-04-10 18:35:03.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bc172a92cea76014bb2394869665ea605b498a42","_number":984021,"virtual_id_number":984021,"owner":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-04-10 16:15:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-10 16:15:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3831df144a1c7c70924ee592b8eadc34e6ddf8db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-04-10 13:11:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec9abb3ee54c0f3d004ea0b84bc83e66cc7a499f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-04-10 13:12:09.000000000","message":"Topic set to kolla-yaml-config","accounts_in_message":[],"_revision_number":1},{"id":"3ca0e1877f064e06f4f5f6de23dc0f80ce2e9c06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 16:15:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b6b955f1af0440490a705122e1bdca9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4db582849127458e81fc7db14ca1af6a : SUCCESS in 6m 33s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/aae5034620f849028ee038d186d491f3 : SUCCESS in 11m 08s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/c3ca72e774374a66b1572ae8cfa72362 : SUCCESS in 1h 56m 00s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/6d543ee95ada45ab887a852564269fe3 : SUCCESS in 1h 58m 00s\n- kolla-ansible-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/2fb7c50cd5b54a39815ca21f13476632 : SUCCESS in 58m 25s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/42ca2de9878b4c6b9e534c42fda66964 : SUCCESS in 2h 25m 23s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/a344c8bd72b84cbdb45dcb31742954f8 : SUCCESS in 1h 40m 16s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/72a8c67cdf0345b69a035833202be80c : SUCCESS in 2h 05m 34s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/517b8255d3ac414d9f134f2b8f8aba00 : SUCCESS in 1h 43m 39s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e3d11ac2d5143a5bcaf2085599ab885 : SUCCESS in 2h 04m 13s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f4f5e4ca4bea4893a21c0af9f18afb01 : SUCCESS in 1h 04m 38s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/07a43a3d2aa64600b36fb5f9af5f5fd6 : SUCCESS in 2h 35m 01s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/aa441139e6d14b25aaba3671d7799c7c : FAILURE in 1h 56m 22s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/857681a1e89946f99f6333f6bfd187e4 : SUCCESS in 2h 12m 40s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/c97ddcb04cc14657b23d64b8663ff25e : FAILURE in 2h 07m 29s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/aed7cc18de0642b4a0c447e02ed6e216 : SUCCESS in 2h 01m 19s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/fde51d18d03d41959867497d5ad7d257 : SUCCESS in 2h 54m 08s (non-voting)\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/6f0ea980be4b4724a0874288dd3ac4bd : FAILURE in 50m 53s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e35c88612bde4ad0979168ebf4431d48 : FAILURE in 20m 46s (non-voting)\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/1012c283fb024ca6978e23f6451b1045 : SUCCESS in 43m 14s (non-voting)\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/573c51f8196a47b78293707ee51ad19c : SUCCESS in 29m 58s (non-voting)\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/b0fe62564595482895a99000190d14c3 : SUCCESS in 51m 18s (non-voting)\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/548eeb8fb66c4adfa3fb1ebccf03fbad : SUCCESS in 49m 23s (non-voting)\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/6b6817f6cbc64a17b87526a3ecfab549 : SUCCESS in 57m 38s (non-voting)\n- kolla-ansible-rocky-10-upgrade-slurp https://zuul.opendev.org/t/openstack/build/c4340b83283f4b20a971e9699bd7c435 : SUCCESS in 1h 59m 36s\n- kolla-ansible-ubuntu-noble-upgrade-slurp https://zuul.opendev.org/t/openstack/build/c7d3efc7cc0648479d5d2127b3993c0b : SUCCESS in 1h 58m 51s\n- kolla-ansible-ubuntu-noble-cephadm-upgrade-slurp https://zuul.opendev.org/t/openstack/build/cf3129e3dae44d74933ca3f2463254f7 : SUCCESS in 2h 59m 41s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bdae464dc0cb4f44a499b35375c3c628 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32e3a8ce89034262a47f162f82f1320d : SUCCESS in 4m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b884b7bbb0774749bf0b9cbccec0b52a : SUCCESS in 2m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/455b63e3fc3247a995007aa2421ad29f : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":1},{"id":"bc172a92cea76014bb2394869665ea605b498a42","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 18:35:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/132ae984ee5b4238a2585d63116a71b7\n\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/47a6425c628443aa8ba631e83cd97ac3 : SUCCESS in 1h 43m 15s (non-voting)\n- kolla-ansible-debian-trixie-aarch64-podman https://zuul.opendev.org/t/openstack/build/cee6da4c6ab74d31ba55e585181a4c00 : SUCCESS in 1h 14m 29s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64-upgrade https://zuul.opendev.org/t/openstack/build/38c4f21f3c1a46098069dc09923d2e73 : SUCCESS in 2h 11m 05s (non-voting)\n- kolla-ansible-ubuntu-noble-aarch64 https://zuul.opendev.org/t/openstack/build/f0e0c1260fd4472b817cdad7a29acb64 : SUCCESS in 2h 14m 29s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e769b1a6cb91b49ffc5a3f258b295abac6afd70a","revisions":{"e769b1a6cb91b49ffc5a3f258b295abac6afd70a":{"kind":"REWORK","_number":1,"created":"2026-04-10 13:11:56.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/21/984021/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/984021/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/984021/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/984021/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/984021/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/984021/1"}}},"commit":{"parents":[{"commit":"1a0d36d9727886173e5d5639c0f228566961ba7c","subject":"Merge \"release: Remove grafana volume by default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1a0d36d9727886173e5d5639c0f228566961ba7c"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-04-10 13:11:28.000000000","tz":120},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-04-10 13:11:53.000000000","tz":120},"subject":"fluentd: migrate container configuration to YAML format","message":"fluentd: migrate container configuration to YAML format\n\nTransitions the generation of container configuration files for the\n\u0027fluentd\u0027 role from legacy Jinja2 JSON templates to YAML templates\n(config.yaml.j2).\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/982275\n\nChange-Id: I61220404ab926cbb13b1daaf5d9395caa0392483\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/e769b1a6cb91b49ffc5a3f258b295abac6afd70a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/e769b1a6cb91b49ffc5a3f258b295abac6afd70a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
