)]}'
{"id":"openstack%2Fkolla-ansible~984321","triplet_id":"openstack%2Fkolla-ansible~master~Ifb725c85d5aaf83ec85e29fd186d4993acf5189a","project":"openstack/kolla-ansible","branch":"master","topic":"kolla-yaml-config","hashtags":[],"change_id":"Ifb725c85d5aaf83ec85e29fd186d4993acf5189a","subject":"ironic: migrate container configuration to YAML format","status":"NEW","created":"2026-04-13 08:59:16.000000000","updated":"2026-04-13 11:06:41.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"70fda349fba9725124f53a113202ce6e13dd5b50","_number":984321,"virtual_id_number":984321,"owner":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-04-13 11:06:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-13 11:06:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5d61c62f83694894672834756669d6f6eaad2fe8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-04-13 08:59:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"918f93cb2462eeb6f107ae858c7f66032f8cd493","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-04-13 09:02:42.000000000","message":"Topic set to kolla-yaml-config","accounts_in_message":[],"_revision_number":1},{"id":"70fda349fba9725124f53a113202ce6e13dd5b50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 11:06:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/486be0d1b8e543298fd746969bab71d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af81ec5bc07944e6aace26464842384f : SUCCESS in 7m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/48a9a4b50ea14209959b341bdac4f99b : SUCCESS in 7m 04s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/4091622982c14dfcaae6ee98747ff05e : FAILURE in 27m 35s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a9c829086fb44f0a1c8bf719730e0c2 : FAILURE in 25m 11s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/7f4c3fd8c19a42d6b13a7517b53d50f9 : FAILURE in 1h 00m 50s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/2ecf2fb0d92149179f6195ebbc2dac79 : FAILURE in 1h 04m 56s (non-voting)\n- kolla-ansible-rocky-10-ironic-upgrade https://zuul.opendev.org/t/openstack/build/f1ac407734a041a4a1492c86f54ed873 : FAILURE in 2h 00m 16s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/2e8ab84c06414d0a99918a1e2b2bd7ee : FAILURE in 1h 02m 35s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic-upgrade https://zuul.opendev.org/t/openstack/build/68773c4e45f34e8195e1b6d2143b000b : FAILURE in 2h 03m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a805e60ac6eb4b09be555a8f59e264c7 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3948fd6b78a34783ad4c9118b29295a0 : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5bae7f4324d745b5bb8250e651014a11 : SUCCESS in 4m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/804e9a01b7434e95b4a9a6c7f32c2462 : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0d0b1a1f64434e617247fd77c6602b5a28d4d55c","revisions":{"0d0b1a1f64434e617247fd77c6602b5a28d4d55c":{"kind":"REWORK","_number":1,"created":"2026-04-13 08:59:16.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/21/984321/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/984321/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/984321/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/984321/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/984321/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/984321/1"}}},"commit":{"parents":[{"commit":"d0664e9f48eeefd9e74cd1e297c8e97d7f18838f","subject":"Merge \"release: Retarget removal of ironic-inspector\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/d0664e9f48eeefd9e74cd1e297c8e97d7f18838f"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-04-13 08:57:52.000000000","tz":120},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-04-13 08:59:14.000000000","tz":120},"subject":"ironic: migrate container configuration to YAML format","message":"ironic: migrate container configuration to YAML format\n\nTransitions the generation of container configuration files for the\n\u0027ironic\u0027 role from legacy Jinja2 JSON templates to YAML templates\n(config.yaml.j2).\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/982275\nChange-Id: Ifb725c85d5aaf83ec85e29fd186d4993acf5189a\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/0d0b1a1f64434e617247fd77c6602b5a28d4d55c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/0d0b1a1f64434e617247fd77c6602b5a28d4d55c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
