)]}'
{"id":"openstack%2Fkolla-ansible~989021","triplet_id":"openstack%2Fkolla-ansible~master~I0595964e6dec3958d939a7bf7c12c4a8d7be82bc","project":"openstack/kolla-ansible","branch":"master","attention_set":{},"removed_from_attention_set":{"37203":{"account":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"last_update":"2026-05-25 19:34:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0595964e6dec3958d939a7bf7c12c4a8d7be82bc","subject":"ci: skip idempotence test for extra containers","status":"MERGED","created":"2026-05-18 19:19:36.000000000","updated":"2026-05-25 19:35:48.000000000","submitted":"2026-05-25 19:34:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"989021","meta_rev_id":"ba16eadb988f3518901c48c6e085a898e103b98d","_number":989021,"virtual_id_number":989021,"owner":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-25 19:34:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-05-25 10:00:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-05-25 12:16:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-05-25 17:58:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-05-25 17:58:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-05-25 10:00:52.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-18 21:11:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-25 10:00:46.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2026-05-25 10:00:52.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-05-25 12:16:10.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"9ef529fff0699d5743fc0ea6f195dadefc5996ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"date":"2026-05-18 19:19:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"94dc275c6bb6c212791d8a3fa12300972b03e63a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 21:11:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0128db0fefbd4ccabf70784b6eea0661\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/926dc413028648bea61bb50b2fe1ba46 : SUCCESS in 7m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fa5f565650df4092b59d55771d4c1ecf : SUCCESS in 11m 01s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/02f59fa09d0f48ef9e86e66d26060101 : SUCCESS in 1h 19m 37s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/9ec852162a9949af9be879716399cb66 : SUCCESS in 1h 29m 56s\n- kolla-ansible-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/2218788cf4954f73866bf09be0014b1d : SUCCESS in 1h 03m 00s\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/59f8da32abf84047a7ea5e7ea9d0847e : SUCCESS in 1h 49m 57s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/f9dfb151cc24402d88fbf6cd5ba442ad : FAILURE in 14m 33s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/b947dc6229bb4c7db982bf6900d1941f : FAILURE in 14m 33s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/bcdf23401efe42118e3aecc51a061d27 : SUCCESS in 54m 45s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/40a4fc0b92e14faab496782830dd99b7 : SUCCESS in 54m 32s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/21c471fe77164388a2f5934809690142 : SUCCESS in 54m 56s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/23f5bf2ac04745b5be4347b3eb71b3da : SUCCESS in 1h 28m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02c3cecc36834acb883d66fba8dfd65a : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cd8a3890a1c4887b42209865abc4a88 : SUCCESS in 5m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2d2a5a6a980c46a4a9307bf56a673901 : SUCCESS in 5m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c9d94f880d8f43a0ad7824475b61f7dc : SUCCESS in 15m 31s","accounts_in_message":[],"_revision_number":1},{"id":"bffab364f0f546e319bfade998964f159c7e76dc","author":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"date":"2026-05-18 21:35:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"af7cbf7360a68236b6176b8ac6c5269f680ac1c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 23:08:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7843fb3a2a704a2ea6c7b17bf31bc070\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c4f8b24f959467f8e103bf2e59651f8 : SUCCESS in 4m 56s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0c865dbc336c4eb8ac63abdd8c9fcb98 : SUCCESS in 8m 20s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/113874b907e44dac93d7723c6360f07c : SUCCESS in 1h 20m 10s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/2f0942dec19140c699385f8158d3c4dc : SUCCESS in 53m 50s\n- kolla-ansible-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/e455646d19c6483eb45b466870371b3c : SUCCESS in 52m 59s\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/042b88f2d0964c83a3fed4621b4d9765 : SUCCESS in 1h 25m 29s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/12e435a72ad54ae5b6d55a010ad61aa6 : FAILURE in 14m 17s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/2dde638fe1e04d79b1515b8c2ff7d7cb : FAILURE in 15m 05s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/74d0a83da901439eb4f7ac03a9e706f2 : SUCCESS in 58m 00s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a8f948a92aa845c9bfb2b89d527b0eb0 : SUCCESS in 56m 59s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e2388da9ac1b480eabb970883a6031c1 : SUCCESS in 51m 31s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/53fd71bfd6a142889869f159d1db08a0 : SUCCESS in 1h 31m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f957c19c62d44a16a3f8eb5eb978a1e8 : SUCCESS in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87b551407f3c4beeb3fa7d2bd49d3e5e : SUCCESS in 3m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7f81f60c9c5641f7bcbb0ccf8d68326f : SUCCESS in 1m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2042439dddce474d901819756654df34 : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":1},{"id":"52813388ba165986c0e02c5b754b1ad0822b808f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 12:55:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b44346ba83b149368f0ed952794cbc1a\n\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/153b87f8fd6646d09e9dd92b5444a329 : FAILURE in 21m 39s (non-voting)\n- kolla-ansible-debian-trixie-aarch64-podman https://zuul.opendev.org/t/openstack/build/9998f38eb08d47249a4064e6bdbedfd2 : FAILURE in 15m 09s (non-voting)\n- kolla-ansible-debian-trixie-aarch64-upgrade https://zuul.opendev.org/t/openstack/build/d68262c528ee4bb8ac26596a383a59c6 : SUCCESS in 1h 43m 31s (non-voting)\n- kolla-ansible-ubuntu-noble-aarch64 https://zuul.opendev.org/t/openstack/build/b1955348155640069d1d0e2eea0d86f5 : FAILURE in 1h 23m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ae5b02de8c6dfee54f291a890f0c479c93ce314c","author":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"date":"2026-05-22 12:18:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6119cc12cc57a0240bd77d42576e170beba4c00a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 13:55:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fe7241f6f8a94941a3218e9d68547be7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cd1790deaf542309c75326c959201fa : SUCCESS in 3m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1193fd18ef8344b0ba182a3d5f4fe69b : FAILURE in 2m 32s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/0544cdc66773400a87806115302f9439 : SUCCESS in 1h 23m 00s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/a0da361d6ff243b78df7b25f0c6af3b8 : SUCCESS in 52m 51s\n- kolla-ansible-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/e43113d7653f45519858985ac3caef08 : SUCCESS in 53m 10s\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/1b3402e7e79449e48782aeda67dff6bd : SUCCESS in 1h 25m 24s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/00204c6ae29f4fcfb2c5bb4d03305068 : SUCCESS in 1h 00m 07s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/7ffa9fe3c75b42eb8e5f6a40c59672e5 : SUCCESS in 1h 33m 40s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/6c890b93b1ac4c67852082a2c39afa71 : SUCCESS in 58m 38s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/42667e75b59c42dca2b0e0ea816cb6d4 : SUCCESS in 58m 29s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e4ba44a69b7e41afaa4c9195657c5bbb : SUCCESS in 53m 18s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/7449e225e72045d28b8314d0baa5cc4b : SUCCESS in 1h 24m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dea23e6fed284d4a9c16e3f3a833873c : SUCCESS in 2m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2137ad57cfb3443c9e79253b24d10701 : SUCCESS in 2m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a9219177d1ba4d608cf51fd006883dfd : SUCCESS in 2m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98b57e5a1f81447eb894947b80d3950f : SUCCESS in 7m 53s","accounts_in_message":[],"_revision_number":1},{"id":"7ecbff4b3cb8308f6e7dd1d5733c172cfe3e7415","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 23:22:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96bee6e220a34633a368d316a06c65b9\n\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/818c2baa2baf499d8a19820edb56ca02 : NODE_FAILURE Node(set) request 53e6f057366b4b14b9b989f29cf2dcc9 failed in 0s (non-voting)\n- kolla-ansible-debian-trixie-aarch64-podman https://zuul.opendev.org/t/openstack/build/b1edcf7964b843ffa749ccf78af4a10b : FAILURE in 32m 15s (non-voting)\n- kolla-ansible-debian-trixie-aarch64-upgrade https://zuul.opendev.org/t/openstack/build/591a9863f67148589b9dbd7a5e9e34b7 : NODE_FAILURE Node(set) request eeb996f981f642f4bde683a88cd1ce45 failed in 0s (non-voting)\n- kolla-ansible-ubuntu-noble-aarch64 https://zuul.opendev.org/t/openstack/build/3ca69b2ab62345aead951b6b1f2d9a01 : NODE_FAILURE Node(set) request bbd5fcba3bb94b04a1eb6ea032c6c516 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c616581ea90955c9c66da3e4e7564de5f83a4b95","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-25 10:00:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b78a7246a401b62635c36dafb577cc0209215a26","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-25 10:00:52.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"1986bea0ebc92a7ec04403f769b4eea17284a3ec","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-25 12:05:54.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"622482a51de5e8a712d9c8b5af8687d6f77d05ba","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-05-25 12:16:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fc282c1e9b9a0a3a6ea4efa9e55e5e194dafee13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 13:11:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0fed724530a4474bac50ce3ec54d9f42\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d305886d59c41d78d37c9d45d187e7d : SUCCESS in 6m 33s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5ecc686d626c4c71b623eb66979fbd19 : SUCCESS in 10m 36s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/bad0c2ad82244d5ea9852c75d449dbb1 : SUCCESS in 1h 26m 07s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/133ae0fad23b41ab86240e8b4eaaddcb : SUCCESS in 52m 16s\n- kolla-ansible-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/6ce9c31837a94de988394a662264c3f7 : SUCCESS in 49m 31s\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/7497922d59b94aa09707224d9cfabbfa : TIMED_OUT in 3h 02m 03s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/3abe7a25db7f463c966bf816cf13fb6a : SUCCESS in 58m 44s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/e32c6faf2ccb4a2390ab14e43cac7778 : SUCCESS in 1h 28m 21s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/d964d5f5d1af4306a24c7dc1b831520b : SUCCESS in 1h 02m 24s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a5f953f5daf94098843c4746ca475999 : SUCCESS in 53m 35s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7f6948fa98664c46aea5910249fa0919 : SUCCESS in 51m 52s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/3e07fa3dc28444bf85b38e932f262ff6 : SUCCESS in 1h 36m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e1ea2103f044aa98588eee6ea321ca5 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6858ab855e6c4d79bb501464207cbc05 : SUCCESS in 4m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7fb8eee351d6413c83c2d5214752c1a1 : SUCCESS in 2m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/884eda7d4688415194bb103a3476fd2b : SUCCESS in 16m 35s","accounts_in_message":[],"_revision_number":1},{"id":"47bb60d8cfb910a87f8db4e1a5debee8440b3291","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-25 13:21:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8d58e01bdd46d0cf37bf51427d428c5d0aeed262","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 15:01:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d47de738b7004e5db732b57dec84944d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ac9a08b068a468686a5aa346205d20b : SUCCESS in 6m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/782308561d2c422e83e626f7fd3a119a : SUCCESS in 9m 55s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/7884d3e74f804ed895b9c5224293d2a1 : SUCCESS in 1h 19m 42s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/27cda6312cdd4e42aee9ae78a9e3f0bd : SUCCESS in 51m 36s\n- kolla-ansible-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/52fefbf354c64881863f9f13f4fe83b5 : SUCCESS in 59m 00s\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/c80edd42333b4cb2b65a5868812e7dd4 : SUCCESS in 1h 33m 10s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/b930cf0623ef4b9cbd5ab0f4d6b75939 : SUCCESS in 1h 03m 40s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/5d7adcdd37274551ab14c6c32480ba10 : SUCCESS in 1h 29m 06s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/0f4474abcd394a979f8f4fe5dd42923e : SUCCESS in 56m 45s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a764219a62a340dc9ef6b7ded382d7c6 : SUCCESS in 1h 00m 37s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/05a5bc2846a943b88574e57a0fe48046 : SUCCESS in 53m 31s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/29320fd6fc034546937d65af612f5945 : SUCCESS in 1h 39m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a7f1c0eae2d4b388a849bd3849cdb46 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64b615e462c542a6a160d29b40f5095f : SUCCESS in 3m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cf08d70565dd44b295c31d119d71fd35 : SUCCESS in 3m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fbd239714b514c73b3d4ad2d0a1ccd50 : SUCCESS in 8m 24s","accounts_in_message":[],"_revision_number":1},{"id":"47c200ec8695dbcef754fe201ba1a72d413bb1cd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 17:29:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c31f7ae0d49b4ffc8f404cc28e66fdd2\n\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/8b578d4b0bc8419184b25b4b634fbbce : SUCCESS in 2h 26m 07s (non-voting)\n- kolla-ansible-debian-trixie-aarch64-podman https://zuul.opendev.org/t/openstack/build/5877f24bfc0046b3ad781c7849ca7f08 : SUCCESS in 1h 55m 42s (non-voting)\n- kolla-ansible-debian-trixie-aarch64-upgrade https://zuul.opendev.org/t/openstack/build/cedd17c117954ec8953a604c9099789e : SUCCESS in 2h 13m 31s (non-voting)\n- kolla-ansible-ubuntu-noble-aarch64 https://zuul.opendev.org/t/openstack/build/da5d7e6ca89f4dac84fc5026c55354ec : FAILURE in 1h 14m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"948feb2f7af4a94f0139587f1a977123e4ac1a31","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-25 17:58:54.000000000","message":"Patch Set 1: -Workflow Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"8ef987be460a317ce67acfd89da190c17f7ed995","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-25 17:58:57.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"08850f3573d12e78990d65ab3cee8c974db00d8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 17:59:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"91fa4dacc23428b00484a308181e3bb93b8c1b82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 19:34:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fb0d21f94ef4fcda845f75b8ae1a94d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/122cf0b9079c4bc9882340d39f423847 : SUCCESS in 7m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f36421da43534ce69280297726c47c98 : SUCCESS in 9m 00s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/fdb6b78f54ee444b8a39555e1a10a172 : SUCCESS in 56m 08s\n- kolla-ansible-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/af0dcc8702ff46c6930da818f9a72432 : SUCCESS in 54m 30s\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/4d34600c379543b5b283c9dadf851784 : SUCCESS in 1h 31m 01s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/9e99aa20b511473eaddfae1a8e2d1dc8 : SUCCESS in 1h 00m 01s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/cc0ccef2b4de41c99ca6d9a7d0e7d0cf : SUCCESS in 1h 34m 06s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/c0d14bd39fd14c0391463fb30614a383 : SUCCESS in 1h 00m 19s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/254dc471b19e439da39a08af99d80396 : SUCCESS in 57m 29s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/06966bd4055b4757863f733302cfeca2 : SUCCESS in 1h 30m 07s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1da09ed8a22d495993142860cfdd93ca : SUCCESS in 55m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa3aa6a0fd7e476090888239405a9a82 : SUCCESS in 3m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2d904ac030d64a93854b0f919ae52a08 : SUCCESS in 3m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7961e076f2cf40d5bd90aff573b34158 : SUCCESS in 15m 09s","accounts_in_message":[],"_revision_number":1},{"id":"de23b5e3ac0bc0b4367dedd2b98cbdb72893eaa5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 19:34:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ba16eadb988f3518901c48c6e085a898e103b98d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 19:35:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cdfc6ca1c3445fc8dd436a034e2adf1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e825aa8ad7847b987897466594c04cd : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a1850f58454b06e936a76a1fbb1b41d4bdae4f7f","revisions":{"a1850f58454b06e936a76a1fbb1b41d4bdae4f7f":{"kind":"REWORK","_number":1,"created":"2026-05-18 19:19:36.000000000","uploader":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"ref":"refs/changes/21/989021/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/989021/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/989021/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/989021/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/989021/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/989021/1"}}},"commit":{"parents":[{"commit":"3269c79a5aa04c218faf7d6472c432e76d665fb3","subject":"Merge \"tests: remove ovn upgrade workarounds from test-ovn.sh\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/3269c79a5aa04c218faf7d6472c432e76d665fb3"}]}],"author":{"name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","date":"2026-05-18 19:18:01.000000000","tz":120},"committer":{"name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","date":"2026-05-18 19:18:01.000000000","tz":120},"subject":"ci: skip idempotence test for extra containers","message":"ci: skip idempotence test for extra containers\n\nThis introduces a proper skip list to avoid checking idempotence\nfor containers that are not kolla-managed. It allows to extend the\nlist by adding more matching patterns in the future.\n\nChange-Id: I0595964e6dec3958d939a7bf7c12c4a8d7be82bc\nSigned-off-by: Bertrand Lanson \u003cbertrand.lanson@infomaniak.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/a1850f58454b06e936a76a1fbb1b41d4bdae4f7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/a1850f58454b06e936a76a1fbb1b41d4bdae4f7f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
