)]}'
{"id":"openstack%2Fkolla-ansible~998699","triplet_id":"openstack%2Fkolla-ansible~master~I0ea5522b1a3a3dbb2071ff93f6800f898858f41b","project":"openstack/kolla-ansible","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-07-27 08:03:35.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"I0ea5522b1a3a3dbb2071ff93f6800f898858f41b","subject":"CI: Don\u0027t fail on Neutron wrapper managed containers","status":"MERGED","created":"2026-07-25 09:28:55.000000000","updated":"2026-07-27 11:09:21.000000000","submitted":"2026-07-27 11:07:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"998699","meta_rev_id":"bfb84e172ef0bb2b280eea6c99b6c7598e333eeb","_number":998699,"virtual_id_number":998699,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-27 11:07:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-07-27 08:03:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-07-27 07:31:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2026-07-27 07:30:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-07-27 08:03:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-07-27 08:03:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-07-27 08:03:35.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-25 12:10:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-27 07:30:58.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-07-27 07:31:06.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"1a0ace40b62f7beb1a704b5c955ab4c570293df3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-25 09:28:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b0df8484923f6f54775af0631babb5651cef3e10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-25 09:33:54.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"1b4de6e83600f4719a24fdf379367a0f6cbc9a84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-25 09:35:32.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"63fdd5c5a29a02017ab2f9f66bda3f14a1485f38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-25 12:10:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f522eb258a0540a09d51c37f7a42c598\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fec797fd42114702b20af9d9e451ba85 : SUCCESS in 6m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/35d4151c4b4742a2a3262ab35c53211a : SUCCESS in 12m 39s\n- kolla-ansible-debian-trixie-multi-compute-ironic https://zuul.opendev.org/t/openstack/build/2cd66298ebc74c11ac1760e690100971 : SUCCESS in 1h 07m 29s (non-voting)\n- kolla-ansible-rocky-10-multi-compute-ironic https://zuul.opendev.org/t/openstack/build/7dfd887b09be4710a154ca52b26cb68f : SUCCESS in 1h 06m 26s (non-voting)\n- kolla-ansible-ubuntu-noble-multi-compute-ironic https://zuul.opendev.org/t/openstack/build/82e4ba9530764172a8d6b3bda09f7d4b : SUCCESS in 1h 14m 28s (non-voting)\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/3b96f4ef653f4d00a9a400969bb842fd : TIMED_OUT in 2h 33m 52s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/40d70b79085e42a8a887d41e0905f4fc : SUCCESS in 56m 35s (non-voting)\n- kolla-ansible-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/76a1d2b1d29044d8bcdb0d3e2fccd3dc : SUCCESS in 49m 00s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/755ae6b8f99445608e4934b7d1539b9e : SUCCESS in 1h 21m 17s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/7d6a47e4b2b141f6a8e57403954469e6 : SUCCESS in 53m 46s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/d41476058b7b4a6ca719af85a97c33ea : SUCCESS in 1h 24m 13s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/3ebfb553342448deb251c342344fb316 : SUCCESS in 50m 14s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9338d0610f647c7b47726094d9cb51f : SUCCESS in 52m 34s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5948e42384c041c790438686e030657c : SUCCESS in 51m 22s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/b3f9ac7aede1490e8bc293748d32f522 : SUCCESS in 1h 30m 00s\n- kolla-ansible-debian-trixie-bifrost https://zuul.opendev.org/t/openstack/build/1cb4d64d991547b2a98103bd2a373954 : SUCCESS in 15m 13s (non-voting)\n- kolla-ansible-rocky-10-bifrost https://zuul.opendev.org/t/openstack/build/e096248d857e4a54ba451ba1ca79e1fa : SUCCESS in 8m 32s (non-voting)\n- kolla-ansible-ubuntu-noble-bifrost https://zuul.opendev.org/t/openstack/build/4cb815d21e1e49dc95d45e8ca117fcda : SUCCESS in 14m 56s (non-voting)\n- kolla-ansible-debian-trixie-cells https://zuul.opendev.org/t/openstack/build/6e7867ecd49d40b59835453a2d1ac29f : SUCCESS in 1h 09m 36s (non-voting)\n- kolla-ansible-rocky-10-cells https://zuul.opendev.org/t/openstack/build/3e9f87f15b5a473ca31419f0e76adf51 : SUCCESS in 1h 06m 14s (non-voting)\n- kolla-ansible-ubuntu-noble-cells https://zuul.opendev.org/t/openstack/build/0c2281dcae21470a95ab5bdbc0c7fe96 : SUCCESS in 1h 12m 15s (non-voting)\n- kolla-ansible-debian-trixie-cloudkitty https://zuul.opendev.org/t/openstack/build/51031ada019049cc9aac910d9d2aa0ba : SUCCESS in 1h 03m 44s (non-voting)\n- kolla-ansible-rocky-10-cloudkitty https://zuul.opendev.org/t/openstack/build/7acb1456280c419ba8ccbed7a658c1df : SUCCESS in 47m 09s (non-voting)\n- kolla-ansible-ubuntu-noble-cloudkitty https://zuul.opendev.org/t/openstack/build/aa29813345644c848ee048cfe31706f3 : SUCCESS in 59m 44s (non-voting)\n- kolla-ansible-debian-trixie-federation https://zuul.opendev.org/t/openstack/build/11b2840246bb42febf7da185b5d4e6f9 : SUCCESS in 1h 11m 51s (non-voting)\n- kolla-ansible-rocky-10-federation https://zuul.opendev.org/t/openstack/build/f5df11e546884182a1df76964b39ff08 : FAILURE in 42m 34s (non-voting)\n- kolla-ansible-ubuntu-noble-federation https://zuul.opendev.org/t/openstack/build/bc68460ead084b8a93c14663d430cd8f : SUCCESS in 1h 02m 52s (non-voting)\n- kolla-ansible-debian-trixie-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/e79f6972783f4ba887904a7b48922ade : SUCCESS in 1h 01m 11s (non-voting)\n- kolla-ansible-rocky-10-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/42a41a698ef445bbb051a6e2075dbccd : SUCCESS in 34m 44s (non-voting)\n- kolla-ansible-ubuntu-noble-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/51a4b33942ff47559bfe6a7bf0414f5d : SUCCESS in 47m 39s (non-voting)\n- kolla-ansible-debian-trixie-kvm https://zuul.opendev.org/t/openstack/build/d8166e5c232a4ab3b255d3215d7c0a74 : SUCCESS in 51m 16s (non-voting)\n- kolla-ansible-rocky-10-kvm https://zuul.opendev.org/t/openstack/build/1b9cc844837442e69b069a724ed49026 : FAILURE in 48m 49s (non-voting)\n- kolla-ansible-ubuntu-noble-kvm https://zuul.opendev.org/t/openstack/build/8b45f08010bc43f3bdc72deb08ed147a : SUCCESS in 51m 46s (non-voting)\n- kolla-ansible-debian-trixie-lets-encrypt https://zuul.opendev.org/t/openstack/build/369a7d254cf54f5dbfd6cb83314cd387 : SUCCESS in 1h 07m 36s (non-voting)\n- kolla-ansible-ubuntu-noble-lets-encrypt https://zuul.opendev.org/t/openstack/build/590016758aea495db843c6e7f056f6b9 : SUCCESS in 1h 15m 03s (non-voting)\n- kolla-ansible-debian-trixie-magnum https://zuul.opendev.org/t/openstack/build/bc6e9f90c74342a4af1d42647892636a : FAILURE in 30m 58s (non-voting)\n- kolla-ansible-rocky-10-magnum https://zuul.opendev.org/t/openstack/build/826cbe18cbd2472798adea7c69165416 : FAILURE in 32m 13s (non-voting)\n- kolla-ansible-ubuntu-noble-magnum https://zuul.opendev.org/t/openstack/build/aa34a2adaf5d462f9cfd846c0dc52f8d : FAILURE in 1h 01m 29s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/080d892b08df47c2b4eb5e4a764d3c66 : SUCCESS in 21m 44s\n- kolla-ansible-debian-trixie-mariadb-upgrade https://zuul.opendev.org/t/openstack/build/54d999e065444430979c9763b3a417f1 : SUCCESS in 27m 06s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/942e1ff9411b43ec91f8763427e2d6a7 : SUCCESS in 18m 39s\n- kolla-ansible-rocky-10-mariadb-upgrade https://zuul.opendev.org/t/openstack/build/3eb8875fe26a4b5baeff42300e5885b1 : SUCCESS in 26m 43s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/5ce30da7c992469d8b09a1a37a32e7ca : SUCCESS in 26m 20s\n- kolla-ansible-ubuntu-noble-mariadb-upgrade https://zuul.opendev.org/t/openstack/build/3bf60e0fa677413bbb5b0d5a84c3c68a : SUCCESS in 32m 44s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/175aff12c5d747a686edba897d74559a : FAILURE in 38m 16s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/d300e6549490402f9bfb5eb4fcd8b5bb : SUCCESS in 50m 46s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/7b4b3bfdb7c04f4fbc931ec076b3a381 : SUCCESS in 36m 36s (non-voting)\n- kolla-ansible-debian-trixie-nfv https://zuul.opendev.org/t/openstack/build/2769acea96764a06817ab3e33073e394 : FAILURE in 1h 00m 06s (non-voting)\n- kolla-ansible-ubuntu-noble-nfv https://zuul.opendev.org/t/openstack/build/4cc36a77ac3f45ee8451524f7ae50193 : FAILURE in 38m 49s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/88fd360679484049946bca236b253556 : SUCCESS in 1h 11m 40s (non-voting)\n- kolla-ansible-debian-trixie-ironic-upgrade https://zuul.opendev.org/t/openstack/build/03f082760f5c4248815b189eac8c70d7 : SUCCESS in 1h 41m 15s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/e77cc2657e684c02b6a8b28b3bd68643 : SUCCESS in 1h 14m 16s (non-voting)\n- kolla-ansible-rocky-10-ironic-upgrade https://zuul.opendev.org/t/openstack/build/9e44da7f4508474fbeac26fb7fff268b : SUCCESS in 1h 42m 06s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/55bec13047ce444d9c5a387eb68cef68 : SUCCESS in 1h 20m 17s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic-upgrade https://zuul.opendev.org/t/openstack/build/a9600c97d7ba4771b5a74c38c18c01f1 : SUCCESS in 1h 52m 24s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/aa92e607887c478dbd85c6bfdda2f1f1 : SUCCESS in 1h 00m 27s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/20d8a90b63174c3dae10a2c7c442b545 : SUCCESS in 57m 53s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/6b241914a427426fa3ad291d75cb2425 : SUCCESS in 1h 26m 23s (non-voting)\n- kolla-ansible-debian-trixie-ovn-upgrade https://zuul.opendev.org/t/openstack/build/8b40fc87724b453584db691d4ff6d49a : SUCCESS in 2h 09m 59s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/46e05603eb7d4bc1a1afdc910f4cf608 : FAILURE in 1h 09m 59s (non-voting)\n- kolla-ansible-rocky-10-ovn-upgrade https://zuul.opendev.org/t/openstack/build/392292b390644143ad2beebcde7d9a84 : SUCCESS in 2h 21m 16s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/e626ee13a0624076a75dc48718d00b74 : SUCCESS in 1h 34m 59s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn-upgrade https://zuul.opendev.org/t/openstack/build/8502fcfd5583457ea9077fec27066f72 : SUCCESS in 2h 13m 35s (non-voting)\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/9fd88a11d73a47e2b40021c8501594d3 : SUCCESS in 32m 44s (non-voting)\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/fb9140dda9f742039948b57433bc27da : SUCCESS in 37m 48s (non-voting)\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/2486333fc17346f8be0324fdd2bb875b : SUCCESS in 40m 42s (non-voting)\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/2f3c5459ccb84ab0b50177a818769808 : SUCCESS in 40m 08s (non-voting)\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/67757741aeb74c1ba9490199d597550e : SUCCESS in 32m 34s (non-voting)\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/73ebd165d2f7490f8a1b06ba1e620b1c : SUCCESS in 26m 07s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f41eff27bd7f4c249ffeb9494981dc4c : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdb4e7d2f0a34a52b371cafa9b2e9daa : SUCCESS in 4m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d53d783af4e84ea494c65d19fd221de3 : SUCCESS in 2m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/61873018d2414e6490d05c512b3e092b : SUCCESS in 4m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cd4ace8367cd41d588ba2a7fafefe019 : SUCCESS in 5m 20s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c475118ac8a4e8a9fec7165ad1d852c : SUCCESS in 17m 04s","accounts_in_message":[],"_revision_number":3},{"id":"9709f39d440a061431d151a71b42181e7e9f6f99","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-25 13:01:26.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"f09c9513980452deb09fbe7a8082792c3269cbcc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-27 06:42:11.000000000","message":"Patch Set 4: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"de67eb1c6e56929dd0152a397255fec7ee6be385","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-07-27 07:30:58.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"ab8e54410820faf185f2a128cd2020efadc06c97","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-07-27 07:31:06.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"f25550148c7f1d2f625815691a22a08b1f3f8fb2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-27 08:03:35.000000000","message":"Patch Set 4: Backport-Candidate+1 Code-Review+2 Review-Priority+1 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"65b2868b3696e5abb02c1b35b933a018c8563f24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 09:19:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86bb8a9e713a4635b84d5d3abc80b59d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c587e2d1712a493d983b974e77428c99 : SUCCESS in 6m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d59fcc72a57943a5a61b6b0fbb339525 : SUCCESS in 12m 32s\n- kolla-ansible-debian-trixie-multi-compute-ironic https://zuul.opendev.org/t/openstack/build/4bcec14fd416448e916970f96dcdbc27 : SUCCESS in 1h 08m 47s (non-voting)\n- kolla-ansible-rocky-10-multi-compute-ironic https://zuul.opendev.org/t/openstack/build/0524b1e14ee546e29b229a5871b43064 : SUCCESS in 1h 03m 49s (non-voting)\n- kolla-ansible-ubuntu-noble-multi-compute-ironic https://zuul.opendev.org/t/openstack/build/1e941c81cd7d43f682ccb938116c0d3e : SUCCESS in 1h 08m 14s (non-voting)\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/72286fe01a5a4bc8bcf745f2b30945f5 : TIMED_OUT in 2h 34m 39s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/a6bcb2f48d5e4d2ab18bcc58db27e5b4 : FAILURE in 37m 01s (non-voting)\n- kolla-ansible-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/3dada50ed0e846e5b9429c01155b490f : SUCCESS in 48m 46s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/de5b10a1ca084cadac66fcf47ae9466f : SUCCESS in 1h 20m 30s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/47535efdb6544a13ab31ec3cd0b38c0d : SUCCESS in 51m 49s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/a34f2ae1778a41c9b4b633da99ce243b : SUCCESS in 1h 18m 40s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/4863067edb8147769b3e9b1bf2c7e9eb : SUCCESS in 50m 30s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4931677487674dd3868c505366df0af9 : SUCCESS in 53m 56s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9d1a3944bf5a4d52861259694a8143cb : SUCCESS in 51m 31s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/094e1a0b76ba46eb96fb28201dee1fd9 : SUCCESS in 1h 22m 54s\n- kolla-ansible-debian-trixie-bifrost https://zuul.opendev.org/t/openstack/build/4b76fdfeb613482c8235b0af4ec58ed7 : SUCCESS in 15m 13s (non-voting)\n- kolla-ansible-rocky-10-bifrost https://zuul.opendev.org/t/openstack/build/df184e2f460e4053872d79bf940b7e84 : SUCCESS in 15m 55s (non-voting)\n- kolla-ansible-ubuntu-noble-bifrost https://zuul.opendev.org/t/openstack/build/46d2196c4397493eb9c54885db1e193f : SUCCESS in 12m 54s (non-voting)\n- kolla-ansible-debian-trixie-cells https://zuul.opendev.org/t/openstack/build/1d1f686852004e3988e447743e808756 : SUCCESS in 1h 14m 26s (non-voting)\n- kolla-ansible-rocky-10-cells https://zuul.opendev.org/t/openstack/build/a8ef14faf6fa4a72b9bd91aceb962036 : SUCCESS in 1h 16m 46s (non-voting)\n- kolla-ansible-ubuntu-noble-cells https://zuul.opendev.org/t/openstack/build/1a48987aacdc48cd8a90bbb415870ccf : FAILURE in 1h 01m 31s (non-voting)\n- kolla-ansible-debian-trixie-cloudkitty https://zuul.opendev.org/t/openstack/build/e1aaf2312f734f7aa753cebba8a56ee2 : SUCCESS in 1h 01m 50s (non-voting)\n- kolla-ansible-rocky-10-cloudkitty https://zuul.opendev.org/t/openstack/build/9249f4fd5f2b4571ac8dd547ab3243a7 : SUCCESS in 30m 47s (non-voting)\n- kolla-ansible-ubuntu-noble-cloudkitty https://zuul.opendev.org/t/openstack/build/53b2e5dcc3e947a78a6631b85ca24b45 : SUCCESS in 1h 02m 33s (non-voting)\n- kolla-ansible-debian-trixie-federation https://zuul.opendev.org/t/openstack/build/4e38448fc8304bb98ec28b9524cc54f5 : FAILURE in 36m 10s (non-voting)\n- kolla-ansible-rocky-10-federation https://zuul.opendev.org/t/openstack/build/e2a157e4ddae479ea14f9e5249603b57 : FAILURE in 38m 26s (non-voting)\n- kolla-ansible-ubuntu-noble-federation https://zuul.opendev.org/t/openstack/build/65ee1261fd204f869cbec4c94493bb94 : SUCCESS in 58m 56s (non-voting)\n- kolla-ansible-debian-trixie-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/9fd66a39b32642e0b1c257a8402e28a0 : SUCCESS in 54m 00s (non-voting)\n- kolla-ansible-rocky-10-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/26ef0e7ca656418d89ee00a16ad16386 : FAILURE in 34m 57s (non-voting)\n- kolla-ansible-ubuntu-noble-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/0aa73188e2e44949b7bc15141bae9f6a : SUCCESS in 59m 08s (non-voting)\n- kolla-ansible-debian-trixie-kvm https://zuul.opendev.org/t/openstack/build/b5665f07306e4e6b96a053456cce19e0 : SUCCESS in 50m 25s (non-voting)\n- kolla-ansible-rocky-10-kvm https://zuul.opendev.org/t/openstack/build/21775f3044ad41438be0be5e9b068b24 : FAILURE in 43m 40s (non-voting)\n- kolla-ansible-ubuntu-noble-kvm https://zuul.opendev.org/t/openstack/build/76f63353e0ec41948ed3a6f84439c2f0 : FAILURE in 48m 14s (non-voting)\n- kolla-ansible-debian-trixie-lets-encrypt https://zuul.opendev.org/t/openstack/build/18bc923928784ec6912bac1024e8f4cc : FAILURE in 39m 09s (non-voting)\n- kolla-ansible-ubuntu-noble-lets-encrypt https://zuul.opendev.org/t/openstack/build/3b1e2fe697954c9194a37b4f0024d152 : SUCCESS in 1h 06m 23s (non-voting)\n- kolla-ansible-debian-trixie-magnum https://zuul.opendev.org/t/openstack/build/f491b16ce7f04f81a5b93452db1533b1 : FAILURE in 18m 31s (non-voting)\n- kolla-ansible-rocky-10-magnum https://zuul.opendev.org/t/openstack/build/a937fbe80a6b458fb5a47bd8b12f8b93 : FAILURE in 40m 30s (non-voting)\n- kolla-ansible-ubuntu-noble-magnum https://zuul.opendev.org/t/openstack/build/096b64c8ee96448b8894c6f999f7641e : FAILURE in 55m 13s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/92f5a6ae63004074bacc1f822330b17d : SUCCESS in 17m 54s\n- kolla-ansible-debian-trixie-mariadb-upgrade https://zuul.opendev.org/t/openstack/build/be731f9763c64059b3368647916d82c1 : SUCCESS in 26m 26s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/22494d0a69484ecab2e29d413642c9ab : SUCCESS in 18m 31s\n- kolla-ansible-rocky-10-mariadb-upgrade https://zuul.opendev.org/t/openstack/build/b9877d12a4654beb90c76132c6b04176 : SUCCESS in 26m 31s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/4214ea6452b543fba138839590fc8837 : SUCCESS in 23m 28s\n- kolla-ansible-ubuntu-noble-mariadb-upgrade https://zuul.opendev.org/t/openstack/build/570948e71728453284838a072a8a6438 : SUCCESS in 32m 14s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/860501d570b64227ab2a4247a446b0ec : FAILURE in 37m 24s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/4585df7600b443beb9afab0492f45b51 : SUCCESS in 1h 11m 15s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/7ccd84b3715e49eeacd4b451a7ab2317 : SUCCESS in 1h 05m 26s (non-voting)\n- kolla-ansible-debian-trixie-nfv https://zuul.opendev.org/t/openstack/build/fd02628548684ad7a69f5073bfed5954 : FAILURE in 1h 09m 24s (non-voting)\n- kolla-ansible-ubuntu-noble-nfv https://zuul.opendev.org/t/openstack/build/92c0a288a52440138d9e97b31c89e946 : FAILURE in 1h 23m 13s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/028f19ea18304bc49c52c04471a87c68 : SUCCESS in 1h 12m 07s (non-voting)\n- kolla-ansible-debian-trixie-ironic-upgrade https://zuul.opendev.org/t/openstack/build/2ee04bad56664b1493fe8bb2edf0b3db : SUCCESS in 1h 42m 11s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/a79b0fe8c2844ccba8b012b118c0f15e : FAILURE in 57m 03s (non-voting)\n- kolla-ansible-rocky-10-ironic-upgrade https://zuul.opendev.org/t/openstack/build/55214720b9e440c18a15109c200dd06f : SUCCESS in 1h 41m 38s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/cfcb29484fd54ec59bc105ee0e631f76 : SUCCESS in 1h 22m 31s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic-upgrade https://zuul.opendev.org/t/openstack/build/7cc83e6bbbd640a0a09984cd13df2502 : SUCCESS in 1h 47m 11s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/4a7b1e5c6f4b4e409966fddbf70faf1d : SUCCESS in 54m 59s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/eb20d5c911df4f8cab80d568b508bdb9 : SUCCESS in 59m 04s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/9cd86cb011ed4d2481ca91f3711ef565 : FAILURE in 43m 42s (non-voting)\n- kolla-ansible-debian-trixie-ovn-upgrade https://zuul.opendev.org/t/openstack/build/7426f1e0e2e44e9db30d3d965085ea8e : SUCCESS in 2h 20m 51s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/de66d7d44584459f88102941253c0902 : SUCCESS in 1h 25m 48s (non-voting)\n- kolla-ansible-rocky-10-ovn-upgrade https://zuul.opendev.org/t/openstack/build/8881eb2be5444ab2b1784d57cdbe6337 : SUCCESS in 2h 13m 23s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/34743d71237c43308bea72b097483ea1 : SUCCESS in 1h 29m 49s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn-upgrade https://zuul.opendev.org/t/openstack/build/22d2ebd8766048be84c7f026b947acb2 : FAILURE in 44m 05s (non-voting)\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/59d27325d58a46aba1689d631ab05348 : SUCCESS in 36m 17s (non-voting)\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/35e54d521c9d438f831db5fa386778ca : SUCCESS in 26m 29s (non-voting)\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/0176e2edf55548acb1ca3dae7392e07b : SUCCESS in 18m 58s (non-voting)\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/9c2991455e8a4cf49a2f8f23d3c8391f : SUCCESS in 55m 11s (non-voting)\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/f09ac4194eba4273b0dccf0543820e2e : SUCCESS in 33m 26s (non-voting)\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/ce342d350d3e42f1b428b55b3a04931e : SUCCESS in 47m 30s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8fd5144a8bd44185822cf17e794778af : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eee6454859fa431ca1d677b1b2d2f76b : SUCCESS in 4m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2ddf2233eef245d4aa1f4efe700fcb03 : SUCCESS in 2m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1365ac7691664c53b75f8c8f78b7171f : SUCCESS in 3m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/800c7302d25846d1ac7a0f487ed5eeb6 : SUCCESS in 5m 37s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e9397795adb43e8bb7e987f1ba62032 : SUCCESS in 18m 50s","accounts_in_message":[],"_revision_number":4},{"id":"3a9a618c516b9d0ebbdaf1e3cef7480e764c16ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 09:20:05.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a8e37802c60147c8b5ccd5362733c0c949f44782","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 11:07:20.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/603199963e674482bd6a3603edafc1d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5eec0b9f1acb4469bea8712c6c632928 : SUCCESS in 6m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d799b8d9da0945fb865bda73fc9a0d79 : SUCCESS in 5m 10s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/1bcee3f0fe4848fd83a05a5e35fa0678 : SUCCESS in 55m 21s (non-voting)\n- kolla-ansible-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/f0810e0417104e449154ed3692195862 : SUCCESS in 52m 51s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/a085431e981c47cfa8fc61a4ffe3feb4 : SUCCESS in 1h 28m 48s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/d01ef72da1694563b2332bf63917e0e5 : SUCCESS in 1h 00m 51s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/b15948176ac84afa83611646d9ce4ce0 : SUCCESS in 1h 19m 32s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/43bb725b9ca047feaf8dc8baf6929eb2 : SUCCESS in 53m 10s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/96b07b6cdbbe4d699bad6bcf6401e8d2 : SUCCESS in 54m 05s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/8178830d7d9b444b84ff9db47a6a9490 : SUCCESS in 1h 45m 36s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5716cf1b8b064f5081acae3791a07056 : SUCCESS in 53m 51s\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/9a89b89d586a45009e6621b7129b782e : SUCCESS in 18m 24s\n- kolla-ansible-debian-trixie-mariadb-upgrade https://zuul.opendev.org/t/openstack/build/6d2540c9882a480caccf397c784ec0cf : SUCCESS in 33m 33s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/1f99a38eb2b245c48e627ccfef42e00f : SUCCESS in 18m 28s\n- kolla-ansible-rocky-10-mariadb-upgrade https://zuul.opendev.org/t/openstack/build/de4bacadd03346d9b390f989f347ff4a : SUCCESS in 33m 43s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/54644b2bba514cc083cd887a984e2dfe : SUCCESS in 12m 42s\n- kolla-ansible-ubuntu-noble-mariadb-upgrade https://zuul.opendev.org/t/openstack/build/8b26d4a188934d1890f103f699d2c1bc : SUCCESS in 37m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78defcfa5d764c5abf355cff0a51b836 : SUCCESS in 3m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/538ec9b6c7084ad0a031406d0729a216 : SUCCESS in 3m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3436b7cb3f6643d9a4ce167153395e89 : SUCCESS in 2m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83a729852daa4e48836383d48ee80ac1 : SUCCESS in 16m 44s","accounts_in_message":[],"_revision_number":4},{"id":"f8e1b069c85d75f2d38fe9f5c2722016e128b249","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 11:07:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"bfb84e172ef0bb2b280eea6c99b6c7598e333eeb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 11:09:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/467cdf5ae2324243b0c3fbee413bd44f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab2c687cee1c465892e48a846678a3f7 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"983379e1e4bc9c0db9965315e0cdfd76aa2921bd","revisions":{"f53d9802352289c5efa56f6490d270a14ac1bb13":{"kind":"REWORK","_number":1,"created":"2026-07-25 09:28:55.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/99/998699/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/99/998699/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/99/998699/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/99/998699/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/99/998699/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/99/998699/1"}}},"commit":{"parents":[{"commit":"ee5b1d2dce8cb54d63d19c4af7f2515f59ea21e2","subject":"Merge \"Make the MariaDB WSREP sync wait defensive.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ee5b1d2dce8cb54d63d19c4af7f2515f59ea21e2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-25 09:28:41.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-25 09:28:52.000000000","tz":120},"subject":"CI: Don\u0027t fail on Neutron wrapper managed containers","message":"CI: Don\u0027t fail on Neutron wrapper managed containers\n\nChange-Id: I0ea5522b1a3a3dbb2071ff93f6800f898858f41b\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/f53d9802352289c5efa56f6490d270a14ac1bb13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/f53d9802352289c5efa56f6490d270a14ac1bb13"}]},"branch":"refs/heads/master"},"b276a6b090ab63b50341d1ec0179bde073be6c88":{"kind":"REWORK","_number":2,"created":"2026-07-25 09:33:54.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/99/998699/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/99/998699/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/99/998699/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/99/998699/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/99/998699/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/99/998699/2"}}},"commit":{"parents":[{"commit":"ee5b1d2dce8cb54d63d19c4af7f2515f59ea21e2","subject":"Merge \"Make the MariaDB WSREP sync wait defensive.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ee5b1d2dce8cb54d63d19c4af7f2515f59ea21e2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-25 09:28:41.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-25 09:33:54.000000000","tz":0},"subject":"CI: Don\u0027t fail on Neutron wrapper managed containers","message":"CI: Don\u0027t fail on Neutron wrapper managed containers\n\nChange-Id: I0ea5522b1a3a3dbb2071ff93f6800f898858f41b\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/b276a6b090ab63b50341d1ec0179bde073be6c88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/b276a6b090ab63b50341d1ec0179bde073be6c88"}]},"branch":"refs/heads/master"},"57edbcd94040d0bc73b41e4d0ddc75ef4f3fe856":{"kind":"REWORK","_number":3,"created":"2026-07-25 09:35:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/99/998699/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/99/998699/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/99/998699/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/99/998699/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/99/998699/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/99/998699/3"}}},"commit":{"parents":[{"commit":"ee5b1d2dce8cb54d63d19c4af7f2515f59ea21e2","subject":"Merge \"Make the MariaDB WSREP sync wait defensive.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ee5b1d2dce8cb54d63d19c4af7f2515f59ea21e2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-25 09:28:41.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-25 09:35:32.000000000","tz":0},"subject":"CI: Don\u0027t fail on Neutron wrapper managed containers","message":"CI: Don\u0027t fail on Neutron wrapper managed containers\n\nChange-Id: I0ea5522b1a3a3dbb2071ff93f6800f898858f41b\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/57edbcd94040d0bc73b41e4d0ddc75ef4f3fe856"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/57edbcd94040d0bc73b41e4d0ddc75ef4f3fe856"}]},"branch":"refs/heads/master"},"983379e1e4bc9c0db9965315e0cdfd76aa2921bd":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-07-27 06:42:11.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/99/998699/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/99/998699/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/99/998699/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/99/998699/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/99/998699/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/99/998699/4"}}},"commit":{"parents":[{"commit":"ee5b1d2dce8cb54d63d19c4af7f2515f59ea21e2","subject":"Merge \"Make the MariaDB WSREP sync wait defensive.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ee5b1d2dce8cb54d63d19c4af7f2515f59ea21e2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-25 09:28:41.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-27 06:42:11.000000000","tz":0},"subject":"CI: Don\u0027t fail on Neutron wrapper managed containers","message":"CI: Don\u0027t fail on Neutron wrapper managed containers\n\nFilter out the ones in state removing and also by name\n\nChange-Id: I0ea5522b1a3a3dbb2071ff93f6800f898858f41b\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/983379e1e4bc9c0db9965315e0cdfd76aa2921bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/983379e1e4bc9c0db9965315e0cdfd76aa2921bd"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
