)]}'
{"id":"openstack%2Fkolla~1001142","triplet_id":"openstack%2Fkolla~master~I61360a9062e365a3ae09853908e8aca8e04597bb","project":"openstack/kolla","branch":"master","topic":"bugfix/ironic-vm-boot","attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-09-18 15:21:36.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2026-09-18 15:21:36.000000000","reason":"Vote got outdated and was removed: Code-Review+2"},"39213":{"account":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"last_update":"2026-09-21 17:50:12.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-09-21 08:19:35.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2026-09-18 16:13:00.000000000","reason":"\u003cGERRIT_ACCOUNT_17669\u003e replied on the change","reason_account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-09-18 16:48:35.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"I61360a9062e365a3ae09853908e8aca8e04597bb","subject":"Install syslinux in debian ironic-conductor containers","status":"NEW","created":"2026-08-17 12:57:54.000000000","updated":"2026-09-21 17:50:12.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":26,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8607cdb8ddc07690bcdf08516f0172b8c42be4aa","_number":1001142,"virtual_id_number":1001142,"owner":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-18 17:21:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2026-09-18 16:13:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-09-18 16:48:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-09-18 15:21:36.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2026-09-18 16:13:00.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-09-18 15:21:36.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2026-09-18 16:13:00.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-17 14:36:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-17 14:42:51.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-08-18 08:32:24.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-08-24 14:18:29.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2026-08-29 22:18:28.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-09-07 12:24:08.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"6ab82cdaa78e6e1e683c116594d99af34e73b68a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"date":"2026-08-17 12:57:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b7bc68078de3298eadc16277943b7be4b219a75f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"date":"2026-08-17 13:15:01.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"460f63c21f5fe02de6fd5c06f3014358b246e0af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 14:36:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4ad4edfed2e4ca598b1ada113581d95\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/25202481756e4dd08946ec07bc87d9f1 : SUCCESS in 18m 11s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/a21104ac78d745759b9e1bc7491af396 : SUCCESS in 35m 28s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/655f6d9b292a4052b3fb89f9abb63c0d : SUCCESS in 27m 26s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/0f48b3a82c4e4c6485c1269b431372f6 : SUCCESS in 15m 09s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/09bc58435c174899bccf1281938c9f0e : SUCCESS in 48m 55s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b253635f48245cc8f3693379cf6d208 : SUCCESS in 17m 13s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e78386f2803647e6a1e2063ae5ebe087 : SUCCESS in 30m 06s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/ad24d259416a4c6daf39ebeb2ec46b1f : FAILURE in 32m 51s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/cd3f8414c55c4d9e840e03cecd5f0adb : FAILURE in 1h 12m 25s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/eac5a9a5b69748d08363ffbbd92965fa : FAILURE in 37m 39s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/5226b7db694c45149f1591e17e76900d : FAILURE in 1h 14m 21s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50281be7763744a19343faf6ddb2a1e4 : FAILURE in 32m 47s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/7645042f161348dc928145213c4489c3 : FAILURE in 1h 18m 28s\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/e3609137dcd044ffbbaf83e7e427b656 : FAILURE in 41m 08s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/e00c0801c35149f3b23ac76fcd81aad8 : FAILURE in 37m 23s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/0275f37ab84c42ea9f6af1b7044fb91b : FAILURE in 36m 33s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/3d3bbb9d534345f5af27d6faf2998fbf : SUCCESS in 1m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c205aa518e3643a1a6b7fe41bf164b78 : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/477b987259c04e5dac7e6508a38719ee : SUCCESS in 4m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a0ab33e853944b7e905a7a1f96401afe : SUCCESS in 2m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/be00811d377e41d7872ea0eb93746f3b : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10a63d6732a4446f90139abf04acf5c6 : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90ff36cbf40a42e99da79900703c7432 : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9311200fe3ac4eb5b61a853bec5b69a3 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":2},{"id":"d89233bb7af6ce7c227740d37a2d81fd66d54bf5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-08-17 14:42:51.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+2 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"809138ee7f48f9e718ed8b615f56aaba45aad003","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-08-17 14:43:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"58c2552134d3b7cc2e425e803d374b92bb167b75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"date":"2026-08-17 14:50:28.000000000","message":"Uploaded patch set 3.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"f6f13a7483ed396635cc595fba1094cabaf1ebd7","author":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"date":"2026-08-17 14:53:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4978a9956eb850fb35b10852da7b6694f602fa80","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-08-17 16:04:05.000000000","message":"Patch Set 3: Backport-Candidate+1 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"88fabe088d61a81af1b76787123dd0271b13ea08","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-08-17 16:06:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"359f9bc3e3abb57a9c9f657839641b50659cdf7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 16:17:18.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/78b0b18e5a5b42b0932c2a0c67d2ca86\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/8d54c35199a44e4e88c2fc4107c5e805 : SUCCESS in 14m 36s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/f3b119c21049422f82dec534462e0f32 : SUCCESS in 12m 36s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/43181db7fe3644b89a59d42ee40581c3 : SUCCESS in 32m 43s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/b2a2848000c64abb9bcc0001e1ba8556 : SUCCESS in 23m 07s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/c4ba5b5462494584b5a5390e7ccd51f6 : SUCCESS in 47m 03s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e01fbb1fa184d73b6d379093c2ba3a5 : SUCCESS in 26m 16s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/180c3a15867e4d82ad6b37ba033e79d3 : SUCCESS in 14m 17s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/b04f7da9635d4066ab9f9e77a536edef : FAILURE in 44m 21s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/6792514d7f7b4052b9e42093b4f64d86 : FAILURE in 1h 20m 34s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/d31398d6eccd4e4a904167a54fa7609f : FAILURE in 39m 07s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/5fec9b35d547439ead9ab03c1504321f : FAILURE in 1h 25m 37s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/501b3bfc0abb44e99a2c987c5c22ae96 : FAILURE in 36m 58s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/c49b2b4aabb842d28ae4ebcd958eafe2 : FAILURE in 1h 23m 27s\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/2148b8554232495191d9df75ec8b38aa : FAILURE in 43m 11s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/54e34d4b46df4285bcdc080f43941c09 : FAILURE in 44m 02s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/e2da58101ab6418a8517a926a36c2e12 : FAILURE in 43m 46s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/0960a5b42a8a4ee2b1187ae5b3b98a39 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4095b59d777f4edeac1bd7aba9199133 : FAILURE in 3m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f4de2c7671ee436dbccade5c456deef1 : SUCCESS in 2m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b65d7dc43da348a4896e9eb19e490346 : SUCCESS in 2m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/87ff691720f649e0abd236603b8d70bb : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d5f373258154882a53228c933c23fd7 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e830e8e73fae410f98056fe4bf984660 : SUCCESS in 6m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4c2f003e97e403d87c35914fcf86802 : FAILURE in 4m 18s","accounts_in_message":[],"_revision_number":3},{"id":"ac69e57a0c4a958004b9173c9912d7d90ecb1f61","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 17:48:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23f973879acf467d8f30edc6cd30c01f\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/f131393d605946d5bc9fa6f879a9cfad : SUCCESS in 34m 35s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/2ad4d775232c440e8f846049521cf141 : SUCCESS in 35m 14s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/4c3a3098aefb4f49beb6794ad134aed2 : SUCCESS in 39m 11s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/195fa68deccf42fa89229a98d25d7c53 : SUCCESS in 41m 10s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/7cd26f51fa474c7f84e59718d5f3f9e8 : FAILURE in 1h 08m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/308b51a4ec474b0b91777f9b7da346d6 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f3af0509fc7341efa33ac242a3a52425 : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b54ac393ee0f47e889b062648c675ce9 : SUCCESS in 9m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"26a3230314d67a23c01035d84523f8666b2c8193","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"date":"2026-08-18 08:23:42.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"f540f06a86249ce3c7f2e7474f04b3f6957f7f99","author":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"date":"2026-08-18 08:24:53.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e6c0a7a7f1a591f7bf8421b06bcea13b2c337ce1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-18 08:32:24.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e6a921af519e375eeb622c49593e8a23cb5aaf00","author":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"date":"2026-08-18 08:37:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"52271fcf6bc1191c5e663c84c0fb8dc107b4f7d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 09:24:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/827687470de64a3ca74fbaf0fbc8ca64\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/c21ed30eb41d4cf5b1735081e4766275 : SUCCESS in 38m 23s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/01b3c29c031a44718689807c0d226b22 : SUCCESS in 40m 04s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/1afcd2217bca4b7585c0684c273f1d89 : SUCCESS in 34m 52s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/cfc3268d13be4b1d976f4bc02eedf1cb : FAILURE in 34m 44s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/5b1f7b3c0637462e9cd482f2260d3f6b : FAILURE in 51m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/86bd7b2ed8294b5c970f5c2e5a47a27f : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/615e9d3ffeed48f697cb3b840cae97ac : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/01624d991a13489e93c14fc07b25969a : SUCCESS in 5m 50s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f983462d658284f3d2afa7f1697f3ef821d85925","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 09:40:31.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/88e8c2df863a4054969de8a13ef1080d\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/67198cd5116b408b835be0b9fa0bfef5 : SUCCESS in 15m 52s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/f75419b6213f415189b6284751aaa15b : SUCCESS in 14m 07s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/4c8968e4e4264e74989cf4a77ce4a449 : SUCCESS in 14m 40s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/01ead220d1e746209193c81bffe45fb3 : SUCCESS in 16m 50s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/93d7f4cff2b04eaf8dc4b3ae6a321218 : SUCCESS in 47m 10s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d759c06a29942fd9281a3d7651a9bcf : SUCCESS in 27m 27s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2ea2efe910764679910e2020f770185a : SUCCESS in 16m 42s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/4b178a910fee459e821dd10541c87213 : FAILURE in 31m 08s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/44d0d250a85d4329857b3b9420d4aa6e : FAILURE in 1h 13m 14s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/8620353de81b485ebf94480d9e44b32f : FAILURE in 32m 38s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/7d4f53b2f34c4e65b64e6c92b9b3cb1b : FAILURE in 1h 15m 41s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7eb18367f4a24767b5a9496c3372ddeb : FAILURE in 32m 01s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/2871ca61e4e546b49678ca6842a4d9e8 : FAILURE in 1h 13m 35s\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/d04a33d5764a45c9b5a9ed41bab8de0d : FAILURE in 38m 31s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/c280532e0b134bb79af2e873e2bdfff4 : FAILURE in 36m 08s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/f0be588ba5f8429f97830c293167e0ef : FAILURE in 36m 33s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/15c9c467cc284e399305f3ea2deaf86c : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77750f579d094087a5e3b23eb22f2626 : SUCCESS in 2m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5ee66f96f43b44afbc044c50194961bf : SUCCESS in 2m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9f2b27c125f44c14a3d5f4d93721e479 : SUCCESS in 4m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/036d3416ec494928954b5f68e13f7310 : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41e75b071e694345a56fe2d6860ec739 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d8d906d2c06485781f74fd9aedf00fe : SUCCESS in 3m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2c9fb8f8bee4824a282639118ff3ae0 : SUCCESS in 5m 56s","accounts_in_message":[],"_revision_number":4},{"id":"134b19cb6795015f034068a0441785a730395667","author":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"date":"2026-08-18 13:57:54.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"46a58b4d56766a6dfa8af4cae129926209163f67","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-08-24 14:18:29.000000000","message":"Patch Set 4: Backport-Candidate+1 Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d3e473abba313521857dff267eff6d5ccd93cd23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"date":"2026-08-26 15:37:47.000000000","message":"Uploaded patch set 5.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1, Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"45b14bfa84b6e53a91de0989110f45dc4620bb39","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-08-26 15:43:05.000000000","message":"Patch Set 5: Backport-Candidate+1 Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":5},{"id":"3294d01bf18d843e0435af7a8b73d0696361846f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 17:15:27.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/94d791d7bdbf4f1bacadfe0583f264e3\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/f9435e2d3fc446e1a79a33b3fa06708f : SUCCESS in 28m 41s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/39d94ce3804b4f1390b3fd44b0938337 : SUCCESS in 26m 39s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/bf758f67dd9e443b80ce0e0236c4e2cf : SUCCESS in 23m 37s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/798f28c5315f449ab34715fa49e21c91 : SUCCESS in 34m 49s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/cc54b1000c09471bbc0fd91065aa7356 : SUCCESS in 1h 05m 06s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/12f5e8ada45a46349ef7a2c10d07393e : SUCCESS in 27m 59s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2190c765899c498eb9ae0526dba99e17 : SUCCESS in 46m 29s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/fb9150f81d0c408290594fd0a50f47f0 : SUCCESS in 1h 03m 48s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/f053df59d23d4fe4bed3f81900d71e78 : SUCCESS in 1h 30m 06s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/026a3e395ef54911aab36f7d80c91e3f : SUCCESS in 1h 05m 26s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/9bb52a58ab7941b7884a154b44bf7273 : SUCCESS in 1h 30m 42s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c01aca9e52a413a9c163b0abb0c000f : SUCCESS in 1h 07m 34s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/a2f8d207d5ee4a5caedc5d6c9912b4d4 : SUCCESS in 1h 34m 53s\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/9ea6996f1d234eff9547fb86efd9f720 : SUCCESS in 1h 22m 16s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/e7055d0084594d47a2af7632e5b888a0 : SUCCESS in 1h 21m 48s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/da8b933eed2945f696b73b6cdd320ea9 : SUCCESS in 1h 28m 19s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/5e308c40c6764ec188c7401ad4266ef4 : SUCCESS in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4de86d36f7254d88bf12db9d7f2956b4 : FAILURE in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9d4583673de54138a34ddbe571ebad95 : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b0f0223d811041079738f6ff35d487f7 : SUCCESS in 3m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/eced6e7a3fb643c8a06b1846d121a36f : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1efb57c50ded449ea2cd97734142c2c6 : SUCCESS in 2m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e83e822a08264b6f8309929ffeb890df : SUCCESS in 6m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6017f3b848b5472e97a347ff142bce77 : FAILURE in 2m 06s","accounts_in_message":[],"_revision_number":5},{"id":"5f708612281fadc89cc1b8c62fc8276ea54f868f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 18:12:26.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4e4ee33b481448c87b2fd9190b9571e\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/a71bb5864aa3494cb95f3a35cbd56d47 : SUCCESS in 52m 34s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/4af1156f207e4dedb499a0cb97b61a4b : SUCCESS in 52m 29s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/21735d9333dd4d0ba6642eef39166e72 : SUCCESS in 57m 47s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/79efc180742f47b28fe11506b6995ee4 : SUCCESS in 55m 41s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/8e23e9cbee9c4ca0990c4ec3814a5773 : FAILURE in 1h 14m 42s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/39ab020bf8274dc3985a897e8ad30671 : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/95fa18becc114e4b8dce97cfbe4d33a8 : SUCCESS in 3m 41s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9a220508fcc549b399502199ffda9b6e : SUCCESS in 10m 16s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6ca086e5ebe6253294d27644735d6fb19aad1bb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"date":"2026-08-27 08:35:11.000000000","message":"Uploaded patch set 6.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1, Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"847d57375c54f89611809847908dc6a01787143f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 11:01:39.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4d7aa88d40e479d8ddc7183e4ded2b5\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/17a129eddc4746f18742953c03669d4e : SUCCESS in 34m 42s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/32d8e6df6dd44456befb243c842d2ee6 : SUCCESS in 45m 17s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/deba31c667a3403c9bdacf5601dc2a07 : SUCCESS in 43m 08s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/7eedb74d440843a0b6dc11343a3650bc : NODE_FAILURE Node(set) request ac3b1de26b864696948fe32c768a5f65 failed in 0s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/6f744f7a391842b2b9641356493e1b0e : FAILURE in 50m 43s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b88d75a61af142f2bf27055ac026fe11 : SUCCESS in 3m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/480d415ac48f4f6ea5c3f92fc61c999c : SUCCESS in 3m 55s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f3170893fa2f4ec99af215fd24662eca : SUCCESS in 8m 26s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2d900c5c87c851421796086a8c38f2f17be47768","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 11:39:33.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a1f76e6606b2472487f7e993025a65ee\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/5e349e9efb5d4c939609968022a2bcdf : SUCCESS in 18m 52s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/5a3685768ff149c88f2eb258b1ac11be : SUCCESS in 29m 20s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/695eef25e06e4615b2fc73c5e54311fc : SUCCESS in 25m 49s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/b2984cd85e0c46788ef81986e993bb02 : SUCCESS in 32m 03s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/a64c5cd3862d47d7b33e627cd6aba52a : SUCCESS in 44m 05s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/12581503af664df687ff678eb6551083 : SUCCESS in 17m 38s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/02c69bacda1f4997b204223c3c6d155d : SUCCESS in 35m 55s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/3eb143769caf4c3db8488873eeb7daa0 : SUCCESS in 1h 10m 46s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/d7b4b67eb129499d90e533241c2535ad : SUCCESS in 1h 44m 08s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/adeaf484885f469b9d733b4610669b2c : SUCCESS in 1h 04m 35s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/064dd5887bb04e5ab92ec99488881d41 : SUCCESS in 1h 26m 04s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/efca853552484c7297871ec65f0d3cd9 : SUCCESS in 1h 12m 04s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/06bacc1172a64871af350e7f53f2bdf3 : TIMED_OUT in 3h 01m 40s\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/ebd82f3be44f44a2ad244772d6c080b7 : SUCCESS in 1h 28m 50s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/cab8366d1d7446a08d9340262dbb9858 : SUCCESS in 1h 15m 19s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/773fd937134b44878be544a0e5c7e0e6 : SUCCESS in 1h 27m 25s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/6ed9c6ac7ed0452f98e101a240afe049 : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5868a38573641f6a19e0d17ba4d01cd : SUCCESS in 4m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0c2262214dd04b31a155a57bfa807302 : SUCCESS in 3m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3c7ba42093db4026a85bf5ad543b2719 : SUCCESS in 4m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/23bcd888281f4fb091fc00ba1fa2e6de : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7521adda46a84d04a4fd5d445f01b7a4 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd794012b8684ff3ab7edab76f56439c : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae7aa720d3754dae8e47411b50f5c93f : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":6},{"id":"b736cd30ab06eddc81099c9954c5cd77e349c90c","author":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"date":"2026-08-27 12:14:32.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1cc06afcf6b8b4cf1331063548c4a6c88984e4b7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 14:44:46.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c909f2925454f36be412b6d46b79a7b\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/d6ababa824d143b7ada0eac3232e3154 : SUCCESS in 36m 03s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/2bbc0837968e4435a027a06cdc3f7305 : NODE_FAILURE Node(set) request ff25948d34454fc19746440ef71c3b87 failed in 0s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/e6e876f2c21c4ba7b91f030515a136f2 : NODE_FAILURE Node(set) request 1d6107041f60473eb93c17db5190af0a failed in 0s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/45c0823bc5a744978ea678934faf63a2 : SUCCESS in 32m 17s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/86f28c25159346f682e975725a69332f : FAILURE in 1h 06m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/03b0b1b1ac0d4b94b09483d86ab8874f : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4afbe7cd88114264a130086eac6d5b3b : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/65eb563067a342a0ab1b23a7606858a3 : SUCCESS in 5m 27s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c5d4673007adfba42c3d7003355fceb57f7692e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 15:19:41.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab107c2facc5429d87acf984a499972d\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/8ca74b17b3114bed847b241b1ed785b7 : SUCCESS in 20m 36s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/de22afffe1b5415d94d0e0ea831ebc92 : SUCCESS in 18m 32s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/cfed5b83daf4433984e75503c79cebfe : SUCCESS in 34m 06s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/9b8b073c79f24ff3a41fd93ddfa4037e : SUCCESS in 22m 46s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/5f3de7dba9c74e4eaafb13e82e74f96f : SUCCESS in 1h 16m 06s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8555fd48f8ae4b408dcdf1f9a9633972 : SUCCESS in 16m 39s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8228172be11e426abc9f8623948ca3d5 : SUCCESS in 35m 10s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/3f47086e2fc941fd8f0037d05031f95a : SUCCESS in 1h 17m 33s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/ff9f2f57acb54cd39ad40453bfa0d803 : SUCCESS in 1h 45m 51s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/f6e084442a4f4edf82c3344c08ef0068 : SUCCESS in 1h 01m 33s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/fc04e7d057b74e08bda346e19cc4d377 : SUCCESS in 1h 27m 13s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/645415b824cb4b6aab08f0c49f52d095 : SUCCESS in 1h 15m 57s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/669c0652dad4420e96b508cf93eaec04 : TIMED_OUT in 3h 02m 08s\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/4b624237a5784560a904976967c869a8 : SUCCESS in 1h 33m 29s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/287a772e8407484691a6d26d5eb9fb6d : SUCCESS in 1h 17m 36s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/fc9986111f724955b92c84271a99ef77 : SUCCESS in 1h 33m 50s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/13f914b1002b43158ab66c367a230a15 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58e8e6d20b984b468d9a1cfb0a466d7e : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/50449edd22b84df39b67b9abec9b92d0 : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c654467db10c444e9a6404b7ddfc2eb5 : SUCCESS in 3m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e7438b2495ec4a6dbed1fe310b27aa92 : SUCCESS in 3m 26s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d283238e5205479ea28b8ba8e4d66a0c : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49767187324b43df8e39f78cab57c418 : SUCCESS in 7m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ba9d44ae08a4e9f85e84244f82f809c : SUCCESS in 7m 16s","accounts_in_message":[],"_revision_number":6},{"id":"bac1263270e0198175fb2ce4c7fe853a04a9012e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-08-29 08:22:00.000000000","message":"Patch Set 6: Backport-Candidate+1 Code-Review+2 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7735ce28ba7dee7e1013f941d0a0ead3ca3102d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-29 10:01:43.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/888bb04d57ac491d9c757156bd96935f\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/376ca55af7cd49b4a4937c1f4d0798ae : SUCCESS in 43m 45s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/c2d880393ab843658686cae2a0d7eed8 : SUCCESS in 38m 10s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/b03ad48b187447b7b166691f9eab03f9 : SUCCESS in 42m 24s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/5453d34291304aa2b866edd7e1e0e52a : SUCCESS in 48m 13s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/081d4b1af51b4b87ab543ed644aee547 : FAILURE in 1h 15m 22s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3d53952f9d4a4b009e81981cfb3ccb44 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7ab7a1695f6b4301b7ec6a8467a1f37c : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b843b4dd3bfc41d988502717a2966864 : SUCCESS in 7m 05s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"fddd216956aeac5069d4e6ed3bbc61ac0252514a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-29 10:15:28.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1bc9a8852c845acb070dd0a0724f339\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/5dabd61b740e43a182e4849736b1daa3 : SUCCESS in 22m 22s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/9072fbc87cfb4897abf97d784d94082f : SUCCESS in 21m 49s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/fbc1f34df4a44c2c9884a71733d2f51c : SUCCESS in 33m 25s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/4c2e5034538f4224856e571596878269 : SUCCESS in 32m 32s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/30353ac7882c48bf881962b8d221d2dc : SUCCESS in 1h 02m 02s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c7e6e0f61dd41f4b64b9c2639bcaafe : SUCCESS in 19m 32s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/50cf8d8665a24a60b10a928901aae66e : SUCCESS in 34m 52s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/b99ef63bcaea4ab498acef397e44a250 : SUCCESS in 1h 07m 17s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/99e6f076e7684637ab9dbb47ef4cb3e9 : SUCCESS in 1h 43m 14s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/a921c832ba55452aba5fd9de143ed0b5 : SUCCESS in 1h 05m 50s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/9efe277fac714a42b7f4341caa26264d : SUCCESS in 1h 32m 38s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed81921022d646b6a1b45322f2b2e2aa : SUCCESS in 1h 13m 47s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/7e9c6c7f1e054cf5b59c3d746ef43348 : SUCCESS in 1h 51m 02s\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/a203500121944992b9603ad2ae89a6dc : SUCCESS in 1h 28m 13s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/9e861afb006f4cd387f8abe2e26cc109 : SUCCESS in 1h 25m 55s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/d38687de3393458f8a3056846108c1bb : SUCCESS in 1h 27m 42s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/6ffde9d5ccb74defb10ef3181d25b581 : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5ccbc7912d149bf9765f0fea09ec19a : SUCCESS in 3m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2e3ef7e5fdf54913bf0d60684a384f9b : SUCCESS in 3m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/95a99ab8f9634dd48bf775d103eb10da : SUCCESS in 3m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ac152cfa236e4a9382434df9d96f249e : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9d9c5fc23f14dd7a0f0b7ccad04f72e : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/199e13fe6b10487482b462c01038f65a : SUCCESS in 4m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa9f2ff03c7a4cf4bc09331bf651c3b3 : SUCCESS in 10m 46s","accounts_in_message":[],"_revision_number":6},{"id":"cc0fc9046840f8fb09de1c3c1f1967034d2edec4","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-08-29 22:18:28.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"e597eafa5ab292ff5b6e05f297094f9e6bfec769","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-09-07 12:24:08.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"f410a9654adb0be289fab51a32a4fefb814b06a2","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-09-07 12:25:44.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4a327e6c5dbd82b52543eb00cee675a4b02b173f","author":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"date":"2026-09-11 14:15:59.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ec4f78b12754bd5549c2d9a8294c8e8ef013941d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"date":"2026-09-11 14:16:05.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1, Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"227a42d8b53ad3260e87a6a2554b3455e738e71d","author":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"date":"2026-09-11 14:16:27.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"4891ba9c17b45a86df899572272b5f2f75a3898d","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-09-11 14:17:39.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"00fcf2c779f6b35f8c59157db49197504321009d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 15:33:22.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fbba03d033d4c5bae589af81a2ca366\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/10371c1972c744d3afea81b3462ee8f8 : SUCCESS in 36m 44s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/89e0c69b70fb49bb88d2f3a651902397 : SUCCESS in 43m 27s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/d2186eebeddc451e9a84a2c68b9587d7 : SUCCESS in 41m 10s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/c1c21661812b4993aa61fe87ca1802ca : FAILURE in 1h 11m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/365e941fa7144c7297ec93ed5dfd2f5b : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7b38181308334dbf95fb0bb9e96481db : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ebedf66b26e74a02941868bd70fdce60 : SUCCESS in 7m 54s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"55c23615696849394f0e163af1525f1ec03ee1a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 16:27:21.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d33675a13cff4f03b3873e7adb03daa2\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/7545cf65f83e468c94fea1c1ab077c2a : SUCCESS in 32m 01s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/67619c81970d4534835f630e28d4109b : SUCCESS in 36m 15s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/b6a7bb442e3c4d94bf3e3e0945373f46 : SUCCESS in 21m 59s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/53131e66fc0542acad8c7cf5c1428d67 : SUCCESS in 46m 51s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c877fdee49644d0ca214eb3c0d084346 : SUCCESS in 35m 48s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6d512add6ea64083ae87eeec3027c2e1 : SUCCESS in 27m 56s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/9a955db82f5748ac9654ed91d566e617 : SUCCESS in 1h 12m 31s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/d1b979a36bf34211a043a2146de9f17c : SUCCESS in 1h 46m 48s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/2be5495993c74aa48dba94cf1b5aba6f : SUCCESS in 1h 05m 28s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/d8d138d501ac4f3f89f3f9bbc79cb06e : SUCCESS in 1h 29m 23s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eed856e7147f4aa3803350f5b876b7b4 : SUCCESS in 1h 30m 55s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/a1fd879ded8245efb67ccc5208b5caff : SUCCESS in 2h 05m 12s\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/f4ece460c21d4e6a9fba4a7cecbaa7c6 : SUCCESS in 1h 43m 02s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/01d0f29c6e874779b2805255840b1a6c : SUCCESS in 1h 18m 19s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/45de331d8b264f6580fcfe4a57d42541 : SUCCESS in 1h 44m 07s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/ff510188737042f88dee8627c94d1d98 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dae5d07b5fd497f9b8de7c57c436b4a : SUCCESS in 6m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d3b231aea39842718806016aa9c1c3d4 : SUCCESS in 2m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/20b733d8d92b47c0b7d8c17453010193 : SUCCESS in 2m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c4e1a18159a74e5f92493d3aca1fd867 : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1390adaa8f7c415aa8d0500f25bf4ce7 : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22efd8987e824d35bd67693ad5a1bd26 : SUCCESS in 7m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0db498c133f84fdf8f3d3d50f8dc2d91 : SUCCESS in 10m 08s","accounts_in_message":[],"_revision_number":7},{"id":"80c0df9fdf1e606cb3f145ec4959342cd155d8bb","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-11 21:22:50.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"16ef8428b4f5f856fff86db4ca53017bdc788dd3","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-09-16 12:27:39.000000000","message":"Patch Set 7: Backport-Candidate+1 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"06f9584084d614fb9b7c7c712d49f5812bcc4412","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-09-17 12:04:58.000000000","message":"Patch Set 7: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"55356c597db78d540c47dcb2e45af72e1541a92d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"date":"2026-09-18 15:21:36.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1, Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"c830ba9886100422282551710770f48b46db5623","author":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"date":"2026-09-18 15:22:50.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"bf62e0bb96085a5bd16ccf9b240ac3132061caad","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-09-18 16:13:00.000000000","message":"Patch Set 8: Backport-Candidate+1 Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":8},{"id":"ffb501e1608e8e4754d9f5c91edaca16fd087511","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-18 16:48:35.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"ed0ce2bdb1369db159c9e0c6bf160f271df686bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 17:13:39.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14bcff5725634ead88e317db946e6420\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/9b1f34238dae4c29907afcd48ecd4108 : SUCCESS in 44m 47s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/f9c89004105d408c928a47ed01663fb6 : SUCCESS in 52m 00s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/bfe472f6b17a4725a317d81587c82519 : SUCCESS in 44m 17s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/ae4b13445b2246a086b4c545b935d183 : FAILURE in 1h 18m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/68fbe18db6ff4f6a97f43490b62d7417 : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b643e954cd36466eb85268dcd5d98a5b : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3b68b783cc2e41278db5a8263ca6d0b3 : SUCCESS in 10m 10s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"589ef52b0b5ce639154dc60de8a6ed92c2012e95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 17:21:32.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/25aa0a3ec2ab49c5a80b0a21d5bb55ba\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/3085cca98db14cf4b5e5e63c94018522 : SUCCESS in 33m 04s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/3818f017d86a48b0a0329fe6bdc12605 : SUCCESS in 35m 11s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/0a76f65a3e6d433f867804d280d7fa00 : SUCCESS in 33m 36s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/26061337b2334c1295f6e103e9303ffe : SUCCESS in 1h 13m 56s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e73c5b78440f4ca689c698ae986dd0f0 : SUCCESS in 29m 10s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5a58ace209e44f77b04f32dbe4b97af4 : SUCCESS in 31m 39s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/24f415e85be54529b0031261011c8848 : SUCCESS in 1h 32m 20s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/a09aaee8ab344b73877a6c24d60b867c : SUCCESS in 1h 55m 52s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/606e058f21504d2eb4987ac2c85ab0d1 : SUCCESS in 1h 02m 58s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/9f6bc12bc1a84c0ebb685f34d9369275 : SUCCESS in 1h 26m 39s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d5f77369b6604eb692583e9dd245e294 : SUCCESS in 1h 14m 49s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/7c8241173aa34e5b95978f58590d80c6 : FAILURE in 1h 23m 39s\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/506ba0a06ffa4ac193150742844aca04 : SUCCESS in 1h 37m 06s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/901aa33f8dac477abfee236855444d3a : SUCCESS in 1h 18m 31s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/4b652b0081274eaaa67d18e37d7a8425 : SUCCESS in 1h 43m 11s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/bd6c1cc06b474ad8a6c0c2fbf75ddb50 : SUCCESS in 2m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/287d47e00c684ff0b7da42fd638fa001 : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/618236f88ee54d548acf29b45e875c4e : SUCCESS in 4m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ba740dee8cdd4079bff7324b840138bc : SUCCESS in 3m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8cb2f2c8f3ea42678f7345831c64e99d : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9876b74224b24948b443eeebe606070e : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9f8ebeede024a75ac20468f063ff155 : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa843bb987a44c83b6489a278810cd8a : SUCCESS in 10m 12s","accounts_in_message":[],"_revision_number":8},{"id":"a1d2e94b5a34e4cabddc1d1e646fbcba45bf0b8e","author":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"date":"2026-09-21 08:05:26.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"dd6ede6b0b6af1c1691cf1c97a6ebaa9de047da5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 09:44:58.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f894f7adc12a44a78d76a0a8cc1c9953\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/cfc5d984b4544a1eb8c475fd0d17254b : SUCCESS in 40m 42s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/ce744b972af64404aadf688456dc84e3 : SUCCESS in 46m 24s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/d510852d98154bc0a49e1ec86fd3a891 : SUCCESS in 33m 33s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/57618020f0ed438dad0ac04859c8eeaa : FAILURE in 1h 15m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5bdd7d67ce664292b851fd574d19271b : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/61ac64942d91499880aaf03285190303 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0cf3f164a9d643c2be659eac8d2b549e : SUCCESS in 5m 52s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"01864197f6a54ffc47ec8dbe4690ae162b7a074f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 10:18:40.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f7cba06ce4064636877a0846a3878b87\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/5c71cf411cfa4c2da0d95a37cb6655d4 : SUCCESS in 39m 57s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/b24ea04ff8e3467ba1213c58822c1365 : SUCCESS in 42m 12s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/80fa731c99df44aeb7647f7a0ac871d1 : SUCCESS in 30m 31s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/71458b6de6e64ee086e2681116117284 : FAILURE in 57m 09s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aeb0951770954703b2029508d50c7fb1 : SUCCESS in 29m 41s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d63b43914aa74df2b3850280a85f1f0a : SUCCESS in 32m 27s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/0e70ff81fc2045cb8cc5e0359594a4c1 : FAILURE in 8m 01s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/134f74fa63c14495ab6891ce7dad4048 : SUCCESS in 2h 09m 34s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/04945cd7fcb440eb9be217753044d875 : SUCCESS in 1h 10m 22s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/631ff54a2a8f422781594eb995b8e8d1 : FAILURE in 22m 41s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2067a53ea8304af5a6a03cd39ae755d0 : SUCCESS in 1h 11m 19s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/95f4f929f22d427a971026ca9acbe8e0 : SUCCESS in 1h 58m 50s\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/478e15ae5b7e47e9bc663efc2ab96f75 : SUCCESS in 1h 35m 45s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/5e13b65b5ebe4d889e8e2e3ec47aeb1a : SUCCESS in 1h 26m 42s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/224b38c5e349445eb714d17cc2ea76de : SUCCESS in 1h 34m 16s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/83ca049fcd2d4b559ad3211f80ce089c : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/141cfdc0b6414aaf8a3f39f5b6f0cab2 : SUCCESS in 6m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1843aec5f78d453aa0ba76906e1430b2 : SUCCESS in 3m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ad5d64e0dab64f019605a725fd282d82 : SUCCESS in 3m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/03a91e1dffa24933b52093dca4de7eed : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba3d30e8fb594947ab603b1e5973dc1a : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f02ccdf4ad734bb0b791309925c16d14 : SUCCESS in 8m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/094b306114fe4ca488d757b05ea87186 : SUCCESS in 11m 40s","accounts_in_message":[],"_revision_number":8},{"id":"52a8b32d4fce17c11f0678c62f7159d91d8f68df","author":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"date":"2026-09-21 10:36:59.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"6d8a78e42b7a82e77480e467400076b0a1e8d9c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 12:50:53.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fc40033f76b42feb6bf2fbb477497b7\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/aede751acab4439690cb72d46495a11e : SUCCESS in 47m 38s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/a5915653d0b340748cddcb11fe66b552 : SUCCESS in 51m 38s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/581b8470881f4b158b832f37d895ee65 : SUCCESS in 43m 18s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/536a3a95a0ed41c8beb7260c74f47271 : FAILURE in 1h 24m 10s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/beea63cfecfe4ee1ad00ce7971bd6dff : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e7a72ccba82c4a0fa02c17f1ded56c6c : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6a7294e119ea4cc489b080c628961c86 : SUCCESS in 9m 13s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"47d462c8aba2a787b58c1ee48db944c74c1562b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 13:15:52.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d9d57ba5768406e9ee9edd7dd529581\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/2d086da7b0b147efb17c980cc46b3229 : SUCCESS in 28m 26s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/9f26f91575674e0bb4ca711a1112b29b : SUCCESS in 40m 28s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/5535ae11f3cf449287025bda6665aadd : SUCCESS in 28m 13s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/ece0b3e8e5b740b59b7a17bf242e7c49 : SUCCESS in 1h 06m 04s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/56629dea3572450aa44a9075a417e45b : SUCCESS in 27m 01s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/635109d51e2341dfbeb58f1816ff6117 : SUCCESS in 32m 13s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/81cbccad97ea4c9da5664e86e0603223 : SUCCESS in 1h 18m 51s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/95c3b7180aef44bb9eaa97f5390095b4 : SUCCESS in 2h 18m 57s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/7d0b6893d0e447b696304191ed6a99dd : SUCCESS in 1h 09m 47s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/c336e9ec63fa44109a8cefef03f528c0 : SUCCESS in 1h 35m 25s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a59ef21a189e4411a7a692d8974cb028 : FAILURE in 16m 19s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/6168c4eb4d874225910971c3471bf90b : SUCCESS in 2h 03m 24s\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/14a8f8b2b1a24f45b6951bcd96d43987 : FAILURE in 10m 49s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/e00d8c3fd72441b2a64d1706620cdecc : SUCCESS in 1h 23m 18s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/c711f5168acc4573a076eba23505c6b6 : FAILURE in 7m 48s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d7982f61bc744884a8643f0100cb3a74 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/273c1591dfd54baf8f5ccdabd7e48693 : SUCCESS in 2m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1b50da7cca5447188b8583a89637284f : SUCCESS in 4m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/38795bd3afd1424caef552155e2b89e1 : SUCCESS in 3m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/44ba288911954297b46ab4c2748d9459 : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c64f71921d84ee6b27dea9e5d8aa431 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dc1ac4f22324976a4cdda52869e8654 : SUCCESS in 7m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/647808940b7d4793b8bd2c23e8d308cb : SUCCESS in 9m 57s","accounts_in_message":[],"_revision_number":8},{"id":"b7695c40ab24eaca2fbf9c37847b87257db18177","author":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"date":"2026-09-21 16:09:37.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"ef7f661794c89d64026e24a9c641485c1b336819","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 17:36:07.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20e821dc92524f15ae5d1f009701517f\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/f8a9c736dd4044d1b841dcec042c863f : SUCCESS in 42m 43s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/d9918123514543e5a1fd693901af02aa : SUCCESS in 56m 49s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/4c5a830c5de040b492ad36a493f26a0c : SUCCESS in 42m 58s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/b7d09b3ea781407da3bf3f2b288ca5cf : FAILURE in 1h 23m 52s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7ec5ac59fefd4be9898d3bb30f728f72 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2e5e65bb1fd645cb9f97698292d2f0aa : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3de2007973c145f0baf3bb9762a9987f : SUCCESS in 9m 09s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"8607cdb8ddc07690bcdf08516f0172b8c42be4aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 17:50:12.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6356caa2c597473eb1305aba0b6b104d\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/2a4f9d6e35ad4e5f97f7e84a2e62a3c0 : SUCCESS in 20m 36s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/2523bf16fb734acf896405928d77925b : SUCCESS in 33m 50s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/aa8475e8fb0141bdaa56fb1fa02ee623 : SUCCESS in 28m 13s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/e8a1ef0928cb46268adad1d877a74701 : SUCCESS in 44m 09s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/78ec1e10ca5f43cab0fa2aba8d26272c : SUCCESS in 30m 40s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ae4408a836c34925b892b7f3720846c1 : SUCCESS in 34m 19s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/73c2fee80f0c42b68e582f9ed9455dad : FAILURE in 9m 13s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/fc0d659a243545d9afe7a07ca84414b1 : FAILURE in 9m 05s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/b8d33d7b11f8425395572b7180466e5c : SUCCESS in 1h 03m 25s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/52a1ea3a2c874f788c27398bda46839c : SUCCESS in 1h 27m 13s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e7ba2f8f048f485da5bf2e4f6ce648ef : SUCCESS in 1h 22m 17s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/c5d37049fd8f47d0bbf29f5cd0f8b866 : FAILURE in 1h 11m 12s\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/0f9d64fd681d4ebc8ba54450b51a8436 : SUCCESS in 1h 37m 25s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/f507bb94b0bd40caaaa32c85afb7fc40 : SUCCESS in 1h 16m 59s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/c717f8738c4c4b4e904d23c1e890d79c : FAILURE in 8m 17s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/74395c7c38c44ec2bda4f4f64ee1f97f : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/006a0ab3798449708dc878808bd1f4c0 : SUCCESS in 4m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b2239c85e01e4b539e04519d654f8e81 : SUCCESS in 3m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ad7d223a65944b4bbaec382aafedc769 : SUCCESS in 3m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f8c0cbea91884660a5bd1632b3a8db80 : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c417c346a6484a5987c756bbf2c2fae4 : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04c85288fa4b41a7848922d81d4e51ee : SUCCESS in 6m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fcaa45bb0d70443293240cb4e1372dca : SUCCESS in 9m 23s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"9a972298bb490d4ead8f2d5ed68f500a2f55f4b9","revisions":{"19fd8a597f56837973dc46a381970a756b3db46c":{"kind":"REWORK","_number":1,"created":"2026-08-17 12:57:54.000000000","uploader":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"ref":"refs/changes/42/1001142/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/1001142/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/1001142/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/1001142/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/1001142/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/1001142/1"}}},"commit":{"parents":[{"commit":"e5b79040d23b80773ca79576ca97040408445c89","subject":"Merge \"Drop leftover of monasca support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5b79040d23b80773ca79576ca97040408445c89"}]}],"author":{"name":"Jed Preist","email":"jed@stackhpc.com","date":"2026-08-17 12:57:29.000000000","tz":60},"committer":{"name":"Jed Preist","email":"jed@stackhpc.com","date":"2026-08-17 12:57:29.000000000","tz":60},"subject":"Install syslinux in debian ironic-conductor containers","message":"Install syslinux in debian ironic-conductor containers\n\nInstall syslinux in debian ironic-conductor containers, to provide\nfiles required for creating disk images that can be booted with\nthe redfish-virtual-media boot interface\n\nCloses-Bug: #2163659\nChange-Id: I61360a9062e365a3ae09853908e8aca8e04597bb\nSigned-off-by: Jed Preist \u003cjed@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19fd8a597f56837973dc46a381970a756b3db46c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19fd8a597f56837973dc46a381970a756b3db46c"}]},"branch":"refs/heads/master"},"2b87a2c7f472a206aa500000fcc3663c96a30ed9":{"kind":"NO_CHANGE","_number":2,"created":"2026-08-17 13:15:01.000000000","uploader":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"ref":"refs/changes/42/1001142/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/1001142/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/1001142/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/1001142/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/1001142/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/1001142/2"}}},"commit":{"parents":[{"commit":"e5b79040d23b80773ca79576ca97040408445c89","subject":"Merge \"Drop leftover of monasca support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5b79040d23b80773ca79576ca97040408445c89"}]}],"author":{"name":"Jed Preist","email":"jed@stackhpc.com","date":"2026-08-17 12:57:29.000000000","tz":60},"committer":{"name":"Jed Preist","email":"jed@stackhpc.com","date":"2026-08-17 13:14:51.000000000","tz":60},"subject":"Install syslinux in debian ironic-conductor containers","message":"Install syslinux in debian ironic-conductor containers\n\nInstall syslinux in debian ironic-conductor containers, to provide\nfiles required for creating disk images that can be booted with\nthe redfish-virtual-media boot interface\n\nCloses-Bug: #2163659\nChange-Id: I61360a9062e365a3ae09853908e8aca8e04597bb\nSigned-off-by: Jed Preist \u003cjed@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2b87a2c7f472a206aa500000fcc3663c96a30ed9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2b87a2c7f472a206aa500000fcc3663c96a30ed9"}]},"branch":"refs/heads/master"},"e81e71e7944ce798db910ff47fa8fdd0aeaf2e2b":{"kind":"REWORK","_number":3,"created":"2026-08-17 14:50:28.000000000","uploader":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"ref":"refs/changes/42/1001142/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/1001142/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/1001142/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/1001142/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/1001142/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/1001142/3"}}},"commit":{"parents":[{"commit":"e5b79040d23b80773ca79576ca97040408445c89","subject":"Merge \"Drop leftover of monasca support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5b79040d23b80773ca79576ca97040408445c89"}]}],"author":{"name":"Jed Preist","email":"jed@stackhpc.com","date":"2026-08-17 12:57:29.000000000","tz":60},"committer":{"name":"Jed Preist","email":"jed@stackhpc.com","date":"2026-08-17 14:50:16.000000000","tz":60},"subject":"Install syslinux in debian ironic-conductor containers","message":"Install syslinux in debian ironic-conductor containers\n\nInstall syslinux in debian ironic-conductor containers, to provide\nfiles required for creating disk images that can be booted with\nthe redfish-virtual-media boot interface\n\nCloses-Bug: #2163659\nChange-Id: I61360a9062e365a3ae09853908e8aca8e04597bb\nSigned-off-by: Jed Preist \u003cjed@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e81e71e7944ce798db910ff47fa8fdd0aeaf2e2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e81e71e7944ce798db910ff47fa8fdd0aeaf2e2b"}]},"branch":"refs/heads/master"},"76f06369e589bc9f28673a2051f5414ab676d038":{"kind":"REWORK","_number":4,"created":"2026-08-18 08:23:42.000000000","uploader":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"ref":"refs/changes/42/1001142/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/1001142/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/1001142/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/1001142/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/1001142/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/1001142/4"}}},"commit":{"parents":[{"commit":"e5b79040d23b80773ca79576ca97040408445c89","subject":"Merge \"Drop leftover of monasca support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5b79040d23b80773ca79576ca97040408445c89"}]}],"author":{"name":"Jed Preist","email":"jed@stackhpc.com","date":"2026-08-17 12:57:29.000000000","tz":60},"committer":{"name":"Jed Preist","email":"jed@stackhpc.com","date":"2026-08-18 08:23:42.000000000","tz":0},"subject":"Install syslinux in debian ironic-conductor containers","message":"Install syslinux in debian ironic-conductor containers\n\nInstall syslinux in debian ironic-conductor containers, to provide\nfiles required for creating disk images that can be booted with\nthe redfish-virtual-media boot interface\n\nCloses-Bug: #2163659\nChange-Id: I61360a9062e365a3ae09853908e8aca8e04597bb\nSigned-off-by: Jed Preist \u003cjed@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/76f06369e589bc9f28673a2051f5414ab676d038"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/76f06369e589bc9f28673a2051f5414ab676d038"}]},"branch":"refs/heads/master"},"3c0f7c388af7d1f453e80132620cd47285fe5320":{"kind":"REWORK","_number":5,"created":"2026-08-26 15:37:47.000000000","uploader":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"ref":"refs/changes/42/1001142/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/1001142/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/1001142/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/1001142/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/1001142/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/1001142/5"}}},"commit":{"parents":[{"commit":"e5b79040d23b80773ca79576ca97040408445c89","subject":"Merge \"Drop leftover of monasca support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5b79040d23b80773ca79576ca97040408445c89"}]}],"author":{"name":"Jed Preist","email":"jed@stackhpc.com","date":"2026-08-17 12:57:29.000000000","tz":60},"committer":{"name":"Jed Preist","email":"jed@stackhpc.com","date":"2026-08-26 15:37:36.000000000","tz":60},"subject":"Install syslinux in debian ironic-conductor containers","message":"Install syslinux in debian ironic-conductor containers\n\nInstall syslinux in debian ironic-conductor containers, to provide\nfiles required for creating disk images that can be booted with\nthe redfish-virtual-media boot interface\n\nCloses-Bug: #2163659\nChange-Id: I61360a9062e365a3ae09853908e8aca8e04597bb\nSigned-off-by: Jed Preist \u003cjed@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c0f7c388af7d1f453e80132620cd47285fe5320"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c0f7c388af7d1f453e80132620cd47285fe5320"}]},"branch":"refs/heads/master"},"8ce86b738906579c241c5837ce4c6a2ee0103e2c":{"kind":"REWORK","_number":6,"created":"2026-08-27 08:35:11.000000000","uploader":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"ref":"refs/changes/42/1001142/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/1001142/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/1001142/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/1001142/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/1001142/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/1001142/6"}}},"commit":{"parents":[{"commit":"e5b79040d23b80773ca79576ca97040408445c89","subject":"Merge \"Drop leftover of monasca support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5b79040d23b80773ca79576ca97040408445c89"}]}],"author":{"name":"Jed Preist","email":"jed@stackhpc.com","date":"2026-08-17 12:57:29.000000000","tz":60},"committer":{"name":"Jed Preist","email":"jed@stackhpc.com","date":"2026-08-27 08:34:55.000000000","tz":60},"subject":"Install syslinux in debian ironic-conductor containers","message":"Install syslinux in debian ironic-conductor containers\n\nInstall syslinux in debian ironic-conductor containers, to provide\nfiles required for creating disk images that can be booted with\nthe redfish-virtual-media boot interface\n\nCloses-Bug: #2163659\nChange-Id: I61360a9062e365a3ae09853908e8aca8e04597bb\nSigned-off-by: Jed Preist \u003cjed@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ce86b738906579c241c5837ce4c6a2ee0103e2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ce86b738906579c241c5837ce4c6a2ee0103e2c"}]},"branch":"refs/heads/master"},"c6a8baadf239bed6037d453a50b18a3c0c9acfde":{"kind":"REWORK","_number":7,"created":"2026-09-11 14:16:05.000000000","uploader":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"ref":"refs/changes/42/1001142/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/1001142/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/1001142/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/1001142/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/1001142/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/1001142/7"}}},"commit":{"parents":[{"commit":"e5b79040d23b80773ca79576ca97040408445c89","subject":"Merge \"Drop leftover of monasca support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5b79040d23b80773ca79576ca97040408445c89"}]}],"author":{"name":"Jed Preist","email":"jed@stackhpc.com","date":"2026-08-17 12:57:29.000000000","tz":60},"committer":{"name":"Jed Preist","email":"jed@stackhpc.com","date":"2026-09-11 14:14:56.000000000","tz":0},"subject":"Install syslinux in debian ironic-conductor containers","message":"Install syslinux in debian ironic-conductor containers\n\nInstall syslinux in debian ironic-conductor containers, to provide\nfiles required for creating disk images that can be booted with\nthe redfish-virtual-media boot interface\n\nCloses-Bug: #2163659\nChange-Id: I61360a9062e365a3ae09853908e8aca8e04597bb\nSigned-off-by: Jed Preist \u003cjed@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c6a8baadf239bed6037d453a50b18a3c0c9acfde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c6a8baadf239bed6037d453a50b18a3c0c9acfde"}]},"branch":"refs/heads/master"},"9a972298bb490d4ead8f2d5ed68f500a2f55f4b9":{"kind":"REWORK","_number":8,"created":"2026-09-18 15:21:36.000000000","uploader":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"ref":"refs/changes/42/1001142/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/1001142/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/1001142/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/1001142/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/1001142/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/1001142/8"}}},"commit":{"parents":[{"commit":"e5b79040d23b80773ca79576ca97040408445c89","subject":"Merge \"Drop leftover of monasca support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5b79040d23b80773ca79576ca97040408445c89"}]}],"author":{"name":"Jed Preist","email":"jed@stackhpc.com","date":"2026-08-17 12:57:29.000000000","tz":60},"committer":{"name":"Jed Preist","email":"jed@stackhpc.com","date":"2026-09-18 15:21:36.000000000","tz":0},"subject":"Install syslinux in debian ironic-conductor containers","message":"Install syslinux in debian ironic-conductor containers\n\nInstall syslinux in debian ironic-conductor containers, to provide\nfiles required for creating disk images that can be booted with\nthe redfish-virtual-media boot interface\n\nCloses-Bug: #2163659\nChange-Id: I61360a9062e365a3ae09853908e8aca8e04597bb\nSigned-off-by: Jed Preist \u003cjed@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a972298bb490d4ead8f2d5ed68f500a2f55f4b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a972298bb490d4ead8f2d5ed68f500a2f55f4b9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
