)]}'
{"id":"openstack%2Fkolla~1001716","triplet_id":"openstack%2Fkolla~master~I55e76d735ea7cc823a0b557684ea8132a1739999","project":"openstack/kolla","branch":"master","attention_set":{"37203":{"account":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"last_update":"2026-09-01 08:28:37.000000000","reason":"\u003cGERRIT_ACCOUNT_27339\u003e replied on the change","reason_account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-09-09 19:57:55.000000000","reason":"\u003cGERRIT_ACCOUNT_27339\u003e replied on the change","reason_account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-09-09 19:57:55.000000000","reason":"\u003cGERRIT_ACCOUNT_27339\u003e replied on the change","reason_account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2026-09-10 13:31:00.000000000","reason":"\u003cGERRIT_ACCOUNT_34034\u003e replied on the change","reason_account":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I55e76d735ea7cc823a0b557684ea8132a1739999","subject":"Set XDG cache directory","status":"NEW","created":"2026-08-20 17:26:35.000000000","updated":"2026-09-10 13:31:00.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":8,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"24963fb1d7411f37d57b8fefcc3f0459a5163446","_number":1001716,"virtual_id_number":1001716,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-09 22:14:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"all":[{"value":1,"date":"2026-09-10 13:31:00.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-20 19:46:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-20 20:03:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-01 08:28:37.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"state":"REVIEWER"},{"updated":"2026-09-01 08:28:37.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-09-01 08:28:37.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-09-10 13:31:00.000000000","updated_by":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"reviewer":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"state":"REVIEWER"}],"messages":[{"id":"b9a55c267bc198dd7cb96ec48c339291382037bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-08-20 17:26:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b5dc10ade1b6b69aa093763eee1ad55b16364873","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 19:46:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c02dda0d36164e3184854ead39cdd926\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/0b86350e9fd644a8b15d41fb3c5c8214 : SUCCESS in 39m 53s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/2a1f250d89d248a28d991fd652fedfbe : SUCCESS in 47m 31s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/4b2074b1588e4178aee7dbde193a57eb : SUCCESS in 43m 35s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/0e3b14eeb95e44ecba1e086751b27a1a : SUCCESS in 43m 48s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/b2c371d4ccba4744939f60d477621d7c : FAILURE in 1h 27m 39s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5bd647ae25cd4fe484cea99bf764567a : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9b51d2a664f045d9bedbd22e3e6bd60a : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/59f7727abf9a49dd8013e6627807f26c : SUCCESS in 6m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a5b84a4ffcd3c9f7f908dd1b1ce6519dcc1adf0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 20:03:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4eee192ce8284299a43b2d51f23e010b\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/599cca9d6e744e5a9e0ec0028b9172b4 : SUCCESS in 29m 09s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/a9629c0ba6d24d50a7ae231b6759fef3 : SUCCESS in 32m 49s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/027c017d875044a6be37e63e159a208d : SUCCESS in 37m 07s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/0b78c5852d7f41a483e0d5f723165e7e : SUCCESS in 20m 48s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/f2e97b08b25f48ec981348cfbfed3982 : SUCCESS in 1h 13m 30s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2dae23beef8544488a99df62d41586f0 : SUCCESS in 23m 55s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/48b0c109d81143058985f1a3a19b296e : SUCCESS in 28m 57s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/f4ded7fc7c1543d080a1525c329d27c9 : SUCCESS in 1h 02m 50s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/5b8ef037a34348b59151e5fd116a62d0 : FAILURE in 17m 38s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/39bed467fc6a48daa785effcfe954959 : SUCCESS in 1h 31m 31s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/06e34f0cada5411e8ed5d7cc25698111 : FAILURE in 17m 48s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7bbe2ee49b1049649fc8522949e24bb7 : SUCCESS in 1h 09m 53s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/da249664534d44588a23b6d426dcaa95 : FAILURE in 34m 12s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/7339d50e5a854c0aa5c8434824b9992f : SUCCESS in 2h 01m 20s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/02ad0ba08acc4e31b86aafcc78b81784 : SUCCESS in 1h 26m 56s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/acfcd290c4194d6989a73c1d4ec666a5 : FAILURE in 19m 03s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/18ad5c0bb3a4454f86016444025b6df2 : FAILURE in 1h 30m 37s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/38b9f8629b234ba3af442227a72a7971 : FAILURE in 15m 33s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/5d75115496fa4ae4b67d08f5b66eb2df : SUCCESS in 1h 19m 14s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/7783197285164913bf09c08a51e3277b : SUCCESS in 1h 47m 57s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/bf10ecab84b144869a7052081039007a : SUCCESS in 1h 24m 31s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/a89136523f0f460ba706c837d7316fad : SUCCESS in 44m 54s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/0f01223f52794e7fa8b2a24773729ed0 : SUCCESS in 30m 15s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/6b73ad84f4cf4f4eb07631664904b66d : SUCCESS in 37m 52s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/fb191fef483e4a309e7e88a6f4084879 : FAILURE in 51m 49s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/3f2f45c0dcff41478f59d5dc5846694d : SUCCESS in 1h 08m 53s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/770f8b7950af47a398b5b1d03aa1715e : SUCCESS in 37m 37s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/6e1cd9504bcf4d9e98664122b9d2312e : SUCCESS in 1h 33m 05s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/e4a48f6ef669473e9dbea91f47279c3b : SUCCESS in 1h 35m 08s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/406b7764c53346eba0860e2c9ed2258c : SUCCESS in 1h 37m 14s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/8dceff466385493799f5491562a005a8 : SUCCESS in 1h 32m 23s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/997d9e7336dd4fe7bbaecb522a15d6cd : SUCCESS in 1h 42m 33s (non-voting)\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/1bd5a80488ca4324863ddfb8540add0f : SUCCESS in 38m 47s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/8db3d0e7ed4f45feb3001522b9279bc2 : SUCCESS in 53m 29s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/483bed9d591249c78ab918218e9fff47 : SUCCESS in 30m 34s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/c845a6c6b6a74484b7f27f7f6e85444e : SUCCESS in 31m 23s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/c57eccb221494c26b4f9bf1fbd450388 : SUCCESS in 39m 22s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/ace16e3b553843aea9aecc5bc84a6920 : SUCCESS in 57m 13s\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/15aba0b042f74449b8fbd66b9d8f8a7d : FAILURE in 1h 16m 08s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/5bf06e3599c84bd4b775f4e91698d242 : TIMED_OUT in 2h 33m 35s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/d900015d8e504beea972687dfd993c10 : SUCCESS in 43m 42s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/935fb994e35642d68d65101e72d83628 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1f743aa49ce415a8bbd4f773512b12c : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0ed4a31d0b7a479d913064b096560b7f : SUCCESS in 3m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a77b948667dc468cacf3b02d9dba39ee : SUCCESS in 4m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4acbeb817c66469aa5578888a67ec650 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd6a2f8817b44776bc92be24b1c86c41 : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7704f329b66e43899b88426fe15e11ed : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a535035e8f84dfdb4122d0fd4949f27 : SUCCESS in 10m 47s","accounts_in_message":[],"_revision_number":1},{"id":"eac4d1cd74d57139cf015749b505dd76f4a92a85","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-09-01 00:02:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"80470ad29d7bbe2a723f10004de460b4692cf77d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 02:14:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4cbef17d7874395aedf58ea6f2d84fd\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/33b143cd692b411aafb81cdf0d49ccbb : SUCCESS in 28m 06s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/6f94aa2c5df44209813bccf49a8f4bf7 : SUCCESS in 36m 35s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/58a7c33d471546c29b4ccbfa4d3aaa17 : SUCCESS in 24m 04s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/ee53aa7349a44ded89f7099c17b03e26 : SUCCESS in 22m 39s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/854d007f417c4a1d926a58e4b1548821 : SUCCESS in 1h 10m 35s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4079880007304a01a5b1a70f6176f951 : SUCCESS in 22m 23s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4fb9de55c41247ea898ca34cb6210063 : SUCCESS in 31m 18s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/861b379dd4e7487482b71a0e11c44f75 : SUCCESS in 1h 15m 49s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/7c3ec3235e484e2d8656483b579ed4f7 : SUCCESS in 1h 57m 01s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/3a17ea13486f4c5282731150d7bce7a9 : SUCCESS in 1h 03m 05s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/d9b6069dae72454f9ac98301eaa08ee9 : SUCCESS in 1h 30m 50s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03d1c7b78f0d41308548c6a050c823a0 : SUCCESS in 1h 20m 11s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/0b6aea686b594647a5df052f2556c296 : SUCCESS in 1h 44m 25s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/d4dcfe349b3f4f48b3b2d0519792930d : SUCCESS in 1h 55m 11s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/33a2e938515d4d03bd4d0e5701811e74 : SUCCESS in 1h 30m 22s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/ccd108ca03c9487a842108152ef298f0 : SUCCESS in 2h 08m 57s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/4eaf7f6f2e03492a80fb121c77c039a3 : FAILURE in 1h 14m 08s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/334cb97e18b749fe94bd2724b9419029 : FAILURE in 1h 07m 53s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/834047ec46a3401aaeb9e2aa7d95bed4 : SUCCESS in 1h 25m 35s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/f3d0002577f045bb89c806064ec7771b : SUCCESS in 1h 17m 31s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/cf92091a123549fb80a740ff620152c7 : SUCCESS in 1h 30m 54s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/7c3eb0558f1544aa880c0fd3fc9ddfa3 : SUCCESS in 43m 01s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/bd51ce71f493420cacec38cc479dfddd : SUCCESS in 25m 40s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/2c81389084e140da9b3ba603078214e1 : SUCCESS in 34m 02s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/276ec328d33b47d093c70e763a061052 : FAILURE in 43m 06s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/b6fa6c2ff7464750bc26e4a81093d953 : SUCCESS in 1h 30m 52s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/6f8ae0652fcc4db698c60aa263ae25d3 : SUCCESS in 1h 19m 30s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/7d5f97463d454f189e06cfd184448fef : SUCCESS in 1h 11m 43s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/eee9fdf5479d40239e18f37e0fcc1d4b : SUCCESS in 1h 22m 38s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/9ed1b6ba898b46e0bf56e40fdeb9b5e7 : SUCCESS in 1h 56m 02s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/b3768f12097349348549879ed821f3e6 : SUCCESS in 1h 46m 38s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/1f1dba08f06c490b850ceb4837ef202a : SUCCESS in 1h 59m 07s (non-voting)\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/5b4e1934fefd44c5b76990b3d5d99210 : SUCCESS in 38m 46s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/bf6a1a1eb8264e1d960ee624ffaad68f : SUCCESS in 55m 32s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/e16ada1d3cf44695a9ce25044a63ef8e : SUCCESS in 21m 03s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/486f6b16f14e49cba773f67921271f6a : SUCCESS in 46m 29s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/32654906ac994f9ca2d7826008ec758c : SUCCESS in 23m 23s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/208cfa61f1ef491398d3b32dabc44211 : SUCCESS in 32m 21s\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/c47e171ff8d44632acf74d9d7c28fd6e : FAILURE in 1h 09m 45s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/d6b9c58fbc0a4d8cb6a851024f5345a9 : SUCCESS in 46m 07s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/8d8e9d0f995343ebb67a1176e09dabf7 : SUCCESS in 1h 27m 59s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/029145cd81fa41e3bbf40554f54c0879 : SUCCESS in 2m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e181f11621ec478583d91480f59dd76a : SUCCESS in 4m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/db21ad3b6ffa427ca73a210a347c2158 : SUCCESS in 2m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9d03be03ef1f485094573e9a54d9eca6 : SUCCESS in 3m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/32548f001597482381f9cb738a77e35a : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/288cd9553c3d45cabd164ad9e246c265 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05ea33a61896475398ebba14dda95107 : SUCCESS in 3m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8cde240f3db430cac74420356b7746e : SUCCESS in 10m 38s","accounts_in_message":[],"_revision_number":1},{"id":"7021734683069c682ee66f6a435d8e07339408d1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 02:23:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3a04a50a8a24b3b9090ad1759b9db32\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/8d7df7a2fa554480954ed928553ee608 : SUCCESS in 1h 08m 01s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/99bc28fcc0234156ada99c62c290e752 : SUCCESS in 1h 13m 37s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/1d9629790f58445da1a236a8c22b6bd8 : SUCCESS in 1h 21m 35s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/46053e6c17c040ab9a0bbc028132f58b : SUCCESS in 1h 01m 38s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/839dc216f1f94b08932ff922c3fb49d1 : FAILURE in 1h 40m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bdef31f358064e869c6723db2728d9d7 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9ac6216d5d20442a9bed67a1b89c322c : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b4ee90fd4d0f4dc0bfd3fef81a625990 : SUCCESS in 9m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"02a38941175749f059737648e38c9d2cdd950285","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-09-07 05:37:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e2982d3cffb7c616052865f1e0ebe65bf17f2c41","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-09-07 06:01:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4ddd175fa4a925d017972cb2e59480c30aea2f03","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-09-09 13:24:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b98eeb621e65e71dc7a979d32e561bb7a9b14e7f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-09-09 13:38:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"60be53e9d333661a31c55e3f16e5798bed0b9cee","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-09-09 19:57:55.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"4f7570707fae7e25b67aaafda8fec12c9fa60f8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-09-09 19:58:15.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"c7aba3c49b3e9c2d556635a90685b4fa5261c2da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 22:14:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0afcb4fe466b4c54a657fc0eab600857\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/2a14b381b9214c66ba0b6d7fe8743a91 : SUCCESS in 24m 09s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/c33d8bad3a034635bc69d3a24e3b56e5 : SUCCESS in 28m 26s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/b5306efb76d34f89b0b52b1e7b903720 : SUCCESS in 28m 18s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/40866d11a4774581be1f1db362eb94ee : SUCCESS in 33m 16s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/4a6bf43292444197878bacf5d7c19485 : SUCCESS in 1h 00m 29s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/91d42bb0e45c4e3cb18d645d6e2a3c51 : SUCCESS in 28m 41s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/055057d8df0040b39dfbb61eea6c4436 : SUCCESS in 29m 05s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/32f0133787a6445fac6b060b3e411fd7 : SUCCESS in 1h 12m 10s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/5e1c68986ac646baa13d663f9ae6f387 : SUCCESS in 1h 48m 35s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/474d221df1ba4726905608e6b97ebe23 : SUCCESS in 1h 06m 32s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/4052856694e4452ea393ab83d723177f : SUCCESS in 1h 28m 07s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/61b700f85c454dabb3049d7c52b6eda5 : SUCCESS in 1h 27m 45s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/fd401efbcca7477b8dd18ca04aea0733 : SUCCESS in 1h 48m 25s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/674b9528f4ca4d88ae3ed083e79c1e66 : SUCCESS in 1h 42m 44s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/a9093aab8d6b46bb8e7d9d1eaaea0d6c : SUCCESS in 1h 30m 51s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/55ca15d99f094f73911a4f17ab9228f7 : SUCCESS in 2h 12m 06s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/3cd9ef6e55f3436bae0b1920d89e1b20 : FAILURE in 1h 17m 17s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/1fec1bcae4bb43a7966766c90ca27137 : FAILURE in 1h 09m 17s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/e5836a04a762491e93ae23b178dc587f : SUCCESS in 1h 28m 38s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/644d669a612649aeaeb7d8c430ca490d : SUCCESS in 1h 16m 32s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/0cde23e5a3624208b096ff115fc1bf8e : SUCCESS in 1h 40m 12s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/e572e277b845498e8fc8d7758d95580e : SUCCESS in 37m 50s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/2b2cc6085e9446bcb6dd10f2083b09a6 : SUCCESS in 30m 10s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/b3887ead962a49b6be5cfa6b296979b4 : SUCCESS in 27m 05s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/d4739ebc7c644994bbb3cb29a239060c : FAILURE in 52m 29s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/60e81071468f4f44be99f466ad73bfcb : SUCCESS in 1h 26m 46s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/5dfa26ff9b764cf58fb6dfc99f361ecd : SUCCESS in 1h 20m 07s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/0aa7095973ba4e79aa5fab181d46763d : SUCCESS in 1h 14m 24s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/bb56a2d900a44ea0853aa1c2dcc0d065 : SUCCESS in 1h 27m 53s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/8b1c3ecf2ed74422be9dfa8451776911 : SUCCESS in 2h 03m 02s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/c7c4ca1e86694a539b76937ca2812325 : SUCCESS in 1h 45m 13s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/3174dfbf7fad4062a07685aed7385cdd : SUCCESS in 2h 06m 10s (non-voting)\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ec7a29a38226427fb6cbb00a51362013 : SUCCESS in 38m 29s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/a03fec4d174440fdac142522610eeabd : SUCCESS in 44m 29s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/52f05bf8994a48889b741d19c58c1fe6 : SUCCESS in 34m 14s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/f4256e574ebc4fb19fe2e5de16c00327 : SUCCESS in 1h 04m 19s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/78f27a56717f47a9a4fdd09d2cc45f72 : SUCCESS in 39m 27s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/36a686ab1d8444e69ebec6e39e44eb16 : SUCCESS in 43m 13s\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/a40b7f2f35a649a098195c7cbd6f3414 : FAILURE in 1h 09m 08s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/6279e83716fc4059a6a06a20580f92a4 : FAILURE in 1h 10m 47s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/2cae754717764c359ee57593f764738d : SUCCESS in 1h 22m 25s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/79fa663e0566455b84e8b16ba3c75dfa : SUCCESS in 2m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edf9df3b1f564b0ebd5ae8294e264808 : SUCCESS in 5m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0a14c67e92e84a6389b944897224bfa8 : SUCCESS in 2m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c920e7a692574ab5b79bdc84f167f5f3 : SUCCESS in 2m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/04ab7961766b409b96cc4a6d9ace4d32 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bafa143bb2fa4f38ab4ea433a75954cc : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91eadc11963f42048f6148c25d6162a8 : SUCCESS in 8m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5000568f44e4a008f5ec1d8c4de9279 : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":2},{"id":"536d7439d4f14f710b507402d140d08de6503a67","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 22:14:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab9a671f5f814e59808233e52de7314b\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/9e7f00964f10467eae4847b2d5b46057 : SUCCESS in 50m 27s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/ba3c50101fb94981a83c0a33cacb1d71 : SUCCESS in 43m 24s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/64e0a2a8e0e14be08f8e9f5ff4fa78e1 : SUCCESS in 56m 35s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/88809c6627fc4915a70f1aa518a3e419 : SUCCESS in 56m 28s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/497922c744264b468fc1c0daeae84283 : FAILURE in 1h 21m 02s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3477ab75dbfe4a54a3e99fce465fd998 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ffe6003ace8042a78f725c5a45b481fd : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3e51e4384b2741ad927ad129ff569808 : SUCCESS in 11m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"24963fb1d7411f37d57b8fefcc3f0459a5163446","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2026-09-10 13:31:00.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2d2ac982271d3229ddedca2b30e7e5942cccc4ea","revisions":{"50489693823b0d00612428039118d93385738bf3":{"kind":"REWORK","_number":1,"created":"2026-08-20 17:26:35.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/16/1001716/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/1001716/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/1001716/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/1001716/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/1001716/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/1001716/1"}}},"commit":{"parents":[{"commit":"3f01507eb215b2162e9815257cbd2ba4091251df","subject":"CI: Switch to build from Zuul cached repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3f01507eb215b2162e9815257cbd2ba4091251df"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-08-20 17:25:23.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-08-20 17:25:23.000000000","tz":120},"subject":"Set XDG cache directory","message":"Set XDG cache directory\n\nKeep application caches in the ephemeral container\nfilesystem instead of service home directories, which\nmay be backed by shared persistent storage.\n\nChange-Id: I55e76d735ea7cc823a0b557684ea8132a1739999\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/50489693823b0d00612428039118d93385738bf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/50489693823b0d00612428039118d93385738bf3"}]},"branch":"refs/heads/master"},"2d2ac982271d3229ddedca2b30e7e5942cccc4ea":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2026-09-09 19:58:15.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/16/1001716/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/1001716/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/1001716/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/1001716/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/1001716/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/1001716/2"}}},"commit":{"parents":[{"commit":"ef8d6f3df6fa97f2227a847c3fdfddd1d2fc1acf","subject":"Updated from check-sources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ef8d6f3df6fa97f2227a847c3fdfddd1d2fc1acf"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-08-20 17:25:23.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-09-09 19:58:13.000000000","tz":120},"subject":"Set XDG cache directory","message":"Set XDG cache directory\n\nKeep application caches in the ephemeral container\nfilesystem instead of service home directories, which\nmay be backed by shared persistent storage.\n\nChange-Id: I55e76d735ea7cc823a0b557684ea8132a1739999\nCloses-Bug: #2166926\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2d2ac982271d3229ddedca2b30e7e5942cccc4ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2d2ac982271d3229ddedca2b30e7e5942cccc4ea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
