)]}'
{"id":"openstack%2Fkolla~1002078","triplet_id":"openstack%2Fkolla~stable%2F2025.1~I375f01701f84dc117a842c344b5f796b9a582131","project":"openstack/kolla","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-08-26 14:21:27.000000000","reason":"Change was abandoned"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-08-26 14:21:27.000000000","reason":"Change was abandoned"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2026-08-26 08:39:59.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I375f01701f84dc117a842c344b5f796b9a582131","subject":"[2025.1-only] ceph: Bump RL9 images from Reef to Squid","status":"ABANDONED","created":"2026-08-24 12:38:00.000000000","updated":"2026-08-26 14:21:27.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8a48dae2af617fef59bf5e2e429508d6ac2c74bc","_number":1002078,"virtual_id_number":1002078,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-25 16:12:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":-1,"date":"2026-08-26 08:39:59.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-08-25 14:22:40.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"date":"2026-08-26 08:40:13.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-24 12:51:13.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-08-24 14:15:05.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-08-24 14:20:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e9bad93e722ef4b8baf83b89d84a61fdb04f0e1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-08-24 12:38:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88127821b62b66af8cce3f0994efe09f51956959","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-24 12:51:13.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d9143fe5544918ed6bbdb313d35286b8eabddae0","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-08-24 13:28:50.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"97fab135092e321459e0d7ffdae4ef9f7ecf836f","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-08-24 13:28:56.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"efe6e0b9a0ab1c00c685210e9f832d9966b3ecea","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-08-24 14:15:05.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"911363df5ad184c88ce95c3d599dfebd11913f02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 14:20:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/83a1397e90d040919a0153f380878efb\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/8561cf7b27944b8bb6bc2e2e0cb94d73 : RETRY_LIMIT in 2m 42s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/4b16b7e1f22a4e358d227578614b30a0 : SUCCESS in 55m 52s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/3885e6e3ee09438abeeda26e3900e10e : FAILURE in 8m 57s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/56560b11f974498f9919c71fa6a9f29a : SUCCESS in 1h 16m 49s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c1bde559df31472daf6d162035995041 : SUCCESS in 48m 18s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/47828208167248c2ac6fbfe8921cfd47 : SUCCESS in 33m 28s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c4e2c7ea5f544bcea21e61a0820a9599 : SUCCESS in 57m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64c2f388ef524bc8981d9aca9bd63899 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/565d9df6bfa245e1bb3068ecf37dc025 : SUCCESS in 3m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ee42a225b7ef46c3b4cae1a98ba9011c : SUCCESS in 3m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13a14dd80c9142f794dcd54fe35b6f27 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a24474a293d4c28a100cf1d4c49b3e0 : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4be1564743cc45bd96dca6fbb9a0cb69 : SUCCESS in 13m 07s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/133bb4f57b604d889f47398f98ac59da : SUCCESS in 3m 20s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b5362d97463c4c0aa3214f30afd07439 : RETRY_LIMIT in 2m 05s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/4c19bd2f7f4c4b118a8d8054ea85137d : RETRY_LIMIT in 5m 50s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/eb4a4036bb014ad392fb8ff81d715189 : FAILURE in 8m 44s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/c7e35773b71a4e1892b3859df3076413 : FAILURE in 12m 54s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/713a325b78664a1f886ba5e4bfc5cc23 : FAILURE in 9m 54s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/aab5551dc9fd4077a7abe75114eaceb1 : FAILURE in 9m 24s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/a232ea3a88ec4ae587ded56f40d919bf : FAILURE in 10m 05s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/5d9011c8f68d4a43906382e4eeaefc85 : FAILURE in 10m 49s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/893fd626648445d1aaff0b3057565bfa : FAILURE in 11m 09s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/043d762f17d74d178749dcdfbaf16d24 : FAILURE in 10m 06s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/3b3062a5ed1c454fade5f24f9d68f53b : SUCCESS in 47m 52s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/98d6ee74888a41b6b58e8b2a381570ec : SUCCESS in 57m 04s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/d732ef6cc3d14ffd9ef2a0662f2df3b3 : SUCCESS in 1h 16m 57s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c81a67593a6d4ef499f8406042810567 : SUCCESS in 1h 06m 38s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/fa54afbb439142a89fb7f17d9ef56a30 : SUCCESS in 28m 01s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/7b971a86f00649949cd43afb3634682c : SUCCESS in 1h 02m 50s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/fa2a11778e1a484884aa92f22492a2e7 : FAILURE in 1h 10m 35s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/0ba2cda78b6e4538aeb6f39314a2f7fd : FAILURE in 1h 12m 52s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/e0a2a58e7e3045ec927773d765f51ce8 : SUCCESS in 32m 03s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/8386c3f5ffb5407e8dc80b8b866bffd2 : SUCCESS in 1h 26m 24s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/d76cbd76db654c6d90f3a7bf02ad77c1 : FAILURE in 25m 58s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/82997950b45a415e9876eb5ae922ff18 : FAILURE in 27m 48s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/c33f3c8d661e46e2bb674c40c4c8ebcd : SUCCESS in 27m 24s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/7ab1ffd64bfb449bb47adf129dc99ec8 : FAILURE in 1h 33m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5c39c2eb3de129008c14dc25a17cb7d0abaa47ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 16:37:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c61fba75fc04000bf33be1c505333ad\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/43b65ef1bd7941c495ee3b938aeee3c3 : RETRY_LIMIT in 4m 27s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/c45749b5adfe46028afd8de4fd9a921e : SUCCESS in 1h 52m 57s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a5c337aec8484a9a9052f540e19c0209 : SUCCESS in 1h 27m 54s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a85bf9ac8fa1411e96f7c4b080b8b863 : SUCCESS in 2h 06m 28s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/341243086fa04f678cf5747a76d623bd : RETRY_LIMIT in 3m 23s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/b5c8f7ae9a684cbf9a42524a2b49f1d8 : SUCCESS in 1h 20m 34s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d46d6c5930a44f10a9642eb27df76311 : SUCCESS in 55m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4799dbcba36f1ed14e217ea2437d99ccd3b00158","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-08-24 18:39:36.000000000","message":"Patch Set 2: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+1, Review-Priority+2 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"9480fa31c83e0da288e39afd5d90a19d09249642","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 20:03:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/10fbc4ed712142a4ac0fd3e29aade9c4\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/90bdc5ce3cde4d53995cf967fe21669f : RETRY_LIMIT in 1m 46s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/69f81880e4e94120ad7e82f6dde76a66 : SUCCESS in 42m 26s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/bbfdc6bb7b024b3393fea7b1ed8e0e9c : SUCCESS in 1h 00m 23s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/eb882dbeb5eb42439468f9655b2606e8 : SUCCESS in 1h 03m 59s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a7babc6cfe5f4c5fa0f76040be843673 : SUCCESS in 49m 04s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/1ab0f3a9250d404683185781a9d77fa9 : SUCCESS in 22m 34s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d4a8b933bda74dae92003f3c7c66a520 : SUCCESS in 55m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/782bd63e6cb44f2fa8cf8cb3299b1c27 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/76c782baddee49d89a322f9818eba2a4 : SUCCESS in 3m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4ab362d4af4d49d58ca93e0f1bb44ed1 : SUCCESS in 3m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1cfcb2b669247139659ac44b3c1e836 : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c154e815c53462c948db8cc60ebfebb : SUCCESS in 3m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35382458cad947f89abe2fcadef0e710 : SUCCESS in 10m 59s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/082397c02e9049f9bb8888b3c38a9f22 : SUCCESS in 3m 21s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/687ef4567a3e427da0712e60602fc998 : RETRY_LIMIT in 2m 25s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/5488d303f29344fca0946191e6c99698 : RETRY_LIMIT in 2m 18s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/abe84da093b44822b19364642192bf57 : SUCCESS in 48m 37s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/c8b6cff98fa84fa189fe41c8d82cccd8 : FAILURE in 58m 59s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/e2312161606e4e55bbe532fa6890d590 : FAILURE in 47m 23s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/b94343c6f87f469697e4c17a316dbd12 : SUCCESS in 18m 57s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/2d2755e84b254a8d92ea81961898bf65 : SUCCESS in 1h 00m 39s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/f57034701d074c34b60b6e5781fc32af : FAILURE in 37m 37s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/c8a1a79b8d144b19a1b80b9849eec3b5 : FAILURE in 40m 34s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/493c75182148460883f205b4dc164895 : FAILURE in 40m 28s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/8b09bd4806a14d47b89ba055a35d6484 : SUCCESS in 52m 35s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/0f542fed06164599a1cda3961331fb37 : SUCCESS in 40m 38s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/4d85474b06814df398e261f69e778632 : SUCCESS in 1h 01m 11s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/33a01269de5144eb8e0d86580b6a8fc9 : SUCCESS in 52m 25s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/9df7bcf81eb240469a7dbaf28f9de980 : SUCCESS in 14m 58s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6c56a16091cc490bb930772269fe8dc5 : SUCCESS in 58m 40s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/f37dd70c8d7a4f28843c466be1c8ca11 : FAILURE in 1h 06m 07s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/87c012419c9e4c5cb0f7a393433c47bd : FAILURE in 1h 16m 53s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/2015bfce4b43468b84bec376f1fe079e : SUCCESS in 31m 06s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/75ae7e4f632f4bc2858a3d3e3dbf1ae8 : SUCCESS in 1h 14m 02s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/1ae3a37d3bfc4e4fb4ca3681cbe675af : FAILURE in 22m 25s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/9abd7907920943aca86e33b1212a1cb6 : FAILURE in 24m 23s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/1c19883b30f041aa85439ad2d492e0fb : FAILURE in 18m 28s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/ef9750ada9ee4125bfa804c63f43e2ba : FAILURE in 1h 21m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"68731279f6e16ed66bffff4b39e658d609b54813","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 22:25:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e443409deb194231abd43b803c8486cc\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ac152bf59eea4945b48bed972a875285 : RETRY_LIMIT in 3m 42s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/6277513dda7143ecb86d6d615e4dc6cf : SUCCESS in 1h 43m 24s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d7fc5f376d3a4ff19945b955261c8101 : SUCCESS in 1h 41m 39s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4b527badc32343c8ace0b85ba00af629 : FAILURE in 2h 03m 36s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/6460fd96e39b43cb9bce5d060d0b6b04 : RETRY_LIMIT in 11m 48s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/be80576220404368ba390ef91cbcee99 : SUCCESS in 2h 47m 14s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f6b64a96b9034885989697dfc5d057fd : SUCCESS in 2h 29m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dfab7cd46d46f675ce0c61862980c8cc7cc77b18","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-08-25 05:52:20.000000000","message":"Patch Set 2: Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ca1f8dbb0f30bf6b89717769293cb01f018deb12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 07:34:53.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9a3cf884550445f299a53c92415cfdff\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d156def40f204d52a28f9cbb3712e9b5 : RETRY_LIMIT in 2m 34s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/ff75d9d1f0684e48a2ef160fa481da0c : SUCCESS in 55m 16s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/0849347fce4d4ebe9cc298c6087605fa : SUCCESS in 1h 20m 19s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/bef32c93cc6149cd902685dea36693a0 : SUCCESS in 1h 30m 52s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/82fd8f679ec042e5bcbf7338413d2bc7 : SUCCESS in 44m 03s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b7ddd7985124462eb705197c26ac5c37 : SUCCESS in 30m 01s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6f2e01f96dc941e5a8e06e0ebb5e86bf : SUCCESS in 52m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51987df2ba284f64b48d1e345d8279b6 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f64e1f929ed643d49841c6a9e8f57387 : SUCCESS in 3m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e55b8fa89ce94c76833f4c01b980b839 : SUCCESS in 3m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f13de46808954b51b26ca6ea9edd69de : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62767a9d4f134e2db2053ff46c49822b : SUCCESS in 5m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c12093c4ebbe4fcc8d506f8f35177254 : SUCCESS in 9m 38s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/93cb715d0e534f2d9b1467d6d1330c28 : SUCCESS in 3m 08s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8b1945c77dc245278c335c690a35459c : RETRY_LIMIT in 3m 06s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/742eb353c8964ed3bd4d5c5e12b6db04 : RETRY_LIMIT in 2m 09s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/09bd564e7fcf4ac2ac129454c90ba8eb : SUCCESS in 1h 26m 43s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/202d31fbb09d407da4fb6f65e06155a9 : FAILURE in 1h 29m 47s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/b9913ac97be447029b0da3cdf860725d : FAILURE in 1h 03m 55s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/2db97ec373724f3aa3a0da99c915b68f : FAILURE in 30m 36s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/adb04de139b9421fb305812d2f3a4798 : SUCCESS in 45m 49s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/8cd7220f31e344f0954f53480e8c344e : FAILURE in 27m 56s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/29795449ffcd4b39b0b8cbd5e2a49b34 : FAILURE in 39m 51s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/26c559a6a1774211961abefbadbc9689 : SUCCESS in 23m 05s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/98138beabb2842e78ab21517aa50b48e : SUCCESS in 44m 43s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/442cadae1efa4b6f8808772ef1fca5d6 : SUCCESS in 42m 07s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/9661e9e95a224500a51831884fd85a7b : SUCCESS in 1h 23m 20s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7638bd578cb447d7b28430a41e8c4f4c : SUCCESS in 55m 24s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b34ff4fffe02406d9966a243232350fd : SUCCESS in 31m 37s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/13ba310ccec54cb482892a90becb4a59 : SUCCESS in 1h 01m 08s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/535e3730513340f38a5bc57fe5c2b639 : FAILURE in 1h 06m 52s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/4e1bcb8e33114cb1a21f5dcf090c6600 : FAILURE in 43m 58s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/67c7458311be41ffa001f89cbe67ccaa : FAILURE in 30m 08s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/eb3a570711ee4c41969302b096aefe75 : SUCCESS in 1h 21m 17s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/5ad5ab9015a24115a91f3157e15039dd : FAILURE in 23m 31s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/c4616ff7ed444711ba8e2412fe1e5820 : FAILURE in 26m 05s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/a498f1a0cc714f26a391f22c70151f88 : SUCCESS in 19m 06s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/da92b11a93c94eb7803f5fd0ea896bb7 : FAILURE in 1h 31m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6a3be5e085024a57e70c4620cdbeb8f278074cc2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 09:00:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47f9632d5b5c41f7a649f25a71a62c90\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ea32ef71f91549758da7d8a96a76f868 : RETRY_LIMIT in 3m 18s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/35079d0d0f304cc7ad915985f92f86b2 : SUCCESS in 1h 43m 36s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/82f0c4704a55413990826c3e8a2c353b : SUCCESS in 1h 48m 38s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/304b97644b7546f69cf0b59e1a658ba4 : SUCCESS in 2h 28m 14s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/e10b36fea2b7440d8e051b24d3b79586 : RETRY_LIMIT in 3m 07s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/bce39098b6fe49ec8067bdf527585f29 : SUCCESS in 2h 00m 36s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/951c47518f2049a38bf5a457eeebbaec : SUCCESS in 1h 42m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"387c4b5ca890b5361e9551bde84910cecebd74e3","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-08-25 10:48:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8006dd6dccbbec0c727d038af468842490af3544","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 12:40:27.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4ce8aab3a8ff4ccda02df30a99fd2dcb\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/524f1ba2eae1414e87fb10408607120e : RETRY_LIMIT in 2m 50s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/96466c01ccc94815aeea6322dd1e0f50 : SUCCESS in 47m 23s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/2cf84d515fd24061adf8063f81a0de3c : SUCCESS in 1h 14m 36s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/2777772bc08441bfac8b9371eb1e110f : SUCCESS in 1h 33m 35s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/2d2676b8b51a4740be102857eaeb8534 : SUCCESS in 50m 19s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ea3bd7187c4b4c8cb221a92dfeaf97a7 : SUCCESS in 32m 49s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/409e2bfa3a9a408ea67a805603e57adb : SUCCESS in 57m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b987c1139b504cada87732472602a93f : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4dbf7f1f84fe40d0b1aae459ce13c419 : SUCCESS in 3m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a7a8254069e3409389ef025c96381899 : SUCCESS in 3m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40afb73bfe3b417b96932abab00e402a : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09740065fa05400fa675c737359a1060 : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ec8112f30f24e069f7171d22ddcebb7 : SUCCESS in 10m 42s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/3ee2f0ed24cc45b0b7cd357a7d948a48 : SUCCESS in 4m 37s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8c79ea6ed28f4754a0e6ac5aa45ecf28 : RETRY_LIMIT in 2m 47s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/7bd9bc8bb65c4bc2b96183b08f9d9dbc : RETRY_LIMIT in 3m 05s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5c93d4217e9543bfaac2f3127d482abb : SUCCESS in 1h 22m 57s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/8d7ba7b5031147ee87e3a05d7513611a : FAILURE in 1h 12m 33s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/5bec8cf1c4494c41bd16b1e66ddcc97b : FAILURE in 1h 24m 22s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/57ea77f5c63746f8858b22ac4a81e6c0 : FAILURE in 34m 32s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/902ec03912454b3eacd26fd5c48f43a8 : SUCCESS in 1h 31m 24s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/ac49c16a75d64a6ba90871a9d3cf3c5a : FAILURE in 25m 43s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/71bbb771df294981a419d8d17f9ea7cd : FAILURE in 39m 29s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/bf72f58fee3a416798b5678fc6f46921 : SUCCESS in 33m 30s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/478bf1648b5d48b7803c44db3759f15f : SUCCESS in 49m 16s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/23ada53aa70b48a0a958a2314e5f5469 : SUCCESS in 1h 01m 52s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/2eb5892b3d1c4cbfa4bb1b172ab97177 : SUCCESS in 1h 23m 28s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b8cdf4fefdc541eba4dfdb50a4d5cb8e : SUCCESS in 43m 35s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/cdfb4e1cdc304f4fb550c3df8c11134a : SUCCESS in 34m 14s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1b363971cde44b87adc733b5f518090b : SUCCESS in 1h 02m 48s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/a48781bd0acc4b43b348b9578f3c1c91 : FAILURE in 1h 04m 03s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/8a5fc5c60c5f4d088d91964d5313f64e : FAILURE in 1h 29m 05s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/af95d0ae3f4744158cb952ed739a7f2f : FAILURE in 28m 40s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/3167ae57792f4e679cf782a00cfef634 : SUCCESS in 1h 18m 41s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/28a5d7d10c5848d9a7fd66ff6e08baec : FAILURE in 26m 43s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/e9a91091fc9d4ce1ac439afdc1a89bf2 : FAILURE in 24m 59s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/22dd221782644d309611b2e9e1e2863c : SUCCESS in 34m 27s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/652c602133e54296bfd1ccf4aa2348b8 : FAILURE in 1h 33m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ff10b68f7ccfa388dbaf66c2d1b51c4a96b8e37d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-08-25 12:42:09.000000000","message":"Patch Set 3: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+1, Review-Priority+2 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"b10ef2e4a75ee646401f55ff0924832ae8fd0184","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-08-25 14:22:40.000000000","message":"Patch Set 4: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+1, Review-Priority+2 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":4},{"id":"5bd757f24fe47a4c2825246b09089c52dc4c92c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 16:12:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad29dcf7916f4cd5a4ab24ccfa4f9a5b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/fb455999dbb2457a9b1f4a2f05044c67 : FAILURE in 52m 34s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/6790efd70a14474a8614780cbaddb499 : SUCCESS in 57m 52s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/3a6b1abf15c94c558d62f168e4795188 : SUCCESS in 1h 32m 57s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/20e6d4e7e66a4168acf4ab0391660839 : SUCCESS in 1h 17m 46s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e32a32a7b6dc4f9eb576bc89c9ff21af : SUCCESS in 58m 59s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/8c919399c6ba4ed39e64871c2a1218ea : SUCCESS in 22m 58s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6528819de99148368bafc52a17f9ad73 : SUCCESS in 53m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97cfef63abb646efa6c51c4eb6571cc7 : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a8029a8d5279485788c8dd4d005e08d0 : SUCCESS in 4m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a5648fcac44641ec8f6a30d0cfb10128 : SUCCESS in 2m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3541bc1679724a06bf1e25d43fd0d7ed : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50ed9333004244f48a87ff9d9d6c939e : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87c82f3c88ca46acaa717a612fcf9a13 : SUCCESS in 9m 50s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d825f2440ecb4774a8e3f20049a5a54b : SUCCESS in 2m 47s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/68cda46247c7460985c6b117f486bc66 : SUCCESS in 1h 04m 00s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/07e563f669884963af3030c2188134dc : SUCCESS in 1h 25m 02s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/79a8e809a4324393a1ec0327d7c09a89 : SUCCESS in 1h 27m 18s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/69a08b86beb643a6bb83b3da11ebb5f9 : FAILURE in 1h 28m 07s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/fdca23bf9c3c4a7fb53379e0dba83562 : FAILURE in 1h 24m 53s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/94a3a8ed62b1428eb2af3b4c782c570b : SUCCESS in 34m 49s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/581d35c71bbc4234bb7663c028ac43a7 : SUCCESS in 1h 20m 45s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/7570ae44906f43a0b55f3efef51128bb : FAILURE in 23m 31s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/25304e82159f44fba4f1d2464fbae289 : FAILURE in 37m 36s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/5b655c87419c4f9392f1f1292d343cb0 : SUCCESS in 21m 41s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/31710937e08a4825b7f20eaa60250969 : SUCCESS in 41m 56s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/6dfbfaf3bef440c59c9df980a7098b73 : SUCCESS in 41m 21s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/0156c60f4ca74fe290aa2b6f6981123b : SUCCESS in 47m 47s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/07b7254fd7aa47e7bdda2d6d0ca2ae41 : SUCCESS in 43m 05s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/3449a4e6ff8c42d39a964cc58d38d600 : SUCCESS in 25m 31s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/62b0bb5aad2641e79bc77112286add70 : SUCCESS in 1h 02m 16s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/a5c1a21bf22841ae8cb62a79b45e1f46 : FAILURE in 1h 09m 35s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/16ab2de5605d4391905eb8e51576a117 : FAILURE in 5m 37s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/b25b03a597184fd3a0b713624bb3d49e : SUCCESS in 19m 36s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/e18de6a5919f448da844bb5b46c100ed : SUCCESS in 46m 43s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/66919876e10042f8b40ec909126c9159 : FAILURE in 24m 26s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/b9acba7adabf4ac3bc65795d13ec1826 : FAILURE in 24m 47s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/2a09cb8ec66d45bfb0e3eda92f717ce2 : SUCCESS in 35m 49s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/5636c33810a846208c92290a074aa1c6 : FAILURE in 1h 32m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d52e824a5bc5188ebe83ffa07853909485e4c999","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 18:00:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6d1cfd8eeaa4e7db1e820f0368f328a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a84740932e404bb19c2d38bd804df200 : FAILURE in 1h 47m 16s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/5f564a6d8a2344aab450ade39caf740a : SUCCESS in 3h 01m 43s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6cd1cfc9c7144d46808451c3bc33aa51 : SUCCESS in 2h 17m 01s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7d1e9dac214946428be5f010c3eea1cb : SUCCESS in 2h 01m 37s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/91d10fe1e83e4a1bb207d1f24f132695 : SUCCESS in 1h 23m 52s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/4c45aea42a6941ad92483af4b8e3c37f : SUCCESS in 1h 27m 59s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/11a488fb7ca9474681b3a799d62e860f : SUCCESS in 1h 09m 57s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"985ac9a200bf0b32f18994662f1d0a965a069679","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-08-26 08:39:59.000000000","message":"Patch Set 4: Review-Priority+2 Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"24a06de1b78e86777466bbbdf89d79bc9ebbad84","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-08-26 08:40:13.000000000","message":"Patch Set 4: -Review-Priority","accounts_in_message":[],"_revision_number":4},{"id":"8a48dae2af617fef59bf5e2e429508d6ac2c74bc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-08-26 14:21:27.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4b6ae8d137c0b195ec41098b36724d06c88f0f49","revisions":{"14ef5260fccb16f54fb83d734cd90b56c1fa14ed":{"kind":"REWORK","_number":1,"created":"2026-08-24 12:38:00.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/78/1002078/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/78/1002078/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/1002078/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/1002078/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/1002078/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/78/1002078/1"}}},"commit":{"parents":[{"commit":"662eebdf470a3c75a76eb50088a65e73df64bb5b","subject":"Merge \"ironic-pxe: don\u0027t fail if symlinks already exist\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/662eebdf470a3c75a76eb50088a65e73df64bb5b"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-08-24 12:36:57.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-08-24 12:37:38.000000000","tz":120},"subject":"[2025.1-only] ceph: Bump RL9 images from Reef to Squid","message":"[2025.1-only] ceph: Bump RL9 images from Reef to Squid\n\nCeph Reef went end of life on 2025-03-20. Switch the Ceph client\npackages in Rocky Linux 9 and CentOS Stream 9 images to the Squid\nStorage SIG repository. RL10 images already used Squid.\n\nChange-Id: I375f01701f84dc117a842c344b5f796b9a582131\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/14ef5260fccb16f54fb83d734cd90b56c1fa14ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/14ef5260fccb16f54fb83d734cd90b56c1fa14ed"}]},"branch":"refs/heads/stable/2025.1"},"c933b15aad60c81e271f733c8ec3ec3e06520d7b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-08-24 18:39:36.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/78/1002078/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/78/1002078/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/1002078/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/1002078/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/1002078/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/78/1002078/2"}}},"commit":{"parents":[{"commit":"662eebdf470a3c75a76eb50088a65e73df64bb5b","subject":"Merge \"ironic-pxe: don\u0027t fail if symlinks already exist\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/662eebdf470a3c75a76eb50088a65e73df64bb5b"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-08-24 12:36:57.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-08-24 18:39:36.000000000","tz":0},"subject":"[2025.1-only] ceph: Bump RL9 images from Reef to Squid","message":"[2025.1-only] ceph: Bump RL9 images from Reef to Squid\n\nCeph Reef went end of life on 2025-03-20. Switch the Ceph client\npackages in Rocky Linux 9 and CentOS Stream 9 images to the Squid\nStorage SIG repository. RL10 images already used Squid.\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/1002087\n\nChange-Id: I375f01701f84dc117a842c344b5f796b9a582131\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c933b15aad60c81e271f733c8ec3ec3e06520d7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c933b15aad60c81e271f733c8ec3ec3e06520d7b"}]},"branch":"refs/heads/stable/2025.1","description":"Edit commit message"},"737cb8152535c0531d2aeb12685c50222b335b59":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-08-25 12:42:09.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/78/1002078/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/78/1002078/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/1002078/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/1002078/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/1002078/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/78/1002078/3"}}},"commit":{"parents":[{"commit":"662eebdf470a3c75a76eb50088a65e73df64bb5b","subject":"Merge \"ironic-pxe: don\u0027t fail if symlinks already exist\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/662eebdf470a3c75a76eb50088a65e73df64bb5b"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-08-24 12:36:57.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-08-25 12:42:09.000000000","tz":0},"subject":"[2025.1-only] ceph: Bump RL9 images from Reef to Squid","message":"[2025.1-only] ceph: Bump RL9 images from Reef to Squid\n\nCeph Reef went end of life on 2025-03-20. Switch the Ceph client\npackages in Rocky Linux 9 and CentOS Stream 9 images to the Squid\nStorage SIG repository. RL10 images already used Squid.\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/1002272\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/1002265\n\nChange-Id: I375f01701f84dc117a842c344b5f796b9a582131\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/737cb8152535c0531d2aeb12685c50222b335b59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/737cb8152535c0531d2aeb12685c50222b335b59"}]},"branch":"refs/heads/stable/2025.1","description":"Edit commit message"},"4b6ae8d137c0b195ec41098b36724d06c88f0f49":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-08-25 14:22:40.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/78/1002078/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/78/1002078/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/1002078/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/1002078/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/1002078/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/78/1002078/4"}}},"commit":{"parents":[{"commit":"662eebdf470a3c75a76eb50088a65e73df64bb5b","subject":"Merge \"ironic-pxe: don\u0027t fail if symlinks already exist\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/662eebdf470a3c75a76eb50088a65e73df64bb5b"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-08-24 12:36:57.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-08-25 14:22:40.000000000","tz":0},"subject":"[2025.1-only] ceph: Bump RL9 images from Reef to Squid","message":"[2025.1-only] ceph: Bump RL9 images from Reef to Squid\n\nCeph Reef went end of life on 2025-03-20. Switch the Ceph client\npackages in Rocky Linux 9 and CentOS Stream 9 images to the Squid\nStorage SIG repository. RL10 images already used Squid.\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/1002272\n\nChange-Id: I375f01701f84dc117a842c344b5f796b9a582131\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4b6ae8d137c0b195ec41098b36724d06c88f0f49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4b6ae8d137c0b195ec41098b36724d06c88f0f49"}]},"branch":"refs/heads/stable/2025.1","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
