)]}'
{"id":"openstack%2Fkolla~1002108","triplet_id":"openstack%2Fkolla~master~Ib381c5226b0407deea11ae70188ab2f2239d08eb","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-08-24 21:07:29.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2026-08-24 19:59:58.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"Ib381c5226b0407deea11ae70188ab2f2239d08eb","subject":"CI: Don\u0027t build from Zuul overrides in upgrade jobs","status":"MERGED","created":"2026-08-24 13:11:10.000000000","updated":"2026-08-24 21:08:26.000000000","submitted":"2026-08-24 21:07:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1002108","meta_rev_id":"c0ba838500a5d4ea51c326bfcafa83dcd0d18f75","_number":1002108,"virtual_id_number":1002108,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-24 21:07:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-08-24 19:59:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-08-24 19:59:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-08-24 19:59:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-08-24 19:59:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2026-08-24 19:59:10.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-24 13:11:36.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-08-24 13:50:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"058cbde39d4602eb7c98b82062a51500b1d19ea6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-24 13:11:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"564c69b2840f9d0e9b8f29dd7d7934c8227b3a39","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-08-24 13:11:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ffaba3d3dcf0f0030cc5822864e0bd731a625bf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 13:50:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d344e05b93df4cca91c5b54844329904\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/16144c85e6a340bd8240d7cc5d73ff5d : FAILURE in 6m 56s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/b307b58bbca2445e92210d47b1823de5 : FAILURE in 6m 44s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/a1dbd495e3af45ff8e6225ffdc7c432a : FAILURE in 6m 25s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/e064a39a10bb4ca4acdd01a577931b56 : FAILURE in 9m 27s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/7d18dd4c93974d0bb6628f9be14aa9df : FAILURE in 7m 29s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9437db26cdf241a6885e778253207239 : FAILURE in 8m 06s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1f4bd641c1f347928dbae55a8a71393c : FAILURE in 7m 31s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/89753e7a8b5b4e9da9930ec7e3727016 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08804264e0f74b8a9df7304512f27b8c : SUCCESS in 4m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6a54ca6f581a4138bc2f8909164f4d31 : SUCCESS in 4m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aa3b2e04313043a98d043a2884744ee3 : SUCCESS in 5m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/65572baa57534dae9d70e41968632e7d : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4a538bfcea240a2bcc042b678cedd15 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5032726c7864218b6c396598421a2ae : SUCCESS in 8m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4625f9ef06204972b757e4d9cb1be449 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"3bf50bd25f484d9eb40e5e099525f30decebe753","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-24 13:54:42.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"6be5cbb813b422da0331c8d223801744f0350a19","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-08-24 14:00:56.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1c83c1306a34f1666fc2a9ce038a7a350813ba7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-24 14:08:37.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"83f15abb62d7e236b9ce609ff38812390b7ac7d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-24 14:08:44.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"5145a7193b1c7fb85bebc550f3a8ec1fde381831","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 14:54:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e886311137ed4b9d87e5f41daf8897eb\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/24bd9cbb40f046138fa529725eaf5f92 : SUCCESS in 25m 34s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/feb91eadf288453e9370a783b869aa70 : SUCCESS in 32m 29s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/288dab167c7f4c0aabecf219b2419170 : SUCCESS in 36m 12s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/97523c3c828a4587b502b4b125dde434 : SUCCESS in 31m 08s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/17e23207712b42dbbe079fa7fbe7391c : SUCCESS in 41m 57s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fcede6d6721d4f759214990e6f332813 : SUCCESS in 25m 46s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c0614823c93c408a8670cde3e7f90167 : SUCCESS in 30m 54s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/9fd6638bea904cd3871854626d0a3827 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/155c267a5bb24f9eb29c582c5ea5797c : SUCCESS in 5m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/03206b7f6a2e45d89f6d9e067b042567 : SUCCESS in 5m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/03986616d5504542aea1a2ce09d3467b : SUCCESS in 2m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6a0116264b7b4ef581d2e3e84ec3b6fb : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12ae7cf313b04d868dad7926cf3d1e80 : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19b62e7d1bd54c8aac04a60b573cb28d : SUCCESS in 6m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37926c1168fa491a865f57cd12da84e5 : SUCCESS in 9m 52s","accounts_in_message":[],"_revision_number":4},{"id":"37e48f17c91a1f540a3f7d190ceb08d814903b0f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-24 16:10:56.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"32851c69e9837d10294619f1cb51b9e95a015a30","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 16:59:48.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97505be7121846d2b8934fbe5a7f755e\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/8e27093db522429d8405d2fe19670a5d : SUCCESS in 44m 09s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/a332adde404b47ebbc1d1e30229bb91f : SUCCESS in 46m 21s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/f8e1d6a8ba514ca48d7b3447a745c784 : SUCCESS in 52m 27s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/40312202df244bbd9b9edd10f9755aa2 : SUCCESS in 34m 33s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8c8403c2971d464088ecdb96696ea895 : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a892e25d503b47b39e4b6c5037fab968 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b0436715c44f46ee84d05aa8a856bf42 : NODE_FAILURE Node(set) request 52e33ef7c1de4920ab75dddd166f6857 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8180e0121f0de5acf706e76071730c94b27bbf41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-08-24 18:20:48.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"607d343ad6459ad6ca0690fb44744e8b1fc9b82a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 19:06:31.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e1672adc4f34422aa14e3058b29cca6\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/4aa3b4e15519441f9c1a39e222038d90 : SUCCESS in 25m 44s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/9f772da74b9b474284f6377e44e0a02c : SUCCESS in 23m 34s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/3f9f8a7f11154d63a2f0219275d7567f : SUCCESS in 16m 19s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/db628b1583c8467b959f6e281601dfb3 : SUCCESS in 25m 41s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/baf10586ced1452097aec50bfa225cca : FAILURE in 41m 45s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ef48b78db04b4d55bb113c7e7aba8ac5 : SUCCESS in 24m 50s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/33be5d9ba3c94b03a9a5ad0589ec9910 : SUCCESS in 26m 34s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/56176924641c4f8db21f77db760ef0f7 : SUCCESS in 1m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f19a6e80ba944079e9e7eaa62ec14a9 : SUCCESS in 3m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/af2ec153ca964cbca867aa1a5876f434 : SUCCESS in 2m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6c744e8ddd414c6c88e5a5f075391154 : SUCCESS in 1m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/90c0b336f2a347cf9819a05c353491a4 : SUCCESS in 3m 15s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f494d9112a864ab98f364b10443b38be : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f16596d267844b62aaea347c654d0095 : SUCCESS in 5m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de1d0bf383694333a80e16f72ee982aa : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":5},{"id":"b073258a7a01a49a07f6d722e4992c61a01084f7","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-08-24 19:12:21.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"072b19e80e2d4a75d1b2c9e18c7e08f0f4dab959","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 19:34:30.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1b5df19daba4550813d0ac193367cfe\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/76a9087f112f4674aafc88b9f7e34a97 : SUCCESS in 49m 46s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/444c24991abb42e9bbc610e3b115d8fe : SUCCESS in 50m 08s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/401375843bb846e3a80d161314903d18 : SUCCESS in 55m 47s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/eff69fcebb8c4b91ae6313d40d452ff6 : NODE_FAILURE Node(set) request def3cc06b37e4a75b6b7eed70e558999 failed in 0s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fc71407217394d0699a1be786d2e7fd8 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d6a6b0e507c24289a6b17b165887ba9a : NODE_FAILURE Node(set) request 3a364a6a4c804fa09618aaaeee673be9 failed in 0s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e0191e57d31741cb9ecfd4d9dc5374f2 : SUCCESS in 8m 45s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0eb2099c753d517eadc178d8e925b30095225891","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-24 19:59:10.000000000","message":"Patch Set 5: Backport-Candidate-1 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7ba7f40887a12d5998249d59103033c36a25c4ee","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-08-24 19:59:58.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"0e8ca57021e65283cb6e3e64d19c4aa84d4c4003","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 20:18:38.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca928cc20aac43988dcd4e2d0ab567ae\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/f64419c465ba4758bff5d88adae0bda6 : SUCCESS in 22m 25s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/a1e7070f2061407d9d6c68da096ec95c : SUCCESS in 27m 25s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/4afdfb0831b9440ea0cd3d7782a36167 : SUCCESS in 30m 08s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/7b570e4728904cf4809bbccd13bc4b7d : SUCCESS in 31m 47s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/dbe45adabe0b4b93b9f3c1e4ccd0a64e : SUCCESS in 1h 03m 56s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/66b32705fa6d4f8aafc80b35c1f1786e : SUCCESS in 30m 37s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e5ba7b543fd6428cbc7b4cb49c5bdd9a : SUCCESS in 38m 56s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/46c664e61fb3416b99d7c5e9e25ddef8 : SUCCESS in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ad19aee25d94b5483c55aa95c98c965 : SUCCESS in 3m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5d13cd139afe4cd085ef0b5520aeecf4 : SUCCESS in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/67bba39178d2438fbac2ec04df39dde7 : SUCCESS in 3m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8226cd45fa44427591dd29fd99173023 : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e304db51e8a4c068fb99969423a4086 : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d8a164aa12a44d69574b2717e996f47 : SUCCESS in 8m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ffeddec0fac4ecda37b2231f303650e : SUCCESS in 9m 02s","accounts_in_message":[],"_revision_number":5},{"id":"9fb2e7dfce7eacbf66d8ca268c0b7cbad543d17c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 20:20:46.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"1c287f1acba34354f2e2cc912d79f9f656a4390d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 21:07:28.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4316f0ca7a814a2c9e2f64271fc68d5a\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/648b0489d4c946e68ab4339c883f0bcb : SUCCESS in 24m 46s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/fc7b53c041c84054a815b4f5385072f0 : SUCCESS in 30m 54s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/16cb50a7b64e44d398d4247ee85ebba4 : SUCCESS in 21m 53s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/d98c9f70d95c42e5a55b9190ff31fb18 : SUCCESS in 43m 29s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d755510b496448e4957ab1091d4f81a6 : SUCCESS in 18m 46s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e78024fd0eba44579b8d2b31914b9048 : SUCCESS in 41m 47s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c9831a98ebe04ff6ae0f37c91bd4f6b9 : SUCCESS in 2m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03a29006724f41a59637fd38eb364c81 : SUCCESS in 3m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4520ff64fae645089261904f47f785ae : SUCCESS in 3m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/758c217fecd34d11a3485898fa502812 : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/befe94259838486a9e35a34066c76a69 : SUCCESS in 4m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/921b32c2139e469582a3bbac3b344c8c : SUCCESS in 10m 08s","accounts_in_message":[],"_revision_number":5},{"id":"df5e940fdcdd5811f24885e0b0efe19cc4d64bde","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 21:07:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"c0ba838500a5d4ea51c326bfcafa83dcd0d18f75","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 21:08:26.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6984f6508c764b78af4e4eec42db81c7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/738f9990c4c84adfa19cbe0ad73bee5f : SUCCESS in 41s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"866f53ee7aaf8319a15c5d44625daaee3092808c","revisions":{"d579d2e7816058bbd38514bde736c9e603c11473":{"kind":"REWORK","_number":1,"created":"2026-08-24 13:11:10.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/08/1002108/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/08/1002108/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/1002108/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/1002108/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/1002108/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/08/1002108/1"}}},"commit":{"parents":[{"commit":"b1bfacdf1974fdf7d1abb03542bb10fcac272c9b","subject":"Switch sources.py into sources.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b1bfacdf1974fdf7d1abb03542bb10fcac272c9b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-08-24 13:10:07.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-08-24 13:11:07.000000000","tz":120},"subject":"CI: Don\u0027t build from Zuul overrides in upgrade jobs","message":"CI: Don\u0027t build from Zuul overrides in upgrade jobs\n\nOur upgrade jobs don\u0027t have their own set of required-projects, so we\nend up using master version of Zuul cached repos and previous-release\nu-c - which obviously breaks.\n\nChange-Id: Ib381c5226b0407deea11ae70188ab2f2239d08eb\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d579d2e7816058bbd38514bde736c9e603c11473"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d579d2e7816058bbd38514bde736c9e603c11473"}]},"branch":"refs/heads/master"},"549803f941e78ed2a3681bfb7ebb1b6a4e49f69e":{"kind":"REWORK","_number":2,"created":"2026-08-24 13:54:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/08/1002108/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/08/1002108/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/1002108/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/1002108/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/1002108/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/08/1002108/2"}}},"commit":{"parents":[{"commit":"b1bfacdf1974fdf7d1abb03542bb10fcac272c9b","subject":"Switch sources.py into sources.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b1bfacdf1974fdf7d1abb03542bb10fcac272c9b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-08-24 13:10:07.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-08-24 13:54:42.000000000","tz":0},"subject":"CI: Don\u0027t build from Zuul overrides in upgrade jobs","message":"CI: Don\u0027t build from Zuul overrides in upgrade jobs\n\nOur upgrade jobs don\u0027t have their own set of required-projects, so we\nend up using master version of Zuul cached repos and previous-release\nu-c - which obviously breaks.\n\nChange-Id: Ib381c5226b0407deea11ae70188ab2f2239d08eb\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/549803f941e78ed2a3681bfb7ebb1b6a4e49f69e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/549803f941e78ed2a3681bfb7ebb1b6a4e49f69e"}]},"branch":"refs/heads/master"},"12c43b048a5bb7f7eeae4d77bdc694a2cdc642b5":{"kind":"REWORK","_number":3,"created":"2026-08-24 14:08:37.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/08/1002108/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/08/1002108/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/1002108/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/1002108/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/1002108/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/08/1002108/3"}}},"commit":{"parents":[{"commit":"b1bfacdf1974fdf7d1abb03542bb10fcac272c9b","subject":"Switch sources.py into sources.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b1bfacdf1974fdf7d1abb03542bb10fcac272c9b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-08-24 13:10:07.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-08-24 14:08:36.000000000","tz":0},"subject":"CI: Don\u0027t build from Zuul overrides in upgrade jobs","message":"CI: Don\u0027t build from Zuul overrides in upgrade jobs\n\nOur upgrade jobs don\u0027t have their own set of required-projects, so we\nend up using master version of Zuul cached repos and previous-release\nu-c - which obviously breaks.\n\nChange-Id: Ib381c5226b0407deea11ae70188ab2f2239d08eb\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/12c43b048a5bb7f7eeae4d77bdc694a2cdc642b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/12c43b048a5bb7f7eeae4d77bdc694a2cdc642b5"}]},"branch":"refs/heads/master"},"1f49dc88057a11e6478e85e422723744985952e2":{"kind":"REWORK","_number":4,"created":"2026-08-24 14:08:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/08/1002108/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/08/1002108/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/1002108/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/1002108/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/1002108/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/08/1002108/4"}}},"commit":{"parents":[{"commit":"b1bfacdf1974fdf7d1abb03542bb10fcac272c9b","subject":"Switch sources.py into sources.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b1bfacdf1974fdf7d1abb03542bb10fcac272c9b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-08-24 13:10:07.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-08-24 14:08:42.000000000","tz":0},"subject":"CI: Don\u0027t build from Zuul overrides in upgrade jobs","message":"CI: Don\u0027t build from Zuul overrides in upgrade jobs\n\nOur upgrade jobs don\u0027t have their own set of required-projects, so we\nend up using master version of Zuul cached repos and previous-release\nu-c - which obviously breaks.\n\nChange-Id: Ib381c5226b0407deea11ae70188ab2f2239d08eb\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1f49dc88057a11e6478e85e422723744985952e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1f49dc88057a11e6478e85e422723744985952e2"}]},"branch":"refs/heads/master"},"866f53ee7aaf8319a15c5d44625daaee3092808c":{"kind":"REWORK","_number":5,"created":"2026-08-24 18:20:48.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/08/1002108/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/08/1002108/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/1002108/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/1002108/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/1002108/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/08/1002108/5"}}},"commit":{"parents":[{"commit":"b1bfacdf1974fdf7d1abb03542bb10fcac272c9b","subject":"Switch sources.py into sources.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b1bfacdf1974fdf7d1abb03542bb10fcac272c9b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-08-24 13:10:07.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-08-24 18:20:41.000000000","tz":120},"subject":"CI: Don\u0027t build from Zuul overrides in upgrade jobs","message":"CI: Don\u0027t build from Zuul overrides in upgrade jobs\n\nOur upgrade jobs don\u0027t have their own set of required-projects, so we\nend up using master version of Zuul cached repos and previous-release\nu-c - which obviously breaks.\n\nChange-Id: Ib381c5226b0407deea11ae70188ab2f2239d08eb\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/866f53ee7aaf8319a15c5d44625daaee3092808c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/866f53ee7aaf8319a15c5d44625daaee3092808c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
