)]}'
{"id":"openstack%2Fkolla~1002189","triplet_id":"openstack%2Fkolla~master~I4e01fb50ca92a12812a0ab1c63b44e1aa5d4a717","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-08-26 16:19:14.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2026-08-26 13:07:21.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I4e01fb50ca92a12812a0ab1c63b44e1aa5d4a717","subject":"CI: Post sources.yaml changes","status":"MERGED","created":"2026-08-24 18:40:03.000000000","updated":"2026-08-26 16:22:22.000000000","submitted":"2026-08-26 16:19:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1002189","meta_rev_id":"2d967a003e838e421018f5da796f662a8f96bd6c","_number":1002189,"virtual_id_number":1002189,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-26 16:19:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-08-26 07:01:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-08-25 19:11:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2026-08-26 08:48:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-08-26 07:01:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2026-08-26 08:48:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2026-08-26 07:01:57.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-24 18:41:42.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-08-24 20:52:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-25 19:11:29.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"9490c741910e5c4398ccb741660e37aba9ef60d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-24 18:40:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e4d92a85360b16e8e25ddc097d88aef12423731d","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-08-24 18:41:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"37d98b5c815236b8d0856b39ba1216d82d5686a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-24 19:17:40.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"39db503dc2cfd1ee8ceb63c203d5930a24e1651e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-24 19:17:52.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"2bc44b7364526af77c8c5ac9350e366f7b15ada0","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-08-24 19:40:17.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e38dacc033be37432f673d17c95a8a1be63ca078","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 20:52:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f7a379c8271b4817a4bbfe9276b0b9d0\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/f3f7f8440da6405a9783db83fee15815 : SUCCESS in 23m 06s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/cfbbb617e20e4431b4e481ecd47800c6 : SUCCESS in 27m 47s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/1c4d959726f34e7cb3f12c79fe00a217 : SUCCESS in 27m 56s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/1fa765dbfe1c4eecabe20a3e60581794 : FAILURE in 28m 16s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/fc2d969845144f97947d3a2e2f04eaa3 : SUCCESS in 48m 43s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ae7d6d6c539d46cbbaeb4475e57e0e7a : SUCCESS in 34m 32s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d33ce621e2634917a08778541462063a : SUCCESS in 38m 55s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/074cbc7da01d4548a5e296de78dda90a : SUCCESS in 59m 07s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/feda5ff8469e4a1daa0f1dbbd6236316 : SUCCESS in 1h 27m 15s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/f8abf8e752594944bdbc005e01b1b12d : SUCCESS in 1h 00m 39s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/d5e52681a4bd4cc5b2f0e441d29a85b1 : SUCCESS in 1h 24m 48s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f79af24f994049f8850a730316080d62 : SUCCESS in 1h 05m 56s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/1b1c49944ccf4cf2b66d66a56f7563dc : SUCCESS in 1h 31m 38s\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/bf93eba670a64feea2ae9a45c4c8e49d : SUCCESS in 21m 15s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/c3da16e9fd674564bd2c2155fa3c8c8f : SUCCESS in 24m 00s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/b73d298ee7714c7bb447c69d0c195c1b : SUCCESS in 31m 17s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/396e70395a574380870a0cd4ad42d7c5 : SUCCESS in 21m 59s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/dd4c5b8cfa6f45158ce9ec0f7223ff91 : SUCCESS in 25m 37s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/6ce5a64ec7724d4e906d2f9c4e243836 : SUCCESS in 32m 44s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/3d39360c88574714ae2e7ac387ae648e : SUCCESS in 1h 02m 25s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/f9324aafcfd9432c88288ac957529434 : SUCCESS in 37m 19s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/9404d0a3ab75466f872566896b77b407 : SUCCESS in 47m 14s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/816e7d36dceb411188563c24a1067fa7 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf78f6fb8be94041a765d7b5af379bd3 : SUCCESS in 5m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/572ee955da9b4060b5fabc46b9c7390b : SUCCESS in 2m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/55cc80839e654b8a8d2d9ef7969b9e4a : SUCCESS in 2m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/564e177f81124dd89df5f9bddcb5f7cc : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8943210492b412a95ed95b559c5d250 : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b9a8d329bc642c1a6c82446da6c0634 : SUCCESS in 7m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cfc553a95ef54f89ac9fefac19a2cae8 : SUCCESS in 8m 08s","accounts_in_message":[],"_revision_number":3},{"id":"ae095b2229a5ac5d3bc7e158e497f95e66f7aad5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-24 20:53:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6167f6ee7cdaf754dd235e7a5659798b29c3a107","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-24 21:08:42.000000000","message":"Patch Set 3: Backport-Candidate-1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1204b73a76d852838c7308c36f648a789bddbde3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 22:26:01.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9574c8a570dc470e8d1f5834edb1e855\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/60dce5f44e2547eb8f4f6405d9404887 : SUCCESS in 28m 36s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/ed6025f467724d49b35066c80796cdaf : SUCCESS in 25m 24s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/07be6514057b40c582805e8136656a58 : SUCCESS in 18m 10s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/4eb63618ce4c4608acc48c8f0a7631a8 : SUCCESS in 20m 10s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/8ec3e6d081bc4035b1d618be3a1bbdb8 : SUCCESS in 1h 11m 47s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/832e30eda25f4a9882c09fb9d0eefe40 : SUCCESS in 32m 02s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fa1e1a0da34c4bf2a9a21c364d6e6319 : SUCCESS in 20m 15s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/9fd7c3dadf724025b719bc87b03453ba : SUCCESS in 1h 03m 01s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/a917b9e70fca451bb137a69fd80cbdba : SUCCESS in 1h 28m 03s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/1bd72ea056144158a5aab556265d721d : SUCCESS in 1h 00m 12s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/528c8c9b768847f5af5c57f30a1462bd : SUCCESS in 1h 29m 12s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2592012053d425ca49b2620856c4d7e : SUCCESS in 1h 06m 23s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/36286708ea6e469da5a5d6efc222d77a : SUCCESS in 1h 26m 28s\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/c06597f76c1b4439af85af1a8190f4b5 : SUCCESS in 24m 05s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/3031e05566c04d5ebe6842bdbe425aa1 : SUCCESS in 26m 01s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/a853e4d685b84bdeb56814e11d61f1ca : SUCCESS in 30m 19s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/0944c4f698c347feba423ffd064a6b80 : SUCCESS in 37m 53s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/1732f149424c43299ac3614453e974aa : SUCCESS in 40m 08s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/904c8296b748465d8b29c0526cc31002 : SUCCESS in 28m 35s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/aa7881d3c71f476aaa6d173b2c1fd7bf : FAILURE in 4m 18s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/12fe22329aee4b34aaacafe908845f8f : SUCCESS in 40m 07s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/9cf7f474b58c4dc7ae0b27b6f13cab0f : SUCCESS in 54m 07s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/85cd8812e2c7456982bc53bc8aaf230c : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7c4739f49a34a7b9b11582bf5a21c22 : SUCCESS in 4m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8f72213c8ff5421c988d823d49ae360c : SUCCESS in 3m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ae584559ca3346ac9ab81504cbc68e2a : SUCCESS in 2m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/29f0e077ac8b452ca3826276eafcad04 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e85e0a3cfdc49d3be87e7b44ec15733 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7277f3e7e4d64b4b97614a934b54fab5 : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78e8572b7d76459cb49af15e3b14f16c : SUCCESS in 11m 46s","accounts_in_message":[],"_revision_number":3},{"id":"5bca397269d59f43104577211ef9119899f3c97c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-24 22:31:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1cac548890ce48469f8a7adc24ea92158c7d2f09","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 22:32:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95bf9c5c3292425da4faac8dd9f4ed06\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/7b59b8778829419c8d0b85e4578910b2 : SUCCESS in 1h 02m 27s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/a27762110ac5440eb54527269131a77d : SUCCESS in 41m 08s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/97b895d4dc3342fea023540973c32c8e : SUCCESS in 42m 18s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/7059a66725124e6e968cef81b35fdcb7 : SUCCESS in 1h 12m 26s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/28498e52046a4d06a5fa301bafa14c7a : FAILURE in 1h 27m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5cda914c90034355ae4613918b46ab91 : SUCCESS in 3m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3a07dc646b0c49ccb54d05ec8b7e515f : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/379bfd6f58f5472fbdf008cbc51f12fd : SUCCESS in 5m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"22f01910e540ed4d2bbcb40d9d9d8738baf64051","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 00:04:11.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8937d67a69fa42c7aa04a87277eb1865\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/c4eea9e5e0474fe18d7261a6e47592dd : SUCCESS in 20m 12s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/b3eb6d32fc254b94ae4a2791550b0737 : SUCCESS in 27m 27s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/78098c7ab0334735895767694977db71 : SUCCESS in 36m 09s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/928a6b61474747de9139570c81a08eb7 : SUCCESS in 21m 35s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/5b5c1777236e401ca33ab2f18c2b1729 : SUCCESS in 46m 10s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5013bac2dec7493e880265028f10a4e9 : SUCCESS in 31m 27s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ea6bfe2dea8f4f46b25bb77275739b1d : SUCCESS in 25m 01s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/03fe7187625d4979abf8727cc866c8c9 : SUCCESS in 1h 00m 33s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/478b2fd5c0ce422ba117aebfb6552787 : SUCCESS in 1h 27m 30s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/9e60d084619d4e4ea69a69044bf9d219 : SUCCESS in 1h 00m 42s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/f0de365de4334743935d65bb702c0c69 : SUCCESS in 1h 27m 40s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed9595c9ed9d41809ea69a07c5414bb5 : SUCCESS in 1h 04m 16s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/21b5d9ed39a84e6a848289c4a461cfa6 : SUCCESS in 1h 30m 43s\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/00e90c72315f403690902697b6c1794d : SUCCESS in 22m 30s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/803bf9e404044dd5b4bc181ea04eb409 : FAILURE in 22m 51s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/43f9493a8cd04805809821f61ed09ffb : FAILURE in 28m 22s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ca6b05d88a7f46219a7523e481b24d29 : SUCCESS in 40m 38s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/405e7cb52088435f83615632b209aadd : SUCCESS in 58m 23s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/f715dfbabe61464a9587cd9d7b765aa9 : SUCCESS in 41m 50s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/018eb1a1eb7941948939e27633eac8fc : SUCCESS in 30m 39s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/05ebe49061ca4896a76b73fe9174cea9 : SUCCESS in 18m 50s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/17dbb97a508846ed81e2e860d1a39b83 : SUCCESS in 53m 23s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4c17e9cb21f94c46b6f58f6b2bcfb19f : SUCCESS in 2m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96c11bbe3bc743c0afabcf280202d1d7 : SUCCESS in 3m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/618e94a327f44f529976005f22763389 : SUCCESS in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b07167a849eb4dae9f41f77b38b206b6 : SUCCESS in 2m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2ac9c96ad65e41288f554f74f446ec51 : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed293eb71dd74449b6b21761154caf94 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67fe2231c91f4903916c81eadb816fb8 : SUCCESS in 3m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df831a2441f348059cb189b6d8cf4f05 : SUCCESS in 10m 26s","accounts_in_message":[],"_revision_number":3},{"id":"106f6b30425db93b371da1d1c38338c09ad801d4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-25 04:32:38.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e4cd92a9c5dec91b46205023aa29ba941ddce63c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 06:11:00.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/57a3112b2c9a4e3baefb285f1960c88b\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/88efd03508a24a5b9b0f3e13dc77b97d : SUCCESS in 33m 46s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/00123433e4c04b41abc3b13ba7da729b : SUCCESS in 22m 02s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/7e2bc74c5f3f47748e4207965f482fdb : SUCCESS in 22m 16s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/0d897a7c097745c3862c395653c3cc0e : SUCCESS in 19m 56s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/5516922c4e6240c68d9a976f32903e79 : SUCCESS in 40m 25s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/394aa813bb9f47ce995646152b4b57f0 : FAILURE in 29m 24s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d83e5be9e1164dea9bb2646e40962c05 : SUCCESS in 36m 11s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/a4cf4a4e081b44bc82253265bf9e9f49 : SUCCESS in 1h 00m 03s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/7bccd4dfe6dc47db8a798c045e6e1415 : SUCCESS in 1h 33m 31s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/1a21cff69bfd4f5bbb00a21939925d2c : SUCCESS in 58m 35s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/15f8de4954ed406aa65e402054f883a1 : SUCCESS in 1h 26m 30s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bdc0f354ba6b420481fb0896561f28c9 : SUCCESS in 1h 03m 47s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/34165dcf9e294c85bc987f5275af0581 : SUCCESS in 1h 36m 19s\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/b1a1a8d6099d4e2bbdafacb57d1dda4a : SUCCESS in 22m 29s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/219afad8cf944bbcaf8432fb9b85f382 : FAILURE in 20m 13s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/65fffb87e1ab462a9d2da51035b0b4f2 : FAILURE in 25m 40s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/51724a49c67a4a01b65f836a3d95e555 : SUCCESS in 27m 17s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/65e67786ff9d44bc96eb0f81ab166ff9 : SUCCESS in 45m 59s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/cc4d456e4ca14e618d4758faf6d3dac2 : SUCCESS in 48m 31s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/670dc832a4e94738a3279b50f2879249 : SUCCESS in 31m 50s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/088934a176204f69aae4eee3b6feaf29 : SUCCESS in 46m 16s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/7a13c2d7cd114062b0fdf438c3332feb : SUCCESS in 55m 51s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/289c0895243447b5adcf725ae55a8a7e : SUCCESS in 1m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82790263fee7499c8dad38b4b87e4205 : SUCCESS in 2m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9724628a74cf4f41b8cc3a6ec1eb49c0 : SUCCESS in 3m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/459a90ab6f1940d9b640f97e904e427f : SUCCESS in 2m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/198301dbe4c6420796c669e2e2995a08 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1deecfc638ae4afba1e8fa3b0523c105 : SUCCESS in 2m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29a9ff068a274a7991f62b3c635c6b4e : SUCCESS in 6m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd3c9e1637834f419d522401662638a1 : SUCCESS in 9m 02s","accounts_in_message":[],"_revision_number":3},{"id":"a123627a822f9063aa056f630c37d8f98dd1b47a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 06:27:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45af396d03274941aec312e7148668a7\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/c819062c4f854bc9b69e4f38d25b9d1b : SUCCESS in 42m 47s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/c54facd4798f40fea9e6739acafce133 : SUCCESS in 44m 47s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/b8db7658efd5435cb2839b6399c72f76 : SUCCESS in 49m 31s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/9bfbd97b86d54a2eb2f9c24c07614ce8 : SUCCESS in 30m 45s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/a957275f32dd4a6ba2a3ed716c88796e : FAILURE in 1h 20m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a98a02bce50e4134a8a40fdaf8cf3264 : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1f6887e80aa2429b9591ebdb849fbc73 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b2233103b62d44769470d9e6fb6e7590 : NODE_FAILURE Node(set) request 2223b74359004a28ae23e7cffbcd7aae failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"14c8ed6b2e1ed72ba1134769f469319c9ab119c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-08-25 08:59:44.000000000","message":"Uploaded patch set 4: Commit message was updated.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"634365c23407aa3af4d7b38650524fdc5c9f40ea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 09:40:36.000000000","message":"Patch Set 4:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/kolla-ansible/+/1002239 is needed.","accounts_in_message":[],"_revision_number":4},{"id":"ececee7d5c4607f3689cffd837dafa6005734339","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 09:40:55.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/kolla-ansible/+/1002239 is needed.","accounts_in_message":[],"_revision_number":4},{"id":"9545490ec8617dd64e9bc7540660396c4a19122c","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-08-25 09:43:08.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8b4d4080827b6b475c881b1d1c76c6e211475004","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 10:21:23.000000000","message":"Patch Set 4:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/kolla-ansible/+/1002239 is needed.","accounts_in_message":[],"_revision_number":4},{"id":"a2d0c537c935aa0203e56cfd14fefae28bad20a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 10:21:26.000000000","message":"Patch Set 4:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/kolla-ansible/+/1002239 is needed.","accounts_in_message":[],"_revision_number":4},{"id":"90e2f4808470fe6ab966a3f0f4d199c28696eaba","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-08-25 10:21:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1aee9b91060f9b9c5ebce4945d84b7cf9b95fd02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 12:06:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04daf9145a5d41bb903d35d63f2a336b\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/eb0f82371bef4115a5e87e2d447efc2d : SUCCESS in 22m 26s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/7152019d54fd485b9c953e0ca8f7b31b : SUCCESS in 26m 52s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/e23cece46cc14a7a8ae08ddd16936e16 : SUCCESS in 29m 32s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/8988b8e28e00449583475565884bb641 : SUCCESS in 26m 25s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/8b9fc243f8524e20ab8e12c13006c9ba : SUCCESS in 1h 11m 54s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/982ad8ecf85444b9b5aec6c86fe3cd4b : SUCCESS in 31m 06s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cbba300fc2c74afe972bf7cd2552db13 : SUCCESS in 33m 32s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/f9568d9b776f4e6faab4c4f3478e3b96 : SUCCESS in 1h 12m 54s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/dda74523eea94f3089ea12f5aa2140bb : SUCCESS in 1h 41m 45s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/2016ce311bde431bb4a71933f28c9c23 : SUCCESS in 1h 09m 44s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/e29bcef96a2e43afb01b88baf823f3be : SUCCESS in 1h 36m 36s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c91f851846ba4e7283d69195acec6cd7 : SUCCESS in 1h 11m 36s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/523889394bf04cd4822a1b7e34aaecbd : SUCCESS in 1h 33m 49s\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/0cd96c43e2004d509b610affe81daae4 : SUCCESS in 28m 33s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/41769899e07f4e35ada6c1083995a50b : SUCCESS in 28m 21s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/136d23f97be04671bb453ed89ffd5888 : SUCCESS in 31m 38s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/7456f7b0bb594de3a92914b913d73255 : SUCCESS in 43m 23s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/a869d2d77b13485ba868c4430b440089 : SUCCESS in 55m 41s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/8958276a89de428bb9e60705d3084980 : SUCCESS in 34m 37s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/90214613caad4379aea44732c56f9939 : SUCCESS in 47m 37s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/989da6058790445c970347a5eb9e00b3 : SUCCESS in 42m 08s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/681ae587fab74e64a32f8281921574c0 : SUCCESS in 46m 26s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/cc58dce95d8346f6b9a37290cb36b5d8 : SUCCESS in 2m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6baac86f84d4c5795551f65475df048 : SUCCESS in 4m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/50b0683ce23446038117b48dfc208e3c : SUCCESS in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ad62fe49f2fd465aa7e79da7f787ea12 : SUCCESS in 3m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/31f0c1e200db45759e7572696451ecf6 : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9fc6804803514dab99728f2155db15ea : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56a135bd95fa4431ab6ec0f013298ffe : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf3f333492ca402d9cfd9e2f10b19fca : SUCCESS in 10m 22s","accounts_in_message":[],"_revision_number":4},{"id":"413fd7ba709dadc6385b323b70721d501bb21da7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 13:33:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/587eb66929fd4c4987954048c5024650\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/bfd19ddd0b02489c9dbec26213a080ef : SUCCESS in 32m 57s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/6eec1916abe44300be7422b4bd7dabea : SUCCESS in 32m 44s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/07e4c6e059df4f30aa8a8c7f20812925 : SUCCESS in 33m 25s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/8fb744330b8046b280220d88a7f104b6 : FAILURE in 21m 42s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/9372e2a5fa014da69825de79ff6b25f2 : FAILURE in 1h 45m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/448c6f198e744520a3002a9c8c6fffb9 : NODE_FAILURE Node(set) request c976f4c32fc64fe49b0662e9881cc7ca failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/23ca77ce833746cba24df2a1183f557c : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7913cb8093b649dbaa74675f8ef47308 : SUCCESS in 5m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c6ffba2953502f4f5be4d27072185f3f9c02208b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-08-25 19:11:29.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"60a36fcfa822516ce61c56fe191646d1db63e77c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-26 07:01:57.000000000","message":"Patch Set 4: Backport-Candidate-1 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"66894aac0dbebb0e56c079c9934cd9e9afd3010e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 07:03:18.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e48687d6d11af6c0cfa54c824d24c1e52dede777","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 08:46:18.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/24fe573f17ad47ca90db88194bc22462\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/767522539ea9471688e8ded4d296dfe9 : FAILURE in 11m 01s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/f6755bf9fe304ae38e52cb8e6bc847c7 : FAILURE in 19m 44s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/a307e99d0a51410b80196a8661d6f5fc : SUCCESS in 20m 55s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/d0b8a67712784e66a8762f817e3b5ed0 : SUCCESS in 1h 16m 48s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f7a286fb3b2462281e893b8232db531 : SUCCESS in 24m 53s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7a31dd7b0ea34221a635a931ed726b31 : SUCCESS in 39m 46s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/a4fa06d1db2f4bbaae3b1ec6ed424df3 : SUCCESS in 1h 03m 01s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/07313189f33f40189345ea3601bc6d3a : SUCCESS in 1h 30m 34s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/2a0f683b35934d04bb3dfd2a06aafbca : SUCCESS in 1h 01m 28s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/239a5f8a73234b84ae2a58c1e7ca40b1 : SUCCESS in 1h 34m 28s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c95d170908546debf46c14189492e8b : SUCCESS in 1h 06m 11s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/b5740e9ab1ad429bad89ca8d61ab9d47 : SUCCESS in 1h 31m 59s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/bfd7ede1a06d4fb3b5ba0140e8f33ed2 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32d796ad03c14ad6822ce708b3c6ac7b : SUCCESS in 3m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/80b97f5eb17b4de7aa838312fdec97bb : SUCCESS in 5m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/214fa141bf114c59b59ff54827cc7433 : SUCCESS in 2m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f43a708e47de4056bb687d773aac7f58 : SUCCESS in 6m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e9e5deabc884e60b43a2a724abb0aae : SUCCESS in 11m 22s","accounts_in_message":[],"_revision_number":4},{"id":"7003739518f3b6a14739727ed24c32a7e1c742db","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-08-26 08:48:09.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"24e1b0abb96ff51625a4c93c8439503d09439521","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 10:26:48.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b97cf536503e48e194ee7e647b108f59\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/a860871947a64e9f963b0494120d5258 : SUCCESS in 19m 14s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/b37ec644422e422db9c28f9b277f54a2 : FAILURE in 10m 23s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/8d3daa3e42244477a01c2b52fc34be8c : FAILURE in 18m 59s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/7588992aef894ea59b87df90fc2304ce : SUCCESS in 23m 59s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/9c00f97f09a649c4a52cce3d6805c243 : SUCCESS in 1h 08m 57s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7575c2f9b1f74101845aa2d55f0b5f21 : SUCCESS in 31m 01s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/31444756a226426a84ffb5898d4dff84 : SUCCESS in 36m 02s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/7c09b03f143e442e819ed82476e8fd23 : SUCCESS in 55m 29s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/a81eac44b46e4dd79495246090d5874b : SUCCESS in 1h 24m 57s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/1efcc93b58184795add5d540e23f87e8 : SUCCESS in 1h 08m 17s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/50b24be4001f41728359a306f33ab47b : SUCCESS in 1h 36m 15s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6fe4732b5b804ebbbd49f89648d639ce : SUCCESS in 1h 02m 40s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/23f083a89f7c4905b37068b45ea7e062 : SUCCESS in 1h 25m 29s\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/cb0f8a7252124efcbb304bd23b7b8e45 : SUCCESS in 21m 44s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/f178bbd47e534353ab0864a6384ec9fa : SUCCESS in 29m 38s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/bbad9701263a45efa5b88b741c46b2e6 : SUCCESS in 17m 05s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/037e4d1844774194896fd9628b094ef6 : SUCCESS in 20m 17s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/31b5667e73b649319ff33390761d5e4c : SUCCESS in 58m 19s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/8ec494eb746b4fccbcae82fd929b0c02 : SUCCESS in 47m 55s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/cf537d52f0c84e35a9658dc103f13b37 : SUCCESS in 54m 36s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/05576cd119294245b49bb2b11de14b52 : SUCCESS in 21m 41s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/95386a34d96c4bf596cc2e5407fde4d7 : SUCCESS in 28m 38s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c8585fd7a4f34b53a32e50bea9475126 : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8eeab46000f4720930c113fc57a30f4 : SUCCESS in 2m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/140da9324dc94aef93355b0d22b1d3e1 : SUCCESS in 2m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3556fa0705d5445698ee608f116c803e : SUCCESS in 3m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a54571730cf443e8aff4adb64b996fde : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/394dfbeb5a884ddeb2a0b59f8c9b46eb : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5aa18db0561142c786cfe3b107236215 : SUCCESS in 4m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1fab682fad14d28a937b9b113029db9 : SUCCESS in 10m 06s","accounts_in_message":[],"_revision_number":4},{"id":"a7f62c1bbf2358d897d4b5bf29386743c6257c5c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 11:22:25.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d81fd244897c4869bda60984444f0e0d\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/7310f4592e034b489b3b10abb92c2e96 : SUCCESS in 56m 00s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/a8b035eeaafa4b1e9c881c7bfcee5fc4 : FAILURE in 24m 43s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/41127423ed7044ccab349d9f5cdea196 : SUCCESS in 1h 07m 56s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/bc4c25cc535240b09a62c7472ea8f0dd : SUCCESS in 57m 30s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/54f375ed9fd341c885d5d4a49d9d9237 : FAILURE in 1h 17m 26s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c625ed3425514c92bec14c0ac3c7ff9f : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4ad0e10dc8704956b9f2104fda5248a5 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ce0ae6d343ff45a28496f7f502951474 : SUCCESS in 8m 32s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d29b1a6cdbfd071c8613310fd8666ef04e2ac80b","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-08-26 13:07:21.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fc4f8916b62a3024aa97e62ee814c0e32e2c0d82","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 14:21:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4e8f896e24841b89cadc1420ee87fae\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/8724dc8ddf734df288af056de92fc793 : SUCCESS in 51m 07s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/48b2e976d66d489fa83ace390a5d5acd : SUCCESS in 48m 48s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/84bab494a9a745a180e34adf35bc3cd4 : SUCCESS in 42m 03s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/9878fef62d6940298ecf6344d52f4527 : SUCCESS in 40m 19s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/91182c58d6a04e31b0b298f5ec542c1f : FAILURE in 1h 09m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/22d0c65efe8f4bc2929e353c951e2f27 : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4d3e17f63dd54c19b9871df4998e8b2c : SUCCESS in 3m 29s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b5d9954cf17a4b1886048ebae3b90b7d : SUCCESS in 8m 25s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"189f383028e2f054909c0a066bd60378212308c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 14:45:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a28dfa1696749e78664d2b6912d58ff\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/daac9e56ec244c82af323b6dc1f42379 : SUCCESS in 29m 07s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/2122f1f2216747b9bfa1ee5d33bd8206 : SUCCESS in 34m 19s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/40478a1d3ee6463db3501738fb4b0c78 : SUCCESS in 40m 26s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/cb3f0a5a5607413db38fd2e74a96d5f7 : SUCCESS in 21m 07s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/d5426803efa946b9b0d0af80bbbb83e3 : SUCCESS in 1h 18m 16s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/caac188f5f5b48fea4af817d14797877 : SUCCESS in 30m 02s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a0b28f94283740b5ac9cd66211cb7a4a : SUCCESS in 34m 41s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/bed4f450a1a947c696247091e0378698 : SUCCESS in 56m 57s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/e44c33b5b99347d8ac64d74fdca8214e : SUCCESS in 1h 29m 39s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/03a62ce617514aa6b47558aedd0af1b7 : SUCCESS in 1h 10m 47s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/b1738f6b5b9b4ae2a547bf84e2b64664 : SUCCESS in 1h 26m 51s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4cf283b0662f404f85a4230a109a35e3 : SUCCESS in 1h 00m 46s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/fa0bf41dcaa6489cbc5f46680aa55dc7 : SUCCESS in 1h 35m 01s\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/6051889f50de4391972e7a23728a084a : SUCCESS in 25m 19s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/f602108a21df4593bf9eb2be3dc1f995 : SUCCESS in 27m 12s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/8efd22e69a2e4f3e95f2ba3382eac84f : SUCCESS in 29m 55s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/afbaa7cf77814eeb85674ef16e3d7493 : SUCCESS in 40m 24s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/905c80d921464a0b8e288cbffe0b5854 : SUCCESS in 27m 28s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/db7c914cc38a4a808d876c2d3cd7133a : SUCCESS in 24m 34s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/b66106b612d04af9a3408b3089c1d723 : SUCCESS in 1h 02m 04s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/a62af80361134c449c0766b336ce3dea : SUCCESS in 33m 37s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/2b54b55fc7a34c2bbf7baeec4badd536 : SUCCESS in 58m 52s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/3118be48960c415da4de2d5482a2562d : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/626dcfd444ff4bd89dc778bea2e3408e : SUCCESS in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4b88f6a2a6e64b0695b8d237c41b56f6 : SUCCESS in 3m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3961b721cadd48ac9dbadecbd0e5ca53 : SUCCESS in 3m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5560434801754336913381c8f88e8cdb : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42ed2c5ebd8943b6be4b91b560d428cc : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4368890b9afc4f7e9d65eb7b6649f7cf : SUCCESS in 8m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74bd8a33d8a248a0a0ec171c4e5413ca : SUCCESS in 9m 54s","accounts_in_message":[],"_revision_number":4},{"id":"fa308d82331fad29382fd678dcd6973a725d915e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 14:46:37.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"517c74e7af4a5c2f45f8262106588524b7784f90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 16:19:14.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e830d7eb2514b37a774f60573da1af0\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/246ddb4a85f547faa37a7a5ca487eec8 : SUCCESS in 26m 27s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/4cf01cf20d5f4d3aa2208cfb3ceccdbc : SUCCESS in 31m 03s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/388bf7bc9909451493463b4573bdaf79 : SUCCESS in 22m 49s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/c8861e414f6341a596b26fd4848a2afa : SUCCESS in 58m 45s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e1cd011dc864329a74b59ee29e4d60c : SUCCESS in 31m 46s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f8e0dba0427846ffab78a1c0fa581dbb : SUCCESS in 36m 10s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/ebeebb00e7a74ae2bca893ccad348b0a : SUCCESS in 59m 37s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/43af0cd875f84e9e8256a142523305b2 : SUCCESS in 1h 31m 50s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/0502d4e0a0954c20978a9a8398534a7b : SUCCESS in 1h 02m 26s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/2d30974849eb4f4b9fdcd40b50f4b06c : SUCCESS in 1h 24m 51s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2c8410ec28284ec6afb78af3ba5b02e8 : SUCCESS in 1h 02m 38s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/f24f1ecb0bf64ca88c6237a36a62e88b : SUCCESS in 1h 29m 36s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/14664c74f3a94dd0863cdd98b7b109b7 : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f301964d0fd145a3acce3e1aebf39a23 : SUCCESS in 3m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d3d7798ed27b409b9d3f77cc59426579 : SUCCESS in 3m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3f198a553e8e497984b1830205b3cb91 : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4c3953db5234a6e8553b0dd50c01b07 : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a1a5a6cb87a4f7e9a7f5b7ac4f98e80 : SUCCESS in 11m 31s","accounts_in_message":[],"_revision_number":4},{"id":"63ead671bab2d88d758a5102421508c79630d228","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 16:19:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"2d967a003e838e421018f5da796f662a8f96bd6c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 16:22:22.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/284b8900bc0448cca6a9b0b1e9a86b82\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b6f849fdff34c2a9c4c0817ef20d6b5 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"81f896141426f66de522ba3af07e17bda6e18792","revisions":{"4190f60d8aeee7192051f113bc331315f089f4a4":{"kind":"REWORK","_number":1,"created":"2026-08-24 18:40:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/89/1002189/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/89/1002189/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/1002189/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/1002189/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/1002189/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/89/1002189/1"}}},"commit":{"parents":[{"commit":"b1bfacdf1974fdf7d1abb03542bb10fcac272c9b","subject":"Switch sources.py into sources.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b1bfacdf1974fdf7d1abb03542bb10fcac272c9b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-08-24 18:38:41.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-08-24 18:40:00.000000000","tz":120},"subject":"CI: Post sources.yaml changes","message":"CI: Post sources.yaml changes\n\nAdd sources.yaml trigger on aio/mariadb/prometheus jobs\nMove out yamllint header from sources.yaml to yamllint config,\nbecause the kolla-check-sources.py does not ensure the header is\nthere and it\u0027s easier to fix it in this repo.\n\nChange-Id: I4e01fb50ca92a12812a0ab1c63b44e1aa5d4a717\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4190f60d8aeee7192051f113bc331315f089f4a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4190f60d8aeee7192051f113bc331315f089f4a4"}]},"branch":"refs/heads/master"},"81709e7ec3e2094b731587d395c528f879b87c6d":{"kind":"REWORK","_number":2,"created":"2026-08-24 19:17:40.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/89/1002189/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/89/1002189/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/1002189/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/1002189/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/1002189/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/89/1002189/2"}}},"commit":{"parents":[{"commit":"b1bfacdf1974fdf7d1abb03542bb10fcac272c9b","subject":"Switch sources.py into sources.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b1bfacdf1974fdf7d1abb03542bb10fcac272c9b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-08-24 18:38:41.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-08-24 19:17:40.000000000","tz":0},"subject":"CI: Post sources.yaml changes","message":"CI: Post sources.yaml changes\n\nAdd sources.yaml trigger on aio/mariadb/prometheus jobs\nMove out yamllint header from sources.yaml to yamllint config,\nbecause the kolla-check-sources.py does not ensure the header is\nthere and it\u0027s easier to fix it in this repo.\n\nChange-Id: I4e01fb50ca92a12812a0ab1c63b44e1aa5d4a717\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/81709e7ec3e2094b731587d395c528f879b87c6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/81709e7ec3e2094b731587d395c528f879b87c6d"}]},"branch":"refs/heads/master"},"6ef8404ad295e89703b5136ea2e83ffb889df996":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-08-24 19:17:52.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/89/1002189/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/89/1002189/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/1002189/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/1002189/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/1002189/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/89/1002189/3"}}},"commit":{"parents":[{"commit":"866f53ee7aaf8319a15c5d44625daaee3092808c","subject":"CI: Don\u0027t build from Zuul overrides in upgrade jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/866f53ee7aaf8319a15c5d44625daaee3092808c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-08-24 18:38:41.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-08-24 19:17:52.000000000","tz":0},"subject":"CI: Post sources.yaml changes","message":"CI: Post sources.yaml changes\n\nAdd sources.yaml trigger on aio/mariadb/prometheus jobs\nMove out yamllint header from sources.yaml to yamllint config,\nbecause the kolla-check-sources.py does not ensure the header is\nthere and it\u0027s easier to fix it in this repo.\n\nChange-Id: I4e01fb50ca92a12812a0ab1c63b44e1aa5d4a717\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6ef8404ad295e89703b5136ea2e83ffb889df996"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6ef8404ad295e89703b5136ea2e83ffb889df996"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"b1bfacdf1974fdf7d1abb03542bb10fcac272c9b","ours":"81709e7ec3e2094b731587d395c528f879b87c6d","theirs":"866f53ee7aaf8319a15c5d44625daaee3092808c","merge_strategy":"recursive","contains_conflicts":false}},"81f896141426f66de522ba3af07e17bda6e18792":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-08-25 08:59:44.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/89/1002189/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/89/1002189/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/1002189/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/1002189/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/1002189/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/89/1002189/4"}}},"commit":{"parents":[{"commit":"866f53ee7aaf8319a15c5d44625daaee3092808c","subject":"CI: Don\u0027t build from Zuul overrides in upgrade jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/866f53ee7aaf8319a15c5d44625daaee3092808c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-08-24 18:38:41.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-08-25 08:59:09.000000000","tz":120},"subject":"CI: Post sources.yaml changes","message":"CI: Post sources.yaml changes\n\nAdd sources.yaml trigger on aio/mariadb/prometheus jobs\nMove out yamllint header from sources.yaml to yamllint config,\nbecause the kolla-check-sources.py does not ensure the header is\nthere and it\u0027s easier to fix it in this repo.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/1002239\n\nChange-Id: I4e01fb50ca92a12812a0ab1c63b44e1aa5d4a717\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/81f896141426f66de522ba3af07e17bda6e18792"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/81f896141426f66de522ba3af07e17bda6e18792"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
