)]}'
{"id":"openstack%2Fkolla~1002489","triplet_id":"openstack%2Fkolla~stable%2F2026.1~If19d55a88975e6717eec2b9cac9fd28b050028e9","project":"openstack/kolla","branch":"stable/2026.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-08-31 16:59:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If19d55a88975e6717eec2b9cac9fd28b050028e9","subject":"stable-only: Update Prometheus to version 3.5.5","status":"MERGED","created":"2026-08-26 13:12:11.000000000","updated":"2026-08-31 17:02:12.000000000","submitted":"2026-08-31 16:59:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1002489","meta_rev_id":"6c9009a8bb969c3b01d984895a2672dc809815a9","_number":1002489,"virtual_id_number":1002489,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":38857,"name":"Kurt Bendl","display_name":"kbendl","email":"kbendl@tool.net","username":"kbendl"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-31 16:59:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"recommended":{"_account_id":38857,"name":"Kurt Bendl","display_name":"kbendl","email":"kbendl@tool.net","username":"kbendl"},"all":[{"value":1,"date":"2026-08-26 18:12:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2026-08-26 13:14:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38857,"name":"Kurt Bendl","display_name":"kbendl","email":"kbendl@tool.net","username":"kbendl"},{"value":2,"date":"2026-08-31 15:22:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":38857,"name":"Kurt Bendl","display_name":"kbendl","email":"kbendl@tool.net","username":"kbendl"},{"value":1,"date":"2026-08-31 15:22:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":38857,"name":"Kurt Bendl","display_name":"kbendl","email":"kbendl@tool.net","username":"kbendl"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":38857,"name":"Kurt Bendl","display_name":"kbendl","email":"kbendl@tool.net","username":"kbendl"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":38857,"name":"Kurt Bendl","display_name":"kbendl","email":"kbendl@tool.net","username":"kbendl"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-26 13:14:55.000000000","updated_by":{"_account_id":38857,"name":"Kurt Bendl","display_name":"kbendl","email":"kbendl@tool.net","username":"kbendl"},"reviewer":{"_account_id":38857,"name":"Kurt Bendl","display_name":"kbendl","email":"kbendl@tool.net","username":"kbendl"},"state":"REVIEWER"},{"updated":"2026-08-26 14:44:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-26 18:12:21.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-08-31 15:22:00.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"c66cb99fa1872852a78bb8c942f3c97eb46a71bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-08-26 13:12:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bf1e6a3c50f909fa6d4d208ece19a0c41cdd111a","author":{"_account_id":38857,"name":"Kurt Bendl","display_name":"kbendl","email":"kbendl@tool.net","username":"kbendl"},"date":"2026-08-26 13:14:55.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6c543785a6ff976c0b5eb99c8730e3b17f1c46bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 14:44:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf03a74b51bc4fdbab5750cf129d3c67\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/f0c03d8a056d4aecab188aecafe66b7d : SUCCESS in 1h 20m 14s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/62b945488d7c4bdb80640ad0de5b5ac6 : SUCCESS in 1h 09m 02s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/bcc756c9875147ceb564bb6dfb83fe04 : SUCCESS in 31m 16s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/9920e055d322408d872c4cc7630385ed : SUCCESS in 1h 22m 02s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/afbc7022864c42478c72ebaff916adcd : SUCCESS in 32m 34s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8693b74179e1441a859439b3363240ba : SUCCESS in 1h 16m 01s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3017690ad928410b8b1bb3f37b801761 : SUCCESS in 17m 13s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/133c6cb09bf543deb9ad63c265a655b0 : SUCCESS in 2m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/533032dc48444351be9671e0ee14a1c1 : SUCCESS in 4m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ed8a56efcce94249875853b07b1e2b9a : SUCCESS in 2m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/70f82054c2d448ae9a27415d8e573439 : SUCCESS in 3m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0ae8eb3b22804f02b2c68e86dfb7297b : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7cb0d55441244e29857d9d59fa9097d : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cacdf65abbd34ad09b0d079547046d8b : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50542886d82f42149e587bdf0a13dae9 : SUCCESS in 10m 52s","accounts_in_message":[],"_revision_number":1},{"id":"bad8dca24262085c1525abe12046e86640b2d864","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 16:47:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7e18aa94ef04023a48ee51c31006288\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/7df833380cf34bd1b085167b89679bce : SUCCESS in 1h 54m 50s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/a6556fde622c4bc8b4e9b0ad538da00a : SUCCESS in 2h 14m 54s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/0ffcbe84a2694f1f9212f98977d12f4c : SUCCESS in 2h 29m 05s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/04f0fb343895464185cfe8c731c80266 : SUCCESS in 2h 47m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"05a480c8703062639a6bfdced5f3e8065a146133","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-08-26 18:12:21.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ab68f7c646b165b0f5d92516600464e291eeaf76","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-08-31 15:22:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6838fa3d41c74df31c24526c5c65c2254c8554c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 15:22:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8e44582a3010b7bce508927418ac26cb5ec9bffe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 16:59:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3738269321904b479eda57852afde84c\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/cccdacc3e1434556b0082a759a18b440 : SUCCESS in 1h 32m 44s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/406aa99b89b347d1ba378ec927dade03 : SUCCESS in 20m 45s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/f2869922ad2a4b07b7cdd36b2f90cbd7 : SUCCESS in 1h 10m 35s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/ee8286845257443b9381ef6e64c50475 : SUCCESS in 46m 11s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c8aad3df1978429d81198db1e1d86352 : SUCCESS in 54m 59s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/18041366d9224cac896e3c5a3dd2d011 : SUCCESS in 26m 35s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/27784bd7ccac4ddb9bf0aa5adabc2a75 : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e463de2e422748eaa7b9c14821e69c46 : SUCCESS in 4m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/05ce8fa7cb724706a16f02ee6ea5fe1f : SUCCESS in 2m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/907705f93957496cbc8e72d1563dc955 : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6730b770edba4f4b9df68e1616dcc61e : SUCCESS in 6m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0a9b211f3a847d6bd490b2d236fe15a : SUCCESS in 11m 55s","accounts_in_message":[],"_revision_number":1},{"id":"69607fcfad5155756574334f614d82fd5b1cdd01","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 16:59:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6c9009a8bb969c3b01d984895a2672dc809815a9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 17:02:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b829ae83b8447ff926483d3470716df\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d983a01eceb492e81939ec2a0ca6da7 : SUCCESS in 1m 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc99dc1fc6484ca7901f93ec66f8f5af : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c4e23082aaf2329ffd1b1ddb3992a881aa53af3f","revisions":{"c4e23082aaf2329ffd1b1ddb3992a881aa53af3f":{"kind":"REWORK","_number":1,"created":"2026-08-26 13:12:11.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/89/1002489/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/89/1002489/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/1002489/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/1002489/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/1002489/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/89/1002489/1"}}},"commit":{"parents":[{"commit":"e40da0d4a7a73212cb16698a12eaeb5799cc55c7","subject":"ironic-pxe: don\u0027t fail if symlinks already exist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e40da0d4a7a73212cb16698a12eaeb5799cc55c7"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-08-26 13:11:34.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-08-26 13:11:34.000000000","tz":120},"subject":"stable-only: Update Prometheus to version 3.5.5","message":"stable-only: Update Prometheus to version 3.5.5\n\nChange-Id: If19d55a88975e6717eec2b9cac9fd28b050028e9\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c4e23082aaf2329ffd1b1ddb3992a881aa53af3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c4e23082aaf2329ffd1b1ddb3992a881aa53af3f"}]},"branch":"refs/heads/stable/2026.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":38857,"name":"Kurt Bendl","display_name":"kbendl","email":"kbendl@tool.net","username":"kbendl"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
