)]}'
{"id":"openstack%2Fkolla~1002512","triplet_id":"openstack%2Fkolla~stable%2F2025.1~Ia06f374aee4932e36d13c4944a0f7fd6861e1341","project":"openstack/kolla","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2026-08-26 19:24:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia06f374aee4932e36d13c4944a0f7fd6861e1341","subject":"Don\u0027t build releaser image on CentOS Stream 9","status":"MERGED","created":"2026-08-26 14:37:19.000000000","updated":"2026-08-26 19:26:59.000000000","submitted":"2026-08-26 19:24:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"1002512","meta_rev_id":"357a80038d5c447528b8c2413e96713a25fac8f2","_number":1002512,"virtual_id_number":1002512,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-26 19:24:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2026-08-26 18:14:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2026-08-26 15:49:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2026-08-26 15:49:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-26 17:38:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-26 18:14:41.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"6d2a746b7a261710f9eb2f5dfead36ab92f6674d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-08-26 14:37:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55b561179e0b0c44b45b26d0de6aea9642f02346","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-08-26 14:37:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a7226c0dfe46e29b07f00fec7f07c613dfd82bde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-08-26 15:48:17.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"9699542d0ce2c407d2afdd3833a1d5a6e582c389","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-08-26 15:49:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fdf2b1dffa18371849ee69540f8da44a5fcbf4fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 17:38:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e2a4f26bf9342468b8b6e0bc96989c2\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/cb076158ff7245eaa1c5f915187221f1 : SUCCESS in 46m 57s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/ad52e303efd54e41bf5b3b7c9dc8cfc9 : SUCCESS in 52m 36s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/e73ea3c0353741f9ba67c23c6a4f80d4 : SUCCESS in 1h 32m 40s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/fb6429367c0a4727a38451b9f726c925 : SUCCESS in 48m 01s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/fef8264bca3b498ea7aececb45c4be76 : SUCCESS in 37m 28s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/1e817d05dfc64135a979cc9ef8af2dc5 : SUCCESS in 16m 28s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c20fc2f10e8f41c5917c75816620b79b : SUCCESS in 54m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d427858196f4a31960e451398f66453 : SUCCESS in 3m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92b4acdcc66244e9a5f336719768af96 : SUCCESS in 3m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3a37908b6af441dc9cd5be9d76228ee4 : SUCCESS in 2m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a2301eb2ed24d7698ec1566b25eac39 : SUCCESS in 2m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10c1d770b67c4c94a1d03739fb376165 : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9df882655b745d0b8f75e1d7698a611 : SUCCESS in 11m 24s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/670495ef21e84a64874bcb55eddc6622 : SUCCESS in 2m 40s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/60a46203faab4b7eb22bff1124cb097a : SUCCESS in 44m 57s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c6cb813531b6435aa78d81b33911022e : SUCCESS in 1h 44m 20s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ae9723f9667843c699c0b6e27e4476b2 : SUCCESS in 1h 42m 57s\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/b32fe82015a9498fba087e483cfd018c : SUCCESS in 43m 15s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/736da2439aa6457a985996ad3dcc8a8f : SUCCESS in 56m 47s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/98aa4d1a798a4177a7a8d65901ac6103 : SUCCESS in 48m 02s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5435d4ca266e45a5bf776b978d5a6987 : SUCCESS in 58m 50s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/32a1c1ee9cc748c0b4ba587455c0705e : SUCCESS in 29m 41s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/53637e3901a14bc1a719474f5d99361c : SUCCESS in 59m 37s","accounts_in_message":[],"_revision_number":2},{"id":"a13f79b1717fcadffb75b4fb2f1d865b4848cd50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 17:38:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a6d64b491283c0764527498c4324b0a4aea80e5f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-08-26 18:14:41.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"98ff5b0c2e1228e840419a628ce34139dd4bce41","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-08-26 18:15:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f06ec54b758147bceccfe513819078d4862ce3c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 19:24:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/843b17db3d3c451999cc4fd0482901c9\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/33f04f45983a4e6587b8bcd19548ac2b : SUCCESS in 42m 41s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/fabf4474fd344a499dff53ab4aac548e : SUCCESS in 17m 03s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3988f186283a427fb6d8c92ee5aa8918 : SUCCESS in 53m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b746b8e0d1743db8ed680587a8ae18e : SUCCESS in 2m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad136a9c784640d6825345fbbcc454f7 : SUCCESS in 3m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dfc26a53075544afb0cce34a8b44a860 : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9184439dbb734bd783a2c82d889ac365 : SUCCESS in 8m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70b3c87aa4d742a7a1047381314bd025 : SUCCESS in 11m 05s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c9e26f7fab334d15aebc05ca35c2cba0 : SUCCESS in 3m 13s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/64868b09c69741b69038367277cf9945 : SUCCESS in 57m 09s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/52f737c726d84f1cb482ec6aa275c6d0 : SUCCESS in 1h 10m 24s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/3e67b32a19424386abdd59334d9ce621 : SUCCESS in 1h 06m 49s\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/f4b45a2e9cbc439caf1a412833618994 : SUCCESS in 52m 11s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/a93d464934a24c1e9151d176525cfb48 : SUCCESS in 43m 40s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/7594e17f7a31445582809f8d33f52d98 : SUCCESS in 1h 32m 22s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5c7d5a534dbe4a168f64b3ae343fce1c : SUCCESS in 1h 13m 19s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/3fcc79372928442f9c261fcb0e9ac193 : SUCCESS in 32m 32s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a021799569874e11ac3cae4f6d4937fd : SUCCESS in 1h 00m 37s","accounts_in_message":[],"_revision_number":2},{"id":"b2ff8228909b79d62f5991c6e0ba4aeb22cdfd5e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 19:24:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4518d96fffd52975a379af9b5e5f5c818304b85a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 19:25:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b73d964b6844a618298b1bc157f10c6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/594556a958184ac293b60b65c4bf6f22 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2},{"id":"357a80038d5c447528b8c2413e96713a25fac8f2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 19:26:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9dcf1e5aae4344348015102e19b3ba23\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/34740f03d2284082b9e82cc18911c6a5 : SUCCESS in 57m 44s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/f13a7a734d104a7da4baa4c6efcdf018 : SUCCESS in 2h 23m 01s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6470a3cec0614af79d855e292bedbbfc : SUCCESS in 1h 25m 32s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f9769d8c67a64499980f816b03b541a0 : SUCCESS in 1h 11m 11s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/5686cbc4c27b435780024c7da0afa8db : SUCCESS in 1h 36m 03s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/a96b1f9b5311408494565b537d240451 : SUCCESS in 1h 25m 03s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/dd5872c4eff64bcc80d8ea3605d3c822 : SUCCESS in 2h 06m 04s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"59814ad828c4e40d77463e242769d8df5edd6342","revisions":{"d65440f67562c5f3c6dcd2d199c26980cdfa8ac0":{"kind":"REWORK","_number":1,"created":"2026-08-26 14:37:19.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/12/1002512/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/1002512/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/1002512/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/1002512/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/1002512/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/1002512/1"}}},"commit":{"parents":[{"commit":"662eebdf470a3c75a76eb50088a65e73df64bb5b","subject":"Merge \"ironic-pxe: don\u0027t fail if symlinks already exist\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/662eebdf470a3c75a76eb50088a65e73df64bb5b"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-08-26 14:37:05.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-08-26 14:37:18.000000000","tz":120},"subject":"Don\u0027t build releaser image on CentOS Stream 9","message":"Don\u0027t build releaser image on CentOS Stream 9\n\nThe releases repo dropped support for Python 3.9, so let\u0027s just stop\nbuilding the image there.\n\nThis is a follow up to I70e244cb289202080a78066f899b9eb24faa5cdc, which\ndid the same for Rocky 9.\n\nChange-Id: Ia06f374aee4932e36d13c4944a0f7fd6861e1341\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d65440f67562c5f3c6dcd2d199c26980cdfa8ac0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d65440f67562c5f3c6dcd2d199c26980cdfa8ac0"}]},"branch":"refs/heads/stable/2025.1"},"59814ad828c4e40d77463e242769d8df5edd6342":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-08-26 15:48:17.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/12/1002512/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/1002512/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/1002512/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/1002512/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/1002512/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/1002512/2"}}},"commit":{"parents":[{"commit":"662eebdf470a3c75a76eb50088a65e73df64bb5b","subject":"Merge \"ironic-pxe: don\u0027t fail if symlinks already exist\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/662eebdf470a3c75a76eb50088a65e73df64bb5b"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-08-26 14:37:05.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-08-26 15:48:17.000000000","tz":0},"subject":"Don\u0027t build releaser image on CentOS Stream 9","message":"Don\u0027t build releaser image on CentOS Stream 9\n\nThe releases repo dropped support for Python 3.9, so let\u0027s just \nstop building the image there.\n\nThis is a follow up to I70e244cb289202080a78066f899b9eb24faa5cdc,\nwhich did the same for Rocky 9.\n\nChange-Id: Ia06f374aee4932e36d13c4944a0f7fd6861e1341\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/59814ad828c4e40d77463e242769d8df5edd6342"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/59814ad828c4e40d77463e242769d8df5edd6342"}]},"branch":"refs/heads/stable/2025.1","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
