)]}'
{"id":"openstack%2Fkolla~1002594","triplet_id":"openstack%2Fkolla~master~Iddb87bf62891ceaa6cdf481245e8944d404377d9","project":"openstack/kolla","branch":"master","topic":"test-ovn-repos","attention_set":{},"removed_from_attention_set":{"35264":{"account":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"last_update":"2026-08-27 10:36:58.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iddb87bf62891ceaa6cdf481245e8944d404377d9","subject":"DNM: test openvswitch repo override in CI","status":"ABANDONED","created":"2026-08-27 09:05:35.000000000","updated":"2026-08-27 10:36:58.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f53879a60e64d0c2b852b2bb7e26f79767b0f641","_number":1002594,"virtual_id_number":1002594,"owner":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-08-27 10:10:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-27 10:10:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"27b1975d55104440eafb1456f405e6d93b0fffb2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2026-08-27 09:05:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eac05411f62953c33fbb1ef048a2e8a726a55221","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 10:10:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7334fb0ff587404aa2b88fa7d9f42cad\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/0f96864db0e948a1a0a2dacb4a95173c : SUCCESS in 19m 55s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/4a77be00421e462aa4ac906332c43678 : SUCCESS in 28m 24s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/e257fbdce6d2421eb825da1b768e7343 : SUCCESS in 29m 43s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/93c8f7055d2548c09872fe9786fe96aa : FAILURE in 21m 58s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/48daaa9726434a6bb21d98f9ccd24410 : FAILURE in 1h 01m 41s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/86db5d71aa6549a79842a9e443a5b1d9 : SUCCESS in 28m 10s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c18cad39a76c47d6bb3387c26081f407 : SUCCESS in 37m 46s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d4ceba8c36db4c06afaef691f11a9181 : SUCCESS in 2m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ff1452f9d444042a1cc78816e6d7dda : SUCCESS in 2m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/89b20a289c5a42199ee366f57e862aa9 : SUCCESS in 1m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7dc045fa209244eb85a0822441c6015f : SUCCESS in 2m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e89df412f7474d1db782c34a9b96ab0b : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e776d99cd0548559d2641254fe0fa41 : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/694091f8c5344cba934f28752ae9eb43 : SUCCESS in 3m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76c94d227f434810b7db4dd836aae517 : SUCCESS in 11m 20s","accounts_in_message":[],"_revision_number":1},{"id":"f53879a60e64d0c2b852b2bb7e26f79767b0f641","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2026-08-27 10:36:58.000000000","message":"Abandoned\n\nCI test complete","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e1fbe82e18d456fa11a520f7adfa31ade9430254","revisions":{"e1fbe82e18d456fa11a520f7adfa31ade9430254":{"kind":"REWORK","_number":1,"created":"2026-08-27 09:05:35.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/94/1002594/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/94/1002594/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/1002594/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/1002594/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/1002594/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/94/1002594/1"}}},"commit":{"parents":[{"commit":"866f53ee7aaf8319a15c5d44625daaee3092808c","subject":"CI: Don\u0027t build from Zuul overrides in upgrade jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/866f53ee7aaf8319a15c5d44625daaee3092808c"}]}],"author":{"name":"Alex Welsh","email":"alex@stackhpc.com","date":"2026-08-27 08:55:37.000000000","tz":60},"committer":{"name":"Alex Welsh","email":"alex@stackhpc.com","date":"2026-08-27 09:05:27.000000000","tz":60},"subject":"DNM: test openvswitch repo override in CI","message":"DNM: test openvswitch repo override in CI\n\nUsing this change to trigger CI and see if it \"correctly\" fails before I\npatch it.\n\nSee also\n[1] https://bugs.launchpad.net/kolla/+bug/2163314\n[2] https://review.opendev.org/c/openstack/kolla/+/1000710\n\nChange-Id: Iddb87bf62891ceaa6cdf481245e8944d404377d9\nSigned-off-by: Alex Welsh \u003calex@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e1fbe82e18d456fa11a520f7adfa31ade9430254"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e1fbe82e18d456fa11a520f7adfa31ade9430254"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
