)]}'
{"id":"openstack%2Fkolla~1002718","triplet_id":"openstack%2Fkolla~master~I5b1271480c4c3c17e202a81379289730f39ba567","project":"openstack/kolla","branch":"master","topic":"openstack/kolla/check-sources","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-09-01 07:17:54.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"I5b1271480c4c3c17e202a81379289730f39ba567","subject":"Updated from check-sources","status":"MERGED","created":"2026-08-28 02:50:32.000000000","updated":"2026-09-01 09:59:54.000000000","submitted":"2026-09-01 09:58:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1002718-openstack/kolla/check-sources","meta_rev_id":"984d18ab065ed636c1800f17c9c43bfbada3610f","_number":1002718,"virtual_id_number":1002718,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-01 09:58:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2026-09-01 07:17:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-01 07:52:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-01 07:52:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2026-09-01 07:52:50.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-28 04:53:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-28 13:36:03.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-08-31 07:53:07.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2026-09-01 07:52:50.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"fec9332a0b35645bcdec61112ab7db0b3d5461b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-08-28 02:50:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32b4f55bbb64580783d45c862698b827903ecd2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 04:53:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ffe5fe724e94e45bcaf0a2ceb877e4c\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/6ef11d8c9c07436283154d0f2b2c0b9f : SUCCESS in 32m 53s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/293ccd1d290445568423e6f7ae7c5c64 : SUCCESS in 31m 26s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/80b027d7c8cf4c05a185cf2a9ec3b0e7 : SUCCESS in 30m 36s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/3550cf51e4b84a688629623369d2eea8 : SUCCESS in 35m 04s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/019365ffe54a4a0ba5c1c20d249e0edd : SUCCESS in 1h 13m 07s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cbd88d6797f2450e870882a4d38aa54a : SUCCESS in 31m 10s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fd6dfd471df24c14a738a1482a32f6dc : SUCCESS in 32m 46s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/fd37bdd1e79d4b178ad4d4a999585dc8 : SUCCESS in 1h 11m 28s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/81069fb0b4284cbd9437a07652cccafc : SUCCESS in 1h 39m 11s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/69f492eba8764c1bbb6733f8380448b4 : SUCCESS in 1h 06m 24s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/7e4427620954424187d585cd500ff52a : SUCCESS in 1h 32m 39s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a0dbbc6dda741549c06bd982e9ed028 : SUCCESS in 1h 26m 45s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/61d654260da44980bc56654ae57b8b3f : SUCCESS in 1h 47m 45s\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/5eb2b4f593734744acf27788feb71cce : SUCCESS in 34m 56s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/1d98de638c17499980062ab818adc719 : SUCCESS in 26m 25s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/229a6d3eaf0d4040b492e0bd727b6f03 : SUCCESS in 29m 01s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/a0aa6b4a7bb8431a9c2b4e32409ae3a6 : SUCCESS in 38m 47s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/9a338153f6174a8a8cf0c5821545873b : SUCCESS in 26m 39s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/faf00565c05f45aa8bdb5992a55de152 : SUCCESS in 20m 45s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/da83231f4cea4a76acd5094526150a40 : SUCCESS in 31m 14s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/20dc39ac1b854df1946b9ae57ed83aa5 : SUCCESS in 38m 26s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/70f6c8dff4e2405ab4a3d29ee0cbd04c : SUCCESS in 58m 08s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d020637846354c9888301fdd6b183451 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e5f398f76a744e68f644bffbe772b32 : SUCCESS in 4m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1e97594beb304b35b5643413b81974c2 : SUCCESS in 4m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3928c17230284f8daa47ef027bd99632 : SUCCESS in 3m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1edc938fffc242618999d5a7b08ef3c8 : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a3c933a0c9445ec9b346e98a81d0e2a : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d1becd0e2ba4c0d8ed3669d6387cf23 : SUCCESS in 5m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d036fc783854888bb4c0b8b720d17bf : SUCCESS in 10m 11s","accounts_in_message":[],"_revision_number":1},{"id":"eb22f824fa9b2dcea056015c01f5320fe9ba6542","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 06:56:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02a192e74a2b4614a8c76e1a05d395d2\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/a2a0943676134627ac63d06a1af826ac : SUCCESS in 52m 46s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/0883d7c5c9f946c5b2fa4ff8f1f3bcf2 : SUCCESS in 37m 53s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/f819c5df09e14248ba24d4364933b72c : SUCCESS in 47m 53s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/f7013a1f54ee4cc2be104ee33a3270b2 : SUCCESS in 41m 46s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/3d68263b73694daa9f4798a5ba5a6a1b : FAILURE in 1h 35m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/eb83ef368db845a3b533d0ea58f6d7d1 : SUCCESS in 3m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/27ba10ce0517465985fb578c47d029b5 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4baf0eddc0d64b3d9257c70859023909 : SUCCESS in 8m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6884dc9b33b3518e07053a8b127ad1392d228b87","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-08-28 13:36:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"94e307fdf518b856b170536a1aa16b367654a2d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-08-29 03:00:13.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"e4f4fd2d474ef23f0e25fe4bf7ea44069ddfed1d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-29 04:50:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/082041ccda1d4727b0c44fa1967b0d87\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/8ede86d6e6e544fc94994377b1375eb3 : SUCCESS in 55m 46s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/f03cf3c318d34adcbe65d16820a3f6f0 : SUCCESS in 52m 50s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/ccd6e87c6b9d47ce9af930ca36478c62 : SUCCESS in 1h 01m 04s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/20c4acd1c4bf40869e820a62690aeb5e : SUCCESS in 1h 00m 30s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/ba43aab3358244afa3600058d6e7add0 : FAILURE in 1h 23m 10s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9e3972b20677460d81579d8b46eea812 : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cc89f2fe7a1f44e586432dad4ea6986c : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5fb322513bf44c148884ee4d68c1c9d9 : SUCCESS in 6m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"195fd64c573f938599c3789ed7fd8945c310551d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-29 04:51:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56350d38924142bc947172406e872cf6\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/6cf56e79202f418da1b9a9292982a40f : SUCCESS in 32m 14s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/6c22cfcece614479b420a781a29c08c4 : SUCCESS in 29m 03s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/e2e5f820135e4f2d81f96a96662fbb38 : SUCCESS in 31m 30s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/5acbb03a97ce4cf0b3055705700c0cc1 : SUCCESS in 26m 49s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/a8887d9c0b1d477ba59c7aa2e95ec7ed : SUCCESS in 59m 44s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d98f766f938145679aeffd79b5b7a2ea : SUCCESS in 24m 42s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e7e0b20dd8484d97a6cb101bdbab5193 : SUCCESS in 30m 25s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/ef1e3e30444842e99f9b1fb29f326d57 : SUCCESS in 1h 19m 20s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/82b9ee1c765644129efeac2bcb11c8f1 : SUCCESS in 1h 41m 53s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/4d3aa823cb5945abb757b35e39daa629 : SUCCESS in 1h 04m 36s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/2fca16da91ae422ea77005b17e28042b : SUCCESS in 1h 31m 18s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/938827abbf2847f6918459dd9ebbeda0 : SUCCESS in 1h 08m 53s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/a95079bcee2b471c80687af011ea2f46 : SUCCESS in 1h 43m 03s\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/59a2361122f54c339a72fba19ca16c9f : SUCCESS in 40m 58s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/40108337590e47a98d553497ad4f85d3 : SUCCESS in 24m 14s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/5dc5ef9330964439bb8663606cb31431 : SUCCESS in 26m 13s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/3a8c8392995a4f52bc119729398cc9a1 : SUCCESS in 40m 24s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/72dc499f3d624662be6dea26f67a9bc8 : SUCCESS in 58m 55s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/021ce896f78a4b3c9add88970378609f : SUCCESS in 21m 21s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/ba6b522744254432a5a6ae34c75094bb : SUCCESS in 30m 43s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/6a7a5cfcdde44589a347fdc034e0efbd : SUCCESS in 30m 56s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/86f0aac2aa644a9b98be9ddbaa038a0b : SUCCESS in 28m 01s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f5267b052c794b00a28dc5843e35d53d : SUCCESS in 2m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff69ec1c3ac14033a1f2900562dbc1a5 : SUCCESS in 4m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/216eda8eb42b497893ecae6ef155ec5f : SUCCESS in 4m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/37e4e36c8bab4d9ab385131a47cedc2f : SUCCESS in 3m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c4bded7f6b574516b4b95477a68e568a : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd5151f98d1c4a67a316816ef4f8dd68 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca32ba1bec234623a0b99a1908e18c30 : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a0079f084bc4020893a5736fa86c9b6 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":2},{"id":"7824d3f8209b4c1b474fcbe356c058f329e536de","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-08-29 22:18:13.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"feda8479000ffd97fe06f17e734de0fd01fc9f7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-08-30 02:58:46.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"04d3be6fa06d0cb22cab8be7e4ce2592543de974","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-30 05:02:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a942e468878b45338fa6ea828d85bd00\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/db8e9fea48b944369f00c1c38f50b840 : SUCCESS in 58m 03s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/ce353efc366f4f2cb32a9eb515294d3b : SUCCESS in 1h 05m 07s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/6ce090a60bc542b49af8213fb44530f9 : SUCCESS in 1h 10m 45s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/99b75db930804faab166ca277d87353e : SUCCESS in 53m 54s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/de3533aa6f6e4107a78dc5965a71168f : FAILURE in 1h 37m 28s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/512f20f710e84a4aa7db6dbef1e62c6d : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/be141891bbf64e809b1df2942b53ee7f : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/818bfc9ce69a4d069253e852995d2432 : SUCCESS in 8m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0975ed3adf6ddb166299e01049f0aff205b714bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-30 05:02:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4b24bcada144460b100cf1893243856\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/558c650a6baa423abc165ecbacf30b2d : SUCCESS in 27m 10s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/dd371c82ab694e87b56c86e6c17b05dd : SUCCESS in 36m 13s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/85d360fa13f244c088dc27461206301b : SUCCESS in 23m 45s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/fa3fbe72695a4d3791c296c06661c164 : SUCCESS in 30m 09s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/fc953676d4f54608876999281f0cdbdc : SUCCESS in 52m 00s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03ee537956b048a4ba21b2abd8421ead : SUCCESS in 23m 15s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/69a7ccfb78df4d059f56aa5f676a53de : SUCCESS in 37m 25s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/31d5741261624a978991b27a66532527 : SUCCESS in 1h 11m 45s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/b2a04aa1e86d49c98637a5308dc9edc2 : SUCCESS in 1h 47m 35s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/996ee9c48c3043e2acade75ca3e1d207 : POST_FAILURE in 6m 50s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/47f11841624943f39fbb77434cd1fe98 : SUCCESS in 1h 28m 06s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b8d28471f9fc41489f601641b43798f2 : SUCCESS in 1h 30m 42s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/61150a97cfb647efa964da6f63ab449e : SUCCESS in 1h 41m 29s\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/44dda244202d456b9359013a3c3f5fb3 : SUCCESS in 33m 40s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/6466a9d9cb564d51893712794b7c7048 : SUCCESS in 25m 13s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/6ff0b554f9d540d9b99ab686ff7ad4ed : SUCCESS in 36m 16s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/cbce8240f92f48568b8e2c6701e2b592 : SUCCESS in 43m 55s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/f7679b9204e440d4947ad98b2fbdb17e : SUCCESS in 32m 36s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/656d833cd0284f6c837b39ba7e071d08 : SUCCESS in 36m 46s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/005c6791292949c0887cdd0fe92245a3 : SUCCESS in 48m 15s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/3d7f621d71d34600ab4c04a6113f4f5c : SUCCESS in 27m 35s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/a8761d7957624e5fa3d05f0f53676711 : SUCCESS in 44m 43s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b72b3f90d3e7420b9ba331b35ab6782b : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15d84e501d284b289331db14e464154a : SUCCESS in 2m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4546cd071f754f17b035d20355f2cf52 : SUCCESS in 4m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/163392677810408d8a84f1091f35b20a : POST_FAILURE in 5m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1396176ed96c46cdb3c2aa0016a95ff1 : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49b6a06710354d858069fb1e7c696238 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/450b8c0772354f2a91842bbd2a2f515b : SUCCESS in 8m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb62c33a9d724c22b9c2f944d8cb6a3e : SUCCESS in 10m 38s\n\nWarning:\n  File .ansible/tmp/ansible-tmp-1788059470.6047223-70-108948265634112/AnsiballZ_copy.py found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job kolla-ansible-rocky-10-mariadb.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ansible/tmp/ansible-tmp-1788059470.6047223-70-108948265634112/AnsiballZ_stat.py found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job kolla-ansible-rocky-10-mariadb.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ansible/tmp/ansible-tmp-1788059470.6047223-70-108948265634112/.source.yaml found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job kolla-ansible-rocky-10-mariadb.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ansible/tmp/ansible-tmp-1788059470.6047223-70-108948265634112/AnsiballZ_copy.py found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job kolla-ansible-rocky-10-mariadb.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ansible/tmp/ansible-tmp-1788059470.6047223-70-108948265634112/AnsiballZ_stat.py found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job kolla-ansible-rocky-10-mariadb.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"184d959fe6326207bec10a296195baf7a339b6f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-08-31 02:11:35.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"23eaebf39f1f46407c0b07a7e51f3016eae3e2bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 04:47:06.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84e1efa32bbd4b5a8ee44592c0c66b08\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/d77f055cf5ee439980f6ba4fe3f1fb9b : SUCCESS in 54m 25s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/880c5c8828ec481683c73b6a2f94754f : SUCCESS in 44m 08s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/8bb392945124449285f563dab5c33914 : SUCCESS in 34m 54s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/dc9daa8cb18c4666a29de11a6192cff6 : SUCCESS in 32m 48s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/9628c578ee2c436ca6318fc7a3132b1e : FAILURE in 1h 36m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e31a6a422fd84296adf6075209089a19 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e9f57f8e807a4bbaa0771c41cdd51744 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a18cd991aa7e4ab49e325fb2abb8d38b : SUCCESS in 9m 14s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f40be47c8d439df9ab2ac7d6043e2e2760a9e6e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 05:31:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/792397ea19d946548eb792389aed1aca\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/158b420202aa4e1cada77dc534e8bd84 : SUCCESS in 21m 43s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/e4d2ec46800f4f57ab43d9efe9d4cfdd : SUCCESS in 27m 13s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/263d791e2e6741f8b544c642de6bec0e : SUCCESS in 33m 14s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/52a06a9f0ba047d79a8fa47172df9fd8 : SUCCESS in 23m 31s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/534e84f713ed4164bb3c77023f9cde7e : SUCCESS in 41m 18s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/35ebb7b494a44f51a66210b55b4e0327 : SUCCESS in 19m 28s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3e47eda628b944cfb949c0a20a228e04 : SUCCESS in 34m 17s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/3cfad12c128d4a63b3fc7a52e55ed632 : SUCCESS in 1h 16m 30s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/91c639340a9543549738b5aa1c95e699 : SUCCESS in 1h 50m 17s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/91b54041660c46af9c1362091d255ceb : SUCCESS in 1h 02m 50s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/3da11ad5888c40b2b9a75eb4f8539a73 : SUCCESS in 1h 28m 22s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/647df2310b194903866263ec704e2e83 : RETRY_LIMIT in 1h 16m 52s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/3e0fe5b13aca4f5db813b783116b8d45 : SUCCESS in 1h 51m 28s\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/af84608b555d4422b16a5dbf8a76b960 : POST_FAILURE in 36m 37s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/dee03d2a14f04eaa95e292572a4cfee2 : SUCCESS in 23m 10s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/7d12a0c3de794c298b5282414f8ccbdd : SUCCESS in 28m 59s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/d388362e435b41af85d0561c17e6ef5d : SUCCESS in 37m 04s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/1ad7ec10d128408dbaa9588e927a9fb8 : SUCCESS in 55m 48s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/78b1c2bd3d4d4a9c87ad5d046abd3e13 : SUCCESS in 21m 11s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/55e7c1bd04fa46f490d8b5f231b4ffd9 : SUCCESS in 1h 02m 41s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/a3be205323e84110b7b1626a907ba515 : SUCCESS in 40m 16s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/286d1fc81bf74d42a46f1f0bbe81482a : SUCCESS in 31m 16s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/bd6b2b88b9d948ccb7b6c9fb40fe4e28 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4142682cc3d540768b54d40a7e3fa0c7 : SUCCESS in 3m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/796a9bfa124d4280a8d79ab83eb55092 : SUCCESS in 4m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fb297ab360fe4c3c8bfc88b2c77b84dc : SUCCESS in 3m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f75d0f68527c4349a51cc0db52a1d49c : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/945c816fccec45b88eddc1e3a671c81b : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e753a1e37c45437a9cc8299009fc4c92 : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/101fd132518e49109aaa28b47d7d191f : SUCCESS in 7m 37s","accounts_in_message":[],"_revision_number":4},{"id":"b75cece5cd9c5bfefd391834784ec41d98320175","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-08-31 07:53:07.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"14a03f06f187a54c93033cc45e7386d218ce45d5","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-08-31 08:20:33.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"2c66e050ab4999fbed3dd33a5c95bc02b9d7b391","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 09:04:19.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f57424a352c94a698805cf25a619a1ff\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/c1418ca888cb4983afc96b07263c7493 : SUCCESS in 50m 18s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/e3a837a3daa644a59bc016c074bd6270 : SUCCESS in 55m 39s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/1990e3bf7e1d4e1db764d16c89362690 : SUCCESS in 1h 05m 48s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/6b1db49f0a73481ba52e9862217163ae : SUCCESS in 59m 41s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/6369bcb45bb44807a127d5272eb82d26 : FAILURE in 1h 06m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c0d5c65596604caea09050ea2f7391a2 : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e508ddf795ce465ba114e114d676363c : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/68f5cd5afed842c686657f507ab67ff0 : SUCCESS in 5m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bd08e893e8a355cba098d8632e730955a0614a71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 09:56:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0e8d42a64054717a5fc44f55ced76c8\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/19d001a487ef44ecb6134ef623eb6be4 : SUCCESS in 18m 34s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/63dc4fcd84ae428eae37005d54cdf0f8 : SUCCESS in 21m 22s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/d111be727a7d482685afe632b921befe : SUCCESS in 17m 03s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/57a3979d387a4183946bc40341c0e4bf : SUCCESS in 19m 56s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/b0f92459c47e4a4b913be59673520e9f : SUCCESS in 1h 11m 43s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/172ce17536a440e98d89c8985fca8976 : SUCCESS in 31m 18s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/47e077f5c5ab43c59c10eb6f09401df0 : SUCCESS in 18m 09s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/c83ff6b9726541b8ab7236820c3f3c51 : SUCCESS in 1h 11m 29s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/156b7d9aa47645b1b0753d3104792527 : SUCCESS in 1h 44m 16s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/0f08179285904b4e8c7c908234824914 : SUCCESS in 1h 01m 55s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/8a13032dfec84e9bb626f6e1d0979ddf : SUCCESS in 1h 27m 32s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f6f4f2e88ec4a1b807351a8f62d91de : SUCCESS in 1h 07m 16s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/f7e38a88e2b7450db55dcb9af762efcd : SUCCESS in 2h 01m 53s\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/5b13eb33462740259ebd5db33e3860b0 : SUCCESS in 38m 14s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/421b4117c9df48d09f31d4adfb953602 : SUCCESS in 24m 02s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/3b41b40c859c4fd3857b8e2f4c67ee03 : SUCCESS in 27m 00s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/be96434a6e42408dbc2157ee5247ca3e : SUCCESS in 28m 37s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/5a03715b69a84bba9918f3960339b9c3 : SUCCESS in 46m 20s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/6696503488d44d56857d528d563bf5bf : SUCCESS in 34m 54s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/3f0da29fd0984885851d792bf712af2d : SUCCESS in 48m 34s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/f6fff79872464aaf8a62aad7ae275a20 : SUCCESS in 38m 23s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/5b17d16c04db497f816dc3b81ad52572 : SUCCESS in 48m 33s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/5802c7c80f034cf18b9ea637e6ecd974 : SUCCESS in 2m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f35581cd6c94c07a0c4910093935090 : SUCCESS in 4m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/df1b368a156a470f9d0ed4fee2fef400 : SUCCESS in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3b1f3313f3af4835a82b5245688bf4ae : SUCCESS in 2m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ee5ced2d69f144cda3ad17559226faf4 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a750dfa5c01448988d5f668cf1b8a462 : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27ec7a55bb6540d4bac56416cd7a44f8 : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/521a025426914d1b9ce0d0234dcd23a1 : SUCCESS in 10m 51s","accounts_in_message":[],"_revision_number":4},{"id":"da7586722fa79d1c2df5391c872f51e5e6691c3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-09-01 02:26:28.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"5faaa17fafd4b2fac332e7ef3d765d5b2924b4c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 05:16:54.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39316b512e2c4f40896ceb75241c563c\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/2ea33e1e6db046768c055bbc9ccfa436 : SUCCESS in 20m 03s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/73aaf06da5ab472f984ab048a7e63b77 : SUCCESS in 22m 25s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/41e773df53b74a0a80f920d74237bd2b : SUCCESS in 38m 26s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/871b371d012e46e1b1cec5fadd6acd00 : SUCCESS in 32m 19s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/165ec2f2342d49af99e81198707a5458 : SUCCESS in 48m 55s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9f92581e7ea4c8cae42058a7e0792b5 : SUCCESS in 26m 57s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b6a5444a77f44239bdd2e4263e73dd6a : SUCCESS in 29m 30s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/96cd8a234729463cb3df512529c649c9 : SUCCESS in 1h 24m 52s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/800cbe8691bb498db9be460b1807ffcc : SUCCESS in 1h 49m 11s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/c98c093c78db4d2a86a075f411872c97 : SUCCESS in 1h 08m 29s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/238f61e20d884489b54cfa20bbb87a04 : SUCCESS in 1h 35m 26s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ac6192d823243c1844173a9af804a9d : SUCCESS in 1h 29m 28s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/e9320ecbb2cd48e6b0a1a8fed8f6ba4e : SUCCESS in 1h 49m 36s\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/cb864b532fd34f638a79941ef5ec8e7a : SUCCESS in 37m 48s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/82b865db211f45569daff99d94080e0e : SUCCESS in 30m 49s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/e372d02df39d409da9def738438ec337 : SUCCESS in 29m 55s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/fe0af30981b44492b282c0811c5b6be8 : SUCCESS in 32m 34s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/8a84be9857e546d5bbea7547a0fcf36d : SUCCESS in 58m 27s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ea12dea293b7446eb6a7bb5a60171c9f : SUCCESS in 33m 29s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/794e4061f348425a8afc46128d2bf53d : SUCCESS in 30m 04s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/0f2d968bf43444c187968d63bd99bc85 : SUCCESS in 34m 39s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/d2bd07ca8ce44b2fad1174a9f8f483c1 : SUCCESS in 59m 36s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/9b082ba3fbf540c0a8c3d03053a19c95 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7470affcb5994dc8b91369fde218d0aa : SUCCESS in 4m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e867cb6f59f041d8975e9b6b665a1ea6 : SUCCESS in 6m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f21283a2219d440db42ca79f7d6b2657 : SUCCESS in 3m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ae76730f1e3b4898953db2a81342987f : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ec4a3598a424f7b9750621ea39ec160 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42b7d41b9759401cb3ebb01422f84f8a : SUCCESS in 7m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/297087a4532246549c4cab8e649aa6f9 : SUCCESS in 12m 20s","accounts_in_message":[],"_revision_number":5},{"id":"666a7500c88614f5ed7af6e3c265c5b884e2eaa1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 05:21:04.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81380c24d54448689469ec8cb8910fb8\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/be4c7cb2aeed4d57b84b9a8fd49005e2 : SUCCESS in 48m 11s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/64a5493a68a344ac93f4688ae6883743 : SUCCESS in 42m 19s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/0bc3cf902ce24f66b1bc6dd8ea5952a3 : SUCCESS in 37m 41s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/0902ce7a812b4d3aaeea02ba416ab556 : SUCCESS in 49m 03s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/502982e1479947c5b0f0871c2ae16a66 : FAILURE in 1h 03m 03s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b93982afb31b492ab67830919166fe1e : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/099223540fcf45c8b5379d7612cad021 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8eee366f98c44144afaad2ebbe4d12f0 : NODE_FAILURE Node(set) request 58c37a0127564c668004ea6d95643551 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2db89f45c52e143d4f18730f3fe6aca392901e65","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-01 07:17:54.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"fb20b4b6f18b7471cec11f7cdc54011766ea0920","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-09-01 07:52:50.000000000","message":"Patch Set 5: Backport-Candidate-1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"fce16868b6c97b5c672ae2a20db79000cf143c0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 07:53:52.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"003b737a97ba2aabca53abb9e91a89cf30b18b5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 09:58:12.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/343bc02bbaa04fc79e3a6e3355042ae4\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/2fc5ec524056411d97c42ef9bdd0a960 : SUCCESS in 25m 45s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/2507d8f19b3e43ad86b5b940a68a79cf : SUCCESS in 33m 28s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/6a7bec705dbc46f0bcf8d436746a252b : SUCCESS in 18m 05s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/ce3d3de8a0b34e3cb3e896fa10b254af : SUCCESS in 1h 09m 32s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70bc86a8bad24f3b9fd27ef97bc79888 : SUCCESS in 29m 24s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/041d44c9d0bf4fc89bf3dafe14e9cd9c : SUCCESS in 29m 22s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/c66542492ad7440d9ce7c2e71ad0d103 : SUCCESS in 1h 13m 25s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/f0bcbc67ba2a4aeea33c5297d3a59705 : SUCCESS in 1h 50m 19s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/60a3ebaafd7c422aa79474fc52f19ffd : SUCCESS in 1h 06m 02s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/9487869538314864911172b8b50bd0f7 : SUCCESS in 1h 29m 27s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a04b360f5bc4ae3816ec64dac18dbef : SUCCESS in 1h 22m 11s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/8ffbcb8e6a9846d7b1960dc9e38aa6ab : SUCCESS in 2h 03m 45s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/31825f908a464d0aa2a39405529faf8c : SUCCESS in 2m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81052450d32549bcb31af229f31ca910 : SUCCESS in 5m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/949b957f208042ee9caa9dd2c1533c2a : SUCCESS in 2m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cd18769ea0554feebd551990c75dfd84 : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/043f3f6a33764f449d181f45ad1f9fcf : SUCCESS in 6m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f7a4a2874e74cc5a261c0352d8a7c8d : SUCCESS in 10m 25s","accounts_in_message":[],"_revision_number":5},{"id":"115457f1c7ce93121b8be9bf9701211359faf968","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 09:58:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"984d18ab065ed636c1800f17c9c43bfbada3610f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 09:59:54.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62abddd4542140aeb23f8a3f4a8b16c9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/528ac0210bf24d2ea3b8a948c40599e6 : SUCCESS in 1m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/471fa7bc60c5409484e3af2910e0d1dc : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"a2b2b2ca197a7eb16d12f47637c78b7903a83fe1","revisions":{"e4e18e5c814b14d145038a9817864886d83079e5":{"kind":"REWORK","_number":1,"created":"2026-08-28 02:50:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/18/1002718/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/1002718/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/1002718/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/1002718/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/1002718/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/1002718/1"}}},"commit":{"parents":[{"commit":"201438c78025e1364cd0e874560de14065bee428","subject":"Merge \"letsencrypt: Add ACME dns-01 challenge support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/201438c78025e1364cd0e874560de14065bee428"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-08-28 02:50:30.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-08-28 02:50:30.000000000","tz":0},"subject":"Updated from check-sources","message":"Updated from check-sources\n\nChange-Id: I5b1271480c4c3c17e202a81379289730f39ba567\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e4e18e5c814b14d145038a9817864886d83079e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e4e18e5c814b14d145038a9817864886d83079e5"}]},"branch":"refs/heads/master"},"ac0786c60dca11375560ba36048664ead85e1095":{"kind":"REWORK","_number":2,"created":"2026-08-29 03:00:13.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/18/1002718/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/1002718/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/1002718/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/1002718/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/1002718/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/1002718/2"}}},"commit":{"parents":[{"commit":"201438c78025e1364cd0e874560de14065bee428","subject":"Merge \"letsencrypt: Add ACME dns-01 challenge support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/201438c78025e1364cd0e874560de14065bee428"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-08-29 03:00:11.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-08-29 03:00:11.000000000","tz":0},"subject":"Updated from check-sources","message":"Updated from check-sources\n\nChange-Id: I5b1271480c4c3c17e202a81379289730f39ba567\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac0786c60dca11375560ba36048664ead85e1095"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac0786c60dca11375560ba36048664ead85e1095"}]},"branch":"refs/heads/master"},"b6346f84062bb0823a45ad2fd425e2042f7af270":{"kind":"REWORK","_number":3,"created":"2026-08-30 02:58:46.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/18/1002718/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/1002718/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/1002718/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/1002718/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/1002718/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/1002718/3"}}},"commit":{"parents":[{"commit":"201438c78025e1364cd0e874560de14065bee428","subject":"Merge \"letsencrypt: Add ACME dns-01 challenge support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/201438c78025e1364cd0e874560de14065bee428"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-08-30 02:58:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-08-30 02:58:44.000000000","tz":0},"subject":"Updated from check-sources","message":"Updated from check-sources\n\nChange-Id: I5b1271480c4c3c17e202a81379289730f39ba567\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b6346f84062bb0823a45ad2fd425e2042f7af270"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b6346f84062bb0823a45ad2fd425e2042f7af270"}]},"branch":"refs/heads/master"},"3045839e6d5b2cfb3e0dad3f004defd2b5dfe290":{"kind":"REWORK","_number":4,"created":"2026-08-31 02:11:35.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/18/1002718/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/1002718/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/1002718/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/1002718/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/1002718/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/1002718/4"}}},"commit":{"parents":[{"commit":"201438c78025e1364cd0e874560de14065bee428","subject":"Merge \"letsencrypt: Add ACME dns-01 challenge support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/201438c78025e1364cd0e874560de14065bee428"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-08-31 02:11:33.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-08-31 02:11:33.000000000","tz":0},"subject":"Updated from check-sources","message":"Updated from check-sources\n\nChange-Id: I5b1271480c4c3c17e202a81379289730f39ba567\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3045839e6d5b2cfb3e0dad3f004defd2b5dfe290"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3045839e6d5b2cfb3e0dad3f004defd2b5dfe290"}]},"branch":"refs/heads/master"},"a2b2b2ca197a7eb16d12f47637c78b7903a83fe1":{"kind":"REWORK","_number":5,"created":"2026-09-01 02:26:28.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/18/1002718/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/1002718/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/1002718/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/1002718/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/1002718/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/1002718/5"}}},"commit":{"parents":[{"commit":"201438c78025e1364cd0e874560de14065bee428","subject":"Merge \"letsencrypt: Add ACME dns-01 challenge support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/201438c78025e1364cd0e874560de14065bee428"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-01 02:26:26.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-01 02:26:26.000000000","tz":0},"subject":"Updated from check-sources","message":"Updated from check-sources\n\nChange-Id: I5b1271480c4c3c17e202a81379289730f39ba567\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a2b2b2ca197a7eb16d12f47637c78b7903a83fe1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a2b2b2ca197a7eb16d12f47637c78b7903a83fe1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
