)]}'
{"id":"openstack%2Fkolla~1003068","triplet_id":"openstack%2Fkolla~master~Ib4a0d7185ccc64900b4068dd74995dbf1ba6823c","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2026-09-09 15:54:19.000000000","reason":"\u003cGERRIT_ACCOUNT_17669\u003e replied on the change","reason_account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},"37306":{"account":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"last_update":"2026-09-09 21:18:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib4a0d7185ccc64900b4068dd74995dbf1ba6823c","subject":"grafana: install Prometheus datasource plugin","status":"MERGED","created":"2026-08-31 15:03:41.000000000","updated":"2026-09-09 21:19:15.000000000","submitted":"2026-09-09 21:18:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003068","meta_rev_id":"43c714aeafde52e368ecc725aa4b580f81d83490","_number":1003068,"virtual_id_number":1003068,"owner":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-09 21:18:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-09-09 13:50:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2026-09-09 15:54:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2026-08-31 19:29:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2026-09-09 19:23:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-09-09 15:54:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2026-09-09 19:23:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2026-09-09 19:22:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2026-08-31 17:46:22.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-31 16:38:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-31 17:02:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-31 17:46:17.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2026-08-31 17:46:22.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-08-31 19:29:09.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-09-01 08:09:02.000000000","updated_by":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2026-09-09 13:50:00.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"d41e26abd22f8cb706269b080e782432550b00a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-08-31 15:03:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c300591660c36873476c0d0e2af44f99979d9df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 16:38:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2135dcd2a0a496c909f93d862fedbf5\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/25e5c5aaa58c4896aa448e4797b87dcc : SUCCESS in 58m 08s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/09e36d65a2c64a70a534e8bb3b2af4fc : SUCCESS in 53m 15s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/f2335306b9d144da9760234a9fd8e56c : SUCCESS in 58m 12s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/e44f879adabf4fe1a1432d8889405b17 : NODE_FAILURE Node(set) request cadc99328f5e4f5e81b1fabd8f0e8f65 failed in 0s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/27c18d40675a400ab7cf9b250476e168 : FAILURE in 1h 21m 32s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e7129fcc98174aefa33d75bbd7e831d2 : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d0b991b5d0724f19ba9df8dcf4a22cac : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7d621f9de5dc4737a870324fa49aa795 : SUCCESS in 8m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"461ae37014db4216adc0965d88b02f7ca8f06b38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 17:02:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f10f2bd0aa054cf7a7ad6e745a4043a4\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/a0bcaee61f574cab9cf2287afe3c31fd : SUCCESS in 20m 23s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/5f1f2351fbb247b6a51644a78a3e2550 : SUCCESS in 19m 45s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/bb835298716345499f940f92c623f528 : SUCCESS in 24m 51s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/59ad42e450dc48c3861ac47cf8175492 : SUCCESS in 25m 47s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/5efb3f0f4d99463a8cb591fd0c8995f3 : SUCCESS in 55m 31s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c913fc160f224063b42e33e1908febc1 : SUCCESS in 30m 46s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8d4a79591de5420297b27a0147db1b96 : SUCCESS in 19m 26s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/37a5aca66c354f1eb4e65f2de42824aa : SUCCESS in 1h 05m 08s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/1b9b5240dac24a4fac2122e9d562318e : SUCCESS in 1h 56m 48s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/fbf267b6e4024683bce8fe2c5f362813 : SUCCESS in 1h 02m 47s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/5902d33e0147479d9fd1cfa77be0beb8 : SUCCESS in 1h 25m 43s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0627ecdd03014b4fbe1e87cbe27733bc : SUCCESS in 1h 12m 51s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/74d4486b758147ad8b8867155891a767 : SUCCESS in 1h 46m 36s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/5bcd09fd445e40869c00b7b8fad1ab32 : SUCCESS in 21m 08s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/f5c6a569a9534112a86ee7229bb59166 : SUCCESS in 59m 40s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/d67856d507b74662b31ebaccad165032 : SUCCESS in 38m 09s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/8fb7c8fe22e94a89b31e4337104350bc : SUCCESS in 32m 38s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/00ecc2f6d006411b89c0d4a06c0eb2ed : SUCCESS in 43m 57s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/dd3fd20a7f1d430c98e08d279debce73 : SUCCESS in 30m 48s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f86cf3c6dee54893b2bb4642a124240e : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dba972c853214511b422605f929e07c4 : SUCCESS in 4m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5008afcbef684f14aebb46d94fa77910 : SUCCESS in 3m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/93583c82983948e88a22ef49d58eeb2b : SUCCESS in 2m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/499d803d16be414ca33b8dfed9276831 : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dfe9619db7d445da8eeaf2ddbae69037 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/256ba5c42dca4c8c82cffc68e7986f05 : SUCCESS in 4m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f23403924cbc4fa0bac962ba690abbd6 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":1},{"id":"02107c1cba5eb14f747242e0c06125dfc8a34866","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-08-31 17:46:22.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"9e77391f91ec759618b44b82a297292ac3cafc2a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-08-31 19:29:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"acc687ad0c1c6f5eac19c0a6370d946b938cbe4d","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-09-01 13:11:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bb79ec862ce7900281a91930f9642fc0dca4839e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-09-09 13:50:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"20986ff1ba87e39dc849864c3912481a55599cff","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-09-09 15:54:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8be0993d92bd53a6650c93f08997e4efb5ade73f","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-09-09 19:22:33.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"480f9bd0459ee16d852c754ee87df7db98a811de","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-09-09 19:23:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"98f154121eda3bc752e981c6d18b0eb566cc29bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 19:25:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1e33d432e416797efd4b33a8346356fd445854a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 21:18:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2368cf5310434bdebefd2bccd82951ec\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/97d05757ce1e4d95a6362631f4ccb300 : SUCCESS in 17m 05s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/be91aa829ee54e7a98996cd2027a6ecd : SUCCESS in 17m 08s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/1a6a5d36165f452ca2c2d3329a9daf00 : SUCCESS in 22m 01s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/0467258efe364f3fb305b7b0256e274b : SUCCESS in 54m 34s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e9c7f8b27f8349e6bf9670dc445c8198 : SUCCESS in 19m 11s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/17856f60b018467dac8a4e71a889cfd0 : SUCCESS in 27m 05s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/d5d8aedd8fce444f8c878253aa5a6c61 : FAILURE in 19m 07s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/99010151fde74c6394bd10006ea7cf27 : SUCCESS in 1h 50m 14s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/6caf4b5f988f44968f1ec89ff04fc597 : SUCCESS in 1h 04m 52s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/f8ef15567cc04a818c77edd82289f347 : SUCCESS in 1h 29m 53s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b07725b22b0a48ecb962366983620a38 : SUCCESS in 1h 18m 51s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/2c74af718c8441d392437ebe892dddc3 : SUCCESS in 1h 51m 37s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f40069e7ea3d4aec8a0a96181660d199 : SUCCESS in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec2356107a844658a075dade09f261a3 : SUCCESS in 4m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1b0cd320a0a54987b017004bdaae0c8f : SUCCESS in 3m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c9de8909276b4ec898edf91fb40e00d9 : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0a9f16d67f94e4d841336343b01e23c : SUCCESS in 4m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8692e71f06f848b3a64d570f35308c3d : SUCCESS in 10m 32s","accounts_in_message":[],"_revision_number":1},{"id":"6327f1c4c81d8ab28bc6b8f89efa6393fe1331f1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 21:18:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"43c714aeafde52e368ecc725aa4b580f81d83490","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 21:19:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6084087006de4c4f8bd4f34d1ab850eb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5ace5daec40432d862c8e440866a84a : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb438cfe4f7c4632a771bd26a9a75803 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aece83cc507ac587769ae61d011161d462dafb1d","revisions":{"aece83cc507ac587769ae61d011161d462dafb1d":{"kind":"REWORK","_number":1,"created":"2026-08-31 15:03:41.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/68/1003068/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/68/1003068/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/1003068/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/1003068/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/1003068/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/68/1003068/1"}}},"commit":{"parents":[{"commit":"201438c78025e1364cd0e874560de14065bee428","subject":"Merge \"letsencrypt: Add ACME dns-01 challenge support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/201438c78025e1364cd0e874560de14065bee428"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-08-31 15:02:48.000000000","tz":120},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-08-31 15:03:30.000000000","tz":120},"subject":"grafana: install Prometheus datasource plugin","message":"grafana: install Prometheus datasource plugin\n\nCloses-Bug: #2165921\nChange-Id: Ib4a0d7185ccc64900b4068dd74995dbf1ba6823c\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aece83cc507ac587769ae61d011161d462dafb1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aece83cc507ac587769ae61d011161d462dafb1d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
