)]}'
{"id":"openstack%2Fkolla~1003404","triplet_id":"openstack%2Fkolla~master~I68bd3b2dabe9631f7f15b801aaf020380f95fa7d","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"37306":{"account":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"last_update":"2026-09-18 09:41:02.000000000","reason":"\u003cGERRIT_ACCOUNT_37306\u003e replied on the change","reason_account":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"}}},"hashtags":[],"change_id":"I68bd3b2dabe9631f7f15b801aaf020380f95fa7d","subject":"grafana: replace deprecated grafana-cli","status":"MERGED","created":"2026-09-02 07:00:06.000000000","updated":"2026-09-18 13:48:51.000000000","submitted":"2026-09-18 13:47:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003404","meta_rev_id":"ecba83625b474a01aab2893ae84f9b39a181c2ad","_number":1003404,"virtual_id_number":1003404,"owner":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-18 13:47:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-09-17 06:52:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-09-17 08:15:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-09-17 06:52:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-09-17 06:52:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-02 07:53:47.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2026-09-02 07:56:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-02 08:57:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-17 06:52:56.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-09-17 08:15:51.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"030e435fdbf060ae42b43116c2bb0bd49558fabf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-09-02 07:00:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ebc285eeb8f8d18d8e73a65d6bf10ade78ad97d1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 07:56:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/374fb0705a2b4a6d9dfbc99306255731\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/1ddf5d3bf94546fea1ecc76bb2a75886 : FAILURE in 9m 39s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/14a7050111be485c99498e9b6411fdad : SUCCESS in 31m 07s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/4882c18fd6924218a09737d353031bc4 : SUCCESS in 44m 01s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/5d0ca3df07ce41fc9378dcbf8e0d3037 : SUCCESS in 38m 03s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/32730ab3432c430796123369096c181f : FAILURE in 54m 54s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/39235175f5c14608b5cb22c2f0636a04 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7ba7f2f3b5d3472fb06d8eeefecf4668 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9aea374ec2a5424794d3ca0957c768ce : SUCCESS in 7m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0d862f5ebc87c48e5fb44c7925ad97b268727893","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 08:57:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52f44a1408e14313a732f42f3716e1df\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/90ae57951e524dcb864209dde0689b96 : FAILURE in 6m 09s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/d99a937d462d4e3fbd24f1ae770b8367 : SUCCESS in 32m 58s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/ad3788f362a642e98191801a17740973 : SUCCESS in 31m 42s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/59e579ee5a2442f1bbb42e9b13a1ea10 : SUCCESS in 30m 52s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/17fc3bafaedd4261bd129712a2896ed7 : SUCCESS in 1h 10m 06s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa6a9271f2464b899c3cb8e07f2170de : SUCCESS in 28m 35s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/39fa4ce67fd246aba45170be66a3d5ac : SUCCESS in 32m 35s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/7996891d786f4958b1d808b69ae9f49a : SUCCESS in 1h 37m 32s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/142f109a95c347898cf69f0810f18c6e : SUCCESS in 1h 51m 15s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/cfceb39cf74744549447aeef1b2120da : SUCCESS in 1h 05m 59s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/f90f875146964680b430aaf664dcced4 : SUCCESS in 1h 30m 36s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b5b71c3d70b346fb9ae2218db3bcf970 : SUCCESS in 1h 19m 03s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/716da0767b354112b1a8772dced59054 : SUCCESS in 1h 55m 12s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/24f99ed86fe7414cb28f70b46ef0c47a : SUCCESS in 41m 39s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/f9d68beae12d4e31bf3951cf55733f75 : SUCCESS in 48m 59s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/728db515c2f5456290bfc092500c51fe : SUCCESS in 22m 54s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/0f5d96191d34467f8867498b40eecd8e : SUCCESS in 50m 58s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/69f131f2e65b4970843b2212a055f8aa : SUCCESS in 29m 30s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/367a7eefc5e04297b1796efbe2a4e3cd : SUCCESS in 56m 07s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b8ea77f3f0324411b80394b2b9d8e000 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baf9b0df2d444cd995c5fef4dc4269b1 : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e83de75f4fb146029ab04b10fd9c0faf : SUCCESS in 3m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/39b65742394544fd8a4473a2cfb305d4 : SUCCESS in 3m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6597f6acdf2a4380babf8e5cc10a3945 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/213fff75f1d548888631280f382ace6b : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20842863029545bfa7cf8634aebdf007 : SUCCESS in 7m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/575e065abf8f43a2ae7fe259fdc4871c : SUCCESS in 9m 04s","accounts_in_message":[],"_revision_number":1},{"id":"c3d989bc12f2163a91d4fea923c7fc333f5e2f93","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-09-17 06:52:56.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ab3c7c92fcf236250e47229e5efd07265e031b2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 06:54:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6aa0a42d63a2911154fc84ff32d64b550d63889f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-17 08:15:51.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5bb013d47989812babd68508552e9c3f8eb54f09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 09:34:03.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/30c821317adc4a148b86ed233ee1cdee\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/e04a8613e65f424ba5f3a7f36205b75f : SUCCESS in 27m 35s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/9fa50a59bd6b4256a26355aaa7895626 : SUCCESS in 31m 04s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/8a26228723344982b31dc612a34b84ad : SUCCESS in 31m 51s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/d7906380b40d438b9f9e53fa437d7928 : SUCCESS in 1h 03m 35s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c9aa02fe2a8b44e1a02ced02afbb44e4 : FAILURE in 31m 29s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/27f73c0483664de2a9cd5578416a390a : FAILURE in 31m 26s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/6d6086644b2c452d8028a24da20a09eb : SUCCESS in 1h 13m 03s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/3d0ea8cdcbb94876a2f14911ec32e164 : SUCCESS in 1h 44m 42s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/1851adbe8c86458588141f1ee36d7720 : SUCCESS in 1h 07m 51s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/e180e68048f3442e846c22f176dc13ba : SUCCESS in 1h 32m 50s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0f038ff1d60482989d46ac7d398ddc0 : SUCCESS in 1h 07m 02s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/cf1fd757db844d67bab720af75d2d3d4 : SUCCESS in 1h 52m 20s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/dc308470d0994dd7be96d00615b25953 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f65c6f103d6747aeb9c75e1e841e12a6 : SUCCESS in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8a8cf63a1d5e4372a0587a2a8b44d551 : SUCCESS in 3m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4dc6165a36ae4d81876886261f41bd7b : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1fd736ee61e4b4fb8f3f07ffa60496f : SUCCESS in 7m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b152dd8f20684db8a1c5b83211ce2532 : SUCCESS in 8m 53s","accounts_in_message":[],"_revision_number":1},{"id":"c50e4a56b1619928a83ec0f1dba9e20053d7d0cc","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-09-17 09:42:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7a2e7bcd84c9058c4e701abcdb5f64b99e444076","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 13:09:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b0e49d723664a4c94d4c588b877e6a6\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/8222561fcedf4985b3801ac3f7da71fd : SUCCESS in 44m 11s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/69d080bbe3694fc3a0a40b4b0fe321f5 : SUCCESS in 1h 11m 25s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/8f5f6d2a704d4a768949a11ae6c768f0 : SUCCESS in 54m 07s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/9c48039565bc45e1b95176e46c58eccc : FAILURE in 1h 32m 35s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d23d0e163c5049a2b94241ca542d2c09 : SUCCESS in 3m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b43e7446e19e47ef99c06bdf62980e2b : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8d6004dc02c84f0d9bbc0dca7bd045d2 : SUCCESS in 8m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7acd755619f1192d719a3eaaf6ca9f905ee7d682","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 16:02:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a66a05e16264bc4888913bbafbc3bb4\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/302c85cb853e459fb0a04d9167462f5e : SUCCESS in 29m 37s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/e50231ec5030431c936bd74bac31f8eb : SUCCESS in 35m 19s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/cce8d6613cba4a35aa810794e7f7ee40 : SUCCESS in 31m 04s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/2ba8f1bb58634d45863ed82d8ac0179c : SUCCESS in 1h 12m 09s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b35cf51fb9cd41098da19b1cdcff3944 : FAILURE in 28m 49s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5c48ce231dee4b9b9667438a2b9c163b : FAILURE in 37m 28s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/466cf2838b4a49beb2869a15ee3c8c3e : SUCCESS in 1h 11m 16s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/da8ee164e36f494e8adc6baf5ef29951 : FAILURE in 21m 19s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/072aa4582d894fdd98ec7d450762f598 : SUCCESS in 57m 23s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/ef1482e044d14470a3b1e6643b90039a : SUCCESS in 1h 10m 35s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1506c2a4eab34c28a5ffacf572f50e4e : SUCCESS in 40m 21s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/3a7c493ccdb042bbb923f3dec45bf6d5 : SUCCESS in 1h 04m 38s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/40de7bbe115640f480745073fd0fdeaf : SUCCESS in 38m 53s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/1b7cc19de6ce46629440aa6c23cca5d2 : SUCCESS in 48m 43s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/8aac13057bf542689707d10cb1a2a4c7 : SUCCESS in 45m 48s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/e20ac39df2f64665b5f2cfa3ec9ed402 : SUCCESS in 42m 54s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/7df2aa79c81c4443bc7cd13a2a3889f2 : SUCCESS in 36m 16s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/3d18e7665ff84be5bc7c437189764224 : SUCCESS in 1h 01m 51s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/3722ee8f0ac34104a327eb0c252a223e : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58537b925d284a38a39db8b693fe8f45 : SUCCESS in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/962a41a6ed12449fa731dbccd6ed3ece : SUCCESS in 3m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2174cfbc2cca43beb1f03b4a81391cf1 : SUCCESS in 3m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4fd9fa3ff0b7419a8729f36284548869 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/814f03aee957458e899b7678d4ef5198 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08cb4ed90822487e8486c2db0626f6a6 : SUCCESS in 8m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe252f0e128541c2b9ec712f6896bdef : SUCCESS in 10m 46s","accounts_in_message":[],"_revision_number":1},{"id":"e78b0b322fe99a4927254dc08bf4cb4dbbd8a1c8","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-09-18 09:41:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a5c3f9c417e85af3f3f286e163012f9e939b455a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 11:30:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e288b707e9d5489788d61f8dddb6c27f\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/3b0d7f73246441fb8cf080790a01ebff : SUCCESS in 45m 39s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/53b5e5c995354b4aa6b7402506b67e9d : SUCCESS in 52m 04s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/7f6e6a54266547398cb5aeb4950122dc : SUCCESS in 45m 06s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/62671e7067354bd48b0de64e163cd413 : FAILURE in 1h 18m 52s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3e456150a05e4095961324d94d103ed4 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4cec88c53e4740e58cc04b2cb32dc838 : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1f49c416116e4c52b3dba1f633370bf9 : SUCCESS in 7m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b3c3b260443cd5d4809ff547b1d2b82db5c80c09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 11:48:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa5a6284761943529f28bcddda795eb1\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/0a0d1b96fe544b96a8931c60ffda848c : SUCCESS in 32m 35s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/e95495fbb8454e57baa18833818d6a47 : SUCCESS in 35m 24s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/08c9c32cbf5c42a899ee6a1822e92804 : SUCCESS in 36m 11s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/88f0fad3605a4dfd9c99f13937f9a3a2 : SUCCESS in 1h 15m 14s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ede515f5ff94e138fd296da74018fa2 : SUCCESS in 37m 58s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1950d1fa6aa44c889a200dd4af6b2b13 : SUCCESS in 29m 33s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/a967056f5f914a4388b19868a0ebf0f8 : SUCCESS in 41m 08s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/ee791a060a064ecabbf05b5bd0579be5 : SUCCESS in 1h 33m 27s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/de1d20c09c0d428c98c91f1181a51587 : SUCCESS in 1h 02m 12s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/365a5a5f55914d56a496cd6c92b5cd82 : SUCCESS in 1h 20m 41s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b4a8376660443e1a9adc1d405e9feae : SUCCESS in 40m 44s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/18ca9484d01d4263bbca264bebe72344 : SUCCESS in 1h 21m 15s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/af091fe57d4b48e79862120df61c0fb9 : SUCCESS in 31m 23s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/9e11a3eb054249d1b4b8c1ddd067cca6 : SUCCESS in 56m 47s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/b3135bc3d4e846b08955630bba25f825 : SUCCESS in 33m 08s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/b6c987d5f31d44418383197fe8da00dc : SUCCESS in 45m 04s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/6fa87472a7794f65b74e7f80818372e3 : SUCCESS in 42m 35s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/8527b9629b4445a9a3b0e4b184fdf429 : SUCCESS in 1h 04m 44s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/1b1c6216e676437bafd3b8f19fa1b05f : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29c0e477b4804ee39494a8cc9cc8c401 : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aa1a295294fb47aabc162ced1021c0f4 : SUCCESS in 3m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b193f1bd19e7409782af39558282bb09 : SUCCESS in 3m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bc230241e1ef4dcfb75d324bc45ee8c5 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56d9e62363c348dda08f1f00cffeaf81 : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37694f7bf5c44877950748c08f2ae1cd : SUCCESS in 7m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af39dde9b47d45b28d297d0e7b269dc8 : SUCCESS in 11m 35s","accounts_in_message":[],"_revision_number":1},{"id":"98a43a4bb4d071f74d3521ea618b628c2182a67b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 11:50:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c680751a72d3ffe90fb24778c8452490ee8c5ce7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 13:47:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76480b78a29f4e309123b9a2e43c1361\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/0086290ea2d948ca80efd18ca1761e55 : SUCCESS in 28m 52s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/d51689d358a44f659fc101f6352f7ea5 : SUCCESS in 33m 03s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/37b101a06ce24059891565899711d72a : SUCCESS in 21m 39s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/9ea1ee7a66d74bc384ff2a34c8431974 : SUCCESS in 53m 47s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2608a3f8a124410ea8343e9e97c07a18 : SUCCESS in 32m 45s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/efbbb5c50a9a454392f90de9ee0a7e83 : SUCCESS in 33m 43s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/50ae407178394c8082d63dd24815216a : SUCCESS in 1h 07m 39s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/18ef5aa5377d4b83b134a9604c8d350f : SUCCESS in 1h 56m 40s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/b89b491fdd204415ba696f8570a8b737 : SUCCESS in 1h 15m 35s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/31925c09e2a448cf82da620f0797bc3c : SUCCESS in 1h 31m 11s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2335bdd4cf054ec7ac472864aed1bce9 : SUCCESS in 1h 17m 14s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/d7abae65a13f44c38e1021f8605d3368 : SUCCESS in 1h 50m 05s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/cff07afd0d2a4425905eff6bb2653550 : SUCCESS in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceb8a129a21a4800adc24e2f4ccdc25d : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/730982e752354031a47b8c08c1e6b88a : SUCCESS in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/53a116b79add4c44a442b17fe24d5eb9 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d97be5f50b34a858c4f93d679b7ebd5 : SUCCESS in 8m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4c9a77397d748fe87e9d782417228ee : SUCCESS in 10m 06s","accounts_in_message":[],"_revision_number":1},{"id":"ae32453e7ddbd9e62885a0dfa38077d71169e67d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 13:47:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ecba83625b474a01aab2893ae84f9b39a181c2ad","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 13:48:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e10bfcebcd8c49a683943aae4f0bbdd2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d758d3baf0014debb368ca43bcbf8b2b : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91165abbd58e4688b2fecbb9db2fc2e4 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ee8b2a2095389567ea393f1b364cad02ec78b766","revisions":{"ee8b2a2095389567ea393f1b364cad02ec78b766":{"kind":"REWORK","_number":1,"created":"2026-09-02 07:00:06.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/04/1003404/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/04/1003404/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/1003404/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/1003404/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/1003404/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/04/1003404/1"}}},"commit":{"parents":[{"commit":"455303aef30d245485b34ac22cabd5ab3cab562c","subject":"Merge \"Upgrade Prometheus Server to 3.13.x LTS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/455303aef30d245485b34ac22cabd5ab3cab562c"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-09-02 06:59:29.000000000","tz":120},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-09-02 06:59:56.000000000","tz":120},"subject":"grafana: replace deprecated grafana-cli","message":"grafana: replace deprecated grafana-cli\n\nChange-Id: I68bd3b2dabe9631f7f15b801aaf020380f95fa7d\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee8b2a2095389567ea393f1b364cad02ec78b766"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee8b2a2095389567ea393f1b364cad02ec78b766"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
