)]}'
{"id":"openstack%2Fkolla~1004161","triplet_id":"openstack%2Fkolla~master~I32526f218776214f1f9516b126e10816769eb220","project":"openstack/kolla","branch":"master","topic":"openstack/kolla/check-sources","hashtags":[],"change_id":"I32526f218776214f1f9516b126e10816769eb220","subject":"Updated from check-sources","status":"MERGED","created":"2026-09-05 02:18:19.000000000","updated":"2026-09-07 07:22:11.000000000","submitted":"2026-09-07 07:19:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004161-openstack/kolla/check-sources","meta_rev_id":"eb6a5b0b6a977b0223a9f666b9f1ddca14364566","_number":1004161,"virtual_id_number":1004161,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-07 07:19:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-09-07 04:58:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-09-07 04:58:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2026-09-07 04:58:15.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-09-07 04:58:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-05 04:38:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-05 05:26:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-07 04:58:15.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"af59faad91a94e7dc285d6f6d6fd91be1210a2fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-09-05 02:18:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a243df7852d3503ac4851871fbddce7a7857dcf1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 04:38:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0953a4b7d52641b8a0831156cbb94272\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/29d7753038694f75a268b1929194bd2f : SUCCESS in 39m 06s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/a856339de65f48768c1298fc50814a29 : SUCCESS in 36m 19s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/a8d2c8f000104b6c8f10132a2d971ee2 : SUCCESS in 36m 21s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/9915a399489e43f6afb3f7dc7715a13f : SUCCESS in 48m 14s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/ab539d3e61a94b0cb89e9ba614ec6c36 : FAILURE in 1h 18m 32s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0dfee8ba75634f81971719d6d2d800ee : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6093ae3974a747a89e3535f6d300195e : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/82ed1088309c4ab59dfecd6539093b5a : SUCCESS in 10m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a1074434e83d8147e4d345edfa5b5012f646ae77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 05:26:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/088dd402b54c4373a5761288da299eae\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/dcfbe179796c4ec9ab864a347b71cae8 : SUCCESS in 20m 50s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/8e927ce578bc4ffd954ecb41a1637f46 : SUCCESS in 17m 09s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/04ae957dd28a4662a247203711e845d2 : SUCCESS in 28m 36s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/76f98203879741c98967786086fac81f : SUCCESS in 30m 53s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/7ad0b77d522b463bbee62dbbf1c533c8 : SUCCESS in 50m 58s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c4dd302efcec405ebe24c5d443ed2f25 : SUCCESS in 35m 59s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1b20f3a9b9d447558583f5d8d1b4203a : SUCCESS in 25m 42s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/b7309db2843341a0a2b0057511b7cda9 : SUCCESS in 1h 07m 48s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/ae5a91121b7d48c6808a8f34c0b57748 : SUCCESS in 2h 09m 09s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/6c55e546aa684c97b6ba8c23f22af611 : FAILURE in 15m 59s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/3fc7645ba81d4928810f86be46799e2c : FAILURE in 17m 34s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a34ccce3f5b48a88ddecee837c94dd5 : SUCCESS in 1h 20m 25s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/185a1680d52a4d3ba264ab76d1645a23 : POST_FAILURE in 2h 52m 00s\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/18ff5a46a417468b9b539f44efb94230 : SUCCESS in 40m 55s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/4b4d56682f68447f91e1647716e58619 : SUCCESS in 25m 14s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/6e3d6405fe304024986e3c3f3af29be8 : FAILURE in 7m 15s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/4040992152c142799a8d14e7f4d495ca : SUCCESS in 40m 23s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/ef36c80d955a4a1a8933b14651f73631 : SUCCESS in 53m 44s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/dfb286667b704f56970c6dda3dd8e1f1 : SUCCESS in 41m 17s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/cc7640398dad4b4cbb3411ed72fce7cf : SUCCESS in 51m 04s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/8fe6001e29d74f25ba6b4a4715e53557 : SUCCESS in 36m 59s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/c9493bfa9c644f0a94cc9b29138b1e2c : SUCCESS in 59m 11s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/294cbf76aa424093a80f03d8b95090a7 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/613d3b7acaf04c1da9d08fcf046a1319 : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ec9ab2124787497a802bdcfa46753d59 : SUCCESS in 3m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dc85a0e44a1e4870bc0aec60dda4fc42 : SUCCESS in 5m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/05118c06af7344ed917190095ec5bef5 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46ed294ee0384340864d9607716e660a : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f6bcf6a2ced4cc68cab3aa5e541203e : SUCCESS in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c616ea0d2fb64328910d46b36befb4e2 : SUCCESS in 10m 07s","accounts_in_message":[],"_revision_number":1},{"id":"e6b4b6268e06c97dab657054ff9b272be4e00809","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-09-06 02:22:01.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"5f793725a15b511a6c8f00d6fdfeb348e6f13006","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-06 04:59:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac575e948f314e4f8bbac0caf31658d8\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/51d6195e103b4762a836ded8e03f442e : SUCCESS in 49m 31s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/93dc28869b6040fe85066ebf346b9b8a : SUCCESS in 1h 06m 35s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/bec540f8952d4f49844bb9f4798a305c : SUCCESS in 1h 16m 28s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/d0c4c97bfc96429eb365a64924f11003 : SUCCESS in 48m 05s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/7cede256a1e74a7fb356de2af233d6a3 : FAILURE in 1h 34m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d1725e9219b3446a84c9a8e391d14885 : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d62942f63a2d48dab925a95883220ff1 : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3d5824433d6e40aa863346a4129e9bbc : SUCCESS in 11m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c439715163e0704466c4f9077959f59958531451","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-06 05:07:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e7311d24f184b3eb925f2b69a8770b8\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/b2ba975397744b8bb4f393480c3b4029 : SUCCESS in 26m 48s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/28ac0fb32d5a46999967c4ef509e16d1 : SUCCESS in 18m 15s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/9d82d0c9bc0544bdb83c13f587895f26 : SUCCESS in 23m 26s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/1b2cebccc4da417e847b61e0ff66db01 : SUCCESS in 31m 44s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/45ae2e2412584abbaa43bf4e40086076 : SUCCESS in 1h 16m 16s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d031bad65b24fa7bcae63d3e087b435 : SUCCESS in 27m 35s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fbf1db1ee4014bb8b3b89ae17c296f54 : SUCCESS in 43m 30s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/3eb1b90e2e89478d9a0fb93dc8eb04fb : SUCCESS in 1h 12m 30s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/e37d4bd690464e0cbad8e1479b4197ac : SUCCESS in 2h 08m 18s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/7aec8ad1268c4aa19c98f043f1bee3f2 : SUCCESS in 1h 07m 33s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/30bc9c1fa1d14ebd89e7644599e9b198 : SUCCESS in 1h 26m 29s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a845ab3a0ae4ed5bbdd24256892af51 : SUCCESS in 1h 20m 05s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/caba71922da94bd28f0ee0f9ce6fdcc9 : SUCCESS in 1h 56m 56s\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/108f21db01474af68ba3695ec828f3d6 : SUCCESS in 37m 28s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/6e7da6fe4d4f4c5fa4b3969dbd196f10 : SUCCESS in 26m 27s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/a47efbc0d6cb4a36a9cd450bdf437af7 : SUCCESS in 30m 32s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/adff2469094543bc8cdfecfa5fa9f4de : SUCCESS in 40m 19s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/6cb4442721c444e0895ca75bfdf832b0 : SUCCESS in 52m 22s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/35a0c0d332554d5c80c770b4b31051e9 : SUCCESS in 23m 31s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/5b6cc20fd968485dadc86867de5998e0 : SUCCESS in 38m 23s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/a7a7ec505dc041aa8feeaa8d28ae2782 : SUCCESS in 34m 11s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/36833c64d67f4d07bd0c2787bd3b3f36 : SUCCESS in 39m 22s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/42a04a5a6ed0416dbdd8bf95a39c49eb : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55a535c9d85641f2a462b6ad7ea10a46 : SUCCESS in 4m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/09258b4c5a744ceda981f7cf316c580f : SUCCESS in 5m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3ab3637614c84a0aa382657add45b11d : SUCCESS in 3m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3868ae8d4bab449298d59280aab97386 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da89e072ca1f4a2f91063496a76b3ae9 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb71925019f84a4f9da02bbc06ee8cdf : SUCCESS in 7m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1dd4b738a3e345b6813f6e2642806769 : SUCCESS in 10m 53s","accounts_in_message":[],"_revision_number":2},{"id":"5286be6dfd0138442e54e32da613591498d3a7ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-09-07 02:21:19.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"0f4894574dbae674846f0c5f22e4394df641d1b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 04:28:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbc22c00ebb648f5ac13c34eee7d854c\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/085c1036e1934779a19e000f8acf3d49 : SUCCESS in 43m 16s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/8d103cfcb2854ec49c7aa4131bd25670 : SUCCESS in 41m 38s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/3e304867aec344de8b8b5918e876c41d : SUCCESS in 54m 56s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/b83d78fe43a048708ff3cbfb4cac2f49 : SUCCESS in 43m 17s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/c46369d698e1445fa293ce40aa973e12 : FAILURE in 1h 19m 55s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a12fad753e61449f9c3465b5af468add : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/47b27e3c0d594354a55dbb33c00c4b43 : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/01149ea00efb48bd80aeca6105fba465 : SUCCESS in 8m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d4001ad736ffb735a5b72cc56dd15738c64c4459","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 04:50:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c7de11aac5e4086a0716a36c9c5e029\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/4e8a3d6d24f84385b5e7ee7610172e15 : SUCCESS in 32m 07s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/73a89b29c7b54b5b84da10cbb45b0aec : SUCCESS in 29m 39s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/c719f9c07fe7415da4a0ff0467ab8351 : SUCCESS in 23m 00s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/5eb60ac0518b4a648067a44e33126c81 : SUCCESS in 21m 26s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/c5c56fcbf63e4432944e34d8b5396b11 : SUCCESS in 53m 20s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/32427f2febd647ac85913efc5d7352fe : SUCCESS in 25m 07s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/92dbe39b55244f618a5d6a5ddf47917f : SUCCESS in 20m 15s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/ce92122be7184c96a9c3ab8407313043 : SUCCESS in 1h 12m 05s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/765a3d40983e4d9b9cb2a02ae979a8f8 : SUCCESS in 1h 47m 12s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/0c62b61963d04e56ba72d9c89cd8b495 : SUCCESS in 1h 06m 27s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/208cf8daf7094c9fa6522b82ad366399 : SUCCESS in 1h 31m 00s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b28535042154fdd8743d3fe8fdea636 : SUCCESS in 1h 14m 25s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/fdf734fd71924e7093f4be5aaf3a8fa7 : SUCCESS in 1h 51m 59s\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/bee42c6e3c954f36a3619662258a6511 : SUCCESS in 44m 13s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/852e7672c4b044af928b383453916a0e : SUCCESS in 25m 27s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/0a67e8fa7a6147a8a35762851a283eff : SUCCESS in 30m 12s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/244889715aec4dfaaf2fb762eff74db8 : SUCCESS in 23m 02s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/3ed5e3b92ac24361be5cf84b3096103d : SUCCESS in 59m 47s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ccdd22e989b041069cd59932e2d2990c : SUCCESS in 43m 44s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/2739dd6a7fd445448bbf03b0936f9913 : SUCCESS in 1h 06m 06s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/90aba7d02c874e2bbf3a35941e1db1e2 : SUCCESS in 38m 56s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/523c0b2ade8b4e3197de6a3e79fc6e49 : SUCCESS in 59m 14s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f5e7306b52464b1d80d93e1e8db40d7f : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11cf6d94d20246d38d705615e42736e1 : SUCCESS in 4m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1f136b040e1348a6b3e44f7f0d4a750a : SUCCESS in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2ca5cf02868e42aea1b51b58eb248d1a : SUCCESS in 4m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bf92ef91f8564eed80f9c03b99e39bbb : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38d8f35110f642b79eed39a17ba3b948 : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8207b2a13c9e47139b827036bc4ffcb6 : SUCCESS in 4m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/068eb9aae19e4d96a9d6aa5fa5b9aa87 : SUCCESS in 5m 56s","accounts_in_message":[],"_revision_number":3},{"id":"86942c32bd0920b5917226677275ef9a2678b45a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-09-07 04:58:15.000000000","message":"Patch Set 3: Backport-Candidate-1 Code-Review+2 Review-Priority+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dafa24b26c5dfcdb0f882ca6b6fd1226e1f43224","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 04:59:56.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"eda3fdde370b74fcdc96979dec1aeb62fa492d51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 07:19:39.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ac95a4c69374dbeb342fa0d1f278371\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/740287b051dc4fc3ace131f801cb927c : SUCCESS in 21m 39s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/3335df1417924b458f7ba96fde259554 : SUCCESS in 21m 00s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/5b35bc052b394ff0a29cd2f4e80cb775 : SUCCESS in 20m 12s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/3174cabf54114051ae5e8c8a927a3b47 : SUCCESS in 43m 05s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b0d1d7892194e7292488397be97d294 : SUCCESS in 27m 47s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d36b71af9fa141d1bd423f50efe6ace3 : SUCCESS in 21m 59s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/6910ec13de9c4af88a8d05c822f4f70d : SUCCESS in 1h 05m 43s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/a8b8ef33b5004febac7258ae1efbecc9 : SUCCESS in 2h 05m 41s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/3ab2370d71a94fa789de99940673d8b6 : SUCCESS in 1h 04m 52s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/a37d35ad3baa45a4a767bd86380075d7 : SUCCESS in 1h 30m 36s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f7eac31ec9c49888c96706efc4a3832 : SUCCESS in 1h 17m 01s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/c74ae580c48b4836ad027f330e67ece8 : SUCCESS in 2h 19m 04s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/72f06b7c988a49489a09cb64362707fc : SUCCESS in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6340418f55694196b4c8585a20942d7a : SUCCESS in 5m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3b2151fa67564a4abeda7ef1aae3d815 : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a30ce332195b4f5a96505a5991ffb794 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c917241a499247a9a1290227527842b9 : SUCCESS in 4m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c716ad14fd949388642ab53ecb6e7d4 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":3},{"id":"47f2a0369b39e07d088f5d010c8ff175882e2017","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 07:19:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"eb6a5b0b6a977b0223a9f666b9f1ddca14364566","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 07:22:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa4d0223a283454dad87bde7330ec3c6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36e884b1073d4fca98e0bb72e3b71d61 : SUCCESS in 1m 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1bd0c32ad44c454e95494b0b41b374a0 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e30a8df63633177915b95b0c94a03abf6c0b86fc","revisions":{"fa4a4dbb4963e94a3cec3564b0b2850fa24423a0":{"kind":"REWORK","_number":1,"created":"2026-09-05 02:18:19.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/61/1004161/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/61/1004161/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/1004161/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/1004161/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/1004161/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/61/1004161/1"}}},"commit":{"parents":[{"commit":"a2515e086b43d859be67c8b1f4af82d4a75a5a1c","subject":"Merge \"Ensure openvswitch-base children get ovs repo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a2515e086b43d859be67c8b1f4af82d4a75a5a1c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-05 02:18:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-05 02:18:17.000000000","tz":0},"subject":"Updated from check-sources","message":"Updated from check-sources\n\nChange-Id: I32526f218776214f1f9516b126e10816769eb220\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa4a4dbb4963e94a3cec3564b0b2850fa24423a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa4a4dbb4963e94a3cec3564b0b2850fa24423a0"}]},"branch":"refs/heads/master"},"05e530df26fc4ec8a7d381186d93ded6c26417cb":{"kind":"REWORK","_number":2,"created":"2026-09-06 02:22:01.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/61/1004161/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/61/1004161/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/1004161/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/1004161/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/1004161/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/61/1004161/2"}}},"commit":{"parents":[{"commit":"a2515e086b43d859be67c8b1f4af82d4a75a5a1c","subject":"Merge \"Ensure openvswitch-base children get ovs repo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a2515e086b43d859be67c8b1f4af82d4a75a5a1c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-06 02:21:58.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-06 02:21:58.000000000","tz":0},"subject":"Updated from check-sources","message":"Updated from check-sources\n\nChange-Id: I32526f218776214f1f9516b126e10816769eb220\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/05e530df26fc4ec8a7d381186d93ded6c26417cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/05e530df26fc4ec8a7d381186d93ded6c26417cb"}]},"branch":"refs/heads/master"},"e30a8df63633177915b95b0c94a03abf6c0b86fc":{"kind":"REWORK","_number":3,"created":"2026-09-07 02:21:19.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/61/1004161/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/61/1004161/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/1004161/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/1004161/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/1004161/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/61/1004161/3"}}},"commit":{"parents":[{"commit":"a2515e086b43d859be67c8b1f4af82d4a75a5a1c","subject":"Merge \"Ensure openvswitch-base children get ovs repo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a2515e086b43d859be67c8b1f4af82d4a75a5a1c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-07 02:21:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-07 02:21:17.000000000","tz":0},"subject":"Updated from check-sources","message":"Updated from check-sources\n\nChange-Id: I32526f218776214f1f9516b126e10816769eb220\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e30a8df63633177915b95b0c94a03abf6c0b86fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e30a8df63633177915b95b0c94a03abf6c0b86fc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
