)]}'
{"id":"openstack%2Fkolla~1004817","triplet_id":"openstack%2Fkolla~master~If7b5d045200d40dc9f49d26f2833af9a54f7ca96","project":"openstack/kolla","branch":"master","attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-09-09 17:51:15.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"If7b5d045200d40dc9f49d26f2833af9a54f7ca96","subject":"base: Drop redhat binary versionlock block","status":"NEW","created":"2026-09-09 14:15:19.000000000","updated":"2026-09-10 13:42:53.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f51fb528cae54fe83b55b2672f1e12dbe651c3fe","_number":1004817,"virtual_id_number":1004817,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-10 13:42:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2026-09-10 11:20:25.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-09 14:49:01.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-09-09 17:51:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dfb220826abd5c18c348c7b89687d7ec93c73e51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-09-09 14:15:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d0b559aaebf36434df019e8967d5073c1c1706b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-09-09 14:30:07.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"908757d252792411c49194c8014280fbecb53973","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-09 14:49:01.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7756c3c8d38dd78ca270203601b7cbd8402b5e8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 17:51:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9723ff94965440688b3c1d4da2a4d578\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/52574bc1dd0d47e0a74a3261c8f0806b : SUCCESS in 22m 32s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/485e2cec8f36439e886ce8f103140edb : SUCCESS in 37m 37s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/590f0842f8d94b61a1d5367ff2b52fa9 : SUCCESS in 31m 38s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/cd88218d887e4327a21143fdc63858cc : SUCCESS in 1h 02m 40s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b9ba0e8213d49adb6cc8c72dba36920 : SUCCESS in 39m 49s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d867871d183b4f768c044bad71c34ed4 : SUCCESS in 36m 09s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/6ea0d226b3e4407a8122521701896970 : SUCCESS in 58m 54s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/2cb4c5352c4c4b79b2913e13be2f310b : SUCCESS in 1h 08m 46s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/d50861630f034378b35ca8dd78fa5062 : SUCCESS in 1h 26m 10s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/57eb9b401db241e9a566182e39cba512 : SUCCESS in 1h 40m 11s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3d841329503b4d52aa668de7b0e749d5 : SUCCESS in 47m 23s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/d186d382bdc14b66bdefaa4ded55e42e : SUCCESS in 1h 33m 36s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/6da174c6b8f142fe84b9c686f4a7ef83 : SUCCESS in 1h 43m 10s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/5a9f7600a52f4be399da9c0402b9ec6d : SUCCESS in 1h 14m 54s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/6e9a03da58dc4108aac5cde2033a1868 : SUCCESS in 1h 45m 08s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/dede2e1002d74d93b409a0b6c1850461 : FAILURE in 1h 17m 45s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/873c2c20230a435bb646760ee5175d47 : FAILURE in 41m 43s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/25b53e2c3c6347f79b43eb9b31c86b62 : SUCCESS in 1h 19m 08s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/303c97c059d14369ae9e88a841046aad : SUCCESS in 1h 46m 06s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/50a2c2792c3845ceb89c5554ff8a0434 : SUCCESS in 1h 19m 28s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/a30a92eb05134fae879a31cb148c53de : SUCCESS in 21m 08s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/26ed24147a6947b39223ad3c2df27c16 : SUCCESS in 22m 07s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/e3c4cfa4f44d4c64aa2ddce6432fe525 : SUCCESS in 33m 16s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/900c4c0c412f4bbb81fb381071f51630 : FAILURE in 54m 41s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/fbc32aa48a88485ab727b97d16ab274c : SUCCESS in 49m 14s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/c4ecac405f3a4140a52d6506e58016e3 : SUCCESS in 1h 23m 32s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/dc8d37b7d5ff4a3985caea09481269c9 : SUCCESS in 1h 23m 03s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/85cd6721bdec4bf4a2b58e86d810e77c : FAILURE in 19m 18s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/faa48245ccac4510af769bb7bf21a142 : SUCCESS in 1h 52m 59s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/e6affb2e67ea465397c37ac49cf1eead : SUCCESS in 1h 19m 34s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/b1225453296f4b69b4ff3b00d7b545f7 : SUCCESS in 1h 35m 07s (non-voting)\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/57af695c79194a4b90873681dbd40dbf : SUCCESS in 41m 35s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/975ff64717184f3ab97e727eb034a2ed : SUCCESS in 1h 04m 17s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/57745ec6de4148e7b090f1170b97e2e1 : SUCCESS in 39m 09s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/6351a03bb80b4a828944c46ed35ebc0c : SUCCESS in 54m 25s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/d0ef0af1a69541ebb553bff88b5f1c87 : SUCCESS in 46m 04s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/cfb2a7f0eb834d6f87ce06a889b75871 : SUCCESS in 59m 23s\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/896f607761324f1ca003aceba59b790e : FAILURE in 1h 14m 27s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/68c4f966b8a64fc1a72ab050410d7845 : FAILURE in 1h 20m 18s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/669b3b1b63b94630b30db913a0fbc647 : SUCCESS in 1h 35m 56s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/af3d8d45d12e4da9a2d972cbf6bbe428 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e87b4f1c7846499b8a546705a0381cfd : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9c2586bfaf0d4ad99d2f870004b1a884 : SUCCESS in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/836f330af6e54ae38a2f410da19756cd : SUCCESS in 5m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/03bab85e73da4102ab19c1bfdb2c0ba2 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dffc9982a15f4c8589d891f1f47e8961 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6db83fed25a546d7b7339eb0829def98 : SUCCESS in 8m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50d67ef2125b42b382cb4bdd3f9bc7b1 : FAILURE in 3m 25s","accounts_in_message":[],"_revision_number":2},{"id":"d55cf5a4d180b820dafedeb3e742a65d15c9fb88","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 19:54:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29d0ee0eb1f84b3a8fe03330430a51ab\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/f84cc8687d8d47f680bc5875af115c35 : SUCCESS in 47m 26s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/99a50c6207c64e1f9c0ee884c997f9b6 : SUCCESS in 57m 01s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/8eef9fb694cf40bfa16a0fb47b2fd97f : SUCCESS in 50m 56s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/c0c5d1d3626d4e1282fe7715235a9b41 : FAILURE in 1h 30m 24s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/108e681b44ce4a2eb5a23e242d50fe7c : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/62d711852de146a594f14b34d8f46b06 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5cb780a833274a9a8c1a719394d847e6 : SUCCESS in 11m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"47030caf22891eb1be97d3f0a9a2d9c33d7879c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-09-10 11:20:25.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"bd9d8f1277f8afd1942432ece578bd41456656a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 13:19:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c023e9107d7648d28ad88b021418e442\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/a56accc03f384d17a942bd40dda57448 : SUCCESS in 43m 54s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/a90b7290de2f4048aff4dd8fdf6a2a3e : SUCCESS in 48m 19s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/bff10350085248f0bd21e048294683b6 : FAILURE in 37m 13s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/095854e4ed0b46c5afa9b8d5fdc82c10 : FAILURE in 59m 54s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ec6a3dc6c0544a51a8b008fb98a41ad8 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b5b586dece854b54b6e04f54437afb6f : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/03754200311d4b229adf61887b905be0 : SUCCESS in 6m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f51fb528cae54fe83b55b2672f1e12dbe651c3fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 13:42:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a30a4188ee5440cb95ea9bcaa81c62b8\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/fa6ef45affac4416836d258113809441 : SUCCESS in 30m 24s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/aaf32a4076ef4249a431f4d61a6d1b30 : SUCCESS in 31m 14s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/b50c750f1d63412e997f73672f221d15 : SUCCESS in 30m 11s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/ca8fe172cac14938ab08f305596221c8 : SUCCESS in 1h 16m 30s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b37dcf7a3a6d41f1adc07cd2e73c526e : SUCCESS in 30m 22s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8f383b7546654c0698e3e6f3139c4987 : SUCCESS in 35m 04s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/ea929c6d54d149648799db572e4eb728 : SUCCESS in 1h 24m 04s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/d8cb5b8b57384317b85abab74f266292 : SUCCESS in 1h 45m 06s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/bef7596b08694db5828ca878eade1f5f : SUCCESS in 1h 03m 55s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/8b6938aa50ac42ee92b994cd9f6ca563 : SUCCESS in 1h 30m 59s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3da693fc9c4a4e24ba82dea92784b545 : SUCCESS in 1h 20m 39s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/f53fb124a23b43339cb77c1350b27035 : SUCCESS in 1h 07m 35s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/b8b7fcd66fa54ba8b56d820dc45d3b0f : SUCCESS in 59m 08s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/e69c1e24656d462cae51611d99fefdcc : SUCCESS in 1h 26m 29s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/fba1a1a1acd549bdb7e37aa499e9a48f : SUCCESS in 2h 17m 32s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/21966ca55b5c4dac92ce3cd27d5359b8 : FAILURE in 1h 08m 41s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/24a449ade1d942b3b6a3972c187ab3d7 : FAILURE in 1h 08m 02s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/7eeb1df23f5e4ddb90f4e27db9ae870f : SUCCESS in 1h 00m 46s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/4ca4eaacd82a4757b63364ddffc8b85e : SUCCESS in 1h 20m 11s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/a44b84a2f29f49ae875cb4563f4a8eaa : SUCCESS in 1h 04m 37s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/5bd14758e81643daa45ae79322de3f6b : SUCCESS in 59m 28s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/4a58354269684d53bb14bae6e12bca38 : SUCCESS in 28m 18s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/a1b7af8994bf4c2693ec99d1dd78ddcc : SUCCESS in 32m 47s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/2e4e211638dc420d96003992aad56b4a : FAILURE in 30m 49s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/2074e38ef66c4cbfac9281a4d661b227 : SUCCESS in 43m 55s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/ad552c87f37e4cecb116068b04fda247 : SUCCESS in 1h 21m 55s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/9321c59e8f384c11b8c1990dd0398c0e : SUCCESS in 1h 12m 08s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/dece3e83aa4e46789c2387e599972fd4 : SUCCESS in 50m 36s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/ea76f7b6c9f04051bca15ceb2c75ab53 : SUCCESS in 1h 20m 07s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/9293ba0ba9444f2da45d09918f7d302c : FAILURE in 59m 20s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/9a16d9a7e5b54102b0487550fcd6387d : SUCCESS in 1h 54m 27s (non-voting)\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/20e82ee8870945ea9b643a0c15c5e325 : SUCCESS in 39m 44s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/648cb828bcef44f3832104611b277fea : SUCCESS in 55m 58s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/075adda274c842e48a8511973b8e9ee9 : SUCCESS in 36m 58s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/0f0b021af2e149c098fb06b6b89b2112 : SUCCESS in 57m 06s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/e54e27ae868248e8b73b597431a8e0c9 : SUCCESS in 43m 00s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/7b33b1bec4c343d8ac20c2b0dd302cb5 : SUCCESS in 55m 28s\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/d5fa447c647941479e607fbbd0d1c105 : FAILURE in 1h 13m 08s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/429c8e1821ac424ba434b602a09a7587 : FAILURE in 1h 26m 32s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/0eabc14baa684350bdcf601cca029a06 : SUCCESS in 1h 49m 17s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/136a2ea480a74f39aacf8ceec295d34e : SUCCESS in 2m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edda8d67253b423d8f4a28c462b8963d : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/05290044753f41a7af111537100cf1ef : SUCCESS in 3m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bc61a885d9f8475f8007554c9500d644 : SUCCESS in 3m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f0dec851d3e748f086b898214fbd5e2c : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7211edd2b1804738aef7c21a9175553d : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85e171e57306458887a405868656c79c : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee286693401f497590a79e3e2ecea830 : SUCCESS in 8m 40s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"49100a10984b21c52249420afd1b48d077543152","revisions":{"a9cc4b1b474883c6f449655ae0027878dc24e03b":{"kind":"REWORK","_number":1,"created":"2026-09-09 14:15:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/1004817/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/1004817/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/1004817/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/1004817/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/1004817/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/1004817/1"}}},"commit":{"parents":[{"commit":"3e55f21365156d67af32421dddee5e1d7d551acb","subject":"Drop support for CentOS Stream 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e55f21365156d67af32421dddee5e1d7d551acb"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-09-09 14:12:59.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-09-09 14:14:04.000000000","tz":120},"subject":"base: Drop redhat binary versionlock block","message":"base: Drop redhat binary versionlock block\n\nChange-Id: If7b5d045200d40dc9f49d26f2833af9a54f7ca96\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9cc4b1b474883c6f449655ae0027878dc24e03b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9cc4b1b474883c6f449655ae0027878dc24e03b"}]},"branch":"refs/heads/master"},"ff9241ac9d044d795039d7245578171fc35f38e9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-09 14:30:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/1004817/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/1004817/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/1004817/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/1004817/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/1004817/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/1004817/2"}}},"commit":{"parents":[{"commit":"87792b780f5817b311c7edf783a70b978f6b6dbf","subject":"Drop support for CentOS Stream 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/87792b780f5817b311c7edf783a70b978f6b6dbf"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-09-09 14:12:59.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-09-09 14:30:07.000000000","tz":0},"subject":"base: Drop redhat binary versionlock block","message":"base: Drop redhat binary versionlock block\n\nChange-Id: If7b5d045200d40dc9f49d26f2833af9a54f7ca96\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff9241ac9d044d795039d7245578171fc35f38e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff9241ac9d044d795039d7245578171fc35f38e9"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"3e55f21365156d67af32421dddee5e1d7d551acb","ours":"a9cc4b1b474883c6f449655ae0027878dc24e03b","theirs":"87792b780f5817b311c7edf783a70b978f6b6dbf","merge_strategy":"recursive","contains_conflicts":false}},"49100a10984b21c52249420afd1b48d077543152":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-09-10 11:20:25.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/1004817/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/1004817/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/1004817/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/1004817/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/1004817/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/1004817/3"}}},"commit":{"parents":[{"commit":"f1d7da742d058d893f6abf3c0a21efaf940187b4","subject":"Drop support for CentOS Stream 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1d7da742d058d893f6abf3c0a21efaf940187b4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-09-09 14:12:59.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-09-10 11:20:25.000000000","tz":0},"subject":"base: Drop redhat binary versionlock block","message":"base: Drop redhat binary versionlock block\n\nChange-Id: If7b5d045200d40dc9f49d26f2833af9a54f7ca96\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/49100a10984b21c52249420afd1b48d077543152"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/49100a10984b21c52249420afd1b48d077543152"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"87792b780f5817b311c7edf783a70b978f6b6dbf","ours":"ff9241ac9d044d795039d7245578171fc35f38e9","theirs":"f1d7da742d058d893f6abf3c0a21efaf940187b4","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
