)]}'
{"id":"openstack%2Fkolla~1005189","triplet_id":"openstack%2Fkolla~master~Ic188d411ed0dd36e1703c6ff4e657872703670a4","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2026-09-11 13:48:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic188d411ed0dd36e1703c6ff4e657872703670a4","subject":"CI: add letsencrypt scenario","status":"MERGED","created":"2026-09-11 07:52:24.000000000","updated":"2026-09-11 13:53:42.000000000","submitted":"2026-09-11 13:48:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005189","meta_rev_id":"7bacae05a1eb869d2c112dd3b6a227c9627e123e","_number":1005189,"virtual_id_number":1005189,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-11 13:48:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-09-11 10:41:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-09-11 09:23:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-09-11 10:41:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2026-09-11 10:41:52.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-11 08:02:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-11 08:51:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-11 09:23:51.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-09-11 10:41:52.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"43c0db904dd6e8c5d21b136b0f4623d7c7cd322e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-09-11 07:52:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4bf78e1a26f324e8513abb46496ba541a46ee7f6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 08:02:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f39e042ddb8e426eb831920ff1e69c44\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f0a22d29e5624f6fa27628e074c72fc2 : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4ac288128e9f45abae29fb26648dcdcd : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/de8829589d1849dba72cbec9d0280b96 : SUCCESS in 7m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2b366493529b44052b2239f788c1457c8ffda2fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 08:51:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e895672dfce4705bf0ca432cde7f6fe\n\n- kolla-ansible-debian-trixie-lets-encrypt https://zuul.opendev.org/t/openstack/build/c3838657d7b94232bd75758cd546ba22 : FAILURE in 56m 11s (non-voting)\n- kolla-ansible-ubuntu-noble-lets-encrypt https://zuul.opendev.org/t/openstack/build/47938b6c3502478792e60fa768058d4f : FAILURE in 54m 06s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4f4bb4fce23c476fa4761dd8a85919ca : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fcbfbd07eaf425d9e8b38d8b31b9b97 : SUCCESS in 6m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/71ee09ee993c4763931c3e2072bb1a5b : SUCCESS in 3m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/037f96f31e86406f8bd607319daa926f : SUCCESS in 3m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b27685f598b84f3382b5cff808b3bd0e : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d8a819868b84aafb4466db8e8daa1be : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23d7df7ddad54f66b4677ef98165d136 : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d04adf27e3a441ba047798c868c9227 : SUCCESS in 9m 45s","accounts_in_message":[],"_revision_number":1},{"id":"3e437eddd9d6594d9e79ccb2054217f02b8ee0d7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-11 09:23:51.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0d398d5e726c3c1a21ae99eb36e41fb5bd7cbee0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-09-11 10:41:52.000000000","message":"Patch Set 1: Backport-Candidate-1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dd46a0ef862dc89fb4bee5f1f77d7443ab3f1e20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 10:42:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c07dac5bf7f5b037ba21c7cd68278379ce5dd335","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 13:48:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8767ec19a1154147b0ae016735c96580\n\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/7db5816d7a27450bb31ea7492ec6a950 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e55c068cdeb47558bebc789a8d02da7 : SUCCESS in 5m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/544b170c22a94e919fe8d900d4e1ecf9 : SUCCESS in 3m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/33347ec3478a4675902be611dcd4186b : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb0c2cc008e040e0b2409654b09a8ec7 : SUCCESS in 6m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5058b6efd3b40f298b6db0aea339d00 : SUCCESS in 10m 53s","accounts_in_message":[],"_revision_number":1},{"id":"2ddabd9e5ada5d40e2d65d6bc75e20e95a63dbdc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 13:48:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7bacae05a1eb869d2c112dd3b6a227c9627e123e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 13:53:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b3a208be4c44310902545075fd0ea44\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7471ad8200b4bc5982082ee7aed4d34 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f584ba56bc5169655cece65995cd6d6dc7664cde","revisions":{"f584ba56bc5169655cece65995cd6d6dc7664cde":{"kind":"REWORK","_number":1,"created":"2026-09-11 07:52:24.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/89/1005189/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/89/1005189/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/1005189/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/1005189/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/1005189/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/89/1005189/1"}}},"commit":{"parents":[{"commit":"d9c5e083132c41fe03af0d256ae883ba1219d1c7","subject":"Updated from check-sources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d9c5e083132c41fe03af0d256ae883ba1219d1c7"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-09-11 07:50:16.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-09-11 07:50:16.000000000","tz":120},"subject":"CI: add letsencrypt scenario","message":"CI: add letsencrypt scenario\n\nChange-Id: Ic188d411ed0dd36e1703c6ff4e657872703670a4\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f584ba56bc5169655cece65995cd6d6dc7664cde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f584ba56bc5169655cece65995cd6d6dc7664cde"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
