)]}'
{"id":"openstack%2Fkolla~1005192","triplet_id":"openstack%2Fkolla~master~I4c481c0664de36171ae54efb304ee5790d428a6a","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"37203":{"account":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"last_update":"2026-09-16 00:32:36.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2026-09-16 00:32:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4c481c0664de36171ae54efb304ee5790d428a6a","subject":"Pin Lego to v4","status":"MERGED","created":"2026-09-11 08:29:35.000000000","updated":"2026-09-16 00:34:08.000000000","submitted":"2026-09-16 00:32:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005192","meta_rev_id":"d88f52e72689653900783f26f1027a311d3c3b25","_number":1005192,"virtual_id_number":1005192,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-16 00:32:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"all":[{"value":2,"date":"2026-09-15 15:53:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":2,"date":"2026-09-15 15:53:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-09-15 15:53:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-15 20:18:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":-1,"date":"2026-09-15 15:53:46.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-11 09:23:40.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-09-11 10:09:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-11 10:41:40.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-09-11 10:41:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-12 09:23:27.000000000","updated_by":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"reviewer":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"state":"REVIEWER"}],"messages":[{"id":"6bdc0682269c934a0f6d439ed7ec6ef7b857a76b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-09-11 08:29:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f9436b4e7b7966108a8505b02eaf681732bf8f59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-09-11 08:29:54.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"1378d28a29db31a8bbbec7a625c64bf3eedb1eec","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-11 09:23:40.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"6c362c99198d515b35c31ac5fb53f2b4b78c7c62","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 10:09:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f520803b6f84d1bbe9f9c9907a8df7b\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/be242cefef9c40f7a57cd8e710354b20 : SUCCESS in 46m 04s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/9241893510714da7aa8f9d849cf46f35 : SUCCESS in 58m 56s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/980be1ac7cca4666882e5de91e8b23e6 : SUCCESS in 48m 02s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/a60b50bb8bee4637b708cbc5a4729659 : SUCCESS in 51m 55s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/b62b0479267841978363415a8f7e40d1 : FAILURE in 1h 22m 27s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ec898b62f5e3464db45656f303d7966f : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/187ea7d4d6a24bdba131c210f5b04547 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/090460200df84b0fa0f888f87d1bd503 : SUCCESS in 6m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"661ce7859b240edd0ceee0e8a98a6854f0a7f233","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-09-11 10:41:40.000000000","message":"Patch Set 2: Backport-Candidate-1 Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ffca3512df0b401a8d731e2a3a6688df2af6603b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 10:41:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99a2f52ff97e482cb44d8e5343eddf57\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/e2820e89f768479bb211a638bc75b09b : SUCCESS in 29m 28s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/800d8a6a08ed4898a71a15c7180e9735 : SUCCESS in 17m 17s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/e2026794a8ad4a328cb62459b47dedd4 : SUCCESS in 22m 27s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/c1427508de6849a1854dc21210fe68cb : SUCCESS in 28m 57s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/02d40bc3881b41feaeeef5c8dc75cca8 : SUCCESS in 48m 10s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ef5cbc8cf9784f06aaf0c345c100a6a8 : SUCCESS in 24m 27s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/61002d307dfe4232bb5b953695205da7 : SUCCESS in 33m 55s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/ee9dbf80742c4d57be87c8142af00bd2 : SUCCESS in 1h 16m 43s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/b846c0fe2c0f4b08a71ba4171d7d45ba : SUCCESS in 2h 01m 37s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/5cec7371da734878bf632db1001f8f83 : SUCCESS in 1h 03m 33s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/0870def9542440ada19c09d1c20e5d3a : SUCCESS in 1h 27m 54s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e266cd5e86d42a7a905144cfd97ae8c : SUCCESS in 1h 21m 20s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/9e544af5a6044eca9bddc1e1e178505e : SUCCESS in 2h 04m 14s\n- kolla-ansible-debian-trixie-lets-encrypt https://zuul.opendev.org/t/openstack/build/b143636bd72846d78459bf1cd734d906 : SUCCESS in 1h 28m 46s (non-voting)\n- kolla-ansible-ubuntu-noble-lets-encrypt https://zuul.opendev.org/t/openstack/build/3f348c5ce2cc47a4a6bad168dcde5378 : SUCCESS in 1h 33m 43s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/73d84398449043dca6a7e2679394179e : SUCCESS in 43m 12s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/6305120484a540fdbc8e7447f8fc398e : SUCCESS in 26m 00s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/a1857aa7707042388d837fd2eac97f21 : SUCCESS in 28m 55s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/5862b241de83458fb8a4df2bbd288154 : SUCCESS in 33m 21s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/84efd04696bc4ae591e21caae7b22969 : SUCCESS in 1h 01m 33s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/e2d4aa9b87994482a37f2ed908ca0913 : SUCCESS in 26m 33s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/8ea474d90aff471fa554dceeb8d37a1e : SUCCESS in 1h 04m 11s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/4575eb88b5ff4e39aeae26841fe483fd : SUCCESS in 40m 20s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/a4b831ed6165419e95ef0657aaab2319 : SUCCESS in 55m 48s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c1061bf9a4a6458399db5e67aeb3941d : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9654974e9e0944a38f9f7969828d498a : SUCCESS in 4m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/68ff82f488e84fd58d064271f29ecbc0 : SUCCESS in 5m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8201ca514e924cd38887673fcae9577b : SUCCESS in 3m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/81b8bfedfe8c40f69771f9b935439815 : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2bcf3614c0954c0c8568b6c8ea940c15 : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c5db24992b64b52b5868b74226a33c8 : SUCCESS in 6m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d39e031025344ecba67c39b6192fb60 : SUCCESS in 10m 05s","accounts_in_message":[],"_revision_number":2},{"id":"7c8753bad4fa1e1f19344c47271633dcec7cdfa2","author":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"date":"2026-09-12 09:23:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"90ccd7e2ed15cb09d7160311b089f03437e47d27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 09:26:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4dfed05b0c1fcb67a5c053f904551873ede84330","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 11:11:53.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0d3e05fb9ec407dbead49ef2ef0a10d\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/a3fca4b3ef034160a500fab5216417f8 : FAILURE in 18m 33s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/c26677ff2f48468f83a8b984372bf099 : FAILURE in 28m 38s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/4e2ab06669a34e3d9c43e27936ec0a38 : SUCCESS in 24m 42s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/1daf2438cfe54111bf6c786fa70794d4 : SUCCESS in 47m 49s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2791b03bfb7549daab033fe13147678f : SUCCESS in 27m 52s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cb52f769a7b64200832df0a016eb0052 : SUCCESS in 24m 57s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/5d35275e9dde412792ce797501addf17 : SUCCESS in 1h 23m 59s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/a262dccda4e74248822a8522b1dd98f4 : SUCCESS in 1h 44m 41s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/b2c5ec2fb32048caa045c393c3f7194d : SUCCESS in 1h 06m 34s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/7ab4fa5b5c214c54833d771f0262d018 : SUCCESS in 1h 26m 56s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/87a35a3f0b164eb7a87033cf2b1c7a59 : SUCCESS in 1h 14m 04s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/c1d352a2856649ae9c7d54727a99b6b9 : FAILURE in 11m 04s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/45aaa3212478496b89528e234e416f84 : SUCCESS in 1m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f2dd4b1adca4373993e4a563b8fc7e2 : SUCCESS in 4m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cb51fa2867774c4091ce946aa6326b95 : SUCCESS in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/901ba348622b4141b55cfb80d418c74e : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6fcb64d1ac0484a8c375e883c1b1a57 : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c078a5a89dd84777b6175dbcfe193a13 : SUCCESS in 8m 44s","accounts_in_message":[],"_revision_number":2},{"id":"052443faf840a655a7422e625a3a2fa55fd6ab27","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-09-12 11:28:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"53cd1acfd942d2668cc8462826a4e72e2b981c27","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 12:20:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6799f5454c1d494bae4630c27de437d7\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/3f45107ac36a48d0b83ca157c5883022 : FAILURE in 34m 08s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/b6dc0a348a644309a1a1283296a077c8 : SUCCESS in 42m 47s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/c75b7135531c4362aa7f0589a1fb18ab : SUCCESS in 30m 14s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/b0c29662dfa4450d99da989752179c14 : FAILURE in 48m 43s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b1eb166550904fc19dd5982c89e643ac : SUCCESS in 3m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0971c95ca4b74fd8ba30ecdb9a75f430 : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/34032cbacd674c3da865e82a1d4e03b0 : SUCCESS in 5m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f7578405b6971c808096f29aefbdfd7227cbf6dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 13:36:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d4fe6c5e088447aaeb03ce4de675563\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/53fa1a9af77f4db19454da9eeb47e3ce : FAILURE in 18m 03s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/08357686ed4946df87611106dbb5c8b5 : FAILURE in 28m 31s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/85be1dd299a344c29e4b71efe62df52c : SUCCESS in 19m 07s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/7a7dd599a3934a5d83730e54773ed49c : SUCCESS in 41m 21s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/96e410919bd84409ad531600f5314113 : SUCCESS in 30m 06s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c7c9b62b0b7f4f829d097dd72df7dac5 : SUCCESS in 32m 24s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/a0639738e07b4b6495c0ae82dd934efb : SUCCESS in 1h 07m 39s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/aaf21a5520174b4f81dcff38c92f73dd : SUCCESS in 1h 57m 12s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/fb3bfd1323d84e9684597761d14ecbaf : SUCCESS in 1h 02m 04s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/fb0ee5fa3d384d8bb83b7d56664ecb71 : SUCCESS in 1h 26m 57s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/858052401a6b4efdb171f82580389282 : SUCCESS in 1h 21m 52s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/1b9e5f9adb9b459a9a9b79ad272f14e4 : SUCCESS in 2h 05m 37s\n- kolla-ansible-debian-trixie-lets-encrypt https://zuul.opendev.org/t/openstack/build/12b08346ad2c4dd6b1e5878d3dcc4edd : SUCCESS in 1h 26m 34s (non-voting)\n- kolla-ansible-ubuntu-noble-lets-encrypt https://zuul.opendev.org/t/openstack/build/b66e024bc52c403ea8d7d3f67bf161cd : SUCCESS in 1h 28m 39s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/a2981960520f4512a5e783d3fcc78ea4 : SUCCESS in 40m 15s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/f6ede6dcefee4e5bb7ef7b87c0aa9666 : SUCCESS in 24m 40s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/b94018200a5f4f1d9c56520db3af21f4 : SUCCESS in 26m 09s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/984a02ea3bfa46e08ff02c50d0afbf2d : SUCCESS in 30m 27s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/70accb6b45bf4fc4b110d189785e1cea : SUCCESS in 30m 52s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/6cd68c17aed3434b920c710d9293f3c2 : SUCCESS in 25m 00s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/509013e20931402085cf6667c8a9a7b2 : SUCCESS in 49m 34s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/9a33ce778a2a4261a00d1b7e471eca00 : SUCCESS in 38m 49s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/32a07e20dfbe4e16bcbb77f346b35244 : SUCCESS in 51m 33s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e99bfc6701974b5aa4be739f00c374e8 : SUCCESS in 2m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/366e2319ba3d4f2b9c1167ab8e9eb322 : SUCCESS in 3m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/acc0a498390344579d99cd2af403fa7e : SUCCESS in 4m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1ff133cae1f94ca79c91c60f6bfec221 : SUCCESS in 2m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/363f77b231cc4351a71172824b39a228 : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74135eff2e9142d5a82546f5b1753619 : SUCCESS in 2m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a80a85bf270145dfb83e1f73fb248c36 : SUCCESS in 4m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d85e525e3e8c40d0b10c2ce1088536d9 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":2},{"id":"f9b0f8e5b84d80efdcf35db7931016dc73fc4c23","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-09-14 04:46:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"05393760a3fd27f67b27e5e823957d53f2c77f62","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 06:08:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e538ef2ce524c77bc92f50093be3c11\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/fb5f17841ee54d32b6f0d854a7984120 : FAILURE in 40m 19s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/343c6f71a4c24ac8bb0cfa8d7b6808fe : SUCCESS in 33m 18s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/912a1f948924481e94e901a1faf4f4ae : SUCCESS in 36m 36s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/49a7ed4d054f4a85a7d734beda5f43fe : FAILURE in 1h 19m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a06227a867c249b99284184aea46cac9 : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/94c1f1060dc44e05ab47f3f3ee3b8c5d : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0cf155a228e840e6976d2d2a14cec604 : SUCCESS in 8m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b89516c1f36a53d61d058f4b1c914af2748d96aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 06:48:42.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/353e63cff0314910b7f4bf460939b847\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/21c8aa836e774090abb0f5c71a2834f3 : FAILURE in 26m 17s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/188bc80ef6c646ca8e7019e25d911dc8 : FAILURE in 27m 07s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/79c64289a2ae4c35ae3631bc3dbd1023 : SUCCESS in 20m 40s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/8bcc0372e25e4d27b272ee980af29189 : SUCCESS in 1h 10m 00s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5466a0a89c0444f19a3a1187cd5927fb : SUCCESS in 18m 44s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d46750ebbb454257befb73d58d6c00f7 : SUCCESS in 17m 47s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/8150794aac5c4a5381409db93827d05e : SUCCESS in 1h 26m 02s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/ab4c31745c3d4f649e13f242135dc35a : SUCCESS in 1h 59m 29s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/bfd790b5902042b6b197d0df46c1e70a : SUCCESS in 1h 07m 29s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/fc2d85d5b6764c539453301dcf45389c : SUCCESS in 1h 31m 37s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/97167e089ff64aeb82e7971e1656c10c : SUCCESS in 1h 21m 58s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/4363f17baeb64f77a1d76a957d978df4 : SUCCESS in 1h 59m 25s\n- kolla-ansible-debian-trixie-lets-encrypt https://zuul.opendev.org/t/openstack/build/2e9bc1f6ba344238b7aa161f8a2a83de : FAILURE in 1h 21m 41s (non-voting)\n- kolla-ansible-ubuntu-noble-lets-encrypt https://zuul.opendev.org/t/openstack/build/5364c799b45a48f19c06542024622b07 : SUCCESS in 1h 36m 57s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/846aa7bacccf4fd6bdc04319683bca5c : SUCCESS in 36m 47s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/591856e8053b40b498c663820bed061f : SUCCESS in 31m 43s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/c03dd547d6e64d65b7f87d26e416fd93 : SUCCESS in 20m 36s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/2d8c2df724934e7dbc2dbd4d5459aeb1 : SUCCESS in 39m 47s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/e6ca8b5c6ce94384a0516be4fe674d69 : SUCCESS in 34m 16s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/f5c9c7c336ca4fd2a5297c6629691fd6 : SUCCESS in 50m 03s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/b2fcf57e61fe42098878fb1e29627123 : SUCCESS in 35m 49s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/8816fb48d8e747e2901926404c23f310 : SUCCESS in 21m 06s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/3a96abad8f034462b17ce8fb8205d658 : SUCCESS in 30m 51s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c296c9a1835e44ed83e4fc540aa33fad : SUCCESS in 2m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa64c8143a034563b1098dffa057e179 : SUCCESS in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/44553a86846c45da825abe6c0cb8d998 : SUCCESS in 2m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/46ffa5144f6748ac88bbdc0d64ca77e5 : SUCCESS in 2m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dff20be9c0bd466d9de8f5435a7b049d : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f115af7dbc154b749105c89881c9eb98 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6129494e48243299f5876361d6d6b0c : SUCCESS in 3m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d3873856e5b416ead5bc05b3d7e059a : SUCCESS in 10m 45s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/kolla-ansible/tests/pre.yml for job kolla-ansible-debian-trixie-lets-encrypt.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/kolla-ansible/tests/run.yml for job kolla-ansible-debian-trixie-lets-encrypt.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/kolla-ansible/tests/post.yml for job kolla-ansible-debian-trixie-lets-encrypt.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job kolla-ansible-debian-trixie-lets-encrypt.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job kolla-ansible-debian-trixie-lets-encrypt.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"7e5e166f47772a255b55f0c4b1652e003df0e588","author":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"date":"2026-09-15 08:43:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"00280ef7c69a4b3e33bb2f4860669828035a5e80","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 09:55:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/048916c135344f429c3105df9422a664\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/d5fe4d5b44ea4402a2dbfb48b2520ae0 : FAILURE in 52m 14s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/5ea9eae5ee30433e898d68627e80d4f7 : SUCCESS in 50m 58s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/b177640e3a6c42bfbb80881b0febc7a6 : SUCCESS in 54m 58s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/9cd7330cce054c9ba7f7800d307ff10a : FAILURE in 1h 08m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5724de57727040c3b373543395eb5933 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3a5d4b850b45423a84d53fb3a094945e : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/247003d8b6dd46bd95094ce1ee8425c8 : SUCCESS in 8m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"860fb24b1cde4af79fcb5879e3c8e4290f0f93f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 11:08:16.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7fece8cc50544e5c9e0438639686b5f5\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/776660fb8d3645e4985a4972c686be5e : FAILURE in 28m 33s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/bc1f2ad5f9de4aa2aaa701fb59d3f76d : FAILURE in 28m 30s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/a0dfe015d7fb476c8767009ca8614375 : SUCCESS in 25m 43s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/be6f991e88a341bd8d72c8ad796c81eb : SUCCESS in 47m 13s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6ccf9f241bf44df7b642fc21422ad360 : SUCCESS in 28m 58s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7adada5db15a4dd698bd18ce8ce2fe1a : SUCCESS in 32m 36s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/fd9941da1cac4769bf89b2999e90e69b : SUCCESS in 1h 14m 53s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/7d302d66f105454dac7b9989c25cc7d5 : SUCCESS in 2h 19m 09s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/8b907c18394945c5a9cf7054448e1412 : SUCCESS in 1h 11m 04s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/bbbd36e19fb94e76b6d28f09543ce0e9 : SUCCESS in 1h 39m 39s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/58a6173a3a9d495ab5814668563ccf6e : SUCCESS in 1h 23m 23s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/ad755f74ab9643bca21be4bdc7226362 : SUCCESS in 2h 07m 57s\n- kolla-ansible-debian-trixie-lets-encrypt https://zuul.opendev.org/t/openstack/build/4ef8d6dba962424bac5c6ce4a9c3e49d : FAILURE in 1h 20m 25s (non-voting)\n- kolla-ansible-ubuntu-noble-lets-encrypt https://zuul.opendev.org/t/openstack/build/742c53b213c146509ad1994b5a14482d : SUCCESS in 1h 37m 39s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/85f4930aa634499795912506450a78d2 : SUCCESS in 55m 31s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/92648e52048743e992172f1bea649c39 : SUCCESS in 34m 48s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/fd81c75cb33846aaa988f899b36688c7 : SUCCESS in 45m 40s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/e01cbb3c26cb4a12b6941a547b7fd072 : SUCCESS in 31m 48s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/868fb67edf374599a44f090136c07c3e : SUCCESS in 46m 09s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/49126e3bc8d34f41a999841611014455 : SUCCESS in 53m 45s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/31f044c6052147d689dfc42e0678ff4e : SUCCESS in 1h 05m 46s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/0f165c73079c43d6816aae006ebcec39 : SUCCESS in 48m 58s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/f332aa2f3f704fdaa338861cdeb138ba : SUCCESS in 1h 16m 19s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/8e19b5cad4a94be78b777d6d40764de3 : SUCCESS in 1m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cd97beab49643788ef614b3450238b2 : SUCCESS in 5m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/93dd1849b799431d846f845f94ea2e68 : SUCCESS in 3m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/08556da0d49b4bfb9f8ca0050b17edb0 : SUCCESS in 3m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/50832ec041724c22bd431f5f44cdcf0d : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96ada710259d4f2387a4b19ac91a48b6 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af76ef5cc4674bfc846137682254d354 : SUCCESS in 3m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08754912b2fe4782b21c3a23c4537b28 : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":2},{"id":"6f0f4efd20e94777427e73e2b538fd4b8ca18b06","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-09-15 15:52:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3948d08e5ae6fd50fb64e7707b99a4604a1576de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"real_author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-09-15 15:53:46.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_23084\u003e\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"_revision_number":3},{"id":"337f446bb19f5567f49d16b087a96c6fdcf947e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 17:58:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc6e7ae289ab45c984bdd01c17d08971\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/fdb85cafa020457e8d7112389a9f8f14 : FAILURE in 35m 55s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/15f826197eac4b9c9f0e27e2b0755c6e : SUCCESS in 44m 37s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/fc4572a4d8a14850bd412db34d9ca942 : SUCCESS in 32m 07s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/7e570cfd678945b2bf2d09f974563d03 : FAILURE in 1h 05m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4c1f5d9295024f5e9c354025a05395c5 : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/df0245a69dad477f988129ad7a5dc26b : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4f51db9452e5476492a529cec50d11a0 : SUCCESS in 9m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"468322a4f678fa5cdc4c0217daff4e645a520460","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 18:17:34.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/870eef4217734c4ca27d01629d8399a0\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/3a39f4057cb14af4b24b86e5d62d202c : SUCCESS in 32m 10s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/bdfb63598946437cbca1d7b5f1906a4b : FAILURE in 8m 23s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/e3ab83fd85cc4d41bdfdabc87c854634 : SUCCESS in 32m 46s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/c0330b002bf646f3af05f91e32df8087 : SUCCESS in 47m 26s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c0f7287ad8e4f399dd141de28b44bda : SUCCESS in 34m 40s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7d2845a8235b48d9ac97b5adbb5c07ea : SUCCESS in 33m 51s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/c7c4660f56b24988afa8fa2dd8a977f2 : FAILURE in 27m 49s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/162997b0cec1439cbcdf9bd19eef659c : SUCCESS in 2h 06m 50s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/bd2848b527024a39a84148ac2bd7f2e7 : SUCCESS in 1h 13m 51s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/2a204937ec4048bc9b26b0958e634f75 : SUCCESS in 1h 41m 19s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb448414a9be43c8bfb93e47da51b3f0 : FAILURE in 20m 11s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/6a0c8b31be4d43b3a5e971bbe69fa2e2 : SUCCESS in 2h 09m 49s\n- kolla-ansible-debian-trixie-lets-encrypt https://zuul.opendev.org/t/openstack/build/d448d678f5e14cf699b1013fe982e92b : SUCCESS in 1h 55m 10s (non-voting)\n- kolla-ansible-ubuntu-noble-lets-encrypt https://zuul.opendev.org/t/openstack/build/3dcb1e1495ad4cb7857bfd83213ddc50 : SUCCESS in 1h 52m 03s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/333df4a4aecd4d7b9b06ccc674ad9bcd : FAILURE in 36m 08s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/0415973679564de7aaa708500cd2c0a6 : SUCCESS in 41m 25s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/a7a75331c36b45909c59aa4bd4822be7 : SUCCESS in 45m 17s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/df813fb5b9af4311b13d9519da969381 : SUCCESS in 35m 38s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/12a3b442b16f42949909206d874afd05 : SUCCESS in 1h 10m 40s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/bc2ee000128647d9837149a9a0f592cb : SUCCESS in 32m 39s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/44c1a3bc9f0b4da89cd525f387ac50d8 : SUCCESS in 49m 13s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/030eeda6c8f64dbb80914795f4287aeb : SUCCESS in 53m 38s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/e0b8a8a71eb544c8ac655d5a9360be64 : SUCCESS in 1h 13m 07s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/dd6dca5d781d40a3b5847c908c1f798f : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b19b03e749794bb0a7e65c6f6425d538 : SUCCESS in 9m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/02924a476da14104a31cdbf23a6eff58 : SUCCESS in 3m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0ae04c2c3bca42978915954256504c77 : SUCCESS in 2m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c80dcf14507042b9992268c1c82e1a46 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a3a16412de34fa5a1e6b49871dfed70 : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98b523dfc23b4b2c93e062cfb068fbd2 : SUCCESS in 4m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f61bd687de744559ada5564bcb2875b : SUCCESS in 6m 53s","accounts_in_message":[],"_revision_number":3},{"id":"13efd5c376d91759c7a109eba02791786e6c290e","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-09-15 18:29:39.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"486e93cc89a168715931238b4743d7f853b3001f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 19:35:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27c1727cf11d4365988767fda3888686\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/5641856d41334a4da9918d32b19ca5f8 : SUCCESS in 42m 56s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/31bd9e86df1241b3bf685c473e960edc : SUCCESS in 51m 14s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/0b707afc182443928557c88185b5f8be : SUCCESS in 40m 59s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/7955c862ee354e1ab4c68ab2d96ec3e8 : FAILURE in 1h 02m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/885867a4f9c94ca49468772107cd0783 : SUCCESS in 3m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2ba7509c796a49648fa31eacadb748a7 : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/845ce1727b664eab9b717dfc2bd5088d : SUCCESS in 8m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6a200fd62662d8cf7c16d0c8dcef05e0380807d4","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-09-15 20:18:18.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a50a21a14cdfa7febdb6a08720318a73857cb12e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 20:20:59.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/90255b2fb8e54759869cb9846ffd1e18\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/7b80d93ebeda4a539afb0d1c188c0ec6 : SUCCESS in 18m 53s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/2a90fd7c37264d7e9a045e97ffa2100a : SUCCESS in 20m 37s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/374f4de9fc574ed7a513ebbc7a97c785 : SUCCESS in 31m 45s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/f20a90b2417d41998a3f04011d943801 : SUCCESS in 44m 30s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/12c73c81b0e147bfab4fc3f2551fc4f1 : SUCCESS in 20m 05s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b8ab9b12335f42fe8a1f50aa366139e8 : SUCCESS in 20m 08s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/e8a996e505914e05a3fb852bc1d9367c : SUCCESS in 1h 24m 01s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/732eeb01ae534e71ac51b0640a1f5847 : FAILURE in 1h 22m 05s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/2fee5e07a37f4eabbb56337620b85d1d : SUCCESS in 1h 10m 56s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/51cc31f14563425bb4dbf452e247e050 : SUCCESS in 1h 39m 50s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4bca5917d0f94b8193d41027cdc163c9 : SUCCESS in 1h 32m 23s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/6e89cc87af494102afd5aa8163c69d27 : SUCCESS in 1h 48m 37s\n- kolla-ansible-debian-trixie-lets-encrypt https://zuul.opendev.org/t/openstack/build/27b8fd93dac1489496e6300e686ab1fa : SUCCESS in 1h 40m 04s (non-voting)\n- kolla-ansible-ubuntu-noble-lets-encrypt https://zuul.opendev.org/t/openstack/build/b766b75583fc4908b97c4583befc78a9 : SUCCESS in 1h 42m 10s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/dd205e4358ec42609294cfa65a926196 : SUCCESS in 57m 03s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/2d52bb54a00b499b9441cc7415a8598a : SUCCESS in 35m 35s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/13c7fad06dd146e5b20342c68deb1ad7 : SUCCESS in 34m 08s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/f36360a781bc4664a4da629617dd7dd2 : SUCCESS in 47m 04s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/382601d5eb354bfca962ab956366ee65 : FAILURE in 30m 08s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/980835e404614da387d079e8c1ac3824 : SUCCESS in 36m 58s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/cba75d53fff744fba8e39678ad116359 : SUCCESS in 43m 26s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/5fd0bd101ffb418b94d2fe7be2683e4d : SUCCESS in 32m 00s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/c1025f94051341f9ac6b40282804925d : SUCCESS in 1h 05m 08s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/1f7f41b2b39c485b97626a68ef6c2058 : SUCCESS in 2m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4be803213da846b8bb7d0f4b6287a993 : SUCCESS in 9m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a84b5660b333464393a59e0bd34dbb10 : SUCCESS in 2m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/38e7be6642464c699c71d82289623604 : SUCCESS in 3m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1e3becbc051242fcbad1492d4e942813 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ed0498ab18942198062f99458e479a4 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/413a520a64984972a37934325767ca17 : SUCCESS in 6m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a6261c7dd434b5a8b1a37d9a1a9253d : SUCCESS in 10m 33s","accounts_in_message":[],"_revision_number":3},{"id":"02bc05300fd0e7b759d974bdefd036f4e7cde1d2","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-09-15 20:25:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f913d5fae9b4966a7104fb3d742ce840067a9f9d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 21:48:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1d16695d0034016bad8d9b03d176e9b\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/cd6ef7136a114df49c2f94df05153caa : SUCCESS in 33m 05s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/ed1be0bdee934ef080bb90a186946e4a : SUCCESS in 46m 22s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/edc2d30f41ae4668894c552c7df13bca : SUCCESS in 33m 35s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/06d511d60faf4820bcbebb06be34df02 : FAILURE in 1h 20m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/eeb0e94838d94ecf9dde1f1d125d2f78 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/888141afcffa492596b3ce78aa7abfdd : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/dcbeeb5bffc7443c9e7759526d880977 : SUCCESS in 9m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b53d1f6c339aaf3a81ef95e4158490f4391a4c00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 22:39:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0e170768dca4f8bb28d71cc0a37578a\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/b201169d5e2f478c94ca53a0e03518fa : SUCCESS in 18m 50s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/f6e96475c7364c5dbee6bc2d22ca69fb : SUCCESS in 22m 36s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/8a3c157cea704977addd338b852263dd : SUCCESS in 20m 45s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/adea5e37808b476cb593e57d4155dee7 : SUCCESS in 57m 54s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c59ba5bdf550426792073729dc17401f : SUCCESS in 30m 55s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/66156e00c1584a8a9b81a449456d82a0 : SUCCESS in 18m 29s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/8dd0024d002942e58fb1e687da3743fd : FAILURE in 15m 01s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/0cea440d04a14107a44f955f94f3d3e4 : SUCCESS in 2h 11m 56s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/743c848236ba48bf8c40872648e837f5 : SUCCESS in 1h 06m 56s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/704a31a836f34524852964a337181d74 : SUCCESS in 1h 34m 04s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bbbbe24e7ce8431c91373bcbfcd5d757 : SUCCESS in 1h 26m 29s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/26486ce412d945ba9b0df1d0c8cb566d : SUCCESS in 1h 50m 17s\n- kolla-ansible-debian-trixie-lets-encrypt https://zuul.opendev.org/t/openstack/build/a87fc71f0ba943d59030f8ae60ee8507 : SUCCESS in 1h 40m 47s (non-voting)\n- kolla-ansible-ubuntu-noble-lets-encrypt https://zuul.opendev.org/t/openstack/build/5ba9f336916542869faedfd86a8a9ad5 : SUCCESS in 1h 29m 24s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/f38f5322a20a442f96daf4e10141f86f : SUCCESS in 46m 01s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/bd9f620af6cf4be29e9d53c9ecdefa5b : SUCCESS in 30m 20s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/d5244aa879104afe918582357a99adb9 : SUCCESS in 24m 05s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/227a51667504435c9e726da7614b7174 : SUCCESS in 43m 41s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/f466754f6b964b47b8f8646ce0fecd1e : SUCCESS in 1h 06m 27s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/0b371cb228ea4c37be33a1988a2dbb34 : SUCCESS in 26m 32s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/00c2d578b3ef40d1a7ff5906bb40a827 : SUCCESS in 55m 23s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/9b311bd9b095401fa08583cbf996236d : SUCCESS in 50m 41s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/be7e825097ca4d28ae8877ff5d6ed097 : SUCCESS in 1h 07m 32s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/3276bb6bd41a47e9a79b1a930be2e579 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64ad58155ccd41c5b4f61265d02bb65a : SUCCESS in 6m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1c1b464cc09f4f989623c85b5a53b681 : SUCCESS in 2m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/784abfe21b2a419c8326ed31e94c3639 : SUCCESS in 3m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5926d3d746cc4b988ec798aa6dde436c : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1bd3733cbd914099bb3c515181f3b4a8 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3145a08b1987470883bd258ee48b5812 : SUCCESS in 6m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45835c21dca1416f8278aafcabbfe62e : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":3},{"id":"95fcba346ad8850e94cb03fe7196ddf36f814263","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 22:41:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5c71dcc7d9909327a36553dd95893b92c3499729","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 00:32:36.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70680608fbaa4d6b9e5f39d28abc2e04\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/8a4e37867c164f10a2ae234c0a7c80f2 : SUCCESS in 35m 21s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/f37803b54fa14d6a84f5c81ad9796d9c : SUCCESS in 39m 33s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/4c4c07698c324e4da33989c763ba26df : SUCCESS in 21m 46s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/93cdff660b074dcdb0f45766cd39f2f1 : SUCCESS in 45m 37s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf1f9f24c4be410bbbc4f7b4bbe927a5 : SUCCESS in 19m 06s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d9cd2b31696a4d51bd0439acb2a96a1f : SUCCESS in 16m 20s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/f7ff73ff6b8d4ee4be1bac4b91891266 : FAILURE in 16m 02s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/a361c42fe3b54ad2bdcb7b52af6dce35 : FAILURE in 1h 07m 42s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/26d5511c09c9472b9d57191d0dffbaf8 : SUCCESS in 1h 06m 11s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/af83bf5479294985bd95d5e1347a6ee0 : SUCCESS in 1h 32m 31s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9ea2c4414fb2463aaf4ea2400b12b4c8 : SUCCESS in 1h 25m 31s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/418a4f44bd8544e2bb81d4f2f4285d8c : SUCCESS in 1h 50m 38s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/24822ee5b5cc465d94289fba04d132ca : SUCCESS in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8f173b202b74a72b3ce4c0a472b5482 : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c8dfbab1238048a0b5dec6860bc5ce97 : SUCCESS in 2m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f87bc7496cbc418e93a5dd1e3a25babe : SUCCESS in 2m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/490669f657f240198a3386e4ea362f6e : SUCCESS in 6m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f23ae091dd0439eacb0f9b7caa4b70e : SUCCESS in 10m 25s","accounts_in_message":[],"_revision_number":3},{"id":"6e88c4f50e989277f0427afb8e6e02ba6def8e1d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 00:32:36.000000000","message":"Change has been successfully merged\n\n2 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":3},{"id":"d88f52e72689653900783f26f1027a311d3c3b25","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 00:34:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86c44a86e6a44ad88ec2eb1ba26d935e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cefa575544a241848d8f79e4f2d3f3f1 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bfccf861b1c12716b99d345a143d10b5ea74805f","revisions":{"7ea8a1e52e9c4cf416321a87167b7386e69152be":{"kind":"REWORK","_number":1,"created":"2026-09-11 08:29:35.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/92/1005192/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/92/1005192/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/1005192/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/1005192/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/1005192/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/92/1005192/1"}}},"commit":{"parents":[{"commit":"d9c5e083132c41fe03af0d256ae883ba1219d1c7","subject":"Updated from check-sources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d9c5e083132c41fe03af0d256ae883ba1219d1c7"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-09-11 08:27:12.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-09-11 08:27:12.000000000","tz":120},"subject":"Pin Lego to v4","message":"Pin Lego to v4\n\nLego v5 changes the CLI and storage layout [1], breaking the current\ncertificate scripts. Migration is not trivial.\nUse lts_branch to stay on v4 for now.\n\n[1] https://go-acme.github.io/lego/migration/index.html\n\nChange-Id: I4c481c0664de36171ae54efb304ee5790d428a6a\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7ea8a1e52e9c4cf416321a87167b7386e69152be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7ea8a1e52e9c4cf416321a87167b7386e69152be"}]},"branch":"refs/heads/master"},"8cc65e0cef3c567de69a1117db406879c9829be0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-11 08:29:54.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/92/1005192/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/92/1005192/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/1005192/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/1005192/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/1005192/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/92/1005192/2"}}},"commit":{"parents":[{"commit":"f584ba56bc5169655cece65995cd6d6dc7664cde","subject":"CI: add letsencrypt scenario","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f584ba56bc5169655cece65995cd6d6dc7664cde"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-09-11 08:27:12.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-09-11 08:29:54.000000000","tz":0},"subject":"Pin Lego to v4","message":"Pin Lego to v4\n\nLego v5 changes the CLI and storage layout [1], breaking the current\ncertificate scripts. Migration is not trivial.\nUse lts_branch to stay on v4 for now.\n\n[1] https://go-acme.github.io/lego/migration/index.html\n\nChange-Id: I4c481c0664de36171ae54efb304ee5790d428a6a\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8cc65e0cef3c567de69a1117db406879c9829be0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8cc65e0cef3c567de69a1117db406879c9829be0"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"d9c5e083132c41fe03af0d256ae883ba1219d1c7","ours":"7ea8a1e52e9c4cf416321a87167b7386e69152be","theirs":"f584ba56bc5169655cece65995cd6d6dc7664cde","merge_strategy":"recursive","contains_conflicts":false}},"bfccf861b1c12716b99d345a143d10b5ea74805f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-09-15 15:53:46.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"real_uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/92/1005192/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/92/1005192/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/1005192/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/1005192/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/1005192/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/92/1005192/3"}}},"commit":{"parents":[{"commit":"58b01f5a162bae158632765a81dea92e8545e06a","subject":"Merge \"CI: add letsencrypt scenario\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/58b01f5a162bae158632765a81dea92e8545e06a"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-09-11 08:27:12.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-09-15 15:53:46.000000000","tz":0},"subject":"Pin Lego to v4","message":"Pin Lego to v4\n\nLego v5 changes the CLI and storage layout [1], breaking the current\ncertificate scripts. Migration is not trivial.\nUse lts_branch to stay on v4 for now.\n\n[1] https://go-acme.github.io/lego/migration/index.html\n\nChange-Id: I4c481c0664de36171ae54efb304ee5790d428a6a\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bfccf861b1c12716b99d345a143d10b5ea74805f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bfccf861b1c12716b99d345a143d10b5ea74805f"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"f584ba56bc5169655cece65995cd6d6dc7664cde","ours":"8cc65e0cef3c567de69a1117db406879c9829be0","theirs":"58b01f5a162bae158632765a81dea92e8545e06a","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
