)]}'
{"id":"openstack%2Fkolla~1005479","triplet_id":"openstack%2Fkolla~master~I386a9056a6802beaa0d8c110b9fc2f9a7fe8ce5c","project":"openstack/kolla","branch":"master","topic":"openstack/kolla/check-sources","hashtags":[],"change_id":"I386a9056a6802beaa0d8c110b9fc2f9a7fe8ce5c","subject":"Updated from check-sources","status":"MERGED","created":"2026-09-14 03:07:07.000000000","updated":"2026-09-16 07:49:34.000000000","submitted":"2026-09-16 07:48:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005479-openstack/kolla/check-sources","meta_rev_id":"9dd3e644591985c0d35ee2521d63d4eb75dd3393","_number":1005479,"virtual_id_number":1005479,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-16 07:48:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-09-16 05:41:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-09-16 05:41:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-09-16 05:41:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-14 04:04:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-14 05:21:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-14 17:00:01.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2026-09-16 05:41:21.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"e100f512f324b6e451c987bba8fbbcd4ca654ec7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-09-14 03:07:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"116f23da45c3136cf83d27e62e16b231df4b5e94","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 04:04:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97e03cc6bd7d4f3ab8ba149343ef1c10\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/de7317d603fb4a82b1e6b73c7d659186 : FAILURE in 40m 27s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/8bf97fd7429e4c1d90ef9d696c02ec11 : SUCCESS in 36m 45s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/dcf9e9c109894a3da787384b975f5848 : SUCCESS in 36m 52s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/fe30887937d748debd87ec38699de8fa : FAILURE in 54m 23s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/00b90a8422844c22afe2fb31c159b85f : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9a00df4096c545bd9f1a4da1732387c0 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ad63164c8fd94a3391bafb30aeac7445 : SUCCESS in 8m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c90b29cb1b319374e0dd1935c7418435a326b019","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 05:21:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/520b308340f3456ea8c117619f06d8d5\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/01570123d34c42a59ca4f5581f08b29f : FAILURE in 31m 00s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/e66156eda25748c79b1bfcc04a1bd3c4 : FAILURE in 19m 31s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/ab6992494492478088a3cb5050ca55be : SUCCESS in 32m 25s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/71ce6662c48542dd8fac594eb7516f3a : SUCCESS in 1h 16m 15s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00f6b4ec6e49455092bdeb6f2889bbc9 : SUCCESS in 19m 50s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a9df40bfbc1045919f19b0cd614fc8c9 : SUCCESS in 34m 09s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/83925728f7734f779daac9ccbf7061fe : SUCCESS in 1h 20m 29s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/0489e77b067d4812aa19564083b1801e : SUCCESS in 2h 11m 17s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/c4e3d8d56dfc4df789bebc5ed59c24da : SUCCESS in 1h 09m 42s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/c7b295c22fb542ae94cfb0896c9de42f : SUCCESS in 1h 36m 33s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/71bb84fc0d274ac78f3244518c7475cb : SUCCESS in 1h 22m 10s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/e5754ea60f31465e992e0d5086e1ca52 : SUCCESS in 1h 48m 52s\n- kolla-ansible-debian-trixie-lets-encrypt https://zuul.opendev.org/t/openstack/build/0e7a2cd639dc4d2c94c8c23e5093ab34 : FAILURE in 50m 49s (non-voting)\n- kolla-ansible-ubuntu-noble-lets-encrypt https://zuul.opendev.org/t/openstack/build/a737e4369b2f4de3b79a29b9907b86d0 : FAILURE in 48m 46s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/fece1c9b746d4215990ce4e00c5ff30d : SUCCESS in 45m 50s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/3a43ee0908ac474a8fed9edc95402df0 : SUCCESS in 30m 05s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/9f707046ed914e0f99538dfe6ccc596b : SUCCESS in 31m 30s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/9a4bb28272b84a9cbc1f735c72426b53 : SUCCESS in 41m 34s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/391f0e670f234b558825c2cf4cc347ed : SUCCESS in 1h 00m 08s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/5837c5f2cd274a4599ef94352e4ee2e4 : SUCCESS in 27m 40s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/0b58d5e9462d4a38bdbf367e02e10c66 : SUCCESS in 1h 10m 56s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/7749570ac995495dbe5e5350239e8e04 : SUCCESS in 25m 38s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/4352445ad9984eb589e574d75a0a26b1 : SUCCESS in 36m 34s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/2d5053e9ee0a4360b10f1f631d94f29d : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/608f0587f09a43b390e2723766f5e852 : SUCCESS in 6m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/01f6f21140754dc292bca83e66f46865 : SUCCESS in 4m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fd4969d934504cdb82cf62c1fee096f8 : SUCCESS in 3m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d03279fd0b1a450793d9f462f80b881d : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0959eb8765fc401d8da92cbaaa3183c6 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11b2ee45a97c4124b4c70378a4f04780 : SUCCESS in 7m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bcb9d6caabe544ee907627b49de32b35 : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":1},{"id":"c1d106d5516d6aa4cda4e9267d2d96e579de956d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-09-14 17:00:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f0d40aabac38b2b2443b99f0eb5758312ffcac4d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 18:28:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4450d65308fa4cf6a6d308ea695cd536\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/e3bdbb032de445899555a2462535ac21 : FAILURE in 41m 38s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/c827dc89976443e2a98a4cfbbe3f38df : SUCCESS in 48m 27s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/095d832008af470aae15c2de4ea8ee7b : SUCCESS in 43m 28s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/9b9eca8037dd48369f78b278596fc102 : FAILURE in 1h 25m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ab68b5cb82234584bb40a8a1f5343ecf : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/23ed07492d874b5da75c17844a6475b0 : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d60bc2fdded24a6aaea65ed41ae96020 : SUCCESS in 9m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9f3d64b166025cb014a2c583b095f501a3692ec8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 19:09:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eedede0e4059463aa6ec2d46763ba41c\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/104eb68182304422b3d5510db24ec78a : FAILURE in 25m 16s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/89275909cb184910b2877c6aee9d7f39 : FAILURE in 22m 27s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/81b189c6977a4101adde6e59c3326b5b : SUCCESS in 22m 20s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/2a93d8625c3140e1bf0c8f10f57d00e6 : SUCCESS in 40m 34s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/55e5d0f207f1452c80f5debe49cbee51 : SUCCESS in 29m 16s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/995f1c2ec104481bbd68f13917c13f18 : SUCCESS in 23m 32s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/64ece6bc4d3145cd93bfd5ca9600ef64 : SUCCESS in 1h 34m 12s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/20bde88ccae149cf840480c3e18a86d7 : SUCCESS in 2h 05m 27s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/d609edd5a14f4d4d904f8e2f9964bd03 : SUCCESS in 1h 10m 53s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/56f3ef3c6ae249739c979cbf12747610 : SUCCESS in 1h 35m 12s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6872c0402894edf9ccb1c609807f084 : SUCCESS in 1h 26m 24s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/6f2a11339f6d43cc9dac0e38b41a8393 : SUCCESS in 2h 02m 15s\n- kolla-ansible-debian-trixie-lets-encrypt https://zuul.opendev.org/t/openstack/build/77842899244a4e74b2011020d574418e : FAILURE in 1h 21m 47s (non-voting)\n- kolla-ansible-ubuntu-noble-lets-encrypt https://zuul.opendev.org/t/openstack/build/f4c345cdb6074a07bec8d495df389ad4 : FAILURE in 57m 12s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/3ae48ddce0954101a2e5b7abadf9a170 : SUCCESS in 54m 32s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/4bce611251b44d18a04262b074e4b334 : SUCCESS in 35m 16s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/2ff4f0e5a8f8430284345e85d319f410 : SUCCESS in 27m 34s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/bb37f9008f3c43a08a7574e9de1db2f5 : SUCCESS in 37m 39s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/708a3f8011a948359a8bfdcebdf162c4 : SUCCESS in 44m 55s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/b469998ba8404a20921c9a18bf487085 : SUCCESS in 31m 41s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/6570aee046c644f3a4f6b8b22e60f100 : SUCCESS in 55m 56s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/af15259fa73b4ff79571b6bbc1469085 : SUCCESS in 45m 57s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/9a9940e909d1436eb3f73494bd26ce31 : SUCCESS in 42m 07s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/fd03e5d1d6e046bc8c436f441e2246ae : SUCCESS in 2m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/158fd08a0aa94285b20552d5a7a15a33 : SUCCESS in 6m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e5686cf4e6b54dc1a7ed2b7a0b4947f3 : SUCCESS in 3m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6a2656eaab4e47f2a7831b13d505b121 : SUCCESS in 3m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b9570295121c457d987953d4b817f26c : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8f230cb089d472aa85d95488a6cfb14 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b77140ece2c54fd798fd54c28e3e55f2 : SUCCESS in 3m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d49ef2bb6ec43dca41f0140e257df91 : SUCCESS in 9m 32s","accounts_in_message":[],"_revision_number":1},{"id":"f9f9418502adecb6b9ceba13bea51f5bd464410c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-09-15 03:15:36.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"9ef253bc000cef7abf3f5a1ed367455aa3a88ea7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 04:21:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7533339924a146879fe11e6e36157350\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/d3fc00903e8742c393f197ef33f8c124 : FAILURE in 41m 35s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/423ad2334a4d4ff48dea173c70fe0f98 : SUCCESS in 45m 56s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/fa280bd06e694febb7e0d2af68fdec80 : FAILURE in 34m 47s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/8b4dd5d8bfda429a8b601e644aef1781 : FAILURE in 1h 03m 03s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/51d6671432db426b9cfdcb173513ac72 : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9e449fa79a634dbb9452f3ea8696ff44 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/89ba9a5a523e42b18692141dcd2d413b : SUCCESS in 6m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"428a586c84edf85e38a5a7dabbd25940cea9e16b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 04:56:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7bec5196160f4dcebc93845a07160c81\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/7816fd2bcafb44d79cd420478be38c63 : FAILURE in 23m 20s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/a4fc454d6af54980a68c152867a7b567 : FAILURE in 7m 40s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/81c400419cc04a2c8372aa5140e806df : SUCCESS in 30m 00s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/f48f30e013af46899cfda4967e562122 : SUCCESS in 44m 25s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a9a8c2c6b0949f792371be0072cae4e : FAILURE in 26m 06s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3faa12eaa4c84842847bc7a5b465eebf : FAILURE in 32m 21s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/0e9bcb037fef40578bec4afdea6ec7b0 : SUCCESS in 1h 24m 12s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/cdf7cb0d66ee40c5b30ecfdb864369c0 : FAILURE in 18m 36s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/a959a0c132a94e93a0600dd362a78f30 : SUCCESS in 1h 05m 57s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/fea72c6b1ab742bc80f1081a1defad94 : SUCCESS in 1h 38m 14s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/04cd6c8f30ca4a50b2e95551a384e000 : SUCCESS in 1h 33m 06s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/25daac2f7b83460ba0a478376f2daf3e : FAILURE in 12m 21s\n- kolla-ansible-debian-trixie-lets-encrypt https://zuul.opendev.org/t/openstack/build/43f8f6c225614823a8670cf8a66eac46 : FAILURE in 46m 53s (non-voting)\n- kolla-ansible-ubuntu-noble-lets-encrypt https://zuul.opendev.org/t/openstack/build/9197dd5404dd4f49b8cd90ebab037dc5 : FAILURE in 47m 49s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/6c0c66f3dd06447d9bec8505985b61cf : FAILURE in 27m 30s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/6de5f78122ac472d9b1704a24d2c066a : SUCCESS in 28m 38s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/8784acf669dc4a54ba85aac96e071d4e : SUCCESS in 35m 36s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/84988e307e9f422295578e943cc16c5f : SUCCESS in 42m 15s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/3324a4027cf14fc89acc6636a9da6931 : SUCCESS in 49m 39s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/b5dcfbfbcd604e0da15d4293da085e6f : SUCCESS in 27m 13s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/9489b3df3f3d4308bebc7b1d9bd724a2 : FAILURE in 12m 54s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/c2df53ba47864064bc6df1870aac892a : FAILURE in 9m 32s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/50dc64fa7d9b4f759ae200d2d8d1092d : FAILURE in 19m 44s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d7bbd9fbdb6d4ceabdf92e9cff43e0b0 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4a63c331a354273a262a4021965866c : SUCCESS in 4m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b9dc0031f1324b57b247b501a0617997 : SUCCESS in 2m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/28a7667bdf5b452cac2515867da6a9a8 : SUCCESS in 2m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e805520b0ec6463184fc9ab07b1f7034 : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a195e4ca80a4c56b9b24f4177f814a4 : SUCCESS in 2m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ef8b69c290c404eaf3aa198b77e77ba : SUCCESS in 5m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4ec09f276e249b4b48bbe23b5f26a58 : SUCCESS in 10m 54s","accounts_in_message":[],"_revision_number":2},{"id":"9db875767e84c5ecf5489c445952b488eb4d4afd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-09-16 02:54:39.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"f750c2c278ff51d2af7c59b24466f0d6fc0da5c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 05:31:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab8430ec475343c29f36b85869805fd5\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/6c3cdf168b114857b21a566cc7eab374 : SUCCESS in 29m 52s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/8c63c2c9a3854f14a7701ff1ff93b59e : SUCCESS in 32m 10s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/05dac0eac3ff4eee84dcdda404f773d4 : SUCCESS in 33m 45s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/684e1ec5544b4c8dbf34f0b4cbb96be1 : SUCCESS in 1h 10m 48s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e7c06d2db0e24545b2b9eb3ab99f08d7 : SUCCESS in 31m 43s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f5fefd8f9b024e3f992a65a1d4506d7b : SUCCESS in 32m 35s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/e80b7264f766401bb2ba958fe0bd45af : SUCCESS in 1h 13m 18s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/54ae3c9924054c4c9a4ce291e8c07b69 : SUCCESS in 2h 18m 29s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/5163052766cc48ca94fdce2407bc2f6a : SUCCESS in 1h 07m 36s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/060dc4862ace43bebcc5e64787b801b8 : SUCCESS in 1h 34m 31s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c786ee137eb48f4a6eff8dcdc5e4270 : SUCCESS in 1h 27m 38s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/b0a56652976d41abb2f9610cde72132c : SUCCESS in 2h 16m 07s\n- kolla-ansible-debian-trixie-lets-encrypt https://zuul.opendev.org/t/openstack/build/fcf1af9ebb5c422a80e7cd4e4539a01e : SUCCESS in 1h 40m 05s (non-voting)\n- kolla-ansible-ubuntu-noble-lets-encrypt https://zuul.opendev.org/t/openstack/build/45a0a02c61ad403da41ff25c4cdc14af : SUCCESS in 57m 29s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/e53f6028d1ae4828b1a47f9fea976006 : SUCCESS in 24m 49s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/46ac713a0a4b4828b6365e1ea06368c4 : SUCCESS in 24m 50s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/5b94ccfe5b8b4098b2bbd6643b288f97 : SUCCESS in 34m 54s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/3bc601b5c9eb42b493fe6d71dafaf7de : SUCCESS in 44m 01s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/655b7eec1915421cb90b478f0a7db8a7 : SUCCESS in 1h 03m 39s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/0fb09de24ba84b9c98769c79f10414fb : SUCCESS in 28m 56s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/ae21a06670374f4fbc82c34ca9414102 : SUCCESS in 1h 03m 30s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/a0f250ad507741209f61eb1bcc0cec0e : SUCCESS in 44m 57s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/38e57a16a79f4115bcc5cc9467c77884 : SUCCESS in 1h 07m 23s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/066f972335f04b3e815ffdaf4c1cc831 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68fceab3a54842cbb3130ffa5cdbf614 : SUCCESS in 7m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3bc2de0c48b749449b000dd6ce6fe59a : SUCCESS in 3m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7c47b983c18c413d8acd03350cb77d02 : SUCCESS in 5m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c7536890d00943e596ffbc936b768feb : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42f44fe0c36144be862f53789fe7cf04 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a410095866d41948bd1d25a01bdb508 : SUCCESS in 7m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf20fad87899482dab5ce0633d2186fc : SUCCESS in 11m 15s","accounts_in_message":[],"_revision_number":3},{"id":"79791ecbfb672b48cf69e839d86cd032a5538c90","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-09-16 05:41:21.000000000","message":"Patch Set 3: Backport-Candidate+1 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0fe7a6c330bb2e1682d21722f98d2075831a4cd1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 05:43:09.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"031dc82b9a7e71fb1b6685fccd9b8196d6d8a668","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 06:10:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d55c41f791454487a0f80848464abf75\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/63e1c6f20f074e0b962e682023d0560e : SUCCESS in 38m 54s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/8c22488fd44a4ff4b9801a1cb67091b4 : FAILURE in 33m 42s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/9896b5510e024b12aa24af0d63b2e7a5 : SUCCESS in 46m 44s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/51707e2553d4487eb2c371283c4c6f6d : FAILURE in 1h 23m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f4cb40a1f13144e89ef7e9dfcfbf8abe : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/68b101c54ba94ce18cf60db3639c4af7 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/858fda7ab3244018a729d63cc60ad3d5 : SUCCESS in 7m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"731c48d5d62a2d457d7c59e202a579d240e9ceb1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 07:48:03.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/821ae9f0685f4946b106a1c858afe36c\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/0d64ac065dc24f2f801e39ed65e65bd4 : SUCCESS in 16m 46s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/3c2a5211e4094f5bb6ed3d57dc13a6c9 : SUCCESS in 22m 36s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/bfa0ca097f2445b3a181279a5e44e95f : SUCCESS in 20m 55s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/7327f9e9c8d24ef2b820a5e6e0a6b47e : SUCCESS in 41m 00s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b42dd62af19f41ce9d092a98ffc2de68 : SUCCESS in 21m 49s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/227a722d92a24fc4a41750fa9f6cf7af : SUCCESS in 30m 04s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/f734bbb3672a4495bee4bfd47b6efe0e : FAILURE in 24m 45s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/dbd06b11c05945178b5315204cdf06f8 : SUCCESS in 2h 04m 17s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/98ac292c13354a7ead9f5268f0744f9e : SUCCESS in 1h 06m 58s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/9834fcc08fdc420d9df046fdc9649db2 : SUCCESS in 1h 37m 00s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/51c69618c8964f77ba66f8541d19c662 : SUCCESS in 1h 37m 10s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/910038c0405d4c5d93d0ad793f0d7a86 : SUCCESS in 1h 18m 06s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/73fb15ffeb964ce5b0ad0c78d7595686 : SUCCESS in 2m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/185a4164e3ff47bfbdaf21a7f36b9a32 : SUCCESS in 5m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e45be3783b2d4604833707a2977801ac : SUCCESS in 3m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8e3dcfaf2b2a42c4adf04cac78577642 : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5808187925664511af7b8d2c568678a2 : SUCCESS in 8m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0185d42bf5e04624b877b70ecff1fcd6 : SUCCESS in 7m 44s","accounts_in_message":[],"_revision_number":3},{"id":"8d80e30258c86846c56754c1fd258f39bec7525c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 07:48:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"9dd3e644591985c0d35ee2521d63d4eb75dd3393","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 07:49:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6dd271c1531d435f875c23e407cb5097\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8db361d7bad74954bc4b4cad16791f56 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/263bc5bbbe434446afc924ca7bfc3dd7 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0cbc63835b7d24c726da1bb44a46fbdd3e09680e","revisions":{"ee4a1290747c8069864760a09eb918d5f172c4b3":{"kind":"REWORK","_number":1,"created":"2026-09-14 03:07:07.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/79/1005479/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/1005479/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/1005479/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/1005479/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/1005479/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/1005479/1"}}},"commit":{"parents":[{"commit":"58b01f5a162bae158632765a81dea92e8545e06a","subject":"Merge \"CI: add letsencrypt scenario\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/58b01f5a162bae158632765a81dea92e8545e06a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-14 03:07:03.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-14 03:07:03.000000000","tz":0},"subject":"Updated from check-sources","message":"Updated from check-sources\n\nChange-Id: I386a9056a6802beaa0d8c110b9fc2f9a7fe8ce5c\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee4a1290747c8069864760a09eb918d5f172c4b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee4a1290747c8069864760a09eb918d5f172c4b3"}]},"branch":"refs/heads/master"},"966ab663f39ff4c8ff3c4e60fe400e78cf059d61":{"kind":"REWORK","_number":2,"created":"2026-09-15 03:15:36.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/79/1005479/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/1005479/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/1005479/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/1005479/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/1005479/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/1005479/2"}}},"commit":{"parents":[{"commit":"58b01f5a162bae158632765a81dea92e8545e06a","subject":"Merge \"CI: add letsencrypt scenario\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/58b01f5a162bae158632765a81dea92e8545e06a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-15 03:15:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-15 03:15:32.000000000","tz":0},"subject":"Updated from check-sources","message":"Updated from check-sources\n\nChange-Id: I386a9056a6802beaa0d8c110b9fc2f9a7fe8ce5c\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/966ab663f39ff4c8ff3c4e60fe400e78cf059d61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/966ab663f39ff4c8ff3c4e60fe400e78cf059d61"}]},"branch":"refs/heads/master"},"0cbc63835b7d24c726da1bb44a46fbdd3e09680e":{"kind":"REWORK","_number":3,"created":"2026-09-16 02:54:39.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/79/1005479/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/1005479/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/1005479/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/1005479/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/1005479/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/1005479/3"}}},"commit":{"parents":[{"commit":"bfccf861b1c12716b99d345a143d10b5ea74805f","subject":"Pin Lego to v4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bfccf861b1c12716b99d345a143d10b5ea74805f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-16 02:54:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-16 02:54:36.000000000","tz":0},"subject":"Updated from check-sources","message":"Updated from check-sources\n\nChange-Id: I386a9056a6802beaa0d8c110b9fc2f9a7fe8ce5c\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0cbc63835b7d24c726da1bb44a46fbdd3e09680e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0cbc63835b7d24c726da1bb44a46fbdd3e09680e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
