)]}'
{"id":"openstack%2Fkolla~1005850","triplet_id":"openstack%2Fkolla~stable%2F2026.1~I5fa8ec4dd5d1d4938e0b47c99d07d328286ecda9","project":"openstack/kolla","branch":"stable/2026.1","topic":"fix/zstd-issue","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-09-17 10:36:52.000000000","reason":"Change was abandoned"},"38330":{"account":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"last_update":"2026-09-16 12:20:08.000000000","reason":"\u003cGERRIT_ACCOUNT_38330\u003e replied on the change","reason_account":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"}}},"hashtags":[],"change_id":"I5fa8ec4dd5d1d4938e0b47c99d07d328286ecda9","subject":"Fix: Zstd missing from ironic conductor container.","status":"ABANDONED","created":"2026-09-16 09:42:48.000000000","updated":"2026-09-17 10:36:52.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e05e0e00efbd51c1c911f9a58890f8138e9b08d6","_number":1005850,"virtual_id_number":1005850,"owner":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-16 12:03:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2026-09-16 10:29:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":-1,"date":"2026-09-16 09:46:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-16 09:46:16.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2026-09-16 10:29:20.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-09-16 12:03:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d74ce7a1bb2bd9566c6616acfe131a014944638c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2026-09-16 09:42:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d4ca91047d2dcb9ba2bc9e7f444f9a88473f8d8","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-09-16 09:46:16.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"229b99b54ce4672d0f48e2105dabf71d7c7cf5e1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-09-16 10:29:20.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9e9ded20810c9aee2314c2b2867b3d9c8a058eed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 12:03:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78e481e0263f49c99fff023ac80b3719\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/f695fe9c59a2423bbb8429cc1351886a : SUCCESS in 1h 12m 18s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/eff402f933e644bdb1e9836115cc6a72 : SUCCESS in 1h 03m 36s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/f1e96d3570c44a40a368372d2f754931 : SUCCESS in 27m 49s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/df62d54e6cf248d4b7eb8009eccdd504 : SUCCESS in 1h 05m 35s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/417465ffb9d54d33afe695f7530c6427 : SUCCESS in 32m 12s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8dc30acb0dd14a16b90182d08e2e752e : SUCCESS in 1h 12m 04s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d9012ead4cb5402787cab53c43baabe1 : SUCCESS in 27m 44s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/5c15f57a1b58401597e1483d92f0a13f : SUCCESS in 1h 18m 10s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/37284ab91bd84ec69456142372e5941b : SUCCESS in 1h 40m 37s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/ac9e7e64e719470a8e7d68571b21f2de : SUCCESS in 2h 19m 38s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/c4c21bac796349329492db5e4068f27a : SUCCESS in 1h 18m 54s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/c6869d744948493b87017e35170dd623 : SUCCESS in 1h 46m 32s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6bf84ad5bae84c528a763b5e6ece941a : SUCCESS in 1h 33m 34s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/ca37c876a8ae49c9a0be4017c94758ce : SUCCESS in 2h 17m 28s\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/db533e014f344f8cb79fd216c8e5cfb4 : SUCCESS in 2h 07m 51s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/c0655d89b29a4686b08092fb066c503f : SUCCESS in 1h 41m 22s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/e1183adeee94418f843dcc4d2fb3f69b : SUCCESS in 1h 59m 43s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d22f1ff1cddc4f4cba84aeb392ff8126 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c08ea113c84406db5975f9e7b5813af : SUCCESS in 8m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a278b485706249e2b46d2842eaa6024d : SUCCESS in 3m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e5434972e1ca4d5d800a495421c8d2f2 : SUCCESS in 3m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3171a8f9f18d4e66b9c4b39794fe0f4f : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6fce77ec12649659b03f5a35724edd7 : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4527afac1a7f452f90e9e86c8c9a70b1 : SUCCESS in 5m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e212f6f5d1ed4b22b76681fde51d8964 : SUCCESS in 10m 13s","accounts_in_message":[],"_revision_number":1},{"id":"635c4f22be79ab0ad0de962e5d8d1fc1fa5f0a04","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 12:17:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00c8dd6590754eda8ec986e5d0390505\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/9a810bc207584280b3b110f5c7a18ded : SUCCESS in 1h 53m 28s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/e983d5cd40cd453189d171572c0e0dd7 : SUCCESS in 1h 18m 05s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/ff3eb140dcfa4a3fba1c87fc4bca00e6 : SUCCESS in 2h 14m 53s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/aaaa7be8fcf14b5a83532b2afca4d97e : SUCCESS in 2h 32m 14s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/7de0eac36fdb40dfa379c8c9fdeb594e : FAILURE in 1h 38m 10s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/18d366e42a95494facd583b6c45591c8 : FAILURE in 2h 23m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"81c2741b25fb134231f5ca64ed7d25bb1d0d0959","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2026-09-16 12:20:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e05e0e00efbd51c1c911f9a58890f8138e9b08d6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-09-17 10:36:52.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ae24ca9b4c52c5adca7e96237917b6a86688d8c8","revisions":{"ae24ca9b4c52c5adca7e96237917b6a86688d8c8":{"kind":"REWORK","_number":1,"created":"2026-09-16 09:42:48.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/50/1005850/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/1005850/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/1005850/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/1005850/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/1005850/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/1005850/1"}}},"commit":{"parents":[{"commit":"c4e23082aaf2329ffd1b1ddb3992a881aa53af3f","subject":"stable-only: Update Prometheus to version 3.5.5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c4e23082aaf2329ffd1b1ddb3992a881aa53af3f"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2026-09-16 09:38:41.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2026-09-16 09:38:41.000000000","tz":60},"subject":"Fix: Zstd missing from ironic conductor container.","message":"Fix: Zstd missing from ironic conductor container.\n\nIronic-conductor needs zstd installed to be able\nto decompress zstd compressed files. Patch installs\nzstd to docker container.\n\nChange-Id: I5fa8ec4dd5d1d4938e0b47c99d07d328286ecda9\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae24ca9b4c52c5adca7e96237917b6a86688d8c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae24ca9b4c52c5adca7e96237917b6a86688d8c8"}]},"branch":"refs/heads/stable/2026.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
