)]}'
{"id":"openstack%2Fkolla~1005869","triplet_id":"openstack%2Fkolla~master~I5fa8ec4dd5d1d4938e0b47c99d07d328286ecda9","project":"openstack/kolla","branch":"master","topic":"fix/zstd-issue","attention_set":{"38330":{"account":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"last_update":"2026-09-17 18:28:32.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-09-18 09:41:42.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2026-09-18 08:39:07.000000000","reason":"\u003cGERRIT_ACCOUNT_17669\u003e replied on the change","reason_account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-09-18 08:58:44.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"I5fa8ec4dd5d1d4938e0b47c99d07d328286ecda9","subject":"Fix: Zstd missing from ironic conductor container.","status":"NEW","created":"2026-09-16 12:16:49.000000000","updated":"2026-09-24 12:57:19.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":15,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"19f397c6255a3b3d7e89798228d34ef9c22232e0","_number":1005869,"virtual_id_number":1005869,"owner":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-24 12:57:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-09-24 10:30:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2026-09-24 10:30:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2026-09-24 10:30:16.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2026-09-24 10:30:27.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2026-09-24 10:30:27.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-16 12:22:43.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-09-16 14:04:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-16 14:31:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-16 17:08:46.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2026-09-17 08:18:05.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-09-18 09:27:55.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"}],"messages":[{"id":"8106d615ebfb247c70c86b2e10b40213e61a39c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2026-09-16 12:16:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"695e4d26f0e9a90d3e6851d60fb63283f827f36f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-09-16 12:22:43.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3f2421e488d93c3a73d54fee15fd887516a2ea81","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 14:04:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec1766c8b413478191fec8b60bf2f6b4\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/2921b6c7d7594b659ed56090ef650f6d : SUCCESS in 36m 30s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/397be3de3fa240b28a9ba5db40fab7f9 : SUCCESS in 59m 12s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/eb9be46ea80249c2b5f38cd6e4382b4d : SUCCESS in 52m 18s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/f743dfa0da904ef8be4ead6dddfd7440 : FAILURE in 1h 44m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0d7f2289c9874581b1241bf4982a1a2c : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/495a4ebd801c4061b68df5366847ceec : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0c776c5bbe2f42618e1614c5ec053f83 : SUCCESS in 9m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5f41846f75c1612405ac73a28b1cadc8e602a555","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 14:31:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2595b4ac3cfb45b7a4b7aa723e829da2\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/8177110acdef48e8b75339e6efdd4de3 : SUCCESS in 27m 19s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/843f61e227ef41abb493a545f27b92f1 : SUCCESS in 31m 05s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/404c3b4d44424dc59a309f44ae8ae341 : SUCCESS in 24m 32s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/da97f036aae440abacb919df09374c3b : SUCCESS in 43m 18s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c4de2c49f7d249819ecb557895f0a039 : SUCCESS in 30m 40s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/414e50731cb1420abbc610a55c504b04 : SUCCESS in 25m 28s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/9854ea2f9bd24feaa78d40d1f69f27fc : SUCCESS in 1h 31m 52s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/6a8230cd7342456985bd02b6b1861709 : SUCCESS in 2h 11m 36s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/3be322a92cfb4b23ab529db05cd4a223 : SUCCESS in 1h 15m 30s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/ef14af75433c4d428347c4c18119e855 : SUCCESS in 1h 46m 45s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9ed3309fa5104084b73c73d84894a177 : SUCCESS in 1h 37m 54s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/9e7b0709fe114b9db4a9959004262181 : FAILURE in 1h 42m 30s\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/bf1adf1cb4b3419fb069c59bb85f525b : SUCCESS in 2h 00m 56s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/8dcafac8dd5941cca711d7c92feb1588 : SUCCESS in 1h 38m 07s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/1325d7aec3b140e5962180ee28b13940 : SUCCESS in 1h 53m 31s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4a18cd3f521e496893cbcda5a185c82d : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f17ca20ade949099dfc75365847fb00 : SUCCESS in 10m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/60fdf8eebc824a9caa799259e759a657 : SUCCESS in 2m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/271d6d40c7c1452985967b90dc693033 : SUCCESS in 3m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9770990122064742981929a360de2631 : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12899d9ebdd445679d13e99a90220e78 : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/112c78df3239448eb8a55785fbfc464a : SUCCESS in 6m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/265285d3a3a04f69b74e1c1fbc57eab6 : FAILURE in 3m 52s","accounts_in_message":[],"_revision_number":1},{"id":"8a671fe87b4ffcb31c41441b9bb5245a81d67b84","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-09-16 17:08:46.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"80b347866ed09745af77bc2a76342380d2059ab5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2026-09-17 07:49:51.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"465f2ed4c5a12a53b12f7f7611187cd4a91f7d0c","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2026-09-17 07:51:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b26e7dfeafcdc16258c0c920ad7c9b1726be9af1","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-17 08:18:05.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9d7f4eb97fbbad2a1116b01f1a72bbd87a9fb382","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2026-09-17 09:35:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2c6f86e714da539c0584ac857f5f44dd44b1d1d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 09:37:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af88c6a3dfb142ea8259970ff7017a18\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/838f167fe8314b9a8dc915065811c3df : SUCCESS in 37m 18s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/bb4db583ca544e498a4de977dee8ef76 : SUCCESS in 57m 21s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/2c1ce5a9ed7d491288689f7204c89784 : SUCCESS in 43m 07s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/37b67a2fb39f4fdfb2c29efbdac90f2f : FAILURE in 55m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/511134fb9cff4596a8cc18d235431c76 : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/98c550ad3ef14f87ba015f5057a7e650 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/251a99c775dc4bd7af06693d38be68e9 : SUCCESS in 8m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5784777dd1b9dc71a5c572685e8160fb13479361","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2026-09-17 09:39:23.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"95322a4e5d5c537e500d99649e32a0b86870a0a6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-17 09:42:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cd9ce7d64f85df7ca7e567cd18b1108c1c5034db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 12:51:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8630b33c4ea492f95b0553a1f598ccf\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/222d3ca45c044b12984db02fc538f48a : SUCCESS in 1h 00m 15s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/5600930ff2e84062b6fe1463c9a18451 : SUCCESS in 35m 30s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/cb45e96c37694f998d378d0a87dfa662 : SUCCESS in 33m 45s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/902d271c80e448288452320b1858fca2 : FAILURE in 1h 32m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e7ab43f3116e4418bb6b45b86be6ea72 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b102ad4a27974ed097fe5ed48ea95df0 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1dfde294142640c78bb92b8069430e22 : SUCCESS in 5m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"eb667687401d8c5316657bb7b8d0b4b87588fb0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2026-09-17 13:05:00.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"48835491a590f52a8340093bbc71bfd3dc2d4e05","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2026-09-17 13:06:28.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"86fae004ea08ac71294f8558a817be3062259510","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 15:47:14.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70e6e79daa6649cfae90f72a009b7b7d\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/1826fe6a441947bea5b2ac2bddfd9d51 : SUCCESS in 51m 56s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/933e663b091c4905b1c6500022ea9a68 : SUCCESS in 59m 17s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/b05f070f1e384dc0a11d40e5e3b4228d : SUCCESS in 52m 44s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/b583c25691e0480e875fa32ee882a359 : FAILURE in 1h 34m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b312005b92244d45b11a62d01b66da6d : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/990c14a306044501a4bc1c2baff77a10 : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/38cec023a3584bebb79a2b7a75310ce4 : SUCCESS in 5m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b318061c7a8c3424a45e2aed1aaea997585f45ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 18:28:32.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6896518cfff64580b9a272d011dc555f\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/0a97a14245594521b030cfc642248b30 : SUCCESS in 30m 00s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/39f50495374a4564981059adc95b9c9c : SUCCESS in 24m 52s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/e615cdf76a174f89bb834b861aa96bf5 : SUCCESS in 30m 08s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/ebca2207487a4a6cbdd3ed6c79d4532e : SUCCESS in 56m 51s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/778eac1b02ab4a8e93dde0dd36f385cb : FAILURE in 33m 26s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ae709db8d25d41ddb21cf431914531c0 : FAILURE in 35m 40s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/51b597c17b7843d4a480ad508dc5a6de : SUCCESS in 1h 16m 26s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/32ee673296ba4985827c38feca2be234 : SUCCESS in 2h 05m 49s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/a8e380d94ced42acb4f7e5bc9a7ea6ec : SUCCESS in 1h 02m 28s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/363721f086f64ebfae5b559f0e443fbe : SUCCESS in 1h 30m 59s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/181ae76b8c8a4b30be6484f255bdbb0c : SUCCESS in 1h 24m 55s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/6b9ab92db2774fe68ec86699954093a8 : SUCCESS in 1h 56m 02s\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/e4702dbc7eec40c4a4ce5037b06d6f70 : SUCCESS in 1h 35m 08s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/894363fea7a1417c88400b30c900c983 : SUCCESS in 1h 16m 44s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/a00d9a4db51541cfbd42b10613fe3deb : TIMED_OUT in 2h 31m 55s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4701fd33fde04c55b41e9cd5c13f9267 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37cfc1ce05ce4f8588fde0624ef9a381 : SUCCESS in 4m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5400db0b6ed24e3b816a2e052be4cc02 : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cfb1bce1734d48438f8e0cefd19b18e2 : SUCCESS in 4m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/079b1202fca949bb89ee407c1f70c4a4 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8457aa4829074e7fbac52a3ca5593995 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b097d30a0c39456d9b7162739aaec284 : SUCCESS in 8m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85da010eeab04237be7b724def39fe20 : SUCCESS in 11m 03s","accounts_in_message":[],"_revision_number":4},{"id":"da82199dce51a87aa0628434707e2a9772b076ce","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-09-18 08:39:07.000000000","message":"Patch Set 4: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"8e44aeea08c3a5ef17b69ac16089f54917c240a2","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-09-18 08:39:22.000000000","message":"Patch Set 4: Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"8fb5277845563f85f19b6cc5db0d235f9f74ce82","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-18 08:58:44.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"6b9156c5cebebf02e523d52db3fe4992c8b254f7","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-09-18 09:41:42.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0d0ccab453f82a0a1f72d576a0858061dc73629e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-09-19 20:04:40.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cd82c61b9457cb0278387c8c9a93a5ebb2f43418","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-19 21:30:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3167fa4177084c6797d8076b7801ecf5\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/9edffeed75d748c9b2ec48af6223ae48 : SUCCESS in 48m 33s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/167a4dd07914418eb07e8a7caa23c1d1 : SUCCESS in 49m 41s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/d91b0c2751d54df9b31ba61b09575de0 : SUCCESS in 46m 47s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/70796973970645ab9fc7748deea77a31 : FAILURE in 1h 22m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/64af371dffdf49a2891df7b7437018ff : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0087061cec9447a9978a7947bfb2246e : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/132462db85af4d13b9cfd4ea55168139 : SUCCESS in 7m 57s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e2324c9504f762e8b2d571be299ce69c02fcd1dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-19 21:46:40.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e7b85cabb3884e8eafe633459ec23a72\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/eafe50d80c814e658a536875eb62ac3d : SUCCESS in 30m 58s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/8b1e480d181a41f0810f6b9de1e37139 : SUCCESS in 30m 15s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/1eb8946e617c41e092230a59b000c655 : SUCCESS in 23m 19s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/894e48539010428ca34060a10c3a5f83 : SUCCESS in 1h 13m 08s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c838c006bd65498eac26a5453cf277a7 : SUCCESS in 18m 56s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d68828364f82461fa8df1dff68c179f4 : SUCCESS in 29m 52s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/8f421e9668b04e4f9082f8e7bd7ce2ec : FAILURE in 1h 07m 08s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/5954fd13bbcd444d8fe0585c38d8b2c3 : FAILURE in 11m 48s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/3de2bed5d6b948b9b600ec75b9ccaa1a : SUCCESS in 1h 02m 10s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/c92f84d3e7294a749e351a834e144afb : SUCCESS in 1h 31m 15s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec5a25a06fbd44d0bc397877bce20d26 : SUCCESS in 1h 10m 59s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/a2851e9b26c74a648abed1c38d6e5aff : FAILURE in 4m 19s\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/617d9558745a44e8ab6cc17a0908a23f : SUCCESS in 1h 29m 12s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/edba110233aa4a91ae46e4aba0056d02 : SUCCESS in 1h 17m 00s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/6c890b55d0634c97a948e598e62ea148 : SUCCESS in 1h 36m 14s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/af968f5e698d4dceb55ccd5da2f33f1b : SUCCESS in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecc80977fed84f58a48d24cdb3d6aade : SUCCESS in 4m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8bc42df39bcf43e8bf281a02c71df04b : SUCCESS in 5m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a96a354c315845a1a2e3b12d63f41f63 : SUCCESS in 2m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c0338e20bf70449ab96cb0494a46ead6 : SUCCESS in 8m 52s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31426eb2d48f4632b6942d298f1263e6 : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab4bbb0e78d349f79f188ee8e5128a75 : SUCCESS in 6m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/558a53826f41485bbe781cbae2661516 : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":4},{"id":"9496a921923a132c392c425283135dce7dc90626","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"real_author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-09-24 10:30:16.000000000","message":"Patch Set 5: Patch Set 4 was rebased on behalf of \u003cGERRIT_ACCOUNT_38330\u003e\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"}],"_revision_number":5},{"id":"881235cccd54f5d9db34c0690d6fd0de546e4ffb","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-09-24 10:30:27.000000000","message":"Patch Set 5: Code-Review+2 Review-Priority+1 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"57aa79c1bec9c553ccae0bd514d7fcc93b404184","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-09-24 10:31:29.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a217ac009e0314456db8af882b4c6b0bd9f09d2d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 11:57:16.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/110cf627114842eca05aadf2b3e23d1f\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/1303e5c59723410389e9acf6d5f7ab84 : SUCCESS in 56m 41s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/2e1f001fef8b403c8754ab49127e80be : SUCCESS in 1h 00m 24s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/fae3eea237c047729c58aa1777919b52 : SUCCESS in 59m 48s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/11569f7888964adba92cf45e57cffef8 : FAILURE in 1h 24m 23s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/30aaccb7c6de4284b48bd0e10014134f : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/93f039e267a54962bdfd3519984f185d : SUCCESS in 8m 57s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"19f397c6255a3b3d7e89798228d34ef9c22232e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 12:57:19.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c8cfb97ec9714ea48c761acbbdd9c4f2\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/9c500def8a394daaa1bc6cb2cbfde6c1 : SUCCESS in 27m 44s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/eb28f97522e34be7aad981c08e609050 : SUCCESS in 37m 00s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/91f607a2206141239f6a2baa5d3a724d : SUCCESS in 29m 32s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/7bf24a29c08146c683f7210e8efb00a8 : SUCCESS in 54m 52s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4bbc7a99c8d6428798c5827224df08f8 : SUCCESS in 27m 08s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/67289c7e19be45858a00ad2981f57211 : SUCCESS in 34m 40s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/aaf5b7b3e8524cd99d45f37bf1de3608 : SUCCESS in 1h 15m 39s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/de2abbd506454159af9126e4a0e35b0e : SUCCESS in 1h 59m 54s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/115128ea85cc42a79df73302fd93a2d1 : SUCCESS in 1h 02m 52s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/2aaea7e0ef4a44baa721a5495509a9c2 : SUCCESS in 1h 28m 03s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aee9440fc57747dba97e7035c7b3298a : FAILURE in 10m 02s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/1a52bd041ebf46c5ac8f8f7db6f04642 : SUCCESS in 2h 22m 17s\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/fe873e6abf3e4e1abdb9b3ed5addbc4a : FAILURE in 9m 37s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/76a830222b8f4a39abba9c3f9c0fd012 : SUCCESS in 1h 16m 36s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/7cb62a9a4143429eab6ca9d6f972bf95 : FAILURE in 6m 04s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/2750a1f3c2924e86884ac07d51e42d33 : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07b4f1b4a07149b391ca33c67c991df5 : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/af4ce1d1a14846c4964e84b2d9890c0a : SUCCESS in 3m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c890c4a93b09412f8b090feb17387013 : SUCCESS in 5m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d049ac270fca461b97f5c4485c99e2e4 : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/741d116bd9d64dbe9dc7894456d629b9 : SUCCESS in 6m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3df49248c7b84f68b350f22966dfd1c3 : SUCCESS in 10m 50s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"3534f45859f5a073126b5d52a44ee8f5c25710a8","revisions":{"f095468e5719a3436d266571edaf64df674a3851":{"kind":"REWORK","_number":1,"created":"2026-09-16 12:16:49.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/69/1005869/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/1005869/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/1005869/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/1005869/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/1005869/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/1005869/1"}}},"commit":{"parents":[{"commit":"0cbc63835b7d24c726da1bb44a46fbdd3e09680e","subject":"Updated from check-sources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0cbc63835b7d24c726da1bb44a46fbdd3e09680e"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2026-09-16 09:38:41.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2026-09-16 12:15:34.000000000","tz":60},"subject":"Fix: Zstd missing from ironic conductor container.","message":"Fix: Zstd missing from ironic conductor container.\n\nIronic-conductor needs zstd installed to be able\nto decompress zstd compressed files. Patch installs\nzstd to docker container.\n\nCloses-Bug: #2167459\nChange-Id: I5fa8ec4dd5d1d4938e0b47c99d07d328286ecda9\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f095468e5719a3436d266571edaf64df674a3851"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f095468e5719a3436d266571edaf64df674a3851"}]},"branch":"refs/heads/master"},"45a64aa6240b06f7aaa2041174abf40348bbded4":{"kind":"REWORK","_number":2,"created":"2026-09-17 07:49:51.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/69/1005869/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/1005869/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/1005869/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/1005869/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/1005869/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/1005869/2"}}},"commit":{"parents":[{"commit":"0cbc63835b7d24c726da1bb44a46fbdd3e09680e","subject":"Updated from check-sources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0cbc63835b7d24c726da1bb44a46fbdd3e09680e"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2026-09-16 09:38:41.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2026-09-17 07:49:42.000000000","tz":60},"subject":"Fix: Zstd missing from ironic conductor container.","message":"Fix: Zstd missing from ironic conductor container.\n\nIronic-conductor needs zstd installed to be able\nto decompress zstd compressed files. Patch installs\nzstd to docker container.\n\nCloses-Bug: #2167459\nChange-Id: I5fa8ec4dd5d1d4938e0b47c99d07d328286ecda9\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45a64aa6240b06f7aaa2041174abf40348bbded4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45a64aa6240b06f7aaa2041174abf40348bbded4"}]},"branch":"refs/heads/master"},"54234f5b7d997998b654278bf212b019db987186":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-09-17 09:39:23.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/69/1005869/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/1005869/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/1005869/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/1005869/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/1005869/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/1005869/3"}}},"commit":{"parents":[{"commit":"0cbc63835b7d24c726da1bb44a46fbdd3e09680e","subject":"Updated from check-sources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0cbc63835b7d24c726da1bb44a46fbdd3e09680e"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2026-09-16 09:38:41.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2026-09-17 09:36:15.000000000","tz":60},"subject":"Fix: Zstd missing from ironic conductor container.","message":"Fix: Zstd missing from ironic conductor container.\n\nIronic-conductor needs zstd installed to be able\nto decompress zstd compressed files. Patch installs\nzstd to docker container.\n\nPatch merged in Feb 2025 ( https://review.opendev.org/c/openstack/ironic/+/938904 )\nenabled the detection and decompression of zstd compressed\nfiles.\n\nZstd is a popular choice for compression, therefore,\nwould be useful to support.\n\nCloses-Bug: #2167459\nChange-Id: I5fa8ec4dd5d1d4938e0b47c99d07d328286ecda9\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/54234f5b7d997998b654278bf212b019db987186"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/54234f5b7d997998b654278bf212b019db987186"}]},"branch":"refs/heads/master"},"c34626ff8efb63e266306ab87fdc324eaac23df2":{"kind":"REWORK","_number":4,"created":"2026-09-17 13:05:00.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/69/1005869/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/1005869/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/1005869/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/1005869/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/1005869/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/1005869/4"}}},"commit":{"parents":[{"commit":"0cbc63835b7d24c726da1bb44a46fbdd3e09680e","subject":"Updated from check-sources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0cbc63835b7d24c726da1bb44a46fbdd3e09680e"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2026-09-16 09:38:41.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2026-09-17 13:04:51.000000000","tz":60},"subject":"Fix: Zstd missing from ironic conductor container.","message":"Fix: Zstd missing from ironic conductor container.\n\nIronic-conductor needs zstd installed to be able\nto decompress zstd compressed files. Patch installs\nzstd to docker container.\n\nPatch merged in Feb 2025 ( https://review.opendev.org/c/openstack/ironic/+/938904 )\nenabled the detection and decompression of zstd compressed\nfiles.\n\nZstd is a popular choice for compression, therefore,\nwould be useful to support.\n\nCloses-Bug: #2167459\nChange-Id: I5fa8ec4dd5d1d4938e0b47c99d07d328286ecda9\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c34626ff8efb63e266306ab87fdc324eaac23df2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c34626ff8efb63e266306ab87fdc324eaac23df2"}]},"branch":"refs/heads/master"},"3534f45859f5a073126b5d52a44ee8f5c25710a8":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-09-24 10:30:16.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"real_uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/69/1005869/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/1005869/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/1005869/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/1005869/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/1005869/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/1005869/5"}}},"commit":{"parents":[{"commit":"3f51ec474be2fcd493753c6b78fd26de4927f6fb","subject":"Updated ansible-lint in lint-requirements.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3f51ec474be2fcd493753c6b78fd26de4927f6fb"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2026-09-16 09:38:41.000000000","tz":60},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2026-09-24 10:30:16.000000000","tz":0},"subject":"Fix: Zstd missing from ironic conductor container.","message":"Fix: Zstd missing from ironic conductor container.\n\nIronic-conductor needs zstd installed to be able\nto decompress zstd compressed files. Patch installs\nzstd to docker container.\n\nPatch merged in Feb 2025 ( https://review.opendev.org/c/openstack/ironic/+/938904 )\nenabled the detection and decompression of zstd compressed\nfiles.\n\nZstd is a popular choice for compression, therefore,\nwould be useful to support.\n\nCloses-Bug: #2167459\nChange-Id: I5fa8ec4dd5d1d4938e0b47c99d07d328286ecda9\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3534f45859f5a073126b5d52a44ee8f5c25710a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3534f45859f5a073126b5d52a44ee8f5c25710a8"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"0cbc63835b7d24c726da1bb44a46fbdd3e09680e","ours":"c34626ff8efb63e266306ab87fdc324eaac23df2","theirs":"3f51ec474be2fcd493753c6b78fd26de4927f6fb","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
