)]}'
{"id":"openstack%2Fkolla~1005871","triplet_id":"openstack%2Fkolla~stable%2F2026.1~Ib4a0d7185ccc64900b4068dd74995dbf1ba6823c","project":"openstack/kolla","branch":"stable/2026.1","attention_set":{"35264":{"account":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"last_update":"2026-09-16 22:16:49.000000000","reason":"A robot voted negatively on a label"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-09-16 22:16:49.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-09-18 10:47:19.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2026-09-16 20:14:31.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"Ib4a0d7185ccc64900b4068dd74995dbf1ba6823c","subject":"grafana: install Prometheus datasource plugin","status":"NEW","created":"2026-09-16 12:36:58.000000000","updated":"2026-09-18 14:07:51.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":1003068,"cherry_pick_of_patch_set":1,"meta_rev_id":"622f033b5d6ef32b7614ecb409fe002035d0ad84","_number":1005871,"virtual_id_number":1005871,"owner":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-17 00:24:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-09-16 23:48:06.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2026-09-16 20:14:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2026-09-16 20:14:31.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"CC":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-16 12:36:58.000000000","updated_by":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"reviewer":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"state":"CC"},{"updated":"2026-09-16 12:42:10.000000000","updated_by":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-09-16 12:42:10.000000000","updated_by":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-09-16 15:48:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-16 23:48:06.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-09-17 10:10:43.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"}],"messages":[{"id":"0796a7428374c4046d22770f39e8bb7a7e8e5f03","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2026-09-16 12:36:58.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* kolla/docker/grafana/Dockerfile.j2","accounts_in_message":[],"_revision_number":1},{"id":"c1b405387a8567def083cc95309529d4cbad584d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2026-09-16 12:40:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5088ce190b14a6c58bcd61baef60f997ac21e1cf","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2026-09-16 12:41:47.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"85de569b13c7bb1f4c00a628e3469a8f6bf2158a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-16 13:07:10.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"858de12d3de585d1f955d49dfeb5462ab1d31763","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 15:48:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/890f3c514ee0453e8908cb5a510a3f97\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/4ebec8d2ad844d72bbd5b354387da2c1 : SUCCESS in 1h 00m 53s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/5822a923865e4ec78dfbf6c842a50b66 : SUCCESS in 1h 39m 41s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/a1289efbb4c7478dbb3e981a3bfdf073 : SUCCESS in 14m 47s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/1b91638499b549efbde3a009a1a61125 : SUCCESS in 1h 05m 39s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/33406ecdbf214a56a2772e78d5b9e796 : SUCCESS in 58m 50s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d0487cd3e3f4b4587e1976fb8595e62 : SUCCESS in 1h 15m 37s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/359420bb46de43209c33f7a23cb41fdc : SUCCESS in 31m 18s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/24a86d647920437d9bc97020b6706f55 : SUCCESS in 1h 24m 03s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/064aaed56ad44bf6951e299a0266a704 : SUCCESS in 1h 34m 42s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/e615ac3040c74a99af31283bb530e5d6 : SUCCESS in 2h 39m 47s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/7b2e8deb12b247d0a708586c89d10791 : SUCCESS in 1h 28m 00s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/2d1538f40e134c1a97f47f5a329467b7 : SUCCESS in 1h 54m 46s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/43cf054543454794964fe8f38333932e : SUCCESS in 1h 47m 55s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/d37874a7a0cc4eeeb4652e4076e7144e : SUCCESS in 2h 28m 29s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/adbe5e2633544d56b386e1477f9911f4 : SUCCESS in 2m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33a257ef3cbb4a65ba449926d1fa6afa : SUCCESS in 11m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/00ad5f410ca047dfab3d416ef200afef : SUCCESS in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ae2378cc12ee433b9fab1a8861291f75 : SUCCESS in 3m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e5ee903f25294e23a4546d89d41b61f1 : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5309d11e11f04f94991f3b16ca68bb60 : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/147356c0cac84ff0814fe7948823fc45 : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c0a108a90b0445aae3787c8da2fe1ee : SUCCESS in 9m 25s","accounts_in_message":[],"_revision_number":3},{"id":"df637176f41f6dc4f22b133e6d22ed9891ea72b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 17:25:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b90f810bfe6a4bf3b35eec1188cdb18f\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/fe79582d7c924a8585ca116141eee51e : SUCCESS in 2h 52m 11s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/8e70a91c6a544e73b52fd3416ea1c9ff : SUCCESS in 2h 42m 53s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/c0e3d1073a5c42d5be6586504d9b0ca8 : SUCCESS in 2h 45m 35s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/ba11f0885e654d7f8fb8194a0d0ac766 : SUCCESS in 1h 16m 10s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/a3f9f69450a34d03b454318338d4f8f5 : FAILURE in 1h 09m 12s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/f41e82bdc90f496aa7da59fcd5484e76 : TIMED_OUT in 3h 02m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2882b93bff9d215cfcf6759c4d42f3ffa099a083","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-09-16 20:14:31.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"54499451f253951368a6ce5e928776b6c3d4e189","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 20:14:52.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ed6a2a08ee572aeb6727918052a8dc9d42020f35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 22:16:49.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d29105fdedd842529b318cd6f34e9583\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/483d16eca21247f5b7b4782d3806ca2e : SUCCESS in 1h 01m 45s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/a055d3c298de48539885ebc5c1b5b1aa : SUCCESS in 26m 37s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/cd6015a67d8042d4a33156fd62515abd : SUCCESS in 1h 22m 00s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/dba98c246e0b4beb90a94f90193cae34 : SUCCESS in 38m 04s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be990fb0232540c7a3b30a859cb0154b : SUCCESS in 1h 12m 19s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/30630cf38f824cad8c53ed6ffe48933a : FAILURE in 18m 16s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/43be3c66b5a5499581ad542b083cb56d : SUCCESS in 1h 27m 50s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/8d3e84a5f9074b5299cb0eca59a07419 : SUCCESS in 1h 56m 43s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/bb18bb5a0d0244748f93c885c5e716f1 : SUCCESS in 1h 09m 21s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/d8b7e9e7e9bb4b7583c815fee4137704 : SUCCESS in 1h 37m 38s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70e945fd8802404fb13760040318ba93 : SUCCESS in 1h 27m 17s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/144a8af5e46843ac977f5f319f948558 : SUCCESS in 2h 01m 22s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c4ebe6b0ec264bbfb57022effb588167 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7febec9a668749fea425c2ff6a272708 : SUCCESS in 2m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/671e7d797f194c2ba0c7ba98b66b5a3c : SUCCESS in 4m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e370b7ea2ff8456392d07053158ab300 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d0ce55910df47df8ee93c6561f46ff5 : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1a93d56cc534fd89046ec8abbb58d13 : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":3},{"id":"782e03bdfdcba03c0093936718818e56af8a6908","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-09-16 22:27:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"47006c907de6b7e6ef6ea1a09e650f8ebaada6d1","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-16 23:48:06.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"679cc343c58421b3a88ca3da7e8f755c28fefe0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 00:24:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/031bf1e835fa4a769a596cb1bd567d09\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/fb91b98973e54f24a83b4181ea82faf1 : SUCCESS in 1h 12m 36s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/f3f6d7a832234310b49dd12f057b60c0 : SUCCESS in 1h 26m 03s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/1a54a24b3e0d4503969d6f0d77cadee5 : SUCCESS in 35m 44s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/084f5e52c3444a7d91c517007d4c3e9a : SUCCESS in 1h 11m 18s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/26d6d6452962453783a15c8dc1131dc2 : SUCCESS in 42m 58s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f4d64b864620446caa733e953b43a710 : FAILURE in 1h 13m 58s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/31a41d9fcc584bc59e13668e306ab4a8 : FAILURE in 36m 13s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/153a074e489e4a16ab411c795bb1bba0 : SUCCESS in 1h 11m 29s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/0c13e82bada344ffbbb9e3006d0a832d : SUCCESS in 1h 24m 43s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/7299a13963a843fa93a860c969b38197 : SUCCESS in 1h 55m 52s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/188acf093ec64dabab240941cbe323ae : SUCCESS in 1h 11m 18s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/514a441b9a404f07aa6311ad9001a138 : SUCCESS in 1h 36m 12s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/279c99c30b1e4d6bbea043a8665094fc : SUCCESS in 1h 19m 34s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/fd63629ef62b47cdb939b6521c61ad3f : FAILURE in 1h 36m 45s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/395529300dff4399ae88f3c883b92b41 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d7dcf1c63f247ecbb3737e47b2d6cae : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e7c6f268d9e54669b3233c01203cb068 : SUCCESS in 2m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aa1f9631a91545fcb80a077640043536 : SUCCESS in 3m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4bc8787b68de4b25886ef0788297b912 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b7e0ad6d0f242ef956462dd9252bd4a : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11a4f2656dae4f11bd2da8706d6b910b : SUCCESS in 6m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a8a932eb8804bc984a6a7d817dc8b56 : SUCCESS in 10m 53s","accounts_in_message":[],"_revision_number":3},{"id":"103b1ecb198e37e8fa5b7c43b2b6efca36c7d86e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 03:42:43.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ae43d3ec61548ecb29a99ec117d02b4\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/a016b49202fa4cb88365883763feca01 : SUCCESS in 1h 23m 42s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/0135351ce6db4f0aacfbfa8004615f11 : SUCCESS in 1h 20m 27s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/431a8f087e5942dca98b8450572f12c0 : SUCCESS in 2h 33m 55s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/b7591a4c0570475aacaf770d839c1f22 : SUCCESS in 1h 31m 16s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/7604d544217148e3b0f4f532b4634040 : TIMED_OUT in 2h 32m 36s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/17d8a1f66eaf413a87e7c066766993f3 : SUCCESS in 2h 13m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"745f923e8320837a9e892e50926ba5a48debbdba","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-17 10:10:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"767824a113940ce55c73f0f0415bd8deea7e06a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 14:08:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff0f92869e29436587679146a477f2df\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/362fe9e754c64ede85fafbe8fb8424b1 : SUCCESS in 1h 40m 56s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/7bec22340a2c432eab06b30f0be56a15 : SUCCESS in 2h 03m 50s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/816cf746fa834dacaf30ff4509d8f1ce : SUCCESS in 2h 44m 34s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/57e3707ff7bf474f8c6219a871183bb5 : SUCCESS in 1h 36m 09s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/9c21b84ca212497da65cdaed3dd681cc : FAILURE in 2h 18m 19s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/84386ff1d0f34581bf120ee943d6b826 : SUCCESS in 2h 18m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"44464c8819785e34a75c1bb3839815318c9b5bb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 17:21:26.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/135fe1dd9402480eb2f8f98b72cbf039\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/c54926d6d2dc4c1fa67363f4f423b594 : SUCCESS in 1h 23m 40s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/a50a7e8634c54af287f8e99c51585db6 : SUCCESS in 57m 01s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/ef043b65baa24f42b978bd6791c21615 : SUCCESS in 28m 49s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/8c89833d56eb478da1d2dda7ae8bbfbb : SUCCESS in 1h 12m 53s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/03cb54d107f1433a9a893aedde5e7c13 : SUCCESS in 42m 45s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de0495c823dd49ae8b900f0d6ad9eb11 : FAILURE in 1h 18m 05s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/df91666c5e1c41fb8c9d247f6da7c8e5 : FAILURE in 25m 46s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/f20d8f8d08f24a0f85da74b8db7f8894 : SUCCESS in 1h 10m 49s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/78d6e9d6460b40478ee42a31b16db981 : SUCCESS in 1h 23m 24s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/589f2310120f43a28e6658dcace49bda : SUCCESS in 2h 01m 04s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/804284af6c61465d98fc777c0ab9fbe9 : SUCCESS in 1h 09m 29s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/422a0ce1407448308744e0e234794c58 : SUCCESS in 1h 24m 29s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ef777482d744fccbcfb1a694073b5c5 : SUCCESS in 1h 25m 48s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/e707a6165ac84fca9d52ac9ad36c3318 : SUCCESS in 1h 35m 26s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/5425b61d361c4a7a841594a5f76be55e : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69c1b5bfd8be47aa890c94b7ca5e9c78 : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/235b2784190e4a7384ee0e9b1094009b : SUCCESS in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/61ae4415f3d64f53a364b25cbbe0ee84 : SUCCESS in 2m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fcb651369c204f91b4ea21a36ef1197a : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af003ab8c63c46ac966faccc4ecd0055 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8abc120917cf4c11933df51e637c4900 : SUCCESS in 6m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c2333dc4bd847e498aa9bbb4bede3a5 : SUCCESS in 9m 59s","accounts_in_message":[],"_revision_number":3},{"id":"d9f6ef2ab1de5cb17b4df551b3eaaab5708b3025","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-09-18 10:47:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"35550415340a6fcec60ced00514d71888dcba1a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 12:48:38.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/afb97528896c40dd9fbee79068b1f94d\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/7cca8d2961c54363824f2a28692db630 : SUCCESS in 1h 12m 57s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/67bd7d8a8ff14db782066c1633bbbb24 : SUCCESS in 1h 22m 17s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/dc968758040c46338c27c5f52e16bb3f : SUCCESS in 33m 19s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/866f379bcd7f443996d2833afe6ba333 : SUCCESS in 1h 18m 54s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/37bfcb2e5fc04333a1a8fdc13a90dda4 : SUCCESS in 1h 00m 33s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b3b8a17ba8a4a689e48d1a682305db2 : SUCCESS in 1h 16m 19s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/24a7741dd4eb4e6989bae3e96a5febae : SUCCESS in 30m 06s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/0654495c45b8414a871e5d6a15c1221d : SUCCESS in 1h 09m 55s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/5cab986530474e92878d454199312a46 : FAILURE in 19m 51s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/888731c650034d70bfe4571f81ee4d5a : SUCCESS in 1h 55m 13s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/00e2b8ec12394c82b1e62953b4d1fea9 : SUCCESS in 1h 12m 17s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/c857239581f74d76b1da0ec1928d99e7 : SUCCESS in 1h 38m 10s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1baeaaddc8d14dcea7efa9c7471cdbcc : SUCCESS in 1h 21m 46s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/acba2ece7cba440089855397083984d7 : SUCCESS in 2h 00m 06s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e00dfbfb05d34019a258543c6e9bfb1a : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e334e66ca3ee4a27b39be746993cc85a : SUCCESS in 5m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6029c0bed7414718a63118745a2af1b7 : SUCCESS in 3m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e13b4121075c41a1866f766862e73c51 : SUCCESS in 3m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/02427d5b90fe4b1da742fb9f18bc50cf : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/873742667ee9493e9f35118ca8bd1648 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a08936f5382a4fa89c9b26a549776602 : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6850f7b1b62b4ec682727ef1ac4addf2 : SUCCESS in 10m 49s","accounts_in_message":[],"_revision_number":3},{"id":"622f033b5d6ef32b7614ecb409fe002035d0ad84","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 14:07:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf70163fc93e4417985b9180bf324511\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/5aa436f20dcb4aef8280e2ec6bb570ad : SUCCESS in 2h 11m 23s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/26145564603e442293c40bda2fefc363 : SUCCESS in 1h 58m 03s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/909b802ceb834262938df63dbaec2d26 : SUCCESS in 2h 11m 56s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/fc6ec4dc34b34133be9605e9633883db : SUCCESS in 1h 58m 26s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/d532fdf844dc489592d795ce14b9dbd5 : FAILURE in 2h 05m 19s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/c61cb557565541299f0eda2398b8862c : SUCCESS in 2h 18m 58s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"415b5b4633939df0be4b1259bcddf31d65b186b9","revisions":{"bfe56890fd264e84e57c5c137103d11364f65573":{"kind":"REWORK","_number":1,"created":"2026-09-16 12:36:58.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/71/1005871/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/71/1005871/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/1005871/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/1005871/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/1005871/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/71/1005871/1"}}},"commit":{"parents":[{"commit":"c4e23082aaf2329ffd1b1ddb3992a881aa53af3f","subject":"stable-only: Update Prometheus to version 3.5.5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c4e23082aaf2329ffd1b1ddb3992a881aa53af3f"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-08-31 15:02:48.000000000","tz":120},"committer":{"name":"Alex Welsh","email":"alex@stackhpc.com","date":"2026-09-16 12:36:58.000000000","tz":0},"subject":"grafana: install Prometheus datasource plugin","message":"grafana: install Prometheus datasource plugin\n\nCloses-Bug: #2165921\nChange-Id: Ib4a0d7185ccc64900b4068dd74995dbf1ba6823c\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n(cherry picked from commit aece83cc507ac587769ae61d011161d462dafb1d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bfe56890fd264e84e57c5c137103d11364f65573"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bfe56890fd264e84e57c5c137103d11364f65573"}]},"branch":"refs/heads/stable/2026.1","conflicts":{"base":"201438c78025e1364cd0e874560de14065bee428","ours":"c4e23082aaf2329ffd1b1ddb3992a881aa53af3f","theirs":"aece83cc507ac587769ae61d011161d462dafb1d","merge_strategy":"recursive","contains_conflicts":true}},"318f794e61a9ea21a18e1096fb0c982a773bb976":{"kind":"REWORK","_number":2,"created":"2026-09-16 12:40:45.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/71/1005871/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/71/1005871/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/1005871/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/1005871/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/1005871/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/71/1005871/2"}}},"commit":{"parents":[{"commit":"c4e23082aaf2329ffd1b1ddb3992a881aa53af3f","subject":"stable-only: Update Prometheus to version 3.5.5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c4e23082aaf2329ffd1b1ddb3992a881aa53af3f"}]}],"author":{"name":"Alex Welsh","email":"alex@stackhpc.com","date":"2026-09-16 12:39:53.000000000","tz":60},"committer":{"name":"Alex Welsh","email":"alex@stackhpc.com","date":"2026-09-16 12:39:53.000000000","tz":60},"subject":"grafana: install Prometheus datasource plugin","message":"grafana: install Prometheus datasource plugin\n\nCloses-Bug: #2165921\nChange-Id: Ib4a0d7185ccc64900b4068dd74995dbf1ba6823c\nSigned-off-by: Alex Welsh \u003calex@stackhpc.com\u003e\n(cherry picked from commit aece83cc507ac587769ae61d011161d462dafb1d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/318f794e61a9ea21a18e1096fb0c982a773bb976"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/318f794e61a9ea21a18e1096fb0c982a773bb976"}]},"branch":"refs/heads/stable/2026.1"},"415b5b4633939df0be4b1259bcddf31d65b186b9":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-09-16 13:07:10.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/71/1005871/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/71/1005871/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/1005871/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/1005871/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/1005871/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/71/1005871/3"}}},"commit":{"parents":[{"commit":"c4e23082aaf2329ffd1b1ddb3992a881aa53af3f","subject":"stable-only: Update Prometheus to version 3.5.5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c4e23082aaf2329ffd1b1ddb3992a881aa53af3f"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-08-31 15:02:48.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-09-16 13:06:37.000000000","tz":120},"subject":"grafana: install Prometheus datasource plugin","message":"grafana: install Prometheus datasource plugin\n\nCloses-Bug: #2165921\nChange-Id: Ib4a0d7185ccc64900b4068dd74995dbf1ba6823c\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n(cherry picked from commit aece83cc507ac587769ae61d011161d462dafb1d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/415b5b4633939df0be4b1259bcddf31d65b186b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/415b5b4633939df0be4b1259bcddf31d65b186b9"}]},"branch":"refs/heads/stable/2026.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
