)]}'
{"id":"openstack%2Fkolla~1005872","triplet_id":"openstack%2Fkolla~stable%2F2025.2~Ib4a0d7185ccc64900b4068dd74995dbf1ba6823c","project":"openstack/kolla","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"35264":{"account":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"last_update":"2026-10-01 08:56:37.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-10-01 08:56:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib4a0d7185ccc64900b4068dd74995dbf1ba6823c","subject":"grafana: install Prometheus datasource plugin","status":"MERGED","created":"2026-09-16 12:42:38.000000000","updated":"2026-10-01 08:57:49.000000000","submitted":"2026-10-01 08:56:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005872","cherry_pick_of_change":1005871,"cherry_pick_of_patch_set":2,"meta_rev_id":"4c927806549b18acb9a6d497371cc1d0a65c97be","_number":1005872,"virtual_id_number":1005872,"owner":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-10-01 08:56:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-10-01 07:30:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-09-16 23:47:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-10-01 07:30:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-16 13:09:07.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"state":"CC"},{"updated":"2026-09-16 14:10:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-16 23:47:52.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-10-01 07:30:09.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"9a72bb516c04e088cec3fdc043fd618aad72b08d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2026-09-16 12:42:38.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2026.1.","accounts_in_message":[],"_revision_number":1},{"id":"cc51fd41ec42113b6f851b2c9f773c8c7511c3b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-16 13:09:07.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ec106b8a5da5375c339ed2b9098dce15000e506e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 14:10:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c266a7788d0a434d923984cd7640c5d0\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/85f0420d0db940cba4a976ce80df7c33 : SUCCESS in 42m 14s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/42cc9dd6837c4180acaa879fef93ce02 : SUCCESS in 40m 41s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/3c92a3791db1432a848e837366c436e6 : SUCCESS in 34m 55s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/c7dbe6e9d16c4ec892fdc9893fb7ecee : SUCCESS in 42m 47s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/ca95c0a91b3d476ba4eb71cd8f82c290 : SUCCESS in 37m 51s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/acce336994ff43aaa47d8c62f054084e : SUCCESS in 54m 58s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cf1e85cae643401e96f2f435b7ed171e : SUCCESS in 25m 54s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/5a6107edc3774d19bc186057b30f79c6 : SUCCESS in 2m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/169adb8af97d43ad9158c508f4bacda2 : SUCCESS in 9m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/503924e3cbea49de9b3fadec48e610b1 : SUCCESS in 3m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ca395098845f492e907fec180ddb39a5 : SUCCESS in 3m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3627507e85664e0b8370c48dd6de1943 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eff45acbc6b04cbcb8824ddede3cdbcc : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fffaeae00c5d446189c5fb302a09b4a6 : SUCCESS in 6m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/047b5c7f2fd24cadb2f835e01cc00545 : SUCCESS in 11m 22s","accounts_in_message":[],"_revision_number":2},{"id":"38817839b0e214a2cbd2d969a11f357900622a7c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 15:37:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0940391a401a471b8539869c40097acd\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/2b312fbc681648fb941af95937ce5f5f : SUCCESS in 1h 14m 31s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/09d8f1f652644c29ba3a72e5d740dfb0 : SUCCESS in 1h 16m 58s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/833bfc48f7884a378b2b8afd68812e78 : SUCCESS in 2h 05m 36s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/7909a0f993634e7c9e402b3da63ec661 : SUCCESS in 55m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c00aa3410972b57a5e5554e9cde6526f47f62abd","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-16 23:47:52.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"210ec8fbf457235b0dcd6175663f60693b4da43e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-10-01 07:30:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a6ba9bcecce36e436b6589ad32609d6d156ae7c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 07:30:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"13dd35d36e71aed3bc708ab70063277fb04b0279","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 08:56:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5073b0a605104fb5a24876b6a4a13516\n\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/0d523832a22e46819d5207bc05723d4b : SUCCESS in 47m 11s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/f76c1d117f10425f9d86688436533aa7 : SUCCESS in 36m 02s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/adec0ed945b848dab82376ff5e99efb5 : SUCCESS in 40m 35s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/cfc8b0df103748498881832f6f3d4627 : SUCCESS in 37m 34s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/91541c2373ec4e5e9eb014e63cb3c7bf : SUCCESS in 51m 56s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1b1ac7f340c149688ad43a4b24ab3488 : SUCCESS in 14m 53s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d0a7e31cd6cd4b5cbac1583e64fcbde4 : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a158d22ee3764b098ad10060138636f4 : SUCCESS in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4c26198509ab4943aa99410ffd874031 : SUCCESS in 2m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bcaeb42240df4eec8ff64bd62bbf9311 : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e646d8896634375b24adff3efd0764b : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c3a92378f5443fa8fb75b5083877f48 : SUCCESS in 10m 05s","accounts_in_message":[],"_revision_number":2},{"id":"9486c3574abc7e39c6c39018d04c8d2de3783af0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 08:56:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4c927806549b18acb9a6d497371cc1d0a65c97be","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 08:57:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29bf9ebed14849579e20866eabb22831\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54a3b3bef53a47958c0efc56ccbe58ce : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c723601894d847118bd1ce45c9f511aa : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4f5656aeb22c337ffd4567646fe470aa3db4636e","revisions":{"c3392f5afd947745bab1d65fede3981a8dce32c7":{"kind":"REWORK","_number":1,"created":"2026-09-16 12:42:38.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/72/1005872/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/72/1005872/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/1005872/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/1005872/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/1005872/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/72/1005872/1"}}},"commit":{"parents":[{"commit":"7fdda0ca724eeabe28ecfc24f7e2df8a3a980876","subject":"stable-only: Update Prometheus to version 3.5.5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7fdda0ca724eeabe28ecfc24f7e2df8a3a980876"}]}],"author":{"name":"Alex Welsh","email":"alex@stackhpc.com","date":"2026-09-16 12:39:53.000000000","tz":60},"committer":{"name":"Alex Welsh","email":"alex@stackhpc.com","date":"2026-09-16 12:42:38.000000000","tz":0},"subject":"grafana: install Prometheus datasource plugin","message":"grafana: install Prometheus datasource plugin\n\nCloses-Bug: #2165921\nChange-Id: Ib4a0d7185ccc64900b4068dd74995dbf1ba6823c\nSigned-off-by: Alex Welsh \u003calex@stackhpc.com\u003e\n(cherry picked from commit aece83cc507ac587769ae61d011161d462dafb1d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c3392f5afd947745bab1d65fede3981a8dce32c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c3392f5afd947745bab1d65fede3981a8dce32c7"}]},"branch":"refs/heads/stable/2025.2","conflicts":{"base":"c4e23082aaf2329ffd1b1ddb3992a881aa53af3f","ours":"7fdda0ca724eeabe28ecfc24f7e2df8a3a980876","theirs":"318f794e61a9ea21a18e1096fb0c982a773bb976","merge_strategy":"recursive","contains_conflicts":false}},"4f5656aeb22c337ffd4567646fe470aa3db4636e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-09-16 13:09:07.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/72/1005872/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/72/1005872/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/1005872/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/1005872/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/1005872/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/72/1005872/2"}}},"commit":{"parents":[{"commit":"7fdda0ca724eeabe28ecfc24f7e2df8a3a980876","subject":"stable-only: Update Prometheus to version 3.5.5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7fdda0ca724eeabe28ecfc24f7e2df8a3a980876"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-08-31 15:02:48.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-09-16 13:08:13.000000000","tz":120},"subject":"grafana: install Prometheus datasource plugin","message":"grafana: install Prometheus datasource plugin\n\nCloses-Bug: #2165921\nChange-Id: Ib4a0d7185ccc64900b4068dd74995dbf1ba6823c\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n(cherry picked from commit aece83cc507ac587769ae61d011161d462dafb1d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4f5656aeb22c337ffd4567646fe470aa3db4636e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4f5656aeb22c337ffd4567646fe470aa3db4636e"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
