)]}'
{"id":"openstack%2Fkolla~1005873","triplet_id":"openstack%2Fkolla~stable%2F2025.1~Ib4a0d7185ccc64900b4068dd74995dbf1ba6823c","project":"openstack/kolla","branch":"stable/2025.1","attention_set":{"35264":{"account":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"last_update":"2026-09-16 23:48:20.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-09-16 23:48:20.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ib4a0d7185ccc64900b4068dd74995dbf1ba6823c","subject":"grafana: install Prometheus datasource plugin","status":"NEW","created":"2026-09-16 12:42:54.000000000","updated":"2026-09-16 23:48:20.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":1005871,"cherry_pick_of_patch_set":2,"meta_rev_id":"c3334c27e3f88ea97938412aa5e6180a057c67ab","_number":1005873,"virtual_id_number":1005873,"owner":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-16 15:07:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2026-09-16 23:48:20.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-16 13:10:32.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"state":"CC"},{"updated":"2026-09-16 15:07:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-16 23:48:20.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"b80150de0275f13634d07c2fec73949cb2aa149e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2026-09-16 12:42:54.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2026.1.","accounts_in_message":[],"_revision_number":1},{"id":"b42b75343f7ddb7010604d84755961bf63f60fc3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-16 13:10:32.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"48aad1f00971f683133793d37412b5b03c500409","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 15:07:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/236d2ced3f9049859ce28c936b75d929\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/4fa2cfc93796462a91d4f3b47f251f7f : SUCCESS in 54m 07s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/18f082ef3bae4a29acef7ed1a1e51357 : SUCCESS in 39m 22s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/0c6d1efda40a4397a33706d208846c09 : SUCCESS in 1h 50m 18s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/83a17e491c674b7fbfc39d15ca77c059 : SUCCESS in 1h 32m 57s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ef70aa1b9843450498d728de9844de2d : SUCCESS in 39m 56s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/fcf22ba5d08d4f95af8048ec4ffd6c93 : SUCCESS in 30m 37s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1016a560b5234643bcdca3d5b117f138 : SUCCESS in 1h 32m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2aaf550d7cc84d67a0155d06107a58a7 : SUCCESS in 11m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62f92bd887d148e9aea6f9f988ff2f88 : SUCCESS in 3m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0457cd75f0cb40eea26c6966814fa723 : SUCCESS in 4m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b7d148888514eb2945a3fcf81ba534e : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19e41e8c445b4090a7be186f7473899a : SUCCESS in 5m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/262ce3203afd4daaa610137c4392ef83 : SUCCESS in 8m 30s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/18487ff9f438410092f2bc460f1539b9 : SUCCESS in 4m 01s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/dbf17b7feba54ec9ae7e1171fd761229 : SUCCESS in 1h 21m 18s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/f91a103bea584a5eab1b7134ca5df7cb : SUCCESS in 54m 32s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2d9d72ae27b64e718922dd41660481a8 : SUCCESS in 1h 13m 48s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/8cc7470824684011a8986202bc0fc354 : FAILURE in 56m 41s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/2300e80bd43c4fd5b71369e821e79b1b : SUCCESS in 45m 52s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/3ee63c6bfe94441bbd3fab2fb7af9e6b : SUCCESS in 1h 10m 12s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/c45ef8151c484ba888b753af6ed52f87 : SUCCESS in 1h 14m 50s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8bfd1eeaaa634abc80b890cad1f40a6a : SUCCESS in 46m 29s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/637141af8f2b4ab49180824ffbb0b537 : SUCCESS in 29m 27s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4eab7567c07b4f5e97efde27e5c21278 : SUCCESS in 1h 36m 44s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/63ed08a0cd7f4607a7626730f938f1e4 : SUCCESS in 58m 26s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/900859a77a9c41d887d7ac8d71b23e00 : FAILURE in 1h 32m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"745b6421b32fb7ded3640a90f7d1afe80d713d85","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 17:09:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6dd6ba4cc3c74e2d844f20a9a32fbc2c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/023b6afae63b4b9dbe6b76663c6515a4 : SUCCESS in 1h 08m 52s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/cd464e3611884465ae6a43a8731b4314 : SUCCESS in 1h 47m 18s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/48fa6be5d72543adaef1821bd387e9c1 : SUCCESS in 1h 56m 01s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d58dd9d067e84a81bd3c5607ae57f901 : SUCCESS in 1h 15m 15s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/c1cb4159a84b4672a4f857d67f3b6d5c : SUCCESS in 1h 38m 48s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/9cae7879fcfd42bbb0ebfa5f2818832a : SUCCESS in 1h 33m 20s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9c58585b6ed149ed806c4665d0dee4c6 : SUCCESS in 1h 29m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c3334c27e3f88ea97938412aa5e6180a057c67ab","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-16 23:48:20.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c4b53a28e878e4bedc2a3a9c05d685859a90d087","revisions":{"c737da6578f15d1c105b7d282c433f4a4ff4c3aa":{"kind":"REWORK","_number":1,"created":"2026-09-16 12:42:54.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/73/1005873/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/1005873/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/1005873/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/1005873/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/1005873/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/1005873/1"}}},"commit":{"parents":[{"commit":"d14cef9bbafa0db561abfb0c0299d1d6bbbf8f0c","subject":"Merge \"Don\u0027t build releaser image on CentOS Stream 9\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d14cef9bbafa0db561abfb0c0299d1d6bbbf8f0c"}]}],"author":{"name":"Alex Welsh","email":"alex@stackhpc.com","date":"2026-09-16 12:39:53.000000000","tz":60},"committer":{"name":"Alex Welsh","email":"alex@stackhpc.com","date":"2026-09-16 12:42:54.000000000","tz":0},"subject":"grafana: install Prometheus datasource plugin","message":"grafana: install Prometheus datasource plugin\n\nCloses-Bug: #2165921\nChange-Id: Ib4a0d7185ccc64900b4068dd74995dbf1ba6823c\nSigned-off-by: Alex Welsh \u003calex@stackhpc.com\u003e\n(cherry picked from commit aece83cc507ac587769ae61d011161d462dafb1d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c737da6578f15d1c105b7d282c433f4a4ff4c3aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c737da6578f15d1c105b7d282c433f4a4ff4c3aa"}]},"branch":"refs/heads/stable/2025.1","conflicts":{"base":"c4e23082aaf2329ffd1b1ddb3992a881aa53af3f","ours":"d14cef9bbafa0db561abfb0c0299d1d6bbbf8f0c","theirs":"318f794e61a9ea21a18e1096fb0c982a773bb976","merge_strategy":"recursive","contains_conflicts":false}},"c4b53a28e878e4bedc2a3a9c05d685859a90d087":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-09-16 13:10:32.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/73/1005873/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/1005873/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/1005873/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/1005873/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/1005873/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/1005873/2"}}},"commit":{"parents":[{"commit":"d14cef9bbafa0db561abfb0c0299d1d6bbbf8f0c","subject":"Merge \"Don\u0027t build releaser image on CentOS Stream 9\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d14cef9bbafa0db561abfb0c0299d1d6bbbf8f0c"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-08-31 15:02:48.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-09-16 13:10:25.000000000","tz":120},"subject":"grafana: install Prometheus datasource plugin","message":"grafana: install Prometheus datasource plugin\n\nCloses-Bug: #2165921\nChange-Id: Ib4a0d7185ccc64900b4068dd74995dbf1ba6823c\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n(cherry picked from commit aece83cc507ac587769ae61d011161d462dafb1d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c4b53a28e878e4bedc2a3a9c05d685859a90d087"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c4b53a28e878e4bedc2a3a9c05d685859a90d087"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
